)]}'
{"id":"openstack%2Fswift~972988","triplet_id":"openstack%2Fswift~master~I6e21eb94ababb01d1676f7e4aa9064a454eb0597","project":"openstack/swift","branch":"master","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-01-12 13:58:23.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6e21eb94ababb01d1676f7e4aa9064a454eb0597","subject":"gatekeeper: ensure X-Timestamp is set early for all requests","status":"NEW","created":"2026-01-10 00:29:11.000000000","updated":"2026-01-13 17:14:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f78d2d51b2f1851ee9b5991fff5d20e357e41bf4","_number":972988,"virtual_id_number":972988,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-10 01:26:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2026-01-12 02:57:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":-1,"date":"2026-01-13 17:14:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":-1,"date":"2026-01-12 13:58:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-10 01:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-10 01:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 02:57:36.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-01-12 13:58:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2026-01-12 13:58:34.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"86150edfe40c82e0c220a293b8a4e37189682062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-10 00:29:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77c0241063c69da76c5d5709a302b2e456fd8b56","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-10 00:30:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d68ebdd5d4db4a9962bfca40f63951d3f0d6e706","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 01:18:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cdad3f4376844f7b99a3e2b4218aa33\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d65dcadba034acf8f2ef7d99c915bec : FAILURE in 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"756e95275c00888d488fe94cbbb2a1222b44d056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 01:26:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cb536c148644b5e9849f8309e60cb46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6790d97994fa4d56b9db3ba85ddb988f : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/88d3c6facfe14651aeb58b2e8fc47a09 : SUCCESS in 34m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/685c5048d1b44833a3fda9286eb98dd9 : SUCCESS in 34m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a773fcc4d3b4a69a29438f5b1589149 : SUCCESS in 32m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acdc16897f8d4612aeabb18853dac5a8 : SUCCESS in 35m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a719ccee4184c91beef59e1c514da08 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8586bb789dc5462d8d66f7035e642eaf : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9791f6eecbf4836897b3969e05dc81d : SUCCESS in 14m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcf3c478a67f4f88b2c83537d613a125 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ee10af2b89ce4c2e8817e815c5f1a8e4 : FAILURE in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92e02802ecb441b39c5fd932866478bf : FAILURE in 10m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2d551dd2f9e84ed2865454784ee5f29e : FAILURE in 18m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5095c64c8bc64c3595a5571d08d700e3 : FAILURE in 29m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/988abdd3fcbe433589b3598b48aa420c : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f55030c2d7b4ca7bcea32fe13845679 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7012e2d1b10b4624b97159bd0dd861bb : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9217211234874ae5922fe0d63b127b2d : SUCCESS in 5m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93b68b1750764e31a88690bba8e58a64 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b01d68dfa43c4a5e8e8e258d58aa0c39 : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50b2c5f23c6f42fa9d00b454efd00ca7 : SUCCESS in 33m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/edabb9905f0a4d0ea0d59aefea434287 : SUCCESS in 55m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e3bcfc4a0d2473086d4894e342c213b : SUCCESS in 28m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5270353eb8d94a5599b4f65d7c4dc310 : FAILURE in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d61ac9f398f4a9c8681257baaa0d93b : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2a57f0dc90d4d54bd87d46f122ec1f8 : SUCCESS in 18m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f9cea699f5a4da89104f47f22b57b6a : SUCCESS in 32m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83876cdd973c43f0866efc134bcaf880 : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"735b338e190ffdeac040b619d51043d18ce686e1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-12 02:57:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThere probably should be some tests. Including if an existing X-Timestamp in shunted or not.","accounts_in_message":[],"_revision_number":1},{"id":"09f7e946bc49eafe0db25e269e93bc3dc9fa9b5c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-12 13:58:23.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"e5559a9e2cca89332630997abcba187e709f5f12","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-12 13:58:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a3c83ef9cf02844b58deb3c270f031f7e57596ae","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-12 16:47:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cc37876642b3491aa74fd23c78e8f517f91dd0f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 17:14:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f78d2d51b2f1851ee9b5991fff5d20e357e41bf4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 17:14:35.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b232da1cb1ac9be2685e6e17dc0b5126fe3e83b","revisions":{"1b232da1cb1ac9be2685e6e17dc0b5126fe3e83b":{"kind":"REWORK","_number":1,"created":"2026-01-10 00:29:11.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/88/972988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/972988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/972988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/972988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/972988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/972988/1"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-01-10 00:28:23.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-01-10 00:28:23.000000000","tz":-480},"subject":"gatekeeper: ensure X-Timestamp is set early for all requests","message":"gatekeeper: ensure X-Timestamp is set early for all requests\n\nChange-Id: I6e21eb94ababb01d1676f7e4aa9064a454eb0597\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b232da1cb1ac9be2685e6e17dc0b5126fe3e83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b232da1cb1ac9be2685e6e17dc0b5126fe3e83b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
