)]}'
{"id":"openstack%2Fswift~973116","triplet_id":"openstack%2Fswift~master~Ic3ea785dd902c51bd7a112716f0f98f012050dec","project":"openstack/swift","branch":"master","topic":"remove-replication-count","attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2026-07-31 21:39:28.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-08-03 07:39:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-07-31 19:29:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-03 07:39:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ic3ea785dd902c51bd7a112716f0f98f012050dec","subject":"DNM: remove CleavingContext.replication_count","status":"NEW","created":"2026-01-12 16:44:49.000000000","updated":"2026-08-03 07:39:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":41,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b2ed66fe7c6c2de6aff9785d37d29806d8de9bb5","_number":973116,"virtual_id_number":973116,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-31 22:46:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2026-07-31 21:32:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2026-08-03 07:39:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2026-07-31 21:39:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 17:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-12 17:53:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 21:41:58.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2026-01-12 21:45:19.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-07-30 18:11:35.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"REVIEWER"},{"updated":"2026-07-31 13:20:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"a5e1ff6d73ae882a8f988ef480ab75af5d380739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-12 16:44:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a36f6e3585872015c45bf8c519aaf4dae7e3c7a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-12 16:48:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06210eec6bf34cf0d1be41533c4c0a4496b21498","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:22:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e84f4da5ea4d4b9d8942120420fdd0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbeb81db348c4316b573dad2de34f358 : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0c9f757b8453cc6cffd57975152b4dca8349c80a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:53:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf4ab9230654507ab197eaad8cfb676\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9cc11ddd294a70a16243ef76c0ce8c : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b5f4ef56f44a4d8db90382e20bab6b9a : SUCCESS in 1h 02m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41d9c0d619b94d1d9393fa9fc63dc597 : SUCCESS in 31m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/779b2843acf94bf7b6d1a73987c5dc6c : SUCCESS in 44m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e40e050f22ce4f1bafe0b636c5e562ee : SUCCESS in 1h 02m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d442ca2a4f2642588e772637cd5470f5 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26c9a6f9b098435c8e6e50af971dc942 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a81101acd3af48948dc6c3ec227975d8 : SUCCESS in 11m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a743a66af040447bb625d4feb4b63fa0 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04e94a7f79db467ca6b5c59a4eb8d95a : SUCCESS in 24m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b01d5047641f49f9af126cf23285727c : SUCCESS in 17m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/65ae169533a04d5faa10fd8320928d56 : SUCCESS in 27m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3222248347284387b5d851ec1173a887 : SUCCESS in 21m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c4b2a691b6a4c5ca1a32a352a321da1 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/734018448b2745bb8e6a997f8c30634d : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/07d44b1ebc92455585b6f6e3f429febe : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/017b5c5c71a643548abe196a14ca6d3a : SUCCESS in 5m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/229f705450864185aebe487f7ae6a7bd : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e55e63cddcc342578950d92520a31c4a : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a7cd8f26d8d4ff7b0a519325df99d81 : SUCCESS in 34m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de23b93e8fba4539a559a8e1b74a829d : SUCCESS in 53m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/86276a028f7849dd8fe6c690c489ad52 : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50896089aea049fb9b6804055bb9dd95 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674ec565615f456db88f3c58a47ae3d6 : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93659abefa13428197546b6db5e25265 : SUCCESS in 11m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6336d4e875654c85a997a76693ac0d2c : SUCCESS in 37m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c357da882dfd477cb7164701654d36ec : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"88f31a3d61ff67b1fd29ff2af5b8e654a6b1c88d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-12 21:41:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"75ac7dddf6a3c429bfe2963fcb76c56305b76431","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-12 21:45:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb525289830ffb6bb619a181d3acb64ffc1c327e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-30 17:38:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"58ebe33787685628761d955414c9503169077870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-30 18:06:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2f308c0563bdfdc59f2ee4c7824f0f3fbf2bf4f","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-30 18:11:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f9db1177702db3cf503a49bb0aee2ea84843795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 19:23:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327995e13a3c479aa3dfc3defcbbf6aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd01640fa8f45a6a67d71090c58926c : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fd07b95c9994097a862789681e1f630 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4c613eec61a94fd4bb1b8c6a4be431ec : SUCCESS in 59m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c7799c36e2c4c9eaf1f950a4502ad91 : SUCCESS in 1h 11m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/720e369fc843482fa16fc367233bdac9 : SUCCESS in 58m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68e8f372f1ab4e20b67de66ca63b43c2 : SUCCESS in 58m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01d409a473cd4319a917cbae5be95f69 : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0f1d5d0d0264c8b8450a948d19562e1 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61b27d4a69524aadae6faa89412162f6 : SUCCESS in 15m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86bab911f56d49cfab058a9cad53c3b7 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c65501ea0274047b178ec4803fc398c : SUCCESS in 23m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/91a882e01de3416c8d89f296c71b8697 : SUCCESS in 21m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e93b42ea561d47e0bb12a8c041a28538 : SUCCESS in 34m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2f0d7dd35adc4be5aab1319d790b5871 : SUCCESS in 21m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a51c50b80c7437aa84bd8890ceeb6ea : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/75aa961ddf19463384b32fe934e07308 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f43c8590fa7b4166954504a1c867b638 : SUCCESS in 15m 10s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa034be40e224e749afd6766362988ae : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0a212564ecc6449dae1b6f26226a5c1c : SUCCESS in 7m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6478f1da987435796806808075bf6e1 : FAILURE in 7m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d160e102ecc44eb4b5047fad8028ba8a : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c5f7445f20f487384dd422dc3e3bc12 : SUCCESS in 48m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0cb75b2563bf47dc981d396c0c6d1f7c : SUCCESS in 55m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8148686066040ad94be348d32f6edf0 : SUCCESS in 57m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d08db2f4e67b40ffb6465edc00e351e6 : SUCCESS in 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bea85d3b06d4ed5b4efb5e754ef58bd : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6f7515d6b454c238fa9187d3fe149bf : SUCCESS in 15m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/703f39e85c4c41c280a51cdf69370bb1 : SUCCESS in 29m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"b35f526a98ebc893c3b8a3e9ce26f12a90c2c424","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:23:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79163ecf02094abdacc38af696ace11a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/858d41028dd746b899f4fc9a2d927bd2 : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":3},{"id":"89cba4c0c0fbfd5ff41eb92d9bbe03b313850d26","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-07-31 04:45:29.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"710c8f1c651243aa1e4d63f4964c09cf8e3bdf70","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-31 13:20:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3e495a3155864d71621a69b8e9d338ac1ecfa7c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 16:20:38.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6b65f14cd8d15d750947d127c6bf813dd013ad71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 16:20:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ba81c300fc83af10341519394fc46194f439fb0d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-31 16:32:36.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2dfa51105343bfbaa4a48462165e920249f36344","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-31 16:41:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d5179c4759a507c28eb8db039410fc2ae82ba907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 17:02:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"25865b0c332df190bc7c0281d0e82c260aa7d382","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 17:31:12.000000000","message":"Patch Set 6: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"180d361ffd58c19c0440b35a5340bd4d8405e803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 18:01:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f97bdf4b2e4cb99ad12154999a57cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79170ccf226c419b8ae83668fac39124 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e355f37bcc14ceb86d2dab9f4360d25 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/d67906b884344c3295f1b4ad673ae565 : SUCCESS in 37m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fcc0c7c225434bcca5f7d64ef5f27de7 : SUCCESS in 39m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2c3a66019134d0ba1c5e8f4726d62be : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d88ffc4d0c444e7a9da2f893f0cf9727 : SUCCESS in 29m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12d0b902c9464489a82d22ca2852d1c9 : SUCCESS in 22m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/887d7ef702e34fb2a21e0e017587b64f : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92a48b77413b4224babec3cca435ae06 : SUCCESS in 22m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9ec2d2d832447e8a6df2bb3e5fbf16d : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6442b18dfefb4db3ac21211c10dc83fa : SUCCESS in 25m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8cb65af19a12486fa4a107d94d608327 : SUCCESS in 13m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1e51adf884b24eb8b97ac8c4d196afdb : SUCCESS in 31m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5c1efdf8c8c5462c84ce08a1fd6f0c71 : SUCCESS in 21m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/85a0a5fdbe034f56b1f950951b1bb3ff : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eb5335e7f01e4ed0970f08a5169b8a57 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5c6d30f0e564b2b85eda2f669c33a15 : SUCCESS in 9m 01s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dc31282a9364a5db5f6c790219e84ef : SUCCESS in 12m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96c7a7c2b68540c38de9dbfe5c24e984 : SUCCESS in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a8fd140d62548dbbdf6acf51c17bb2a : FAILURE in 4m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/669f8371e37e478a87d905d9f48226e4 : SUCCESS in 5m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9945bc68a12c48d7bff245d2c29b1980 : SUCCESS in 53m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8cb6be9a3004dfdb69b814af68976fd : SUCCESS in 50m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/699a4ab3fcf14d7981181ae049681063 : SUCCESS in 43m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b075a72cc9a4bb390c1550102198ad0 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eddb44156ec449ba869a1984e758f56 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b71267d6b0b4883a127b663763c752a : SUCCESS in 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/298bdffe64694317a0166a84b6ff1b2b : SUCCESS in 53m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"3460edb7cd655f94d0e7f5f7db8cc25f1a54d672","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-31 18:15:51.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0bfc8b2d9e499d871842ebeaeec6e29aba7f4c3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 18:42:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311be7b5d233489788ecde9402e468b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/157d4e444fd541e7b7557ee9c4d45d47 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":6},{"id":"a1d72fb6389a64bdc45f73f144ac622f026aa715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 18:56:23.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"cd191dac4247742222e356b8e6a91cd165afef45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 18:56:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0655412ea067f20a3febfada27af587fa69bd721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 19:19:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c1f3905aca5978f356e12f09cc0b5db60500d1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 19:26:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0ea1c7d71f5b71c9108d6a76259e3ba6724ed937","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-07-31 19:29:10.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c82782c2b9b762e1d47a3b7ab68a41ddc39d0d29","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-31 20:34:01.000000000","message":"Patch Set 10: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f6e15d4189bde21b878b719ed6158981a2a1eeeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 20:41:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa54b0aacac407ea43da944527f13d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623ba696b97a4717b9c8fcb37e858e4e : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2bc463907ca43889620e20051988f50 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d897dc63aab54bb1a1a41608b3a1031a : SUCCESS in 38m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00d8114feaf943ecbbfd9a2f5098dda6 : SUCCESS in 1h 03m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fda8b72120324496b825d4f864a52afa : SUCCESS in 50m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df55237396aa4de0a27e9a0f36348982 : SUCCESS in 35m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6023e00a0f084f26b15d354db4474c5c : SUCCESS in 22m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00329ec2e5b6484aba51fb0a321ad852 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/476c57e5c75942758676f181cd4d31fb : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64a07d62f4d7401cbf6918ab9ab0ff1f : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fc1b7b2ab8274d55bddcc1ad7f928411 : SUCCESS in 26m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/968b7d2854ac405396fc527246376807 : SUCCESS in 12m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0e811d4e31594bf1bbc25f87118b642a : SUCCESS in 21m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9d6fe69f224a41ac9126b7941847cc5d : SUCCESS in 21m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9d5a65ea100431babcd311aaa7c4cb4 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6365c26d7d8e4e6ebbd4fcf4dfb8bcb0 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d099e5a3869d43b89be2c0b1eb015a67 : SUCCESS in 15m 39s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc690e9f40ea41e9ab7fe20b58e30305 : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ac3d7d6e8f249c1afb1abcb3f8e749b : SUCCESS in 5m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c33134cd1c04ddabb2ac868e17019e0 : FAILURE in 7m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fbb525589e2c47c0b4efaf2e1fc5cd3a : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c464af95a274b949d29ffc25f9c48a4 : SUCCESS in 1h 04m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0d5e33d23134ceb88075defc73ed06b : SUCCESS in 29m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45ba3616e64c44ff88f7b06ef17fa959 : SUCCESS in 58m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf13df434b204d78a120af667bfa4641 : SUCCESS in 32m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad3a4e4ad6146eca871f8e823205403 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c52a6a48c3d742a5b126a43d7a9c2d43 : SUCCESS in 18m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2ec4450a4704e70aa037bb9796927ae : SUCCESS in 1h 07m 51s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"a76f877f5dc54bc2a5d6efea7878dead20410775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 20:41:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e6b6aae3e0b4f6da8e394b29646c67d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/735091a685a1426f94feb417425b784e : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":10},{"id":"5efeb0ca3a5186d23ec2fd56ec62b92a55253425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 21:31:20.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"49659ed7d0fbe597a26eb7a9c9de4cdf069b5acf","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-07-31 21:32:35.000000000","message":"Patch Set 11: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c9f8f1a197424098e51c5defda00e2e26e492a83","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-07-31 21:39:28.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"de5e615c45ebd29bc2c173c802e309107f79967e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 22:38:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/491306d5d58844ddb1c536319e6fa8ff\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1331b5afaf54969b3b1e7bdddccb5ca : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":11},{"id":"29f2906ee686ca264320431f644220fd51e2b93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 22:46:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e7e3c47940b4e0598870e5545c5b138\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c5b391aaa0454c990e42d38ee3ce6e : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db7ca923bed14485b012bf54061fc254 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/b31661e92a564414986974f36b1f58a6 : SUCCESS in 31m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35ccd35b777a4459b19b495ac835b896 : SUCCESS in 1h 04m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ab2cc91bd734a1fbebb2415358085d3 : SUCCESS in 33m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e92769ee1fbb4e15a8ae715d5f1d0436 : SUCCESS in 1h 04m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b07089169d8044c6bb046501ff541acf : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d95d729cbb66492e9921778a43ea60ad : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/544c279cccf340528381dcd1e73dc937 : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f203e50f3f304cfeb6d8dbc27c08cc1d : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b6523489bfe646798ece337ded87952c : SUCCESS in 26m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3245a4c5be72405084e3f8f69e1c2faf : SUCCESS in 18m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1b7935dfd51444e941f44d1633ef1ed : SUCCESS in 32m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/be2e10bfc22a4ee49a6ed2a7ac286118 : SUCCESS in 22m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/42bffc4d66a8485284f895ad0758883a : SUCCESS in 8m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d776c573d0b747cdb33d73d66fffd1b9 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/90839b5888bf405fabc53db549f3f60a : SUCCESS in 16m 24s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a14603f68a184b458c8b538f2fc49252 : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38f6260cad54484c8b0ce2d0acfe8c25 : SUCCESS in 8m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ceb3c927ffdb405f80b6fd29ab0ca7eb : FAILURE in 8m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/766398167067486e8a1f2997d0fef7fd : SUCCESS in 9m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b93cad0975043f6bd8fed7949a4d895 : SUCCESS in 1h 01m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e449944f21ce4afcb595668281bf2341 : SUCCESS in 59m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f84565a5e4b433a9260ad93e94a7931 : SUCCESS in 51m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83617c46eae147afb143b8159a0e4f40 : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d57410a89547e58f1d3259b9324f94 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7ae042569dc40ac9b396bd65fb14046 : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4438f42ccae4e1d939d4107e64d0eb7 : SUCCESS in 59m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"b2ed66fe7c6c2de6aff9785d37d29806d8de9bb5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-03 07:39:20.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"929fe198ef33476404b8d3609fb99097fd921b55","revisions":{"250ad4c5e8f4e3f3a3e2dc74fb5db4f2269131eb":{"kind":"REWORK","_number":1,"created":"2026-01-12 16:44:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/16/973116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/1"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:35.000000000","tz":-360},"subject":"DNM: remove CleavingContext.repliation_count","message":"DNM: remove CleavingContext.repliation_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/250ad4c5e8f4e3f3a3e2dc74fb5db4f2269131eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/250ad4c5e8f4e3f3a3e2dc74fb5db4f2269131eb"}]},"branch":"refs/heads/master"},"ba709b6e0a7b1f776d31f0dedfcc3cf282d39ffd":{"kind":"REWORK","_number":2,"created":"2026-07-30 17:38:57.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/2"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-30 17:37:21.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba709b6e0a7b1f776d31f0dedfcc3cf282d39ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba709b6e0a7b1f776d31f0dedfcc3cf282d39ffd"}]},"branch":"refs/heads/master"},"5e1cfded35733f32e6e55619eb3b9441c2923b6f":{"kind":"REWORK","_number":3,"created":"2026-07-30 18:06:56.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/3"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-30 18:05:40.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e1cfded35733f32e6e55619eb3b9441c2923b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e1cfded35733f32e6e55619eb3b9441c2923b6f"}]},"branch":"refs/heads/master"},"9cb4fd74e606aebd4d7e9b99bbcd44166afd1112":{"kind":"NO_CHANGE","_number":4,"created":"2026-07-31 16:20:38.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/4"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 16:20:27.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cb4fd74e606aebd4d7e9b99bbcd44166afd1112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cb4fd74e606aebd4d7e9b99bbcd44166afd1112"}]},"branch":"refs/heads/master"},"b97875acbc88069c20ad7405b866bc33a041e289":{"kind":"REWORK","_number":5,"created":"2026-07-31 16:20:52.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/5"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 16:20:47.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b97875acbc88069c20ad7405b866bc33a041e289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b97875acbc88069c20ad7405b866bc33a041e289"}]},"branch":"refs/heads/master"},"52c7a4d055e799edef13fa779b2dcf2286f03797":{"kind":"REWORK","_number":6,"created":"2026-07-31 17:02:46.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/6"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 17:02:40.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52c7a4d055e799edef13fa779b2dcf2286f03797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52c7a4d055e799edef13fa779b2dcf2286f03797"}]},"branch":"refs/heads/master"},"3279a707f43d43614cb6d9382e08610af5172ede":{"kind":"REWORK","_number":7,"created":"2026-07-31 18:56:23.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/7"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 18:56:16.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3279a707f43d43614cb6d9382e08610af5172ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3279a707f43d43614cb6d9382e08610af5172ede"}]},"branch":"refs/heads/master"},"52cb85e727651282ba520898c2c29fe084a7d771":{"kind":"REWORK","_number":8,"created":"2026-07-31 18:56:37.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/8"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 18:56:32.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52cb85e727651282ba520898c2c29fe084a7d771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52cb85e727651282ba520898c2c29fe084a7d771"}]},"branch":"refs/heads/master"},"f7e8826ade78413ae1c7ac486b61d1cbc7f45a92":{"kind":"REWORK","_number":9,"created":"2026-07-31 19:19:59.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/9"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 19:19:54.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7e8826ade78413ae1c7ac486b61d1cbc7f45a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7e8826ade78413ae1c7ac486b61d1cbc7f45a92"}]},"branch":"refs/heads/master"},"3aa77e68ee23eb6f20e125e44e442725b86b5a02":{"kind":"REWORK","_number":10,"created":"2026-07-31 19:26:58.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/10"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 19:26:46.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aa77e68ee23eb6f20e125e44e442725b86b5a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aa77e68ee23eb6f20e125e44e442725b86b5a02"}]},"branch":"refs/heads/master"},"929fe198ef33476404b8d3609fb99097fd921b55":{"kind":"REWORK","_number":11,"created":"2026-07-31 21:31:20.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/16/973116/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/973116/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/973116/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/973116/11"}}},"commit":{"parents":[{"commit":"149f8f33e645a2ffb09de7fcd307a83f96877455","subject":"Merge \"Remove duplicate sample config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149f8f33e645a2ffb09de7fcd307a83f96877455"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-12 16:42:03.000000000","tz":-360},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-07-31 21:31:14.000000000","tz":-420},"subject":"DNM: remove CleavingContext.replication_count","message":"DNM: remove CleavingContext.replication_count\n\nCleavingContext instances are constructed using key-value pairs that\nare decoded from JSON blobs persisted in container metadata. If the\nCleavingContext interface changes then it is possible for the\npersisted JSON to have obsolete key-value pairs that will be passed to\nthe CleavingContext constructor. Previously this would result in a\ntraceback and prevent cleaving progress, e.g:\n\nTypeError: CleavingContext.__init__() got an unexpected keyword \\\n    argument \u0027replication_count\u0027\n\nWith this patch, the `replication_count\u0027 kwarg will be removed when\nconstructing a CleavingContext.\n\nTo the best of the author\u0027s knowledge no arguments have been removed\nfrom the CleavingContext interface in any *recent* release.  However, an\nexperimental additions to the interface was carried in a downstream\nreleases, and the arguments subsequently removed from the interface [1].\nThis has resulted in obsolete key-value pairs being persisted to\ncontainer metadata (but only in our environment). This patch is\ntherefore remedial and not necessary upstream.\n\n[1] Related-Change: I4ad1e22fd011652d80269a76353e5fc14db249f5\n    (patchset 37 vs patchset 38)\n\nChange-Id: Ic3ea785dd902c51bd7a112716f0f98f012050dec\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/929fe198ef33476404b8d3609fb99097fd921b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/929fe198ef33476404b8d3609fb99097fd921b55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
