)]}'
{"id":"openstack%2Fswift~973194","triplet_id":"openstack%2Fswift~master~I2d1c0443dc1685e3133a8f720181fc0ae7b11c19","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-15 00:16:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I2d1c0443dc1685e3133a8f720181fc0ae7b11c19","subject":"Update dockerhub secret","status":"MERGED","created":"2026-01-13 01:05:13.000000000","updated":"2026-01-15 02:34:23.000000000","submitted":"2026-01-15 02:31:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973194","meta_rev_id":"b19f3b7f78481e8810928a961d61d5a6400951b7","_number":973194,"virtual_id_number":973194,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-15 02:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-15 00:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-15 00:16:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 01:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af08e234ecb9ad358a7eb13bb025a619f5eca3f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-13 01:05:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0145da45e5a737d538b947b8f075c1a93353c71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 01:05:53.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch feature/mpu.  The\nproblem was:\n\n  Secret swift-dockerhub does not match existing definition in branch master\n\nThe problem appears in the the \"swift-dockerhub\" secret stanza:\n\n  secret:\n      name: swift-dockerhub\n      data:\n        username: screamingfrenzy\n        password: !encrypted/pkcs1-oaep\n          - ruMizg1iVvKm4ABLQ8GshZMwt3EzxOyjPZsInL20+ZS+TQxhEwRbLFGzSxnrChIOdioyl\n            7TMW1PxQeJ5T/mPIsV7TBsSsnIMKYRcDSbKjnC0hjILpKfQXLFw4/rV/d3jeB6oLDSTW1\n            fIt4NmJqhsjlvst+VwT1JnFHLdrRaGMWYkjRU8rEmH82jDM7Wk7J+selykvTrlRQ7RpQR\n            6huzniL6PJPOZ7I5VsQcCmEWYKwd/u9Ifhe50yjgxmKR7Fi+wl0nBSOzt38f9ZEXTB6So\n  ...\n\n  in \"openstack/swift/.zuul.yaml@feature/mpu\", line 486","accounts_in_message":[],"_revision_number":1},{"id":"27d3c0ab4609151c87473e0e50c63eb7bf0b3005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 01:06:25.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch feature/mpu.  The\nproblem was:\n\n  Secret swift-dockerhub does not match existing definition in branch master\n\nThe problem appears in the the \"swift-dockerhub\" secret stanza:\n\n  secret:\n      name: swift-dockerhub\n      data:\n        username: screamingfrenzy\n        password: !encrypted/pkcs1-oaep\n          - ruMizg1iVvKm4ABLQ8GshZMwt3EzxOyjPZsInL20+ZS+TQxhEwRbLFGzSxnrChIOdioyl\n            7TMW1PxQeJ5T/mPIsV7TBsSsnIMKYRcDSbKjnC0hjILpKfQXLFw4/rV/d3jeB6oLDSTW1\n            fIt4NmJqhsjlvst+VwT1JnFHLdrRaGMWYkjRU8rEmH82jDM7Wk7J+selykvTrlRQ7RpQR\n            6huzniL6PJPOZ7I5VsQcCmEWYKwd/u9Ifhe50yjgxmKR7Fi+wl0nBSOzt38f9ZEXTB6So\n  ...\n\n  in \"openstack/swift/.zuul.yaml@feature/mpu\", line 486","accounts_in_message":[],"_revision_number":1},{"id":"3ae7edc5ac05862fb75bc650252ba57d803ba3e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-13 01:14:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2cb63785bb8f649c1291c903795b67c2b89c8986","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 02:12:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7f99a0e8dcf4e8787dfd7611147ad17\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/703a178b6b5c411da8bc70694d8f9b8d : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":2},{"id":"073fe18922e09111b0f3c9da711ba4d940a13a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 02:20:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72c3e8bee6144bd8713abb596d0df34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec89e5fe17f4bdca0c3839390934ad9 : SUCCESS in 6m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/f727b9f1bcb749d791bc4fb05a719572 : SUCCESS in 33m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/824d0c435cf745f7857bef0b043c43fe : SUCCESS in 33m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b52732f581054890a604be9f0d9ccc64 : SUCCESS in 35m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2b78cdc103b4cc0b64fda8f66c1861e : SUCCESS in 48m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b7eac5c80514071b086589aaf247fef : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/903ac06b7f1040feb47f4aa3c7680a18 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73364f37ec1a42e6a445183bd6cb2057 : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/199456d72861475b8bfbf85f8a47b5e0 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f67ffb9feca548d8aede688a10d70d2a : SUCCESS in 24m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d88dc2de19f34ca49b2a5d8e9eb10008 : SUCCESS in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/30529c2ccf6b4d6fa63ac588ec4f354c : SUCCESS in 21m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/afcdf9a66aed4c42b1f74fa5a4820860 : SUCCESS in 12m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7102c194c2944e94bc9da276d1efde62 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dd5bc403f99141e48fddcb17c0e364d7 : SUCCESS in 7m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad22299cf55f4806a6360a027b9872b7 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3cd9ecc856ff48e4b796a74749dfaa3d : SUCCESS in 3m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b7c8e984f19847059446a2946dbc7046 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/94635e4492f94680b5290de8abb7d0d0 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c20e5a76b73f4a778092f5434652993b : SUCCESS in 55m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77bb3f32d54c4adf857fdd7a07375be8 : SUCCESS in 24m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9099ec9dbed44f8fb6bfc5c23a9a981d : SUCCESS in 47m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/652bb75617ca4e1a8474f4247b0dca7f : SUCCESS in 24m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d8f1a6e8fa4819b3bece84913aaa55 : SUCCESS in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ffa2620df5f54eb68b2f1783dbf9ca63 : SUCCESS in 18m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1d9977cb8f44e3cb98f1ea9719f550e : SUCCESS in 34m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e03b454ec8a54b22aea0961828885385 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":2},{"id":"77a8ceec9dcb3469653e9fe98dd03253e0fa521b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-13 04:24:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0bd174956da6d41d70d8704be238a23fa541104e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:25:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87ada4794e84fa06fbd7cb60c8e05da224c68dc5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-13 05:28:54.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bd487713095c1dd4307b2c84a77f1714aaddca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 05:28:58.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"be66ceb3ae7f6683524eb725e1e5c0d4c4145707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-15 00:15:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fdbf47bce66f0e644fa8f917ddba2c877e1c4895","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-15 00:16:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6595c059937a88df53fbbf3474c381cc7aef2941","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 01:17:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c1884051914e15b9336a45200900b3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5ae8d27fa8740b3952b7a7fce6e5f16 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":3},{"id":"bbf03f49311ae08a47706f49a275b842a01a1331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 01:23:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8473ae3abf0c4d479e6577a29e10ea90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0967fef1262e4671a0de829c3ea273ec : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/5c3a735d94124b6f9b1cf8fa6434f1b9 : SUCCESS in 32m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51812c4c0d654d6c8cb583a8abbb674d : SUCCESS in 1h 02m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d4adbc1b3b1f4b44b15426b38ae8461a : SUCCESS in 50m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a572676cf4124138be39f547f6577df9 : SUCCESS in 1h 01m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb5041f310034466a3c839a270aa5952 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a30b76835a24a4d85950e067c414346 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0712d892abe84f94b5d3089863b5d4f3 : SUCCESS in 14m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce08be34bd7f4694b96915da6b77dba4 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9da7a619ec14c1bb577817a7924cba0 : SUCCESS in 27m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/602245c74a7d4c13b91dcf649cf46858 : SUCCESS in 20m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1d198b3056694254be5bced9654d5f66 : SUCCESS in 28m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0177f13de4e94616b14a2a6f868345c1 : SUCCESS in 22m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e538c912e7d4ebdbdaa6396c1685e37 : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c57f44c80e54ad4977aac7e00edb599 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dfa783ca607b4c0e993667554fa7fb31 : SUCCESS in 11m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cba1ff5431894bbf8519b0dcefd82b6d : SUCCESS in 5m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc552af940f846239e593804256d6df8 : SUCCESS in 11m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4112ed3996a4ff6bc1267144873be63 : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad8218849c59403f91d206850b0d1761 : SUCCESS in 37m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9627a4dc842643ef82d719ca9495ab34 : SUCCESS in 24m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c61adaebc8a2438f9266a7124bb593ff : SUCCESS in 51m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a76902e53304158ade143125cadb639 : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680609a5f2b64ab4b33fe8824c14970a : SUCCESS in 2m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96e9a390afe54bd08c7e8c82de98d120 : SUCCESS in 25m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43ab3fcafb254199b4e88fc5dd23d874 : SUCCESS in 36m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7059400df42a4a24b06cd3d4820e8b65 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":3},{"id":"910e25027beb26c95246b7717766dce3b99800f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 01:23:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"852739428377ecd11076e63d39cfdbacf8e54dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 02:31:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8719e90781124b77904890c7421989a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f2ba3fddf54f6a8326ce0c48c186d8 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/33be6d0f6a38417fa2bd5b1a6c3e94ef : SUCCESS in 1h 02m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/690bc948985047da98b1b539a5d7bcad : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c50ecef45896488883f1f2f910763b50 : SUCCESS in 55m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ff2d61951354736b57013f1fbf8ca49 : SUCCESS in 1h 00m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3d47b84a505c4984ab1cd52d37eaa69b : POST_FAILURE in 5m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d9f68521ca74b0489d2dc9fe370ca7b : SUCCESS in 25m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f994146ec5f471a9d6206671bbba259 : SUCCESS in 20m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/33787ded02854e708814631f9cf36d92 : SUCCESS in 29m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9f6afa5e68d342a5a383007cd8d0e229 : SUCCESS in 18m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0979aa982c084d2a95b09a30ff6c3107 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e29ed016934c4c168b7768a819a15e39 : SUCCESS in 9m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a848f1e6e09e4fb5ba1a3c214b40aa70 : SUCCESS in 8m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c621ddf3c661424a8c3a7fcfeef7d2c9 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/53d3278f50164e109b705b834de0211e : SUCCESS in 5m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/631cfe081ab74dcf980b33c8d8421d81 : SUCCESS in 48m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dab00af4642447da67586f65c9e0a54 : SUCCESS in 57m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c55a5405e8f497fabc364e0ee97aee3 : SUCCESS in 59m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7975d046a7944eb8549fd41db40663d : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7df968214964dc69a9d76ff392e3cad : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ebe2bf5dd8b4edbae62d668464cc641 : SUCCESS in 17m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94aa3ffc57e3465cbad238cb92914ede : SUCCESS in 37m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e7799ccb56f450baacfcdd33352cbd3 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a06630b2fe44b9fcfb88ed5bf49f0b1daffdc0fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 02:31:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b19f3b7f78481e8810928a961d61d5a6400951b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 02:34:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2022b566219e47b3af46f1bc0fe1b915\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7910687e6d764e61a6ad568db4d02b69 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/46145329e9bb4dedbf5970436ed58eaa : FAILURE in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86d1dee2ad13772989e0d0e0ccfe3fdcf7fcf3b0","revisions":{"2831bfdb9815a904fd2c16c401de98137b5cdca0":{"kind":"REWORK","_number":1,"created":"2026-01-13 01:05:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/973194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/973194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/973194/1"}}},"commit":{"parents":[{"commit":"ff79e136704a689f72575277be9c17f205bec8de","subject":"Merge \"docs: Clarify circumstances for using clobber_hardlink_collisions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff79e136704a689f72575277be9c17f205bec8de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-13 01:04:58.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-13 01:05:05.000000000","tz":-480},"subject":"Update dockerhub secret","message":"Update dockerhub secret\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2d1c0443dc1685e3133a8f720181fc0ae7b11c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2831bfdb9815a904fd2c16c401de98137b5cdca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2831bfdb9815a904fd2c16c401de98137b5cdca0"}]},"branch":"refs/heads/master"},"def73534d20935ddf1763efb785b163f3f3428a2":{"kind":"REWORK","_number":2,"created":"2026-01-13 01:14:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/973194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/973194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/973194/2"}}},"commit":{"parents":[{"commit":"ff79e136704a689f72575277be9c17f205bec8de","subject":"Merge \"docs: Clarify circumstances for using clobber_hardlink_collisions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff79e136704a689f72575277be9c17f205bec8de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-13 01:04:58.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-13 01:14:35.000000000","tz":-480},"subject":"Update dockerhub secret","message":"Update dockerhub secret\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2d1c0443dc1685e3133a8f720181fc0ae7b11c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/def73534d20935ddf1763efb785b163f3f3428a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/def73534d20935ddf1763efb785b163f3f3428a2"}]},"branch":"refs/heads/master"},"86d1dee2ad13772989e0d0e0ccfe3fdcf7fcf3b0":{"kind":"REWORK","_number":3,"created":"2026-01-15 00:15:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/973194/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/973194/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/973194/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/973194/3"}}},"commit":{"parents":[{"commit":"1971983dece9f889ccb4dfd5dde628727390f074","subject":"test_backend.py: refactor TestUpdateNewItemFromExisting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1971983dece9f889ccb4dfd5dde628727390f074"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-13 01:04:58.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-15 00:10:52.000000000","tz":-480},"subject":"Update dockerhub secret","message":"Update dockerhub secret\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2d1c0443dc1685e3133a8f720181fc0ae7b11c19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86d1dee2ad13772989e0d0e0ccfe3fdcf7fcf3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86d1dee2ad13772989e0d0e0ccfe3fdcf7fcf3b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
