)]}'
{"id":"openstack%2Fswift~973272","triplet_id":"openstack%2Fswift~master~I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-01-27 20:53:49.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-28 21:15:31.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-28 19:53:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5","subject":"s3api: do not set x-timestamp header","status":"MERGED","created":"2026-01-13 16:38:17.000000000","updated":"2026-01-28 21:16:33.000000000","submitted":"2026-01-28 21:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"973272-p-timestamp-collision","meta_rev_id":"abc76a2d73d3f0de3d53f5f2758ad0fc13390a51","_number":973272,"virtual_id_number":973272,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 21:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-01-28 05:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-01-27 20:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-28 19:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-28 19:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 17:30:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-13 17:54:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 23:46:35.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2026-01-15 00:50:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-01-23 00:59:45.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2026-01-27 05:38:30.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-01-28 05:42:12.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-01-28 19:53:44.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"c2e1e9727839c70d6c161948d3bc217d1b6a9bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 16:38:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d0e90b9734605206de4320a48bfa9c023656c9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 16:44:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b3ec801feb4f8f852478f0436b42dc4f75168be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:30:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68e717c8273d41bca3eea269d7a6ce96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6897512217a44ee48ec8e8b0269bdb7f : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":2},{"id":"413a7408a75f57edba890920471fd0a4ffa37a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:54:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb64cb45c4bb4fe98651e435c7f06bf7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98db5c54b1fd4e56b46b60c1e329b716 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2eac5ebfef7d4bc5b00ebfdfb818feaf : SUCCESS in 1h 02m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef1fae66a2ef47fab6d17d45e99e1189 : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a4d3579d9b374832ab3e6d0c96713f0b : SUCCESS in 52m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eeeb7d6bea54470cb9d53f7971b3f81a : SUCCESS in 33m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d3e6ed264d54bad9a8548b3d1755ba2 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4da7ec3eba114a3cb54d8ee2220b9302 : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce81fb9d77b64a38a653280f54bc6d1f : SUCCESS in 16m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e94b439ef4b645778135670ed5950424 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/489906da1ffb48bd96122994f84b8323 : SUCCESS in 28m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a68b06edb31d48968591bc1cc58d255b : SUCCESS in 8m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9e3bf77dc454de193b33fae4e118d33 : SUCCESS in 28m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0164566ba28e4a7ba53a8f6d08b79297 : SUCCESS in 11m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/272523c520cf48debc6e860fd1a4c2c7 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e1de864360904134aae10588d8bd1a25 : SUCCESS in 8m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/74a56b114e86422eb55ca0fa62490f1c : SUCCESS in 9m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfe62526f2aa4ef49b94037a25114a14 : SUCCESS in 7m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4977872936e7471c9a448e43098919f4 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a5aade49aee441aeaafc7b1d70986a26 : SUCCESS in 5m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8874785e2b64b15849eabdde7bd754b : SUCCESS in 1h 01m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77dd2eb9584b424cb7b7829208979e0d : SUCCESS in 44m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cabcd109988749aaa931905b5421d7bd : SUCCESS in 47m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/614b1f107a7e4f1cba8edd4e4c41cfe9 : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffce110d7b36434493fd2e770a7892fb : SUCCESS in 2m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e076c1e24e84a8798dacb28ba7a78eb : SUCCESS in 17m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e67cf2114e5e4cbbbe4730f35db7a059 : SUCCESS in 50m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3cbce751da3471780f00a5df3871b5a : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a3fb04764d794f49b6fe29cc557f3294620ea6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-14 14:51:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"03a9c331589f724bcdb7449e016b71442658c55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:00:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb643e8a33649fab5a2102c90765475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7808fc693a9e41fb91abaf96d7a916ac : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/54a92cb826574f39a70ae2f8cd810e3c : SUCCESS in 43m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/175fc1dce1f64887ba8bf52a86daea4f : SUCCESS in 32m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0dc5012b56446f2a61517d2413c6bf7 : SUCCESS in 32m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa1e1fa82b414f4287e38f346d9169b6 : SUCCESS in 49m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33f7430d733a42e8ab5c7b2f3a16a353 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dd870eeab5d4b25876be3a08a3aebae : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca6ff24a021b427cb8d18747a909d012 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/125439b9e88145daac12963d258f97fb : FAILURE in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88d7c16cf5414065b115ed3bf1f72d42 : SUCCESS in 22m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/121f75aa60ca4cd684ce8666dd55e108 : SUCCESS in 21m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e2906768aa5d413c8367945f95b51fba : SUCCESS in 34m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3f6d7f619edd4feca61b1d04ce05caf8 : SUCCESS in 30m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6afe73e3036c4dd785f8b7f0977c107e : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/24b6cdea79e14e3b9cb5c37cc77d76c7 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/52bf5b8dff4d46318023b3c5e69208c7 : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8be3ec174b674801ab8f34254e47908d : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22c609d183054578a8ed13ac2d468b13 : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e82b555f899452ea064d7aa29a81e89 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/753361a5ae834110bd53b36bce31a6c8 : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5964b50f5c540a59f7a9945f485fa08 : SUCCESS in 50m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0df4f8d8dbf4c42addb9e74df15cce2 : SUCCESS in 27m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa17867e539b421f95f664f50ef3a6d7 : SUCCESS in 28m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e0b4c5204b496b9c0e367972018ce9 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d85fbb040eba4161a74f2277c7b8ba34 : SUCCESS in 17m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8af9ba43ba0499ba3c597464feb5c83 : SUCCESS in 57m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b47996bf177746c390e9d13ed2283425 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":3},{"id":"193cd8ebc7ff711b3c3c66cc92900e0b8dfdece7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:04:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beb27b0fb1414a5c825f89ae5e728587\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c382bf87be0d4d119580d6a2cf77ef2c : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":3},{"id":"7742d501b7da75238128038e7e2fb427ec5df8d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-15 00:50:03.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6b60a63cdea966574e9e2fd68eeb9b7ffd4d8a67","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-16 17:30:57.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f8c77dd2985cdb63192fe8d478b38bd26d74c2d4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-16 17:34:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b1638984d247f7c1b1e2b4c891e340b7d85511d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-16 17:43:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ec94d05144c64fb39c4288a386d7c98bbeb94f33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:36:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5752a5ce68c74dc7838b390b8483bd29\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88db197a33bf4058b5da2faf211cdb9f : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e3548a5528a764f1172dbb18d7c40e7b33581467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:25:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"5a8facc83d046409ff58f35bb158516085668244","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:50:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17f0e63e223749aa8265ee4b7dff7ac1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41913b7783e34d5d9a6bb430c5b48a25 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":5},{"id":"3ca59dd7884cc74a2834fd748436789c000bd8f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:52:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70b39bf061ed4dc9a9d566066c16579e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67ff4634d294239b5cb6ab5e57bc524 : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/c0b449f1b6aa4d57a3f902cd639accc4 : SUCCESS in 49m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b44eb324349346f0a287abd1116290ad : SUCCESS in 1h 11m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c7bb86cffe84705872de69a02dbaf0c : SUCCESS in 36m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0886cbe58f8e49d2824b42ed110504a9 : SUCCESS in 34m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e77dae3b7c74752954763807438247d : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a0ddd3b9d5f4acf8b0b087bcbb75600 : SUCCESS in 10m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef0afd7120774b3db4fe680e3569a6f2 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1664febb7b446feb88cf01a12db06ca : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7927f8a19cc74b259719a25d4ce8b855 : SUCCESS in 25m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f414aa995cc46209950a73ed3a42ef0 : SUCCESS in 10m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a74208a35f744612bead2ff6493db21d : SUCCESS in 24m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/00789146b3d947cf8f0cbcbf149f9705 : SUCCESS in 31m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/719e5df5081f4d44b4a306db52fdd121 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c47a87838a994a188b4eb061edfbbe54 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e658879430147f787c38fedf5c9fc87 : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee044eecb1354ca882c64f64e037212a : SUCCESS in 6m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/862b5e22d7e743db9eec7dd8f90f03e1 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0577d0b281ea4e03a367241c5b702739 : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/960e50e418be44088245ab8cdec5b995 : SUCCESS in 54m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c17c1dbbd56e44678fee330bcf4adbc1 : SUCCESS in 55m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a58698fb69c4a46ac19bb1d77d0fa18 : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55f0d6fba27940dfb9cdd27b74bfa899 : SUCCESS in 26m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1c346ab8d04eb9ab2435c8712d4cb1 : FAILURE in 2m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ffb45bbdbdb4359b4471ebd8619bbc2 : SUCCESS in 21m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce129ff06a9040e09e02571b13e9a3b2 : SUCCESS in 51m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88a2a35586754aff931a081d29c3906f : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":5},{"id":"c4ee18e581f388461a4df7c1eaa5bde3517efd3c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 11:07:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb2708835b56e36354e86a8ad37d8a9efcd67878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 12:17:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b177f82900674c908ec7bf87a2b9cfaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d280912d148e4925a5ed09ed3618a967 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/99aaba4250c442bb9edc0d66d87c8da7 : SUCCESS in 38m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee68d41abe2a45b6bbec6024f5b11c91 : SUCCESS in 44m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b503553e2f3d474d8068113d494af99a : SUCCESS in 37m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffe0e12c3ce24092b1e091a1425ee061 : SUCCESS in 1h 03m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc51b4fca5684eda8a1d9d965436249c : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad2645811c404625b9dfa9396aede878 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05b531461acd431abde0c5c17fa45c54 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/502ad686f60f4ab5b20ecc6746b3b423 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1806b4dcd03f475aa692835431f8d5b7 : SUCCESS in 26m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59fb19274e954f3b83439728a51ba460 : SUCCESS in 8m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00da44cc7b74411faad74f3018d44f6f : SUCCESS in 29m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/549103c647ba4c439e32c4f78598a2f2 : SUCCESS in 13m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/71e9abaa53c044379a6895712017bee0 : SUCCESS in 12m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1cdd2675560641bda7816f98e661052c : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/15f83ddbff96473d9d343f105e6f08a8 : SUCCESS in 8m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f05309923a6640969c0a33b5518f4d67 : SUCCESS in 3m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af6dfaad68c74c0c8d4cb0cc47e3c093 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ecb6f8d15174f07852909f64d4f34a7 : SUCCESS in 5m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/655da928bc6c4b8495f9267770f8708a : SUCCESS in 56m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dc0ff8c4e8f45d292e40aeb22198b4a : SUCCESS in 52m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e565911422f4206bc15d67f42504a61 : SUCCESS in 56m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15354ea1d5a74ee9b910211797b615ed : SUCCESS in 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c3b84f919d48929af3bdcfc781e935 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9dd840ecfa7d490bac55f7f9e5f60e4c : SUCCESS in 14m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d3f82355f604e6f9e3dde2bc32c0218 : SUCCESS in 53m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a666aa1f8634a98ab6cec7fdc278d5b : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":5},{"id":"1498bbc5fba34fa3c1ea006758144a3eecfb349a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 12:20:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7100586d3f04efaab414a199681478d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d03c6ff7b1644738b0063bce39fef5e6 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c35556486cb08639ca2e204750ed19eac59d7d57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 17:05:12.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c994a78dd9f4ffe749f08cbd68efea9665f8ccc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:57:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61fea8a36ca8454283088ccbd332f2e0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fcaea01ec23242008c07acf8c72420e3 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":6},{"id":"d30cdaf741bd468b721bbc9e3d1735c0526e0081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:31:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dc45cfa8655414cb0da566148983bd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1c1f1c10234c9a8dd1646ad8f6c621 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/6c307ae667c748f18d87a45b6b30f60b : SUCCESS in 1h 03m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c0931317dfe44ae9347999b0a80ef68 : SUCCESS in 59m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8559cbf758b442d8bd71505b5dd9730a : SUCCESS in 41m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01c8f67b69d9490fabb1a9349cd077e3 : SUCCESS in 1h 19m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d854f088d5042539e733150e07a37de : SUCCESS in 10m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b576e018432d4e36bda2c41efa3a2f0d : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/906d475bf19441a5a82c3d654721d30c : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d42ece5689c43f2a9adadae75d1e795 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2481e00255594c0c9fada25c5f8567d6 : SUCCESS in 24m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc8a63162c946c98c80f3b230b8c5c2 : SUCCESS in 8m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eb223178049e4e28a99f848677ef3a74 : SUCCESS in 31m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c4f5525436994386a3ddac30852d28a3 : SUCCESS in 29m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4d8e9e4e1c274c449023b36e5cfbdb30 : SUCCESS in 8m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e86a9703a0774bf09f548c84b16e33c6 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2df1790ae17a40ac8b155de86b6c3c39 : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81cd405b1b0b417fbe0b3fc6a6bca775 : SUCCESS in 7m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1885eb290744a699801666de0709ccc : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/465621cdb50e4137af92705aa9a2264b : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6996c754fec495f87076c23be20e7c9 : SUCCESS in 1h 02m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5db0e90c711947b39e06c9a15358ae89 : SUCCESS in 22m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03bc934c2b134422bcc19ee00c1c6ca0 : SUCCESS in 35m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f6a1c10880d4881b5ee2bc43efda2c7 : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a583ca6ad8694307bfd3eef0d0f9510b : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02ab26d9575f4249bd2a75aafc29a9b4 : SUCCESS in 9m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4c648b0426147ce8c739b923d875fdf : SUCCESS in 53m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6976e93464184556980feec0334442cf : SUCCESS in 45m 48s","accounts_in_message":[],"_revision_number":6},{"id":"d7c8da8cd740de7f8d2e38ce1dc7e5f383813e1a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-23 00:59:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7e8b4fc900a605545022e95985fdabbf9cdfc7a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-23 09:46:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6995b0ab1a789063d5f139ed587ae6d97ec7921b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-27 05:38:30.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3e13e2adb73dabad39166878c942ef280f73ab04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-27 11:58:03.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"571a2c7a3fbb6a7c5752c5252358a7700fb94a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:20:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f326c7431c7f49359bcde348fdf6983c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61eaa98fdff4781abc37f5b4afda298 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/f5b3ea3312494502a1b5d1a9e608c191 : SUCCESS in 1h 02m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ef1652c0c314ecc87609b5f96e2d2b9 : SUCCESS in 51m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c44cd942008845c09444705291035921 : SUCCESS in 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79d497525fe8469297699f00ef1a9dfc : SUCCESS in 49m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faaad91f7f4e455e857306eba065dbe2 : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b780363e6f34873b329d4e18241998e : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0d113780ce54ae8b77e91adce60c0dd : SUCCESS in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36e6f6ca31924718a3b1f863c03b57f3 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb06255e0099438f815418a8b8ccbd76 : SUCCESS in 23m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd00a8834e44bafa52740347705515d : SUCCESS in 21m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6dc9b646951c4e2bbab2112a3048e92e : SUCCESS in 21m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dbb92a594b9045a4a5db365eb02427d4 : SUCCESS in 12m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a6ae2e74e22d46859a180f3d18c01b37 : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6196f4f533da478c9566bd0110cf8d4a : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/45bbc2e28c2348b98faca64d1fe5a582 : SUCCESS in 14m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c70bdf218b254fc68f6e612dcba559d4 : SUCCESS in 5m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/380db38c0da2459f939ed32e653b5ef7 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b97396db29044b07b8b12e5c29a2600e : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dd77970de5d4bb595adcb45ab59f266 : SUCCESS in 1h 02m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9ada083a9514a048c036b7e470053d0 : SUCCESS in 37m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a15693e6ee6842c192040b170f4cef17 : SUCCESS in 54m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46fa172422724e1b99afc101d1aac11c : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c517dd6f1d94f53bf5cac84f460f38a : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c031e1e93f584b1bbf3e3ca53a0e34df : SUCCESS in 15m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e200d767dada4003ac6dc07539b05a1d : SUCCESS in 40m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b732adbb24064fc6a7ca704db2b5b99a : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":7},{"id":"55fb3b012e5765e2ec2c8004d65a78035c966374","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:37:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fac7d57b2ea48a4a29d0ccb92b2de70\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29f1e06ed61e4367b242d77270fdd4f2 : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":7},{"id":"99aecb48b923d783716e9a2c68e48f974a127703","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-27 20:53:49.000000000","message":"Patch Set 7: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e3ff8459385a35e0acc48716505163550b5f8998","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-28 05:42:12.000000000","message":"Patch Set 7: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c86afadf2fd0c680f102e9a9500d5e261225fd82","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-28 19:53:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"be13eda037d6aa207cdef1d12caae2f771354866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:54:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2bd9fa1cb2329349598dfe28d577c2c104c4e79f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 21:15:31.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794bafd1ea01451285d39ccb7ce9871f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e5493e42f148b990ac46e4ca27a939 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/95598956b17e4a58bc6d0d4706fcb555 : SUCCESS in 58m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de4bc313a4304bb8808508f310a789d8 : SUCCESS in 50m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a8da9f0eaae4fbebf9489645d9cc4a8 : SUCCESS in 54m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c33eaf4d962e47e2aea4a1685eb3d85c : SUCCESS in 45m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/082965d6e3174abc9481513632d5cc4b : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff283571295e4fbda6d9273f617dff50 : SUCCESS in 11m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a725af970234113ae372f929cc25a1d : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8a00f9cbf4d54f29b3c6fc7504b7c3ee : SUCCESS in 19m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/22701664b36244d49cc3b88c1e8d2be7 : SUCCESS in 25m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b8e3c76c6e6b409c98c68420676c7412 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/078fe73f4c9c41198b80c62cd7e55013 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/16443ae651b44ed89a75494889e5caf7 : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5180543fd5b4b25a0e31e41612f49b4 : SUCCESS in 5m 57s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccbe6bb6fdf0407784a1e427f2e53011 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2affb83028ab458dba7e625b7ce3004e : SUCCESS in 57m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ab156cac177438798a67302936b5cdc : SUCCESS in 1h 00m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e399d40845d141b8989354df0aee8018 : SUCCESS in 51m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cd905b1b16c42a2b16b8741663370b9 : SUCCESS in 23m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b517ae94e745119e155278fb60b4e3 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd725da20e544a45bfacdfe40c26cfd3 : SUCCESS in 19m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a68052b23b57432a8673d65ce7cd4af5 : SUCCESS in 46m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dadbaf1ebe3497388c2364a0f2d05dc : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":7},{"id":"05bc44d30d837fa5e236b3f496b3e83f5bccab09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 21:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"abc76a2d73d3f0de3d53f5f2758ad0fc13390a51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 21:16:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ad1b38b80b45ffa422084a8cd7ecb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf8eca0972d4018b8d37045dec1bd6e : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/64d7b3f9dc8148a897974b0a711eec6a : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0007353a3985e4b3db38a1a159f51b8601af12b3","revisions":{"4d19487114fdfca86c4e2d8114b5b506fbc7e6f9":{"kind":"REWORK","_number":1,"created":"2026-01-13 16:38:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/1"}}},"commit":{"parents":[{"commit":"949f0fdac8624c1a40381edd1948e82b3092d350","subject":"tests: assert that X-Timestamp headers are valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/949f0fdac8624c1a40381edd1948e82b3092d350"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:57:49.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy request response.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d19487114fdfca86c4e2d8114b5b506fbc7e6f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d19487114fdfca86c4e2d8114b5b506fbc7e6f9"}]},"branch":"refs/heads/master"},"bad2772e1d073d0b62daa36ed4644822383d7b07":{"kind":"REWORK","_number":2,"created":"2026-01-13 16:44:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/2"}}},"commit":{"parents":[{"commit":"949f0fdac8624c1a40381edd1948e82b3092d350","subject":"tests: assert that X-Timestamp headers are valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/949f0fdac8624c1a40381edd1948e82b3092d350"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 16:44:08.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy request response.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bad2772e1d073d0b62daa36ed4644822383d7b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bad2772e1d073d0b62daa36ed4644822383d7b07"}]},"branch":"refs/heads/master"},"de7c88432bcfefbccb6ae161998b375a6898082e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-14 14:51:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/3"}}},"commit":{"parents":[{"commit":"1971983dece9f889ccb4dfd5dde628727390f074","subject":"test_backend.py: refactor TestUpdateNewItemFromExisting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1971983dece9f889ccb4dfd5dde628727390f074"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-14 14:22:22.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy request response.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de7c88432bcfefbccb6ae161998b375a6898082e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de7c88432bcfefbccb6ae161998b375a6898082e"}]},"branch":"refs/heads/master"},"26b610880972484d84e9d6e3f132323a7d7ba246":{"kind":"REWORK","_number":4,"created":"2026-01-16 17:43:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/4"}}},"commit":{"parents":[{"commit":"1971983dece9f889ccb4dfd5dde628727390f074","subject":"test_backend.py: refactor TestUpdateNewItemFromExisting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1971983dece9f889ccb4dfd5dde628727390f074"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-16 17:31:08.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy (PUT) request\nresponse.  The proxy server app copies the PUT request.timestamp to\nthe PUT response.last_modified.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nRelated-Change: I4dacc11ec086fdd2e9bb3f68295f2ebaea68bede\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b610880972484d84e9d6e3f132323a7d7ba246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b610880972484d84e9d6e3f132323a7d7ba246"}]},"branch":"refs/heads/master"},"cced07999fca875375d883b13606a87981a5c587":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-19 18:25:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/5"}}},"commit":{"parents":[{"commit":"72191716d26291e17f46c9e2bd3c0e169a6f3004","subject":"Merge \"Recliam db_dir/*.tmp files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72191716d26291e17f46c9e2bd3c0e169a6f3004"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 16:02:58.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy (PUT) request\nresponse.  The proxy server app copies the PUT request.timestamp to\nthe PUT response.last_modified.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nRelated-Change: I4dacc11ec086fdd2e9bb3f68295f2ebaea68bede\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cced07999fca875375d883b13606a87981a5c587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cced07999fca875375d883b13606a87981a5c587"}]},"branch":"refs/heads/master"},"8a51dc88fd8766a6e58512ed90e0c32a770deafd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-20 17:05:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/6"}}},"commit":{"parents":[{"commit":"64bde7e05c258322864703b55042455d4d3c48ad","subject":"proxy-logging: Add \u0027api\u0027 labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bde7e05c258322864703b55042455d4d3c48ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-20 11:13:44.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy (PUT) request\nresponse.  The proxy server app copies the PUT request.timestamp to\nthe PUT response.last_modified.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nRelated-Change: I4dacc11ec086fdd2e9bb3f68295f2ebaea68bede\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a51dc88fd8766a6e58512ed90e0c32a770deafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a51dc88fd8766a6e58512ed90e0c32a770deafd"}]},"branch":"refs/heads/master"},"0007353a3985e4b3db38a1a159f51b8601af12b3":{"kind":"REWORK","_number":7,"created":"2026-01-27 11:58:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/973272/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/973272/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/973272/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/973272/7"}}},"commit":{"parents":[{"commit":"64bde7e05c258322864703b55042455d4d3c48ad","subject":"proxy-logging: Add \u0027api\u0027 labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bde7e05c258322864703b55042455d4d3c48ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 15:54:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-27 11:57:47.000000000","tz":0},"subject":"s3api: do not set x-timestamp header","message":"s3api: do not set x-timestamp header\n\nThere\u0027s no need for the s3api controllers to set the x-timestamp\nheader in requests. Furthermore, it may be inappropriate for the s3api\nmiddleware to set the x-timestamp header if in doing so it overwrites\na value set by a previous middleware.\n\nIn the case of an object copy, the last-modified time required for the\nresponse body can be obtained from the swift copy (PUT) request\nresponse.  The proxy server app copies the PUT request.timestamp to\nthe PUT response.last_modified.\n\nRelated-Change: Iab68f4b596dfa7f23bc587fbde859d854faff87f\nRelated-Change: I4dacc11ec086fdd2e9bb3f68295f2ebaea68bede\nChange-Id: I4f46ae51579aafb9f511ddec8f2bce5fd5498ba5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0007353a3985e4b3db38a1a159f51b8601af12b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0007353a3985e4b3db38a1a159f51b8601af12b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
