)]}'
{"id":"openstack%2Fswift~973658","triplet_id":"openstack%2Fswift~master~If05f93a74f576b485118944a06a51b09f8fdd54a","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-30 01:38:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If05f93a74f576b485118944a06a51b09f8fdd54a","subject":"proxy unit tests: add more x-timestamp assertions","status":"MERGED","created":"2026-01-16 17:43:41.000000000","updated":"2026-01-30 01:39:27.000000000","submitted":"2026-01-30 01:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"973658-p-timestamp-collision","meta_rev_id":"f1a256e81b70a05a6042c1971c6301e1174b0001","_number":973658,"virtual_id_number":973658,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-30 01:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-01-30 00:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-01-30 00:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 18:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-17 01:30:34.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2026-01-28 06:37:43.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"f2fa8c06ea676b96422fc31fec0e3f35b3335ad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-16 17:43:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ee23aa6c63e4ef4ce616f5a84832d7190d45934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:38:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac986718ec74831bed698e9bd5f5ff4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d26086aede4781a35edf66e10a3ee9 : SUCCESS in 10m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d92f50653e44df58142589f92a00f96 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/846a361096354af694845ab1f420e898 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a8d0c5fd5924ba8b34a910576495284 : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9e59d8602174c89b7075367c1cf9c0b : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3e97792ddac427e92153241aa8bcef2 : SUCCESS in 24m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64cb13a9acd84a3282eb19fa03875e04 : SUCCESS in 21m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8563e48ad584ab29c259ab462a11e9a : SUCCESS in 34m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7293158b394245238b21e964fe221603 : SUCCESS in 30m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/85d73382daf2484da1d7688e101a2d0f : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/713dfafb0ff741b985851730e39c3d48 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a64550f8fa1c40a3a6fc7fcb349da724 : SUCCESS in 14m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/795eda267b274ce18bf10b4a62dac238 : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16ee651f4f284412b7c42d9946d5658b : SUCCESS in 48m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/331709e13764499bbeefb1877a829e27 : SUCCESS in 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e939a1b53d1643839f7c8f6726bf6754 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38ce73dc3ee54e9bb8dc88ceb23dfa7f : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1b9fcd9b235f0dea4b61c88ab1581a655a213d3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 19:58:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f9488736e0455ab713e34a58e1b027\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0b5d9449a5c4f1d8293655454c64dc5 : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":1},{"id":"eba0e7dbab87c4d2c3da4b54c4ef902ac5e5171a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 16:29:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4351c57c978febd1d7bc533ce2151b9ab724f98f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:25:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"902ef5ab6baac8116f126bd7f4eac3c01620408d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:23:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96a1c229fa36408ea44734d85ab715a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf45b07a7134b9eb924ec5745cb2126 : SUCCESS in 10m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d71a9526dbab496ea729041fba6cd78b : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/088934bc5ebb43f881c457430f5893b0 : SUCCESS in 23m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b7d45538d3f4a44b87b32cd63591df1 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3010072dab7d4f8583cddb8e8888099e : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e9df4e376fa4b77b5c68b3d02becc7a : SUCCESS in 24m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5233ac5dec25472683262dad0ddbfd1b : SUCCESS in 21m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fa806ee7ef3c4bb080e611da083ceacd : SUCCESS in 11m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/484a3f56cac340518257ac5ce62a3b3b : SUCCESS in 21m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b60c8f74a749436981033105c8371c2d : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2f4f5c428c464578983d3f81617cd79c : SUCCESS in 16m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd14482c44374eef9461bb9a92767885 : SUCCESS in 15m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e125dfb9704d4eb585252e0e2ca5fae0 : SUCCESS in 41m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec23b2772af6484599a63399e6d79e7e : SUCCESS in 50m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2445ac97f9b4f79a01d4b654447f866 : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd83f387a1e4d78929bb84bfc377264 : FAILURE in 2m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b363dac4551420aaf04a947fd77ae47 : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":2},{"id":"05e33098943e08086675ba4af005b6e2c0d96f72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:44:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2ad700db0554d288a89e0788c7c979c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b1031a80e2a48aabed6f813645757a4 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":2},{"id":"11956dd04c5f78b14357a01bd2e0687fc408bcac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 17:05:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e22b0be07b8d69a1e2befaa4df1831d2a63dc759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:04:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1005d61207f54f4e8caf585c5279b3af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903836c102f345eb80a94f7041dbe1ea : SUCCESS in 5m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b06ffd8fa57a434a9daa1b5c4c11a0d8 : SUCCESS in 12m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c07ef0e0af0e4a7cbc49e162b7991676 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cff4a69342a142fb957f52961ea67b35 : SUCCESS in 12m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/118d1159f07543fca74161c006511375 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5e767b4953514de5a9c11f1d523fa9c5 : SUCCESS in 12m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48b13b2e54114a93acd2130ed202f463 : SUCCESS in 21m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b862f0f868aa42e4a2daf8913990f39b : SUCCESS in 26m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/80319ba5e4d243389f75dd632b3a1678 : SUCCESS in 29m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8fa8ae59349f4bd5bdfe432755ce9243 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/29de9c6fbe954f3eafe90dc15ea85aba : SUCCESS in 14m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/32c3612b47a04a1aa8f0aa7b44cccf88 : SUCCESS in 15m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f3b9ff8925546ee8b98ea3fc273a6d6 : SUCCESS in 53m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/efc9b6fea6ed4e8c9ea6cd6bdde3f810 : SUCCESS in 37m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c2a6c7b32a4a58940a4fc4ed8f6083 : SUCCESS in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72870fe836cf42c98a074873b2d4c94d : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28fea25f23f24563be485f3253f66bef : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e9a306ea7ff397eca5b99694f8b2facad3edfc93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:21:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870674c618e043078043ce9393c8cbcc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da2bb3301af34daea1c3907d563bf1bf : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":3},{"id":"19b3c4a365d50cad9092070ef17f5426cd5c2e78","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-28 06:37:43.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14b6b6c4d030d2e205c395cf0189570abeccbb09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-29 03:56:58.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0fb57f1bf3718db56739899b301bd88372eda3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:52:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/456c98b8567349b69b839f0ce59c6757\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3267e06cb09945809a499bda826409a5 : SUCCESS in 9m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/295baa731e394e6ab2289898bbaa8152 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a182c804bf14ee8bd7aba433b2c041a : SUCCESS in 13m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b9be46a83e1416f86f0f86d5ae90a11 : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab2f9aa15a274f4388058c44094623d7 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/64e7e9be813741789eea5ef4c9ed149d : SUCCESS in 21m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d053b1119ed24ff2a0e4ff01d44829ba : SUCCESS in 20m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a742c9fb220f41f2a691fb1e645e6198 : SUCCESS in 27m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ffd662ff4fff4fadbec4aaff830afdb6 : SUCCESS in 15m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/80922638b3f247f7b862fde1ce3d8812 : SUCCESS in 9m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/37b1403f0db44ae88990978abc2a8762 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/95189c08532343068d766924ba36c2e2 : SUCCESS in 10m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df6aaade04084b61978fed7faf5d38dc : SUCCESS in 28m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6639d2c5d7449feb84bb5cef22551e1 : SUCCESS in 50m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eab2f20d2424219825ba4509e7f38fb : SUCCESS in 26m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d699e7ccdf4835b8bee99147f11e8c : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a7eaaa1ea394ae883dbd00befeb84db : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6974fea5c60273743e046165af1d660135ae6df3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 15:24:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d35776ad0964d848fbdf08875d6ed8f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b20f37ef04eb4492851f73d6a83aa647 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":4},{"id":"d08e3fca7981bdd9e8fdf3c975a35dc67b54b99d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-30 00:44:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4e08ac0ceda0f09ce886383115c5dfd09c60bf9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 00:44:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"88ed51775fc94f7029f501c0d9fadb824307ca14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 01:38:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b38d0c9c5c68459f8cb27d776bb5d036\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02bc8125e0547c8a11cb705afc7f3ca : SUCCESS in 12m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ba7533ae2e7e4aa18ba0a97d0b0937f2 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a52568c6293247828d3153f7671befcd : SUCCESS in 24m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f120c56ed2a476a8cd16a423e5e24c2 : SUCCESS in 22m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8753af682a9a48c1b30abc4c2acb0891 : SUCCESS in 19m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/840c80828a474dbf9266880075bb5bf4 : SUCCESS in 29m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e71abbab15554b58bd957c51be375d22 : SUCCESS in 12m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1a5b64c9a56e456890d3b01461d63a8c : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/44065f8f2c3c4893ba39f2e365be4055 : SUCCESS in 13m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4999158d0674a6f98970f429069ccd0 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4197c35f4524426eb58e0ca70c67ce3e : SUCCESS in 8m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/489b66c3c32041ceb508bb49048cf15e : SUCCESS in 34m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12027f139f1b48b898d0cc47f6324f45 : SUCCESS in 49m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f189a432aabb42acb31eadc51c49d58c : SUCCESS in 26m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd252be2d8a0462b9273c8b83c275d74 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2aa88528c284c68a93452f8aa7b6b06 : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":4},{"id":"e111f749dd26fdf9767143c9916c6e15b7860276","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 01:38:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f1a256e81b70a05a6042c1971c6301e1174b0001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 01:39:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a27e8237f24a95949a09a7308fe907\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc7748671ff4e7dae17be9fe623f87c : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e27d8046c6954baaba84104e6bb13c33 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4502b707e5531562eb9e41334490e7efb183d43e","revisions":{"9fb5bb50285c2b02005520b70a6e3a728bbb9c81":{"kind":"REWORK","_number":1,"created":"2026-01-16 17:43:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/58/973658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/973658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/973658/1"}}},"commit":{"parents":[{"commit":"26b610880972484d84e9d6e3f132323a7d7ba246","subject":"s3api: do not set x-timestamp header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b610880972484d84e9d6e3f132323a7d7ba246"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-15 14:23:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-16 17:40:01.000000000","tz":0},"subject":"proxy unit tests: add more x-timestamp assertions","message":"proxy unit tests: add more x-timestamp assertions\n\nChange-Id: If05f93a74f576b485118944a06a51b09f8fdd54a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb5bb50285c2b02005520b70a6e3a728bbb9c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb5bb50285c2b02005520b70a6e3a728bbb9c81"}]},"branch":"refs/heads/master"},"159ca1a18cf9a9fa6863ea7e483f22d4af4b70d7":{"kind":"REWORK","_number":2,"created":"2026-01-19 18:25:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/58/973658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/973658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/973658/2"}}},"commit":{"parents":[{"commit":"cced07999fca875375d883b13606a87981a5c587","subject":"s3api: do not set x-timestamp header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cced07999fca875375d883b13606a87981a5c587"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-15 14:23:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 16:15:25.000000000","tz":0},"subject":"proxy unit tests: add more x-timestamp assertions","message":"proxy unit tests: add more x-timestamp assertions\n\nChange-Id: If05f93a74f576b485118944a06a51b09f8fdd54a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/159ca1a18cf9a9fa6863ea7e483f22d4af4b70d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/159ca1a18cf9a9fa6863ea7e483f22d4af4b70d7"}]},"branch":"refs/heads/master"},"997a4d310b0dc71d182c93ecab9d526902510b78":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-20 17:05:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/58/973658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/973658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/973658/3"}}},"commit":{"parents":[{"commit":"8a51dc88fd8766a6e58512ed90e0c32a770deafd","subject":"s3api: do not set x-timestamp header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a51dc88fd8766a6e58512ed90e0c32a770deafd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-15 14:23:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-20 11:13:45.000000000","tz":0},"subject":"proxy unit tests: add more x-timestamp assertions","message":"proxy unit tests: add more x-timestamp assertions\n\nChange-Id: If05f93a74f576b485118944a06a51b09f8fdd54a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/997a4d310b0dc71d182c93ecab9d526902510b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/997a4d310b0dc71d182c93ecab9d526902510b78"}]},"branch":"refs/heads/master"},"4502b707e5531562eb9e41334490e7efb183d43e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-29 03:56:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/58/973658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/973658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/973658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/973658/4"}}},"commit":{"parents":[{"commit":"de22f8efb6047b534a334442c423316fd4b474f6","subject":"Merge \"s3api: do not set x-timestamp header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de22f8efb6047b534a334442c423316fd4b474f6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-15 14:23:06.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-29 03:55:27.000000000","tz":660},"subject":"proxy unit tests: add more x-timestamp assertions","message":"proxy unit tests: add more x-timestamp assertions\n\nChange-Id: If05f93a74f576b485118944a06a51b09f8fdd54a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4502b707e5531562eb9e41334490e7efb183d43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4502b707e5531562eb9e41334490e7efb183d43e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
