)]}'
{"id":"openstack%2Fswift~973826","triplet_id":"openstack%2Fswift~master~Ief15bc339cddad2cc28111b37e77c7904d2f48ae","project":"openstack/swift","branch":"master","topic":"s3api","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-20 11:02:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ief15bc339cddad2cc28111b37e77c7904d2f48ae","subject":"sq: proxy-logging test tweaks","status":"ABANDONED","created":"2026-01-19 13:12:39.000000000","updated":"2026-01-20 11:02:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b643bc13b68eb8071e563a306db9a300de379ab","_number":973826,"virtual_id_number":973826,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-19 15:20:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 14:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"600dad8ad08d4ea9583e8918e39c7dd68c1be108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 13:12:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0600d1ddd17ebcfef56af577d51036ddd5895ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:12:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/633f3ace803a4fcaaf505f649f81f106\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44eeb5ac6894efe9628487c3ef2246f : SUCCESS in 8m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bc373b042dd43ab8ae344988938cb64 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53aa80d0fe194618b923ff7befb429f0 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/837484ee441c4d56ad12e2fde6629219 : SUCCESS in 12m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d03b70d0f1e64081b5bd6f4e1b55aa5f : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/49a5f01fd84644dcbb8d048193f1902d : FAILURE in 29m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5521d0c96c414b9590d46312bfd3c01c : FAILURE in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/05816e7f928e4606807f2ce5ff66bfad : FAILURE in 28m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/aebcd7f85ef0464b8f5582d60b6bf4c2 : FAILURE in 29m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/658ac084d589446d915051f360624439 : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7918a2555ff3421c8c7f038cc2490b8f : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1e2b9843c1640a58e0f35c1f14100bf : SUCCESS in 12m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96dedaf4f05e41238cc521a5176b5de6 : SUCCESS in 22m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f5a8bf1028345a4aef96440f150c694 : SUCCESS in 54m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f147a527fa1345c78e977b6559e00ebc : FAILURE in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0509854d9d76445bbf76f8067b33262e : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/183d1260d94147a88b9d8b010d4e12e1 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"160c4fc88f7ef9b386e16e6520ca3d3ecf59e2cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 14:24:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"30563683abf34c671d2851babb34b49a2b052639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 15:20:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93ae5fc09dfe44b4b1f3fe15d86211e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/826447eb076a44bd806ba50b4742473f : SUCCESS in 5m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf22e52b8cdf4fe89a686c8f9b5f37fd : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49b83168d6744329bf6043f49699b249 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a114ee5b6653417eabc638bd854426ef : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2949752c4609431faaf802c929d91806 : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/356412da475f41e48fbc79d947dcc753 : SUCCESS in 23m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06297aca6903449e80d44f0d5b50a3a0 : SUCCESS in 16m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a4201fe32cc407cac3216be262c2307 : SUCCESS in 29m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2ceaaed073db4ce09072e643db814cd8 : SUCCESS in 18m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/daffb9517574410385aafe3be5b77fd2 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dfe54b2b195342c18f19359eedcc11ce : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/347d0f9900644af4ac40781022f9c3df : SUCCESS in 15m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6632206007194b4d97a67a6fb5ca0795 : SUCCESS in 50m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c1069af181141e5a45536aa1b504ba3 : SUCCESS in 47m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee2dd8ce175f4fa28bd2384c768223d1 : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd02a832cf344f3cb6bfa67d0590e9e3 : FAILURE in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b05dd07f137f4607a96cf76e5f8ee28a : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":2},{"id":"944eaaa8c5a576623871b520cc7708bf333d40cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 15:56:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca71849708c24da19e3fdeeb6a46257f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bee8cb4150554c88a44bdcb7a3e12270 : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":2},{"id":"0b643bc13b68eb8071e563a306db9a300de379ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 11:02:24.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fdac785f4c240af48fa9306c635333005697a75","revisions":{"7d085fb56f77ec7bf11acdc440c79ae8b002d463":{"kind":"REWORK","_number":1,"created":"2026-01-19 13:12:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/973826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/973826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/973826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/973826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/973826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/973826/1"}}},"commit":{"parents":[{"commit":"510200509cbb99c11a1e578543a42a11415eacbe","subject":"proxy-logging: Add \u0027api\u0027 labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/510200509cbb99c11a1e578543a42a11415eacbe"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 13:09:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 13:09:24.000000000","tz":0},"subject":"sq: proxy-logging test tweaks","message":"sq: proxy-logging test tweaks\n\nAdd a few more s3 sigv4 specific tests.\n\nModify some test names to indicate they use sigv4.\n\nsimplify import statement.\n\nChange-Id: Ief15bc339cddad2cc28111b37e77c7904d2f48ae\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d085fb56f77ec7bf11acdc440c79ae8b002d463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d085fb56f77ec7bf11acdc440c79ae8b002d463"}]},"branch":"refs/heads/master"},"7fdac785f4c240af48fa9306c635333005697a75":{"kind":"REWORK","_number":2,"created":"2026-01-19 14:24:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/973826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/973826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/973826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/973826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/973826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/973826/2"}}},"commit":{"parents":[{"commit":"510200509cbb99c11a1e578543a42a11415eacbe","subject":"proxy-logging: Add \u0027api\u0027 labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/510200509cbb99c11a1e578543a42a11415eacbe"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 13:09:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 14:24:38.000000000","tz":0},"subject":"sq: proxy-logging test tweaks","message":"sq: proxy-logging test tweaks\n\nAdd a few more s3 sigv4 specific tests.\n\nModify some test names to indicate they use sigv4.\n\nsimplify import statement.\n\nChange-Id: Ief15bc339cddad2cc28111b37e77c7904d2f48ae\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7fdac785f4c240af48fa9306c635333005697a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7fdac785f4c240af48fa9306c635333005697a75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
