)]}'
{"id":"openstack%2Fswift~973927","triplet_id":"openstack%2Fswift~master~If26c961c2133ec48c32a29e37a4a427aa8a4c818","project":"openstack/swift","branch":"master","topic":"p-db-replicator-test-fix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-20 16:21:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If26c961c2133ec48c32a29e37a4a427aa8a4c818","subject":"Fix flakey db_replicator unit test","status":"MERGED","created":"2026-01-20 12:21:05.000000000","updated":"2026-01-20 16:22:48.000000000","submitted":"2026-01-20 16:21:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"973927-p-db-replicator-test-fix","meta_rev_id":"38a6642f8a9851d8ff749ea612270732ad04ffbd","_number":973927,"virtual_id_number":973927,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 16:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-01-20 15:05:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-01-20 14:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-01-20 15:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-01-20 15:37:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-01-20 14:10:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-01-20 15:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-01-20 15:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 13:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 14:07:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-01-20 14:10:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-01-20 15:05:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-20 15:17:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-01-20 15:25:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-01-20 15:37:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a8f19b41cb98b41a88e041a817d190973faa31ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 12:21:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69f5d0c27fc38770e260ef2388f19e020e139f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:18:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4d512f289e4718b4bacb92ab383c56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be2b2a451a84680bbda60c4095d01b4 : SUCCESS in 10m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34569e697b2f4375823072755e71f731 : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64068871fff14091b82c853a8274dd18 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e731a2f9b474f06af8e335a2b63d56b : SUCCESS in 12m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bcce2c61c9474bfeb6b6d1008f8051b1 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/97588111f77f4ca59d553a15fc11b49e : SUCCESS in 12m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/745d3023058c49418120022cd3b8304b : SUCCESS in 15m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d9becc4fef5d48c49e2dd4d8e051a697 : SUCCESS in 10m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/145fc422e11b41cc873484269f519deb : SUCCESS in 13m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/424bf41553f34c6eba119c15762895f8 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7863d7b18d64419fb2aa9a026e25cdad : SUCCESS in 14m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd04f2a477ab4203aceb22b3abb8e266 : SUCCESS in 16m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1dcd080c501140c2aa99f7a8ef14c8b4 : SUCCESS in 24m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adf764a4bf484c5badfdc4675043bacf : SUCCESS in 51m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fb4993ed27e416d9c573f300a79635a : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74cc5e1456ce470a948e80b878de658d : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b58739910cb0454585281e115dcbb58f : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"41e1675aeb23c303be78b33b0c7aa2d60490391e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7484527dbfdb4ad5bc282ba24730a98b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39593bc3e9254dfd9e907750fa55be4a : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b2414549848f15201ae6302ba33f835a446fef86","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-01-20 14:07:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2cde9b581dd58e086306b1d5cec0668667ef914","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-20 14:10:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"358aed432d6f6592f18e42befa58a20b7fe7443b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-20 15:05:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a47a6fdbf50593aa5b4232e8e6a077be4787940f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-20 15:17:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e986aea24de380aa8e3d248612a605cad2182f78","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-20 15:25:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4560760f76bfd2baf9bc63344b1aab0134f69fb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbfc8dc5fd1ab572532e88df9d95c6979e0ec194","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-20 15:37:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8d7351680b9278d15359134b0915722586d2b64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:21:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad3f3c82bd14470cb9c21a516f342915\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ce924b10ce41aa907b8a29e3a47753 : SUCCESS in 5m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/155cf0b87f1a4fae8d12a2ba110768a2 : POST_FAILURE in 6m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bf6175157cb1402083ff8f7a607f6039 : SUCCESS in 11m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4992681c65b4bbb8d6b58b3d7a5c2c5 : SUCCESS in 15m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/804679e4bd584b19aede632845082b97 : SUCCESS in 13m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9ededafd5f4c46da82a2aafd894bc7d7 : SUCCESS in 16m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/438c2dd9fa9e4cdd852fd72b63fd9670 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c6d4c93a06b43818d9c305a9663533e : SUCCESS in 13m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5fa4eafa39cf4b2c8d797f5e7c7af1aa : SUCCESS in 8m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ef09799d38c4abcbb206538e6d921e3 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/828cf4af3bf74ea5a98606dde1c42d2b : SUCCESS in 5m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/315ffb0d25134144b6197eee46ac41b0 : SUCCESS in 39m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/046cf6cba16940a5b3df3bb5facc2cbb : SUCCESS in 22m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac02f5f185534213818207df032d8e80 : SUCCESS in 28m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2135b4b9748478ea093672aab7f9a8b : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5895b04d5e0443808b78a419f7dba432 : SUCCESS in 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"779f43bdafb394d345a32ce1470b2747ba5ad876","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:21:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"38a6642f8a9851d8ff749ea612270732ad04ffbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:22:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ec81128eb2f4b18ac04f4568c735ee0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54fb100ad0dc46319bb977b4af099be5 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/be940facba6d475bbfee0e98e5b8e0f3 : FAILURE in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"732fd859d8e0a0b795fbb42ed8390ab4cd494227","revisions":{"732fd859d8e0a0b795fbb42ed8390ab4cd494227":{"kind":"REWORK","_number":1,"created":"2026-01-20 12:21:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/973927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/973927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/973927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/973927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/973927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/973927/1"}}},"commit":{"parents":[{"commit":"64bde7e05c258322864703b55042455d4d3c48ad","subject":"proxy-logging: Add \u0027api\u0027 labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bde7e05c258322864703b55042455d4d3c48ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-20 12:17:13.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-20 12:17:13.000000000","tz":0},"subject":"Fix flakey db_replicator unit test","message":"Fix flakey db_replicator unit test\n\nPreviously, test_reclaim_tmp_files assumed a particular order to the\nresult of os.listdir. However, no order is guaranteed from python\nos.listdir and the test would fail on some platforms (including\nmacos).\n\n[1] \"The list is in arbitrary order\"\n    https://docs.python.org/3/library/os.html#os.listdir\n\nChange-Id: If26c961c2133ec48c32a29e37a4a427aa8a4c818\nRelated-Change: If30005269d40a1a3f711008b5d3b863efc9fb683\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
