)]}'
{"id":"openstack%2Fswift~973990","triplet_id":"openstack%2Fswift~master~I53445f574953d9d47dd93d6819060e4b3cabee75","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-23 20:58:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I53445f574953d9d47dd93d6819060e4b3cabee75","subject":"CI: Update dockerhub secret (again)","status":"MERGED","created":"2026-01-20 19:41:17.000000000","updated":"2026-01-24 02:32:57.000000000","submitted":"2026-01-23 23:16:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973990","meta_rev_id":"65b52cc3acff58632ffdb461aadf9d4c9024403e","_number":973990,"virtual_id_number":973990,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 23:16:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-23 20:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-23 20:58:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 20:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f633c21e48adc98e99cb9964a304026262049c83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-20 19:41:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e76e5beac23bd85211043b2648b89a569a965b57","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-20 19:42:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e52146c9729a48f1ddafdb4bbfb4fa333bd9f0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:50:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643fcecd7aac4773922f4f546b4f91cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217e898d2d4144a6906a4f3bbd01afc0 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/ec9b0f2a4060444bab713005b6a8eb7f : SUCCESS in 48m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab05927ddb21482faa5eda8e8f00d740 : SUCCESS in 57m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1beafcf79c64ec3a20b8f3efb29cd1d : SUCCESS in 58m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57fff9f713ab42d0addd6c5ede8eca85 : SUCCESS in 1h 06m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5493247cb8e45b2ad4603bfbb6ffca8 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b6202f250f24d16b8352269af563f33 : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3d381ba36c94f40aed3aa260029a7f8 : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4341138abc7a4523bb243197291ec74c : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1fe5af6210c41f9b2eebfe098e57128 : SUCCESS in 25m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/51a610e3ec014acbbd94e6816b08e935 : SUCCESS in 14m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/269863e159c4470895bfed1372dd4f3a : SUCCESS in 29m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7c9c427e76084bfe80cd66440691900e : SUCCESS in 13m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/12427ac3f8de4e04986774d3c0149be8 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b97faa18eeb8408ea2066f21d15a3160 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ce32f33953784fc6af9cae589fded5bc : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96eab553a7ac46459c2e959ce5cc5904 : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9ab25cda25c4ca693a2551e0946f16a : SUCCESS in 12m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c346d8f30751414b9150cfc7aa99e2da : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/526f5fa201f3420584217d4410b714d6 : SUCCESS in 54m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0168486e61345329336b1dc1ec7b5b8 : SUCCESS in 50m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e6a523feb1f4de4bb8000e80e55f5e4 : SUCCESS in 51m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85f84abf550b4731a0830411318ccd00 : SUCCESS in 29m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7636f9dee3ba4793922d22b970e0599b : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc12a46a9c9d4fefa1314dbc992266f8 : SUCCESS in 9m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b4156e702694e55bb55f3cfe37ad6a9 : SUCCESS in 42m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b51e7ca427d647eca7904b056af1777d : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6628a33ac5600aaac14b831639a5a5b25822db06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:51:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09c1958ddc82290a4a625514209ef1d0780425fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 21:15:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52e9ab764caf4c36a787332e4a733fed\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e466bd3c37c348af8ee51e191f3e6289 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cc4eab7491efbc0f00f25d80cada86b04caba40b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 22:00:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d47cbe3109a744cdacea8294d395f1a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3252d3c484814769970bb340d52fa561 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/cd62830936ad414f9a0f4b527907ca94 : SUCCESS in 1h 00m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b497392e5ce4f61bafcbea111d770bc : SUCCESS in 58m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38412ed2e27c4d2293ed8e93bfb802ef : SUCCESS in 55m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fefe5ec1c42e4e9ca0c81016481e2785 : SUCCESS in 34m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b6abdffeb38244b08bddd216cffd91e4 : POST_FAILURE in 3m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/44f8886555e642b4b2992c891941cbaf : SUCCESS in 26m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48d02176badc412aac0bc631875b9d6d : SUCCESS in 9m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/84817a878a26406d82ba01e8b4a3b291 : SUCCESS in 12m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a60adaf89e0a4d50b70fa54dd71ea5ef : SUCCESS in 12m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b4d62a0247cf4256a7672a4d5ee62d6f : SUCCESS in 8m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f55c845884714421ae05d5b6cc887282 : SUCCESS in 7m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0be8e173da4f4eb29554963ad1e3ddb8 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9b08e7174a44f67be5b3210e5f899d7 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad8d031d1cd141daab666aa3a0b80cd3 : SUCCESS in 5m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f7e176fbc3b44dcb69fc63200c1728e : SUCCESS in 36m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46b3e7b418ed4e8aa2a267ddb0c8a278 : SUCCESS in 51m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65a5aa03902a44eb8b9d37459a4e1e44 : SUCCESS in 22m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df8b490a83284390aa10bfb0f3f8681d : SUCCESS in 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c277c491efe7419ab15d4674d3a54e30 : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b33402af18fc4184a963ad68ad450666 : SUCCESS in 10m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ab56f03e6b64bccaa168b6dbffdd7bf : SUCCESS in 31m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d7ea7eb1f0347eeb7537174bb965cda : SUCCESS in 39m 57s","accounts_in_message":[],"_revision_number":1},{"id":"aa298c27191863558ba8c59f42c8a1220ddc20e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 18:06:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"28adc670d1da53a9c051a3f3e839636c718fb80c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 18:08:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e34e837a571869ed4f0708e0fa3f2ced5c347b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:23:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae3f3441d60411e824c0835cc71ad46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91087db4e7354f67a7afa5485e5e523f : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/4ddd24159bd84ebb9956353fec964bdf : SUCCESS in 1h 04m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e718f41adb54066936f08d1dd3675eb : SUCCESS in 31m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2603f0a6570e4fd8a949d349833527a3 : SUCCESS in 53m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f2079499b14f98a1f4ec18e079cf5a : SUCCESS in 37m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e2b05edb510405e9e310c21f7e30dd7 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d00fac9848d4407bd26e02cc270494b : SUCCESS in 10m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4b8eb0d05f740c583e4a7cea39bdd6b : SUCCESS in 12m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f12d7b1d3764a0085b5344efa90c856 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/07e568814dd046879bca3a7ab5bb00bc : SUCCESS in 25m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02f045a5977a47dd92bcb92ee7177b52 : SUCCESS in 20m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/85247a3fb48a4c9eb1ddb841fea3c686 : SUCCESS in 34m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8dd269b8f7f842c497c2a9221c6879fa : SUCCESS in 19m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b8971c7de5fa46fa89d894582bf952c5 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ed2eb59e8da54aeb8eb7dd1baa832aea : SUCCESS in 13m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a6957a91b329442abc1efe20c47e29f2 : SUCCESS in 9m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/336ea7f638ba49a79fe7a3167ee6c97c : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5dacb1d4c3fb4c8d9133b92be1e56fae : SUCCESS in 15m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9d86791ff0c9480d9e7792f603c22c64 : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/77b3d0f81b8940ed9218fce9f1b638a4 : SUCCESS in 53m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfb1d8f3e02d418a8efe3eb7bba3448c : SUCCESS in 37m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c01ce3be8d6406da095f413637322d5 : SUCCESS in 41m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/459989cb0a834311a867f93365b8d1e7 : SUCCESS in 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89fbc82cc734f579f254986e5c5e061 : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5984380d6e02407385baea253d5f41ab : SUCCESS in 16m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f48ec0f65444cce9b99a313dd64a17f : SUCCESS in 58m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58b8f9b18ccd45af9c6b20a8ce85486d : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e93e6116af0959cd00e0da3a715f662d37acdb1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:24:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"451927af3406b80b31562655f99295c617b5b12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:42:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ab0ffb0ea34ec79273bd697fc774a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507466b32529447a8e0968de4b2fd965 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/19fe1ffa3bf14d3893e3a843e06ea691 : SUCCESS in 1h 00m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d911f4139b54115952b11d60b38d4dc : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/686ef58e83924801b55ec8f861bd421e : SUCCESS in 50m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/407c0e345f6a4c11bd91e19f046de2f4 : SUCCESS in 35m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/845fbd6bc25940db99040974341b7f0e : POST_FAILURE in 8m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c44c7d3875a3457693ad4a4997e6321e : SUCCESS in 19m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad86b2664684fff9357f5514035283b : SUCCESS in 19m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ec43bb6615b486b96505c9cd1949f29 : SUCCESS in 27m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/838655f1c5f54f5889059c2c88a7796f : SUCCESS in 28m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/075671d6025a463a9860742a07b6195d : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84b6cf2a0f6648ef83f80e38d1b1abec : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3240b43b31854284b78f34b98513ad22 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/031da290ef89433db4aa6f75719567a2 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b689dc56fe4f44b4b1d09e5f1e9e8de4 : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8573285309b34841b26f0932195a866c : SUCCESS in 35m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/302da8633e474ad9a63c01258107dfeb : SUCCESS in 23m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/533409b13d374e338d173680a3c35fa9 : SUCCESS in 54m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6df3806d67444963a80a69e651fbbd98 : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982e24553f864fef8859f2793cee8c9d : SUCCESS in 2m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/660f3c2841964dc09c9134986ae49e1f : SUCCESS in 9m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa2f7576876f467a9c2297f254e58dc8 : SUCCESS in 52m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf6404e9f94f4524995661e7a93a96cf : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":2},{"id":"39faa59ad5f7425c237c102500277a14a4c32583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 20:57:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"70ebb288e29d3337797d182fc34649ae0a1b823f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 20:58:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf31eb7e979d359098c59eda426480f6383183d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:02:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b094abe1f444708a218ad85db20a5d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfadcc7b2c74c7e9f1a448fd84c1e64 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/0a3611c1087b4cf4bb78ceb6cecc4ac9 : SUCCESS in 29m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf3a3d9c1fad4231b36f28fdf45942e6 : SUCCESS in 30m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d87e605ab484926be276934ec372dc0 : SUCCESS in 54m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcd573a080f14753995feaf62b058524 : SUCCESS in 55m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75dc824e25b44c6b814d49d201daf8c6 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f58832dbfbd74afe9898e8ee6aa61e25 : SUCCESS in 22m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99d6f94d51ac46e8aa849eba7e53e6b9 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6fc3eaa58aa7478689238d389aa97c59 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f2e6ce0a91a0466a8219312ebe966c1e : SUCCESS in 12m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e004ebeff2f4417f8a9e1d960af54d65 : SUCCESS in 18m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7e969a5bbff6448ebcaa3f232b82123d : SUCCESS in 29m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d64480126d3048e6a3df9dd79e75aa9a : SUCCESS in 24m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/30ac77825c864cc2ab15dfe1002b46f4 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/af872c0403f645c6a9a9110dd699bbaa : SUCCESS in 8m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c3b100821cf49c48dd05f24ebfe94aa : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea09a70789894e9cb7dc940be3df9241 : SUCCESS in 6m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c280c85652ef4319b64e266f9bd7e6c6 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb450c251fb34d4fbbf95c9aa178339c : SUCCESS in 5m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/64c73b1b019e43909060c4f55e3bd6f1 : SUCCESS in 55m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d90f133807d4b7c88746f5b07eca882 : SUCCESS in 22m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2f0164f854a4c9d99a98dda7fa20d5e : SUCCESS in 33m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/043b493868f843178e040f777436bfdb : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f31007ceaa4ae0863d99fd9b6c280f : SUCCESS in 2m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f622bace585a4f5e8268f3cb3a5e1320 : SUCCESS in 18m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4450680595d44d4f91de17a1b967dd8c : SUCCESS in 54m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/974660cae68a4c9c8278a6f222af58ae : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":3},{"id":"19518bbd4136d77ba083ad8764cafc55fe20aa7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:02:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd945e7e552b562c6e74cd0964726f48ae8cc10a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:16:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aff500fbd2a4c0ca6ca861424efcd4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ffc8b5db6c48e8a861b1948f794b94 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/1c9c16623ee74b0799d88ead934e7ee4 : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c243fe4a810d4e30927a51d37c2c518e : SUCCESS in 58m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49357baf6dac491db4c1e0916134ebd7 : SUCCESS in 38m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b33df80eea74688b3acb60ff00ea7fe : SUCCESS in 1h 07m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/80c37e19f3ac4ad8b65c2d5b61d92aa1 : SUCCESS in 4m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce879c0c161549d2ab5617cf71dcb635 : SUCCESS in 10m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d0c98f466f5f4ec8b339d1ecf48714df : SUCCESS in 20m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f35d020519364b128340177cdec7daa1 : SUCCESS in 28m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7ff73b35c8ba42afb21a7b1c4d59dfd1 : SUCCESS in 30m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1ab2e62059114811b726e2fb31a28972 : SUCCESS in 12m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2cec9fc6f6c9474fa1ca9d27cad26a59 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7772d5bcb4f4461eb26605cd375a1efe : SUCCESS in 9m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b098c865a85741e680535f58d3771e10 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1a3299375fd408285ef61fdc580835c : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/55fa674700f7411a9645143a96c98c07 : SUCCESS in 51m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1096aae388e549d5a76e1a529cb9c48a : SUCCESS in 40m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d927dba8ddd2433e970c734904e79bee : SUCCESS in 38m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b73d9e668c44a5fb2ae19bc9092f7d5 : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45067cfa583a46deb239730ff8eb3cfa : SUCCESS in 2m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/753c3afbb86d4f64865342ce73f03d09 : SUCCESS in 18m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37a03d082bae412eb4efcb4b192324bf : SUCCESS in 51m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24c80b036b324e599ca422ddf531c01b : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f7257022fcdf09101cc9c2d5aeb048fec5c71272","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:16:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"48b05e6e831c26835b61425e1e8c2f929b8d4856","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:18:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78854652dfaa4e338c605857b6fa1d21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/911c02999ee14c8193aa2919d856a701 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/be5d2bb25cd44b288c7eee3f6955b5b7 : FAILURE in 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65b52cc3acff58632ffdb461aadf9d4c9024403e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 02:32:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f33d2cd49d4bfbad7aa9da6767a1c7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/978571df2d6d4b77b922e390299dcb6f : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b52363394c5a705177f1e55be43f9a6afbfc92e","revisions":{"a907caff9232a86e42b037777dd93f97e8268e60":{"kind":"REWORK","_number":1,"created":"2026-01-20 19:41:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/973990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/973990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/973990/1"}}},"commit":{"parents":[{"commit":"732fd859d8e0a0b795fbb42ed8390ab4cd494227","subject":"Fix flakey db_replicator unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:40:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:40:38.000000000","tz":-480},"subject":"CI: Update dockerhub secret (again)","message":"CI: Update dockerhub secret (again)\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I53445f574953d9d47dd93d6819060e4b3cabee75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a907caff9232a86e42b037777dd93f97e8268e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a907caff9232a86e42b037777dd93f97e8268e60"}]},"branch":"refs/heads/master"},"ebdd9bf2456c3965824362e909e0ad1b2d425116":{"kind":"REWORK","_number":2,"created":"2026-01-23 18:06:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/973990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/973990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/973990/2"}}},"commit":{"parents":[{"commit":"732fd859d8e0a0b795fbb42ed8390ab4cd494227","subject":"Fix flakey db_replicator unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:40:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-23 18:06:22.000000000","tz":-480},"subject":"CI: Update dockerhub secret (again)","message":"CI: Update dockerhub secret (again)\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I53445f574953d9d47dd93d6819060e4b3cabee75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebdd9bf2456c3965824362e909e0ad1b2d425116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebdd9bf2456c3965824362e909e0ad1b2d425116"}]},"branch":"refs/heads/master"},"9b52363394c5a705177f1e55be43f9a6afbfc92e":{"kind":"REWORK","_number":3,"created":"2026-01-23 20:57:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/973990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/973990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/973990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/973990/3"}}},"commit":{"parents":[{"commit":"732fd859d8e0a0b795fbb42ed8390ab4cd494227","subject":"Fix flakey db_replicator unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:40:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-23 20:57:18.000000000","tz":-480},"subject":"CI: Update dockerhub secret (again)","message":"CI: Update dockerhub secret (again)\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I53445f574953d9d47dd93d6819060e4b3cabee75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b52363394c5a705177f1e55be43f9a6afbfc92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b52363394c5a705177f1e55be43f9a6afbfc92e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
