)]}'
{"id":"openstack%2Fswift~973991","triplet_id":"openstack%2Fswift~master~I5ad18f1ef0ef040fc65d34904fad2d324b451d0d","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I5ad18f1ef0ef040fc65d34904fad2d324b451d0d","subject":"CI: Make swift-upload-image non-voting again","status":"MERGED","created":"2026-01-20 19:41:17.000000000","updated":"2026-01-26 22:28:24.000000000","submitted":"2026-01-26 22:25:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973991","meta_rev_id":"f8f303feaa3c7b04ec23d991bad8e4eb110a1ee0","_number":973991,"virtual_id_number":973991,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 22:25:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-26 21:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 21:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 20:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07c4f078679fe2aa29f8f33345151230796290e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-20 19:41:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4082273bd8175c23b4f32fb117b315d07993e29d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:44:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39be364c0f2346b380208ea803cb95bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b49e7b15684f2f8f122008a2ff3af5 : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/affb18a88b5a46708d0fbe0ad954f726 : SUCCESS in 43m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/710574c1f2cd4a3d863e1c24e10c772f : SUCCESS in 45m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68cced8d3c1f42b89bfed7683c6ab683 : SUCCESS in 56m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ef8b19007cf49dc8b7e28a43d32540a : SUCCESS in 46m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97b5b77411b34360a181d23dbac69aeb : SUCCESS in 10m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9a19556618d49bf996a15c400edcddf : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c20a192c928407aaef044ff907a3f66 : SUCCESS in 19m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f33867bd9694167962813529adfe5eb : SUCCESS in 8m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/872e65cb5bc5426899b5110939627c89 : SUCCESS in 22m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fffd60db25cf45e08bacd46674d64326 : SUCCESS in 9m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cc3fd82b4a784964b2c05d18f0dedb50 : SUCCESS in 11m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4d24af8a14224771bb747783a48e67ea : SUCCESS in 13m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ec125eef7dde44cb8b6d0fd54fb3c2c4 : SUCCESS in 7m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ad18e6070efe452f89ec70e4c50b1a71 : SUCCESS in 9m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e24f0de935eb4671be290837f9b53fff : SUCCESS in 12m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/322486a14d5d455c857b5640542c63e0 : SUCCESS in 8m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/167f7d86fb3c4695a771f6e4a9691714 : SUCCESS in 13m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/87eb37a7b5a847d7b19d12afcaddae35 : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/09677cb576034bfa9911eacfccab8fcc : SUCCESS in 36m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2fc8df2ae38748a2b96c76e25650235a : SUCCESS in 34m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff340caf02c54c73886824bf5ea43a5c : SUCCESS in 38m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de38fe126fd44d0ebfe4073f6d0c9651 : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b31810551e9480abca763309a32bd85 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f67fdd51f564db2b31b4d2eedbbcbc0 : SUCCESS in 18m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6b08cebb6a44d5abc7660301041755a : SUCCESS in 51m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10a8c4cb7fd04b2a87cca062cdd69bd3 : SUCCESS in 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4a2d9ba2d482e1021b31071616b5b1e2e7cc9978","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 21:31:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2075317bcd6c4c8093f8f5b4007ebfc5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9313aeb7aef4497ac5249183848fc4f : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e9b3dc6efb1980a8e8188afde1255cfe0cae996d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 23:18:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0a4f608800b66cd921f0b47527b7ec0afcd07dc2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 23:18:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8445e70f1f1906112057595f490890fa9d4daab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-23 23:27:34.000000000","message":"Patch Set 2: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56b25e8237bad1147c6d70a2998cbcee9a896795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 00:21:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2d689ecd514498488d138344bee6772\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b775e04878462da4914c819c94bff9 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/0a74d2f8312e43e1824e4e769effa85a : SUCCESS in 35m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/daf4d69589024ab9b58074c86e8fc223 : SUCCESS in 32m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3368239e7eba4b9e89f9fe2fb04d4912 : SUCCESS in 35m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49e2d39f9f184c2f8295cfef06f9a5fc : SUCCESS in 35m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee0dded75071471fabc5e29b797b9f21 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f04eccce23e4443d937b55f65fbfa33b : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5feabcecfc1046c9b80aa5b31ab1c487 : SUCCESS in 17m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e35557847ec3450aa0ade9bb459308a1 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/faf73001084d4d20829d8936ed4abc52 : SUCCESS in 9m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a2a19077bd834c93ba475bd4e89b63e5 : SUCCESS in 20m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7eb8e48816b24ce0b44dcbf2655643b6 : SUCCESS in 20m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/844c01b670c54ee5b72f0d83e24a3776 : SUCCESS in 10m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c686d8a7f969477b8e54d2df2d920875 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2b3ed0bd0310403f87e8d0e07a28a9a2 : SUCCESS in 11m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7d312225b67e496e864064a39c50c91f : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4326cf0accc4acbb6d828a562028891 : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/693ec21db54f4fc0af2bdccde04e1f67 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/af74d8c6c8a74554aaa0a7eae54498bb : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e3de3a5709846c4b2a2b5dad0b0b41e : SUCCESS in 55m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebd9cb299cfd4a348df25720dedefac0 : SUCCESS in 49m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e546bfbeddae45e3bcf6abeb6d5ad7f2 : SUCCESS in 22m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/866f54bec2964f359322dde3e9f615aa : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8e1b24b6b74d9aa8ead18a292f236e : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c77d837ec3348739de2f7f3292be228 : SUCCESS in 18m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f19a3ced239e454d938fe80cf28fc178 : SUCCESS in 28m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05ed0c221bc8460cadd617598e707c0f : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4fa7c89322e1bb13f8739b16f0f9a0a445461c0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 02:41:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd196c14c937407d9a78086392179679\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49408b979ed5463b9630fa4957b9d0cb : SUCCESS in 1h 20m 02s","accounts_in_message":[],"_revision_number":2},{"id":"bc1249e6afbb1a2060b38c62dd684bf321b292b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-26 18:17:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"906bccf4858ae2d18877522f363356565723b63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 19:30:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b26a2c43966342179b0b829e9c0185bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f59b170808d478d80b5cc6996e4e93e : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/a0aa0a0b208b4620842b82c9e512751e : SUCCESS in 31m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67d13ee92a614f34a6ad8141fc2c9d2c : SUCCESS in 53m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a89c21ef347c4f35ae733d6a35fde810 : SUCCESS in 33m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/050a4a7d73c942229bd79385fb64959a : SUCCESS in 1h 05m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08444122e7a242e2917f593bb56d72c3 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/854ddc8daca34643a75caf0845013130 : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f3226c7ec5345fc85a66bda00ce33ff : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a1c2422946544d8bd3153a05e08141a : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36fda17c0ccf40d8a7e7fa1ab5cd5656 : SUCCESS in 25m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c56c3a1d94914d7c86fd670e61744589 : SUCCESS in 21m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b9c6eb4271ed4d439a3d2f0ff016eece : SUCCESS in 27m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0571381ba5454a4586a7aadb33a3f80b : SUCCESS in 13m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2c6d5f6f200d4ab2a70083180deb6f5a : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4b7847fbdd94712832f3b7ebdbdf6dc : SUCCESS in 8m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/64bbf6794e334a0fa31419b861389b81 : SUCCESS in 15m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2bd113a1993d47118635d61c66344171 : SUCCESS in 3m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c53335af3ed147e4ae61caef5557daf3 : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eea9322a7b294ea6853e6e2922b5e87e : SUCCESS in 5m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/682760ea5c3a47bd986236c7a1509eef : SUCCESS in 34m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11c43f44ba7248769cc2f4fee8ec5ec1 : SUCCESS in 58m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e96aab0e17fa48a68ed7ee3098a33569 : SUCCESS in 22m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c830c36584e47c09d449cb996b32188 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1078aa77d8ed47eaa2e8e5df83972fc3 : SUCCESS in 2m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58b3aefd068849df883d53cea5f3dc4c : SUCCESS in 16m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9e06882e3f24522a0f008b666cf5270 : SUCCESS in 50m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49cefd0e5b6049fc88e6c95297d3b4ff : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2e2d79e47a5806c0e54c984cdeda3baed7632475","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-26 21:10:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c34d57e8d8fb7c83933d0c6d5906ab6ae23c006c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:11:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8acddd290d67d8a47fb679781cf0070c6931e855","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:23:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c24c96b8a034b69b4ebc5c153ff53a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/234d5bb094b04431a275108a81f369ec : SUCCESS in 1h 30m 28s","accounts_in_message":[],"_revision_number":3},{"id":"823b9402c44902753ccdad5847277db95c805d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 22:25:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81ce90824a04328bbec8c81959daade\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4feeed3f2d4a69b8a6f96cf151423a : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/9389604e73f94e24bcd09d1fd6eb89dd : SUCCESS in 1h 05m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5e2f92d81a34d1f9f380dff560ca4e7 : SUCCESS in 51m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea64a7b493944b97b54fbe97ea796a07 : SUCCESS in 40m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71ca38880af14d5495b4d14c3343dcb0 : SUCCESS in 1h 02m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/77bdacd9f93b49d2aa1eb95a2fbe537a : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0dbf0505f0354c70a0ba4bfc778556a7 : SUCCESS in 13m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe087dbefb5a4b829c66c01a03b43860 : SUCCESS in 23m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e88c1bbaa37242eba12c6453c10eda47 : SUCCESS in 27m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1aeb26c630cf49f8afcb0a93324f3e84 : SUCCESS in 22m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5d547b3deed64b87beb12413f9cedcce : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b3a9c40954a4eaba1f6caeede60db2b : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/08af39a16ec1437e94b4363cd4322d00 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bef12529531e49abb93bd6a3bcdaf4a7 : SUCCESS in 9m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/62b2de54199e40a88706dea751465ce7 : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d72b6a0515444cf1b0f836869086c097 : SUCCESS in 1h 03m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ffd965b61bf4c0ca039febd015c1aef : SUCCESS in 33m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/098ad6291e9d43e9add7bc2716b940cc : SUCCESS in 55m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cca4a0c0b5846e790e2b118d7699ec3 : SUCCESS in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb1f0af343b40e0b58963b4e223a936 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d06a6173c65d4a66b14d5168b399b378 : SUCCESS in 21m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d73f779d77e4c419b89ef0b0845ef19 : SUCCESS in 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136b1a49a7fb48cd81877b2d5012a2bb : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":3},{"id":"13c6a01057182c8843da1492d13992495d21335a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 22:25:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f8f303feaa3c7b04ec23d991bad8e4eb110a1ee0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 22:28:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e354c371dee456da42789c33495db42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37aa4fb3cd694c6ba77020a096d6b7fa : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f735fbd4c943432a9392901159f75a70 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef0d8591c0c2e57fd3f2a683e874a40ca87e5901","revisions":{"8e7ed08b8ae68412f9cf651fce31f61771746c1c":{"kind":"REWORK","_number":1,"created":"2026-01-20 19:41:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/973991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/973991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/973991/1"}}},"commit":{"parents":[{"commit":"a907caff9232a86e42b037777dd93f97e8268e60","subject":"CI: Update dockerhub secret (again)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a907caff9232a86e42b037777dd93f97e8268e60"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:41:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:41:06.000000000","tz":-480},"subject":"CI: Make swift-upload-image non-voting again","message":"CI: Make swift-upload-image non-voting again\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I5ad18f1ef0ef040fc65d34904fad2d324b451d0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7ed08b8ae68412f9cf651fce31f61771746c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7ed08b8ae68412f9cf651fce31f61771746c1c"}]},"branch":"refs/heads/master"},"429e84e9c37bd6100061469ca12c7b1018b5528a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-23 23:18:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/973991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/973991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/973991/2"}}},"commit":{"parents":[{"commit":"5a7239f4bf85be0313cbb610115f8046fcc1243c","subject":"Merge \"CI: Update dockerhub secret (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7239f4bf85be0313cbb610115f8046fcc1243c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:41:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-23 23:18:11.000000000","tz":0},"subject":"CI: Make swift-upload-image non-voting again","message":"CI: Make swift-upload-image non-voting again\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I5ad18f1ef0ef040fc65d34904fad2d324b451d0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429e84e9c37bd6100061469ca12c7b1018b5528a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/429e84e9c37bd6100061469ca12c7b1018b5528a"}]},"branch":"refs/heads/master","description":"Rebase"},"ef0d8591c0c2e57fd3f2a683e874a40ca87e5901":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-26 18:17:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/973991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/973991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/973991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/973991/3"}}},"commit":{"parents":[{"commit":"ee682f4c2f642e5c4b990e34090e8ee54b2b2f16","subject":"CI: Update dockerhub secret yet again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee682f4c2f642e5c4b990e34090e8ee54b2b2f16"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 19:41:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-26 18:17:05.000000000","tz":-480},"subject":"CI: Make swift-upload-image non-voting again","message":"CI: Make swift-upload-image non-voting again\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I5ad18f1ef0ef040fc65d34904fad2d324b451d0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef0d8591c0c2e57fd3f2a683e874a40ca87e5901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef0d8591c0c2e57fd3f2a683e874a40ca87e5901"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
