)]}'
{"id":"openstack%2Fswift~974666","triplet_id":"openstack%2Fswift~master~Ieaf9ca9a67e40a7133f56f4fde86a88ba4666ac4","project":"openstack/swift","branch":"master","topic":"update_dockerhub","hashtags":[],"change_id":"Ieaf9ca9a67e40a7133f56f4fde86a88ba4666ac4","subject":"Updating the dockerhub secret (again^2)","status":"MERGED","created":"2026-01-26 01:33:51.000000000","updated":"2026-01-26 04:20:30.000000000","submitted":"2026-01-26 04:18:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974666-update_dockerhub","meta_rev_id":"8a3cbeebf9cc476ac2642cd6b16a2f1877d65a20","_number":974666,"virtual_id_number":974666,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 04:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-01-26 03:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-01-26 03:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 02:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a537827703c5139a263eb45c86484c2193feca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-26 01:33:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"263b16c51d00b143fed3e9fe3e2b35e86b4f80aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:48:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abd5038abdd4568aec0dd056155356d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecfab62f72e42e4bf1ea48b08a174f7 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/0ca0effd64a944459d7e6c432ec4429d : SUCCESS in 37m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f70208cfbc44d2bbbcbb0440aaa013a : SUCCESS in 1h 03m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39376a65b9d84293888a85b6e6220e3b : SUCCESS in 55m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33dd3a394c874a75a4fff7b10c9865f0 : SUCCESS in 32m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc6e6cabd86048c1b7efcd93dfab10c5 : SUCCESS in 11m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/267d93be4d9e46279539e29961ebc74e : SUCCESS in 10m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2a244de66c4406aa1f5514d525180fd : SUCCESS in 14m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c4c47f5375e4cfab34aa225ff5faa6f : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ea37589999f4d45b51674a487915cee : SUCCESS in 23m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aeb747bdd55147938fe043d94818c8e7 : SUCCESS in 16m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1403b3eb9fc4ce18ffd94fb5673a433 : SUCCESS in 13m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/90181267627949029d3b5c40f6023cff : SUCCESS in 30m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b525a4e1e0814dc6bbd7b1134d5a7f6b : SUCCESS in 9m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/10d7adb3f99f422888e0e754e893b710 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b40ece6200964c2da194afc48c68bdfe : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6b7ebe08a7b42fbbace34ca423257ce : SUCCESS in 5m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0bcbd050dae4d2cb04b9dea37d2646d : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/68b1638c72f94aaea55826c0466e9f97 : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0099e1d4aac746d5a04dd32e8cdf914a : SUCCESS in 1h 00m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b04147544c9462b8e287e4eda52eb1b : SUCCESS in 25m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f81bfb9c6b0c47cf824365c95b5eeaa6 : SUCCESS in 35m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ba8df2029b44343a606e922834307bc : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e8b1d7eba744be2b99eaf410b9951ef : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4862027d120409c9a0c5f49565c3ce0 : SUCCESS in 17m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7afd0b65e4240128c6eb30fab04a9fb : SUCCESS in 51m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d9122e33d4048658500bddcef437e1a : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7a835f85fce46140dee9d3c00b3ebd10218ee7db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:57:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecfb9632445040049e3127abb6b76c9f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/363301e2c3d34a4f8bb7238fa7ae3a3a : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a0b818a3867f15995c52d4f7cacf6cce0cf79cfc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-26 03:07:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"caa69a58777189c35a5100b6c2fd15935947eece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 03:07:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"344020b15da2d542784fe1cad21d40084cc3fe79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:18:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9667d50da93540558081cd75523cd5b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eefd36e214f54e569978348f7090de14 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b641681d85af4ebd989ee2a6c7befe20 : SUCCESS in 58m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b3636b4860944c0b17535e31225a7c9 : SUCCESS in 29m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5840dd1fde354a3e95927447cb1644f1 : SUCCESS in 58m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c8016ebcb704b6692b135a8d46893c8 : SUCCESS in 36m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6fcf720fe7bd450a855a6bbe01bcdf2c : SUCCESS in 8m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2d112f59a5943da8b69ceae1f69ebe5 : SUCCESS in 24m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9de320e4a84486aaa1176d75c87dbf7 : SUCCESS in 15m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eb358d9a4e8148aba21481706b103590 : SUCCESS in 29m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f3cf3a1c6ec3434ea477b1cb7ab026d9 : SUCCESS in 12m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ced59b54df6a48fc90efb594a2bf0b22 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0b60f01d02d94f9a96b721cbd8524750 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ce19ec2df144519a72e3193827e7472 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f507869bfd9a49d7bd96fbaafb4a6497 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6054f3d9ee6d415fa1cf798ffbd59cfe : SUCCESS in 10m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/685e858d8c7047b2a1fc6b888012a105 : SUCCESS in 59m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57bad34bb79f4165b77764e2c3f05bda : SUCCESS in 50m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f772b87f976d447ebad551c553864d1a : SUCCESS in 51m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7f6738faf214be4bd101d382e8b44f3 : SUCCESS in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabd85fcf0e142dba829ab1f4d8fc66c : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/66eb79aa09a742798c444a5dbc444c3a : SUCCESS in 18m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a0f92933c5d4a4bad8fe749c25868a8 : SUCCESS in 49m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f62b7df4aea4af28c4bbf0d98e60580 : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a1f819f437f0839b6b37751bf1b54194e05c8659","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:18:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a3cbeebf9cc476ac2642cd6b16a2f1877d65a20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:20:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2cc933a013347cd8be22944cd79a282\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c72fa9b040844919ab5520729d2f147d : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e451b7da85604c7c84ebf17f823fecc1 : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b0c11b68fe264fad2ed2e5d8e4db083986c9e6a","revisions":{"1b0c11b68fe264fad2ed2e5d8e4db083986c9e6a":{"kind":"REWORK","_number":1,"created":"2026-01-26 01:33:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/66/974666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/974666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/974666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/974666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/974666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/974666/1"}}},"commit":{"parents":[{"commit":"5a7239f4bf85be0313cbb610115f8046fcc1243c","subject":"Merge \"CI: Update dockerhub secret (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7239f4bf85be0313cbb610115f8046fcc1243c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-26 01:33:07.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-26 01:33:07.000000000","tz":660},"subject":"Updating the dockerhub secret (again^2)","message":"Updating the dockerhub secret (again^2)\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ieaf9ca9a67e40a7133f56f4fde86a88ba4666ac4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b0c11b68fe264fad2ed2e5d8e4db083986c9e6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b0c11b68fe264fad2ed2e5d8e4db083986c9e6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
