)]}'
{"id":"openstack%2Fswift~974756","triplet_id":"openstack%2Fswift~master~I202ac65afd0fa1812cc7b5e0b6775489ac1daedd","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I202ac65afd0fa1812cc7b5e0b6775489ac1daedd","subject":"CI: Update dockerhub secret yet again","status":"MERGED","created":"2026-01-26 18:17:12.000000000","updated":"2026-01-26 23:14:08.000000000","submitted":"2026-01-26 21:05:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974756","meta_rev_id":"95dea028c7dc038b107853ce2fa3cdd9f4b31d1d","_number":974756,"virtual_id_number":974756,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 21:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-26 18:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 18:18:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 19:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2ed14fe97f2d89eba65f1621e9d9dda56edfa31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-26 18:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3159ddfb77e463febe540b45be84fa1bf37738c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-26 18:18:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6389d1374841854aa7c1b3008d42f04e848c423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 19:26:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdcadefa1e6945dab86fef0814bf796e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc4ef45fd6748d09c540f8c4e4bbaac : SUCCESS in 10m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/dff4664734da4fbd83ea468fcb9cd32d : SUCCESS in 55m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/986c41d7d13d4b098e3e050d909b5d6b : SUCCESS in 1h 03m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c500b2bcde654f7a83b03d86ac9e752a : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6698c25ae5f546dd816121adf923fd5e : SUCCESS in 36m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5180c4ddb1de45abaab2007f1dc111fd : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcd278f832bd46f1baab502872d46ff6 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48bf370d706d43dab2c9b484ce251395 : SUCCESS in 10m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccd11439b768497c98925fb4d28734e5 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/659382cc4ef04cd881e114bfc7c045e9 : SUCCESS in 24m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e501c226ed0c4e8bb0b289c9d4a20199 : SUCCESS in 21m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2cadcfcc23664a4c8786674a062e50bb : SUCCESS in 11m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/65f793647b584a759f018466095f2a5f : SUCCESS in 12m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c7bd31fc21394a79ab0c9c5e796c80f2 : SUCCESS in 8m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8bb5d3b2f3854e67b9dba55a7195ac5c : SUCCESS in 8m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/05f832aeebf9419c815275aeece1a69b : SUCCESS in 8m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa435d480ccb47aeb4b11558960aabb5 : SUCCESS in 4m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a97493f6e58141b69c32eaf74f0fccd8 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae2dfde3576e4c4f93f32fa28b7a61c0 : SUCCESS in 5m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02669a62977b4d20b5f8f358bd11dfb7 : SUCCESS in 1h 01m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f92c6c59e3544c6967a876b2783ed91 : SUCCESS in 57m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a280eade56b4af1b918d2be48ca3086 : SUCCESS in 54m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4929f265bf9442e59d304ed5a8199661 : SUCCESS in 30m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda924fd61f34c4aa5f3b5dbab54e0ba : SUCCESS in 2m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2599b5117764c869c7a730644c685d8 : SUCCESS in 10m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74cd8cec55594ed68f78afb90b471376 : SUCCESS in 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25b8c1656d1947b9b4df16aa5c0c5101 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4b9dbaecaeaa6b74266890e4634644ca16acf9ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 19:27:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"193c939807eaaf66a4f425c716b3fdd108262df5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:05:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb8ef43b82724ef6b6ab955afdc604b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3525afe679b54d60ad476c0d0f755f19 : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/1f456e879f284edc912e4ac51d499ba1 : SUCCESS in 59m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a18760b1043e49bcac0cf1ff666fa1a5 : SUCCESS in 1h 01m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b735a40cc2a44e38c09c6fe39dbe810 : SUCCESS in 42m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d20c2300ef8b425a9b83f12a81d54d34 : SUCCESS in 1h 02m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/adee059d3a434797bcd8b76aeb01bc56 : SUCCESS in 9m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff6044afd3ab446bb2b839a8b167e101 : SUCCESS in 23m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2909011c3e3442108e1bd5d85963424a : SUCCESS in 20m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f088b8ef6d6f448b8598c1c3769bd551 : SUCCESS in 28m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/561eb6e773ae4268b1d4188dc39d83d8 : SUCCESS in 26m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fe9b799cf8ff4f359299224d86448076 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/23270582ec444a9a9af6e7bd42e34a51 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c51301e83cd4de48b235b4fc6e638cb : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27295a4fc1fc4813825f269ac46248e6 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16bef95c72a74f1991e0890c90ec2773 : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/088da43f39fe470aa63b8e5cb2bb9360 : SUCCESS in 34m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b629a36e2d8840ab95e1ebfc4a9dd6d5 : SUCCESS in 55m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c5b809aebc94ad8aeb43af918b7d014 : SUCCESS in 21m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6dd01fe2ba74706b702766420a16bc8 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed8d70ced23f4f5c9ecd3cae8c32e309 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45dfd687870f4c0aa63f7110c37e0769 : SUCCESS in 18m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/504cf4ddd8f74d83937100378ac7cf54 : SUCCESS in 1h 03m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b10486a9d8a24226a06dd2d2c7af23d8 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8ef1a41b55884ddd8f9881b828800266c0c24f24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:05:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"811395973bb5f401aefe09185c60a11bd855e6d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:08:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77baa7fd1f684a558f7126fd2379f815\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ccb5e14e404b75bc2479f5fd3d90d4 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f3f1ab35963e4c31b82cb0f8e5450d44 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95dea028c7dc038b107853ce2fa3cdd9f4b31d1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 23:14:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120e83ceeb1949cbab2bdb085189d935\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4e7dcd2e552499b8657e5873e414d8d : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee682f4c2f642e5c4b990e34090e8ee54b2b2f16","revisions":{"ee682f4c2f642e5c4b990e34090e8ee54b2b2f16":{"kind":"REWORK","_number":1,"created":"2026-01-26 18:17:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/974756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/974756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/974756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/974756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/974756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/974756/1"}}},"commit":{"parents":[{"commit":"1b0c11b68fe264fad2ed2e5d8e4db083986c9e6a","subject":"Updating the dockerhub secret (again^2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b0c11b68fe264fad2ed2e5d8e4db083986c9e6a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-26 18:10:02.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-26 18:16:56.000000000","tz":-480},"subject":"CI: Update dockerhub secret yet again","message":"CI: Update dockerhub secret yet again\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I202ac65afd0fa1812cc7b5e0b6775489ac1daedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee682f4c2f642e5c4b990e34090e8ee54b2b2f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee682f4c2f642e5c4b990e34090e8ee54b2b2f16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
