)]}'
{"id":"openstack%2Fswift~974910","triplet_id":"openstack%2Fswift~stable%2F2025.2~I6e0c76dd9157e8dc265eb67cec486e0db782b9fa","project":"openstack/swift","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-28 18:08:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I6e0c76dd9157e8dc265eb67cec486e0db782b9fa","subject":"[stable-only] Ignore bandit false-positive","status":"MERGED","created":"2026-01-27 17:57:33.000000000","updated":"2026-01-29 01:17:54.000000000","submitted":"2026-01-28 20:47:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974910","meta_rev_id":"96e18f7acee6cdf99a7fc6d1edcfc34d1ebc1360","_number":974910,"virtual_id_number":974910,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 20:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-27 19:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-27 19:28:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 19:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fa587140faad5e3b2ec4ef9f44f4297e6d154d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-27 17:57:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cba86391e39852c8f2d871649ce7226df0540d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 19:05:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ce6b232ba8042299293a09888fc37cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014b08a1b7e942b49873dbdd0d210e7e : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/fd0b1a787fed40718f74cc80f951be1f : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4679960d35044d9cb1223ad1294c21ce : FAILURE in 32m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0b656862be548d593851fdb3cb9ea24 : SUCCESS in 50m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dde57dab921400fb959c9765327cd02 : SUCCESS in 34m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bea9d2f3ea942ef84008aae2846ca11 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8191616d3c8148d091e86dafd6a6b96b : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea8aceb4aca94dccb47f9270097481c6 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5f7214ae984414c922922999f49a2f9 : SUCCESS in 8m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36cce9e624fc469a84ff019646fdbe9f : SUCCESS in 26m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4af8c440ef2243f59a473c97c01e3f67 : SUCCESS in 22m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0766dff3458246bb93cd686080a9ac02 : SUCCESS in 18m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/196b6487c1f941ada8a3692396aef8ef : SUCCESS in 11m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d8002b5d4faf4c679ac7bb9c3227b630 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/14eab2fdf9394fdf8a63ae7e72f788bd : SUCCESS in 15m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e24c3b25f3e84bf99b9899b700e8282b : SUCCESS in 17m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83189a43254f42abb1d8448d2d7ce65e : SUCCESS in 6m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d0621a1536d4575a6579161445185f8 : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/283cd271195e4829b4daee5fed17e951 : SUCCESS in 10m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e602d607229049a3af7311b4f7231f09 : SUCCESS in 57m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef01e08fbb754b588c27911cd9a6506b : SUCCESS in 34m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/481af61d307640a1b324ea5af2184e01 : SUCCESS in 33m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8d81c3dc33a40ee81df1ea9f32eb5ea : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ba7138534c495783e256d7e091ba67 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c74fd4f6dfa94e55b20d389ce7523a1a : SUCCESS in 27m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3cd2b5e53a04af3aeef041287ff6244 : SUCCESS in 53m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d41870f654a4443b43d049218c1fc13 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7d7c43f5146a7f428b479caa5e3cf1ef566485c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-27 19:28:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c08dc4a523e59d3bff17202b4685632ff6d394b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 20:37:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/159e1612922c488ab42ec8d47f10870b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083e0e4d28ca46fd97fbf38cebe1792b : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/74779b6048a94223b79f75093aa0dcc4 : SUCCESS in 30m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5914362895b4d1ea9431fc65da26d23 : FAILURE in 26m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8870a2c7eb484dfa9539ca9337117017 : SUCCESS in 31m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04f8b4be150d4d9580d1b685c2e41140 : SUCCESS in 51m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62cafdef14bb4e87aa45469407f4e0ee : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e5d2722fb3548ef94d52d210ecc8e67 : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fe989a067344154aca9086ddc0276ac : SUCCESS in 22m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a194e1c22ae04338b9a7ebdbe0c5c6b5 : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/41a01e5fe30c4d6fb1123c9ecac82303 : SUCCESS in 26m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4338aa94aa34cfd981ec1776670ad17 : SUCCESS in 20m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ceb842f876634e5f804d864d334e52dc : SUCCESS in 12m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/02e9bcd571724597a421db7bf1f11241 : SUCCESS in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e828f9e5ee74f33bf690c549756106c : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5f71d39648fe4169a34c0c988970ccb1 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c5abb0254344a32b1158d87f5708f27 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62ac2380c7454477808846e021f829a1 : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b5df8784eec4b3a82232d082e0ced32 : SUCCESS in 8m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/775066fda35241018a8228803209c9c4 : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e61808730c05474b878ce94dae44fcc6 : SUCCESS in 34m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec4f820b65614981a295b1d594b08580 : SUCCESS in 35m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5adae54a2e534a828efddaea13c85b40 : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2183b6ccbbd490384d53777846e91e2 : SUCCESS in 28m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba16d275b6ef4deda15b409a33cb0212 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/89cd8be8a450403ea50997b1289d71ce : SUCCESS in 9m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92b68c50d4f0407eb3db04737d4b816a : SUCCESS in 50m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8145532f8b1481ab9657be6be7bc549 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"add30e26bf6f18d6aee4d29210087f10292dec93","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-28 00:09:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52e880f5f05c84a442fb67590f4126c49da9a6de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 01:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14f577d175474b69b57cbfabaf24bf2f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2461454c582b4db99ecdfaadc55f14ba : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"97e742dd93a8bc6847d54aae49643c863f4f37cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 01:20:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f61f899425a4f559a63e828df7dd557\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415e3059c0024564afd74dbd013e0ec7 : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/f63a948009b94319856a0451ee5bd899 : SUCCESS in 1h 01m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c04efcca59e54ffe9f249bf90b9f8d56 : SUCCESS in 47m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3608d36e41a749e3a0dba02b6431e29b : SUCCESS in 1h 01m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f918749e03da421ba8c193cde35daf19 : SUCCESS in 1h 02m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aefb779628cc437aadbaa2c10451f15d : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9b76ec6c85441f680d6942f8a92425e : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f916d14a6d54454bf0bdca29dedba70 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01d6b81df55d4c178770afc99db9922d : FAILURE in 4m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05d8dd3bcac844f6b97a8eb08928eabc : SUCCESS in 24m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1594b2ac79a94919a2a4a69e9a756cad : SUCCESS in 16m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb5b24c379a94e199fecd605fa6f5145 : SUCCESS in 29m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0b9f91dbdd2e4716834aae703358e39b : SUCCESS in 28m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6fc6ed49cb3d4630b7e834dfb828e923 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8b4ccb9cd10045a18cbd7cbfaa496d1f : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/52a869edb97d4367ab58cb437974f0ac : SUCCESS in 8m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0ec24e87cd24aca875a8b847828f6fb : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3602d3a3181e4295bc472f3f40b2f013 : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a721b6ab87d04c5b8fba7b600e7e3aa0 : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7861f549879645afb143e12f217552ea : SUCCESS in 33m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/703de0c9f28e49e188fa3f095ba1f7b0 : SUCCESS in 21m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47e0afc2229c4e6ca10a7f9889ad3ca4 : SUCCESS in 50m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd86f05c00c041829ab7c19221cbfa0c : SUCCESS in 27m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc6b969d76fb481585acf5856d12de25 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fde97f170d2648ce9e5282b5edd3bc7d : SUCCESS in 16m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7d3aa6d74c6442186706d1674afbf16 : SUCCESS in 49m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d4c1dc1e4848b8b30ee62f6f3aa3a6 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":1},{"id":"949444f07c2baabac9bae9014784b218e2bd0660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 01:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20b7263d4c6422370f07ba572991518ead791594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 02:27:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c618a2c51be45229361faeff60bb400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf12cc3fd8094fc28da3ba8de7a2c1ae : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/33e4a2d3287541bea9a639fe4a7e9cbf : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d311ff2333534a639ff9c4264e86975f : FAILURE in 27m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3fc7694097bc461582c9ad30bf92b154 : SUCCESS in 37m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8296bea5fad4e8cb1f846e09296983c : SUCCESS in 32m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7d8fdf5d1080417cada257cfb4015cb7 : RETRY_LIMIT in 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a3b6ca0a7ac40398620432f211eb774 : SUCCESS in 20m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b8cb30bebf8456da5fb45fb2e654863 : SUCCESS in 16m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/af312045bfe640c3a861950ee19bd980 : SUCCESS in 22m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1d16c6d3452e47d98f6fc6ae6a0babcb : SUCCESS in 13m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f699dc8249c74870b5de36a51eee653a : SUCCESS in 12m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fb98e748813147098a0b84d71748fe9d : SUCCESS in 16m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2133fedb8b8b4e919336d0353e169b2d : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8f20626b2eb49cea04f954a2f336237 : SUCCESS in 5m 21s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/77a81bbc2fef41e7b38613c17bfc2da5 : SUCCESS in 10m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28fba2c499334d3da719649df8a9ca6b : SUCCESS in 1h 02m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29c8cee625ba4d53a8cf18341182dc47 : SUCCESS in 51m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89321845ae9e4366af5685b6f62c1204 : SUCCESS in 21m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f6629f251fa457b9176afdd723b7cfb : SUCCESS in 26m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd658291ee054fc8add81036ce39f6df : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72515c98610f49efb45ed6962306382f : SUCCESS in 8m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6deb2ce20a6145e7ad6b14d0301971d1 : SUCCESS in 35m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/639698a289284a02894de7de17877478 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5304f8aa77643d282315d436e55ddbfcc3e5e8ab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-28 03:42:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"950892f15b08fac37a1d8380b0cffee90bfe7391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 04:43:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fc0408c19f548dc994d48ac8f5cf550\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d87234444c425e8e8ad672ad0dc85d : SUCCESS in 6m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ae471034a2d144968e429b81dc6eb752 : SUCCESS in 50m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db6ac2bcc68747cb856dfe63a42e58b9 : FAILURE in 18m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c44d11046d74becb0537737bc823629 : SUCCESS in 43m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f921a99b10246eb903f311d505d0499 : SUCCESS in 32m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ee33ef6a71d4d799997f03a7c6ca6e1 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa6b53eb791749ef88b33259bbc4a65c : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4af904be9de4281abd654ff39382f2f : SUCCESS in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a3630e5d81a4ccd8be89ae3ddc6d905 : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/39e7dfec650b41c08d0f941c370518f2 : SUCCESS in 24m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e2a291728024e5c938e124b142c5785 : SUCCESS in 10m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5a5577113c5449f48a525ed58bcdbf0d : SUCCESS in 18m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c616b4851faa45109579a317e55822a9 : SUCCESS in 29m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/89c8c7224a0945869c6517e73dad5e59 : SUCCESS in 12m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7d1cc537a4944ae9b32c4eacb9e1a149 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3a2b420f751c4efaa5c4eb4b5c2d0833 : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7a11fc457c3407a8ca44c28a495f3cb : SUCCESS in 8m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b166098d01554aa9a98ed7ab921a63fd : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/835d12151e81433e91dc438a7a5e128e : SUCCESS in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5ba24997c8f4f249a754248de2ba1e7 : SUCCESS in 53m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4712ec0726414849a78f743e913a06a4 : SUCCESS in 55m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/efbac3d82d5446fb8cebaf02159eb207 : SUCCESS in 48m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c4027f170e44989a6acd27c867b5a5d : SUCCESS in 23m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80bb9bbc9e449bdbdb4f65578d93873 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0e87a008d6346628e56cdcd34fb94a9 : SUCCESS in 18m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38ee66c6c92f45f69dc1fec79835400f : SUCCESS in 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da1ff7d9c45d49eea190bd9d074b5798 : SUCCESS in 47m 57s","accounts_in_message":[],"_revision_number":1},{"id":"50d8bd1943a7ceafeaadba2d0ca62b19e07a2a98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-28 04:53:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fd81668b71610249192baa7bb6c68ba65eff75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 06:03:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1daff8cf73d43fe8d060b76840cb7af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78de1d2913224cfaa839c9e7e1a4d0df : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/6d3021689aed452cb3387e750dcd3dfc : SUCCESS in 58m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd7d1ae756864a16aba813770cd34239 : SUCCESS in 1h 04m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e7ab5f38f7140c4831737ddc3b4f35d : SUCCESS in 41m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/872971d8dbe241eea89804c3765f48c7 : SUCCESS in 32m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d8269f4dc9a4fbdbae0c8c55887274c : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c57d920b9a054fad90166a48edf40e04 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aee32de6c3a04c31af56ad04ac929a86 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c18b5fc441674a03b38a6a6ff789684b : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5fd036ed559d40d3bee4bdb6afa066e6 : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e76feb74a9de4c7e8fd24ff5592ab142 : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db69dc9e29dd48949d3e8cffc981f0cd : SUCCESS in 27m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2cf3eaa9d2104a30a706196e1f28a6d9 : SUCCESS in 28m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f615b8948e524b3395a8c573c6f8d3fd : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3235628ac3f84d77a5de3e3bb09e14da : SUCCESS in 12m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ec5633816fea4b8e80f593a13da791f6 : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a53b9b7aea6544ffbeac45cd2e43db3e : SUCCESS in 5m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/197812acf05346c0ac73f551e0226187 : SUCCESS in 12m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f11b9dd11b645b89dd826387a6b944d : SUCCESS in 4m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6976ed295f944e18353258bb49c6c06 : SUCCESS in 34m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1144674e44f845e9a3deee72eb8ba902 : SUCCESS in 47m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3cf7c335693a4c3391adf278ee480fc2 : SUCCESS in 49m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfbd2a6d66a54ac284c6c0f6d9956816 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b608c9203c4eb19d8b41710c059777 : SUCCESS in 2m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e5f389c05b44db0b0ba760a9462269e : SUCCESS in 15m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdce95d43430429396143bb352847d92 : SUCCESS in 42m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7c5889518b64ae5b3f0c8b4bfa0ba2d : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1f2ee758c08f433c87f32ca45fe6405bb689c55b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 06:04:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e727877f1c54aae67c0b6902244508d0dbb631c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 07:22:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd3ee1b5c154ff6b65f5b883562f5bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/958210590c764d0f88be1845af19eb3e : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":1},{"id":"610193366aeaa20edbf6809ffc87b82ea61d0c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 08:18:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47f24022ae2f448b9b216d59be560729\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1168cb309f4e5c8f588b560d93460c : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/fd01b80dbb7440d2a656ec8ec78e7c7b : SUCCESS in 33m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/edb4dca855fa4ddeb2ec7c31ccdc7414 : FAILURE in 26m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96a1ac9619ec4b789ab1a9c4c1db0d96 : SUCCESS in 36m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/324bbf1bccd940ce85b80ffe5e949cf9 : SUCCESS in 1h 00m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/03a585931d7746cc990efaa9586a512c : RETRY_LIMIT in 9s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36e8f3cb77ea4a76bc39e02f75245e6e : SUCCESS in 17m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c58ed00b8834ec988a71cdefd4f68ea : SUCCESS in 10m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2accc4b711494e4a97a7323b90e25d4b : SUCCESS in 12m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3f575c84d683480ebc623ecd1b4b4f52 : SUCCESS in 20m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/24f67e01e66a40e2a44227597e84fee3 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/59540e3d6e43451b889faead4f580342 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5a6dd35653394794af7f641bcc786e11 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89084beea2934af3913427c7bfe41c88 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/159ba333c7e74d698fc5fbdf29c8be9a : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9dcc72c36e6431d87ba68a150043843 : SUCCESS in 56m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbb11607ee444f579a8db11dd534a69e : SUCCESS in 47m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b937e3e3ce5c4f659f771de970ed016a : SUCCESS in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62449d6120c74c3bae02d0345ad595d7 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a49608569948a091c38acfb2432b9e : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/029d055d1b594aefa20cc487cfe47fea : SUCCESS in 18m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0927e6742784afe8a36c939fd573f04 : SUCCESS in 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64cd4fc06dfc4bc38121db400a8db03d : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1419475ed9ad6a41f7f194ed313db92397724b84","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-28 15:18:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef683ed7c79f74555a4f2ea18478549a6e5501b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 17:28:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2229926d70f4a87a7df0eb4c1dd46a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de597431e0aa4e8fb632b8d9aa07d342 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/022ada07174f430e8506d3684cc12459 : SUCCESS in 1h 04m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adfa61dfba524478a276e659bcdf59d9 : FAILURE in 26m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6fbeb7713025419d90142024616fe92a : SUCCESS in 59m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73299ed765c54001b7f36dce4213d372 : SUCCESS in 1h 07m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f250313739ba4928b42738f5602057ac : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffc50b49930a42b98b494edf39b24026 : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40c7de50bbb640fa9078c53bd0fc84d4 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68710e4dd0ac4f82804fd5f6719632d8 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4302818b54c439e87e8c7498932bf52 : SUCCESS in 11m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7bd689892e8a45d0a17be2737d8bb9ec : SUCCESS in 17m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a7c764179ab4668a13f1d175a6dc579 : SUCCESS in 28m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/043c4b06194e41a7b83e6f4059ee5e08 : SUCCESS in 14m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bca47dca2dda4c3284e3b24b9724bfce : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c58b9be35c1446d0bdae849e9672ffd4 : SUCCESS in 13m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b301b5ee69fe48a183c18e726fc298ea : SUCCESS in 8m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c4cd78e1bf44a7fbef5b0fee372caec : SUCCESS in 7m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dab89945177b4c178fef0a70bd1f43af : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e137f09d04ab4c3886fa0b3b1ae95739 : SUCCESS in 6m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36f3a5b940e6463483e383e52ae3e706 : SUCCESS in 54m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/982df9df991c4fcabb211ed8609faba6 : SUCCESS in 52m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc88317f774049cba7a86b7e8b4217fd : SUCCESS in 39m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b64f1a779cd473f876e67f6ee872d00 : SUCCESS in 28m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145ea06072ed4f9386a009980f395e7c : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1934b042c3cc41e4806153429846fba8 : SUCCESS in 10m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a7b87c579f1432c81d2bd5ef617c180 : SUCCESS in 53m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2825bcbde194d61bf4a39d98a43d228 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":1},{"id":"860d5e192cad7173e5a9eb55ff7576cd54edd036","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-28 18:08:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6fc30a3701e6adef9c6d19acd23f9555130ba4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:19:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2835a222e060499ba559a762714364cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b518d3f9ad004a49a97b16fd92836543 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/9b41a0816d114859aa2eb0b63393719c : SUCCESS in 30m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ec176a911af4eb88c0312dd5b34f0d9 : SUCCESS in 46m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ff4d51122b741e9bcc65015c4a50b06 : SUCCESS in 51m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3339cd8420644c7bd0cad4671ffbec2 : SUCCESS in 1h 06m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a78cb41cff324e488232c17277dcae3e : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d5258e5efcb46c0a7f8f60e75dfeedb : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/369c8dad5b294562b9f2b1cea26bdd60 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7512b3c776e14a2aa7d39fea3902f1b5 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf8db36e05684963a30adc6af5c43dca : SUCCESS in 23m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e67882e1964f04aac6821f83d20b48 : SUCCESS in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6806a7f6fccf40ccbe5c2f7b1df2b72b : SUCCESS in 20m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/786e32ce19ca4b9e95c9bbd96c919977 : SUCCESS in 28m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e475e9c09bf44c490699e9bd6c1db38 : SUCCESS in 7m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e034fbec4b46444eba81079d71608406 : SUCCESS in 8m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8ebd2da280034612beee2e6e1b0494dd : SUCCESS in 9m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c7d36541fa0a4414b0ac96bc87247458 : SUCCESS in 8m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/781b0bf21669487bb853f9791d09905a : SUCCESS in 9m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eabd3273433043e8a61988ea4bd5b822 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c87c3c29b56447f2a0cc56a9f508510b : SUCCESS in 59m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e4c254892fa4f73953240f592e6b54f : SUCCESS in 52m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4848847146f94a33bfb925f2f55824d8 : SUCCESS in 45m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce9932ed886a415c9c2e8aecc25ee876 : SUCCESS in 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/204054dcbccd41ae9c7f96e0d680e586 : SUCCESS in 2m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3bbc8afa7da441048f5c1f3352ce32ff : SUCCESS in 17m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8e2b3a44a3a4ffb910e28b331214a3a : SUCCESS in 52m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/768a238af72a4dfaac82f41dbefb8176 : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b640417c7f285f70010b643e146409cbe695469d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:20:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1f2a8e4e8cddc5402d89d4cd47c0d0b4ee300e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 20:47:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fe7ab4b57a4d64b7cec65c7b08737d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fd268a86b64901808d48d20eb46e6c : SUCCESS in 6m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/30d307ce2b544f04b8f376c5eac1ed64 : SUCCESS in 1h 06m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c71f6159ef5246d0a51409a45609a6b0 : SUCCESS in 1h 01m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0dc3800965914701a7f77c4a00085ec9 : SUCCESS in 1h 00m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6ff5b561cff4b6ea066c6d33e1bda46 : SUCCESS in 38m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b01418f13cce43b08eefa53d6e1e85cc : RETRY_LIMIT in 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca3c2ecab10b42f39fb591002958446e : SUCCESS in 19m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c3398ed14ba6433cab7577e04f62c4a6 : SUCCESS in 17m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4911581dc73d4e2e858d00c6ce6049d8 : SUCCESS in 19m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6ab83baf7a7448b7bcf546391b525bcf : SUCCESS in 13m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9fec2623c01a476a8bea3d7b8a2223a6 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/570cdbe4237f49c3bb4a27367390abf7 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7bf098f106934aaeb4accdd920b4bf1e : SUCCESS in 9m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d97305f833c74ab4a10e52bb687e0c9c : SUCCESS in 8m 22s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/542dd5664e3e4c29964102abe1debe01 : SUCCESS in 10m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68cc28a2605b4ea39e8f81f27bb86f18 : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0aa6ff9b13dc4c1294c5c6b8bc2adf75 : SUCCESS in 54m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61912728499e402eb6e410552e5d85dc : SUCCESS in 25m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f1300c9e07e4e7f8d0424cf6714ea86 : SUCCESS in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3269e29698b4c2aadd94b9a6c6cfae6 : SUCCESS in 2m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42f276cdd66342abb1289b67fc72223f : SUCCESS in 17m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b894ae0976c44aea315adab9e06bd97 : SUCCESS in 58m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f16704e2c5f84358b092441f6df6769c : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dccd382bd25ba4dfc96bd0a98055b111ed4b4677","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 20:47:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b8da30e46ac60314bdcacd84a36f5307454b79e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 20:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ff1717132a946a1ad87a91f41a2393d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603f205ef2ec4e279b23f507270bfc96 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4aedb518a3514f06a6afd214219153d3 : RETRY_LIMIT in 5s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96e18f7acee6cdf99a7fc6d1edcfc34d1ebc1360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 01:17:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a723e73e4b14d36b7969ac2fe06fc9f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32631ba296224a9a8a6c4b8dd1220ce9 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14a470928f8b90f10c6f70f69c50a1ef584cb549","revisions":{"14a470928f8b90f10c6f70f69c50a1ef584cb549":{"kind":"REWORK","_number":1,"created":"2026-01-27 17:57:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/974910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/974910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/974910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/974910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/974910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/974910/1"}}},"commit":{"parents":[{"commit":"1922c2da2a642ec77ba0c74461a53f2e280cec7a","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1922c2da2a642ec77ba0c74461a53f2e280cec7a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-27 17:49:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-27 17:56:37.000000000","tz":-480},"subject":"[stable-only] Ignore bandit false-positive","message":"[stable-only] Ignore bandit false-positive\n\nbandit 1.9.3 started looking for hardcoded passwords in dicts, which\ntriggered this new false positive. We fixed it on master in the related\nchange, but ought to fix it on stable branches, too.\n\nRelated-Change: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I6e0c76dd9157e8dc265eb67cec486e0db782b9fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a470928f8b90f10c6f70f69c50a1ef584cb549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a470928f8b90f10c6f70f69c50a1ef584cb549"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
