)]}'
{"id":"openstack%2Fswift~974919","triplet_id":"openstack%2Fswift~stable%2F2024.2~I541ca3270870cd46952045ce28fb443f31e8d145","project":"openstack/swift","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-29 17:04:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I541ca3270870cd46952045ce28fb443f31e8d145","subject":"[stable-only] Update rolling-upgrade job","status":"MERGED","created":"2026-01-27 18:29:39.000000000","updated":"2026-01-29 18:27:34.000000000","submitted":"2026-01-29 18:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"974919","meta_rev_id":"365683220a7322f8e2876f022db0a967dc1f5e4c","_number":974919,"virtual_id_number":974919,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 18:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-29 17:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-29 17:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 18:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-27 18:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 09:36:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"0f15048cebb69d4b7f046527af43c353273e4782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-27 18:29:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31a70a8a65dbf2c06d6713a436c1b126c47758af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:30:15.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch stable/2024.2.  The\nproblem was:\n\n  Job swift-multinode-rolling-upgrade-victoria not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - integrated-gate-object-storage\n        - swift-jobs-arm64\n      check:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@stable/2024.2\", line 618","accounts_in_message":[],"_revision_number":1},{"id":"534e1237e07b9ef09b16a9b1d5ee5a7a0e9ef44e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:30:46.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch stable/2024.2.  The\nproblem was:\n\n  Job swift-multinode-rolling-upgrade-victoria not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - integrated-gate-object-storage\n        - swift-jobs-arm64\n      check:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@stable/2024.2\", line 618","accounts_in_message":[],"_revision_number":1},{"id":"a49caea9a5dedb2c9f83bb46d4f3dd425f742db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-27 22:05:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d9ebf09c47f103bdd9fededa64ac5c4ef4cd2fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 23:16:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85dc1dc275c24761b5fcc2707d77469b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ee3f79a1f9421aaea3827da057619d : SUCCESS in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/178fca4434a14bbc83b246a56356bcf1 : SUCCESS in 1h 01m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/136cd3a2c923491caae0f900bbc0fb35 : SUCCESS in 39m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58d30283f0964cbb8f94b3410c1e2fca : SUCCESS in 1h 02m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19b9be2a1e664acfa481b3e062b52b56 : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68a037371ead4c9c97f8cbded2799c8f : SUCCESS in 11m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b248972caffb47c1917ae81377f3fe5f : SUCCESS in 11m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/694677802c8c407ca909e0f5907d9d42 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/264fd14fde064a59b56c45882defdc28 : SUCCESS in 26m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/731f0247b2ca449a820cb08d1bbed664 : SUCCESS in 17m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/59a9384b78ba4bda99c4a73c1bc0932f : SUCCESS in 16m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cc43acac35af4eddb5f8278c1d479bdd : SUCCESS in 22m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/23b90927de954a80a6df1bd173db16b5 : SUCCESS in 26m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b6e33b0ecfe4254875ecbb83017663e : SUCCESS in 23m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/485607da5480413594f38d9106b94e5d : SUCCESS in 21m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60aee62da7c945c998a4ee05d704a15c : SUCCESS in 23m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d3d052bf462d43b3871d3c103f1cebfc : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9839fb86ecdb444db3663e9598c060f6 : SUCCESS in 12m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc66bcd0e64143c18da645ba6836e3fc : SUCCESS in 18m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/852195c59d9d43298a2e521c4cbececc : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f75f1ac55af41c480d4793978e2d554 : FAILURE in 17m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/08c85f765b984a098d71667651bd35be : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f71cf4bae3ad403b9783240de0045783 : SUCCESS in 30m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b14ce6714c8241449ec2bf3a952ae98b : FAILURE in 13m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c993b714296d4523ace01204dcced421 : FAILURE in 23m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc837b07e4784fcfa39b469589617da5 : SUCCESS in 25m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9323947a4394391882b8d9e27343416 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29e2ab21c9f6468f95fa19c39206ffc0 : SUCCESS in 9m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44e2e246d3a243d390904cfe9ea1b7c0 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":2},{"id":"fb6f6c3163a106ce940087705d620088ffde30b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 07:20:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8129f5f6f3b14d73a7d4d50d2d3f6db9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/825fbf9f98504d818528c4cedd57c20a : SUCCESS in 1h 11m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ca7276cd4454c95b2575cef7f4b237e : SUCCESS in 1h 09m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/03c7706fe09141028078ae60402d7fdf : SUCCESS in 1h 01m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/695ee3bd9fdd456b9fbfa2a97cd1ad7c : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2f916c38f42c6fe4096a793083c053a24b3ad6f9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-28 09:36:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9c9f213549950fa07888512d68bfa00c1b3c165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-29 04:55:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"419c16a633bef3d198b7f6ef3e4f5ae8e6762d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 05:58:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/746a248a60f24f028c4486403c25cdab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d6a0276ca54efca263a90481af7a6b : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/9d02bf8a914e4062bf1b17d5c36258e9 : SUCCESS in 56m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9c58871c09f8419f95a343850f43b5fa : SUCCESS in 40m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aa79b7c5ac14c4d9e721c8505cb5c71 : SUCCESS in 56m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/122a060ac0b049ab9f885653c288145f : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e50be0a02a574749ac9815edca14c5ee : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3508210adb744494bdc5af190fe1418a : SUCCESS in 12m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5bc359b186a46b1a6baf9bc9d5bc19c : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/178dd0e41e414910832bd939e13e521e : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c755bd4f66374af4affe97298efa77e9 : SUCCESS in 9m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30ebf173971043368bb1c7d4e228da75 : SUCCESS in 16m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5757e5142cbd49c989541353a4139245 : SUCCESS in 15m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b99d4b038c540e0bb9ec010369e9c29 : SUCCESS in 9m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bbe15ec21ed9405b991c615dcfe3cf2a : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df1f5155e6dd4255a8dcd0543b53bd39 : SUCCESS in 21m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22ef8e443e53450b9cfbbb47c902e405 : SUCCESS in 14m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/53fbad1363224ffebac8bcba90fa82be : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d95c485d359e4f0c86e399117eba55de : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b47af56072744c25a5fda83763724d36 : SUCCESS in 18m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62d35c4de93a4c1fa2bdecadf4c9217f : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69b59995f9544e58b7d2544fb77a59a6 : FAILURE in 20m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e386acbf23264fa7b71120022728f024 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae564f847f074dcaaa1f0b5e3d8e9dc5 : SUCCESS in 29m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52b32663a682492ba054ab0191ea14d5 : SUCCESS in 24m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/caf08b8cceb44d348767302e88fe4508 : SUCCESS in 24m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fe64b6b908a414c91e9ef2e8834d507 : SUCCESS in 24m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283058e7e4e64c22845d9c3b263760ea : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9186ee6b77aa4c218f2c52fb7f35bf27 : SUCCESS in 20m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b005156f474b41c2943fcb3cde284ce4 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f121c24fb47dba367ad9b83fa2c2fb0976178c94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 16:55:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4556974e9d4dd3962ffd4e25c53521\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a68bf61fe61b4c15be4f80edc6eed6db : SUCCESS in 59m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/517da4a814f54f72a08cc1265609fc81 : SUCCESS in 48m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6b380a701a94ea38d56a1a24118473c : SUCCESS in 52m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1ce9818898443ba83ed53f9ae893f0f : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ca69b91b47ac2094712109c4e5b9bace764514f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-29 17:04:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3faeeee7ee43b180e8efad6e9b3f745556789179","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:05:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b2179ac916025793e5221ffbfbb1c379b5dcef1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 18:25:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1bf0dc8b5246b78b5c7453cb4dcd96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14b87ab5378452b8430f8f54bd29446 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/8520a23fabd84eb39ef9d6252de8df24 : SUCCESS in 27m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b51881595f9f4e5995866793f810e07d : SUCCESS in 35m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/106014340f824c5c908daf619f54f8ba : SUCCESS in 54m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4139e200f6cc4f0db438d02b03eba103 : RETRY_LIMIT in 8s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/758cafc61ffc42f0ad36a2a7502fdfd2 : SUCCESS in 22m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f31f832ad02945fcab656098f3492a56 : SUCCESS in 19m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9dfe89f1db64ab68bade54e569df5ef : SUCCESS in 21m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a8392793e0764ce498742d5563f29f5b : SUCCESS in 14m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44486923c30e41c4a9cfdf662b6e5814 : SUCCESS in 19m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ac03112f7a14282938458497b59ac61 : SUCCESS in 22m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/04bea0b9d10447489550a485237e9dc5 : SUCCESS in 22m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5dc7f2ff39e846cd947b71612bb29fda : SUCCESS in 20m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/616c4060c2b041469666e32c79c75dc9 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7987fd97f4fd4f08bc9fc7553ae74d6b : SUCCESS in 17m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/59b8c377e67649baa285b1eb48830754 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e87740a10ec425aa6a5850353396b8f : SUCCESS in 8m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3f0b5e9d38ce4fb6b6ccc1ab9bc9b4ac : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/85835d92f1a543f89fb061cd3d3d4b79 : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/675b34a4060e4609acbf17423be8ff44 : SUCCESS in 23m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4aa4aa5b0872488b9685883de94744b7 : SUCCESS in 30m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef29f9f625b545df9c525bffa0a370f5 : SUCCESS in 22m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a93cf560032e433c85be3161551a14d0 : SUCCESS in 3m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/007c36d7995941259624c603e0084a1c : SUCCESS in 18m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cf58b62b72a43bd99568d3f89a02bac : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9f4f197b47524b7d62f8424c22e2aa5b5aab7523","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 18:25:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"365683220a7322f8e2876f022db0a967dc1f5e4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 18:27:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff4e17b225294768af59ae0436f88d67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4851cbae784ef18354bc4d6308a864 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f178041239e44f63aa95e6477e7a1274 : RETRY_LIMIT in 11s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8292e7c72f6487dce97e127fbf907e8a9e248987","revisions":{"8b2eb9ac3eae106bb5cd5fe726ba6154cedcd0e1":{"kind":"REWORK","_number":1,"created":"2026-01-27 18:29:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/974919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/974919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/974919/1"}}},"commit":{"parents":[{"commit":"90ff83b613080ec90743c5af0c9fd18de0a0aa9a","subject":"CHANGELOG for 2.34.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90ff83b613080ec90743c5af0c9fd18de0a0aa9a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-09 17:22:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-27 18:29:26.000000000","tz":-480},"subject":"[stable-only] Update rolling-upgrade job","message":"[stable-only] Update rolling-upgrade job\n\nCaracal has gone EOM.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nRelated-Change: Ic50ee837492e84750b4ca6edd97595c761725c42\nChange-Id: I541ca3270870cd46952045ce28fb443f31e8d145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b2eb9ac3eae106bb5cd5fe726ba6154cedcd0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b2eb9ac3eae106bb5cd5fe726ba6154cedcd0e1"}]},"branch":"refs/heads/stable/2024.2"},"02a60138d6f5084c791d0371f6fd8767e4978f54":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-27 22:05:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/974919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/974919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/974919/2"}}},"commit":{"parents":[{"commit":"ac926959ddbfca66610088ee05282f0aa5b0b2c7","subject":"[CI] Remove old experimental rolling upgrade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac926959ddbfca66610088ee05282f0aa5b0b2c7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-09 17:22:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-27 22:05:12.000000000","tz":-480},"subject":"[stable-only] Update rolling-upgrade job","message":"[stable-only] Update rolling-upgrade job\n\nCaracal has gone EOM.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nRelated-Change: Ic50ee837492e84750b4ca6edd97595c761725c42\nChange-Id: I541ca3270870cd46952045ce28fb443f31e8d145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02a60138d6f5084c791d0371f6fd8767e4978f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02a60138d6f5084c791d0371f6fd8767e4978f54"}]},"branch":"refs/heads/stable/2024.2"},"8292e7c72f6487dce97e127fbf907e8a9e248987":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-29 04:55:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/974919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/974919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/974919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/974919/3"}}},"commit":{"parents":[{"commit":"f82513938564691c570cd7972f3c64b03055b7a2","subject":"[stable-only] Fix CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f82513938564691c570cd7972f3c64b03055b7a2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-09 17:22:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-29 04:55:37.000000000","tz":-480},"subject":"[stable-only] Update rolling-upgrade job","message":"[stable-only] Update rolling-upgrade job\n\nCaracal has gone EOM.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nRelated-Change: Ic50ee837492e84750b4ca6edd97595c761725c42\nChange-Id: I541ca3270870cd46952045ce28fb443f31e8d145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8292e7c72f6487dce97e127fbf907e8a9e248987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8292e7c72f6487dce97e127fbf907e8a9e248987"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
