)]}'
{"id":"openstack%2Fswift~975483","triplet_id":"openstack%2Fswift~master~I0ce4d64acfd3ab5e41f5f4b3e255cadf27b8b063","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I0ce4d64acfd3ab5e41f5f4b3e255cadf27b8b063","subject":"CI: Only promote \"latest\" docker images from master","status":"ABANDONED","created":"2026-02-03 01:49:06.000000000","updated":"2026-02-05 05:01:57.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"536cd8c11e95bc3f6a553ae0d1e839018316603a","_number":975483,"virtual_id_number":975483,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-03 17:27:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-03 03:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 19:17:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"274936279bb054cc29ff6181d64576778e161563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 01:49:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"472eae9a5ac7f2f675ba87ea465cf6d9e0c164c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 03:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e3ccd0feba043b39aab8e8fb422dd00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f319e1237d44ee8a3e675fad2264560 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a359e27444eb4ee5a257a6eaa072ca43 : SUCCESS in 29m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14488ae65ce640b690cf942daf1fe4f3 : SUCCESS in 30m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c9c59906f964a29afaf3110ad2f9df2 : SUCCESS in 31m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a16e2c6f97e749258ec9b8d1badda96d : SUCCESS in 1h 00m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3dfb342b709f46eeb8526cd7c5d46496 : SUCCESS in 22m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69416bd98f7842b4ba5c0e9a50ac7917 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3c1c4ce7dcc4529898e743da44ef7b5 : SUCCESS in 22m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c74538cd0d1642bd809c949183348f80 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6025e81eb57a4c8eb96cd09f0244fedd : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb94846f16a6403ca324a14c34504594 : SUCCESS in 7m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/697b7b0a751245068c5a0d81a840568d : SUCCESS in 16m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c038356137704e438b00d7004e12a8f3 : SUCCESS in 31m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b00a6e77aec54f6dab18bd5c5f35bbe3 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f4d224f193d4453a52e6b67f6f800c3 : SUCCESS in 13m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cae7ef2e394e4e729bb2d8a63f345dfc : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d1372761c644a2d86283d18cb51d77c : SUCCESS in 6m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/68a29de99d4e46458d66445e7e1fe6a1 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/50b5d944d72745cf8f2daa014f91cd5b : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/167b5b37dcdd46dd836d02f3af0e161e : SUCCESS in 59m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06fb125ee527463fa1e9fd2de66408eb : SUCCESS in 21m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6bbb32f8f3146a480902d57af39f123 : SUCCESS in 27m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be25063ec07e412085724a3b5cb73b09 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffee763862134ab1945009acbcb7f06e : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8652fc06f21b4bcead7e8db8ff39b177 : SUCCESS in 18m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b88df57474b648ba847fa0ca51d9ac7e : SUCCESS in 57m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a6bf7f201924f8bb488093c4958fd13 : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2598df53279b0b80486f397e40d79255bd5abeec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 05:32:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01aacfd5d95946baa870e6d9b0fecd0a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce4d09bad6b4487799b838fb1558ba01 : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ae8f59567bb6ead4bd02ff53cebbebb75f79c74e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 16:13:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"38d1a7fe22507ddba06031a82f598bbf85f093d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 17:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0637997ebe784c1e839e2a41751bdc95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45fec5869d8845e88968ebf38d4469fa : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/6f7cb6c72b0d4502ab17a8b53f4ec52d : SUCCESS in 58m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1534597e0ce04b54a385614fe1ea9d9d : SUCCESS in 29m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfe89f1ba90b462cab5e4c6ae3b1ce1f : SUCCESS in 56m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50c1a8d9eb9e48af95eed60ebdc87029 : SUCCESS in 1h 06m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/363bf0a39b69487ca5b28468c25a94da : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad852c258f914070857d567bd5268c25 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d4b365a6de04a1292a8c8a81343b629 : SUCCESS in 21m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/581fe0ceb8f74414a8a7f0df8822f7dc : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3666be360b234627a89904516d677322 : SUCCESS in 24m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e3888e2af1a4ee5bb12bb47c4e3719d : SUCCESS in 17m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b79a2a30a3164309ae8c09f4797ba065 : SUCCESS in 31m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c34a8daa8dc54e67b384890f8240b1aa : SUCCESS in 13m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/53dee52d01d549778819594d54376998 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40efa3ed4aa445c7817539dac73010d9 : SUCCESS in 20m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d799d04919349e0bef748e2c9491cf8 : SUCCESS in 15m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8fe75211c97b426da317fce034c455fe : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4421618d88c04e77b6e4188d056618d2 : SUCCESS in 12m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8fa5e11a8814a249e99a1d1e6dd7aab : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bed564ae28044465b00dcb5ff64a965b : SUCCESS in 1h 03m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e366f302af234194bc99fd62dd59e0ca : SUCCESS in 53m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eca26018db6b4753bbfc96ad29d9d328 : SUCCESS in 52m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b34685133e7a471c8ca451c481eff7bc : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039aa022c3c740c890fce12a03dfb037 : SUCCESS in 2m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/173bb37dbc3449769b3d270d4f365c57 : SUCCESS in 9m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daf454b1d279403e88feb37878d664e5 : SUCCESS in 48m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86c0c494e64c46a6bf7e5a0319a0978b : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7f6cc063798a3b10fc7f5bed5cbff9fd10837472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 17:37:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b813e0c0994d73ad996be2368de51e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be11e266d92b49aeb36f2b4c3e353282 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4ad5abd7a5aedfa2326d7b10a47db88be14109ef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 19:17:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"536cd8c11e95bc3f6a553ae0d1e839018316603a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-05 05:01:57.000000000","message":"Abandoned\n\nProbably better to focus on https://review.opendev.org/c/openstack/swift/+/732480","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91cc738878ea68e3c43a4113e3de63111690ea9f","revisions":{"e11c52e4c69d9a00c8bb159afb552ee09d92f14f":{"kind":"REWORK","_number":1,"created":"2026-02-03 01:49:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/975483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/975483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/975483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/975483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/975483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/975483/1"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-03 01:44:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-03 01:48:55.000000000","tz":-480},"subject":"CI: Only promote \"latest\" docker images on master","message":"CI: Only promote \"latest\" docker images on master\n\nChange-Id: I0ce4d64acfd3ab5e41f5f4b3e255cadf27b8b063\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e11c52e4c69d9a00c8bb159afb552ee09d92f14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e11c52e4c69d9a00c8bb159afb552ee09d92f14f"}]},"branch":"refs/heads/master"},"91cc738878ea68e3c43a4113e3de63111690ea9f":{"kind":"REWORK","_number":2,"created":"2026-02-03 16:13:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/975483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/975483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/975483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/975483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/975483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/975483/2"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-03 01:44:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-03 16:13:24.000000000","tz":-480},"subject":"CI: Only promote \"latest\" docker images from master","message":"CI: Only promote \"latest\" docker images from master\n\nChange-Id: I0ce4d64acfd3ab5e41f5f4b3e255cadf27b8b063\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cc738878ea68e3c43a4113e3de63111690ea9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cc738878ea68e3c43a4113e3de63111690ea9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
