)]}'
{"id":"openstack%2Fswift~976810","triplet_id":"openstack%2Fswift~master~Ia9e5782ee8f632951412898895a6134464e9dc3e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-02-13 22:13:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9e5782ee8f632951412898895a6134464e9dc3e","subject":"CI: Remove jammy job which is not required to run in this cycle","status":"MERGED","created":"2026-02-13 19:52:54.000000000","updated":"2026-02-13 22:15:37.000000000","submitted":"2026-02-13 22:13:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976810","meta_rev_id":"bf434dd389f1e394804b397a24e601b43fe1fb04","_number":976810,"virtual_id_number":976810,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 22:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-13 19:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-13 19:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:58:45.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-02-13 20:46:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-13 21:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f58186e9b9edd0ae98d24f5dcac13644c3b086ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-13 19:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcaa6446e9acaeafc78dfdafc96a7bae2f1b8dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-13 19:58:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"298792950d5a12a97936559606b28cbb842de630","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-13 19:58:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"65c055d8a97e0042b1735fe8fd082eb1c236186a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:46:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2fe97dfcce040b29f80ef133530b1bf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2757a368f5c147058bc427f6c7a09edb : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8b42ba44be96d8e0b5b7bb8f94b2c6d6294f5ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 21:09:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce12bd7b6aa48988c4ad0367fc52686\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e786cf19cc934b9ca97c9e42f799f262 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c6c6ded18ab94cd7b3bb23d66285011c : SUCCESS in 1h 06m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ab60e43c3f44db6ba47a05c8b00be57 : SUCCESS in 1h 01m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad2a5ef4be904ddc9f35adfc80abe532 : SUCCESS in 55m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f53da56cc5e437bbbadbe78fafffe02 : SUCCESS in 1h 08m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae6e9d40b7674a54bf8a3fdd94b09681 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40c1c9743e74413ea8cc6aabc0e5f894 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6b0087f25464704bca003b9f5237475 : SUCCESS in 22m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7eea6da378ce4996a0f080529392bd39 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/110ba55f8be34356b95254cae00832ec : SUCCESS in 10m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e78aeb02ffbe46ccbdd1a3896e6fe166 : SUCCESS in 16m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f3c9a8861d447f8a3d7bed7821ff483 : SUCCESS in 19m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/362e1aa5a0994a6faff60991b979fd7a : SUCCESS in 29m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/817c16fd2e274440b630322645f39f94 : SUCCESS in 12m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aed17c78a3d540d1a32483c2d5eb2d47 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/133d70b03cad45088f720247c7809c5c : SUCCESS in 9m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e1db3aa5cc243a9abbe46fb3bbb1789 : SUCCESS in 5m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d6c995d5ed74a58b432ed9fd774dd57 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8ae005a149244613bd3e89e86bc5de9a : SUCCESS in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c24f0ca6c5844cd4b859d255c5b4bf9d : SUCCESS in 49m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0cca88c27c5644739d566cec01641320 : SUCCESS in 33m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36283245c96f4b16aa3903e3e9c6359b : SUCCESS in 32m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faf4f2b613b64ab588328e005c62f3e7 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f28350822948f89bde967bbd9e07a7 : SUCCESS in 2m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90877c8387304cc4915bed4fddee965f : SUCCESS in 18m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c54fb1c71d54ad894aaf7ba500ba5de : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":2},{"id":"194f88ecb20b0c6f5b59f6861926c8ba5faefca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 21:10:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fee4c4448f7a5f1bf0e8168e806a1eab48b62a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 22:13:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54d2971242f54329b035090c71d42dcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5536f91f53df4e21ab301d5ee26b7827 : SUCCESS in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/85e2481e10ff4d5fabe73409fb58ece8 : SUCCESS in 1h 01m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/338d7991dc734daaba6e6105edaabbe5 : SUCCESS in 46m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f878a62a0af54cebba360f7caed1634b : SUCCESS in 29m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5224b9bfa8414eaaafccf2e1d241d424 : SUCCESS in 50m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/05557df6f1be4da6a34cb94183969ec8 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a4d537b812043268aaf1a1564c44eb2 : SUCCESS in 11m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02c6194fae8641a8a86a6054f1338e6f : SUCCESS in 9m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6dc19208fabf4f048bc5e80046c48e51 : SUCCESS in 12m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/af928e7061614e4a99f673733a50bf53 : SUCCESS in 10m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dac29ab1f18b44108dbd0d758b6a062f : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aef445862d5d4271984ac1c24b68e2d6 : SUCCESS in 8m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a570a860bba341d3a04b0d6d2bfe72b5 : SUCCESS in 8m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9a01d20d26a45c59e1e2c5532657d01 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d2e34450b0bb4fee94d56446f3784ad0 : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8a572917546493eb1c73d9d290ce09f : SUCCESS in 56m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4c02d01c6ae444ebbd30e82501f8dcd : SUCCESS in 49m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9ab5913fbb1412f8160180301d60710 : SUCCESS in 24m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc5adbb7881b46aeaa8391bbc7f9bb33 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d1fd63e62e4dad89502b2aaf027eeb : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33cf7499475744dea350d829265da334 : SUCCESS in 8m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c37eff84c47b4041a0dd9579703b133b : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":2},{"id":"c5b85ec8770cb90c9d50c8c6014213aecaece7bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 22:13:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bf434dd389f1e394804b397a24e601b43fe1fb04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 22:15:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df5200d58ed472dad679acb3a916604\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7311db7197d41509b4321d784cab138 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c6d24a937ccd406686e49832517dc4c1 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c051a0b1a0a3a4632305786a4f0c102794dfe7f8","revisions":{"9c445863a070b2e48e21ad80bf5ec70f1bf1ee1e":{"kind":"REWORK","_number":1,"created":"2026-02-13 19:52:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/976810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/976810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/976810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/976810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/976810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/976810/1"}}},"commit":{"parents":[{"commit":"06c182145163de395f9524e8505550f070e4ab38","subject":"trivial: remove print statement in relinker unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06c182145163de395f9524e8505550f070e4ab38"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-02-13 19:52:06.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-02-13 19:52:53.000000000","tz":0},"subject":"Remova jammy job which is not required to run in this cycle","message":"Remova jammy job which is not required to run in this cycle\n\nJammy is not in testing runtime for this cycle and we can drop\nthis job which is also failing 100%.\n\nChange-Id: Ia9e5782ee8f632951412898895a6134464e9dc3e\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c445863a070b2e48e21ad80bf5ec70f1bf1ee1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c445863a070b2e48e21ad80bf5ec70f1bf1ee1e"}]},"branch":"refs/heads/master"},"c051a0b1a0a3a4632305786a4f0c102794dfe7f8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-13 19:58:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/976810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/976810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/976810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/976810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/976810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/976810/2"}}},"commit":{"parents":[{"commit":"06c182145163de395f9524e8505550f070e4ab38","subject":"trivial: remove print statement in relinker unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06c182145163de395f9524e8505550f070e4ab38"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-02-13 19:52:06.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2026-02-13 19:58:39.000000000","tz":0},"subject":"CI: Remove jammy job which is not required to run in this cycle","message":"CI: Remove jammy job which is not required to run in this cycle\n\nJammy is not in testing runtime for this cycle and we can drop\nthis job which is also failing 100%.\n\nChange-Id: Ia9e5782ee8f632951412898895a6134464e9dc3e\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c051a0b1a0a3a4632305786a4f0c102794dfe7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c051a0b1a0a3a4632305786a4f0c102794dfe7f8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
