)]}'
{"id":"openstack%2Fswift~976890","triplet_id":"openstack%2Fswift~master~I4db72dc72a274838cbc4cd54e8ce17770eddfd0a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-02-20 13:54:43.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-02-20 12:53:17.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I4db72dc72a274838cbc4cd54e8ce17770eddfd0a","subject":"tests: move test_db over to BaseUnitTestCase","status":"MERGED","created":"2026-02-16 04:16:08.000000000","updated":"2026-02-20 13:55:44.000000000","submitted":"2026-02-20 13:54:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"976890","meta_rev_id":"9c8ae6a66244855583e8838bea6dc56c36c39337","_number":976890,"virtual_id_number":976890,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-20 13:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-02-20 12:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2026-02-20 12:53:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 05:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 17:24:33.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2026-02-20 12:53:17.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"5dea3f2b985ab6956e7a88d3d82308e226c3fb75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-16 04:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e84308c730d564950a2510ce25b35ff8cf3f626f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 05:01:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e67468bc585c407799f7eec2fa0476dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18836dacfe064769877764a183f9a6b0 : SUCCESS in 8m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/486c3b997dfa42ce81f1b9be1c90ec89 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40732df692e2400abeba2ef1f293b2f2 : SUCCESS in 11m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc9efeabad2a424e82e755f91c102a00 : SUCCESS in 15m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85f4938b202c4921b7a49fbea91da416 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0079fa2a39b5496cb0129043b3fa05ed : FAILURE in 12m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cbb9a73cd73946ccb49e018211b86eed : FAILURE in 15m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/11903b043aa9413baa9471285509da59 : FAILURE in 31m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3eb5fe9e931b490789795092da7e0e07 : FAILURE in 31m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/87ee88d60cbe4272a000e9251c490254 : SUCCESS in 11m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3fb97f6bfae746628b50606926dddb27 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/beadf9cf4d46486b9a14d1bc60914895 : SUCCESS in 13m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5ebb081998240beadbe324abb5ea4d7 : SUCCESS in 39m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a87910aef9d34ee2923a3804979686fc : SUCCESS in 25m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb798000a3c145f9a66022e44b5b2f15 : FAILURE in 21m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be693709f36647c9aafd2009d993a95e : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/faeca000d3c14babb467a66071f60733 : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5b30e47b804cf51116446212cea71ea6602e034a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 05:07:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a75552024324db7a07ba29af761fd25\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f7c4c2315fa41cc898f8bc631a30a21 : FAILURE in 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d92f594f29652425a579d3c198b580efaa7dcb33","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-16 17:24:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"edb43f21ad1fe08f31676637d4484afa5212c8c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-16 17:42:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"88e15e72a6d7f0236e8f930aba730f23f24e6d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:48:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1ff08960f25447f9f55f14500398430\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32761efeca324068896e52aa499aa101 : SUCCESS in 10m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a880af60984346879326b163cad6723d : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ab57d6cf3054680b6e95e82acd6a01e : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c631615c828547f696d00b58b2d39fa5 : SUCCESS in 17m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e965f2908654146b34080bf1de9ac28 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/53216ceecdb04a3a892d0b33de9e9eb1 : SUCCESS in 12m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/872975ea3756491f982a70fe3b6ee450 : SUCCESS in 23m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8aa21ca5f6a74395bacea206f1b076b4 : SUCCESS in 29m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/98b09523b643498683c01ffb2eb15abf : SUCCESS in 32m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/82a1b1e9dd3a4748acd0f8bdc8f631c5 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c482148d561e43bdb64b417badb11a40 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/77c45468a828460880aa6edadbc2963c : SUCCESS in 14m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aec61c901dba4b408717d70fb28088f6 : SUCCESS in 59m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03e173bbb0d7452a80ae101949662832 : SUCCESS in 47m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b00ffa388f141d89b7adbd1ccde70bd : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1ab2f8317b4297bcc7398e5d7666aa : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24b0ebe20db7402f9f3402ccbcd9e93b : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f967823da18693e0c4151a41364450736183b6cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:52:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e25ad15211a41eea6a489cde7715514\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b018990b86b649c6b8e3ee02cd08489f : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2487e65c6f336525c7342c9441ac9742ba3843ac","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-17 04:24:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2022edab08080563a76356abcc7f6a7742dcc3fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-17 19:12:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"737df4570f79a979db127c718213b83b10a0a7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:53:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e0b636b045a43f0be97343a184f47d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2995582c500645b6b3031274700c8bd8 : SUCCESS in 8m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc22249b088847c683c36f474eef223e : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a6f40640242476abecff8906767f431 : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f58ce144e0214b99b4a7553a0e371334 : SUCCESS in 21m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/385e7ea81fac41b6937e8e77b620aa67 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eb9096ee89b143129bbc690ce9970576 : SUCCESS in 25m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8034a7bb5f4a89808cf80a8f83f6bd : SUCCESS in 19m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a8c79ec144564091873482a28934cc11 : SUCCESS in 23m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8ba2dd0929fc4172a13a091ea6022f7c : SUCCESS in 24m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b864b2d189774bb4a20f873e8b475ff6 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c75c89d88ad142f2b937eaa435607eb0 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2db2811586b24a15a074de7f3dbfc30a : SUCCESS in 14m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f562ba7c46d246498a0dcd322d6d7a8a : SUCCESS in 21m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15044034fb6241108b3bebab4b3b8d12 : SUCCESS in 39m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6607188740d34cb1b2c30725b057a484 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae8f1d80aa84fc9986c0f65f64c07ad : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04561144ac48491ea1943e1fe0cc5e02 : SUCCESS in 19m 28s","accounts_in_message":[],"_revision_number":3},{"id":"3c422d2684d6c7c0108e858ac0630999f3b4dc8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:55:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0316d4d39a6344eca2147067ba0411d7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56ce9165b58a4e2abda4d2efb5023961 : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":3},{"id":"18ff44798f46c25f63dbc7310ce5a9da25cc9606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-18 11:18:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"eb8447a80ca95b71dc552144925b439ac15d0126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 12:28:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99817e079a4c45c78bf1734d1a1a07c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06bdc0f42a1048e89db301833aab2bdf : SUCCESS in 11m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35134b914c6d45b98f774a0e98b97bbe : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90a2385d5ccf4f5bab431c08e1427df5 : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d73fbf5374414b1f840b43619a382c1b : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f0b18f814b642e5940f137ad10380ac : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50d3caaecdaf4f759c9248b03a14f430 : SUCCESS in 10m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd35ad05c564a2e8fc019cae52fa15e : SUCCESS in 8m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/80eba320dcbb4ef09b06501bc0bce92b : SUCCESS in 22m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2469036d67cf46c8bbb315e79be37d05 : SUCCESS in 11m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/037efade00b14426977def97e9a2aafa : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1b7fd33e519d48fbb3e0d26d1192d975 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/531e941fedc34773866afb3c97eb11ba : SUCCESS in 17m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc7ef9e5d0164ee599af7b47638e682d : SUCCESS in 25m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f14190d30c64eebacc00787c8b48c73 : SUCCESS in 48m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e926f23c73e54772969d49e8962022b4 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db066bdb5fb6440fa2a0f70e82de442e : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e1c320e67204831839d94f492729bf2 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":4},{"id":"1894cb9bf40edb1e34438d508d9db3f4185a2fd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:21:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81bb752f7a6c49d0b348ec17a363a8ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2dadf312d8414bdcbfa7da4f66e59615 : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":4},{"id":"85415dadc3bbde5f136174d66d4e1f65896aa62f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-20 10:46:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"946e5671c3b364995c039602b4ef1ddf27c18215","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:31:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf519e8b53042f1ab3e84d5c721c0ad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b87c44dde4247f7be78f15e2dab9928 : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":5},{"id":"4560b7fa78575fe645edce9e4bf881de0a2f2050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:42:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf569a09af3f4817adea7f982f4cfe51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2029eee0d4649f8b0dadd8703ab0d4e : SUCCESS in 10m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fc8f76e7ceb4bce8df56015a22d871a : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f711f90ff6654be7bcbdaafc8ee973b0 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a6ce44b02f74b6fb7e49efcff66eb97 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca156006e18c464c9c4501fcf8efcb20 : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f6a5b0d9d72d463daa88cf40eb2e9666 : SUCCESS in 20m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d030a0d3471242f8ae1e0eda346888f8 : SUCCESS in 14m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/53c2f82f65d64f70bd6bb60b78fa9b49 : SUCCESS in 28m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cd28aa7121484530b6f5ec4ff215d012 : SUCCESS in 11m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5e81b26441f242f19e29bf9b31681965 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44605806e9ec4c78a47bc4668f26f76b : SUCCESS in 9m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8b7b16db8b154d3abfa2240c96a70cc5 : SUCCESS in 8m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad4adc3d6986414ea7eab561ca2fd9ac : SUCCESS in 54m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb7cddcf15a04cee8aaccfdd86e0ec72 : SUCCESS in 22m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f8cc6c60f84646984ed5e1eda8c75c : SUCCESS in 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f5ca68b53c441986f890fbd3f57844 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b49640cfbd84d66b8e216b586680683 : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":5},{"id":"51b26f71496211a25ecc3005568260cb49b0d34f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-20 12:53:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d1d1e160b077539f0837a48236487b39fc7f90e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 12:53:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"949452569c197a91e94c9c9935881a875be5960e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:54:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d942288e298646518a64cfcddccc082d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3423c41c2544d20b0719e0c8ce9b7ac : SUCCESS in 4m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/df0ea6c09ec74081a7296d9566895d15 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c59da568da084fd5b098ec555b845930 : SUCCESS in 11m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57df870423c046a19b2ec4b748215cc1 : SUCCESS in 20m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ad85e64f00a14e78b675447c820c4ae2 : SUCCESS in 12m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/be09c6c13aa0422d8d6863e64016b08e : SUCCESS in 12m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d72165c588c42729ba3649401437190 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/68ce583f0d6b4091bdac3c47c954021e : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ba0b642c0eb468a9ebf4c0ec5ff6356 : SUCCESS in 13m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4e80d9163d464e54847ff5379f570765 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/639725e2ee2e49f08efe5dcc8a76b694 : SUCCESS in 5m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0ab2dbd14ae4fcaaabae0e1c3f00640 : SUCCESS in 32m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b30b5efde696442a82a3449c1ca31c52 : SUCCESS in 56m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b328bfc57a24c66ae8280bae2db3ef2 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6b5597d9b045898aa5433b88065abc : SUCCESS in 2m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55131b4795ef4c138d797b3350e9d474 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":5},{"id":"ee508f47ee4080195cb7229d75c330bd446e054d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:54:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9c8ae6a66244855583e8838bea6dc56c36c39337","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:55:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/014885fa9a8246518cb21870c7337d53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9150ba307f564d64880aacb474bd825c : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8981fc6c513c485b9900ce97da9134e1 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f15ccf81b4c4c03db3f65b2cc2174367ee6eac5a","revisions":{"be179c55c1360958950e76aa74588cdc7c1e1fa5":{"kind":"REWORK","_number":1,"created":"2026-02-16 04:16:08.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/90/976890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/976890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/976890/1"}}},"commit":{"parents":[{"commit":"2fe96d6988f58f737565db44fe1ce6679a902567","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fe96d6988f58f737565db44fe1ce6679a902567"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-16 03:44:19.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-16 04:15:52.000000000","tz":660},"subject":"tests: move test_db over to BaseUnitTestCase","message":"tests: move test_db over to BaseUnitTestCase\n\ntest_db was using its own self.ts which was a timestamp_iter whereas\nother parts of swift unittests inherit their testcase from BaseUnitTestCase\nwhich has `self.ts_iter` as the timestamp_iter and self.ts as a helper\nmethod which calls `nest(self.ts_iter)`.\n\nMeaning this code although looking similar is actually behaves quite\ndifferently. So this patch changes test_db to use the BaseUnitTestCase\neverywhere it uses `self.ts` to keep it consistent with other\ntests in swift.\n\nChange-Id: I4db72dc72a274838cbc4cd54e8ce17770eddfd0a\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be179c55c1360958950e76aa74588cdc7c1e1fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be179c55c1360958950e76aa74588cdc7c1e1fa5"}]},"branch":"refs/heads/master"},"fdb004b4eba86d91ac4a6b5166e4f687914997d2":{"kind":"REWORK","_number":2,"created":"2026-02-16 17:42:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/976890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/976890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/976890/2"}}},"commit":{"parents":[{"commit":"3dabe051b7f9b694a2d589eea9b7a2bc340e0dd8","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dabe051b7f9b694a2d589eea9b7a2bc340e0dd8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-16 03:44:19.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-16 17:42:41.000000000","tz":0},"subject":"tests: move test_db over to BaseUnitTestCase","message":"tests: move test_db over to BaseUnitTestCase\n\ntest_db was using its own self.ts which was a timestamp_iter whereas\nother parts of swift unittests inherit their testcase from BaseUnitTestCase\nwhich has `self.ts_iter` as the timestamp_iter and self.ts as a helper\nmethod which calls `nest(self.ts_iter)`.\n\nMeaning this code although looking similar is actually behaves quite\ndifferently. So this patch changes test_db to use the BaseUnitTestCase\neverywhere it uses `self.ts` to keep it consistent with other\ntests in swift.\n\nChange-Id: I4db72dc72a274838cbc4cd54e8ce17770eddfd0a\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdb004b4eba86d91ac4a6b5166e4f687914997d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdb004b4eba86d91ac4a6b5166e4f687914997d2"}]},"branch":"refs/heads/master"},"e543f21daedfd1bc87fa6fc6ecc3abaa99bc82eb":{"kind":"REWORK","_number":3,"created":"2026-02-17 19:12:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/976890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/976890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/976890/3"}}},"commit":{"parents":[{"commit":"361dd1f5f2a205b848974d71ba5b204ea77c28c4","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/361dd1f5f2a205b848974d71ba5b204ea77c28c4"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-16 03:44:19.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-17 19:10:49.000000000","tz":0},"subject":"tests: move test_db over to BaseUnitTestCase","message":"tests: move test_db over to BaseUnitTestCase\n\ntest_db was using its own self.ts which was a timestamp_iter whereas\nother parts of swift unittests inherit their testcase from BaseUnitTestCase\nwhich has `self.ts_iter` as the timestamp_iter and self.ts as a helper\nmethod which calls `next(self.ts_iter)`.\n\nMeaning this code although looking similar actually behaves quite\ndifferently. So this patch changes test_db to use the BaseUnitTestCase\neverywhere it uses `self.ts` to keep it consistent with other\ntests in swift.\n\nChange-Id: I4db72dc72a274838cbc4cd54e8ce17770eddfd0a\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e543f21daedfd1bc87fa6fc6ecc3abaa99bc82eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e543f21daedfd1bc87fa6fc6ecc3abaa99bc82eb"}]},"branch":"refs/heads/master"},"56f13d0511ad5bc9598a96cc3868635c10cb6d0e":{"kind":"REWORK","_number":4,"created":"2026-02-18 11:18:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/976890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/976890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/976890/4"}}},"commit":{"parents":[{"commit":"810478eed645cc896511fd20245a9fe43bc0a276","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810478eed645cc896511fd20245a9fe43bc0a276"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-16 03:44:19.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-18 11:05:58.000000000","tz":0},"subject":"tests: move test_db over to BaseUnitTestCase","message":"tests: move test_db over to BaseUnitTestCase\n\ntest_db was using its own self.ts which was a timestamp_iter whereas\nother parts of swift unittests inherit their testcase from BaseUnitTestCase\nwhich has `self.ts_iter` as the timestamp_iter and self.ts as a helper\nmethod which calls `next(self.ts_iter)`.\n\nMeaning this code although looking similar actually behaves quite\ndifferently. So this patch changes test_db to use the BaseUnitTestCase\neverywhere it uses `self.ts` to keep it consistent with other\ntests in swift.\n\nChange-Id: I4db72dc72a274838cbc4cd54e8ce17770eddfd0a\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56f13d0511ad5bc9598a96cc3868635c10cb6d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56f13d0511ad5bc9598a96cc3868635c10cb6d0e"}]},"branch":"refs/heads/master"},"f15ccf81b4c4c03db3f65b2cc2174367ee6eac5a":{"kind":"REWORK","_number":5,"created":"2026-02-20 10:46:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/976890/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/976890/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/976890/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/976890/5"}}},"commit":{"parents":[{"commit":"810478eed645cc896511fd20245a9fe43bc0a276","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810478eed645cc896511fd20245a9fe43bc0a276"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-16 03:44:19.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-20 10:46:28.000000000","tz":0},"subject":"tests: move test_db over to BaseUnitTestCase","message":"tests: move test_db over to BaseUnitTestCase\n\ntest_db was using its own self.ts which was a timestamp_iter whereas\nother parts of swift unittests inherit their testcase from BaseUnitTestCase\nwhich has `self.ts_iter` as the timestamp_iter and self.ts as a helper\nmethod which calls `next(self.ts_iter)`.\n\nMeaning this code although looking similar actually behaves quite\ndifferently. So this patch changes test_db to use the BaseUnitTestCase\neverywhere it uses `self.ts` to keep it consistent with other\ntests in swift.\n\nIn summary, TestDbBase extends BaseUnitTestCase and next(self.ts)\nbecomes self.ts().\n\nChange-Id: I4db72dc72a274838cbc4cd54e8ce17770eddfd0a\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f15ccf81b4c4c03db3f65b2cc2174367ee6eac5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f15ccf81b4c4c03db3f65b2cc2174367ee6eac5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
