)]}'
{"id":"openstack%2Fswift~977171","triplet_id":"openstack%2Fswift~master~I7411f219890b6571fd8c71ed04bb9ee9ab031d79","project":"openstack/swift","branch":"master","topic":"feature/s3-cors","attention_set":{"38841":{"account":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"last_update":"2026-08-18 00:29:02.000000000","reason":"A robot voted negatively on a label"},"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-08-18 00:29:02.000000000","reason":"A robot voted negatively on a label"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-30 18:42:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7411f219890b6571fd8c71ed04bb9ee9ab031d79","subject":"s3api: add basic bucket CORS support","status":"NEW","created":"2026-02-17 23:10:40.000000000","updated":"2026-09-30 18:42:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":40,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"44929aa297ddbc6580fe12a49d57fff18a1e19b0","_number":977171,"virtual_id_number":977171,"owner":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-30 18:42:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 00:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-18 01:13:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-02-23 16:52:00.000000000","updated_by":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-07-15 23:53:20.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"}],"messages":[{"id":"ba1bba04d4787bb30ebe5c195fad9891974bf415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-17 23:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e77e5d2285c35ae3f522d9c929af0debc9bf5ef1","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-17 23:16:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfb86d9630b3c14f27d410da49c131ae85ea3b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-17 23:22:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7732eed802423d0643f3542760412eba73f59fb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-17 23:27:58.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dd39ba0e3312807acb546641fd3ef22a5457d736","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 00:00:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e734de59cd42c4b4c35dacc3c05587\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7786d6a0fe31456f8e91e7fb4fc3b1a9 : FAILURE in 31m 15s","accounts_in_message":[],"_revision_number":3},{"id":"1753fbd51f5caf86a4ec84109e46101a742fa0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-18 00:06:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c5e8751f1d7f0b039af61d283e114212a1534387","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 01:00:56.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8afa828875514dc891945a25cf4b12d7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eeb59ab3319d447a955cffdd6688d6a2 : FAILURE in 33m 33s","accounts_in_message":[],"_revision_number":4},{"id":"9d3ea21d287d8627d6555b79c6b5db7939b45e77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 01:13:48.000000000","message":"Patch Set 4: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b10a0616eb64cc29df0631aa034759f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e549f8b4944402b28b472970a26003 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/7930b3f23f324e67b2808d07ab384bc0 : SUCCESS in 43m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62e7b24f3fc34f87b994c3244a1129c7 : SUCCESS in 34m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d426a6e6de94c5a8e7cdf8f10dfff57 : SUCCESS in 54m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/498f9308ff00419d8e99d0cc63192bd6 : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07f45726a7fe4acb9ce165fb40f3148f : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/052de6b967e34e2c937dc9daaa861cae : SUCCESS in 10m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5317c616180649dea5d5c24ed10107cb : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a901de0bf2f4fec86ea2fe331118a26 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2d58924760af40bb871d9034d2101051 : FAILURE in 24m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2cab99ca0f4d4b26909b7f9aed2217f1 : FAILURE in 18m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17a4c6e8e6974e0598dac106c0d37b9d : FAILURE in 23m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7ccbf22a51284760b8b1a8665069d603 : FAILURE in 29m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8bdfe171aee44023b2ce072857c107ff : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e53f33ce0c024bfab484f4ddbc150060 : SUCCESS in 13m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ba3d173000c843d29c80c90d5c9650f7 : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/226d72ebf31840a1a8e024b96a6434ca : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b48c265dd2664e08b3851aa36e20f664 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8ecae79536546dab98241c124e3191a : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/027166d2621b4ca69e98eff78ee7b81b : SUCCESS in 34m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/169ca008b8fa437b90c0fc296917734e : SUCCESS in 21m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ce1d5c1759e4dbe9a773d0f5a330b8a : SUCCESS in 21m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d507bef7d8a4098a09f9f78bf01cc3e : FAILURE in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc1e354366e4b688eb55464b038e79e : FAILURE in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/812e1ab2d784455ba7a6217ff65b15c2 : SUCCESS in 17m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b54212ce4d648b784ae9bd2141fbc57 : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":4},{"id":"19617e1556fa197bf02fc8d822f27fbd7546d4fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-23 10:12:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ab7966864250e4b48a0be786be68d0eb7727d10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-23 10:25:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b0d3b1e5ca32ca6de4c933e029a97f2a344b95d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-23 10:28:06.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"cd0c606cbc93ff652ec527f6eafa255ed1b03fa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:03:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1ae5b7675e46c4bfa9d01b8ded1dee\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5bff8099e294c398c9c6f642f8be640 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":7},{"id":"d4f515c820f8a5e10609583f724d4b84bc5a27ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:39:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fbe73e71bf84439a8aa4b94df358392\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1590d31fb7d40ef94c66114b368d69d : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/81f0fc0e6b5b4d30814a20b66bd678a4 : SUCCESS in 47m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b103126b0da4c2b968c8cc6a4f4dc5f : SUCCESS in 1h 07m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f0aac356eff4214858aa4ac3f055d3c : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df299f6472eb48a0b6856ff7b9655e72 : SUCCESS in 1h 06m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/545004ac31d04df28cf52856ef4fb1b3 : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2749c36f8f8944a98bf8b05459bf692c : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62d2142ffe2144318304a1f82a239c90 : SUCCESS in 18m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/973fb17f1eca48edb9eb98820d772ffb : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ebe904def4ff4a88968ff78f48f02c9f : SUCCESS in 20m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eefd352a62814470a68c24cb4f40066d : SUCCESS in 17m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/36a5bd3dc52c47f1b6a28b790d4e965f : SUCCESS in 20m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5f945c8133b04a3b88b326fbee4c690e : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57fd349f39c247a2a5fee243c638c8a3 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/921438e09a3a4671bf6312101afe1947 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aaf0e26908454ecbbc2db20fe95224eb : SUCCESS in 9m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62177717d8cd4532bd64df192e74898d : SUCCESS in 5m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e997bafa6524224bc541321ef6e61e4 : SUCCESS in 7m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9b2eb6f73244179b5162528b8558991 : SUCCESS in 12m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/071bb6cf7e9a4ef4bbf5a7491e63ec6b : SUCCESS in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26ff46dae5c44c92864243b6c273e80b : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49c36eb4ba204bac91abfe5f06ead728 : SUCCESS in 51m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc4a405d2a144e899d921ad3fb06cc9b : SUCCESS in 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97f45daebf54cc3b290ee6c3e602c2d : SUCCESS in 3m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c164d99f35a4ea896fea7a9646c15c3 : SUCCESS in 8m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/921894c68c064bf99151b84669f6d6f4 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":7},{"id":"752818f0ea1d8c817cb2653432a9166fa599cb68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-23 16:45:17.000000000","message":"Uploaded patch set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c3ee280970bdb1c9bc9742c0a1193b4bfd5b3334","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-23 16:52:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"887caa8b3dd086cb85ebf6b46f906c725422ed66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 17:18:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d6fa348fa1242e7a0c4c35fc4c5bd9e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7430e3ab400d49109333aeeef4eeb87b : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":8},{"id":"8043d311d5853510c7a9ce6e58bc163bf679ff05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 17:44:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a174baaace2c482a915ee788ed86cb14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cd205b5d5745f999ecc54f2f5fb3e3 : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/bce1cc5e1c8e48b6a7a7e586a93ce079 : SUCCESS in 43m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74858333242740e0af99ef3e9cff01f7 : SUCCESS in 29m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/310c99b1c57d4138bb93a776715a9c8f : SUCCESS in 53m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2cd1eac67dc4c47a668b7e2edaee90c : SUCCESS in 48m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2664b537f9da4a7ba9661766f6b7e48e : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a2aee787ea64ade80fa6304b9cae0a7 : SUCCESS in 12m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eed951be8f024da39bec845254f413b1 : SUCCESS in 16m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25a33fe6a47948bcac93f3ef228fa8cd : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b133882cce8446c8c7b46ee4f4c9118 : SUCCESS in 11m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b074932ed41146f2b0c6cb9c3e2c66d5 : SUCCESS in 19m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cc46f70570f342108feb93236873b375 : SUCCESS in 23m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/915aa68655ed4ec8acc827dc2f598ec8 : SUCCESS in 24m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28babbd018e44a09b72cafff14c99dd9 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/292d278a79f94938a57a6da0c1ab3ebe : SUCCESS in 12m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/de3b45c141bf48688ad2cbe94162b0d9 : SUCCESS in 9m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9909406546614d3794293e924cd1f1f2 : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9dae91977f5d426986f6e44633049771 : SUCCESS in 9m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2a7a7e7840746789b039a2400fab257 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/307b66f6935a48199fa7c380a05daecc : SUCCESS in 44m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/418ca54fd7164a808466e49cfc0de209 : SUCCESS in 24m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5a0a5bb22414be385e8109965132164 : SUCCESS in 23m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc514c6ae3964aeab05bf76a77f908e2 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1137b87f2f874bd5958ac178827364fa : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d956cdb11a974a0b945ed4863d2bd6da : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/377c6ec01c6e4028873edf4c26167794 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":8},{"id":"e97a5df92be307740855940f3a99ef869e316fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-02-26 11:57:49.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"04707808f21b1b5ad9329eec514a8b9b788c0d0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 12:29:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a23adbe497ff4b0380e8b129fe7b7b8c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2496d175114947e3bdf6e4371a4e5a2d : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":9},{"id":"97fb6eac7387b143a4b9d89263730e12ddd90fce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 13:14:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9617904ef4014778ad4249e5774e004e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bae55cf113e486db5263f06416f53fb : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/76a02380afe0438ebe5ff94aa232982e : SUCCESS in 1h 05m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/72649d9d1afd4ce6b6ffcbea80234af5 : SUCCESS in 49m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fbe3dfcef01246548030ec8a5d358908 : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb719a9c4b58482f885dcd647bbcd177 : SUCCESS in 33m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0289b94006ba4be6b9e8c46b5af47c6f : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cffccc4d3c647b2b000d6df55249d62 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46f64b80ac3a40848759b1bfb57d7aeb : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c49e766356f4b65bd8eb38157b739f2 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/21d4e49b5dd34c4492e5456675bba2bc : SUCCESS in 21m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c02fcf6330b142b18003c38adbcca28f : SUCCESS in 21m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f5639ebf425e4c4b9473ee8635cd4dd7 : SUCCESS in 23m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3fda9c48be904b33bd63202fbcb741f0 : SUCCESS in 29m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9d1f815bf71445f8b7f7b3da739fb4ff : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/74408967168949c78222542b4bf65a1c : SUCCESS in 16m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/60c9233938c84087982af6ba9ff09759 : SUCCESS in 8m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d47a6abe7e344449086dba94fd57771 : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a09dd828ed44d50b6490b05e18871cf : SUCCESS in 8m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5660a0ab081e442892ec4fd8f73b18b3 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f22050275b0241ec81b0756bf78c6d43 : SUCCESS in 50m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2ca59bcef20410c858164e3bc69ddb5 : SUCCESS in 50m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e238b35b07e403f8c1478b7ebb7095c : SUCCESS in 25m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4595bcb1bce4274bafc2ae1e145250f : SUCCESS in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52a83175e7b4d0a87729ec3e4256678 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea683f931f4d4c0393700c6975f6d6b8 : SUCCESS in 9m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a69235017ed941d19256a27496cd4efc : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":9},{"id":"14d8a5ba7f596141808aa679a44da9e7393c5d9a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-10 18:21:55.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bc4d04f58b0187f327faaa4ba250ca9e08daf9fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-11 21:15:09.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ca46f54c4245c191a30c58f8f3b8c0799ffc1fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-12 10:16:32.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"8637fed145bde0ab1e46eb5741db1d85c6475ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 11:08:15.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db05a3692c624808b4f1bb7f6277c3fa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23353d1c1dbf45c69e72750b15fdd4a8 : FAILURE in 31m 47s","accounts_in_message":[],"_revision_number":10},{"id":"6e7a295c6840bcb7f01db709b59d00fe4f7e370e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 11:18:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3fc310ec13941e18b60b5f837b30288\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275d07d2a04143d68693b09e2cba86a4 : SUCCESS in 6m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/80a0320539a84d11bd205ba3ccdd0b6a : SUCCESS in 39m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d450cbee55444ee190f34ebf5ed3cf70 : SUCCESS in 38m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d84788921c694e4d97f157d7f4a40099 : SUCCESS in 35m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6bb2f43635047f79708a36faac4aad1 : SUCCESS in 33m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b250b1823e3b4100a8900bfbdd9a427c : SUCCESS in 10m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fc80342643a4c4f9e93a1d881489cc3 : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19f277dcb21e432fb6b9e080ac377e1d : SUCCESS in 10m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d75d57e9d1f942aca19e901336c28a04 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1084f3cd3ed4418a1b10809228525c5 : FAILURE in 14m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/57010de0a62d4f17b20506c095a7ff56 : FAILURE in 9m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/41fd8cffa8f74d1a99b24200d7833f51 : FAILURE in 12m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fa1d34c7ac474947a36cc5f25f125037 : FAILURE in 9m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a036d61d99394a3b9fb24784fc7cf2f7 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ebe9cb3e452542f59be8c28c9b4d955e : SUCCESS in 13m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f287c18bcb0c469e84568976e5d6f49c : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35fcbbedc05f4a5f824e90f7ef1efd88 : SUCCESS in 8m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4bb897fd217d41d987a012208bf6586b : SUCCESS in 7m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9d3d3b7e4ba4cd795596e4ae33f7996 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/62858e0f9f5f4cf79ce21ea552cacd8c : SUCCESS in 34m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/827f84570d3b4909b9195486c315328c : SUCCESS in 48m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/951fc74418f64514abf60949019d0376 : SUCCESS in 35m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3329846710248f081bbd3f784fff15b : FAILURE in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bc0a1c1a7641e8b77042b2f4bb1fb2 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/517e33615f304948b24a3781abd65416 : SUCCESS in 19m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/233a63d3010047f193d70422aac556e7 : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":10},{"id":"e6699e7169e97d54f1018bd18eb8633b370fc41d","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-12 12:51:29.000000000","message":"Patch Set 10:\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"823d1b823276ac78c44d5e39006c3775663a274d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-12 17:36:06.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"be39c58244f60d4cfcc3e329886ee4e73493c0b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-12 18:05:56.000000000","message":"Patch Set 11:\n\n(9 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8e39eb2a81458a149d57b414e292119293fd571f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:50:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb3ef44e6ae44e1b046dc65d4d3361b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db9f95c6f2742cc99ea8eafb4e9f49b : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/410566e2a6bd472784f90a86519124fc : SUCCESS in 57m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d9d772e5a4c403295529056b0c156d8 : SUCCESS in 1h 01m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c170468306544e8b697c4299bb06836 : SUCCESS in 57m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fe1e7f4f76740f2a05aecf434cb96f7 : SUCCESS in 33m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5b246c73a5245bcbb2b38228c3e4a4e : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbe64d58260d47a59138ad4e7fd7e0c0 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92b47c97f0d04f5eb758043c1cdf809e : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9695e2e874c458eaad33694d36e0408 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4c9c15318f84bb4ba9b3dd59e19e22d : SUCCESS in 10m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d1cdd01c45b94ccd925bef8f4a9e6629 : SUCCESS in 14m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6d5f17bcb20045bc94d0a0cdfc8785c6 : SUCCESS in 27m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a5d1565cc2484b85a0cc31860d4c0fa7 : SUCCESS in 24m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c1e0e5f435d745939eb1eea8fd237fbb : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8cd8336dd1ce44198786db98f8629575 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c3fa268b3f0149f7993066db8fea6a94 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/780288cef83b40859a8b75081c2c6704 : SUCCESS in 4m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67c348ca645349f4a5379a29179d1158 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ff250af68944455ac0fa6b6ddd7d0e1 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0987bb5b5ac14bf19b880ec1d05e49fb : SUCCESS in 56m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/288f9543c9e045b8a90c07dc4030f24a : SUCCESS in 48m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f1e788fba3c4981833c93f9cb33241d : SUCCESS in 57m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63f060ee00e34717b47a954ed6af32cb : SUCCESS in 25m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1932e0fd9748d99d96a2fc496d5f83 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/805a08507b2c45a09a4146d0ee9d5d57 : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab2ebd26afeb485aaa7f43d1f2ba9137 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":11},{"id":"88841c1e06ce8dbcb63408bb800b8df225642492","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 19:00:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a083bbe686545128653a134d7429b98\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0cf8780eb55470e9909f59ef3d6af0b : SUCCESS in 45m 06s","accounts_in_message":[],"_revision_number":11},{"id":"64b3e9d917525b3196663bc31524a2ebbac81b3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-13 11:33:45.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"8d1714e4ba2946ef90913626153b150b2f7cedab","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-13 11:34:27.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"3bcc9f528e66e3266dc7d51d57a43b4878f028a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-13 11:36:14.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"a5342c22040e87dbfd5054dc2addf1b0e8655621","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:20:41.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a147e5b0700455e9208acb0bdbab5e7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8ae10cde06a4485b2187691ecc0bf5e : FAILURE in 44m 08s","accounts_in_message":[],"_revision_number":13},{"id":"b2a33d2ee14dfde139fa53205e7c7e6ed179ca9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:37:51.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb3a468fad8f4ddfba7e226a71bfd14e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99287a458a524b0f8b21376ffb52dfb7 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/5ddb82ece8e2494397d5e5f42fd4c6e8 : SUCCESS in 56m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6bfa66f05acb4696b4cfd6993b4de556 : SUCCESS in 30m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2bdc0680a0d84804b0c1193fc8e6d88d : SUCCESS in 32m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c913b7b072d41c48857f561ba076da8 : SUCCESS in 40m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5862f9d8b2147d1a4300cd9868f8b40 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39f8426560cd425fa189c85daf8b8b25 : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d844cd467bc447f594199f4f93b1a961 : SUCCESS in 20m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19fbe11532b54f1494b8dc1ff027068d : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de877880b8ee427faca9af1c68202352 : FAILURE in 23m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6ee3cee1474f4cc5b4f6ca5944fee1ea : FAILURE in 18m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/29590a4d6e2448ee9205787227a91f35 : FAILURE in 11m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6d7363b7123146a8bb4d75fcc8ddbd44 : FAILURE in 20m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1463a87004a7446188db405bd4a7bd57 : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a8cc75c7e47463dbd91fd8a88753274 : SUCCESS in 8m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5fb2adbe9a714933bcb1c9283e7e65e3 : SUCCESS in 13m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d321597cbc134d17910b5ab32d8dee1d : SUCCESS in 6m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e8984c265ec4b2d8ef9bccc2b2423d6 : FAILURE in 22m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bbe3a627189942ffbc341b0c0405264e : SUCCESS in 7m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c5781ecf3f9422f97f09c01254d8597 : SUCCESS in 53m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fd305fbc0e14455b4a6d9b42184dabd : SUCCESS in 29m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33502337d4ea4b2da7286c721a1469c4 : SUCCESS in 22m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81737694de2543c994f58d1840313811 : FAILURE in 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead225339f0f4962b90ed3dc8c054bff : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca73ca0cf98448a6b8ad0e48c79abf2c : SUCCESS in 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b5a68eab8ec4683aa35ef2672e229be : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":13},{"id":"6488aee59b516a7005580aedd26853180952a0d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-14 10:43:01.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"298960f3926c117572e9e653d22fac79cdc9e0b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-14 10:44:07.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.","accounts_in_message":[],"_revision_number":15},{"id":"b8b31aab266117d446d7ce17ecb36519e52de499","author":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"date":"2026-03-14 10:49:58.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1c00f18a9466a6169071ce8b58f1e2aa58a3d6db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 11:11:31.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2243d687bd64f29ae39820ef65cd430\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac3ab6fbe3b94df78dcc7409aa1f3441 : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":15},{"id":"91acd17a148e57c4b6c6a65560bdeea1d29dde49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 11:51:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50eae414457849bc83ea5ae634ffa99f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1156b4ca4034f2a8ba85c318d2ca950 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/f683b3c97e0b49f29fac792e77dc2298 : SUCCESS in 1h 00m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25203a793ee142cd8a12af771e206715 : SUCCESS in 40m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/87481bb4d92342498cb4eb07c0cc2759 : SUCCESS in 38m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff27e331818b42caa1d3cac5b62ee67d : SUCCESS in 32m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79f8ed31cc06423192511c6914223e87 : SUCCESS in 24m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d27457cc63364cc0a5f8974dfb38919f : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad33c59e440a41498cd23419e6669c15 : SUCCESS in 10m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b0b7df80b4e4c56883c847506500806 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b7416726a694734add31674c9027b7e : SUCCESS in 24m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9bc18c637b90481182b173d4dc048063 : SUCCESS in 9m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1aff5ab05077437ba8752cb083769b1f : SUCCESS in 27m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b0ad486a12424ffc9c82d2d3d1972eef : SUCCESS in 20m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76a52339657c4a61b10d12b8e6719197 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/27571ee68a364793bace2d84e37930d6 : SUCCESS in 13m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/236ee65ebfe1491ab5f89694dae5c414 : SUCCESS in 12m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/df2655dc75d24cff9692e563bc4ddf08 : SUCCESS in 8m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a745e41691e4428a3fe40435831ab18 : SUCCESS in 11m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2c77f8b4f004ef299c0e4e37c580955 : SUCCESS in 11m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf5270430aa4483a8f1b5b42cca77347 : SUCCESS in 58m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82597dd7e28b4522aeac72a9be280bda : SUCCESS in 31m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b0a512068ae476b89d3e759b8206109 : SUCCESS in 21m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/831c583db494460a9ac6285ebf2b5639 : SUCCESS in 31m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6429fcaf62384b398a7cfc08a5d6d6e7 : SUCCESS in 3m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/872e11290b0d4044810d3720f1ece55f : SUCCESS in 17m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85c4598b173745a3a54fda146309f727 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":15},{"id":"2c1e0fdd6090e20b9589e2c9672117ce3b3782dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-17 19:32:25.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"b6d772d303378193abfcffd3e1cd25bbc9c247c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-17 23:46:56.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"8f1f38591e7a21cd3f14a639d7c15aafe7a27f41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 00:25:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1edcc45f8da7434eb91dfb2cc4161300\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/585c6a8b8c24496791771a510101e96d : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":16},{"id":"4f64ab23885de88e0872b20495eee020bb44a54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 00:29:02.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aba3d69af5a844d38ba2c568823bc935\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503df297bcc94c5198c894c441607b4c : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a70ae73c00a94b308ebdf01604619672 : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/bb3ca9ef79bb4c5a860d09d6a93cdc44 : SUCCESS in 33m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a244c3ca81b74bb19fda3443392ab232 : SUCCESS in 33m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/11143acc21c349beb52f5aa459953f93 : SUCCESS in 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5381fa0779f64781a58b169b38b301c7 : SUCCESS in 41m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2ff942e11284fe99595451292d62ff1 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9a33f04b64b40669d6c618fdbc6f297 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ee606ea76e04b209e5552f1cd3e36ee : SUCCESS in 12m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de7f4badbc0f408782109216dbf14b7d : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7149a01cffa64ff7b1a3ff4a33c58931 : SUCCESS in 13m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2134752ad09a47b4a6f78cbc7068635f : SUCCESS in 15m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2f652ccd3e2a4b96b371ccbc49f72ba4 : SUCCESS in 20m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/25d885370e1941b684f1d35093559188 : SUCCESS in 11m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e0e5ea1fae84ceea0e6228346e874ba : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0abbd49e21a8414ba705f638bb1d7642 : SUCCESS in 8m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ebc2041f5fb94c26a563c8ef1aff5904 : SUCCESS in 9m 45s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6616373380f74b76b18885e380247481 : SUCCESS in 9m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/584b50c47dbc4d949cc6c489aa46967f : SUCCESS in 3m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bba8d568184d4be99e66ee7f8ae00597 : FAILURE in 7m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f8e5a7caec8841e4a4e0b6487994fa7c : FAILURE in 5m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c197e3e7c18a467d91c5a210907905c6 : SUCCESS in 35m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e28d15d907ce4fcdb2e5424903ef33c7 : FAILURE in 13m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77b5a9d0983b4157bdd010a5e46db1c5 : FAILURE in 33m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23e39dead0e142d1bf6a2f52750ff768 : FAILURE in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae349aceab854f46a30f7a2fdb0a2b03 : SUCCESS in 2m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1281d7f9ed284443845253762c65fc20 : SUCCESS in 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/393d67e3acb542b7bb042d81d87596c4 : SUCCESS in 37m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"cedaeaa8e759a8da5ba253ef6170530effd39ac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-24 23:23:21.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"70164bb9376d1e22c55a380700139dac26221125","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 00:35:07.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7b6fe40503a4b7c8e5dd68a52ef94e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d57adc97f6584ca0893fdca21df108cc : NODE_FAILURE Node(set) request 3201dbd22a40446aa335ea1e73aa3920 failed in 0s","accounts_in_message":[],"_revision_number":17},{"id":"dff25a510d3f377aa459f44bb5d56dbb752eedcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 00:39:31.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a412c93ca6946249bd693c416e3a9c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02049282dc2e4f19bfa24bf83e0246f0 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7041b9057ebe4eb3870874aae1686736 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/9d6df4ea56324fe383e747af7c41529d : SUCCESS in 1h 12m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2e917df16894489b19ef05ec68cf5ce : SUCCESS in 1h 07m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01438af708e147dc805daf2f9960d821 : SUCCESS in 49m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c738b67638d44b8d95f5e9b929c9fbec : SUCCESS in 1h 03m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/979b5b4e611f405da97c4659a88e3d88 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac0eb163a0d24b49b98a4d0cf533c735 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fae967c283b4a85a6611f732c60dafc : SUCCESS in 12m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bb0c6fe9b04242db9a12976065d8541a : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7915162339124f3b8614377db1ea23be : SUCCESS in 14m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8e1dbe26d2bc4c7a97ce02f21f3f8862 : SUCCESS in 22m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4f9d2dc01016432e8cc043be14f2b3b5 : SUCCESS in 33m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/08c136176d5f4e5083f9f5b170252993 : SUCCESS in 21m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ed3f0bf32b74983b94cb247fde61567 : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/58b6350a37354a63b328a32210924239 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd70970f375b4300a9e675f0291c17e9 : SUCCESS in 10m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e58948acf204808a10aa0f8e5a2e700 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ca4a1743ee64904bf1b8f6096c2806a : SUCCESS in 3m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d0c742fb81f4f8dbae17f34ff13aad8 : FAILURE in 7m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/27046dc925eb4b8a9c2a35aea36757b3 : FAILURE in 9m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5d9e83e693346e0b1765117e34eb121 : SUCCESS in 1h 03m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e34be038fc024a8d995e0cfbb26a0b21 : FAILURE in 31m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/376e13c55d214eb3baa278415525a3ec : FAILURE in 32m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90e369c1dbba416ba33ef0c92bfad9ab : FAILURE in 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5240f2b32b534c5ca198c995fa46af81 : FAILURE in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a33e2cf3852942daac65a87bbcd964ed : SUCCESS in 18m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4411c5971f644ae8a6c79260ac0a181 : SUCCESS in 26m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"2a640b36222c023a3df7460d1be1dd440764966f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-30 17:33:33.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"711ae886004b8e8f666f6fac27b632b62508505e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 18:33:16.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61338313058d417592307dd512b4b252\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dac0c51c21d6401bb0ea738852c7332c : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":18},{"id":"44929aa297ddbc6580fe12a49d57fff18a1e19b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 18:42:12.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdc797b881254d8984203c7d0a75d860\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13955c0681404474be55d83d366f2f38 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c012cc97da0a47c9b5a5c3bf3ba3e515 : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/97da4e1f8fc643968be9bd69b8e15d6d : SUCCESS in 1h 02m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abc7f11a45aa4ddf8ac16d5751399758 : SUCCESS in 57m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52bf0f61cecf47f6a99b4d75770aa087 : SUCCESS in 51m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50c628414b1c4a16a6617d8351ffee60 : SUCCESS in 49m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33694d52d22743c4951c2fb307249bcf : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc4348da1b8a4653969769aef0b5b82e : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c13a462f45144829db95f204fd7e054 : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bb9e32f4de564665bd1eee0f89bfdcd7 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5bd1f249c90442c1802983b34babae42 : SUCCESS in 21m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7820350b99ec4cfeacf2ef428352d180 : SUCCESS in 20m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/439d535edded458d82816211a5f3595f : SUCCESS in 13m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1ec9e96a18b04a8c81f9c9767ec7b180 : SUCCESS in 16m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94bce16da83140e58c9fd020011438b2 : SUCCESS in 8m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e796a864a0154ce4a20dede4780a2a87 : SUCCESS in 14m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/53db068013f140a69b485d1ced9ee77b : SUCCESS in 13m 32s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8880b9846c7944ccb5290abbcc68712a : SUCCESS in 18m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b1b1333443046c8806db5d8d061a9ed : SUCCESS in 5m 24s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/812ed549cd39458f9e790abe204a6172 : FAILURE in 9m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f96bbce1aab840b8bb5c5cdb7068098c : SUCCESS in 1h 03m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0129d8207fe4416a286945dbf7d2e32 : FAILURE in 23m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a87ca9aad4d548aa8945d9f035830136 : FAILURE in 33m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c529b3b64716440086d45bd850d83362 : FAILURE in 30m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9842aa48fc1f4bd0b2337cce18306d93 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77a4854a14514c7b9343999c1a23ebb9 : SUCCESS in 18m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac87d3b296e848b4940d5d465063561b : SUCCESS in 55m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"bde63f9d867f4ee48995df4e4da266bc76e775db","revisions":{"a77d97aa5dca9c8b1c3b5dc29d9fe6ac78185989":{"kind":"REWORK","_number":1,"created":"2026-02-17 23:10:40.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/1"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:10:29.000000000","tz":60},"subject":"Add s3 cors api","message":"Add s3 cors api\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a77d97aa5dca9c8b1c3b5dc29d9fe6ac78185989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a77d97aa5dca9c8b1c3b5dc29d9fe6ac78185989"}]},"branch":"refs/heads/master"},"1902d61a2aaa9f564a93864d8d3edea39a9596cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-17 23:22:24.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/2"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:21:38.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1902d61a2aaa9f564a93864d8d3edea39a9596cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1902d61a2aaa9f564a93864d8d3edea39a9596cf"}]},"branch":"refs/heads/master"},"c79ed754e3a1b0cf92450484dd8e94c9fe84d78c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-17 23:27:58.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/3"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:26:32.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports up to 100)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79ed754e3a1b0cf92450484dd8e94c9fe84d78c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79ed754e3a1b0cf92450484dd8e94c9fe84d78c"}]},"branch":"refs/heads/master"},"392c97a06010c2bd682865d089992bdfa48cf37f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-18 00:06:09.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/4"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien","email":"bastien.wermeille@proton.me","date":"2026-02-18 00:06:09.000000000","tz":0},"subject":"WIP: Feature: add s3 basic cors api support","message":"WIP: Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports up to 100)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/392c97a06010c2bd682865d089992bdfa48cf37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/392c97a06010c2bd682865d089992bdfa48cf37f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c64484e43c3bb24a5dd6fd1486d764b13e9d859b":{"kind":"REWORK","_number":5,"created":"2026-02-23 10:12:54.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/5"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-23 10:12:26.000000000","tz":60},"subject":"WIP Feature: add s3 basic cors api support","message":"WIP Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports up to 100)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c64484e43c3bb24a5dd6fd1486d764b13e9d859b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c64484e43c3bb24a5dd6fd1486d764b13e9d859b"}]},"branch":"refs/heads/master"},"c4f22f325348f46da27137cbb4641867f71df82e":{"kind":"REWORK","_number":6,"created":"2026-02-23 10:25:17.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/6"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-23 10:25:08.000000000","tz":60},"subject":"WIP Feature: add s3 basic cors api support","message":"WIP Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports up to 100)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4f22f325348f46da27137cbb4641867f71df82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4f22f325348f46da27137cbb4641867f71df82e"}]},"branch":"refs/heads/master"},"44e3da06eae246d5dfd8eda225d63930b9ac1973":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-02-23 10:28:06.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/7"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-23 10:27:30.000000000","tz":60},"subject":"WIP Feature: add s3 basic cors api support","message":"WIP Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44e3da06eae246d5dfd8eda225d63930b9ac1973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44e3da06eae246d5dfd8eda225d63930b9ac1973"}]},"branch":"refs/heads/master"},"14a6a26a7dea77559f4ee16379c70d07d0c35dd8":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-02-23 16:45:17.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/8"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-23 16:45:02.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a6a26a7dea77559f4ee16379c70d07d0c35dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a6a26a7dea77559f4ee16379c70d07d0c35dd8"}]},"branch":"refs/heads/master"},"5b7fbdffd946f88b299478004b280972cf2be219":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-02-26 11:57:49.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/9"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien","email":"bastien.wermeille@proton.me","date":"2026-02-26 11:57:49.000000000","tz":0},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b7fbdffd946f88b299478004b280972cf2be219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b7fbdffd946f88b299478004b280972cf2be219"}]},"branch":"refs/heads/master","description":"Rebase"},"a75bec11250e6c90b84ea870267c9eecf055d3a9":{"kind":"REWORK","_number":10,"created":"2026-03-12 10:16:32.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/10"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-03-12 10:16:13.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a75bec11250e6c90b84ea870267c9eecf055d3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a75bec11250e6c90b84ea870267c9eecf055d3a9"}]},"branch":"refs/heads/master"},"807d6a4d6f2ce358c94f4bf65cfb0d6dc1dd92ec":{"kind":"REWORK","_number":11,"created":"2026-03-12 17:36:06.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/11"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-03-12 17:35:55.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"*\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/807d6a4d6f2ce358c94f4bf65cfb0d6dc1dd92ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/807d6a4d6f2ce358c94f4bf65cfb0d6dc1dd92ec"}]},"branch":"refs/heads/master"},"d3754165421e83581c82d01868b76df53a7edae2":{"kind":"REWORK","_number":12,"created":"2026-03-13 11:33:45.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/12"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-03-13 11:33:33.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"GET, HEAD, PUT, POST, DELETE\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nCloses-Bug: #2077629, #1905104\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3754165421e83581c82d01868b76df53a7edae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3754165421e83581c82d01868b76df53a7edae2"}]},"branch":"refs/heads/master"},"35f9ab0216607dea840832746a2a1e7b411154c8":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2026-03-13 11:36:14.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/13"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien","email":"bastien.wermeille@proton.me","date":"2026-03-13 11:36:14.000000000","tz":0},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"GET, HEAD, PUT, POST, DELETE\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nCloses-Bug: #2077629\nCloses-Bug: #1905104\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35f9ab0216607dea840832746a2a1e7b411154c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35f9ab0216607dea840832746a2a1e7b411154c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"df950f67b622d92125a5e4bd558034e0fb6e8cb0":{"kind":"REWORK","_number":14,"created":"2026-03-14 10:43:01.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/14"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-03-14 10:42:51.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"GET, HEAD, PUT, POST, DELETE\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nCloses-Bug: #2077629\nCloses-Bug: #1905104\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df950f67b622d92125a5e4bd558034e0fb6e8cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df950f67b622d92125a5e4bd558034e0fb6e8cb0"}]},"branch":"refs/heads/master"},"f1544e87c894c608a06f737b195c6348a79b91a1":{"kind":"NO_CHANGE","_number":15,"created":"2026-03-14 10:44:07.000000000","uploader":{"_account_id":38841,"name":"Bastien","display_name":"Ph0tonic","email":"bastien.wermeille@proton.me","username":"Ph0tonic"},"ref":"refs/changes/71/977171/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/15"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-03-14 10:43:52.000000000","tz":60},"subject":"Feature: add s3 basic cors api support","message":"Feature: add s3 basic cors api support\n\nAdd basic cors configuration support for S3-compatible buckets.\n\nImplements three S3 CORS operations:\n- GET Bucket cors: Retrieves CORS configuration\n- PUT Bucket cors: Sets CORS configuration\n- DELETE Bucket cors: Removes CORS configuration\n\nCORS configuration is stored in Swift container metadata headers\nproviding automatic compatibility with both S3 and Swift APIs.\n\nLimitations:\n- Only one CORS rule per bucket (AWS supports multiple ones)\n- AllowedMethod must be \"GET, HEAD, PUT, POST, DELETE\"\n- AllowedHeader must be \"*\" or omitted\n- Complex wildcard patterns in AllowedOrigin not supported\n\nThese limitations align with Swift\u0027s container-level CORS\nimplementation.\n\nCloses-Bug: #2077629\nCloses-Bug: #1905104\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1544e87c894c608a06f737b195c6348a79b91a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1544e87c894c608a06f737b195c6348a79b91a1"}]},"branch":"refs/heads/master"},"9f526a41f830a13a105ad5e7dcff1636af6ed6da":{"kind":"REWORK","_number":16,"created":"2026-08-17 23:46:56.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/71/977171/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/16"}}},"commit":{"parents":[{"commit":"3064444bf433dfb7ade3d91130b58f7bceae9add","subject":"CHANGELOG for 2.38.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3064444bf433dfb7ade3d91130b58f7bceae9add"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-17 23:46:00.000000000","tz":-420},"subject":"s3api: add basic bucket CORS support","message":"s3api: add basic bucket CORS support\n\ns3api currently treats the CORS bucket subresource as unsupported, so S3\nclients cannot configure per-bucket CORS through the S3 API. They must\nuse the native Swift API to manage the container metadata instead.\n\nAdd GET, PUT, and DELETE Bucket CORS operations. Store the representable\nrule fields in established Swift container metadata so both APIs apply\nthe same state when it conforms to the S3 subset. The proxy applies that\nmetadata to ordinary CORS responses. Keep S3 preflight controlled by the\nexisting cluster-wide setting because an unauthenticated preflight\ncannot identify a Swift account.\n\nRequire PUT request integrity before XML parsing or metadata mutation.\nAccept a fully verified streaming checksum for this check, and reject\nduplicate expected checksum or signature trailers before verification.\nKeep the public aws-chunked trailer tests deterministic and date-aligned\nby deriving each SigV4 credential date from its X-Amz-Date value.\n\nSwift\u0027s container CORS model imposes these restrictions:\n\n- only one rule per bucket;\n- AllowedMethod must contain GET, HEAD, PUT, POST, and DELETE exactly once;\n- AllowedHeader and ID must be omitted;\n- AllowedOrigin supports \"*\" or visible-ASCII exact origins, not wildcard\n  patterns; and\n- ExposeHeader values must be ASCII HTTP field-name tokens.\n\nCloses-Bug: #2077629\nCloses-Bug: #1905104\nAssisted-By: Codex:gpt-5.6\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f526a41f830a13a105ad5e7dcff1636af6ed6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f526a41f830a13a105ad5e7dcff1636af6ed6da"}]},"branch":"refs/heads/master"},"2de7a57ded2295127294a120ad320904b9fa1529":{"kind":"REWORK","_number":17,"created":"2026-08-24 23:23:21.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/71/977171/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/17"}}},"commit":{"parents":[{"commit":"578c2ece12646daee7db31f20e8478af4210d5a4","subject":"Merge \"probe tests: add tests for copying sysmeta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/578c2ece12646daee7db31f20e8478af4210d5a4"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-24 23:23:16.000000000","tz":-420},"subject":"s3api: add basic bucket CORS support","message":"s3api: add basic bucket CORS support\n\ns3api currently treats the CORS bucket subresource as unsupported, so S3\nclients cannot configure per-bucket CORS through the S3 API. They must\nuse the native Swift API to manage the container metadata instead.\n\nAdd GET, PUT, and DELETE Bucket CORS operations. Store the representable\nrule fields in established Swift container metadata so both APIs apply\nthe same state when it conforms to the S3 subset. The proxy applies that\nmetadata to ordinary CORS responses. Keep S3 preflight controlled by the\nexisting cluster-wide setting because an unauthenticated preflight\ncannot identify a Swift account.\n\nRequire PUT request integrity before XML parsing or metadata mutation.\nReject checksum-protected ordinary HTTP chunked bodies before reading\nthem. Accept a fully verified streaming checksum for this check, and\nreject duplicate expected checksum or signature trailers before\nverification.\n\nSwift\u0027s container CORS model imposes these restrictions:\n\n- only one rule per bucket;\n- AllowedMethod must contain GET, HEAD, PUT, POST, and DELETE exactly once;\n- AllowedHeader and ID must be omitted;\n- AllowedOrigin supports \"*\" or visible-ASCII exact origins, not wildcard\n  patterns;\n- ExposeHeader values must be ASCII HTTP field-name tokens; and\n- joined AllowedOrigin and ExposeHeader values must each fit Swift\u0027s\n  256-byte metadata value limit.\n\nCloses-Bug: #2077629\nCloses-Bug: #1905104\nAssisted-By: Codex:gpt-5.6\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2de7a57ded2295127294a120ad320904b9fa1529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2de7a57ded2295127294a120ad320904b9fa1529"}]},"branch":"refs/heads/master"},"bde63f9d867f4ee48995df4e4da266bc76e775db":{"kind":"REWORK","_number":18,"created":"2026-09-30 17:33:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/977171/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/977171/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/977171/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/977171/18"}}},"commit":{"parents":[{"commit":"238727818d20e871a033cee29ec3e04800738440","subject":"diskfile: remove redundant tempfile cleanup comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/238727818d20e871a033cee29ec3e04800738440"}]}],"author":{"name":"Bastien Wermeille","email":"bastien.wermeille@proton.me","date":"2026-02-17 23:07:58.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-30 17:29:46.000000000","tz":-420},"subject":"s3api: add basic bucket CORS support","message":"s3api: add basic bucket CORS support\n\ns3api currently treats the CORS bucket subresource as unsupported, so S3\nclients cannot configure per-bucket CORS through the S3 API. They must\nuse the native Swift API to manage the container metadata instead.\n\nAdd GET, PUT, and DELETE Bucket CORS operations. Store the representable\nrule fields in established Swift container metadata so both APIs apply\nthe same state when it conforms to the S3 subset. The proxy applies that\nmetadata to ordinary CORS responses. Keep S3 preflight controlled by the\nexisting cluster-wide setting because an unauthenticated preflight\ncannot identify a Swift account.\n\nRequire PUT request integrity before XML parsing or metadata mutation.\nReject checksum-protected ordinary HTTP chunked bodies before reading\nthem. Accept a fully verified streaming checksum for this check, and\nreject duplicate expected checksum or signature trailers before\nverification.\n\nSwift\u0027s container CORS model imposes these restrictions:\n\n- only one rule per bucket;\n- AllowedMethod must contain GET, HEAD, PUT, POST, and DELETE exactly once;\n- AllowedHeader and ID must be omitted;\n- AllowedOrigin supports \"*\" or visible-ASCII exact origins, not wildcard\n  patterns;\n- ExposeHeader values must be ASCII HTTP field-name tokens; and\n- joined AllowedOrigin and ExposeHeader values must each fit Swift\u0027s\n  256-byte metadata value limit.\n\nCloses-Bug: #2077629\nCloses-Bug: #1905104\nAssisted-By: Codex:gpt-5.6\nChange-Id: I7411f219890b6571fd8c71ed04bb9ee9ab031d79\nSigned-off-by: Bastien Wermeille \u003cbastien.wermeille@proton.me\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bde63f9d867f4ee48995df4e4da266bc76e775db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bde63f9d867f4ee48995df4e4da266bc76e775db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
