)]}'
{"id":"openstack%2Fswift~977260","triplet_id":"openstack%2Fswift~master~If4a812fc46d7527b85e1211cd9067b8d9e3ad50b","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-02-19 14:21:40.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If4a812fc46d7527b85e1211cd9067b8d9e3ad50b","subject":"WIP: add timestamp factory middleware","status":"NEW","created":"2026-02-18 18:48:39.000000000","updated":"2026-02-19 14:21:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"e9e7aa45228bfb367b9f3852bdcf73efcf9f8de2","_number":977260,"virtual_id_number":977260,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-19 14:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 19:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06bb454a563f9ac9b31dd97d10a8d2962f2e4180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-18 18:48:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da1b27a9ca2dca11ca06f9e653f540921fea4739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 19:50:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f7f0a1ab507418296397504974a8873\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aacb0a3f9aa749e3b92690b74da4bd1a : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/bb761a4ae75b4d2e8c8e389ec8721ee3 : SUCCESS in 42m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f640fb9cab464bde8961e880ce0640d1 : SUCCESS in 43m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/237ad65efe554a909a33e9badd990de4 : SUCCESS in 40m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/163015082ac64dfeb9e3aca355a6cda7 : SUCCESS in 47m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aefd4849cfc34a88ac7e39b9ae3dc3f0 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebdd014aecfc4cf5a09e686630eecb02 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ee4d4a3e2994e2ea8ada912d0c24b61 : SUCCESS in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87203fb9bf7d49e38f0eb78483c2aeb2 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/777b2f35926642bfbf75e0a485436f7e : SUCCESS in 28m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ca1f012f7e4451955b8cea79523b4e : SUCCESS in 23m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ab411baf20342f9932a9efbd693637d : SUCCESS in 21m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3f0a68446bf64888ac118bbeb78e89e6 : SUCCESS in 23m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/726566edadb3448c9433be844aa27090 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/034d64930e744939a99ac74e18db4a96 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3396b10229f745a0b0d72957692d2fed : SUCCESS in 9m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ba4558245044059a009a6fd2fb532d1 : SUCCESS in 3m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/800ccd3c7b344d6cab882972d73ae81a : SUCCESS in 9m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d388ef54ef0346fa96f45149b9c0e518 : SUCCESS in 4m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2324b7aefd7d4b4c898a193c9cf5ae66 : FAILURE in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca25381d1ff949f19d1c3ba8fc88d759 : SUCCESS in 46m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/216333c3c88e4f4a9790e408a1c23833 : SUCCESS in 51m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11abea14bff846679e0375039385c160 : SUCCESS in 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e847c914f35e4f9eb41325b6d78b8b25 : SUCCESS in 2m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/621e8049e0f24d169e91a93362a0ad77 : SUCCESS in 19m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08cfb0d5c5eb41c38247ff544d380d94 : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cf0899554793730a4d0ffaaa423732ff2a377427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 23:57:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123e938bf9094377be86ae6961bfa178\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/802e90260fb149e78aa916b2f339e7d6 : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":1},{"id":"67cb3fe3fd9c6f5bc30a8ea8eb21c1586056f19b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 11:42:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cd8ebebe27e98d7476ea5653ba00c7aec385c23f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 13:13:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"275be7ca3b9a3e2118cc498e9bd15a076bd6e6e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:11:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4ca13dcd17436fa39e40ccd56fb6ca\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7110d39647d84b7789a04e1631bf427f : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e9e7aa45228bfb367b9f3852bdcf73efcf9f8de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:21:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adfb8d3c6f3640a9a4b9ce545e11c896\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1610d95fc1464118ad246612df999975 : SUCCESS in 6m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/a88f0d92fcfd4e04be8aea9787c05566 : POST_FAILURE in 57m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8346cc3c37d847e981db604bc7be44d1 : POST_FAILURE in 54m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8350f3dc6fd842f8b7849da517ec6791 : SUCCESS in 55m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf6a505d953c40569a08b3ba4128f7fb : SUCCESS in 1h 02m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07383205f92d4ca4a5a0a72383e1536c : SUCCESS in 11m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72432336c54449a2b8c719e4e2dd89bc : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a4ba802bc6c449a9789d370b1aa97f0 : SUCCESS in 16m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e3f1ceac7e04da78c817db4488b4432 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab300775055b460e9ec67a4426cb40b9 : SUCCESS in 27m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d2c62dd500942c48df9c771625f674e : SUCCESS in 10m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0db15c0f6e024e46ae75ed413a35b326 : SUCCESS in 24m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a0d55a9a09c64c7c9db573caed6b93e5 : SUCCESS in 30m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4da2a664f2d049c3b08340c14961b5f4 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ff882661ccd4f429ddcc930b2ffdb79 : SUCCESS in 12m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/32359538886c43a2943ba23a288eeb98 : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c403c18ca74472fad6e048a10ce6f2c : SUCCESS in 3m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a77860bb9ec4b00a273f7ef7ab4fab6 : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fcb4f5ec73e14fc78f44293f29e41be0 : SUCCESS in 5m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7717fdcf3d48440c9c0404f26c2e6e58 : SUCCESS in 1h 01m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c276a425f6e046c1821f83ebf9a13f39 : SUCCESS in 27m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36d5b72791c340ebbf25c187e265ca07 : SUCCESS in 51m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7c10b9743db4e7c82a99b538afdeb8d : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1ee1366e1a46458837d70c16cbb793 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17bdaffccee740d2ba8fa1d14f64df64 : SUCCESS in 10m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70905c6ec2e94cf08f01121b4a188ad0 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b65ad80c5a11b78fed28004d9e07bbd1afdbe94","revisions":{"111feb3278d69d3b87212954a2eae08ac744e0df":{"kind":"REWORK","_number":1,"created":"2026-02-18 18:48:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/977260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/977260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/977260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/977260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/977260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/977260/1"}}},"commit":{"parents":[{"commit":"70256a85180c445a1d513c846d812d125d01a2b8","subject":"Timestamp: add random jitter to the hex field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70256a85180c445a1d513c846d812d125d01a2b8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-18 18:40:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-18 18:48:25.000000000","tz":0},"subject":"WIP: add timestamp factory middleware","message":"WIP: add timestamp factory middleware\n\nTODO:\n- the middleware is not yet used or tested\n- direct_client can use the factory but not read configuration\n- which way should the config option default?\n- more tests!\n\nA timestamp factory middleware installs a configurable\nTimestampFactory in the request environ. The TimestampFactory can be\nused by other middlewares or the proxy app to create a value for\nX-Timestamp headers.\n\nThis allows operators to choose whether to have extended timestamps\nwiith jitter.\n\nChange-Id: If4a812fc46d7527b85e1211cd9067b8d9e3ad50b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/111feb3278d69d3b87212954a2eae08ac744e0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/111feb3278d69d3b87212954a2eae08ac744e0df"}]},"branch":"refs/heads/master"},"5b65ad80c5a11b78fed28004d9e07bbd1afdbe94":{"kind":"REWORK","_number":2,"created":"2026-02-19 13:13:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/977260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/977260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/977260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/977260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/977260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/977260/2"}}},"commit":{"parents":[{"commit":"30b13736f4e091e7ceeb6c472cf10b8ba4631ab5","subject":"Timestamp: add random jitter to the hex field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30b13736f4e091e7ceeb6c472cf10b8ba4631ab5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-18 18:40:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 13:12:31.000000000","tz":0},"subject":"WIP: add timestamp factory middleware","message":"WIP: add timestamp factory middleware\n\nTODO:\n- the middleware is not yet used or tested\n- direct_client can use the factory but not read configuration\n- which way should the config option default?\n- more tests!\n\nA timestamp factory middleware installs a configurable\nTimestampFactory in the request environ. The TimestampFactory can be\nused by other middlewares or the proxy app to create a value for\nX-Timestamp headers.\n\nThis allows operators to choose whether to have extended timestamps\nwiith jitter.\n\nChange-Id: If4a812fc46d7527b85e1211cd9067b8d9e3ad50b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b65ad80c5a11b78fed28004d9e07bbd1afdbe94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b65ad80c5a11b78fed28004d9e07bbd1afdbe94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
