)]}'
{"id":"openstack%2Fswift~977330","triplet_id":"openstack%2Fswift~master~I492926671833c42795d0f8215e40021034e19580","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-02-23 04:54:11.000000000","reason":"A robot voted negatively on a label"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-02-23 04:54:11.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I492926671833c42795d0f8215e40021034e19580","subject":"timestamps: allow BaseTimestamp.__str__","status":"NEW","created":"2026-02-19 15:35:08.000000000","updated":"2026-02-23 04:54:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b17478a70801a4363be58fb3f858624e788914a3","_number":977330,"virtual_id_number":977330,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-23 04:54:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 16:10:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-19 16:52:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff9e5baaa0e4448e0b395abc66a4648f1c914a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 15:35:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ff6a69a8bc97e8ff12ba5cef7a469842f75f6cf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 15:37:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54a567b3daadbe4155494a63c674df1e2746312b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:10:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c5e189edf7408196d9e9cd7a3879e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6f85dc6e5b864491a9554fe137d4552e : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d11951d85ffe9c2de4f55d5793059857b4efc2bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:52:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1304c9c1aa044068acbf49f6fd1e36ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10137c5ac76c4c3386538bab6ca6c99b : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/36698c09925b4039bdb697e1645699db : SUCCESS in 59m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b928605c9b554b278c17bb1b99e026bb : SUCCESS in 1h 11m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95611a7926624452be1ddf6653aa4a72 : SUCCESS in 54m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/778160cb17224d63958bc37658ac39d6 : SUCCESS in 1h 09m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9de615c905d24653b4de569d48c5f0c0 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6f9e69f77b942ff925bd34f230a0bf1 : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a594257f62f49b4a71c45cd8c632558 : SUCCESS in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3f8f492d5cd4f4d99df21de633c016b : SUCCESS in 4m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ac499fbed67c407abd5e60004f878c48 : SUCCESS in 25m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6860dadba9f4b0584847f80db98f091 : SUCCESS in 25m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6630e53a54ef4d7481327ddc27526dc6 : SUCCESS in 33m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e119615c5e89401987a5dee2a5de7b59 : SUCCESS in 33m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/79f83730b6cc4d4f85e5fcea984e3471 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/56cb3ddb213c409182543104c0f1565d : SUCCESS in 18m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e298ff55d3464272b3fbceaca9744ffa : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43f96f5d10b34558a888a5f582d0aba6 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e86fffe064574765858cb5aa5f2b686f : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f6ef7c7648a4c0cb3eb09c67f00a782 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9ffe926931f4415b48957e110c67606 : SUCCESS in 1h 01m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02b0200d375544cebee64587f727bcee : SUCCESS in 52m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6111212617541a2872d531b0ef5bc5c : SUCCESS in 53m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4da7d4e5470342b2bbdbc41078a4c747 : SUCCESS in 32m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb5862881324d49af2b0a5d8e19ddbd : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d59fd7abdae9422f9970c489e6757d33 : SUCCESS in 21m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41fe4f3248364e8fb05450b2889803be : SUCCESS in 44m 58s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_utils.py","accounts_in_message":[],"_revision_number":1},{"id":"7770e51bf1dcdddacb9cb1b3e3a44681938743a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 17:44:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"03887e9af3eabc69005c14a6f28491c8cda4ea8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fade5e4a22042919e74db3ac95ad8f3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07450cba2b9541a68bae593afaa7bb2a : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1b0c570925e5d7ada5c695116a9ae2ff12ee6f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:57:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebeb6f24ad14fd6b7214ea4079a8df9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d219f7487d43e38255c37a94631d23 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/18962edaf5f0496c9fd9c82364a41c71 : SUCCESS in 54m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b27820f30ac4ddf9cbed24a2b2873d7 : SUCCESS in 57m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f84343bdd0624c038ecffb10eb383607 : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc01d5d22244bcfad123681e5c735c4 : SUCCESS in 1h 01m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ffcae7baacc4380b3386379577b65d2 : SUCCESS in 11m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67803712de2145e2adabb915c64f8a25 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d129cd75ef8e4c038227defd2cdc9705 : SUCCESS in 19m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b5c54deb431346df9e9998162ec1c40c : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50d560eb0ae140eebfd5f559d8981ec1 : SUCCESS in 23m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/50e55c57517c4bf18116b2a8c055ed8a : SUCCESS in 24m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/34977acfb66749bdbc198f80dfc54955 : SUCCESS in 13m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d8ca44a81da44ab19eab60bbb4ac8816 : SUCCESS in 31m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7e28a928365c46a9be5d9648334a9f3a : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/52f0682b25574695943c9a63c3c8d302 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c038e06775ee40e39f3c5f547444b05b : SUCCESS in 17m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b12fa1850a3a42f2ab6d0eb88779b762 : SUCCESS in 4m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ae17b5167c24340afd65bce40245f6f : SUCCESS in 9m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/37a9fdcfafad4957a39541a9c2970f20 : SUCCESS in 5m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5eb369a2e89a4ff3be5ab1dd07f1cba6 : SUCCESS in 55m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b3aa9b6220441aa846a322dd5beac6d : SUCCESS in 51m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc9f4d77bde74bed873830fad84b4676 : SUCCESS in 36m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe4f1c069e0b418db9e0db5a9cff236e : SUCCESS in 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40208433ba914d88bb005f84b6d59d5c : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10d6651f6f5547c6a87dd8b790f40ddd : SUCCESS in 11m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dc8d594e08a485e8d77f57deba57171 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e8cbdd6c26069baf74f4e24683c394f78fadce1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-23 03:45:31.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a447a9cc897cf52879a684dbed750b28d8c699c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 04:22:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4764fc6ae14454b488d1dac5933c2b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc1eeda404394a8d83195175da621ab7 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b17478a70801a4363be58fb3f858624e788914a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 04:54:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a863305b6f984d8e953b9dc712e78120\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6ced797cb94b398897a33b77d335c4 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/61f23d0be237473aae3eb4d1afe63bfb : SUCCESS in 1h 02m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34c3b2339c9a46b09214d4263ec305b5 : SUCCESS in 41m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b46def1eeaa4cefa40cc41c1870000c : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/184d521fad6b4846bf1442a54ea6d975 : SUCCESS in 36m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b361dbc63024cdcbdbbcb00d65e650e : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d384cc0f3e744e7787d78bb86dfb8185 : SUCCESS in 11m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59ad7dca8d4741fa9e628e8fde1a9f94 : SUCCESS in 10m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1758740f3ce5415b84988a3fe5dc31fe : SUCCESS in 4m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59ccf3119d7e411e8a9bbd1cd8ac8a40 : SUCCESS in 13m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e8a925c78a74e4f8e16e0d9ea0203b1 : SUCCESS in 9m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/de8f113c2c5b4625abe3fcb76d5be089 : SUCCESS in 21m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bc21790fc1dc48a790815ba54f8bc23b : SUCCESS in 13m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c6f2fc1e9e784586b7fa278d8cd53f9c : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a8327fcb3c24e8196e5847aefe265de : SUCCESS in 13m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0efd0b289a80482ca4b0822ab3c8700f : SUCCESS in 8m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5dc99c4f5b0425996c20b3b371d520e : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fdc9a91f558042bab9ac97d912d5e3ff : SUCCESS in 8m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/77f54f383e67454fb2e07de593344aef : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d29a2ee3d54248c7a26f542bb529d40e : SUCCESS in 46m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/517b0f5c8e0746749a4d6813164e4a91 : SUCCESS in 54m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c49a371ac03f4f478653142c7e9a6ebb : SUCCESS in 21m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e691cbf6bd994fe783b2f8419845d615 : SUCCESS in 23m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea97fd152b0846d9903cfea69b40f225 : FAILURE in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9e4cd1bc94b49a6951c10bc4c3dc2a4 : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7809c56c727041f19f70b63a4697c216 : SUCCESS in 56m 41s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_utils.py","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bc5b239270404bf576ba66e83d8310a9eeb2ca1","revisions":{"f634ee8226cc56880fe816f63f9566eb8cf2a9b4":{"kind":"REWORK","_number":1,"created":"2026-02-19 15:35:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/977330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/977330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/977330/1"}}},"commit":{"parents":[{"commit":"eba8d9d87a87cde66f2f7578cde519c5846a834b","subject":"Introduce v2 Timestamps with jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eba8d9d87a87cde66f2f7578cde519c5846a834b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 14:43:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 15:34:56.000000000","tz":0},"subject":"timestamps: allow BaseTimestamp.__str__","message":"timestamps: allow BaseTimestamp.__str__\n\nTimestamp has two str formats and so historically caller were forced\nto choose by using either the normal property or internal property.\nBaseTimestamp, and NormalTimestamp, only have one format so it\u0027s safe\nto use str(ts) as well as normal and internal - all will return the\nsame value.\n\nChange-Id: I492926671833c42795d0f8215e40021034e19580\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f634ee8226cc56880fe816f63f9566eb8cf2a9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f634ee8226cc56880fe816f63f9566eb8cf2a9b4"}]},"branch":"refs/heads/master"},"841cdffa4eb4ce43b6fa0d961f3932b31d882751":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-19 17:44:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/977330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/977330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/977330/2"}}},"commit":{"parents":[{"commit":"fdda21a7bcfabbd6b5be9c80e6df17ed45886b1a","subject":"Introduce v2 Timestamps with jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdda21a7bcfabbd6b5be9c80e6df17ed45886b1a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 14:43:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 17:44:12.000000000","tz":0},"subject":"timestamps: allow BaseTimestamp.__str__","message":"timestamps: allow BaseTimestamp.__str__\n\nTimestamp has two str formats and so historically caller were forced\nto choose by using either the normal property or internal property.\nBaseTimestamp, and NormalTimestamp, only have one format so it\u0027s safe\nto use str(ts) as well as normal and internal - all will return the\nsame value.\n\nChange-Id: I492926671833c42795d0f8215e40021034e19580\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/841cdffa4eb4ce43b6fa0d961f3932b31d882751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/841cdffa4eb4ce43b6fa0d961f3932b31d882751"}]},"branch":"refs/heads/master"},"7bc5b239270404bf576ba66e83d8310a9eeb2ca1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-23 03:45:31.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/30/977330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/977330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/977330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/977330/3"}}},"commit":{"parents":[{"commit":"48f938c7b9f8a5666f830b1a81552c808376cc7d","subject":"Introduce v2 Timestamps with jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48f938c7b9f8a5666f830b1a81552c808376cc7d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 14:43:24.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-23 03:37:41.000000000","tz":660},"subject":"timestamps: allow BaseTimestamp.__str__","message":"timestamps: allow BaseTimestamp.__str__\n\nTimestamp has two str formats and so historically caller were forced\nto choose by using either the normal property or internal property.\nBaseTimestamp, and NormalTimestamp, only have one format so it\u0027s safe\nto use str(ts) as well as normal and internal - all will return the\nsame value.\n\nChange-Id: I492926671833c42795d0f8215e40021034e19580\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bc5b239270404bf576ba66e83d8310a9eeb2ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bc5b239270404bf576ba66e83d8310a9eeb2ca1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
