)]}'
{"id":"openstack%2Fswift~977763","triplet_id":"openstack%2Fswift~master~I79a0456c3150942f69f22373684141d8c1e56906","project":"openstack/swift","branch":"master","topic":"fix-rebase","hashtags":[],"change_id":"I79a0456c3150942f69f22373684141d8c1e56906","subject":"sq: Make this skip update proxy driven","status":"NEW","created":"2026-02-24 08:39:39.000000000","updated":"2026-02-24 23:57:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96c48702f7b3e3b0dba93d13b947b9d87186afad","_number":977763,"virtual_id_number":977763,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-24 23:57:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 09:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 09:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f3edf8ef06a6450702c4819adeae11969614ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-24 08:39:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5831557b0f5a48ae1c3e7135f960b1f948ba687d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:12:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566922a0104448368475936560c48290\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c0f436766f04337aba1d4b5ddb73cdf : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2a7c86ef4ac6f2901c5aa715ac871740ae8dc7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:51:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ce3393de9b4f62870e777520553773\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8883b98c002416c8d1c2e1ddac6a2ea : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f037bb713d87494fbc8fe81a8132192e : SUCCESS in 32m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/758319965c7343e89145b2e36ab87794 : SUCCESS in 1h 05m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47b03e6587904dba926dccd8bb34177f : SUCCESS in 49m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/111aba1d0efb461a8c6b9602a112bcf1 : SUCCESS in 1h 00m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1e9d2d3d0f74b1b86fb1501533aaa94 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14459adf6eb2479dbb366e4cec654850 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a52b328ca43848769dc3a46ba0175a16 : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c9b4e8dab7c49fc8bfb18f50724e3ab : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e651aa06f98341cc8c1a776b7b3b1db2 : SUCCESS in 22m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/302c0e9765744d20a70021c6519d257a : SUCCESS in 20m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50e08fa9cf2c4eeba85a9c59819f2de6 : SUCCESS in 20m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6a9f038ac12c40d08a09f947ae98374f : SUCCESS in 12m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d678ab3c8724751af75464c3c459b37 : SUCCESS in 7m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6bbcd8b2f5694b7885ecb36d9b0feb0a : SUCCESS in 13m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0f5594e0725347c7a58f341173dc7c0c : SUCCESS in 9m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd0530a0af414891b57e846ea233a297 : SUCCESS in 4m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfbe7d1612474d92901e14130ab2fd93 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3259f94f9c894973a1d738bfba28f6dd : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c895fa9a977d4a829211cf17dd446346 : SUCCESS in 34m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cad9fa8958a7493dbf44d9a23449b1dc : SUCCESS in 47m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/acea4f6dea3d42728afde030a199e539 : SUCCESS in 36m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4415097c4e0d4a24b56fb3fb755deb6d : SUCCESS in 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527cd31d021a4ebeb9ea202c5785d80f : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78623671483144c4a3d1bbfb20bc6426 : SUCCESS in 16m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7f0aeb6f3084f49be9efb9dbce50cd8 : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":1},{"id":"40891462ef0152ab58975f0d3c52b51401cdc910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-24 22:44:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e89ba8b2fcdb41782758102361e766977a681da6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 23:16:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b47817a4a4643f39898f71965841745\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/885ba00a850d43cba0ae5e116da2ff1f : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":2},{"id":"96c48702f7b3e3b0dba93d13b947b9d87186afad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 23:57:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbb257a6940a4c21a7f450f6f113af0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079a8525cb1c4d368314c95a8617c41a : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/5eeb80efea614aa792db8f8503f428ca : SUCCESS in 33m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec9140dc99dc474587608e9d9782e816 : SUCCESS in 44m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df87a843ad454817b685e849d84d090c : SUCCESS in 56m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0d6963ea6164f4fb8ed2bda8f03093c : SUCCESS in 1h 07m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f7a4323a1484c59b7911c35ed826cf9 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d57d238259942f6bae4ba9396ae5d78 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efc203cef5a54632b5bd53f60bf8cf27 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32ba39d9d5a64de3ad8e72b3b67b644a : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a1d7677bf2b548ef8b9ee64233aef1da : SUCCESS in 27m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2d1e9f1924b4cdabad4f317f61bc199 : SUCCESS in 21m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/07ebe3acea714c67a891f7d307c8e880 : SUCCESS in 33m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ff656b7dd16b432788923dbcf1587308 : SUCCESS in 29m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1a55d961161647f38299a03737b95b69 : SUCCESS in 7m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/14f7aea21d544dd3b67c7a926f9f8f39 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d50ed04b87a42c5b095e8b55ff2338e : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17ea0c24fd5d4fb68e1d931123beddb1 : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93a61ba301bd41829365edafb3083196 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dcf5292a2fc74b5494ec87571a352a05 : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6746eb3090c4463e8248493848079ad0 : SUCCESS in 56m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e97c0a9e3354b6a8dfa2f5b1eb15594 : SUCCESS in 33m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e32f4fe5be1846ccba043cf93853053a : SUCCESS in 55m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b715020b1a2e4c5c969852eaecb58082 : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55bc0306c8a4b5f80b870d13fa29a68 : SUCCESS in 2m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81e444be623845db9ada4e69aa2925e0 : SUCCESS in 9m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25826e1205a74f179f2dbe2af3896fbc : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3cca3c749e1be2b43fa7dce720e9d9e368305cf3","revisions":{"8349ecab01ee0122a43baa29f3d50e9c159aaac1":{"kind":"REWORK","_number":1,"created":"2026-02-24 08:39:39.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/977763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/977763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/977763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/977763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/977763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/977763/1"}}},"commit":{"parents":[{"commit":"106202a019bd9787f2968e40474aec287515a59f","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/106202a019bd9787f2968e40474aec287515a59f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-24 08:32:05.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-24 08:39:27.000000000","tz":660},"subject":"sq: Make this skip update proxy driven","message":"sq: Make this skip update proxy driven\n\nPushing this up as a squash patch because I know the parent patch is\ncurrently being carried. And so want to get this right and get feedback\nfirst.\n\nChange-Id: I79a0456c3150942f69f22373684141d8c1e56906\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8349ecab01ee0122a43baa29f3d50e9c159aaac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8349ecab01ee0122a43baa29f3d50e9c159aaac1"}]},"branch":"refs/heads/master"},"3cca3c749e1be2b43fa7dce720e9d9e368305cf3":{"kind":"REWORK","_number":2,"created":"2026-02-24 22:44:19.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/977763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/977763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/977763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/977763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/977763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/977763/2"}}},"commit":{"parents":[{"commit":"106202a019bd9787f2968e40474aec287515a59f","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/106202a019bd9787f2968e40474aec287515a59f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-24 08:32:05.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-24 22:44:06.000000000","tz":660},"subject":"sq: Make this skip update proxy driven","message":"sq: Make this skip update proxy driven\n\nPushing this up as a squash patch because I know the parent patch is\ncurrently being carried. And so want to get this right and get feedback\nfirst.\n\nChange-Id: I79a0456c3150942f69f22373684141d8c1e56906\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cca3c749e1be2b43fa7dce720e9d9e368305cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cca3c749e1be2b43fa7dce720e9d9e368305cf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
