)]}'
{"id":"openstack%2Fswift~977844","triplet_id":"openstack%2Fswift~master~Ic72b0d65348b06815c3ab2939ebed2667603a443","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","hashtags":[],"change_id":"Ic72b0d65348b06815c3ab2939ebed2667603a443","subject":"tests: eliminate some usage of Timestamp(\u003cfloat\u003e)","status":"NEW","created":"2026-02-24 18:04:17.000000000","updated":"2026-02-24 18:58:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"911265c5d553eed5b1c5855bee17c6521fe43f21","_number":977844,"virtual_id_number":977844,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-24 18:58:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 18:47:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 18:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e55765298274469387cecff89245d99e973aba85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-24 18:04:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70885f74e1bb205646a370157505952912e87bc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 18:47:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6651e2265e5048a69a8faaa8191682bf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41e9e41063c849f68a910b6280145732 : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":1},{"id":"911265c5d553eed5b1c5855bee17c6521fe43f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 18:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92f06b2edfb4475a824be85d5b0f1c20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda376036ae0458584f24492ea132a63 : SUCCESS in 9m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba6b9506c35b46368601a51c67e32ee6 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d07c3f6a97414d3c87d91ed16066470a : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b81ea66fc9a14f4ba56ac990f32e5d29 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/026af3c0d79b43788ec2f70781419126 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a600f838f76e45299a17b8689e6e5742 : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2db5b93101804f6d8b5862cf08bc1923 : SUCCESS in 12m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1bcf51bfa9a648d5ac1f58150d51021d : SUCCESS in 37m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2cd4f1be2acc454595683e05d6c33665 : SUCCESS in 28m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4dd3f939c5bf4e21910e6783de0deaaf : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4cf75d2286744d5ab670f6bba9127201 : SUCCESS in 12m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e32978db0fa45188c94ee0e4e4f67ad : SUCCESS in 12m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e229ad4cf12c4f9c99e3f2bd06a6e26d : SUCCESS in 22m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef2ab7a8fedb469bab1c3615e6eb6c49 : SUCCESS in 32m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3086da27657433a80736b587417b27b : SUCCESS in 25m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5256d8c84f7842c6a3290c6d7b092e9a : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa3b14043baa4b84811bf2a9dc197494 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bba8ba8885a8f3c419d185333593c3686b6e77a","revisions":{"1bba8ba8885a8f3c419d185333593c3686b6e77a":{"kind":"REWORK","_number":1,"created":"2026-02-24 18:04:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/44/977844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/977844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/977844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/977844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/977844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/977844/1"}}},"commit":{"parents":[{"commit":"3b4cdeed1c1f9456e04bed2f54e644d342128b85","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b4cdeed1c1f9456e04bed2f54e644d342128b85"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-24 17:16:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-24 18:04:04.000000000","tz":0},"subject":"tests: eliminate some usage of Timestamp(\u003cfloat\u003e)","message":"tests: eliminate some usage of Timestamp(\u003cfloat\u003e)\n\nThe recommended way to mint new Timestamps is to use\nTimestamp.now(). In BaseUnitTestCase we also have a self.ts() helper\nto return sequential Timestamps.\n\nThis patch modifies some, but not all, usages of Timestamp(\u003cfloat\u003e) in\nunit tests.\n\nChange-Id: Ic72b0d65348b06815c3ab2939ebed2667603a443\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bba8ba8885a8f3c419d185333593c3686b6e77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bba8ba8885a8f3c419d185333593c3686b6e77a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
