)]}'
{"id":"openstack%2Fswift~978011","triplet_id":"openstack%2Fswift~master~I1a3d54a290236891c0409c973536112f3eaafb43","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-02-27 11:10:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-02-27 11:10:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-02-27 11:10:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I1a3d54a290236891c0409c973536112f3eaafb43","subject":"WIP Timestamp: source jitter from swift.trans_id random digits","status":"NEW","created":"2026-02-25 18:37:04.000000000","updated":"2026-02-27 11:10:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"65293d1b91a361f94429b3ccfa9f57ee1d9e6338","_number":978011,"virtual_id_number":978011,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-26 19:16:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2026-02-27 11:10:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2026-02-27 11:10:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 19:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-25 20:30:20.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-02-26 05:19:39.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"3439d17e4244f7ac22fe74a8986ee4e328f08865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-25 18:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60666e66b9dd27442a1558a761221ca9ef6177a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:50:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5423f4f3a5034ab1983b48134c443255\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1dedcfda6e443689ce668caab22daa : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/99f86b1918b4462f8d5e5ca3806103e5 : SUCCESS in 48m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f7c678b8bfd4bd1b1c46d84abb5d0c2 : SUCCESS in 35m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b5fae0deb4d43acb08975b939b5dac9 : SUCCESS in 40m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d63469db8b0485897fd7f05f0e48b3e : SUCCESS in 49m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff0d53f90ca64ca58a5af29e685725e3 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae1cca13c59146c99cf48de51edc7a44 : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d3fbc85d4b64071ba4062aa01bfe47d : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0182f31017ca446eb0d2cdec586a275f : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c552316b29e41449e2982e754802f95 : SUCCESS in 28m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/227bcef2dbda48419884f40412638d10 : SUCCESS in 23m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a36d0f71360c4dab9efef8557cf1b706 : SUCCESS in 31m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/03c62c3237574264a286285ce071e416 : SUCCESS in 12m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1f646f4932534e8395ab4ce3253bf2a0 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/059d81ff20ae4278b7be779b841ae42c : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2f6009ce00824bada70e23e58d1e7da9 : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0db0f1ac551d429baa1d516b4621edc7 : SUCCESS in 3m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ee39689fab4472386743cc078b3140f : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e267c10fa5d0475191219728257701d9 : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/defac0cb02304a2bbe55e620e445b6c8 : SUCCESS in 39m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e451068b7e74a66a95541897f42c0ed : SUCCESS in 37m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/432b4817cdcf45a0b6daa4bd6dceee76 : SUCCESS in 51m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02608ba646c345288e0506cce15c50df : SUCCESS in 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808120d673854ac5afa751dcf918cdf7 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f51fbfb5b06647dea6689aec966b07e2 : SUCCESS in 9m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b630abf4fedf4ddba17638e4f222d00c : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b86ccbd6ecc58e0dfc0e343e9edb89c62f178d76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 20:17:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666d74665dcb4556b3f2b0de923b9005\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/969b11d92c8449e7b360fa4cc8f83ed6 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c07e6aa7a65e9edd79be9ceb875090fd920e64d5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-02-25 20:30:20.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8ef35a5424f0cc5347aa5206f009fe049d015420","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-26 05:19:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc794f983cba5e059fa91eb81db14e6610873b00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-26 17:52:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e6b43d92d14174642bd46e72798a6e2d29bb8ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 18:59:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b703118d544cbca4b08c5f8fcc470c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dce524e1131449f2821a0800a16adda5 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b94b4e304a51071982a732fbbd89929ae68cdd2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:16:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a67aab2a5f4754b414e58d52faf603\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a040a362804e5ea242ad764873b898 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/38adc90cb8424e06ba73792e45810b4e : SUCCESS in 1h 03m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed4ea97e4b4249d287509c67fb84308a : SUCCESS in 1h 04m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2119a4be5ded42beb3c76a103a9a1ed3 : SUCCESS in 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0007761dca364f74b6772ea74ccaaeab : SUCCESS in 54m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cc70f11895f432880c927b222196c40 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f9bc3388bbb44fd96384c0d5780cc13 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/928e5972539b497b8f586887513139a4 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/028471c9a7384bdf837483ff41d902dc : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eafa9189fcfb48529ad1e86624bf2656 : SUCCESS in 12m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7aa87653577a43478c9294b1620c13bb : SUCCESS in 23m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2bdb4c9d2d1a4b4eb859943c6543c1db : SUCCESS in 14m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bc8cca341631420eaa048b647b64ab2e : SUCCESS in 35m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb89d07a1113434ead7c86eac3e8598e : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c9692288856d44519674f4c4e7d14ab9 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f0b7b2469fb548009d84794118d36276 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/497cbac3dfa9416fa55f1b0b4b3b74bc : SUCCESS in 7m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f608a054a344abcb9921be5747cb60c : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/78c38e915fd740989c6f427ba1ce7957 : SUCCESS in 9m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e535f80413a432999b567df558a6307 : SUCCESS in 55m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd72ccd78c1247bdab66ab5f3dc30fb7 : SUCCESS in 42m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fecae048476243fd876582b52410c7d7 : SUCCESS in 1h 00m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7448bb0c42849e6bfb618df9b757c57 : SUCCESS in 25m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9509021ee30146ceb7fde65feff0807e : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4b55f61d81144fa817570ac4e73f562 : SUCCESS in 10m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0726b1a28a44934a009cc8ed578b9d7 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":2},{"id":"65293d1b91a361f94429b3ccfa9f57ee1d9e6338","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-27 11:10:01.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ce5dd0efc5f0c163371d2352b961141bda1712e","revisions":{"ab910bd44411b65dd40f9d2710f460680e6c7d28":{"kind":"REWORK","_number":1,"created":"2026-02-25 18:37:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/978011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/978011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/978011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/978011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/978011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/978011/1"}}},"commit":{"parents":[{"commit":"3b4cdeed1c1f9456e04bed2f54e644d342128b85","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b4cdeed1c1f9456e04bed2f54e644d342128b85"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-25 18:09:55.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-25 18:31:53.000000000","tz":0},"subject":"WIP Timestamp: source jitter from swift.trans_id random digits","message":"WIP Timestamp: source jitter from swift.trans_id random digits\n\nTODO:\n- probably needs more tests.\n\n- an ensure_transaction_id helper might be nice, that returns the\n  transaction id.\n\n- do we want to use the \"version\" hex digit (the leftmost) to indicate\n  when the jitter does relate to a timestamp, or not, to avoid\n  fruitless grepping? We ought to always have a transaction id in the\n  proxy pipeline but not with direct cleint requests.\n\nTimestamps embed 9 random jitter hex digits in their hex\npart. Transaction ids have 21 random hex digits.\n\nUse 9 of the transaction random digits as the Timestamp jitter, and in\ndoing so enable the jitter part of an object variant name (i.e. the\nwritten down Timestamp) to be used to grep for transactions.\n\nIf not transaction id is available, make up 9 random digits as before.\n\nChange-Id: I1a3d54a290236891c0409c973536112f3eaafb43\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab910bd44411b65dd40f9d2710f460680e6c7d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab910bd44411b65dd40f9d2710f460680e6c7d28"}]},"branch":"refs/heads/master"},"2ce5dd0efc5f0c163371d2352b961141bda1712e":{"kind":"REWORK","_number":2,"created":"2026-02-26 17:52:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/978011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/978011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/978011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/978011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/978011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/978011/2"}}},"commit":{"parents":[{"commit":"a70327460b04baf94cb663ec9ee6bd91784e7132","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a70327460b04baf94cb663ec9ee6bd91784e7132"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-25 18:09:55.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-26 16:39:39.000000000","tz":0},"subject":"WIP Timestamp: source jitter from swift.trans_id random digits","message":"WIP Timestamp: source jitter from swift.trans_id random digits\n\nTODO:\n- probably needs more tests.\n\n- an ensure_transaction_id helper might be nice, that returns the\n  transaction id.\n\n- do we want to use the \"version\" hex digit (the leftmost) to indicate\n  when the jitter does relate to a timestamp, or not, to avoid\n  fruitless grepping? We ought to always have a transaction id in the\n  proxy pipeline but not with direct cleint requests.\n\nTimestamps embed 9 random jitter hex digits in their hex\npart. Transaction ids have 21 random hex digits.\n\nUse 9 of the transaction random digits as the Timestamp jitter, and in\ndoing so enable the jitter part of an object variant name (i.e. the\nwritten down Timestamp) to be used to grep for transactions.\n\nIf not transaction id is available, make up 9 random digits as before.\n\nChange-Id: I1a3d54a290236891c0409c973536112f3eaafb43\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ce5dd0efc5f0c163371d2352b961141bda1712e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ce5dd0efc5f0c163371d2352b961141bda1712e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
