)]}'
{"id":"openstack%2Fswift~978242","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I194e4173e9451158897928fe910d3b29c73919ce","project":"openstack/swift","branch":"feature/threaded","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-03-06 18:28:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I194e4173e9451158897928fe910d3b29c73919ce","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/threaded","status":"MERGED","created":"2026-02-27 17:17:37.000000000","updated":"2026-03-06 18:30:33.000000000","submitted":"2026-03-06 18:28:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978242","meta_rev_id":"e243475f4c001ed35d9811a0b26d80f0a6dde7c8","_number":978242,"virtual_id_number":978242,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-06 18:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-03-06 11:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-03-06 11:20:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 18:10:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-06 11:20:18.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"95ebfa3270e0bc5fe36081055d04319daeb8f0cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-27 17:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df5be6bd11b8bd8d6062e162a6c9cebe38c0ec28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75175acb3c042719385623baaa6033b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5084151cecaf48afa9ea023a728fe4a3 : SUCCESS in 5m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/739bb0a057bb449db94e8ca916fdd2c5 : SUCCESS in 9m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da1823b6f7a04d89988e39377e594011 : SUCCESS in 10m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6aa05b4150147569857df0a1f5efdd9 : SUCCESS in 11m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a201f20ed27e4525bf07951034f6e2c3 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32fcff3bb8454c0c9b0ab86cc0f47f57 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9cebff33d5a54dfcb82be50e18be7e8a : SUCCESS in 19m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/86b6229108794f72a1c05f08cc52e777 : SUCCESS in 10m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/266eb4b1327b454aaf603ebfe742b966 : SUCCESS in 21m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/37d2a2bfd3ae42df98fc9b135effe394 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8855d37651af4757b54d45bdcd1ca7f8 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/397bbce84d7146309ad04c9c8b4d20a6 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7db5f387588e481e87d8b8d13a4ccaa8 : SUCCESS in 4m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc3648aefac746a58cd76dbb14430502 : SUCCESS in 8m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/07a7d8e61cf34173b9aa5cf07f42fdc4 : SUCCESS in 5m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a16687c3d002419581bb740896d6bf0e : SUCCESS in 45m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d73717cc1394f029e8ce8e2dba68f71 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576c84c4b1964af8a25b80d142008c5a : SUCCESS in 2m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dbbf8e5d2954f9da558a21916c9421f : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"daed13bf94fc69734c6156d334b09f4e78043c55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 19:52:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8180ac2bc3c4aadaa07c184aa14f7c7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b96c324bc41e43f6bb3f669cf5fe50c9 : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f186fe71363e4a487075f8d57d247712d88ac482","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-06 11:20:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f13e1e257fa84efd4b10c3b11a0dfd560d393af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 11:20:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa1f10ab16c5254a217fe064f4cbab16d2424ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 12:20:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa51dd2cba3746d5a1399186463bb169\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6b141edca545ecb05251421267d07c : FAILURE in 3m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b6cc2d0eb19a404d83f0998e1cfaddd0 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/14c8ecb7e8644a3ca00b46f8fa0f470c : SUCCESS in 10m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53dd218c45aa48f88acbfeb760f0f885 : SUCCESS in 8m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fd6040f4a0be403a82532f46f071555d : SUCCESS in 10m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/041a39c22a564e84bcafd2e5a8ae8500 : SUCCESS in 11m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dffadefafc50426ea2ddd023fce030cb : SUCCESS in 7m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5a77235d5255423ca015d25f548baa64 : SUCCESS in 8m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a4a18ef594b940c8aad53463140c2588 : SUCCESS in 8m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a1fbca4700741aa8f3310757fd07541 : SUCCESS in 4m 54s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e25dc35023d48419871085ed4df3135 : SUCCESS in 5m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/05fcd6bfb2184ce08e16629259f8ab37 : SUCCESS in 55m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19e88698fe704b5e8e2e90c1655829ea : SUCCESS in 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cedcc70cd32c481e99a57ea65b9eb76e : SUCCESS in 2m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ffc47c7088d642ab9f0c1bbc8489ee81 : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ab0a3c02c7fd6ab84675ea9a332998d5f28e3db0","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-06 15:05:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc081a306f8a92d4c73aa81625cd1f0ebccc663a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 15:37:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9507cdf74c16415b824da1b5b6fde4b2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8457b06b0bfb4da89a593ee8b7b959be : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fc069a83881b0943318432221c5e200b155b1cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 16:12:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d03c9008cff4356b603db051434bcc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62a66e785d146569f3c0034c976f727 : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66234f0376784784a2a14d0d43d8fe8d : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9640ecdf3a046279ecf524ea03a597f : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ec9a27b44994a93a0ea9d17b583a75a : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8eacb12a6783450696686299051236eb : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/034f1795f70249b493cb6fbec4ef5f7a : SUCCESS in 24m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad4b1f8e62f247a39654ad36d28d57a1 : SUCCESS in 17m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/610548d9134746f1bafd9978282c84d8 : SUCCESS in 43m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/042b139414784d8aacf2a2eb02e67e98 : SUCCESS in 29m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/60a7e40e21384c268367c21dc93d4b3f : SUCCESS in 8m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aa682395d66c4a8488caa00988904bac : SUCCESS in 9m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6723300ca2d34abf9495ca940eb66aab : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/12c3c6015d584b188e32440fb0c3f10f : SUCCESS in 7m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1444a02fb5546ee8afabb50f8dae7c2 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/61c560b95340422f9173b402cfd84a2d : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6514e144794f4d8f22216bbf81d4aa : SUCCESS in 59m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc05ff75689f41bb9c4878c71992db99 : SUCCESS in 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38be59f01b24eae9516b7654a8e10a4 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/020ffcbd5aa54209b485c67a5cf7db21 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8d31b8356faca1541686d82220dff24aed204c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 16:14:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"915bcc5df71a94cfa435150daf57ecb753a76ccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:28:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84294ca596d949cf940972c4d9fd2795\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523bad815b5c4aff864699f8ab61f9aa : SUCCESS in 9m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/131ee2b416654f7c8927b3eb0a07a76b : SUCCESS in 7m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/422a9c3bc8544baf8d64c53048ef904e : SUCCESS in 24m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc25d52d58b54adfb8e9faa41a7c2927 : SUCCESS in 22m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/80a8672049784ab3ab95105101cf8ed9 : SUCCESS in 30m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/16bf27f4f3ee48b9b5ffe78277d18962 : SUCCESS in 32m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ee1f19a099bb4ef0ba8c5c3bb41feca4 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9a0698b2965542f8ab990d18783ba7da : SUCCESS in 8m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5271bcb7772a408aa5b150798be87190 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a3a5ddf382844e39f7aa7e143d45ac0 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/47a9471f7736412b8a2326c9b19b55e0 : SUCCESS in 11m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd0ba891a662438e802b68eb9acfbd42 : SUCCESS in 59m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/273d3495fff946f099c013951bb713b4 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1882d6133ccb4fdca15224d22adfaeab : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eed1e54f642443728045f6e1757fa3f0 : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1bb3d8812887ac1d26f3c16616acdeb9e53fcd2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:28:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e243475f4c001ed35d9811a0b26d80f0a6dde7c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:30:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d870f29bcc8749198caa097d566553ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087671749c1442ca8f84b6d4665f1844 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cf1fd2d6284042e9af98fcdb9e738479 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f68ca11097768226a133526ceef8ec7c445d6dbb","revisions":{"f68ca11097768226a133526ceef8ec7c445d6dbb":{"kind":"REWORK","_number":1,"created":"2026-02-27 17:17:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/978242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/978242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/978242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/978242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/978242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/978242/1"}}},"commit":{"parents":[{"commit":"3a21251a9472acc7d438d7dc1abd4c055b5c71ca","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/threaded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a21251a9472acc7d438d7dc1abd4c055b5c71ca"}]},{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-27 17:12:07.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-27 17:12:34.000000000","tz":-480},"subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/threaded","message":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/threaded\n\nChange-Id: I194e4173e9451158897928fe910d3b29c73919ce\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f68ca11097768226a133526ceef8ec7c445d6dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f68ca11097768226a133526ceef8ec7c445d6dbb"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
