)]}'
{"id":"openstack%2Fswift~978572","triplet_id":"openstack%2Fswift~master~I946636249c61897b82e18451bf7a116261d190ab","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-03-06 11:25:17.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-05 22:00:35.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-03-05 22:00:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I946636249c61897b82e18451bf7a116261d190ab","subject":"obj-reconstructor: use Timestamp.internal in frag prefs","status":"MERGED","created":"2026-03-02 17:46:59.000000000","updated":"2026-03-06 11:25:17.000000000","submitted":"2026-03-05 22:00:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"978572-p-timestamp-collision","meta_rev_id":"391defa9a44e8cdc000689c477ed8a3a6e3e6d9e","_number":978572,"virtual_id_number":978572,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-05 22:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2026-03-05 20:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2026-03-05 16:59:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-05 19:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2026-03-05 20:18:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 19:06:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 15:42:39.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-03-05 19:12:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f06c5295e30d03c9385cbfe654eaf54da7bc7f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-02 17:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3aa0ce3b4ec3f7bf6dda57a39460b06f09af1309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 19:06:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21df118414304273ac8935dae930e297\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b049d3a2874c3d95ef55bc92902f37 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/b6f1c1d90d954e0c8c753698896c0e38 : SUCCESS in 54m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2a1dfde1d53495c975483fbf7908feb : SUCCESS in 1h 09m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1682ec7847c4fed878140b6ce9fea73 : SUCCESS in 1h 01m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99d5fe10047c4223ae6b581a4b8c292f : SUCCESS in 1h 05m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58f7154a5c194f0386e3f2f998f09ca8 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fa6297ac91848f98362467dbf60c562 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efb7d169dd494e0496be71b3d165b57e : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0dde29a79f649149b551ff831bc70dc : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2535681e14514cf7bf0676a987cf6da1 : FAILURE in 26m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aac775376aa64d2ca47be2ae2cc73809 : FAILURE in 23m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bc8dffbe5871430b8c710c5697a544fe : FAILURE in 38m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4e6a599890ac4e91a6646e8371447ab4 : FAILURE in 32m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0a504a131f8545d2b3287a01f260ec1d : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0d46e7cafb2645ed92940d5d8bbf60f2 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85ca921ea89446218a7de9be75f5e058 : SUCCESS in 12m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3082edcdc12046619f4589483c8d9ee3 : SUCCESS in 9m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edbd1e97474c4f02aa8d3d71611d9927 : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9286349c9787411fa65ef13447fead42 : SUCCESS in 8m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa6921f8638144fd97935eea8272e1da : SUCCESS in 1h 00m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f39ad94b191d4faf95e18e6d25b858f5 : SUCCESS in 36m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c47549b5cb004c9aac200539812fb36b : SUCCESS in 50m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f16a346fa6074cd581862a817e394e6d : FAILURE in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c548e267953e416188939f29bc97a10c : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e760dcd88114e0c9e4bbf38d4e77a7b : SUCCESS in 18m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f45480b6f37046f48d7f753d379b7c52 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d1b614b7955c13b85468811db40ce819fbdf7ce6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 22:02:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d3207b1a6742d79fc33a7b36600043\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ed7831ea516409b88608236651d0bc7 : FAILURE in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fd18b0aafb6d88d62e99e3173ba25d5a2e6d4ed2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 05:32:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"98530c93187cfa63550702c6a21a19b3f00a7b80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:03:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1782f74ff7b44b959e5e4ebf5f8f9547\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ae4b9daed8b48edb80b05f4c90ea69d : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":2},{"id":"29a67bd7bbee1007127cf25d216bd5cecb4d9805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:51:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b387b419ec74b44b3397a058746153c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed769c5c99f1481d89f7e957f1281f90 : SUCCESS in 7m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0a7d7469ba37466bafaa5b1b075a258c : SUCCESS in 1h 00m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6359dffd5e304037b61a80fcda767aa8 : SUCCESS in 1h 08m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5190f3cc8ebf49339dd44ce7fe423cba : SUCCESS in 34m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e8f7728046442f1bbd7b0601691e66d : SUCCESS in 36m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4ca52fa8da640708cec1f110fd99f28 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64088591352c43a489e796cfb46191bb : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eced8892cf8140799f209c1844ac2472 : SUCCESS in 15m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38a5307f9bab479a918dc7fc85f81f35 : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d6f9c0bbc7254aeea38172c9577a055f : SUCCESS in 21m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2d76ae4792049848f58cc24568a20c2 : SUCCESS in 22m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/decbf9101665410d8c4a5742889d3adf : SUCCESS in 30m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d40c904d26d34839b61fe169671d57b3 : SUCCESS in 11m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5f98c9b02b9c462e9f7a15b5a3d802fe : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0d205178b88b434c9b97f84df48f8080 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30821315e2154b18bcef48a9e77b2058 : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a41528b424340219aace7bf71646155 : SUCCESS in 6m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2fb9192e95e4073993b9159f4854da6 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e6dc5531061453a8bb72046f804b071 : SUCCESS in 4m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2018e2e8ad7c4074b76ed9e5c3c800d1 : SUCCESS in 57m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1138d927af9c4870b91c8e40251d555b : SUCCESS in 51m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e422af5232e4e4bbb2dae535abf05df : SUCCESS in 47m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fd6494214af4f279151db82a797ecd9 : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9ff2b98302475f9a710ae8b9ff3424 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a33f3d9a3bb4a7da79c880df69603bf : SUCCESS in 21m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6507854f71964b7191ac273e8328ec16 : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":2},{"id":"14c726a798a3a42485074c105a3c401441052eb4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-03 15:42:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d7ff49bbdf6f61714bdca2d74abb55ea7afa80e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 20:59:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e9bc8fee71a4af9ae2d30c1936a12f57ddcbdcde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:10:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a063a83ea640a1b1b330397ba24e27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad714581872740b6aa3d46835acb2b1b : NODE_FAILURE Node(set) request 4b9feb0d5e0c4dfd8125afe0a5dc575c failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"8822e02d47240d3d44032d025c2234c41c0b55d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 22:08:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad9f38f94f24e538079738195e8128c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47944c91cc740b58b604afc0f5114af : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ca5aa918542c47fda0abc935f297b333 : SUCCESS in 59m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34b1bb4050f544b8b9d1627d4e1e9426 : SUCCESS in 31m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4051b327b70b48bba2d705a4fdf72d71 : SUCCESS in 35m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90238645fc93490e840442c9222d0b31 : SUCCESS in 1h 03m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/143da7292c584cdc8364c9ccdc79ffc4 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c08ff9930ece45b69d68ff5755b498e9 : SUCCESS in 23m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b122c43a0b546f69a815a6a17a38f4e : SUCCESS in 10m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d5666dcc598410589e7e0857c211e39 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b24a4ab56fd4e69a9ee7019c50fdd4b : SUCCESS in 25m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14789fcfe6f8459aa7e1b0e3d315c289 : SUCCESS in 25m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d68226e3803c4ee0a2d75357ce4dcf53 : SUCCESS in 24m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/19110f8265d8431da016203afc10c812 : SUCCESS in 11m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aa435bd7393d4e2199d25b39188ac222 : SUCCESS in 9m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b046e1fecf2e480e85979dfd4ac6eb2a : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14bf8e04e3ed493592a490d3036fe40c : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e164d33bc4a5422390419fcdb03f5389 : SUCCESS in 3m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c83c96a70d7949369ad71cfc6d097e3f : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fd10de507d042078e700afe2c3ad6d3 : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3355499860e6420d99b7cc4ffd10493d : SUCCESS in 1h 00m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10b641ab15814989b20921b05994c584 : SUCCESS in 58m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/932b207cdbdb4f85aba1b1c9e73d2af1 : SUCCESS in 51m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5af33720d69b40d780f15d43563ddd16 : SUCCESS in 29m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb23beb1dd264a529673f85bb8b143a6 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5d9f47fe3a54a85861ae7e1f7e0c938 : SUCCESS in 25m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fc4c614eca24452973f91062bb4d751 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":3},{"id":"4a2e462b67719d373a72069f442922ca89a435f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-04 10:19:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"64fab79401dc0b41f0a21d0fb8de83c74d453f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-05 16:53:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e1559d9e897ec14c631f6bb754cc5dc519df443f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-05 16:55:38.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c9181a5b9ae6dccff699d4d0ff5f3eee18df0641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-05 16:57:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cf20ccbde9108ff430dfc6fc93448f98875a7754","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-05 16:59:05.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ced1bd1660fcf467e570753074c67fc69c7828d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 18:02:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd27c2bfe9074802a6abc3e83b3af343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab78ce4adb244d9eaa9e4c92837f19de : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/59069b318e03407ea2252e4e571c43ef : SUCCESS in 57m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be60158c771f4c71b03393581bfb79a2 : SUCCESS in 59m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1961e9c681c4f2091d5d29032d49533 : SUCCESS in 58m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/855b44b858e94143aa3faa3ee57519a3 : SUCCESS in 1h 02m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e049b7d1b8b456891d7fad4877dfbb3 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b001b5cae4c4a5ea4f9c61c1c82d504 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/230e3edc4c8e49dd8f8c0aa02c163d42 : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d342a3b32572443bac85dd5e35a6401a : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1aec2caba925471bab8534b58beba5b9 : SUCCESS in 25m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2fa9f6bb4864459b9404bec37ac9189 : SUCCESS in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a846c5d8e5754335b83760eb4a6c1b22 : SUCCESS in 30m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f5d9eb7e3f5c4961b2892d59cd8448df : SUCCESS in 28m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6c7d12cdb4d54a97a0e00ab46a6381dc : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a4c63d8976345eb83925d9506d36fe3 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d96fa79ac4d44347848f3fca71622501 : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02bdcf6d8f4e4c9da263e4dbed7a5870 : SUCCESS in 7m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e386ca637a74965a4a0199c6aa9f8c7 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a474407fe08f458eaed6fc878995b8a6 : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/953d9c2f0caf4693bf080eabd8915a32 : SUCCESS in 56m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2669d4fd7b784da4a3b982d4dbeea548 : SUCCESS in 53m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85fa517bd2eb430a9ad26fac8d9bea35 : SUCCESS in 51m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93d20e6ad4e146f29872217f32e9abe8 : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3756bade9be4731800b215e00c9fead : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3a9250fe143401ca8024dd06da7cc4b : SUCCESS in 25m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4daabbb7bc24779a33fbbaa8b3c84cc : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":5},{"id":"528ac0a0db9fe5fe81664ebb76f3739fcf1aff08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 18:04:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953eb57c976543a78d3a30765e8ef800\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0e394e2a92340b8abec0c059c841710 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":5},{"id":"04b93f94e3bde4db6cc7db9cdf282bba032c756e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-05 19:12:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b0813c687a1928753f799c6ad519eb71e395cdac","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-05 20:18:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"48eef071cbaf31416848cde2a0049ffbd866e7ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 20:18:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"95ed57fb2b04000fc36bd3b9b178e56c2312b93f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-05 20:26:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3dd17d9375fcffde4cd7b0fda8f60c7f9494a43d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 22:00:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fd3d018297a4a34b667af7132ee297b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99ffa384736481da36e844ca3e60ab2 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/d11440e96ee54cdba32b71608e8c96b3 : SUCCESS in 37m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26e31fd77e3d4cfc9fc1308bb6cdb04f : SUCCESS in 1h 04m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c77d9adf0fe140f5ac4413c244bb972e : SUCCESS in 55m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fc585ef00d84bf498dba88b8c7c003e : SUCCESS in 1h 09m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3e22e5af41ff42eca24ce0ad175b210a : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c35258edd69c4de48009b9e2c6c88e6e : SUCCESS in 10m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f6d4fd6c5794f87a96d6c0a6ae03cb1 : SUCCESS in 12m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/667ebf117ede4ff8a22f2887ac2bae17 : SUCCESS in 14m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/788653ae067f4bfe9c6e1f98944b9bd7 : SUCCESS in 16m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f55e61afb25645f1b0e44d9368ec52b4 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/589bc1cbf10b42aab9a51573702801dc : SUCCESS in 14m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7afcdb7b77a34cfab51195bd162d77bc : SUCCESS in 12m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6cbccc6dbcc944d29a0a5d8d0340aaf4 : SUCCESS in 9m 30s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/72fff3d464d247e3a41d15efc6e21c7f : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4cd4283fa3d4a8aab9decda1aaada52 : SUCCESS in 40m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97826a61b3174c308f8285e94c8c2429 : SUCCESS in 57m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82620412aa3b4775bf1e47b04b2a09b6 : SUCCESS in 36m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59ca18e17d204c4db1fc4aa593f1bcea : SUCCESS in 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0830b7e26eb14d04a857ca1686b041bb : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/efba272ba09f46348c6d8522364c25e0 : SUCCESS in 17m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cef6b9a1a95a4eff97b71f145db73c16 : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":5},{"id":"dcb6b10bad947f2de19e2e0de57b44a6d12164c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 22:00:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4af2f602c14445f01884c0c3b97b1f9732b94f7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 22:02:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d2e8af1aa24e20a17794a267756a38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0c91055e8543d99c4a038db8fc9fc3 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/afdcbe82eb514d05ad1a6afb3702880e : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"391defa9a44e8cdc000689c477ed8a3a6e3e6d9e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-06 11:25:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1f488b2842eb27c9b1caad44caffb18f8f95a228","revisions":{"513b58e697270d8a71c9c0d8477154ea1716f78f":{"kind":"REWORK","_number":1,"created":"2026-03-02 17:46:59.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/72/978572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/978572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/978572/1"}}},"commit":{"parents":[{"commit":"9ea0d0b5909fe47ea1c53fb8b84a350c566b2dca","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ea0d0b5909fe47ea1c53fb8b84a350c566b2dca"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-02 17:31:22.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-02 17:43:47.000000000","tz":-480},"subject":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.","message":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.\n\nChange-Id: I946636249c61897b82e18451bf7a116261d190ab\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/513b58e697270d8a71c9c0d8477154ea1716f78f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/513b58e697270d8a71c9c0d8477154ea1716f78f"}]},"branch":"refs/heads/master"},"0fe71bd40210db1e25096d5471ad1b30a3ae67c9":{"kind":"REWORK","_number":2,"created":"2026-03-03 05:32:24.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/72/978572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/978572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/978572/2"}}},"commit":{"parents":[{"commit":"9ea0d0b5909fe47ea1c53fb8b84a350c566b2dca","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ea0d0b5909fe47ea1c53fb8b84a350c566b2dca"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-02 17:31:22.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 05:31:21.000000000","tz":-480},"subject":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.","message":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.\n\nChange-Id: I946636249c61897b82e18451bf7a116261d190ab\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fe71bd40210db1e25096d5471ad1b30a3ae67c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fe71bd40210db1e25096d5471ad1b30a3ae67c9"}]},"branch":"refs/heads/master"},"886790163c0663421460f7a7497a367c9a524cb9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-03 20:59:26.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/72/978572/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/978572/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/978572/3"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-02 17:31:22.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 20:25:48.000000000","tz":-480},"subject":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.","message":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.\n\nChange-Id: I946636249c61897b82e18451bf7a116261d190ab\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/886790163c0663421460f7a7497a367c9a524cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/886790163c0663421460f7a7497a367c9a524cb9"}]},"branch":"refs/heads/master"},"571155d3f0617bb28e17dd1c4521daf4133b2f51":{"kind":"REWORK","_number":4,"created":"2026-03-05 16:53:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/978572/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/978572/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/978572/4"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-02 17:31:22.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-05 16:44:34.000000000","tz":0},"subject":"obj-reconstructor: use Timestamp.internal in frag prefs","message":"obj-reconstructor: use Timestamp.internal in frag prefs\n\nPreviously the reconstructor used the Timestamp.normal format for the\ntimestamp in the fragment preferences it sent to remote nodes. If the\ndata timestamp has an offset then the abbreviated normal format would\nnot match when looking for fragments on the remote nodes. As a result,\nremote nodes could send a different (newer) set of fragments.\n\nIf this happened, the Related-Bug could result in the reconstructor\nrebuilding a corrupt fragment at the requested data timestamp using\nthe fragments returned from remote nodes which are for a different\ntimestamp.\n\nCloses-Bug: #2143206\nRelated-Bug: #2143202\nChange-Id: I946636249c61897b82e18451bf7a116261d190ab\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/571155d3f0617bb28e17dd1c4521daf4133b2f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/571155d3f0617bb28e17dd1c4521daf4133b2f51"}]},"branch":"refs/heads/master"},"1f488b2842eb27c9b1caad44caffb18f8f95a228":{"kind":"REWORK","_number":5,"created":"2026-03-05 16:57:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/72/978572/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/978572/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/978572/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/978572/5"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-02 17:31:22.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-05 16:57:20.000000000","tz":0},"subject":"obj-reconstructor: use Timestamp.internal in frag prefs","message":"obj-reconstructor: use Timestamp.internal in frag prefs\n\nPreviously the reconstructor used the Timestamp.normal format for the\ntimestamp in the fragment preferences it sent to remote nodes. If the\ndata timestamp has an offset then the abbreviated normal format would\nnot match when looking for fragments on the remote nodes. As a result,\nremote nodes could send a different (newer) set of fragments.\n\nIf this happened, the Related-Bug could result in the reconstructor\nrebuilding a corrupt fragment at the requested data timestamp using\nthe fragments returned from remote nodes which are for a different\ntimestamp.\n\nCloses-Bug: #2143206\nRelated-Bug: #2143202\nChange-Id: I946636249c61897b82e18451bf7a116261d190ab\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f488b2842eb27c9b1caad44caffb18f8f95a228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f488b2842eb27c9b1caad44caffb18f8f95a228"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
