)]}'
{"id":"openstack%2Fswift~978816","triplet_id":"openstack%2Fswift~master~Ida910415f92aff721cda4c78830b1ce52b4e78da","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-03-13 14:11:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ida910415f92aff721cda4c78830b1ce52b4e78da","subject":"sq: fixups for db auto create tests","status":"ABANDONED","created":"2026-03-03 20:35:45.000000000","updated":"2026-03-13 14:11:17.000000000","total_comment_count":10,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"d8accb2c8953f84920525b54f26ebc7ac22937b9","_number":978816,"virtual_id_number":978816,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-03 21:26:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 21:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-03 21:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-06 13:49:53.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"c6073dd7ef8aa1b33e463153a4b498abed469512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-03 20:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f07a3676b9e764ee328b50242947ba4f7ae340d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:06:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e36c983889804bda88be6d4e7db1582b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12f431c2db0540c38bdd39d5b7143e68 : FAILURE in 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c03c035c99da9b0baf6f09865e26898ae7cd7c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:26:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8913e23d835c41e9a7921727b78c6537\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d44c03ea0b4c12a7a4abdd2c210454 : SUCCESS in 9m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bec4ffd7fa2d4492a34e5962367b5312 : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fe0d6a583e349669d02878e0ab2d64c : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/003f14592e1740a7afbfb9d565bed818 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/258cd1a75e3848b99750d0c8556d29a8 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9c5a9795c455404fa5f54141d91eda16 : FAILURE in 23m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2785c841177148ddac241090c6a61669 : FAILURE in 23m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8c24627983d47709acc21cc365a38e1 : FAILURE in 22m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5aa6c18c5ec54089abd615c1b5d9fc75 : FAILURE in 24m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9cb6e6860a7d4e9da3e84030837b6b96 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/361911f3dec445a787a6e56434c6059b : SUCCESS in 8m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/916d397fea574c38a0a108f61cf1fed5 : SUCCESS in 9m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dcaaafeb967f4c8c96af75cd5c1bb2a5 : SUCCESS in 50m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5380bec692184ad498f4f0b65f6e45a2 : SUCCESS in 28m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2c5b54f8979434a9b7b8d0a9bbf2e7e : FAILURE in 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4977c8b592af48cc9d3298de2e8c2527 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ba3460023524d1491fa3c118e9f29a7 : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"aa601d9cbf08372d2cb45e6fb310ed6401afd7ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-06 13:49:53.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c954986934a799854e37f465ce20571c236f4c15","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-06 15:47:27.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d8accb2c8953f84920525b54f26ebc7ac22937b9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-13 14:11:17.000000000","message":"Abandoned\n\npretty sure this is squashed 968740: timestamps: add NormalTimestamp for ShardRanges | https://review.opendev.org/c/openstack/swift/+/968740","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de59d440997920e64c764900fbe1828dc06e47aa","revisions":{"de59d440997920e64c764900fbe1828dc06e47aa":{"kind":"REWORK","_number":1,"created":"2026-03-03 20:35:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/16/978816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/978816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/978816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/978816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/978816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/978816/1"}}},"commit":{"parents":[{"commit":"9ab5474684f71ce8bf9683f6cef738d5865474ed","subject":"sq: ts FIX: autocreate db\u0027s with NormalTimestamps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ab5474684f71ce8bf9683f6cef738d5865474ed"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-03-03 20:32:42.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-03-03 20:35:00.000000000","tz":-360},"subject":"sq: fixups for db auto create tests","message":"sq: fixups for db auto create tests\n\n* cover the one untested status_changed_at i found in container.backend\n* fix my bunk ts_remote test in container.test_replicator\n* refactor auto-create tests to separate create objects/shards\n\nChange-Id: Ida910415f92aff721cda4c78830b1ce52b4e78da\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de59d440997920e64c764900fbe1828dc06e47aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de59d440997920e64c764900fbe1828dc06e47aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
