)]}'
{"id":"openstack%2Fswift~979045","triplet_id":"openstack%2Fswift~master~I766c8229f6ebb62391d2698b232d08828383c1b1","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-03-12 16:42:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I766c8229f6ebb62391d2698b232d08828383c1b1","subject":"tests: Fail loudly when kill_drive/revive_drive fails","status":"MERGED","created":"2026-03-05 18:50:08.000000000","updated":"2026-03-12 18:00:28.000000000","submitted":"2026-03-12 17:59:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979045","meta_rev_id":"9b79816af5813c8a9f47192478101b9518324fbe","_number":979045,"virtual_id_number":979045,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 17:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-12 16:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-12 16:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 19:28:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 19:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-11 17:55:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"86320ff8e3912c0ea14257c62f95f69d609711f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-05 18:50:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65569ae9fb1a03cd83e8aba46ea7552d7a1d6a53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 19:28:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c1d26d01b5346c5b5bdf124179ed451\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/736186aba4794b529d32e95983ee5a00 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c2d559e89a4ace27f9dded34b071b6cec5aaeb2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 19:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3345e6053b3f46bdaacc81d0f4ce74ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7451142b9d5d41c0a4cad8f57b6346e5 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/038e44b9a24e43579df21144a0b06ffa : SUCCESS in 1h 02m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30445b56f33845149f11dd9e08f2864c : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"835368edcff6adb580bdb5407115deb73df2c25e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-11 17:55:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"021199a5921e23b3e020c4221ab3295a2c5a53cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-12 16:42:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae7c43eb81deee372a2a05e7dbcc50970b7b9301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 16:42:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a53f5942a3aaff6edb3ea8f93668fe8495c32b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:59:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/218766b6e4f345ccb64f6447202f5a95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7adbe17744443bfbfa465a1c7a8606d : SUCCESS in 10m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfcff73f09764086953d488140f74a5f : SUCCESS in 25m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ac962dea56c04bf1a4a877925f677d68 : SUCCESS in 9m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/47e99a74e635413bb2cd6a6ae7732600 : SUCCESS in 29m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7523b1ac4e8844f8bf6976e57a841076 : SUCCESS in 20m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c000cf61aeed428ebe855a557c4581e3 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bd810a309aa148c98960ee4ef5afa0d2 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/63e89fce6b5e4f4ca934be70f59643f2 : SUCCESS in 9m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b5244c73e8c4fcebd2320950175ee35 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1284dc77682a477eb3db192c33200941 : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fce5bc8bfd36463996f5e974492b1061 : SUCCESS in 1h 01m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf43ade7631e48aa826b4091b034c6d5 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a65d23b42a6bddf25babc1f1a9bb133d54e7d7d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:59:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b79816af5813c8a9f47192478101b9518324fbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:00:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77db78c2b59143b58a59ddab9c50c712\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c054c4859645a8bb34809b5164605a : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8167816adc374a3d960f56064d8460e3 : FAILURE in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a24e682099a28c5280acdd9171bdeac96aadaf8","revisions":{"5a24e682099a28c5280acdd9171bdeac96aadaf8":{"kind":"REWORK","_number":1,"created":"2026-03-05 18:50:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/979045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/979045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/979045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/979045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/979045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/979045/1"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-05 18:49:58.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-05 18:49:58.000000000","tz":0},"subject":"tests: Fail loudly when kill_drive/revive_drive fails","message":"tests: Fail loudly when kill_drive/revive_drive fails\n\nChange-Id: I766c8229f6ebb62391d2698b232d08828383c1b1\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a24e682099a28c5280acdd9171bdeac96aadaf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a24e682099a28c5280acdd9171bdeac96aadaf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
