)]}'
{"id":"openstack%2Fswift~979576","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I51cf1f6503cfa6044e847030665d18f0c4e0eb46","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-07-23 12:20:30.000000000","reason":"\u003cGERRIT_ACCOUNT_38496\u003e replied on the change","reason_account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-21 07:44:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-21 08:33:10.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-28 12:52:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51cf1f6503cfa6044e847030665d18f0c4e0eb46","subject":"Add replacement for eventlet.spawn","status":"MERGED","created":"2026-03-09 12:26:52.000000000","updated":"2026-08-28 12:52:49.000000000","submitted":"2026-08-28 12:52:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979576-eventlet-removal","meta_rev_id":"9101e1fda830ff8cfcfdc72a818bcd957b412f03","_number":979576,"virtual_id_number":979576,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 12:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-21 05:41:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-21 07:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-08-28 11:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-08-28 11:44:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 14:26:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 16:09:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 14:45:17.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-06-25 06:32:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-07-13 20:19:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"3d5667635620a0d1add1548f227a6f6bb030dd59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 12:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55fdcf05fa690e72283b3aa2fd8f452824c1dfa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 13:53:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"887049b3f2821bc4a111fc6de7b35d58300c2259","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:26:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54dcbcf8f8ef44d689f155530c06749f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e90d3a35f4041818c0c31cf907cfb71 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d763bff66c705d380ad236dc8a229eef8ca74d5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 14:54:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4fc9856e5c98d4dcebced3a6491bb0875637a7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:26:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b55f44c97b4935a80927ff1e6e2418\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/254ef87eecdd4c6d9959bedbe4984085 : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":3},{"id":"3fcd73cd5f2368c153320a202d7a99a77c350aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:09:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2369c4a9baae48d1a6636687806361fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374aeabae7734275bffdd2fd38a8b63f : SUCCESS in 11m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d91ca8962e644ca4a842ae371132f0a0 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a490b5394dd461e88320c872e49b03a : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9374b257b1264f209c5ed644b9d67f68 : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8cd31844ca904784becb504e7f41504b : FAILURE in 5m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d6df63515564b078591b4f516319c5b : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/86ba3e8b689d47fcbf23118a92b4bdc9 : SUCCESS in 9m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/be7537a26b9b48eeb9af5b716903e317 : SUCCESS in 30m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4aabd3b2d59a409d8c36009f4071bd87 : SUCCESS in 32m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5bcd9acb12af460c8739ad868e9fe05a : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c493f8f5e9d1425c84210d9bbf8cb47f : SUCCESS in 8m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c615e521c7494f0e8bdc972f78eada9c : SUCCESS in 8m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc6cb6464acc4262a45b118b42e0e19a : SUCCESS in 5m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4588659bc2a64416978945ce9a3601cc : FAILURE in 22m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd05451db53d4127b08a1232e99ae0f9 : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1ddee4f95824d0585c1b9755edeccbf : SUCCESS in 1h 05m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a3d2bc3cc074ff384e862c93195c653 : SUCCESS in 28m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f688eda9ed141d99f9d5c259546c03e : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a20aec08359c46eda6fd2fb189c3d583 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8c649b132529036faa3dd026b1e8eb89d48f029c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-13 17:03:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"22e07688cde468c68023e7a12f520182444e9410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-13 17:13:09.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f0366ace5a1974615fc488e75d475f1176fc28ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 18:16:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f762bfc2c8dc445db1e8a5a52edd7c2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cd645c82f143ceabdab6118aadfe5a : SUCCESS in 11m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d57d561bec2e43d6b4a801b245c1e643 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61dc18457f3f4e52b51e748dad0dc58c : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9714700d8a0e4630834201c62296bd92 : SUCCESS in 19m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ce7a98923df402b84114e7bec950328 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/690eb6b346bf478a8e1ac495cac82d3c : SUCCESS in 21m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d43b6872f144442844df478110e9332 : SUCCESS in 19m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba46952b78b34c878f8244abd2b1e5e5 : SUCCESS in 31m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/127a5beb9a064276a399be3808c171a6 : SUCCESS in 21m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a115c219939046339adb5ab261b9d8d9 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cb2bb9bd70a249c7a90a67efdafc36fe : SUCCESS in 19m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd9bb3a22ab84a40a5538773d3690461 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b088ac88b5764285bc2affac60f3b221 : SUCCESS in 5m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1eed7156c137432092057de8f8a0c8b7 : FAILURE in 19m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/93bfdf7572d8433f97b3d3091dd0f916 : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a279dff7f7c94205889d898ae578353b : SUCCESS in 57m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6518623668d4dbaa523c350a65926f3 : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ade2a80656f4a9094d7bb73ffab77ad : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36aad48563784913990e848c6831d17c : SUCCESS in 21m 21s","accounts_in_message":[],"_revision_number":5},{"id":"a4bea5015e26c30aa044c5ec1eea36dbf3d408e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 18:36:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4201bc2acf427aa90bcb6b25ff0c51\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc54fcb6894548abaff8ca5470c61ebd : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":5},{"id":"65ef7337baca2a36ed0dc33ce6fa856e4d4ffded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 08:59:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d04843035121e6a7144275cd44ab750d3af83ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 09:33:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea08c4614df474397b3e127e9b5f305\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9092b033bfa64e078ecdabd5c9b2662d : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":6},{"id":"0c778923480619d79ec801961a3f494c9eb6583c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 10:04:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e284e779ebb74ba6a17f1a5d31313c80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a33dfaa51d4e3c9808a8e5918b2300 : SUCCESS in 9m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25e0cc0d712a4ad6b6ab53c1fd0d010e : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/194258ae98c540d89eedddde0d2f0a02 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bb2a1a14f1d4be38f41f8b9129bc5fd : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8813cb5671e043bb9020558bba16d2da : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5bddfa565bdb4459b586d074f54fec88 : SUCCESS in 12m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/21fd10647b1643c2a7ca76738a7070b6 : SUCCESS in 21m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/90a5e648797246eb9c1a14d624374786 : SUCCESS in 14m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/41ebce1de5d1413cb8a51f2613ef89dd : SUCCESS in 28m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/072ffbfad32643f4a784650f9116edb0 : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/08fae9afb6794a9dae7a6ce71dd5dd10 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e8dcdcf07e10440d8f0801473855a644 : SUCCESS in 9m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/740eddf0985b4b9696cc25280e0f22b3 : SUCCESS in 8m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ec21eddd4604de3a2f38c439c5250bd : SUCCESS in 10m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d75ba52255c40b2956914a390b6ac7a : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5561a6ac06449148bcff187a47625d6 : SUCCESS in 56m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea17cbe671d64b8bb43d667c506f3562 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c6704287ff47899e86515c8b918a77 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/500ca55073c0467294fa519575caf36f : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":6},{"id":"9cbe13585f4a0321de8d87019ff7cebb0fcecdda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-12 13:11:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"80deced889680601a784c23d0de79f0ac87085b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:58:04.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0012bbdf12304613b1d1d848eefae3f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2be0d56a446c447e8ed500629b2a1b86 : FAILURE in 44m 51s","accounts_in_message":[],"_revision_number":7},{"id":"08ad771f860f9fe59b0a514162911f51012ac8bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:27:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19965081262c46c7bedd59eae014e420\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c451dd76b3c451e9db8b23e380f0376 : SUCCESS in 10m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/934b1bac01784aafa768271808a39dea : SUCCESS in 11m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93fed8b3b6bf4f66915a58a1489aa1e7 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/052c589532744e6fbab44ad5272289a7 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/621cec93b1fd4575b7811dcceabb6c87 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/17f9ca7cdf18460092849334f6b2b6fe : FAILURE in 22m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97352bc165594c5984a1693c51de6bb8 : FAILURE in 22m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b661a0067d774a539ac05a22f8badbee : FAILURE in 28m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/92cc4f8b38b24aa28df75a5ad9a59792 : FAILURE in 27m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e763436057aa42219117303f025db174 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b69d98bcd4d4013b4304057750ae4a6 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/00d96b2253d64225866a82ebb91fd111 : SUCCESS in 10m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/719cea18191a4628bda679550178d24b : SUCCESS in 7m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7067aaac3b43466ea1fad9d5d35ac4dc : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e41c14c2edd449c3a75572accdd0eafe : SUCCESS in 11m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5829d168838a4b6c8c58bd519ac28597 : SUCCESS in 1h 00m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6417b712fa2248e6ac8ee6c7bd4e10b9 : FAILURE in 36m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab668b8fd5b46ffa84c89570c960876 : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eacd53ecc4d24abb94c3753a9d7b3e5a : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":7},{"id":"038c87ca31e28d3b2df343d8e229ab5ce5d524bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-12 15:42:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0ca3fdf90eaa0823b44b72116171b20a33056b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 16:45:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b84f1e20e9824463b4b3ab709b4dc440\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4861e2dab254370a40d6f2a451a95fb : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/497c9a1be67f4933920e67d9c7a92e57 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b9eb0f962a8404db35152f8705e6963 : SUCCESS in 12m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2b38ecf2e244fdb898db32fcdca2d62 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c33f6ec2eae94d7898e702d992368265 : SUCCESS in 22m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d67b6fe481e0441da27a0171c243939e : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6fb31ef9528e43b0a9d991483fb1eba8 : SUCCESS in 11m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a03f839ce50c468e804a60e502f7d74e : SUCCESS in 17m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/911c0821d4864c3faeac7df9dd68de40 : SUCCESS in 32m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/52fc32616b7547c2a8bd8f9499fc84c6 : SUCCESS in 17m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6124312452624e96b8c38757c5d1eba8 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ea5ae1bd7cab47408812fd0fed195724 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/40dc12783678457f8ed586921c8b6bbd : SUCCESS in 18m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdc8407827a74447ad917b7f907c4777 : SUCCESS in 4m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8328d15080264cf2946686804fae5de7 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a58c0efb30934e4880bde0efed9dd875 : SUCCESS in 6m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7822c2b23f4a490d92b0cdfe1a6e3425 : SUCCESS in 1h 00m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c8f2c8686046bdaf53c39bc8bad8b9 : SUCCESS in 31m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d30b72f1ecab43b4bf795811fcc882de : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/978480b381414411b3470df6e5a5ef80 : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":8},{"id":"ffcd9411a546c935a97d099f46e75cda83ed0903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 17:11:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87771be10fc4c95962d5ffa363c210f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56d07517a15a4af29168469d97694aeb : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":8},{"id":"9f47231f2e219e387af115dc72da737cff6a2109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"00a68bf8e2a5bc9c453d9aa5b4a79ee0d0734267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 17:58:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/095b12d7fbab4823a05a7ca729ac72a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e0609654ac04353860d145b4795c702 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbcf546bcda2441085af357e9e69d75d : SUCCESS in 9m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42f2871d0b064f62be3b75a10163a438 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f897d640b5164cf78aca4cc8c8adddb0 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1dfd04b27c643e2a796ec1b7013ac0e : SUCCESS in 22m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db5a526c374a43feb1a56a6604607f20 : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba5158a116a54babb6364e40c88ff1fe : FAILURE in 26m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/100b36d2aaee4e3298cbfc85dffd7a66 : FAILURE in 25m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/98228ab56e2346d3bebca46018c9e6fe : FAILURE in 11m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3652dfa491554e4192f209e2c7b42e16 : FAILURE in 23m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4a153112372f44599a7561937e4405f7 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/974bc430effa41a8b31f1eb6cc232b0b : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d540343dd3984303864b69b539b532b3 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dde9c18ade3146fca18e7fd4866b4fa2 : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0359c137cf6462a8edf4a629a2a89c9 : SUCCESS in 13m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b0ba007715a444080a3041243b2370a : SUCCESS in 12m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a04e3037ffec4c6298f540273b313441 : SUCCESS in 39m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6bef6020261430f96ca88c99ddf200b : FAILURE in 35m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618be3005f404e9cb7caa850cc53dd40 : FAILURE in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/405e2108b42f421ebb4fd697b164415b : SUCCESS in 19m 08s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_updater.py","accounts_in_message":[],"_revision_number":9},{"id":"2bddb59ca82a46faf5f7c13153ac1a6d86d21af9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:22:10.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959bba0d97564de69779d98bd7359849\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/456ee706ebac43178dbe26c1de115d8a : FAILURE in 1h 06m 08s","accounts_in_message":[],"_revision_number":9},{"id":"3e19aeec5f2abd329cf433b51bfe6bb2856e539f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"838792fea1dfb0f7b4d35d6403ab8b8cb755eb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:27:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f11f8339c3f245f9958c3964d0fbdefe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5473672ea64629a214ab223d76b5b7 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90d6a4168eeb44dfbe16700a6870fd43 : SUCCESS in 8m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4da829b288be45938c70f6e8ce64ed51 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a3aa958143f47a7bdf679112b2848b7 : SUCCESS in 24m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8959094e2674726bab1563e60f492e0 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6a03ff22027a4104b5f7f2f825366545 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f05ae8c1398047bebe28a00afd3cc727 : SUCCESS in 26m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4d859f9929db4e1e8f351bf84638e341 : SUCCESS in 22m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1502ec23317748239f284c8d69b3ba9c : SUCCESS in 30m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d439951d01a44be6b51972b18f74e257 : SUCCESS in 21m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f5d9746c099e4d29ae7c9a8533e24916 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8fdf77781bcb4de58faf4810c8937f0b : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b0dcc9d678a0462cb3af26b153c9e2e1 : SUCCESS in 10m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eecb7e64d60f48c283f2fba7cb2a4b80 : SUCCESS in 4m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d2eddbca7dfb4ea6bcae8635361a6ed2 : FAILURE in 19m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7a7fc57c70e430faa099521c678d1a7 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d75f1b51a2714c9a9e7bb615fcca4a0a : SUCCESS in 51m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea1bf5d0d31a4555bba39996c55aa96b : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5aa05955694b2fa4f89525daa90eed : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45789b8535264f6fa21099eff463ddbf : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":10},{"id":"c39e78a5ff808ff8373d24dc13ed2b9a41237a1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:49:36.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca63fda14aa1408baf1e0179ed8a31a6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24f61a70d9294ed197cc476af8ca1912 : FAILURE in 1h 10m 17s","accounts_in_message":[],"_revision_number":10},{"id":"4b1b218671938f44ef83929bf6d90c6ed8675836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"b3bbc9650c4e4a54b4c2c12caf40fcf5913ccf89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:14:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/841948e07d1845909f100f950ee864a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b84bc593932436cb4126bf805e803e3 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1294fa4e33ff42db999a7cc60f59711e : SUCCESS in 10m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e73d94e6033d4fcb89d65301bd1877d9 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/226b3a3d273a4924aa0ffda0cd7f4d61 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee35803ca5dd4ffc9fc62013fe7a7dfa : SUCCESS in 30m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/454a6229be91486a985355a373948ab3 : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/248ca0ad5b5d4a6782e14fd886153cbb : FAILURE in 26m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cad99c253ff144f29efd1c05951502b4 : FAILURE in 18m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/332e8c127e164065afb63a77f3cf365a : FAILURE in 31m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cf07afe363b6433a8e43975f47e5c4d1 : FAILURE in 22m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ae278cf4ce4d429facff309fc688c1f6 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c49e76d27d934bb1a82c05eacdb14465 : FAILURE in 19m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad1f7fb1224a4bfeb54a0a0a25394e83 : SUCCESS in 11m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7dfc7acb8cf948c6acf61448f1d5999a : SUCCESS in 15m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73aa3cccd5524feb84b1e7188be9aa7c : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/36f4c512b7394e059416d75a61fa4d14 : FAILURE in 10m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc58673c40ce4efe95ab55efddeccc6b : SUCCESS in 42m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f68695b85b904e42a4087cbbbcc4247c : FAILURE in 24m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc52267c564340bb85455324ad5529d4 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b1aa7316fc543939facc84c193beff8 : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":11},{"id":"9723561337a1932eabb552d6e75532796b27de43","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-11 14:45:17.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cbe1d19619c653323287e2602f59bf3720d6551c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"ef8f017ca04b7f4204546f9da2c4339d4ce976b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:48:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9248805a7fcd48da976d9790811e8daf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd33dd0c30e423c81b3e12993ed2f5b : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/953deb12af024e078f5cb8bde51166ac : SUCCESS in 9m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4e504f891cc47759277090d2013b13b : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39915f7b3352442bb3ecf2bf74dc5180 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05d951a144b3446cb562b850ed08ec3a : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a390135911cc41c9b6933b1022a07456 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e40d945a0f24e9abd59d349bee2cd4b : SUCCESS in 26m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/06e62555eee5429c9cd13abde9a3605a : SUCCESS in 22m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/44bc09ebf60d4707a6ae19501a41d661 : SUCCESS in 29m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ce5a5028ccb04f208aa15971efd30b17 : SUCCESS in 21m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/215dc71138714b04a764f49cc2b1dfc9 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/95faeadf2d3942e7a9de80c9ba649253 : SUCCESS in 12m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/71822ba426fa4142ab3116706349649e : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/999c058070a2467083d35a418c626fc3 : SUCCESS in 8m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/79ea21e56f3b472bbc3353cc6cbac564 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aef0fee3624e4bb39e429d79c2ec17c3 : SUCCESS in 4m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/637932ec297648df9999b733f5c6a937 : SUCCESS in 40m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/409ca603c90a4b8e96c170ca84ae05b3 : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f3d099f0a64d2c90ea852820693f68 : SUCCESS in 3m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/335dea9bb4794a69be387893cf251ff8 : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":12},{"id":"c7299c24087a30ad34a4761cb601b905da778600","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:38:47.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a9fd1b0d07a4f8fa50a38a104c45b61\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7114e891b497438791ce7617043cb01f : FAILURE in 29m 03s","accounts_in_message":[],"_revision_number":12},{"id":"134f23fbae65622be7c3a3c68b43ac2d1071309a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-25 06:32:44.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"ffb7fd0d9509c96a9646c6e6b2b8df65d06c0fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"ca68f959ddf6b5132a4c40310dc59ee8054e0c3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 08:53:34.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f109f5ae27440459a98c0c579e65185\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/412d05a44c364c0fb3d48c3d44010bdf : FAILURE in 40m 09s","accounts_in_message":[],"_revision_number":13},{"id":"52ecb097ff40b0a29ef5a5a7cbb8910deca68943","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-06 09:36:35.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"efa6c813f40648a860b17e6b3b6757612bbd979c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:29:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8748cc5588143659730bb959ca35c6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949bcf31af4e458ababc6c67290eb731 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b827e37525e74168aa1a7f7463327ce5 : SUCCESS in 9m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/822ac7761e844c6883639c3e166dd752 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/135e305b03eb486493d17f2aad0b68f9 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eab07155e6ac4ad7bc056fe6c22195d3 : SUCCESS in 22m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73772023476c4daa8c34ef45345cd40b : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96e8570ecc404d21b09821068d70ee4c : SUCCESS in 10m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/54637340d4e942d3bf703ea57590d49c : SUCCESS in 10m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/66fa63cef0f14782a5e495af0fee2116 : SUCCESS in 30m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d348365fe3b34a4f80bbb66f27375548 : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ed5b2616738143cb92c7323708864e22 : SUCCESS in 8m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff567a9ef0ec484f82e9e31b59d19014 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f985e5d43f0043e9b5c12de60e01d44a : SUCCESS in 17m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24263538280c473788c3a1c92a9a1cfc : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c7a51b40e7c4bb8bc7b07c48987e294 : FAILURE in 18m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/88ca95117c3f44d2bbe404cd3518647e : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bc09711290b44e9a94771fea0880f6f : SUCCESS in 59m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7e92d8d643d47cc81b3000b8d55a18c : SUCCESS in 29m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2256b2ebcd495da0fa698a7abfa2d8 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48ebd6d7a9f543c58e6f9b42cf420b49 : SUCCESS in 17m 42s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"c6c18bbd347c051c8a0f9af7e8cf2de9a193b1b7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-07 12:26:26.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c0e3fc58b3e23cd1d55ce9d75db4a892c0b40765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 13:29:48.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"3c1c52ce683812120b78642357fcffdb1aa0c8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 14:15:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19cf123e1b24ba5b31cc934529bf566\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ebf0baf465f4ea1b728c7d098da3ddf : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/483aad45c874445dbd9a9751b2ab4842 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a21554d1bad24287b0b5b169280ab313 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c31ed5a75b24f089b3a3062c9573082 : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70300160694a4d36bf0694e2d45cf9fa : SUCCESS in 22m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/151fa53d1a8a45dfb1f9ca77439aa1a7 : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e496f98475df42f9b92188a940435faa : SUCCESS in 12m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/310cda64e47c41d1803acd455c0064bd : SUCCESS in 14m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/656f8a15456644b1b3e4dd53db077587 : SUCCESS in 21m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0a9a646e6016405a87c1102221eac550 : SUCCESS in 20m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d40c61116a1d42b2a002172a500bf507 : SUCCESS in 10m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/76dc8d1722b34295b67e114fd20af8c8 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7231c75bbbeb444b8561ddfe26287d1a : SUCCESS in 12m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78d4a89667834e908d78319bb585d17e : RETRY_LIMIT in 5m 40s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73c1e76ff55e4a17b2547dcc01d1e955 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/64bccd20bcc84e07b31354d6423addc8 : SUCCESS in 8m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e171c50ee3843dc801555bc8ed341b8 : SUCCESS in 38m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adfc1a89e8b647b8a4b3b7a5393ed1c1 : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3721996772f421caab967dd30f2f208 : SUCCESS in 3m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b17b501be0724f108e2a40dfea09737c : SUCCESS in 16m 38s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"5d7b37e193327bb2ada8d9c50341e6ea6223f4b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 22:31:19.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f14fe5fc59dd4bb68483f754faeb1ec8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5c45283d11e448bbf19672e9b0a0fa3 : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":14},{"id":"3ef8891797b8aeeab4430b5a90319a582e405679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"1e55ae25ff2d12a35feda46c761d33194f7c889f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:29:32.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51350b1547b4007b16e096a67ce2e3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22eeb2426c74480aaad287528c1551a3 : FAILURE in 1h 07m 10s","accounts_in_message":[],"_revision_number":15},{"id":"783760d4f17d3eb28a9491311e44d81a9865e5a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:42:20.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0457e4003e2747dd8724c3dcf16d8229\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc2ba15560894fc7b184cf831056c96f : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15966e3e3fd6404a824e4bb8f17fcefc : SUCCESS in 9m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cef066c124647fcb8d5850c902a1806 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df8f28568dc84d3f981e2a59fbe8e801 : SUCCESS in 23m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c63d870601f4577b02cf0fbabb98034 : SUCCESS in 22m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5feb7a59718b4ea1a1afd2964e4155ba : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ef151299a7402c972773e1b94c24e9 : SUCCESS in 20m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b5aa8ad5146c42f8b2d3e37fb59a0d85 : SUCCESS in 21m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c42e6838b0ea4452944fc35ddae4615f : SUCCESS in 30m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/98a5a8c52ed24bc09338988dc9ed1206 : SUCCESS in 22m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/805840a39e2e4a939a80472145e1a0b1 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0d6b8b55bbb448be8053cd6b7d67ed79 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed4c854f0fd34549b227c21f678ea671 : SUCCESS in 10m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61d07e85acd640e69823f370018211f7 : RETRY_LIMIT in 3m 42s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa6ad00d81e64b2ebe27170f825d5aa9 : FAILURE in 23m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea5a4a4a55a7482ab91afe774a1e7648 : SUCCESS in 5m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e7c102458db4766a87429d475828295 : FAILURE in 1h 07m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1de73f5685e843cbada971edbed158e3 : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98dccab0933d4489a46b98f3527b15e6 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98bb330d6539441287e41d4b4f55f910 : SUCCESS in 16m 21s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"836fcb9e76bde62dfc25c795a868dfead644089a","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 09:16:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5229a58ab553ee15bd86eabc63604db355da3ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 10:24:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc587de5bfde43c4a31254f91d811ebd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741d08adc55846cc813cea4ef043a43d : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94881636f59d430f9717886243385e2d : SUCCESS in 4m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aca9a29e2b3a49e7ac129ac925deacc7 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8cd85a3af1574276923f8f5bc1e815d2 : SUCCESS in 11m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aae97edb67fd4d45bbd7090c15ed9a68 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fa61ccd9b594f76912087c3acb02e0d : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a7c562b2e6e84056936d356e2f010058 : SUCCESS in 30m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fb5bfb39db484fdf990e7fa5d90a67c1 : SUCCESS in 14m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0a9e24b6767648ec8d52513b6a6e40ff : SUCCESS in 30m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a6f070f2d84c48708f087103651356a6 : SUCCESS in 9m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd97c4029b964836860cb0bfb488af04 : SUCCESS in 8m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/610ecf9e87354f82b3952edf83f0179d : SUCCESS in 12m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4258a8371202477a937d407b13d0e916 : SUCCESS in 12m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e8cde4b60de4d09ba9245c92c8bcc9f : RETRY_LIMIT in 5m 04s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb8d5a902e504e87bbd5a7dfbb5f9fd8 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecf4ac6df7c7452f9b85fb5e3ad5e336 : SUCCESS in 9m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a059d9ef62b49b394cfc158861e01ca : SUCCESS in 1h 02m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a85cc5f7815e419d986b10fb72787555 : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c4770eab1c4e1d9212452231ed0981 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c4f316ef0b84aeface39e2e00de3884 : SUCCESS in 18m 25s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"f457b783344aae971b4decc3ea263e79ff6fdc20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"fb3d1a0bb8adb441a5db54addd99a2214eb6ccd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 17: New patch set was added with same tree, parent tree, and commit message as Patch Set 16.","accounts_in_message":[],"_revision_number":17},{"id":"3c6c10f1ea0224eab593236859fa1e8e14f6e327","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:47:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea1cc80eb0994985afa2ce7bb9845896\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1079863edeae4fd9a3e404ce869de77b : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":17},{"id":"7e2bb5948b27b8261a8673dbfdd54e0c567c7dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:50:11.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718ce521109946ac96547e84c82ba157\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b544a7f368ca493c86338a64060f05c8 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03c3beb41a54450db9d580a3f0a5e0d1 : SUCCESS in 7m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4a089d14aa644da87f57454973b30d1 : SUCCESS in 12m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5431a5415aa49a8a40039b06068a51f : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffe4f35f10d94e0c801e9a71b2f18495 : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fec75a45aff541af8456d73f07ad141b : SUCCESS in 11m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fdf5077787f54e989e8c368427064587 : SUCCESS in 28m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/af6ee3365b6f4a83a6ae39da3d68ccf7 : SUCCESS in 23m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b56c01d6d89c48c995fca591cb097d7e : SUCCESS in 30m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/796ab20a088c4740ae005777fc471c63 : SUCCESS in 21m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ae7e14cce75045779cc8b0172fb8d4ad : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b55342c6cd9b4d65b3f3acdf13af9f96 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/764666aea2ed445990ccf7d072465627 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/520ca97ed1074329b5ae5fea40d52b12 : RETRY_LIMIT in 4m 43s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6de89d587554baeb4614538e74886f0 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/33cb6d26fe2243c4a02db787f1e4cdda : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/906f7f8a82894ac9944426fe97f8fb9d : SUCCESS in 1h 02m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb19c639c9bd4bd89c659d80cd63af78 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a113ff18a6d84768af2460b0dd7dba06 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d0b7c7499ea4adca0e5501ad363104a : SUCCESS in 25m 08s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"5a3eae646c802ab70afac934dcf756a0eddffd8e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-13 20:19:16.000000000","message":"Patch Set 17:\n\n(11 comments)","accounts_in_message":[],"_revision_number":17},{"id":"880fd27d5cd321c3f5deae3c852633e2bdec5e2e","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-21 10:55:13.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"e748497bd8d370a732636babebb48f2709dd49e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-22 15:09:59.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"97e5dc5690d8b376efeccac974e03e3fd4ea1679","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-22 15:10:24.000000000","message":"Patch Set 17:\n\n(5 comments)","accounts_in_message":[],"_revision_number":17},{"id":"72da4bd9856f30bbf9701ab0928e83e83f9d3cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:20:17.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43bb30fe25c042109899beda71b19e25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f5aa7439864aadb7cf354003f83338 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/405a3372ca594141827d8456d2f9c45e : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c84e1d2483e1477b866861443da3156f : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99f9ad16e0424cb99325d4aed5282fbd : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c1f58eac2ca4d7d910fbf0f78a50b70 : SUCCESS in 19m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f450b62a70874dee99b987f52a2e8472 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0025664a981e43efb8f91fb0057a386c : SUCCESS in 26m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/71407b05e39946d7a6549cf0f9370c44 : SUCCESS in 23m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fc9033e31fa04d739dd0af05f250b89e : SUCCESS in 29m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/52c91ba79add488eb286c864d0032115 : SUCCESS in 22m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/212d12c3325f4c65af9de618242a7e8a : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ef1ee2cd50b1467ba9eede688f5c5748 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/abf6c67bfd2841ea84a57c039561cb5c : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0bc07d39025c43d5b46ea5f6bcdf8332 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4387530b6d044709ca86b4a977d0de1 : FAILURE in 7m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b8418eae7f6d412b8069e667eefef42a : SUCCESS in 9m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e6082989db24fec8ea6db539c2ee60c : SUCCESS in 59m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24f55470f1da4d9c9e7e2b7fec2e4bbe : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ccd045a81d54c2b886bdd5e09e3ec7b : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74db3e785111457aa046cbcde73c2bdd : SUCCESS in 19m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"bcdc926994b0982fbc856f5128b92cdc4b084796","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 17:21:36.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ed5b246bc9246d399c29ae582feb816\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2520f9a7dc9d43fcad7b9227e4768b2a : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":18},{"id":"06b6e207f9ee8e2dca26bebad8146149e86b68a6","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-23 12:20:30.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"f290e35bb340880466ba0f206db8de4965d2393e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 05:41:20.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"87bac5f1ff930bc1d31bc98aba895f6bd039c2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 06:34:16.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e1f61fe75cb4ab89d37484de66551b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f55b159370c4d7798f6985a41f72528 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1391565190644235ab9ce073f0a136c5 : SUCCESS in 3m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14f3997dc8774cfbb516db77a4eb2f68 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/286d435c34004fc3b1a8bc79fc6827e9 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/332723df5a27442e89c3a134281c37cb : SUCCESS in 11m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a4272ca94034415b9c3d6b6bebcd50a : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f2a9c5397d384e21b810316b09f1474d : SUCCESS in 26m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2740d3e356e743c180749bbbe9efe110 : SUCCESS in 23m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3c48c6571d2a4717a8d21ef4f18f010d : SUCCESS in 31m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/61f4be87d4444d44ab945f8b92d97b6b : SUCCESS in 13m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c1f4e5a83b04416093809ef860fd08db : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e4d3846a51a140f4a893833b43dbb2a0 : SUCCESS in 15m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/15021255e321461e9b4dfd98d9c77adf : SUCCESS in 15m 32s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3882a574381241ab9836f00ef5cc7d9a : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10e8ed5e85134e5ab813ac2b7dbb79bd : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/adb95d0ee1e24d2e9e5a26a8feb1aa1f : FAILURE in 7m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9402e20e87304c94a5ad099d8551caee : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4505e1b7682f4b18b2267f09defcec1a : SUCCESS in 42m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eea97a29e2a43f0aa0e93e4efdf243a : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ec7a678aa1493394d3622cae86f3a5 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5c667fed6a7416f8e137d1f4f8c0444 : SUCCESS in 16m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":19},{"id":"e965d12d2593c0c36135bf7e6c1e0147c2b7a9bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:20:01.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81447fa9beb14c4bb3578827a4a262ea\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61895066dd1240328e19270a3de76591 : SUCCESS in 1h 18m 56s","accounts_in_message":[],"_revision_number":19},{"id":"513b28844d36fc3b2b5e654b3bf561c3b6902f2d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-21 07:44:02.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"d3eb66d8214c4101e8aa585ba026485da8df646e","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 08:33:10.000000000","message":"Patch Set 19:\n\n(5 comments)","accounts_in_message":[],"_revision_number":19},{"id":"ac7918371a76056f74c4fd6e454097ebc467365f","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 11:44:06.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1b09828f8c5ab21f32117ca0c8c1b8f59784e09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 11:44:22.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"77a70b96b872961b4cd29caa69381586befe7db4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:52:49.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbccf67449cc457497944c0fc4e5beb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e28f761cea044a9bcef2d0cf9f5088a : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b92c368c9aed4b988078ef8b0f51a6ab : SUCCESS in 4m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9931499cbd82497d89a3bb0eb247619b : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e91bd30fa3544f590563642c7c4937e : SUCCESS in 26m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/13b304fe40c743eabf327c31f027d374 : SUCCESS in 22m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/21a1c2d741114685a07960a8e9c7a159 : SUCCESS in 12m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4316cf00192e4c0c8b72f6a9dcca6620 : SUCCESS in 23m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/462ac2890d86490598b38f77e30a86a3 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c71f3ce2ad604be882b785c7b200ba9c : SUCCESS in 9m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9e8a650f4cfb4d37a2b09b3429f1e8c6 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a9dec5d5601433abe02ab5826c82fcf : SUCCESS in 5m 52s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6700291958ab4ff28cd577dc38ae8083 : SUCCESS in 9m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/436abacdcfaf47a0bad3a847156f0e5f : SUCCESS in 1h 03m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d2eeef941434e6cb80601cef0672dcd : SUCCESS in 28m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6f70159d2c4e53838d20f37c85fa8c : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f3cd56fe34c4e2ca46095418ac2d86c : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":19},{"id":"9101e1fda830ff8cfcfdc72a818bcd957b412f03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:52:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"75c5545e51c2b2f6457a997e4c44fdd36208be9b","revisions":{"27e125e5367e616eb8ca93cf927860f0521301ab":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:26:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/1"}}},"commit":{"parents":[{"commit":"b0ffc138b17d228afd78efd83103d38d40073fd8","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0ffc138b17d228afd78efd83103d38d40073fd8"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 10:43:46.000000000","tz":60},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nThis uses ctypes.pythonapi.PyThreadState_SetAsyncExc to be able to stop\na running thread using kill().\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e125e5367e616eb8ca93cf927860f0521301ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e125e5367e616eb8ca93cf927860f0521301ab"}]},"branch":"refs/heads/feature/threaded"},"58ca9af7393d2c48681d745c2e4d2104fd64ae24":{"kind":"REWORK","_number":2,"created":"2026-03-09 13:53:14.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/2"}}},"commit":{"parents":[{"commit":"b0ffc138b17d228afd78efd83103d38d40073fd8","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0ffc138b17d228afd78efd83103d38d40073fd8"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 13:42:23.000000000","tz":60},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nThis uses ctypes.pythonapi.PyThreadState_SetAsyncExc to be able to stop\na running thread using kill().\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58ca9af7393d2c48681d745c2e4d2104fd64ae24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58ca9af7393d2c48681d745c2e4d2104fd64ae24"}]},"branch":"refs/heads/feature/threaded"},"4ba4e9e0f02f014183251f3ac9cf155e0eeb2d2c":{"kind":"REWORK","_number":3,"created":"2026-03-09 14:54:01.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/3"}}},"commit":{"parents":[{"commit":"32a3dbb3a7dd4b4c0d3f60b4b46fe652e62f6381","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32a3dbb3a7dd4b4c0d3f60b4b46fe652e62f6381"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 14:06:24.000000000","tz":60},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nThis uses ctypes.pythonapi.PyThreadState_SetAsyncExc to be able to stop\na running thread using kill().\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ba4e9e0f02f014183251f3ac9cf155e0eeb2d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ba4e9e0f02f014183251f3ac9cf155e0eeb2d2c"}]},"branch":"refs/heads/feature/threaded"},"80abdb8beb1cc5cabbc7cedae2e7165c92647106":{"kind":"REWORK","_number":4,"created":"2026-03-13 17:03:59.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/4"}}},"commit":{"parents":[{"commit":"e27b1feba85f96b672249bdc68e45785c1c1b821","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e27b1feba85f96b672249bdc68e45785c1c1b821"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 16:10:27.000000000","tz":60},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nThis uses ctypes.pythonapi.PyThreadState_SetAsyncExc to be able to stop\na running thread using kill().\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80abdb8beb1cc5cabbc7cedae2e7165c92647106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80abdb8beb1cc5cabbc7cedae2e7165c92647106"}]},"branch":"refs/heads/feature/threaded"},"bd85479aa4cc9b4de82587e61f48b85fdc6c8c29":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-03-13 17:13:09.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/5"}}},"commit":{"parents":[{"commit":"e27b1feba85f96b672249bdc68e45785c1c1b821","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e27b1feba85f96b672249bdc68e45785c1c1b821"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 17:12:28.000000000","tz":60},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd85479aa4cc9b4de82587e61f48b85fdc6c8c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd85479aa4cc9b4de82587e61f48b85fdc6c8c29"}]},"branch":"refs/heads/feature/threaded"},"fb13cf3f7fc0bfc2ccdeccc09f7cf95a80ba4806":{"kind":"REWORK","_number":6,"created":"2026-03-23 08:59:05.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/6"}}},"commit":{"parents":[{"commit":"ac076f22e5197d3c9c0852bf1aa638d3c876cd48","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac076f22e5197d3c9c0852bf1aa638d3c876cd48"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 08:40:21.000000000","tz":60},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb13cf3f7fc0bfc2ccdeccc09f7cf95a80ba4806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb13cf3f7fc0bfc2ccdeccc09f7cf95a80ba4806"}]},"branch":"refs/heads/feature/threaded"},"d8a3a52100d6b1753e1b5be93b19f758ba534823":{"kind":"REWORK","_number":7,"created":"2026-05-12 13:11:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/7"}}},"commit":{"parents":[{"commit":"8419dd4c1a0ce3c627106e7c862830a546018e5b","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8419dd4c1a0ce3c627106e7c862830a546018e5b"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-12 12:46:30.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8a3a52100d6b1753e1b5be93b19f758ba534823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8a3a52100d6b1753e1b5be93b19f758ba534823"}]},"branch":"refs/heads/feature/threaded"},"22be55b8fee7942637464ac882267faefa3b6ace":{"kind":"REWORK","_number":8,"created":"2026-05-12 15:42:16.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/8"}}},"commit":{"parents":[{"commit":"4f4236c4381a95f82ccce07a19ca0caedb632907","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f4236c4381a95f82ccce07a19ca0caedb632907"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-12 15:39:39.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22be55b8fee7942637464ac882267faefa3b6ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22be55b8fee7942637464ac882267faefa3b6ace"}]},"branch":"refs/heads/feature/threaded"},"ce6e0c09e75478fab4916e2ff27ecbc0515a03ef":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/9"}}},"commit":{"parents":[{"commit":"cba892207f68bdeff8404b12eef081aef3eb8e6a","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cba892207f68bdeff8404b12eef081aef3eb8e6a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce6e0c09e75478fab4916e2ff27ecbc0515a03ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce6e0c09e75478fab4916e2ff27ecbc0515a03ef"}]},"branch":"refs/heads/feature/threaded"},"c640fe6dda8e236b80891dc573d33e00c98b3853":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/10"}}},"commit":{"parents":[{"commit":"8bc7d98ffe5af1fa67cdccd31d965b8600242a72","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bc7d98ffe5af1fa67cdccd31d965b8600242a72"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c640fe6dda8e236b80891dc573d33e00c98b3853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c640fe6dda8e236b80891dc573d33e00c98b3853"}]},"branch":"refs/heads/feature/threaded"},"3c4ff69310d0b5d73c7444c1f8746a28e9e82b47":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/11"}}},"commit":{"parents":[{"commit":"46f25aa6238d76178c07075a756bdd9c630cdd51","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46f25aa6238d76178c07075a756bdd9c630cdd51"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 12:37:53.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ff69310d0b5d73c7444c1f8746a28e9e82b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c4ff69310d0b5d73c7444c1f8746a28e9e82b47"}]},"branch":"refs/heads/feature/threaded"},"0b86f990f6e0cfc19fc27141c1c3986f150ffcfd":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/12"}}},"commit":{"parents":[{"commit":"f394d21a957c0feeea7110670d7633279c4b5e30","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f394d21a957c0feeea7110670d7633279c4b5e30"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 14:41:16.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b86f990f6e0cfc19fc27141c1c3986f150ffcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b86f990f6e0cfc19fc27141c1c3986f150ffcfd"}]},"branch":"refs/heads/feature/threaded"},"9cb1e0b420928734630d4716ff35cae18debea6e":{"kind":"REWORK","_number":13,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/13"}}},"commit":{"parents":[{"commit":"069df8e8301947cdc2a3c21794996dc4c6d6c396","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/069df8e8301947cdc2a3c21794996dc4c6d6c396"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cb1e0b420928734630d4716ff35cae18debea6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cb1e0b420928734630d4716ff35cae18debea6e"}]},"branch":"refs/heads/feature/threaded"},"1763a319364f1f5c8c7bc59f09f4427ba84f7350":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-07 13:29:48.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/14"}}},"commit":{"parents":[{"commit":"91f76dc63851225a66cd46b5b0a917d426246eaf","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91f76dc63851225a66cd46b5b0a917d426246eaf"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based wrapper is used. eventlet\u0027s spawn is wrapped to support\nthe timeout kwarg.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1763a319364f1f5c8c7bc59f09f4427ba84f7350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1763a319364f1f5c8c7bc59f09f4427ba84f7350"}]},"branch":"refs/heads/feature/threaded"},"2aad24f54112b193b45336834a7e0cb045e945ee":{"kind":"REWORK","_number":15,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/15"}}},"commit":{"parents":[{"commit":"ca6e3c53c4c48904f4587b869469d0e94195cc05","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca6e3c53c4c48904f4587b869469d0e94195cc05"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based ThreadResult wrapper runs the function in a daemon\nthread. The returned result supports wait(timeout\u003d...) for a bounded\njoin. spawn_n is provided as spawn without a return value. The COPY\nmiddleware heartbeat and the object server\u0027s container-update call\nsites are converted to the wrapper.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2aad24f54112b193b45336834a7e0cb045e945ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2aad24f54112b193b45336834a7e0cb045e945ee"}]},"branch":"refs/heads/feature/threaded"},"d75c219eeafe493f37eb2df3f25bb27650203225":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/16"}}},"commit":{"parents":[{"commit":"8fc0af88ca4cd1eb69d2f532ed996621da3084b6","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fc0af88ca4cd1eb69d2f532ed996621da3084b6"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:18.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based ThreadResult wrapper runs the function in a daemon\nthread. The returned result supports wait(timeout\u003d...) for a bounded\njoin. spawn_n is provided as spawn without a return value. The COPY\nmiddleware heartbeat and the object server\u0027s container-update call\nsites are converted to the wrapper.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d75c219eeafe493f37eb2df3f25bb27650203225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d75c219eeafe493f37eb2df3f25bb27650203225"}]},"branch":"refs/heads/feature/threaded"},"c9548d4f68d8ba81386348c24c4dfdb3547b7a82":{"kind":"NO_CHANGE","_number":17,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/17"}}},"commit":{"parents":[{"commit":"51a472c8942f1d1c097519ad56f47ef8ec99a97a","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51a472c8942f1d1c097519ad56f47ef8ec99a97a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based ThreadResult wrapper runs the function in a daemon\nthread. The returned result supports wait(timeout\u003d...) for a bounded\njoin. spawn_n is provided as spawn without a return value. The COPY\nmiddleware heartbeat and the object server\u0027s container-update call\nsites are converted to the wrapper.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9548d4f68d8ba81386348c24c4dfdb3547b7a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9548d4f68d8ba81386348c24c4dfdb3547b7a82"}]},"branch":"refs/heads/feature/threaded"},"3c0ed44a80b29a7375eddc279a537d1b70619855":{"kind":"REWORK","_number":18,"created":"2026-07-22 15:09:59.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/18"}}},"commit":{"parents":[{"commit":"66129e4fabac0a42f4a17467043ff926a781628b","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66129e4fabac0a42f4a17467043ff926a781628b"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-22 14:23:38.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based ThreadResult wrapper runs the function in a daemon\nthread. The returned result supports wait(timeout\u003d...) for a bounded\njoin. spawn_n is provided as spawn without a return value. The COPY\nmiddleware heartbeat and the object server\u0027s container-update call\nsites are converted to the wrapper.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c0ed44a80b29a7375eddc279a537d1b70619855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c0ed44a80b29a7375eddc279a537d1b70619855"}]},"branch":"refs/heads/feature/threaded"},"75c5545e51c2b2f6457a997e4c44fdd36208be9b":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-08-21 05:41:20.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/76/979576/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/979576/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/979576/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/979576/19"}}},"commit":{"parents":[{"commit":"3622e8d5c3abb51a6575c999f10f0ae5a9510114","subject":"Add replacement for eventlet.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3622e8d5c3abb51a6575c999f10f0ae5a9510114"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-29 12:43:32.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Add replacement for eventlet.spawn","message":"Add replacement for eventlet.spawn\n\nWhen eventlet is enabled, the original eventlet.spawn is used; otherwise\na thread-based ThreadResult wrapper runs the function in a daemon\nthread. The returned result supports wait(timeout\u003d...) for a bounded\njoin. spawn_n is provided as spawn without a return value. The COPY\nmiddleware heartbeat and the object server\u0027s container-update call\nsites are converted to the wrapper.\n\nChange-Id: I51cf1f6503cfa6044e847030665d18f0c4e0eb46\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c5545e51c2b2f6457a997e4c44fdd36208be9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c5545e51c2b2f6457a997e4c44fdd36208be9b"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
