)]}'
{"id":"openstack%2Fswift~979578","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-28 12:56:07.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-28 12:56:07.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-21 08:23:24.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-28 12:56:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2","subject":"Add replacement for eventlet.tpool","status":"MERGED","created":"2026-03-09 12:26:52.000000000","updated":"2026-08-28 12:57:40.000000000","submitted":"2026-08-28 12:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979578-eventlet-removal","meta_rev_id":"288445eb2cada37da8f8eba09ffaed6b074eeecd","_number":979578,"virtual_id_number":979578,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 12:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-21 07:44:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-08-28 11:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-08-28 11:44:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:05:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 13:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-13 01:51:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-06-17 14:18:38.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-07-09 07:24:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"1fd1efefa25b5a4b42bd5435082d1d7f45f16fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 12:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0499eb35d0a5dc4b9ddfc8bd91b79eea012000d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:05:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd96a636bb00456eaaad86ed8488790d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7235698deb1428f8e772ce219839374 : FAILURE in 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1386b95941dcc0ff0b4743cee354319bde66356b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:34:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd9b05d2d20342d79d2e9372fd3d22a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15268c057904ab6864e0ba2e2489b67 : SUCCESS in 9m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff49918cb9f1461c808b6e2f53b5e88f : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1b8024e150946beb3bde78bc829c417 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7b6a8da03d549c3a2e6ef6a143ff547 : SUCCESS in 25m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/107b9b8b42114fb3921a39f84f8c2814 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/876a5d9857234bb7bb9b3e0faf64496d : FAILURE in 12m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d3e8a7447eb460c9ea8b7d0792998b9 : FAILURE in 22m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f9ed1531d29444c0bd568a94a6aabccf : FAILURE in 30m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5ead976e008040c7a63e4c9e4388030f : FAILURE in 29m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/014cb9f33bee400bab9f9a1b94bf363d : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/df760ebde6f64db7b134ced3fc7f5e20 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/08ae25a37c7140d1b4c084519e0124c2 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46e9ab5045cb407688e1599a6d0e1be8 : SUCCESS in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d573ca4f2ff34e67a6cf6ca1925636a2 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/72bdabb2e64c4d3baf7b3f714e3b1616 : SUCCESS in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c92fcbb7abf24244a9e1cf9d8d16a9f1 : SUCCESS in 59m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f7ea8ad89ff4b8cb54d63681ebf483b : FAILURE in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16cd373c0914ed289736318b5a7e7f7 : FAILURE in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d1ed439a8dc14374848b1a85cf90be34 : SUCCESS in 22m 06s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/copy.py","accounts_in_message":[],"_revision_number":1},{"id":"a5c9020c020824c13fa58204f73a08792562108c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 16:02:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"738305dc9a0d8df745f3af3197807fe704f89327","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:40:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66a5acf669464799b523fb40721863d8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c25c4f7691714f9a940f8de16428194e : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":2},{"id":"449e987d0c53fa797109810cfa8af4faa3954d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:45:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe240eb7a15448789fdcdcaf67057bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d32c99f3a4470bbd38abbc52371cec : SUCCESS in 10m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca9133da75254199a363a2b358d3a221 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39d67f4784c14554b54650ed941e0ceb : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c7490689a4645c7b7dca39f55cc33ba : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e36d949d872c4f89b2f9681cec242d8b : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3bfb2f3030c74f7ca8a0a6fd90ea678f : SUCCESS in 11m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e96603e90fc24296be4ab780929063da : SUCCESS in 10m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7147541567b048c7813d7f6794466433 : SUCCESS in 11m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04f15ad9ae32426480c55bdbf1b9e8d0 : SUCCESS in 20m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5caff48d7ae24b8fbf5420f4327366b8 : SUCCESS in 9m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84d6207fcb2e42d39e11ea99632c6e3c : SUCCESS in 8m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2e3370274cf94a05953d56f81fd4ff10 : SUCCESS in 8m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c83bd40012064806b9e9a33768d95713 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a2a717381b548558d9d0bf5acc861df : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/96dd47448c504425b671bd6fa851b52c : SUCCESS in 5m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c3a970e55654f339409e20be8bf25d6 : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2894452bea9b420da4d82fa961371d7b : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1b3586c92e44f8ae1cdcea7199cab9 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f14c3710dfa435b9c2c8ff9fa05c881 : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":2},{"id":"99626a5c8d69fa0f7e4c6bbdbfe51b401b40e04f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-13 01:51:32.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7ded645a8c2ff8a1e226bb80d5527e0456eb16e9","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-13 10:48:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2d3965a4a3f8470bce27d2777d106e10c3459728","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-13 15:37:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2da09daa583399d96eacc3c5618711e0430c96c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-13 18:23:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd7d0a3c0f94a72342dc8c07b9db483b2919540b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 18:53:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4335656c86d047afa05d91223d8af14a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07423bc171484d9e896e13d68ffe61f7 : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ab202832b50d5f2c20fd49642ad5adbf05ccc8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:23:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a604562bfed4a9ead4c65a417f5bd9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76572c93dbd0494a8470c2e9234fe627 : SUCCESS in 10m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2e263ff05c54da493ce11338161c55f : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85b56c65b0434d95b61053a770b06894 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85944cf0b334452cbd77ff0dff554b79 : SUCCESS in 16m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33524adefe4346ddb051d3ae384e30b8 : SUCCESS in 8m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a732eb7e8be642fdb7af4da990dc58be : SUCCESS in 18m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/34b1fd7e9449455bbe64dfe4c0de8f65 : SUCCESS in 23m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3f554b12d78c4de9a5bf9592c419edf9 : SUCCESS in 13m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/522d1b99ead64638a7873b4a8fdfd67f : SUCCESS in 22m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d8ce62f6ebf84ceca837324faa5988a3 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35e13d6b4af04ea6a29d7e43ff79ff9f : SUCCESS in 8m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b918879ea8714a49ab4486bd827380d4 : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c26626ea5ead4b249c6ef0c5ea965cfe : SUCCESS in 7m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5db9d619ad144a5c92b185786cb1ed30 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7044435ae61f4666ae62f3792b7980b9 : SUCCESS in 5m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/faa6fb02dba3471c87c71080d296c23d : FAILURE in 46m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c7a70f974364a68816e835059ae0d66 : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91f885e78a240089dfb1fd7ceb2d060 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06dab5f0ccb0439eaeedfc134da4fced : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":3},{"id":"609c8fdffbf33c4e841d25758493ef42439085ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-14 08:49:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a5f129a2d25f47e1030cb91eb2a85e832eaf4c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 09:44:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d90865007c1044489337dcbc146a310e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8cc71629c2416bb40edb0e1d52ac34 : SUCCESS in 10m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ea69ba2b26e4c58b7a5c62c80db55c9 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e76f28f843fd4f92bd83dc454e1b0d2e : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f09887daf8a455b848773aff0c9f375 : SUCCESS in 12m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e22b4c80d4c54639b2d3c873c107cbdb : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d95728488d44b3c87a78619f6315811 : SUCCESS in 13m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/610955ed734d4d8cbb415bb2f8a75049 : SUCCESS in 16m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/551a0bab4a384ac2a128ddf357340579 : SUCCESS in 11m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ac8d6c14614647a599706548f4616a3d : SUCCESS in 18m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac2a97a7655848e098574499a23fb5be : SUCCESS in 8m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4e67cd7c63b495cbc885808083b1a43 : SUCCESS in 8m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fb61b359bcdf4066854e179c4e805bb6 : SUCCESS in 8m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0bb77391536346d6a3a88a0dc883811b : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf5200c9dc77420083fc79aec176cbaf : SUCCESS in 13m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5f01772b3efc40d9b9208410b0abee15 : SUCCESS in 10m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddb78e2eca24484c9089c4ceff187336 : SUCCESS in 44m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/605855fde26c419cad5a1a92ee2bd351 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fa3dd2b31c4b60a2da1e739655c2e2 : SUCCESS in 2m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d42fd58210644b12ae937b42131cfc2a : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":4},{"id":"aaf001f5be06a24ef425be38dc558ee31cdcfc35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 09:56:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a7e8c5d1eb46bbab6e901ee692f23e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b2f1b6fba3c24eea9c9377a46d53320d : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":4},{"id":"1bb85449f2e7ac6fd02c2460de39cf41cb9a3e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 08:59:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"abd4ba3d8ff9200a5183c00eaaf17de8bb98c6d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 09:32:47.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53acbf01fcb542a7bbc53a1e49ff4791\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8004cfc43e03490b98d1c3c260bad732 : FAILURE in 31m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8bb2ca313265c5d7a175a234d16aaccb00f0c85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 10:05:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b979a55d81d40b0ba6e72e9e6f73708\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7269e3998e42f0baf84c31902647af : SUCCESS in 10m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54c88b3fc8a54d9aacdc5dda6e61dacd : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8aa187dee21f4a1ba74bf38e27801fb8 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/373592d709874e89b077df1dcf3a66ab : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f45a791a86864dc8b8ac2c9a3a457bcd : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fea278659ba147d9bc925a767259c290 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/701b802d8db7489484072da14baa1932 : SUCCESS in 24m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a042d0601914c2f8c812bff6c708523 : SUCCESS in 11m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/072af2502342415e9332ca0d6bc31223 : SUCCESS in 29m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eec806b7772e4c678eef208529c55f4a : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7b625a64db5b4375aacdacfb7c7dda1c : SUCCESS in 9m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/831461a7fbaf4484b8d93e61961cf253 : SUCCESS in 9m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64772b4451584d2c8441e9e7b180698d : SUCCESS in 6m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed18a68518e845e6aa9e3a3f8d568fdd : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/013ac22643244687955fbe09f22cd04d : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3913314894a74707bd81e0c0a0ae7599 : SUCCESS in 58m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/189a76cecfb14dea9984f391b47174ab : SUCCESS in 27m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53670e3271c4eea8c5111248c7bc524 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0a3b766d9c743c6a2b397babc3962db : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":5},{"id":"52622edd84c6cf8992f6350f27de661335002603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-12 17:00:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"bdace3f0957a9a9738a8695d320d054ab2b11ebe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 17:45:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a27ecd87bc6145cdb898b4b82fdee201\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3061432382d148128c055dd232a7a884 : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":6},{"id":"d0b76fd0a9c6392f7b978098b7e2b7ca4bffc8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:08:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af7776bed4b647c8b1aeac184120ceed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c792e6c4bd465dabc2177683557926 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54f2b5b079964edb812ac0618722bfd6 : SUCCESS in 5m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88b821e64ae64d82a73b967dc76349bd : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cbff36e6fd84dc38260191cae4e517b : SUCCESS in 12m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8be4adfb20349a69bb6506d18b57d3b : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63d7497fc12c4589b43c5608c3120caf : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68a92a92b7ed43d5b57ccb059466f3d9 : SUCCESS in 10m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2cc08c2238324124bf7c0a463a81b5ca : SUCCESS in 9m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cef0a51dd7eb4d2a979ab667e082cb6a : SUCCESS in 30m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f2770db34d844d9097a174ea48b33568 : SUCCESS in 21m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7da95d6d5a9e4281a5396404eadff67a : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/12afa5d5f4b441d8859d8002a518fc6c : SUCCESS in 8m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/645f3e2160c94ea2ac800b28b3852423 : SUCCESS in 22m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c72330f2256b49d6a85226a5915cd749 : SUCCESS in 6m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/936d42b4458440d09950ec04bd357999 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/60632615fa754981b4c6dd0f72c4ebfc : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c718ce892134015ad57649d1e21e85a : SUCCESS in 1h 02m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f27f4abf60144a699c75da59c8ea5f71 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7e1bcca4284d06a818767eebfaabd3 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8cdad2c537a42cda8e5dd7b29c45ebe : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":6},{"id":"2272df2f0b342e2581f9ac28198425ee589f6667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"98b8c41c20753cfc979d2c798b6d3bb860684c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:10:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7101b0e90e04b7faa327c095b7f0734\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e749bb15aab4ca784355efc02d2c274 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88652967bb44481688decc3d3063d988 : SUCCESS in 9m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11b74e4a4e1e4370acd43bce2fb9cfec : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60422c02cfe146c0aa26ec14a0c6af61 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4140d8eac48b43d484c0152823b06c49 : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60d8faa22d8c40cd844944df0fb98497 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6721152965d34c2382066a479e2e984d : FAILURE in 21m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef2a4c8d00ce47458cdb7155f25ffd86 : FAILURE in 20m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/315115a4adf7424faba0b26350bdfcc5 : FAILURE in 17m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c232b22fce274d65bd93f5518ac923f7 : FAILURE in 21m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/86d101b77a2d453ead02d6c7c8fcdbf1 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/69fb51d5d4934a138a43aaaa0dcc544f : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/33233fc9c2a54d908ff4534c4b0266fb : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17be2907e124482ea87f6cf8fc02779d : SUCCESS in 4m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ee664ab351847ad8dab6b6d300ad321 : FAILURE in 16m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/66e381d4bfb84ad5b8d228439163395c : SUCCESS in 12m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/11aa20555b08448aa7618f1aec5cde4e : SUCCESS in 48m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/016cf53e79714a56b223bafb8447070b : FAILURE in 28m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7a46e595c14118a0cc7d50139fbb0f : FAILURE in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bcf1fd80f1784be7bef5b468b96b941f : SUCCESS in 9m 47s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_updater.py","accounts_in_message":[],"_revision_number":7},{"id":"48e6f9e3126c1096f9833b1765f36245bf87130b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:17:22.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce81575a93f417a9049abe4297c638e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5dd1178016c2445381db715b9105b151 : FAILURE in 1h 01m 32s","accounts_in_message":[],"_revision_number":7},{"id":"380e91126cb51b2aa7323bc1b6b3cf9879fbfb9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"fcdbea6c9c3a07e0829e6adc158b4178fa339b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:42:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3f386457af4e2e8bb5fc6dedb6c21b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f94f984153b46c6b882efc656083afe : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aa02a3fff6a43a3a98682f50d586eac : SUCCESS in 8m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69899199ac894b86a976a2ff7e1a3727 : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d015fe0cfa84ec78a3e5f2f07694b56 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdb359898b0947a1962fdb7101bbcb58 : SUCCESS in 17m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f43510b94d154956a03073ae43048c4b : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b2d4962973248e9bbbfed2518456d86 : SUCCESS in 25m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1855ef952e9b424e8272192e1e58dde9 : SUCCESS in 18m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/653329f6e4b34a13958465eeb0cb1452 : SUCCESS in 11m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3921c12cedc84f278513e9197803956a : SUCCESS in 9m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/381fee941f4a406eaa9a9eeb90ef7562 : SUCCESS in 8m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b3c13c92cbf432888729502c2afb826 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc641f84fc99416fb16b5d5c2ca37a01 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4057f759ca74f02a63633a47f784dbf : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/77a8be5c29ff4b009eabe132eea3095c : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/596c6acf411f49e392002a2423d00003 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a92b0d82e39491692817924748020e1 : SUCCESS in 1h 03m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12e56678f744423eb89ad67ee0756261 : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1708971256a140dfb3f1e608593ac0b9 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/acd6fa80772b43879fefe3372168a338 : SUCCESS in 19m 34s","accounts_in_message":[],"_revision_number":8},{"id":"0abdef8e23ca6c07c5a3a2d90c0d4e8488b6f9a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:52:46.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca1bcdcef39a456da6154a654e0c67b7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c666c5edbaa844c8932e141b20582343 : FAILURE in 1h 13m 26s","accounts_in_message":[],"_revision_number":8},{"id":"8d436b1af70aa8ac705e28c5c4202ad363875b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"714cf5cdf4b88f4217c18a4bcbb77823a950e8af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:18:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/753821f1bbcf40d58c653ab931395127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ae458938d34b92bfc925d16fdd5491 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24d77349cc3b40d08185ec06a00ff85c : SUCCESS in 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/006fccfc2372422eb3a6f0c7449abfdb : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5eb875058a44fe7952c00d4fa380112 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1657053c187f4ec099cb16b96e378479 : SUCCESS in 22m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52ed8ea77e5f4fcabe1cba67cc83bf51 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4b8b016fa0f44daa50f05b63dc715e0 : FAILURE in 20m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5a4731a7909c422e9c1b16752ff064a8 : FAILURE in 17m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ccbdc8dc893c485398890e984f8772da : FAILURE in 35m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/567e1eb4f4574e7ebb492ba3dead1b0a : FAILURE in 20m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2d2b331f16ee49e891cb538224672402 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dd4bdce077054b3ebaf77edf52ecb2cb : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6af0ebfeebf4faaab83ebe1726b8e64 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7df9f2f8f453449d91cfde150a09f02b : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a174593ca8d4b89b18d41ba9e48bdbf : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/adca9cb0c62b4d7080900b183667ac5e : SUCCESS in 9m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d6ad069cb284d65af8cbd4a7bef9d27 : SUCCESS in 42m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35a7254d392447068cbe688fa8d7754a : FAILURE in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6069decaedbd46b5b1598dbc227a6137 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/08771bf45ea947759d7d95de4a835120 : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":9},{"id":"26d8583ec60ecaa513809465acc5afd092c81b88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"62703d370c7c24c41c6f751f6c37a7d982223abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 07:07:17.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d04f36bc3bc64c5baa3bc85ce281120d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6bac2d58eeb46dc87441aee34b0b697 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9167baf5ba7945549e55fecb4aa240c0 : SUCCESS in 4m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ce253c02df44399b040b2f847df12bd : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f4aeaeba6e647919afc1bde7396ac6f : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a20fb00580b496796165ab6eecd70fc : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45cdaf4ed2d94de19df3ac1af633f95b : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c747ef02372e499387b1dae1be683504 : SUCCESS in 13m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7fe80b2c86944c61bbe4163fad970a09 : SUCCESS in 19m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3fbe71202d4b4d8ba266b6813d075fdd : SUCCESS in 24m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/88db2e2e92a8492a9ae69250a28ebbf4 : SUCCESS in 9m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e915fac4c504b78b5b41574242ebc93 : SUCCESS in 8m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/42c71e2cfa484259acc8bb2e199e3e05 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eb916e0b38724247897ecb91c0d5a78f : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7bb9cbd8fed34cf881089a5f65f9ba6d : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e885802a5d1948219db5797a9fa9d99b : SUCCESS in 11m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5863e768ebc5465f94d84d17a5b6a156 : SUCCESS in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d53f2f1d905f4587b174555ae708e8f6 : SUCCESS in 1h 03m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddc2b7d31c394425bb6d1d2bc238ab6f : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/602a9d945bd2403aa91be42c3047c674 : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70257ebf54d24b2f851d2a21717eb619 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":10},{"id":"5a4dccfb9d7adda245300479ebce901791297a54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 12:12:57.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6a11b91dd54e72a1062788ce6e092a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/448788a53af44d06aab26887848e26ee : FAILURE in 58m 08s","accounts_in_message":[],"_revision_number":10},{"id":"af40c24c365a47c33fd2bef619c5dabbd7e3d61d","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-17 14:18:38.000000000","message":"Patch Set 10: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b1fb910c3fe134ed38bd01742f869a235ad519c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"7a2f345238e1e7bedfef3149179db702d69523aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 08:35:55.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21171f02e87b490a9457bfe4c995b2da\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5f0270364384397b6f639b8bfcb3118 : FAILURE in 52m 53s","accounts_in_message":[],"_revision_number":11},{"id":"92a31994075ca5603257ec843354d7fe76046064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:20:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d95afc21734a3fa618366fd84343af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b78b94340e42d4aee3c771697d48f8 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a93c750864943f291fd143ce229d5ab : SUCCESS in 8m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4c599c3cf1d49c187b4fb4d315b29c9 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89f9290c194641289f1c5e1b5706f8f6 : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b57a84e7f62f43439ccf635b790b1e96 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74453f7312494a3e91613a69f54e69e7 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c3dc86c72b94a5795bbf16d93a8b07c : SUCCESS in 23m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3a001769cc21474992f5bf0786530da4 : SUCCESS in 19m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1e033d42da824f81895cd120dda6b0a5 : SUCCESS in 11m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6ed0e560f9c549809c001ffc6f3273f6 : SUCCESS in 21m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd65a9074fcf4999a6865d393523df93 : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1bb80c4931c74bac9f3831004218974f : SUCCESS in 15m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/421da4efe0c14b798994d9a879838102 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a67fc42d332d48c283a9834f796cf2a5 : SUCCESS in 7m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9920dc2e193a415fb7fff188239df9a0 : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4325cb727ad849fa89c60319aae4e9f4 : SUCCESS in 5m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ec95a7ba8ea469fbd6bbe6bfb5cb587 : SUCCESS in 58m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aedbf58dbd34ba0b24731bab837c5c5 : SUCCESS in 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8d6c3033ada4c7eb2b76bf26e0663a1 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9c7aa95ea464591945bf88c0f3eb5f6 : SUCCESS in 19m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"746f8618e019431aae7525499f78ca8c7883a784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 14:08:32.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"fa6de4e02984c8971a7cce9a7a8fbdf1e8080c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:21:09.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0527888051e4049ae80ca38a335cca5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4312e59409148d6b17c1232034aef2f : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eba45bade0d4b6db92307181cec69e2 : SUCCESS in 8m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a13eec14ac684bffb4dfcc741f694472 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9679053e38d3461686b9639e3377dcc0 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b01b276adf84b2297d2fc5f150334a6 : SUCCESS in 21m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6b6fa9a5d804d9a9ee420221e321ff5 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2db83c8bef04ab8ac5890b63fcd33d0 : SUCCESS in 26m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/14cc3e41dd5c4723b55d31d8bdaa48c6 : SUCCESS in 22m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cc29fe493cb644beb8aa3114dc41c479 : SUCCESS in 31m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/95b9ba0b827e4fdd9d70b958e707f8f4 : SUCCESS in 20m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4f9b73057679404fb16fe76ab05e8433 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a98874a8631a4219afe7d8aec14a7e50 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/53f97ea8ccce42c79309ba8ad313a4cf : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b1e851aa6c74ee0bbf1fde6b4c7d30d : RETRY_LIMIT in 6m 40s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2091ca3cb4fd46d9b083dba7dc7ca4ee : FAILURE in 14m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec838086a1c141e4b2cf9ed0e157b0d5 : SUCCESS in 8m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9eb1da13614244029d3727ba62ac90b5 : SUCCESS in 1h 01m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e78663fd78d43ad97e3d90851acd300 : SUCCESS in 28m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a3364103574b0dbebef997f4633686 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ddbe762bd1e14d28a6e7021cc811f2a3 : SUCCESS in 21m 10s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"46bf8fb6bb9578f689c5d842a7b95f0ac8863e63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 23:13:15.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce0de27557d245c9be62797a85b9b940\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50afba8a08a4433db4e49e071410fafe : FAILURE in 1h 10m 42s","accounts_in_message":[],"_revision_number":12},{"id":"59730f49067d680b1caa4531a274718d4df69ec3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-09 07:24:28.000000000","message":"Patch Set 12: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7b7cd37d30d0d88d12495fa2183c2bf3f762619f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"2bbdfd487530e2789f9fc8042dae597ec0e92588","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:25:40.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991d367be56e4598a6f35128d8b40b41\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/446b7cf207104d9480acd884651cdd2f : FAILURE in 1h 03m 29s","accounts_in_message":[],"_revision_number":13},{"id":"dac1f4da88fe0916543781e5ef9ed81df33a8042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:42:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e56c21baa6642468e96005bcfe87933\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2d2b4bb6274e46babd41bb3213b6c6 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b61f82a2d8424fc38149c7a2542b16b8 : SUCCESS in 9m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0581f934ef6648aaaec8edaeb4091fc8 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f096709ca777485db8dab95b9359171c : SUCCESS in 11m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57005f0f7b4048a5a3d0af1405e93930 : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf117d2e47e649a7b2b6c1b9e6dd5590 : SUCCESS in 11m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24262fbd3ce94c0a9a0571e025b0302b : SUCCESS in 26m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b9a27cff83834379b34604e13156938a : SUCCESS in 19m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/76a8ec963e3c44fc97783ccfdf6bfac6 : SUCCESS in 30m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ffad2bbfcd564ed69fb7eb39e4c27552 : SUCCESS in 21m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aceeaa051c60488db3ae5a1ebedb7a47 : SUCCESS in 13m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b95ce3576f114e7785cfbea571142c03 : SUCCESS in 8m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f56bdb0d8b7445558acf5169119aa146 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fecf45ecb1fa44ed8b846ffd1be314c0 : RETRY_LIMIT in 6m 34s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e0e0c6d03fe416abef731610a630996 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a2b6bf2be16c4c36a14b2efb5f8fea15 : SUCCESS in 5m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/392787cf37534777a7edd7879f17215e : SUCCESS in 1h 14m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34f1147deb664341ba384764451b14c8 : SUCCESS in 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa6a1d594944c34a88601aa0eaf627e : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26df979ba02d490483a764a7210482b2 : SUCCESS in 19m 10s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"2433b80a60e2016ce3bf5a13f2522042a4c65fc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"c135a28bd7e6614d9a273100b20365046d1591b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.","accounts_in_message":[],"_revision_number":15},{"id":"51de32b39f408ca3bffe5f2644795cc39f35458a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:22:45.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/328210611bc748b7810b5d1c3c536e63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec9920008b149b6b9c029f7f71c2377 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4afab89fbbf432ea654360c3d62b5d4 : SUCCESS in 9m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6694b10582414445b54ed12130004eb3 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c71846c7974248aeb354aeda337fca17 : SUCCESS in 11m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f50e2b89d1f43c3abbe1dbb6013736f : SUCCESS in 21m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1db16532d55b4bed894bc718b38f442a : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca40bb4ddab946ab9af8c9983987e52e : SUCCESS in 20m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dbf37996f7ea467b8f30aba7e42c2b3d : SUCCESS in 22m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1c0e3b01fd814aa9bd25751c41392910 : SUCCESS in 34m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cb2c481a9150496897c420513c298a87 : SUCCESS in 9m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2ab9e3bc459645d88653b7eadb555661 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7797a0dee18a47dfa329de5f5a317ca6 : SUCCESS in 8m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d48d9c5aa41e48a7b49881e7f64d3aa9 : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c67160c1a5b54a1cb9780c3ecd350a3b : RETRY_LIMIT in 3m 52s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bac98d3362794c80bcee197890192fcf : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7eb6ffea2bd94dd09ef0e25093043379 : SUCCESS in 5m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39db1c15f36a4f7fbfd83a245aceaff5 : SUCCESS in 36m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5558e67d7e06485e83692462caf499fb : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d3a7a895684010b5270d35b54a7eaa : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35d38e6a9ef54fab9ebcc129f393a5f6 : SUCCESS in 20m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"e9367f567c321a23692abd56d3c0ab0008244e4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:48:15.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b4a656ad094aaa8b1938309286ed48\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5edea56a34db4315b519f66a4cfccddf : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":15},{"id":"89b7acb2947cf62a70e8abb645d00abb7dcee246","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-21 10:55:28.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"a57db546b55055112e6f665a5a152cf53cc03263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 05:41:20.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"55c4a6be51674eab5e7992b333f69f1d8416d051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 06:41:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07917dd195a343618b6d919ca322a2b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50912afdfc74f94913debca297104bb : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93a8358f73e74ee6852322413ee78d9a : SUCCESS in 9m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7dea2427a22a4f44b965ee1884cd47d1 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7d8b09d4ee54f81ad63b874431042b9 : SUCCESS in 11m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16b10dbfa7e8400da1e84c1038161db2 : SUCCESS in 14m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68ccb14916c940c8aabfd9c3f4e742ba : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af0a8d55be04479ca6e65a07ffab41f6 : SUCCESS in 26m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/27f3fa9531a844b6a1acf63c97fa8316 : SUCCESS in 13m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/979079c7c930470c97da4a5ded9d874e : SUCCESS in 31m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/79de5f8c40634c5d984542465dde0cd0 : SUCCESS in 11m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/052b1f795ddf4d1d9a36ed2c09521f89 : SUCCESS in 8m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c45b86dc27204acba7bdb63daf8aee2a : SUCCESS in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/12c05f354b8748068e75a702f4a410c6 : SUCCESS in 16m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/487c969dc2664fd996f17f574ddbb79c : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2dcb0e361b0e415bab7f51bcb8a26c0c : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/324e7fbb6799425c842e454408aff767 : FAILURE in 7m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fcc08ad67174e6e91e00ab2c5c252ef : SUCCESS in 11m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f11c57981e4b41cb9d888a124ae2a633 : SUCCESS in 43m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ca05af770694eac88dddcde2625f6cd : SUCCESS in 33m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fdf8d1fce84435bd0192979bbaa787 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edbe0a8ca7844f1aa044ad27d404b746 : SUCCESS in 19m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"7b2033309f94b33af9e19bc45f884f4f49b81934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:24:03.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d736aff91ae4ef2903fd4e650dce0c0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d6360340dbd642fda2221a240fdbc9a2 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":16},{"id":"16d4c209a4b9ddb6de398231bed975348d34711c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-21 07:44:53.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"3656c79bccb183ecdb85caafdc09c05e2780f6d7","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 08:23:24.000000000","message":"Patch Set 16:\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"0bd2f1be1246b44130bfcee35dd21f877e6a28bc","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 11:44:28.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"85893c7325a9fc761952362849172b620b165e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 11:44:48.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"f580eafc47b51aeccae2a4877e1cb0f0b9d051ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:56:06.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b89d826248c49fe8bbcdcdc619959a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065c480503f0404398da4d64e33935c8 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b168d30ad1064084b2f3bbbe367f9da3 : SUCCESS in 10m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b49c6bb1d15c45fda9aee9ae530b9f09 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/632c26cbf5e740c7862d0618a5e0b239 : SUCCESS in 25m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/53398d3dbaf54dc09d3bcbbca48cc49b : SUCCESS in 10m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c2a0f399bf6546eaba402a614e67b1b1 : SUCCESS in 23m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dcc3901da985449485bea06c48a4c63c : SUCCESS in 19m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95e01e777c6f4c818f8339058e0fd1f7 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ef21acc6b7094855b8e86e4a544c9194 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f90952ecbe6843b88412629de2a49f0e : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d818943610a8412fa16aaeca80acf2ce : SUCCESS in 6m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1283507a2f464c529423c1cbb2ce2baa : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb975c1341ca4819b55d33e65e6e6364 : SUCCESS in 1h 02m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f21be703df644ebebd5a3cd604cff328 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cc2054275f452db5e159689df8510a : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28afb9e5d2e14c66ba037925e2fb4316 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":16},{"id":"b230974443076ccdba267ba4612cbbc19ca0c6c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:56:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"288445eb2cada37da8f8eba09ffaed6b074eeecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:57:40.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2154f28a47cf4719a36cc40ca9d0c0af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffaf72253b904d44819afa66464ea6ed : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c32ec88c5edb433084482dc9e9cc21a2 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"4e0e55adca31eb8a804da7696d9dead82d4db813","revisions":{"fbcbecc9b01f65662412a6913584db87e7090d13":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:26:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/1"}}},"commit":{"parents":[{"commit":"972b99e66a033ede12bec997fc59be898f82212e","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/972b99e66a033ede12bec997fc59be898f82212e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 12:12:09.000000000","tz":60},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbcbecc9b01f65662412a6913584db87e7090d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbcbecc9b01f65662412a6913584db87e7090d13"}]},"branch":"refs/heads/feature/threaded"},"c644e226f502b50cd07afa73eef037cb5df7b48c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-09 16:02:22.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/2"}}},"commit":{"parents":[{"commit":"26c8c297fd9d9621d0a8198691c7a9ef7524ce6d","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c8c297fd9d9621d0a8198691c7a9ef7524ce6d"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 14:07:19.000000000","tz":60},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c644e226f502b50cd07afa73eef037cb5df7b48c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c644e226f502b50cd07afa73eef037cb5df7b48c"}]},"branch":"refs/heads/feature/threaded"},"715cc5c1b96df4ecd3dd8b4900716e16429ef7f4":{"kind":"REWORK","_number":3,"created":"2026-03-13 18:23:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/3"}}},"commit":{"parents":[{"commit":"010be858b3cc2b058a1b98d88e3d104ddd0c6f05","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/010be858b3cc2b058a1b98d88e3d104ddd0c6f05"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 17:12:33.000000000","tz":60},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/715cc5c1b96df4ecd3dd8b4900716e16429ef7f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/715cc5c1b96df4ecd3dd8b4900716e16429ef7f4"}]},"branch":"refs/heads/feature/threaded"},"2f4269765bfc2f091fdea0d62ddbe5d4a2e992f0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-14 08:49:50.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/4"}}},"commit":{"parents":[{"commit":"8af323fbda9962e89e8bbe097d2fc1a72b2a27fc","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af323fbda9962e89e8bbe097d2fc1a72b2a27fc"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-14 08:44:43.000000000","tz":60},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f4269765bfc2f091fdea0d62ddbe5d4a2e992f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f4269765bfc2f091fdea0d62ddbe5d4a2e992f0"}]},"branch":"refs/heads/feature/threaded"},"a0588410bc507c446a0d079854749a4d933a9a10":{"kind":"REWORK","_number":5,"created":"2026-03-23 08:59:05.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/5"}}},"commit":{"parents":[{"commit":"e05491938becc769e84a951de9ab0a5a7153d4b6","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e05491938becc769e84a951de9ab0a5a7153d4b6"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 08:40:21.000000000","tz":60},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0588410bc507c446a0d079854749a4d933a9a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0588410bc507c446a0d079854749a4d933a9a10"}]},"branch":"refs/heads/feature/threaded"},"f526d0f21a095a1f336ceb229b23d360660cb74d":{"kind":"REWORK","_number":6,"created":"2026-05-12 17:00:27.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/6"}}},"commit":{"parents":[{"commit":"875c8f40edd7bd4bb8a6cf68adee197234b2891c","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875c8f40edd7bd4bb8a6cf68adee197234b2891c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-12 15:40:51.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f526d0f21a095a1f336ceb229b23d360660cb74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f526d0f21a095a1f336ceb229b23d360660cb74d"}]},"branch":"refs/heads/feature/threaded"},"d867e28857f15d002e83c7cc433ab4aef5b44789":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/7"}}},"commit":{"parents":[{"commit":"bcb5a09d43e5bf115431ad3ae70a85d23f6ecaaa","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcb5a09d43e5bf115431ad3ae70a85d23f6ecaaa"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d867e28857f15d002e83c7cc433ab4aef5b44789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d867e28857f15d002e83c7cc433ab4aef5b44789"}]},"branch":"refs/heads/feature/threaded"},"c850082931f815e79724e8a32b7f25c832c2617f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/8"}}},"commit":{"parents":[{"commit":"16c262422bf2f1373f48966bc8e1d87f255f1558","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16c262422bf2f1373f48966bc8e1d87f255f1558"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c850082931f815e79724e8a32b7f25c832c2617f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c850082931f815e79724e8a32b7f25c832c2617f"}]},"branch":"refs/heads/feature/threaded"},"b469e6652138c714d776e260d4dc21f5d181be5b":{"kind":"REWORK","_number":9,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/9"}}},"commit":{"parents":[{"commit":"906f6eb06b31b2b9f29116c1817f9b034acd91a5","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/906f6eb06b31b2b9f29116c1817f9b034acd91a5"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:13:55.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b469e6652138c714d776e260d4dc21f5d181be5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b469e6652138c714d776e260d4dc21f5d181be5b"}]},"branch":"refs/heads/feature/threaded"},"ad6424b53ee6f9c9c8fe6f2107486dfa5c692491":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/10"}}},"commit":{"parents":[{"commit":"0fd3c0f57b28b8344d753f2792445e37554da3f7","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd3c0f57b28b8344d753f2792445e37554da3f7"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 14:51:24.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad6424b53ee6f9c9c8fe6f2107486dfa5c692491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad6424b53ee6f9c9c8fe6f2107486dfa5c692491"}]},"branch":"refs/heads/feature/threaded"},"c8b1560c38e885a815c96b281d282be6f1026930":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/11"}}},"commit":{"parents":[{"commit":"a797f189001fbf4dfade2466a69c1b2c34797d59","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a797f189001fbf4dfade2466a69c1b2c34797d59"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8b1560c38e885a815c96b281d282be6f1026930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8b1560c38e885a815c96b281d282be6f1026930"}]},"branch":"refs/heads/feature/threaded"},"25dc6bc11a67a0d2c682e7789eb76cec990fa6b8":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-07 14:08:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/12"}}},"commit":{"parents":[{"commit":"c69e6b4339fb347c54c8e01fbd8aa9541b946f3f","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c69e6b4339fb347c54c8e01fbd8aa9541b946f3f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is difference though: running in a threadpool limits the number of\nrunning tpool.execute() to the size of the threadpool. This limit does\nnot exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25dc6bc11a67a0d2c682e7789eb76cec990fa6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25dc6bc11a67a0d2c682e7789eb76cec990fa6b8"}]},"branch":"refs/heads/feature/threaded"},"3870600b490c7d7f53a77361520a110f4f7a7909":{"kind":"REWORK","_number":13,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/13"}}},"commit":{"parents":[{"commit":"38bdcc2a8c5a3de37cb72a22a501449f3c458785","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38bdcc2a8c5a3de37cb72a22a501449f3c458785"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is a difference though: running in a threadpool limits the number\nof running tpool.execute() to the size of the threadpool. This limit\ndoes not exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3870600b490c7d7f53a77361520a110f4f7a7909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3870600b490c7d7f53a77361520a110f4f7a7909"}]},"branch":"refs/heads/feature/threaded"},"ede50f4c6d81fb9d50bea8ff9ba5ea7c1d3b0c5d":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/14"}}},"commit":{"parents":[{"commit":"1e7814cf52a6f41c9d8091600fa2c16ef0c5b540","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e7814cf52a6f41c9d8091600fa2c16ef0c5b540"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:18.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is a difference though: running in a threadpool limits the number\nof running tpool.execute() to the size of the threadpool. This limit\ndoes not exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ede50f4c6d81fb9d50bea8ff9ba5ea7c1d3b0c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ede50f4c6d81fb9d50bea8ff9ba5ea7c1d3b0c5d"}]},"branch":"refs/heads/feature/threaded"},"1ff8fa9f51681bbb4ffdbdac42fd156e3afca4b3":{"kind":"NO_CHANGE","_number":15,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/15"}}},"commit":{"parents":[{"commit":"4adb920e4d987bdaabf25153072be9d123b3902f","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4adb920e4d987bdaabf25153072be9d123b3902f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is a difference though: running in a threadpool limits the number\nof running tpool.execute() to the size of the threadpool. This limit\ndoes not exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ff8fa9f51681bbb4ffdbdac42fd156e3afca4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ff8fa9f51681bbb4ffdbdac42fd156e3afca4b3"}]},"branch":"refs/heads/feature/threaded"},"4e0e55adca31eb8a804da7696d9dead82d4db813":{"kind":"REWORK","_number":16,"created":"2026-08-21 05:41:20.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/979578/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/979578/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/979578/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/979578/16"}}},"commit":{"parents":[{"commit":"4eb692541fcfc209ae6208a08fd7c704c1aadec6","subject":"Add replacement for eventlet.pools.Pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eb692541fcfc209ae6208a08fd7c704c1aadec6"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-08 16:36:18.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Add replacement for eventlet.tpool","message":"Add replacement for eventlet.tpool\n\neventlet uses tpool.execute(func) to run the func in a threadpool,\nallowing the current coro to yield so other coros can run in the\nmeantime. When running without eventlet, there is no need to run func in\na separate thread because it is already running within a thread.\n\nThere is a difference though: running in a threadpool limits the number\nof running tpool.execute() to the size of the threadpool. This limit\ndoes not exist when executing directly.\n\nChange-Id: I33f5a9a698b1de5f4b778e28645a9af9d9af6cd2\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e0e55adca31eb8a804da7696d9dead82d4db813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e0e55adca31eb8a804da7696d9dead82d4db813"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
