)]}'
{"id":"openstack%2Fswift~979580","triplet_id":"openstack%2Fswift~feature%2Fthreaded~Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-28 14:31:13.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-28 14:31:13.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-21 08:43:12.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4","subject":"Add replacement for eventlet.GreenPile","status":"MERGED","created":"2026-03-09 12:26:52.000000000","updated":"2026-08-28 14:31:13.000000000","submitted":"2026-08-28 14:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979580-eventlet-removal","meta_rev_id":"f3852152dfded94f5cb2c331ba4a6a57cc191e46","_number":979580,"virtual_id_number":979580,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 14:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-08-28 13:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-08-28 13:03:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:17:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 13:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-29 12:56:54.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-07-09 07:37:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"717a58e783dc86fa3afd0f758847c4ebbdbe302c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 12:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e955971ae29c405911ab8cdb1d4152ae7d7d276b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:17:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b959258942c47caa2c75d1c9c35e301\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67106619e15b4eecbc135338df7210d8 : FAILURE in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7840a26d7421fe6d8b6f3a4c12a384b13d88d1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:38:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34be4854d5274452a568daf80c112a8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124d8f28064d478ca9adfc91ab99a935 : SUCCESS in 6m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6190c71df0c4d5688c65c4ff9a668c9 : SUCCESS in 23m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/174a5b2282274b13860de2d0f181868f : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/103202ea07ef42a6a64160654cd8c44b : SUCCESS in 18m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec4a17ea3e814ef8bdd05c9e638d8c68 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b44e13d3a87f4d13922eff95c7ed9f03 : FAILURE in 25m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b6e57107022419e8396754ba18ee1f6 : FAILURE in 16m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/14823874ec5f4cf98f52f750ae17dd6e : FAILURE in 28m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/30e0607a8be1469dba6e89692dfb1c6e : FAILURE in 12m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ed693575affe4fd6be44fc0125a8979a : SUCCESS in 10m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44ffd8ad858640f4962fdd86eaa257ff : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/621e0389b3c14cfdbda9ae515027f178 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc12ceaf3aa84eae85c7be8c99f685be : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7322879cc821456ca235306bd4ed7062 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/95c65dd9efe04949959500f37a997c63 : SUCCESS in 9m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a36888039004d66aab2a0fe7f315ed0 : SUCCESS in 1h 00m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8b5c6489a84f90a9c9fa392d2343b9 : FAILURE in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89caa3848c7f4f0fa320509f26d52711 : FAILURE in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bfa4f4490774becb0eadcb29474445f : SUCCESS in 27m 28s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/copy.py","accounts_in_message":[],"_revision_number":1},{"id":"58fb4459c99fcf74d9708d4d650a1ef4c78f241e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 17:18:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"723cf65ead2e9507a62235563227cf5029574bc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 17:59:13.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c194388802fc47f681ffe9423d94e866\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf0e86659e564ddb90fd48008786898f : FAILURE in 29m 23s","accounts_in_message":[],"_revision_number":2},{"id":"02528702ebdef1f9d772674fc4dde2a328bbd70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 18:22:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7e14ae5645543dca4373a84d2e6df96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421da8c2ab1141848cb13f061b7e92f1 : SUCCESS in 6m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd3ffb51712b47baa2b357fc890a4d39 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc00fee27aa3479eb7176766ff18c4f5 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e572ebb1e30c4743ac91f193860dbb29 : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ef4f2cc762d40f48226b75941685980 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/12be81de3d3d4ad2914a9c16babb39c5 : SUCCESS in 10m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9249cad86834e098739a7a4c2057bd5 : SUCCESS in 21m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5737cdfb49f844b89e9471390ba6bba0 : SUCCESS in 28m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/756fdeff3d734207906d88993f62227b : SUCCESS in 13m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f3a538c6bcdf40a7bb7435b0236b1ba1 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7da9851557604a7b918e27f75cb5eba9 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/502f2c8dd0ac4436a3b7592376b779b3 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54d2f15a663b46fe912cfdc247e2dff9 : SUCCESS in 6m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a4ade8bd4024e408c63cb96111d5403 : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/062ef791b6304b11b44bfa6a06a04b23 : SUCCESS in 7m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d675b5815bfa4e19b606f4d7c0278362 : SUCCESS in 55m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408bdc1d81e640b2966869f5cdf63eb2 : SUCCESS in 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4454525d16584687a6a9608a0cfc6fe0 : SUCCESS in 2m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0fd2d31d03d4440b9b67114c64cb18c9 : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8faf050fce63a9364eb9476b0014f6927e5201a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-13 18:23:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8eab69bfdeecc3124cd9ebab1eb45c43851d7b16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:01:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1547a4df2e964a738b93dda88aee5e25\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0cac9851bad416babd194213dff1f8c : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":3},{"id":"f937b81f2defeeb89cade232288857fd1e92a4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:23:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/234b6fdaa8994966882e4533598be5c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e152f3bec9ff4d069c7e7767f57813ce : SUCCESS in 8m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7705687641814cc89baccce534347420 : SUCCESS in 10m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/275a9de1ca6e445294bf0a69b0d28320 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a2f7efeb86942ff97fc031b205829d3 : SUCCESS in 12m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/99c8247642cf4715a243f8958b155683 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0bbaeef733eb4af0b60792908295f655 : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7372eff555444cfd9f1dc95c70f9408e : SUCCESS in 14m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e2494a2c837f41c99288e2fd398c383d : SUCCESS in 18m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0cee20230bed46e1bda0c68d0e2e02b7 : SUCCESS in 28m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd7745ce23bc4706abaebebd2a758a51 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/08b16167e5a64dab854de5fff66d0c8d : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/81f05cb551d549818ee582a3ea5b07d2 : SUCCESS in 9m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/700cc1405b15401b989860d0ede31dfe : FAILURE in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa196205e40c49118f610bccbd3fec9e : SUCCESS in 9m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0183b57550944404b6609bedca6f0c27 : SUCCESS in 8m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a096a18d0c3743c2bbfd500ccd83ab88 : SUCCESS in 54m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dd5d132a30343e3ba984f4cc348393f : SUCCESS in 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22dab0bcb19d4485a85413616efd9cc9 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1729a443487b4b2b896a4290e070eb66 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":3},{"id":"3a9af9e59291d3eb29fa49b4eab44316df62ca93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-14 08:49:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2931f6e546960c0d25db2bf0b5b70c30e66556b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 09:24:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/760cddd732af46c4840ac8d8fffdf10b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4355ecb383214c6c84f059d1ff258b85 : SUCCESS in 10m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2194d3a106e405c881438debdb25104 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bc312ec5ef04c7087d9da7434144030 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/978a5ca42fc645d7a35df8fd31452e0e : SUCCESS in 10m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/251c7c07c10d453b91e387594852c755 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a55f6104e30a449fa374ffff46e5367a : SUCCESS in 20m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c692396cf45745daa81d34590ad4cc3a : SUCCESS in 21m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd21d0efa43c48bba0edee1cb6d9a71a : SUCCESS in 21m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5aeea00bd3b84f8596bce7b5241f1b45 : SUCCESS in 10m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/77b2be3b07814451b29c9a565c8940aa : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/80d92b89dde2458292dd19d764a34599 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fb4499596890404a8ade9cd61d0c8614 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0a6dab3bc0a44e4ac9707ab7fe26090 : SUCCESS in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b300033407f4964befdedeb2b3e2149 : FAILURE in 16m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/12d5d18bc2c8477ba6ea4ffbe9dac702 : SUCCESS in 10m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c001893207d949c8b98a87391a56d9ac : SUCCESS in 33m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62c9845aecaf4a118bb4617febbcc88c : SUCCESS in 27m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/552f9a6a2daa45658085ec2218c82bd7 : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6013faa4e7b3476fb6368e9656236ecd : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":4},{"id":"1114d96d8c7bdc1c3bcc5b5224a9f0e7cd8c8dc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 09:58:45.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b93f436daed54cab8941543effb063b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45f95d9bfdb8456da4540eff86d69ce9 : FAILURE in 1h 07m 05s","accounts_in_message":[],"_revision_number":4},{"id":"d36960b357c48c6f40b837e0f43b64caaa97f88b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 08:59:05.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"08ac1fa12f7e74294c73e80e83c2f55fac4bb7af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 09:42:27.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8717b852411f446f976b002b6d4782f8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/049118ce039a485bb70e91bc95ea95dd : FAILURE in 41m 10s","accounts_in_message":[],"_revision_number":5},{"id":"a1b3e7d9d663b05b471b8befa4659bd8bf36a0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 10:10:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d74088301c2f4e879f0ea9937296d226\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a516d13be3648b1a0bc7c3c5b0c02c1 : SUCCESS in 8m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bd6e2ce2f064709a081292c7f3a9596 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ada4c6c7e99d48ccaff3160ac70ec40c : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/635ca757101a44958d25be402bff316b : SUCCESS in 16m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9148786ebe2340e6bff3c1aa522c5415 : FAILURE in 6m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/499fbac559ea4f2f906a3284e50bb5cc : SUCCESS in 20m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc89c848e7f743eea3f7e79ddd990c24 : SUCCESS in 22m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d2f53e2f03e4dde9decce358153e765 : SUCCESS in 11m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e45f46e1aba748d6bedac282cda15864 : SUCCESS in 31m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/af5b049fdb2e4e9ab9b9cd51bc45181a : SUCCESS in 8m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/20c0ad6b60b441aba061d444006500e9 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/02470cbb172b4b01841a73caedf23f5a : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a6111c44f024820a855b8c479218fa4 : SUCCESS in 7m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32412fd3bee548088e55f54f846009a2 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/39af36cbee6d408fb57474b165cf344a : SUCCESS in 8m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c977c87f8b84320a92f91c35bccd6b5 : SUCCESS in 1h 02m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcb207e0688143a296f1b460ab69e496 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8de1606b1f487b8d8500f925e8c41d : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1761c63038734890a1fe80ba87923c55 : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":5},{"id":"88c4a4571dbce6fca6bab248c4a144f4e766ba65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-12 17:00:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e1d9f34562450b14aabf4e0b8b12e61161f4485e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 17:41:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ab440839dd488e814df57f3c08f11e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d97f491ef33d48aa804222b0a0fbae8a : SUCCESS in 29m 33s","accounts_in_message":[],"_revision_number":6},{"id":"c2582fdd968a132a53f93db977c9c4a0a7a052e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 19:15:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7940b52a4ba547fab9184f6b529e782c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6863a16870145528b10f0bbb52706d2 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92f8287d73cf4ee78c49e923e0e2ddc3 : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac5647d0dbc7422596d071df5622f0b2 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/056a447df749447eacc2cc46963226ee : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f535c8763498496fa8ee19fa1dc94941 : SUCCESS in 10m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40f4892d5baf4568bf71655dc9bdce80 : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/851a2b602f674744afcfc66c815ff7fa : SUCCESS in 25m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/582fcc5e3107435c8febaf2f7939b03e : SUCCESS in 22m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/24585b927f564b62bca91d31e02a53c9 : SUCCESS in 31m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/523962f078e344209a9263135518eca3 : SUCCESS in 21m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b619f32fc51a4da8840291064a97a59a : SUCCESS in 8m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/64a7b8b58e3b4fc2bfa2b4701cb278ef : SUCCESS in 16m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4a6bdfd442fc40e88f15a1e21a85922b : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f690546f11e64aaa892957962237e312 : SUCCESS in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c78a864f304641cbbd526a7d64f05a33 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6834788eb5e4d8590cddd289ccc6d98 : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6fdf96418d74000848fb89547f7f598 : TIMED_OUT in 2h 01m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/717b2bafb5ad46fe9c01dd2e7df6b3fa : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a6ec019d084c54aff9972430adb1d8 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4967621250a74765ae298dc773eed170 : SUCCESS in 25m 55s","accounts_in_message":[],"_revision_number":6},{"id":"4d682dc98cb6dbfcc29013d6b55c12fac9a046d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-13 07:37:42.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4dfa380ed3bddd076c405167afbbb49d3a940c6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:14:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce37184857d4ee28343a80af60edcf2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f7d13f4941a4af8aca24d88b7652205 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":7},{"id":"e216fbfa5dcc1ae7e101f6a7fff9f315239cd75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:47:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6721fc6327c341a39525683ea8c743cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c63959f80c4992861ef2a662e1ea43 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ea6726cfae4be7820ed8aa6315b1be : SUCCESS in 6m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac8aa877f7c1419cac432de140676235 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb50cb552c9e44528f38a206535e2639 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/865e9e23967a42b7a71acb60ea41b70a : SUCCESS in 21m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/606c2fdf809f44cdaf569d3df7ef2fe6 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/959c025c8eba4d2b96cf2aa8b3d44989 : SUCCESS in 18m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dafdd65707f8408eb36b94c0bb7ed8a1 : SUCCESS in 21m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/da5d9843287a401b9c0717e166638e16 : SUCCESS in 24m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3dcd4404badc4484b1b2d5a23da96e97 : SUCCESS in 20m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1a1c835575a94995a707cacb7803d560 : SUCCESS in 13m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e70ba370eaae4c02a451250fd36562d0 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/821bb886dd6143f5a197082da1d37b87 : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64203fefaa204a059aa18d67bb4cf6a4 : SUCCESS in 6m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/620145266346471d8eb747212ad43d10 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fcff68d09764ab2a122b90a27e9bf35 : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be8285be86fd41d78c499dfe3122e813 : FAILURE in 1h 02m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/776c555b53e54b4cb464f9141b7db758 : SUCCESS in 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1642d361f7464fb5b88fbf76c5c7fc30 : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9940c7badc6c46119c8d7dc36928c599 : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":7},{"id":"1575fff899120cc5c7ff62abc6a134fdb8e339b2","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-13 08:50:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f0a9ec38a56cc7680687bebe17702f273b0de432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:13:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf20ac31f05943c6a2e5c61333700648\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efd98f98d524f6dba10fc6b8db7f044 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64bb444f88b415784ea21495d5bee7e : SUCCESS in 10m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47d9e611aa2a4e0eaf37fff3e397c3af : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4284febba9b247da951dcd1be751563c : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54ca901bd8024d9ea984699378970bc2 : SUCCESS in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/419faf2fe7474846beff7103b7ce4b78 : SUCCESS in 8m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8a08a92a25414448bb0eef67bb1b8650 : SUCCESS in 11m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e17fd167f5564ce7a0eba86a955d3b4f : SUCCESS in 19m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/26f1612bb1194bdd85b223791e1a82f4 : SUCCESS in 30m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d43785261e14466b964e490f6718a47b : SUCCESS in 21m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e5b977c0ee64476fb891accb117196bd : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f0cc58afa8441f1b5b3a727d7c1bf12 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76e391824c7b4794b691887367c91777 : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b6815fb8104e4737b137150ec1161945 : SUCCESS in 8m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86abef427a524f08b2e93b41b7b35a9e : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/400bdbf2f2004b5681a6877b159a4f3d : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e4fe95d3845463996adf55781991770 : SUCCESS in 1h 03m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd1de3fc5956447aa77cfb4347903b80 : SUCCESS in 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da24468c6bf1475daa4f21ad7330039f : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/175463f343ce44a591991daf1704b28a : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":7},{"id":"e558d384ccf9a9b77a5478406b8789dcab1bc657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:57:04.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/267277b003564138ad5f14f113c337e4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c772f2935d6f4c8ea4001134654f7221 : FAILURE in 38m 14s","accounts_in_message":[],"_revision_number":7},{"id":"d2b1b6507e1f362dfd9da76c2540655d23abec9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"ed32057c326126b16e9ea88ec13ec67b79fd9389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 17:50:38.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9da5074a9bb4e999b1abb473bfc36b6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08e9864b08a44b039ce9669eb3eb6e7a : FAILURE in 35m 09s","accounts_in_message":[],"_revision_number":8},{"id":"d2b71e30475518c61bfd6c69b8455e3703899f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:13:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fac1bf6b39fd4c169279bb788f1c71fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949c94ee317241919e943b33898aa72f : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f70130a33774001adc026cce300d880 : SUCCESS in 7m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad6905b95204418a8ff27896eea68665 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9fcab2a51e24d15b0eca228d7f13764 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/149f786bcb844cecb4d662cd8f527fc4 : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c2b8bdefa15b4dc59461c74909dfcf3d : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/031e60f04fc24828916924c0fa117ec9 : FAILURE in 16m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ca07e2dd82a243eaa9e9968a581b130f : FAILURE in 22m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c655b31105b5456fbe8e23eb9ea25480 : FAILURE in 31m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3edc9c9b92ed4e769837a0ece65bef9d : FAILURE in 21m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/43d56309ba7842aea994157c065d9a5a : SUCCESS in 11m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4cc88af2f6de44b0abdd01d9d92b6701 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ff50e2f474164144a0098be4ebbd4bf5 : SUCCESS in 12m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92d267e18ea8449b80a7268926d88392 : SUCCESS in 7m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6e8567172394cb78d62a7062318e59b : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a0bbc0d3d3a449a38e6347a41e55cf6f : SUCCESS in 11m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e882e5062ec457a9dd5dbfc9b6ce682 : SUCCESS in 50m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9c5064a210d4d7cb1295fbdc56d741a : FAILURE in 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d189be5c574b33a5631e7df27e72ff : FAILURE in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec966b5ea86449dc8516aa4eda51757b : SUCCESS in 18m 32s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_updater.py","accounts_in_message":[],"_revision_number":8},{"id":"0fb7a413c4ef3b20431af6a53559f90365602da7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"924995de934474ae26f60a9113820f9ad1c810e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:37:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2c7ba9e3b2549a8a1ec6fcc0f3ac76a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f74a26096e634a418733ddc1f1b2c599 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":9},{"id":"25e8bb99f2d031bc77de1229bea4cb193f80f134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:42:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d112294d70aa4445b979336ee6c28e3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4503953c090454db0203208f8d32bf9 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17eec18f54e24b10bae194bbba538904 : SUCCESS in 10m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd9b8dfd7d1f4f58a4d09496e50d78c5 : SUCCESS in 11m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4134296599f74880a8080254f5de96df : SUCCESS in 11m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fe9d22bc73f46be8cd52d41cb2107eb : SUCCESS in 12m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0bcdf7a25a24686b367f95a9fb233b1 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0d47888899842f8a2e9256055616ef6 : SUCCESS in 11m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0dc694b403a34cd2a5066ccc68e43818 : SUCCESS in 23m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c23d49a76f1c4f1da29fc02a6ea4dca0 : SUCCESS in 22m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/25176c4226b84a918e461783e6579a00 : SUCCESS in 17m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9c192264a19a4f6484bc0114c14a8393 : SUCCESS in 9m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da46f079e5704e56bc2e40ca48f2bf57 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/661244a3307c40e0aecfd891ed612757 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/733f402f3fcc40cb85b4047963f0cbf3 : SUCCESS in 5m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b49a12ec457b43f6afe2d36f9a1493ca : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d2573f75284b4ff08bcfd07d9c1e3ee6 : SUCCESS in 10m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b90ce57b8f204614b72bf68633e23e2e : SUCCESS in 1h 00m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6819ca3386c4ff388340dd5e21113c5 : SUCCESS in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7878b1671dd45b0814b9abe7cc60df7 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f3ad43b33d445229c5c928e3adbb2d8 : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":9},{"id":"882c5ae83e92eb354ee1ebb93c3629fda90de520","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-31 09:33:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f5c6f35ea55bba8ddc4c914a2c628c6a170f0d02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 10:13:24.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e863b7c29414359989f58d2adefa3bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ab5a0c8ad66426689885f2e10810ea2 : FAILURE in 37m 42s","accounts_in_message":[],"_revision_number":9},{"id":"ca6f4405ea3e73a850f3c7e55dc2d21bcb4b1c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 10:36:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/507d5912c19a40858011a6f86ebe3ee4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61f0444ea6948fb96c96745c551f11a : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e57785de89f74da780492c8c1f6babbc : SUCCESS in 5m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1bcd6e5ab3540c998f91ffe6b1084b9 : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/275edc41424f4d1ca01fad3e2d29f9da : SUCCESS in 12m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a6dfb5e245049a6a4441c711a942f41 : SUCCESS in 12m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/931982979d044d03a9e0ba0409b7a2c7 : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eb82390f207d44c7ad375af3ce11407d : SUCCESS in 19m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2bb90e7b90064f67b058cafc16b0acfe : SUCCESS in 22m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/67d8e9d34ae841c09e7dccef2085963b : SUCCESS in 32m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/874c8948162e41b1911841bef0df647e : SUCCESS in 20m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d0bdaa52ba404805bdf64d9c8adf9aff : SUCCESS in 9m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a9a0bc600b434cb6b5b5d9ac1bdf9e1f : SUCCESS in 11m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2b3ebe48d0ca466e95bdeda3dbecfcdb : SUCCESS in 9m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a36bd015ff0b4555a6ac49caa53c9cf2 : SUCCESS in 6m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d67c64a5939d4846aea5609282052661 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dcb617a8838a4abc860e4349e188abc8 : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3a6af3f171146d7b4ee5e870a403dbb : SUCCESS in 57m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab3670783a284c5ebe1eeeeaac089572 : SUCCESS in 29m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139b52311c1b4a1095f3a142b732c869 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8ab1bdd87354089ac58a6b98d62aa48 : SUCCESS in 25m 26s","accounts_in_message":[],"_revision_number":9},{"id":"b1af5ec4d3a06f67454a6f8276e9956b20cc551c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"732b6d0946e38f30d01206fbe6f4208814268fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:27:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bc1958e284d45c99a214369434c18fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a08c67067240f3a803b713aed7909e : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ead4136d385a4001a07eb2feaff0687c : SUCCESS in 3m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6c677115749439aa931a6fd5bff1501 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bbf0a01b6ba4ab783f1c928bbab156f : SUCCESS in 11m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2488488f702f4c2f97f78d2b5f9b2306 : SUCCESS in 14m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c4a91f416c24870a1243914c2107be2 : SUCCESS in 10m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddefbc26aa4a4a47a0630c4227c48dc8 : FAILURE in 25m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ba768d625a8b4c908161580929a20d00 : FAILURE in 22m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8a87c065d35a45d3aab069a333f260fb : FAILURE in 30m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/68d7692bac0f4238af48166937bc5451 : FAILURE in 9m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0abbd7e45eea41c08bec6f963bd214ab : SUCCESS in 21m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/799958926a2a4060a5f97f0f7a656400 : SUCCESS in 8m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cb7240dfb913434d8fecc0db7a7a1976 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/125c319a0fdd44aaa97ae84a0c93e62e : SUCCESS in 7m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ca8558430fd4eb286074a04213ff15b : SUCCESS in 13m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea9127effc274ca5a88b39cf863879e5 : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/687d3fb3108444be9ea2478abec9368b : SUCCESS in 57m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a17b57a3444504a52830f749b14698 : FAILURE in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bfb45418464451ac0874bb1ab6c93f : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f0ec3463f714e7ead96c43de4714b06 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":10},{"id":"75e3b9b401860408583e5920aca13ad16f18be2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"3b211f993b1399b2a1472a0faa23c30feca9d95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:34:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f73c3e041114fa3b689ec5b0be1e348\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff853c65621a469691d02d10fb3eb42f : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e69c3f1d1844681a92883e803d3beb0 : SUCCESS in 4m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99c3a2c174c54530892d5e05b7eec7e7 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fc3b428291d4fd5bde5337889b8e1dc : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26736168de474817991771ca9c0c2831 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/802b86dfb077429faa292213436d08bf : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4391e484173f415db36d57e0ecbd2f32 : SUCCESS in 25m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3203fb853cc84cca884b76c3d9cb2ef3 : SUCCESS in 21m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4211f6619ed24d36bf80c0d2269fa0bc : SUCCESS in 19m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/44e7e1dd70a1493e895b28ab85e83c1b : SUCCESS in 26m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e89a19b9b69f42d7999aaa452d23361b : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40b5621c28cd462892f3d5562027ce58 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ecae89bfb3b448909ed147676d37a1d2 : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3bbce69691db4462864230e400feaffa : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2cd6ca12c0b844d184506b75dc39014b : SUCCESS in 22m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0143d469df74beb81760cd31b81e026 : SUCCESS in 5m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bae7a496a5a4078aff062e2f2962b00 : SUCCESS in 40m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a2961698bbb4e04a0598bc4989b75ac : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b036cca16e4deeb85501ab158dc882 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21ddc2e67c894175955c8407d9f1ed42 : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":11},{"id":"dd74ca66de81f29bf4fa08d9defe7767098d4613","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:50:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c47162f9ac40c984c91468006e7f0c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7a2a11e0cb4487bb158ffe57ada1e3f : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":11},{"id":"bdd6c153ce2d33261de7b7466548e821912574a6","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-29 12:56:54.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"fcae4420719fb6592d3bbdb4535f924088732fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"d786bbef16d12baa960d0a8b00c0e50a02a07799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 08:39:49.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbb447c95ca4a7f8ff098d2a4d7e277\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3fb8e52684244dd6a3b60af6cf869b40 : FAILURE in 56m 06s","accounts_in_message":[],"_revision_number":12},{"id":"d1cd63e28add5b5c5cb9f9423a74eda9a5caa10a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:33:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239398dd2fc74f4bbe83e8fb30d0a322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48135f62a944898a222899e47d34158 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19942873b9940f4b74f3b48c612ff45 : SUCCESS in 4m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3ed1e8c74674852abcc0c78a4505ccd : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1e911c656fc42eda26c30b2071aa027 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b6c7682d2ce460eab752709df1c5e2d : SUCCESS in 17m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d218893c8744fb385e05b3408370281 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5983910a50d44adaaef2055a561ae91a : SUCCESS in 10m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81f6ecf20f65426791125a4322740175 : SUCCESS in 19m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/229a5525c09f4c40a6845f49ec7db6fc : SUCCESS in 26m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/abc1101b4712401aa69075f538e0bb35 : SUCCESS in 10m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2d470b2e2670406bb7e23cb4a37e26ea : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/92861a86f68b4c39b98377d6869e8d1a : SUCCESS in 18m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fafa16785ec548afbc5e9761b9dfb84f : SUCCESS in 8m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/254f4b65635c4657a777afbfeaee7708 : SUCCESS in 6m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82c479f5ff414892b33da75da9c1655b : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/70d6ff4708e940c3b82f0c4f4e9975cf : SUCCESS in 10m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7750d6bef2b2459fbc52c39e18e93afe : SUCCESS in 59m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57f350ee29ec4c90b1d4a4dc6bb9ac34 : SUCCESS in 29m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6355eb80c85f46779ee2220ebbf9e0fc : SUCCESS in 2m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e263cf345e53432f8beb2ac1c7d01808 : SUCCESS in 18m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"10e35a3d594d44a76c4eccdc8575746dfdec03e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 14:08:32.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"42dc8c50d673f50465540900331258c40ca0baa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:19:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b91b796fddc34eefb02d2e6d77fa0168\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f58a1e6cd9a438d874f92f86cb64562 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d903e01e6f8c42c490c79a732263798a : SUCCESS in 7m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7445d5617e43484d82c130f62667508e : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04977646c83f478c90e0742698e33f94 : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/591122d652384738b4d61a972d612e52 : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e1183d4a7984be29ac0004a218b5275 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85c7a1a518c142dab6d82faca94e60e0 : SUCCESS in 27m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/438f6d2caab94e34a96236d003aa1264 : SUCCESS in 19m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1f6be9884914a02835e82c5a4517aa7 : SUCCESS in 23m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3b564cbc53fd45a1ab0c3c2271b2f50c : SUCCESS in 17m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95922d3622bb4d2e80591c027dd3d6e4 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b5ea07105406448e92239dd537aa7d6f : SUCCESS in 16m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/18e6b25b6b944c2980a465e723a9cfc8 : SUCCESS in 27m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a9a90a420414fc081de603cae6eabe7 : RETRY_LIMIT in 6m 58s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f4d550a619d49dbb9d886a7778a97e4 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41172f11d1564bb2935cb1efccd3009d : SUCCESS in 9m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/80fa61155d2c40c5a33ef7600b8948fc : SUCCESS in 59m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2adf95b27244856abdbaae9e42a010f : SUCCESS in 31m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5d6d3bf1aa4e7e93088460cf1babbc : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5c00b64f5ae4516ae04d819b5939cbb : SUCCESS in 20m 11s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"ea34010f99f0046f345489e039b53d4b41c7d869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 20:08:25.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183ad28923bf4eadae1e8bf96cd6d07a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/198f3c54617d40d5a63e5e618de0d795 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":13},{"id":"971dde794d8d9eeb52731614840360a6fb839aff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-09 07:37:51.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"5fb281bc5ec7b36f6afaff78990f2d9f1c772b8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"70eb1f0017dca2a51a425fdb7978428b9c0f4173","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:16:41.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dcaa3b55bb7450a900dacc610476f82\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37ffd78a4d2a438db3f0f5d76335c983 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":14},{"id":"3e9f2ddebb7402bfdd1dcfc17869a7ea8d0bba84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:30:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/492eddb4eee54e2b9e13cfb77cee5b1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6cfb141ced45d7afad7cdff2882b1f : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d529adeef674d74a000c1cf965be373 : SUCCESS in 8m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4aceddb6d0a346b88092516f74cdf59b : SUCCESS in 12m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e35cc881409e4109bb130e57d57ea848 : SUCCESS in 23m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33706284317b4e589e3a41cfdb547c42 : SUCCESS in 12m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37d7acbfe85d4050bd0ae23ebc98dbbe : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfc9761cf57248d996f41a8a124a683b : SUCCESS in 11m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7ee62dc39a9d4bd58b27fb5cfda5acee : SUCCESS in 21m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/53792c9e0d054e8d8347d97f4ec14cbc : SUCCESS in 22m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5dd9600263e5466a8b38c22fba179683 : SUCCESS in 22m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e80fa69b34074feb800113f460cc4f5b : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7517442f2f354d188b06fedd858d50a7 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d536dac284c34fd3bffce844671ff460 : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55c5b60cb77c4001b71a67d765c4ed68 : RETRY_LIMIT in 5m 37s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/480305c0e6504704adb300dd44e0759e : SUCCESS in 9m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/26ebb5f1763549fc86fc02f3a85b31e5 : SUCCESS in 5m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3091e3eb974a41a08716f0a6276cccb6 : FAILURE in 59m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6f1f703961b4fbcb06ef8c4414dc243 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf30fdba9844216bb0221221c30673a : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d7034abbb944d5197c15f2d3caeaacd : SUCCESS in 11m 00s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"ed816e9bc19236bdfc6db46161965245839de2b9","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 09:16:03.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f77cd4ab615817d7087ba431ccdeabc2ff15c5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 10:24:22.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3b5c3dafd95446fbf0c09256fbff136\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1abd648a4a4a769e3f5ff2512dcb7c : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a62a230ff2d84a13a5189f8a3ec8814e : SUCCESS in 6m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7078bb1935034bf289a7d8ee68ff68b3 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/837d468a97274a34b45afe293bdc9f8f : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea71f3aa6b4c4bea95dcfc70b6565cc4 : SUCCESS in 11m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d70d2f317c74661ac9e6f7b18f0d145 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/10ad36b8a9be4df8890f071b544aa87d : SUCCESS in 11m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e7c10285cce04aa6aea5c7b0e5e5642e : SUCCESS in 17m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b5189167cbaf49adad0aea542d053b65 : SUCCESS in 30m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8bdb06b5a9134c72a1666163af4080a7 : SUCCESS in 10m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a99e142a9ce942bfbf5dfd3aa242803b : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2433d2d98e574debb8922a047503cd18 : SUCCESS in 13m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d3949f4da88945d48a3a2863da1329c3 : SUCCESS in 12m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80920416f0ef4055974ba2e31093f1cd : RETRY_LIMIT in 3m 13s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d182bd7ee2444fa59bff00be48eeb028 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecc6873222a840f18d32c26b552352d8 : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/938dcb5940284064ab3375bcdd4caee1 : FAILURE in 1h 02m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e7514f286de4ceaa5996a7684f8c2aa : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92fe33f3ec7b40ce8e1d994be52035f2 : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f57b74ecea8547ffb4e833d632098a84 : SUCCESS in 18m 09s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"12088dd4fc7e570b541584d8bb82d5c089167810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"4ef1bd156938ed34bac043b74d7a59d0e95a7aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 11:27:40.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbeed388002d40289e815c633d4b6f28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7b49daebd44bc99556489de19a4cfe : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0708c8e089c48a0934ffbad534900d4 : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc693ec5754848cc98feea7ba072f35e : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5021fa64a424693a55e64b0b3dc63b7 : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5aa299ab0da8455994bd11514ef2ae36 : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a65faf4b4fd427889bc587819d30f1c : SUCCESS in 11m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b51ea86781f4b88a3f62dab0238b0f1 : SUCCESS in 12m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3bdd13db161642f8830f90eae3f36c38 : SUCCESS in 11m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8bec7fc3af6d4a1d831a5dc429a83d5a : SUCCESS in 21m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4e5c22bd33aa4afb8d4892df03bd8989 : SUCCESS in 22m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f93609f9de7c45f69a857bc4fdd41a1e : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/46e5b20c1c3f4aeda6993d5360f4a602 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/be33beb78f214b9a98fa815ba0fd01c0 : SUCCESS in 17m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b80ca957e864232ab95ccbe8e505e77 : RETRY_LIMIT in 9m 06s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55419dd487d54bd2a0cb20133242b991 : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e993061000242cda7a867f39969e383 : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/65ebf90aa59b470da5feb1500abf13d1 : SUCCESS in 36m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ccba97dc82640f886332e71ad79b101 : SUCCESS in 29m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/845554fc253641878d6d57f07a8c1f70 : SUCCESS in 2m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8e5732088824cabbc804c3a1c5e1ab7 : SUCCESS in 19m 16s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"e261a9a8bc22d4edb5fd5639047e2bf8caf0da59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent tree, and commit message as Patch Set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"19eb28f7419b69c12bcad7a0998dfd75d27c6da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:21:01.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75412632f5c84a48b86a4e0276de836a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfaa57ad34148ddba4a29a2d5d15c5c : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61394ca9619042dc802446a82570fcec : SUCCESS in 9m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79f16d72103d40608dbd0139b9f3f315 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30390f8c63a542e6833012630b3fcc45 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5375a6f959684830b1b2bc6215a78fbf : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a23d5e176018400188edce2242be9bfb : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46caee8fc8ba48afb7d864b63e2182b0 : SUCCESS in 25m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5b29f7c967304e3ab89b0bbc72d5944f : SUCCESS in 18m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ba44bab740fd4bc2a25af522f1e54769 : SUCCESS in 21m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/749c72e1cf4748a98c1d40e0f94bb563 : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/300a699634d441199262e329c1d8dd18 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/904184aa76584c059fcb95a2dc995fa9 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/47ec5d12690e4e54837896494d3a7586 : SUCCESS in 9m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f6ec3b381f8493d8c57fa035508bab3 : RETRY_LIMIT in 4m 12s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f8c9e791e924f8293053d655e6ccc1e : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a86792eabb9f4ace874abdabbd0c727a : SUCCESS in 9m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1381fb99705547799efb79e071323d7c : SUCCESS in 39m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd6e6bdd3c6c45a69ec861f960c3888c : SUCCESS in 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b428a558a4440ca776e84f01f55956 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7857d1adc7b54a3ca924d3efb0535999 : SUCCESS in 20m 02s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"f7343124975ec3a45fdad88f5d9eb88305a83a5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:48:36.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb454faf84544c9892513c7a9465e402\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3059c632fd1d412f8a545cf1323070f2 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":16},{"id":"2d7f19fe1f0666b4bf41200c5f8d848a533302cb","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-21 10:55:44.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"5f0507757b79c6bb4d4f590ae127e4c5bb7fa815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 05:41:20.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"6dc00014beb52433a137b739d00543aa48588c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 06:53:54.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdccfc26b470422f82a72ad9498105cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b842ee7d7c4649b35faababd252e36 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1020fdd0cd854cf8a6e02f3175c8194c : SUCCESS in 6m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b44ba0ee38624c7d96f91130f44bc12f : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77ab972e937043339695df3a8395d1a1 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1310c90fda474649a7a9edb73434149f : SUCCESS in 14m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73b6f6c9ec49418b9957f940942c5dc8 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/17a52e144ee84251a4fed9681afc7e4b : SUCCESS in 14m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d8f68e93a4234bfba0d0eef9a393c12d : SUCCESS in 22m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/50d0d6bea2bc42b7b7d0a76985a61835 : SUCCESS in 31m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bab05732353b4a1992a16f82f5c4adef : SUCCESS in 15m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/38b480868ed04178b544004040a3e03f : SUCCESS in 8m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/39f29292b7c942cbbf186bd9b34dfc18 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76d8306fc8a24fafa4bc531aa6668716 : SUCCESS in 8m 39s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e94abb43165418ea844338e6b7ea8c9 : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9a76a4d80ed46dda5b1ba47897108f6 : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e33b76b739c648a9b6589c4a83da254e : FAILURE in 7m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c36f7b512ee04ecca830e3831f41e387 : SUCCESS in 8m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fb961d818d849f9b389258c2eafb340 : SUCCESS in 1h 02m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c6dfe7978164228abc378aebcb81286 : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e5bd932ee64a27bac44ee639f627cd : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52b01e8ee9df403ab2832046c5a5a905 : SUCCESS in 26m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"fa3bb53b8388c7d52179ba283544f96fc8b52e15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:52:50.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b23063c5e5145b7a95d77b0d62fcfc2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dace1b0d2967433a925b6e19fcb535c0 : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":17},{"id":"229e9619a7142800695d6bfcd21eb5a91dfabb57","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 08:43:12.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"6d5dc15e5a6e289067b9828a0634601af4053190","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 13:03:35.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"80f83afa845d3e218abd077e379dac55590171f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 13:04:06.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"fede2fd0a04b14d7534ff3eb5abe4a6648e8af13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:31:12.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f804de3d0ddc4b87bbc07d4f938a8783\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8647a5015a2145e594c0c1d2fd99fcf6 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4ab9ebf6f92467681ce999ceef8322b : SUCCESS in 11m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/12d2d6b7559446f68c2001d76524bb11 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ae5138ed66cd44bdafae17d7da5072f0 : SUCCESS in 27m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7b8169cb7e5a43c7aa2389f1172d0660 : SUCCESS in 14m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/880923dfa3d940a9a1a4b5dbd1385dcd : SUCCESS in 25m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7b77c0a261964d418bf217ffa38472af : SUCCESS in 17m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/427a86e0fd1749b199a2a5a7445262d5 : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87be9cec5dbc4ead94b02a88e5733d39 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c0ccbd31d3c24a948f4947ebca49fd84 : SUCCESS in 16m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6df25bea4e9e452b960c55c22d7772f7 : SUCCESS in 4m 25s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1367961d6abb4b95b00f2719a9061538 : SUCCESS in 6m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/731cee2f217f423f980e8928a27e2d85 : SUCCESS in 1h 00m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27d56818e4074976807cfeb393ba4ede : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f9ad2d5a4b4bbe9487974572dd5e0b : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10becd3ad005438ea8adc946f4770080 : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":17},{"id":"f3852152dfded94f5cb2c331ba4a6a57cc191e46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:31:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"462e3ef89fa670185aadb64b1df45ff88b0d5f99","revisions":{"63c9c99606253e42e3ae0ec0be38fe25e4d45188":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:26:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/1"}}},"commit":{"parents":[{"commit":"61c6fd13066c099739da30fc89d1442c125209f0","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61c6fd13066c099739da30fc89d1442c125209f0"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 12:14:04.000000000","tz":60},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63c9c99606253e42e3ae0ec0be38fe25e4d45188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63c9c99606253e42e3ae0ec0be38fe25e4d45188"}]},"branch":"refs/heads/feature/threaded"},"fa0615536348370181d398594f9d85f3846da094":{"kind":"REWORK","_number":2,"created":"2026-03-09 17:18:43.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/2"}}},"commit":{"parents":[{"commit":"7e7fe96fbb2092d6d3cdad4ceeafa4319e8c73a5","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e7fe96fbb2092d6d3cdad4ceeafa4319e8c73a5"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 16:23:57.000000000","tz":60},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa0615536348370181d398594f9d85f3846da094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa0615536348370181d398594f9d85f3846da094"}]},"branch":"refs/heads/feature/threaded"},"32fb8eb010de89b3e274b96140445ac711cc8518":{"kind":"REWORK","_number":3,"created":"2026-03-13 18:23:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/3"}}},"commit":{"parents":[{"commit":"9bf6207a05f1ade1073e3bed6ca8450451583b1b","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bf6207a05f1ade1073e3bed6ca8450451583b1b"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 17:12:33.000000000","tz":60},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32fb8eb010de89b3e274b96140445ac711cc8518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32fb8eb010de89b3e274b96140445ac711cc8518"}]},"branch":"refs/heads/feature/threaded"},"7d4f69ec250e17d78e53743a4d142528353e255f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-14 08:49:50.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/4"}}},"commit":{"parents":[{"commit":"94fc3b2ffe4e36b66e409d5ef1a191a85e00edd1","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94fc3b2ffe4e36b66e409d5ef1a191a85e00edd1"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-14 08:44:43.000000000","tz":60},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d4f69ec250e17d78e53743a4d142528353e255f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d4f69ec250e17d78e53743a4d142528353e255f"}]},"branch":"refs/heads/feature/threaded"},"32ab7f65d1cf7691d53093aaade5fce55ebb5578":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-23 08:59:05.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/5"}}},"commit":{"parents":[{"commit":"00b64102cb7a4fbf246154fa0cf97966d855125e","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00b64102cb7a4fbf246154fa0cf97966d855125e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 08:40:22.000000000","tz":60},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32ab7f65d1cf7691d53093aaade5fce55ebb5578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32ab7f65d1cf7691d53093aaade5fce55ebb5578"}]},"branch":"refs/heads/feature/threaded"},"485fdee896cad5647664e995fcebe38b7e236a4f":{"kind":"REWORK","_number":6,"created":"2026-05-12 17:00:27.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/6"}}},"commit":{"parents":[{"commit":"b3fabc1d996c9a9fd3027d6041bcbf52180fab8a","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3fabc1d996c9a9fd3027d6041bcbf52180fab8a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-12 16:05:27.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/485fdee896cad5647664e995fcebe38b7e236a4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/485fdee896cad5647664e995fcebe38b7e236a4f"}]},"branch":"refs/heads/feature/threaded"},"01f235fc22a90d6aa728aef630caefa7900f5a7d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-13 07:37:42.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/7"}}},"commit":{"parents":[{"commit":"73155f88f75feb525be799712fb892c2e274ac27","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73155f88f75feb525be799712fb892c2e274ac27"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-13 07:02:43.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01f235fc22a90d6aa728aef630caefa7900f5a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01f235fc22a90d6aa728aef630caefa7900f5a7d"}]},"branch":"refs/heads/feature/threaded"},"f1796cd7e3d8f2d37b2daa33ae9be450eb9dfee3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/8"}}},"commit":{"parents":[{"commit":"8c5e32a24df46a08b02ebe03a74ac3246432aaa8","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c5e32a24df46a08b02ebe03a74ac3246432aaa8"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1796cd7e3d8f2d37b2daa33ae9be450eb9dfee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1796cd7e3d8f2d37b2daa33ae9be450eb9dfee3"}]},"branch":"refs/heads/feature/threaded"},"9d000bebe16ee9379d0be1c936c27e47ea7fd6a9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/9"}}},"commit":{"parents":[{"commit":"1255d0ca75ff46b932bce10a4841bf575779e531","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1255d0ca75ff46b932bce10a4841bf575779e531"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d000bebe16ee9379d0be1c936c27e47ea7fd6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d000bebe16ee9379d0be1c936c27e47ea7fd6a9"}]},"branch":"refs/heads/feature/threaded"},"82d0c7a6ef715a9629983fed5f7dbfaa5c92a2da":{"kind":"REWORK","_number":10,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/10"}}},"commit":{"parents":[{"commit":"b67471cf7e3dcecc7cf09c707828583a501f65f6","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b67471cf7e3dcecc7cf09c707828583a501f65f6"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:16:16.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82d0c7a6ef715a9629983fed5f7dbfaa5c92a2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82d0c7a6ef715a9629983fed5f7dbfaa5c92a2da"}]},"branch":"refs/heads/feature/threaded"},"98c9e5cee4bb7f34251676af50a78a3822d09276":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/11"}}},"commit":{"parents":[{"commit":"2b31276185ee43730c602626c5652c66d5c99477","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b31276185ee43730c602626c5652c66d5c99477"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 15:01:27.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98c9e5cee4bb7f34251676af50a78a3822d09276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98c9e5cee4bb7f34251676af50a78a3822d09276"}]},"branch":"refs/heads/feature/threaded"},"e3578971cc13ab57fcc67168a268402d69165b17":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/12"}}},"commit":{"parents":[{"commit":"eac0ec7da699937078da72718227b03811a74755","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eac0ec7da699937078da72718227b03811a74755"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3578971cc13ab57fcc67168a268402d69165b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3578971cc13ab57fcc67168a268402d69165b17"}]},"branch":"refs/heads/feature/threaded"},"2dbe1572edb4be06a1daf595b767d85bb02ae1bd":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-07-07 14:08:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/13"}}},"commit":{"parents":[{"commit":"8e29f7d821204efb1dde14a37ee4a6dcafaa74a7","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e29f7d821204efb1dde14a37ee4a6dcafaa74a7"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile that wraps ThreadPoolExecutor with\nGreenPile-compatible semantics, spawning jobs into a thread pool and yielding\nresults in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dbe1572edb4be06a1daf595b767d85bb02ae1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dbe1572edb4be06a1daf595b767d85bb02ae1bd"}]},"branch":"refs/heads/feature/threaded"},"b2b7c6ce7d50524dfb9cebf47f532e685496a8de":{"kind":"REWORK","_number":14,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/14"}}},"commit":{"parents":[{"commit":"a8505028c7cb4103bce56e9af392a28c5777d0e3","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8505028c7cb4103bce56e9af392a28c5777d0e3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile backed by SwiftPool, spawning jobs into\nthe daemon-worker-thread pool and yielding results in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nGreenAsyncPile\u0027s result queue also switches to make_pile_queue() here so\nan abandoned pile cannot strand native workers on a bounded queue.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2b7c6ce7d50524dfb9cebf47f532e685496a8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2b7c6ce7d50524dfb9cebf47f532e685496a8de"}]},"branch":"refs/heads/feature/threaded"},"208e850a8766ee78c647152598332d2ab93d6c18":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/15"}}},"commit":{"parents":[{"commit":"0d8f60eaeb97b12cae07ecb02aa4f6edf75b4252","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d8f60eaeb97b12cae07ecb02aa4f6edf75b4252"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:18.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile backed by SwiftPool, spawning jobs into\nthe daemon-worker-thread pool and yielding results in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nGreenAsyncPile\u0027s result queue also switches to make_pile_queue() here so\nan abandoned pile cannot strand native workers on a bounded queue.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/208e850a8766ee78c647152598332d2ab93d6c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/208e850a8766ee78c647152598332d2ab93d6c18"}]},"branch":"refs/heads/feature/threaded"},"16a5d2028cad206b439758d840c4e1cfda095f7c":{"kind":"NO_CHANGE","_number":16,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/16"}}},"commit":{"parents":[{"commit":"e38f13dacfd49484337626589fb10dd2a41f7b0f","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e38f13dacfd49484337626589fb10dd2a41f7b0f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile backed by SwiftPool, spawning jobs into\nthe daemon-worker-thread pool and yielding results in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nGreenAsyncPile\u0027s result queue also switches to make_pile_queue() here so\nan abandoned pile cannot strand native workers on a bounded queue.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16a5d2028cad206b439758d840c4e1cfda095f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16a5d2028cad206b439758d840c4e1cfda095f7c"}]},"branch":"refs/heads/feature/threaded"},"462e3ef89fa670185aadb64b1df45ff88b0d5f99":{"kind":"REWORK","_number":17,"created":"2026-08-21 05:41:20.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/80/979580/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/979580/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/979580/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/979580/17"}}},"commit":{"parents":[{"commit":"da034069ff285bef87bfd9ffdb94f0a94881c8fb","subject":"Add replacement for eventlet.GreenPool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da034069ff285bef87bfd9ffdb94f0a94881c8fb"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-12-09 06:21:19.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Add replacement for eventlet.GreenPile","message":"Add replacement for eventlet.GreenPile\n\nAdds replacement class SwiftPile backed by SwiftPool, spawning jobs into\nthe daemon-worker-thread pool and yielding results in submission order.\n\nWhen eventlet is enabled, SwiftPile is aliased to eventlet.GreenPile.\nWhen disabled, the thread-backed implementation is used.\n\nGreenAsyncPile\u0027s result queue also switches to make_pile_queue() here so\nan abandoned pile cannot strand native workers on a bounded queue.\n\nChange-Id: Ibb03fcf4adb3e279cc4f9b0291439adb9fd350c4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462e3ef89fa670185aadb64b1df45ff88b0d5f99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462e3ef89fa670185aadb64b1df45ff88b0d5f99"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
