)]}'
{"id":"openstack%2Fswift~979584","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I12d464419435f4a9d133445342fb5f608b9e20ab","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-07-14 11:43:21.000000000","reason":"\u003cGERRIT_ACCOUNT_38496\u003e replied on the change","reason_account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-28 14:35:24.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"I12d464419435f4a9d133445342fb5f608b9e20ab","subject":"Finish standard-module shim call-site conversions","status":"MERGED","created":"2026-03-09 12:26:52.000000000","updated":"2026-08-28 15:45:54.000000000","submitted":"2026-08-28 15:44:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979584-eventlet-removal","meta_rev_id":"3c3e1bd1e902b3fba96e3837a5c72b5423abeb88","_number":979584,"virtual_id_number":979584,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 15:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-08-21 06:02:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-08-28 14:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-28 14:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 13:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 12:31:47.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"7c2d3b8d62b0ac1f9cb112ce7cf27c091889fb9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 12:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1819c46337b0fd8fa7dc7ca9009667f4c6a3e542","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:10:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d95526b7a0c4748af999ede58afd37d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b763bfbaadd74544b3627d599086065b : FAILURE in 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"603789f9ac27f05578c703378f66024ef883a38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:37:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88abd18f8f7a4552a2aa6609a9f20e66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a78cd671f24c498cbad12266c33414 : SUCCESS in 10m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7395d2f51c14524b493811b12675954 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40a760b9be2348128c69fe45614347b3 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/122444172f384c8e8ce5e500e2193a5e : SUCCESS in 22m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d80473d310745828ecea18e58edbc3c : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1fbb5e49849a4569a7bc7702e9f13fdf : FAILURE in 25m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/596efd2a6c1d4dd59f0036c5f2b65b2a : FAILURE in 21m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dbe2ac889d9a432d8ce9bd5185a149b5 : FAILURE in 24m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/23e879860a6e4acfb5914783b636ac9c : FAILURE in 26m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b6ac4b0ec6d6482387eab366dba960f8 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f973621be2714f449487d97d4e6b351b : SUCCESS in 8m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/71fee4a4842146b498387a48cff9a90f : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/315273b511ba440887470123f0d726e6 : SUCCESS in 7m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/838ffeb9edf74d4ea26c192e78f90402 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7866e3919e20463d9b3b4229b86e86f2 : SUCCESS in 5m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a25a223dd52b4d67b32be6dbf89c066e : SUCCESS in 55m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c04d36bc06b14dc8812b258da231b241 : FAILURE in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0e3b6e1c524830b262b4e342dda782 : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c529b3bc23234549bc8a2e860629d3e9 : SUCCESS in 18m 32s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/copy.py\n  Comments left for invalid file test/unit/common/test_threadpool.py","accounts_in_message":[],"_revision_number":1},{"id":"0b9eac08904524c226ff85ad7d5bd5c378903444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 17:18:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7eb99ca9849de5572ede6413f5fd99889c4770bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 17:51:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f138689693f47439cc7abb8d5424448\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/860aef358e3a42ce9f9645626ce9efbe : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":2},{"id":"91c34dc41f7d8ef879219ea558e2de12cd8acf4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 18:23:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7fce73d8616486cae1d65f3635b5a22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e54e33ad094098afc2e46f94c68409 : SUCCESS in 10m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/200a6ce706d64cb0bbfea0672a0fd98e : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0de28a3f45604f3282941b3dce6ed429 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/288bc975c9194ffaa15938ffb8d159f3 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1900f92bbed54ea3b05353f3127a5a37 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb1040c26bd5441fbe35b491f8b68168 : SUCCESS in 21m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc25726d4b214b6d88e3fe742ceea854 : SUCCESS in 8m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1fbbb04ff45e4c3ea194d1e608777135 : SUCCESS in 27m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1f5cc9080a254986902d6022c96363fa : SUCCESS in 28m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3c1cf7fbc63f479e83ae87cd4c91c021 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7c5af33c02544203a2ab659bfe1bdf4d : SUCCESS in 9m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8b8b218090b848589dd998a9584b9c7b : SUCCESS in 13m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96ede8e63bb543339f505519ac901cdb : SUCCESS in 7m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3052301b3414df28ea48af1871d93c2 : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e5c5a4326fd41fcacecb6965838e497 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24ecd14a557149abac091e0d5af4d6c2 : SUCCESS in 57m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bedb68212ba3461790289d60555392f7 : SUCCESS in 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356e079fe48a4596af253e2f55648f5d : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e64d06f9d98240b7864ea7eaa28006db : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4f0dc34f02d23bb13950fe1cda1814eb771e7b80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-14 10:22:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e881dfaea0712fd2748b2a341000c1b9ab3beb54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 10:55:14.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f3723fced04a72a15ae6ca3cc2b545\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81cdb67815db421a94d92b0c6da258cb : FAILURE in 31m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4359594767d0d85d4cfb805d4f75197490dcc4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 10:56:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a895344766e463ea20f456e8a6f1646\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5afff74283e434dbef3f5da889a6dda : SUCCESS in 10m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2eed25de6aab4b77adafbf5d92c0cf8f : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d92c29b425742cb9bb725c4497f8837 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/757e60fd06594d9daf81b9045e385a90 : SUCCESS in 15m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab391beac04d4338bd63434fa9abba61 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2a1dd9b9bc947bab422231e2384de3e : SUCCESS in 12m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/42c909bbfb714678b73634db4f3e70e8 : SUCCESS in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b038fc64babc4efeb8c87afb4c139e26 : SUCCESS in 27m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c30988c2ca4746cead1986590b1f48d2 : SUCCESS in 19m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/720efba1152c40218bb6cf8ab877f4d2 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1d592d0222b24fb4bfd54efde65b309d : SUCCESS in 9m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c70681925a964c729c85ef5e3c2cfde8 : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc79f9a48b824e54892fc6899fe93b77 : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b3731dd7efa48e097813a5ba7f0c722 : FAILURE in 8m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ea12aaef4a84c609be2733cb815fb5f : SUCCESS in 8m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/03ea2e5b3adf47d4a15ffc7717282c6e : SUCCESS in 32m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36df25b356740c78bd1b16105003c7d : SUCCESS in 27m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8c34d227a946cfba44e94a27eda194 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e32b59f727564a2aa7843621af464b4a : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":3},{"id":"fa0218a710bd7e75d7cab536ef2f0d6d6a47fedf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 10:33:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"380f39b1ace78d0fcca938693ccba966f64b4314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 10:55:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"954302f054aa4874909851a3a10e2b3015e7940b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:02:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7ab43555ee4325ae6d82b56c4aebd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6b3b2c940b4c4fbc9b47a182387982 : SUCCESS in 10m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7f472b702bd4b148567eac508d10f18 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/081dd57014934097a3bf2dc38610c33e : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1445aa1b1dab47259dd6fe5c6c39ca5c : SUCCESS in 12m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43d65659e3b14248bd794aa70ec48c92 : FAILURE in 5m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ebb36e31aa40389daace6681025a1f : SUCCESS in 25m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a6dee807b5f4889b10cc33b352c301c : SUCCESS in 23m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/77513bd65f054e7396d413d754365c4b : SUCCESS in 13m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d4287a45cd9c48e0abf9c24afd3e91ee : SUCCESS in 25m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cba2bc4eaecd460ca9cbb14eec50a022 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b04b6c4b07f14815ac9dc7c1f7c41e43 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4fa74bb426e14faf8e054047c3e3b553 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/721b616c58e742febaaa9c2e8706e2d7 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3062eb078074611850f250d0664cead : FAILURE in 17m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3db5fa3a8b484dec935a06286286df71 : SUCCESS in 9m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9cbdef9e2c440d0be4932ce2cb6635d : SUCCESS in 53m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c2be03ed2644af0be4b8fb4313781b1 : SUCCESS in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eba0c5ca1414310a2fb2fdabafc0507 : SUCCESS in 3m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e07f9c904def4f60afd6676152ef0855 : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":5},{"id":"ee290c59e27ec303ce068a9d470846ffb2ca7375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:30:41.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e96f9e34b82499e897d145fdab969d0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ff22de0837d4c9e94d573adca287f2d : FAILURE in 39m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f0afe4b25180e016f371dd99afd3131a49f6291f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-13 09:16:18.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"92b8fc0c42d39c9dcea7afa4cc9293597e0ed173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:17:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd9f984e98224f85be8eed40a6a8ec9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c930d21f406c4d44bc8c46ae21ae8165 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec39226380724676828814395ae5b05f : SUCCESS in 3m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a057c60ad1684cb6a89ef5eaf2c977f3 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9abfcfef20d04d748f5163ddd839d377 : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d117b2f33fa940f590d7084ad227fc50 : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44dc984c9025428da52c40f83c140dd6 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4488305d01a64941a4f77822a9e4621f : SUCCESS in 24m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d281adfa85bb4d139a407f015c3a1ba3 : SUCCESS in 18m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1ced511eaca047aaaa7c5595e742167d : SUCCESS in 20m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/983edddac3ac4bccacdbc2c125814b67 : SUCCESS in 19m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94f4c4677a9741bd9e6c18137b4cb2e5 : SUCCESS in 8m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/143ad80a81ca4b12a031ae5e7a6ba051 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/38afe97afe9343f3a94c7a425cd7505b : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac890656538d468e94cb8854d09483f7 : SUCCESS in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f964a542a3e5483391600d825b044787 : FAILURE in 20m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b64fb88de0384f6e85a0308eaf213c8d : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/359929275cd94ae785d52ac341240ae0 : SUCCESS in 58m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0432acd72fc4d4e9f6833d191aa9e5b : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6773733f4b545b9b401400675fd3073 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd249c7dfa4d4ff9b84573ccabca1c9c : SUCCESS in 21m 40s","accounts_in_message":[],"_revision_number":6},{"id":"1dfa2a07a3b4c9db9f66039791fdc63c0129cc85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 11:37:00.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7692a69a28694609b369c2bbf8a945af\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31b9adeea9024dbcb70d757c93d21077 : FAILURE in 35m 28s","accounts_in_message":[],"_revision_number":6},{"id":"d70f5d756e150e6b4f81c9d3ae11c9c2fe39fd92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b316256272e930b3bf9ba84d9dc5524592016ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:21:27.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a127adc35cd42749b2b541cd0848e7f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/017508b0b84349edb0c8c1bafee07d28 : FAILURE in 1h 05m 18s","accounts_in_message":[],"_revision_number":7},{"id":"b0edc168fd0eb51f25d0337464839c12d058395f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"3432fe313adf00e66855885f67b8617ca276fa5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:44:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8954ab24350542e8b197fb033a0eab83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f672bfda0fa14f76a47c441eb7da2100 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5deb3dde74a4881ad5561b30e70719d : SUCCESS in 8m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fd53794945f417cb351fabdf625fe71 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e470215fc61445389ea6fd1b2b113d88 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12c8fdb198224b29aa4027c1e2f0b5fa : SUCCESS in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6016fdac80934fc499afa61be3976b32 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/77c4ae071b324743bbad30930d83a719 : SUCCESS in 26m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9333f7a2e92847cf84188bb2a7a025ba : SUCCESS in 23m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7c9ac6505adf4a13b25db9b063b33e1a : SUCCESS in 30m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0dd759b307d7487fbf7b4bf37b536a8e : SUCCESS in 21m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b04cd473ba9147a09b18b19e1e88c7af : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cb9c8583b2c9477c98249e5e342c5fd6 : SUCCESS in 8m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/50bee3a261314673b672b4095b3fbfe7 : SUCCESS in 16m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ba506358585417580dba5c22680d6bf : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1cb47a83597c41a6884fdc5c478eb3ae : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eed5df95e6b342d4a6d621c696fccfc1 : SUCCESS in 9m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/219aa374a41c4c49a61d017bd4b3b303 : SUCCESS in 59m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0a18f16968343868bb628e7c8d7e7ed : SUCCESS in 29m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7b9b6056a0a494bb365bdaaab50d2ed : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5a91e0876074c8fbce389f2ee2af865 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":8},{"id":"eaeed21ba5318a9d50141693bcdc2ba6c91a7f5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:58:32.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f25d441e0c7475b9af74d0b43654c1f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/790abd99a569438f93388a51fd97eeea : FAILURE in 1h 19m 11s","accounts_in_message":[],"_revision_number":8},{"id":"7ff5dc86807bbb54ed3d9389bf1e8be98fd998d2","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-11 12:31:47.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"08cc196dacd1f736ac27ecd53165bb3fb218db28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"e3dc64e618ae9fd26b90a81148724059646a981e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:31:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81fc9fd279a34fb99045df335d43e7ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b79b9e401e4f239196b03999356882 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74dc22d4f9aa493ba2f00e18c0101594 : SUCCESS in 9m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d6e71e8224342b39c0f44e99d660120 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a112682d381f4cbaa0878b63e24094db : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62cd8d4f45994197a137cae9ab58b1df : FAILURE in 26m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2df8ae8626cd4489952ccc8e5d32f93e : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40300d9ef10c4e48bd289eb22626a47c : FAILURE in 26m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1744527acfac463c9325c0f4a1d71e1d : FAILURE in 9m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e93b696b9384de1ab24af3605407f70 : FAILURE in 26m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/41ac36f2f73a4195a76004ae3f33f5f8 : FAILURE in 10m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6bb786a3f28041669c89206757dcd768 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f3e8889aa6bb4327b1dc80c74606bd19 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4eb35fab06164d7884f29d8cbeb09842 : SUCCESS in 17m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/00b99429e37e4e3aaee7408a3be138d7 : SUCCESS in 7m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44405d0d0bf44bad8a14378e3835f849 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7feec05cc729492f946e43374c7ab27f : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/81d63f29a0d54313811696ca709ade17 : SUCCESS in 59m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/454a894055a24405ade74fac5c2a41e7 : FAILURE in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03bf96da376c4dea8782c787ff8f4b04 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7218ea521ef146bab0807ccafa437251 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":9},{"id":"8de9ab3f2042570d84b0c7f238b588d00a9240da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a0b7f2e203765c1bc56a75564b136fab42383f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:44:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559f6dff3a7941fdbba86c348f5c421a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caaf0763ab28427c83873d688298aabe : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1115e40a530495a9fc7be25a6c58116 : SUCCESS in 8m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/797d9fa2ab6d49b3b70e646c00e6b491 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/628b65546a754405bd5dc581920b2b46 : SUCCESS in 21m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/748551f4a2dd403e886f68e2b716cb1a : SUCCESS in 11m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a99c422da564a729a086d83e7ae3f5f : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dd71ba452b6a45078fe5ee42ae7452e0 : SUCCESS in 25m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1e729fe3401d45f3af118af5b5e84cf6 : SUCCESS in 21m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/321f96e9d99d492e8e13800961ea06e1 : SUCCESS in 15m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6c9106827dba4a2fa87a61711e2dc525 : SUCCESS in 9m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/00bdfe3c5a594e298ec20402360c2db6 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d0ef86c35584e5aae96dc31c26d1acd : SUCCESS in 9m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/efc89975ccd0405d80d1661a04e6a22f : SUCCESS in 8m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e6ab48b1b79461b865f877c9882f3b6 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc847d4fad4d4ee5af33d432b86b4de6 : FAILURE in 9m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3bce275210d64440a5093d7fc0e1ff9d : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e2ac7375d7449a788e62332fb37a7bf : SUCCESS in 36m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc3ba1fb7d954df9aeda443f91b2e90f : SUCCESS in 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ccb5bc19bda441b950ec021a35c84aa : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/362380ef687c4fffae4bcd1d24f0ce0c : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":10},{"id":"db340b11aea9b99d8b61e7c51177cbb35343a0dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:57:44.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb11541b8ad24c5eb8be503b6adef876\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8a4206f2d5a488fa16c89bb6f520701 : FAILURE in 58m 36s","accounts_in_message":[],"_revision_number":10},{"id":"efad8641900f1494d1e8f50d6710d8b6ff6ea67e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"a8770a16b08fd1363af8b46ebdc0528e5a29ebce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 08:54:55.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f955af330c74d11a243c00826834642\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d9e2619e894a461099c9d53aa6c0e0d3 : FAILURE in 59m 03s","accounts_in_message":[],"_revision_number":11},{"id":"e77e04dfefe8db76678e3b4c02685a704845e5d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:36:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79bae71bccdc4e34b0f8f1342cabdc8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f6b77e1ded4a889fdf208725b4d8d0 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0ec84cb18744c1937ea0fbfbac92aa : SUCCESS in 8m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab6cc3215957467dbef1938cd7a08fb1 : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c267ab4eea6d43ebabdf2f0924d644d3 : SUCCESS in 12m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fa8b8db6b654207925f6db3ddaf9cc0 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0017cd7dc03c42dbb6377a850b62ee10 : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f5830e23518e452187ff2422a2bd2545 : SUCCESS in 18m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f561b8b911b143e9b3ba6c74008a76f9 : SUCCESS in 10m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/33c2a5f7e36b4b189c7b4d101bee57aa : SUCCESS in 31m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5163f06f601a46139eae2fa0b17e9095 : SUCCESS in 21m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/32612d884adb45debff25ea2dce5636b : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40de7d1c6ab74b99bf4e4810c8655248 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d1e9ce62aa46441ab5020cb0af4de9e7 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0f4f39523854224ba79a3d5f469840c : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1899a7cc93404ab29fdfc0e328c69655 : FAILURE in 13m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9499e4d3a2884f7894696c15c5d134de : SUCCESS in 8m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/faedd1b670d74f4d9648d2fd38fd1a6e : SUCCESS in 58m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c72490e80d344e4284f267c581dac6fa : SUCCESS in 24m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923825138e1a4ae79a519c4c8e99211d : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1bdeb5e56b046328787fcaa9c023ef7 : SUCCESS in 11m 29s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"ab084617f279097bac352a7c56cf06390dc4b6c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 17:22:32.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"e77d54609b8a52c75ceb9f6cd87dc46376c02eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 18:52:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf695248717a4c299a0575d3824368e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be468b1dbc434f70a41a9ca74285de63 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b463354e09b432091962451fbdf4065 : SUCCESS in 11m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4bcb4a055734606b25072078d4590b2 : SUCCESS in 24m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7027e9db69354c33b3bccec89bafaa84 : SUCCESS in 22m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2f1f4046d604a03a25e7939f05e92df : SUCCESS in 12m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce9160e1b57e407aa911511d43cdd059 : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0213e33b46a2406fb3d52a552093835a : SUCCESS in 30m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/86ed86edae9c49489a6263b52827bbb0 : SUCCESS in 21m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78487b2311464d13ae160c59dd1d6d82 : SUCCESS in 32m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/39bdc0a538ba4a8db6ad064b47e7fe8c : SUCCESS in 22m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d8c18947f5254c8ea6ca1a5638144996 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/23fe2d72e9c54004afacccba5b2761bb : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/60ef64f05b2e44bfacdd30085bb2cfef : SUCCESS in 18m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cfead86f12474c3ba0cd6bb71fa0e5ea : RETRY_LIMIT in 4m 47s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b851dc44e7a6443cbea2369688b78886 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1eedac78db3144988e4ee1847da361cb : SUCCESS in 10m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4eea2b75e82a41b3b4d7a1fc32e9b86f : FAILURE in 1h 06m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41941f912bbb47c0a31312cd86fb7b53 : SUCCESS in 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf8f8ebf2044b418653a04c10e884e1 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ded9393700f4c56a54e9c11c3ab6bb0 : SUCCESS in 20m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"d5798b64bda287dbb7598b765f3cfa63e9843ed0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 00:33:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b661ccbf4e4122a048f05e1765be47\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/992992082cdc4d4ba1609ea4c02e3625 : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":12},{"id":"265685247cbe94ad9b337bcb91b19f19ec7fe8a1","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-08 06:41:33.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"444701c7c7b24ee24eebdff0c8d16babda4ffd96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 07:47:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16e3b653d8fc42588f33b84895e4663e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc90cf063fc34fe28cb8fc676827dc66 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bee4af9118d04e3897d04eb6321603aa : SUCCESS in 9m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe65993499f1481685caa3718686598c : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e3540e4f0fc4be2bd7853e3af37f22c : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bf780cfb3644fb88bb312ff3457e672 : SUCCESS in 15m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/06e0f6a0d468480885e08f53759429b8 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dd8bb6e76d424fd3b45958d04a298e93 : SUCCESS in 24m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0e5f803ff09c4078a7bdd53b450e8024 : SUCCESS in 13m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c419a8c634b442e89989f5f5452f812d : SUCCESS in 12m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/91cb289ffd5049b3a5b0353afc6e3c7f : SUCCESS in 21m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0624f03c521f42f8aaeb07f333270964 : SUCCESS in 8m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b2b8fafa2f0c4e9193823c696f883c49 : SUCCESS in 16m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c935d57e5294b7f934c44a80298d91b : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9b02842a970486cb1fefaced3035ac3 : RETRY_LIMIT in 5m 15s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/80f8e346838547fd98ed4b49f154c0b7 : FAILURE in 20m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad318efe0a3f437a8b803101793bff5d : SUCCESS in 5m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d02bd0513f0496ba243f18c6ab8ada4 : SUCCESS in 55m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/648f96e1e2784f2b825116091f23a6b1 : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a543b94b46b4037a4e01c26c8f143b7 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2586db1973a24e51aa50aeed4364fa51 : SUCCESS in 18m 37s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"21aeac6312048f19a40931b0c51165c42e31afea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 13:10:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cea998d84aa4242a8a0882a0d377d95\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/240086b78c764dd5adc8a2c1c7d8c986 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":12},{"id":"41293d18d2de6dc50f97f545c2a3abf625adbf7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"5a43dfa52b3e32a8a183c7e638cb5e2d72d396ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:18:01.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc9a778fd704fb9893a4fccdc584483\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebc9683e863a43e3afe05315fd70d5b2 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":13},{"id":"a5b84d5aafe5398c45854701ba87acbf358652ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:27:50.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f69a442e0845779476c071eb600c22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe1155aa0db4c428842536cb2c236e5 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c36a8a2bfb74a3d8ad2a30dec3e774d : SUCCESS in 4m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3e2eedea05740b59536a0ae84ad06f8 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72b5fae755d247afad535d1d05b26a04 : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df53599224b748c3862dce00126bc0b0 : SUCCESS in 22m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2024692113c84a739fb44a8a0e29ba97 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f3337af94c334bcfac56e2dd1b0350fb : SUCCESS in 28m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef4c2210222e444f9a47347b0f885db2 : SUCCESS in 21m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/49651bbbd05a427391989142da4412e3 : SUCCESS in 12m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/756025facf0d493e95b7c5657a055cc5 : SUCCESS in 10m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/596669acb73444edaae06cefebf46359 : SUCCESS in 8m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/15720275b69f4d9ca47775e12660d236 : SUCCESS in 10m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3499b620618e4b789758fbfc63a7e5ff : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe515df4e61041bc9c7091c6f4d07326 : RETRY_LIMIT in 3m 35s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81ed1373568d453189b236b264d6d061 : FAILURE in 10m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1477a8f159854289918cab83e88822c6 : SUCCESS in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cf09399f54441c2b3fb10f151110c82 : SUCCESS in 53m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe2209f84cef4ceab22365ec24c608e8 : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2225ff80c7544f2a5c7486a23fd6ad1 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a635d4aca524881848230bf09bb65bf : SUCCESS in 17m 57s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"8223a689ac31e0057c4d089315b590f12756ad2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"4bd2987acf4d1d04d9e94a445468223f9d205381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.","accounts_in_message":[],"_revision_number":15},{"id":"b7b5a5db3cf67d3d2a9ab79a1797669a31e3be25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:39:09.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3141fc7084d44e189f28caeae52903c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ce431f5b81d4cfca02db5a97e57e530 : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":15},{"id":"101ff5ffcdfc005906217e21f81dc7ff3fd8d42c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:39:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85099bd26bac4059a0704c267865516a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de763d34f89f4bb498335695c44b174d : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8676398df48400f97adbee96a7d5b22 : SUCCESS in 9m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f406de96f0b4f0fa47fa80768c331ff : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1deb2a9b28db41efb4ee5b5f650485a6 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9afcc277817438dbe3414190aa87a13 : SUCCESS in 13m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/804c558f03aa45e1895a466a5f756d85 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/91eccd228bbb46d5a7a6b09e5dd91bb2 : SUCCESS in 25m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/808b4ac87079481ebff3408d8e633125 : SUCCESS in 9m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f4250f0bb93741d3902696d3a24b72bf : SUCCESS in 14m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ebf233071b9045efaebc81ce25f85855 : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c8da0be7ce7a4de59663e9999eee7ba5 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0819eb9fa7e942cdadef5a7de8eefe82 : SUCCESS in 10m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2f8706c8e9c640b7b10cb502fcdeb85e : SUCCESS in 13m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9132b9e28b448558ecec5647d090a77 : RETRY_LIMIT in 3m 22s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/897fa345b38444a189358d5fd887a9cc : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/68b2e36ded784164973eb52e44c07177 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c832e5a2f4b48b09e00cdb98b51fd23 : SUCCESS in 51m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63a70430433a48e192584708ee939c64 : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89316dba4fd54f73898f2c65ab0d64ce : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/232381b7dac64973846bac3a9aa0649d : SUCCESS in 18m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"4e0d4691e40d7955b603057a64f1b7e9c3dee27e","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-14 11:43:21.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"f9738f415fd734ed501b036764def0b75081a654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 06:02:18.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"dffa97f815ea7fc1ac2b3159d97b46e37735d417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:15:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98da3b53bcf24f0aa4f16f39cc7ca0a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cdb92682b14725b94e4ce91600e9f4 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5917aabe95b4126bf14e75a13156b92 : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08a56e2ce3a34f0e9169e53d3fc1ea0b : SUCCESS in 10m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3fca42af8fb4cb2818908ae15d84703 : FAILURE in 5m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6367cdcea543480d8a86838fc0821fb8 : FAILURE in 5m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9282a5d6bccb4805a899244a955c863c : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e9b944cf0b74fafad492866d0ec4bfd : SUCCESS in 11m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cc47e1f1f22f4456854ae93d14d0c0d3 : SUCCESS in 9m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c07bde5ae5c9469ebfd63c2429522a58 : SUCCESS in 16m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d7db3d97142d42e4acc8c251d6fd7b7d : SUCCESS in 22m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c29251affba84e17a52d76d4668363b1 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1bf593d185fa4e969361ac280fefd91b : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4827af1d42c644d3b41852947c885a08 : SUCCESS in 15m 37s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/47bd6e2523e146e9ab0ae1f1f834b959 : SUCCESS in 9m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84607feabaa04fbfb7bb402330786b15 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9bf546d2e01c4864abe3bb72fe3d5ad0 : FAILURE in 5m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8dec87a8e2cb42538f5b06fa8e2d32b4 : SUCCESS in 8m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5aca1d13b87241f893e4e5c5bae6bbef : SUCCESS in 59m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/571db689c60840a094684095454cdd1d : SUCCESS in 28m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac724b6237d48fda46b5ed2a4b22bd7 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fec9a664556242d7827bfa4bc46e2442 : SUCCESS in 20m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"b7e208747cfc331ea89da0cfa7af2bfbd8bd0c64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 08:34:43.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c84d4274644b97b0dfe3a67462a359\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce8f08d8ce1c47488f264b544ac30a3a : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":16},{"id":"28cd84df856c3f414d5700f9f3da8a793b77f01d","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 14:35:24.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"dd1ae8b97bd46424b91a034c30babd4bd8dd028f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:35:44.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"c77beb7c8ee63408019832858618994c26a46bdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:42.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a26522fcf3a94a8d9e950c81024e9cf7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17201c20db084e27a55cd9f33e9dbccf : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44be888170434f1396236d72c16ed4c1 : SUCCESS in 3m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fa77560d3ffe4145b5a011db7f3d089c : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b54df30839164752bd2ed76f066a865f : SUCCESS in 13m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/44fe5e57106248e1847b327f4cf05f8d : SUCCESS in 18m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46e35343cef74f9c9c000481608721cd : SUCCESS in 24m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/69ffe0d8f4154c53aa7b9c6fa665ce45 : SUCCESS in 21m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/18107a801a18496aa5eeb6b87b00fdb1 : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f58e9de418bd4972a4fd704059972fb5 : SUCCESS in 8m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/40e58b2223004992b48dc0f77c989ee8 : SUCCESS in 9m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8db1a411dbd04bbfa3efbb98fc95ab4f : SUCCESS in 6m 21s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0832a1e4ddba48dcb13947972b5f4db6 : SUCCESS in 8m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f411d6062bae4897a4533eaa576c753a : SUCCESS in 56m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2600d5d8392f4c7189a0e783ddb3a7a6 : SUCCESS in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078b51af2e844acdafef8494ad40bd77 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5dfd1a552c674972bb5799aa586695e9 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":16},{"id":"5ce2dab2978dd85a109a306597de94a0cf4a26af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"3c3e1bd1e902b3fba96e3837a5c72b5423abeb88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:45:54.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c46bc6618a3544ce96eab186d4dbae28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1597caa9367d4957b7d759e0ce5b947f : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/164eedb8de4c4957bd40212962addc3a : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"77f55de88c5e2cd930c12c4a0d0400a64536889c","revisions":{"054c188bae4290f3c20b5af82e86ef395d4a5f20":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:26:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/1"}}},"commit":{"parents":[{"commit":"2839d209c20c057f39947536e820a9ce70e7c8fb","subject":"Add replacement swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2839d209c20c057f39947536e820a9ce70e7c8fb"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 12:22:36.000000000","tz":60},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054c188bae4290f3c20b5af82e86ef395d4a5f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054c188bae4290f3c20b5af82e86ef395d4a5f20"}]},"branch":"refs/heads/feature/threaded"},"0acc733f9ab70db5bca87e37923d05a9cba62541":{"kind":"REWORK","_number":2,"created":"2026-03-09 17:18:43.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/2"}}},"commit":{"parents":[{"commit":"2b608337f63f2624ec855f81c589ac95744a1bdb","subject":"Add replacement swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b608337f63f2624ec855f81c589ac95744a1bdb"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 17:15:26.000000000","tz":60},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0acc733f9ab70db5bca87e37923d05a9cba62541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0acc733f9ab70db5bca87e37923d05a9cba62541"}]},"branch":"refs/heads/feature/threaded"},"32be266249603caf9b000a90d95fb8e2706ae84a":{"kind":"REWORK","_number":3,"created":"2026-03-14 10:22:00.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/3"}}},"commit":{"parents":[{"commit":"043344c185b0e5269c3544665d09d79558f9e8a2","subject":"Add replacement swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043344c185b0e5269c3544665d09d79558f9e8a2"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-14 08:44:43.000000000","tz":60},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32be266249603caf9b000a90d95fb8e2706ae84a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32be266249603caf9b000a90d95fb8e2706ae84a"}]},"branch":"refs/heads/feature/threaded"},"5df0dce88aebdaf12a18dc761f182421de317964":{"kind":"REWORK","_number":4,"created":"2026-03-23 10:33:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/4"}}},"commit":{"parents":[{"commit":"056b43c08dea4878b6fcb20b56212d242d7f49cc","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/056b43c08dea4878b6fcb20b56212d242d7f49cc"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 08:40:22.000000000","tz":60},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5df0dce88aebdaf12a18dc761f182421de317964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5df0dce88aebdaf12a18dc761f182421de317964"}]},"branch":"refs/heads/feature/threaded"},"a9229de9a654522cca84d256c34236072dd67228":{"kind":"REWORK","_number":5,"created":"2026-03-23 10:55:39.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/5"}}},"commit":{"parents":[{"commit":"056b43c08dea4878b6fcb20b56212d242d7f49cc","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/056b43c08dea4878b6fcb20b56212d242d7f49cc"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:51.000000000","tz":60},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9229de9a654522cca84d256c34236072dd67228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9229de9a654522cca84d256c34236072dd67228"}]},"branch":"refs/heads/feature/threaded"},"c81dcb722a729b6725ddfcfd6649292f8f040c77":{"kind":"REWORK","_number":6,"created":"2026-05-13 09:16:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/6"}}},"commit":{"parents":[{"commit":"900f9620d8c97c791d02429b312a1920a6bbe56c","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/900f9620d8c97c791d02429b312a1920a6bbe56c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-13 07:02:43.000000000","tz":120},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c81dcb722a729b6725ddfcfd6649292f8f040c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c81dcb722a729b6725ddfcfd6649292f8f040c77"}]},"branch":"refs/heads/feature/threaded"},"4f2e1814d9cbe83c4c29183e025f6f90f48f77da":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/7"}}},"commit":{"parents":[{"commit":"01883c944fb93a6cf37e2c8fe09631ddbdd28008","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01883c944fb93a6cf37e2c8fe09631ddbdd28008"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2e1814d9cbe83c4c29183e025f6f90f48f77da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2e1814d9cbe83c4c29183e025f6f90f48f77da"}]},"branch":"refs/heads/feature/threaded"},"e816f9971b5f513e95e5fa1263d9329b88d30e24":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/8"}}},"commit":{"parents":[{"commit":"ad60873ca5959f6533317e1787564bc293807a87","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad60873ca5959f6533317e1787564bc293807a87"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e816f9971b5f513e95e5fa1263d9329b88d30e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e816f9971b5f513e95e5fa1263d9329b88d30e24"}]},"branch":"refs/heads/feature/threaded"},"a6682f3e86c655ce985c00467bcdf53fc146276f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/9"}}},"commit":{"parents":[{"commit":"f04de053b786c93db7572fc5c462ff0db0f4aa18","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f04de053b786c93db7572fc5c462ff0db0f4aa18"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:17:02.000000000","tz":120},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6682f3e86c655ce985c00467bcdf53fc146276f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6682f3e86c655ce985c00467bcdf53fc146276f"}]},"branch":"refs/heads/feature/threaded"},"0d1b63517d2b67ad8925a302e0e9f1e3058c9474":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/10"}}},"commit":{"parents":[{"commit":"5dde207a03c391b099d4440583c17793b4a9af79","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5dde207a03c391b099d4440583c17793b4a9af79"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 15:22:14.000000000","tz":120},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d1b63517d2b67ad8925a302e0e9f1e3058c9474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d1b63517d2b67ad8925a302e0e9f1e3058c9474"}]},"branch":"refs/heads/feature/threaded"},"a2fb65494aa62246a30e23b106124bff21a2762c":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/11"}}},"commit":{"parents":[{"commit":"0ebe96cc3e3ddda2243053113b58a1e18352a733","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ebe96cc3e3ddda2243053113b58a1e18352a733"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaces (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2fb65494aa62246a30e23b106124bff21a2762c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2fb65494aa62246a30e23b106124bff21a2762c"}]},"branch":"refs/heads/feature/threaded"},"042c0b9331ef7516207cd52002676d72d842672c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2026-07-07 17:22:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/12"}}},"commit":{"parents":[{"commit":"dc4621f534bfbd9022f0fe8d82c54c06f1f47fa7","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc4621f534bfbd9022f0fe8d82c54c06f1f47fa7"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Replace eventlet imports with standard modules","message":"Replace eventlet imports with standard modules\n\nReplaces the eventlet imports with their counterpart using mostly\nstandard library imports. Some imports are not replaced (eg \"os as\ngreen_os\" in swift/common/wsgi.py) because they are only used with\neventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/042c0b9331ef7516207cd52002676d72d842672c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/042c0b9331ef7516207cd52002676d72d842672c"}]},"branch":"refs/heads/feature/threaded"},"5aa8334a7bc00894504a135964fab77411cc9634":{"kind":"REWORK","_number":13,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/13"}}},"commit":{"parents":[{"commit":"f8dc0ba882196f1e292c2d1cf58bed213f8f9587","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8dc0ba882196f1e292c2d1cf58bed213f8f9587"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Finish standard-module shim call-site conversions","message":"Finish standard-module shim call-site conversions\n\nConvert the remaining call sites from eventlet imports to their\nstandard-library counterparts; the bulk of the substitutions landed with\nthe earlier primitive commits, this finishes the stragglers. Some\nimports are not replaced (eg \"os as green_os\" in swift/common/wsgi.py)\nbecause they are only used with eventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aa8334a7bc00894504a135964fab77411cc9634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aa8334a7bc00894504a135964fab77411cc9634"}]},"branch":"refs/heads/feature/threaded"},"9f133fa677699b9b2e1e50c1193889583c96c3fc":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/14"}}},"commit":{"parents":[{"commit":"d2fb17442ec0e38cc552290ba9b05041c48c277e","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2fb17442ec0e38cc552290ba9b05041c48c277e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:18.000000000","tz":120},"subject":"Finish standard-module shim call-site conversions","message":"Finish standard-module shim call-site conversions\n\nConvert the remaining call sites from eventlet imports to their\nstandard-library counterparts; the bulk of the substitutions landed with\nthe earlier primitive commits, this finishes the stragglers. Some\nimports are not replaced (eg \"os as green_os\" in swift/common/wsgi.py)\nbecause they are only used with eventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f133fa677699b9b2e1e50c1193889583c96c3fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f133fa677699b9b2e1e50c1193889583c96c3fc"}]},"branch":"refs/heads/feature/threaded"},"cd248c42af3278db01ecd9e731a9519bbbed379b":{"kind":"NO_CHANGE","_number":15,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/15"}}},"commit":{"parents":[{"commit":"6552cea59e863765c329976fd944a6d2f64f2716","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6552cea59e863765c329976fd944a6d2f64f2716"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Finish standard-module shim call-site conversions","message":"Finish standard-module shim call-site conversions\n\nConvert the remaining call sites from eventlet imports to their\nstandard-library counterparts; the bulk of the substitutions landed with\nthe earlier primitive commits, this finishes the stragglers. Some\nimports are not replaced (eg \"os as green_os\" in swift/common/wsgi.py)\nbecause they are only used with eventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd248c42af3278db01ecd9e731a9519bbbed379b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd248c42af3278db01ecd9e731a9519bbbed379b"}]},"branch":"refs/heads/feature/threaded"},"77f55de88c5e2cd930c12c4a0d0400a64536889c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-08-21 06:02:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/84/979584/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/979584/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/979584/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/979584/16"}}},"commit":{"parents":[{"commit":"58e7c7c5ff1ef3868c519815a9a8af36aa0222df","subject":"Add replacement for swift.common.utils.Watchdog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58e7c7c5ff1ef3868c519815a9a8af36aa0222df"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Finish standard-module shim call-site conversions","message":"Finish standard-module shim call-site conversions\n\nConvert the remaining call sites from eventlet imports to their\nstandard-library counterparts; the bulk of the substitutions landed with\nthe earlier primitive commits, this finishes the stragglers. Some\nimports are not replaced (eg \"os as green_os\" in swift/common/wsgi.py)\nbecause they are only used with eventlet.\n\neventlet.green.http.client -\u003e http.client\neventlet.green.socket -\u003e socket\neventlet.green.ssl -\u003e ssl\neventlet.green.subprocess -\u003e subprocess\neventlet.green.threading -\u003e threading\neventlet.green.urllib.request -\u003e urllib.request\neventlet.queue.Empty -\u003e queue.Empty\neventlet.queue.LightQueue -\u003e queue.Queue\neventlet.queue.Queue -\u003e queue.Queue\neventlet.semaphore.Semaphore -\u003e threading.Semaphore\neventlet.support.greenlets.GreenletExit -\u003e greenlet.GreenletExit\n\nA couple of \"# nosec\" statements were added to pass bandit tests that\ndetected possible issues due to the use of urllib.request; these were\nsimply not detected before when using eventlet.\n\nChange-Id: I12d464419435f4a9d133445342fb5f608b9e20ab\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77f55de88c5e2cd930c12c4a0d0400a64536889c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77f55de88c5e2cd930c12c4a0d0400a64536889c"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
