)]}'
{"id":"openstack%2Fswift~979585","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I650d08d638479c9795e55bde7f8c58fd47af152c","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-28 15:44:45.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-21 08:48:36.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"I650d08d638479c9795e55bde7f8c58fd47af152c","subject":"Skip importing eventlet in threading mode","status":"MERGED","created":"2026-03-09 12:26:52.000000000","updated":"2026-08-28 15:44:45.000000000","submitted":"2026-08-28 15:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979585-eventlet-removal","meta_rev_id":"87ed326bf5a04393692b76391ce049cefadce570","_number":979585,"virtual_id_number":979585,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 15:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-08-28 14:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-28 14:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 13:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 12:27:31.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"9f39f45ad737223fcb3741d2d5d5568169c2e48c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 12:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca169d9a3bbe4fc0cfbc5032eb83dcae4c77e514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:12:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42bfcdbfac7f44bd80dbaec6360899fb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16a35ce55ca043ec82d51a024f852d70 : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"00891bae476b699468e9a892d06c90455be645f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:35:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b26fd821948d47f3905a328829f1c407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda1a3ded47d4b9d9e7e4a50f8f145a8 : SUCCESS in 6m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/256d798078944d59a3ae531d1e42b0ed : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b936aa5adf0b40b6b1fb99df4404beff : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97fd9202f2c541feaa34b86341b33e4a : SUCCESS in 11m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eeab9bd062f84441ab2a56e9a7baa3ea : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55cfda8a57a642a698846d0ccf05eef9 : SUCCESS in 23m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a75941b952524f90ac88fa43342e337f : SUCCESS in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/da82d77768994d94b51cfa4cb288e513 : SUCCESS in 35m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1fdb86ce7a4246f4a5c9378540011183 : SUCCESS in 29m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/238e4be50407407dba3d77fa7b30ed19 : SUCCESS in 8m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da7b8ffcf57b40608d72d129d3f0bfcb : SUCCESS in 16m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4b81a4a598ec44a1a8aa61e83e5fd279 : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9409fdc96e9d429eb0a8ae139ac2a4dc : SUCCESS in 6m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffc71bb10d29409c8387539d7bf3101c : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/177951dbbbf949f69ca57ffb8cfdf609 : SUCCESS in 11m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a949bd8a60084b24bfb7895ccab19f6b : SUCCESS in 59m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ec5e930427f4b9e860931b8f551cfbc : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c85a793410438c9ab9cad6ca77eda9 : FAILURE in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f4c1acba96b4df78aa954300d7ad1de : SUCCESS in 12m 50s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_threadpool.py","accounts_in_message":[],"_revision_number":1},{"id":"f0e4b04ee476eb7555ccc391cd280f73396e032a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-09 18:52:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c5bc716e4ce54d543608888a4a2cc8b614af8324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:28:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e62b011e5aa34d6e80053230bddb1448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433ef569741f4ff5a00790dcb6ae82e7 : SUCCESS in 5m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8029dd01ee51447b84400bb82a297b8e : SUCCESS in 22m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd2a02f9830c44e7bbfda0f951264378 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b1ef1524fa947798401dccfccfef767 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de2ecbcb2e0d4d7bb594e7fe2ec2e729 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e168209f771e4cb1b92930f29d0d1e58 : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02691654a0ae438e9cc3581ec6667f23 : SUCCESS in 20m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2e508e2244284bd5b05873e1a196e456 : SUCCESS in 18m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/407fce4a33fd4905b203ead003a133af : SUCCESS in 29m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ef23af08a33745c09d7b6cacb002618a : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa3f26117b8b4b5ba36f4336c6260c1b : SUCCESS in 8m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2b8db6c9bdc74c419090672ff4cdcc95 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a312d44c6e4741538cd2576526da11ca : SUCCESS in 10m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/daa18aad4cf24da0aa4bcd1312fc6f64 : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9c4433154394dba9e4ef978e9135913 : SUCCESS in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/98c4457424e3459d89c9dd7a32b94cc8 : SUCCESS in 33m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b7ddc9680ca4fb594f47b5f1c744f54 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e578e7631ace494ea8b5966b707cfc4c : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68e976d4e3d74df58dab3a0e76577b63 : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5f0ce54ef878efaad114457ff710c06558bbf14b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:29:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89487376ef3741119ccc5ddc533684a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6998ef07d1e4510b6d8206fe1050bec : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1c5b6b3f34f64506fc474bd671bb18eea776f707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-14 10:22:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ccd693157b2fc23ff8de21f0db67ba6196161437","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 11:00:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a96790682193418a93ce7fd352cf51f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f3868a6eb6c4a288c06212cce703bc2 : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":3},{"id":"137c1a67051bc5fa36a0144b8a377f2cfc342723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 11:11:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f29fa9848a4926bbe1283305437eb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a2c71c0b184084b3398744dcba47f4 : SUCCESS in 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c1c4edf0be54184a7f8646649837b8a : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/460a169c75824d339d45b7926833c41d : SUCCESS in 10m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44a0565c4dd642bba68449cd820c28dd : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ddccd0c5beb54323a188878d61179997 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a4412b28f5bf440580a2ca54ac15341a : SUCCESS in 24m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4f8c7497bdf430d8ee65863c06efd52 : SUCCESS in 20m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/32308a18ac5f4ea89ae582c736a8d650 : SUCCESS in 12m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e7a44a769c3499ba4e051e5ce2a595a : SUCCESS in 12m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ea86687e5be74c39a6edcb233262f2bb : SUCCESS in 11m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/763f7bedde1a4845ac6bba250059fcd0 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/18f6d4c9a09d47baa0a88cdee097ad75 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d1d3ff81b04447d9f0aa10cc909c1a2 : SUCCESS in 8m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31ebee87147242b08f2bc68eecd6e7d7 : FAILURE in 15m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/493c518dd64e47b0b384a1d7a21d9974 : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c80a7f0e3d64bcc9be72f7b327f19a6 : SUCCESS in 47m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/619f323d492d428cba2ec5340a627395 : SUCCESS in 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a79059dfc940508f9706a43d9503f6 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a113f681e3114946a1498739915e88cd : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":3},{"id":"32c5fbbf49b3132f7299b05ff64d3f8507539037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-18 16:58:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"13107e0d81b94bff7e2519d787f345be8fe95816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:00:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793cca7ecf4e464ab4b7adefa8a7988e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf3417fb2f44e62ab46eb3772011aa9 : SUCCESS in 5m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbbf6d37bdfa400abfceba891960605c : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4820b3294f794b97adaef7b38f156847 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2d477deea914100a13368908adedf6a : SUCCESS in 13m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23bf5b5b07a44c34a2911520c2f11bf8 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/80e18efb25c846c48cc5199bcccda3e5 : SUCCESS in 23m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6c3ff261b21f4ff7abf908ed16a54027 : SUCCESS in 20m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ff332158c0084be2b08f8a2a350ec29a : SUCCESS in 21m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2c36b87b0dcd425bad42221210de9bca : SUCCESS in 25m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e2774b9e01624eb287790742228005ce : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/238c87e34b8d4d7298c0efa515b05abe : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c27fd31f35384936ae340618dfc3f8fc : SUCCESS in 8m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c8060a1e42454fc4bd49c5074780693a : SUCCESS in 6m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e222cdd11ab74288832fc20967638ccf : FAILURE in 7m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0cf62dc8a55641b2afc58874382baf5a : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3501c9f280c84f33a3d3107e74e9d934 : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aadccb1000964fda82a31055014499f2 : SUCCESS in 24m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aed7dbd93c040deb4050e458b96a87e : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9fd9fde1cfab40b6aa5fd2b49c63f900 : SUCCESS in 19m 07s","accounts_in_message":[],"_revision_number":4},{"id":"69cdcf8dc70b6c0d4a60071630a9449ad1915eed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 21:59:17.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f8e1d2da7384fb284155b665a174f64\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95006ca6f8474ed4a385c22dde96cdf2 : FAILURE in 1h 13m 20s","accounts_in_message":[],"_revision_number":4},{"id":"6f8602ea08866b99471fea9ff25575e30ca20412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 10:33:18.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a366f4573d24865071b30a37d534bd750f8d529f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 10:55:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"560072f6649f81c6fad03be2798ffa081796f6da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:00:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7dbea9711684f98862d08cf4d191204\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c5e9bf46674c779f72516f9132fa41 : SUCCESS in 10m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b11f831bec21425f95f5b25732c0ed97 : SUCCESS in 12m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1a090737f7d47caa7b95e8c980ed016 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aea95f00f20c454e8048b6ed26217672 : SUCCESS in 11m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3626636b6c9b4baca7e4b8b7d24ce301 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d360dc7daebf4def9c44f453b791ba2c : SUCCESS in 24m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e96aa2c1e7a404089cafea939dfb35b : SUCCESS in 10m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f1dcd1db90ea4976a8cfe047f0b769c8 : SUCCESS in 26m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/56c1cf25488c44d1902900557a553395 : SUCCESS in 21m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/50a51e4504644a6fa37335c1a5171b66 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c4d940d911347318f66f8bd6fd2ff4c : SUCCESS in 18m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5ed48d5545d340ca82d31ad88b622721 : SUCCESS in 10m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c69d429accd4335b5d92ae699e747f4 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84b32ee4bf5a4dd1911f6d815b94ec83 : FAILURE in 14m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e8443f6626074ba48ee61a1d9a88cd05 : SUCCESS in 5m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c4cfc7a802419584b55833c0d53c2f : SUCCESS in 51m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5325d32fd0e64042b71520d0d1a00290 : SUCCESS in 29m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850cde675dd94e9db31bfefb324fb3fd : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21d11d424dcc4872ab71ac834dcb17f4 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":6},{"id":"898047305c132c46af729c968145486c2d3b4593","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:34:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec3c83f817a4b07adbe7d43c7f92ade\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e2545ac2db44c07a3c0fdfd6575317e : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":6},{"id":"221e88702ef17012e0f70702de098cb14733bb41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-13 09:16:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"65a4a27ad9eebd6b88fed382cac7557a3de358eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:59:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dddfa43483814cbd924d6fa96f41d113\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d829d1329677488c95571d66229d553e : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a7228e6760044dabd5573c4a1bb0307 : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffc652e86bff43b692270f221ca7d39e : SUCCESS in 10m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/900db5f1a11b405887c11d3f097619dc : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b00836ee4b624fd6bf47e26bdbd585e3 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70bf683087874c2b8891d8c3a6ba6561 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b99c489878054c1b91a220bcbb714a1b : SUCCESS in 11m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aae725ac8e0f448e9b8fe610b7064253 : SUCCESS in 25m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/199f24aa040b4e9e8deb2c37e39dbcd7 : SUCCESS in 25m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5f63a2b3ea1e4f59b1d07a9c952f087a : SUCCESS in 9m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/78cb6b1dd2844c87b5a3ad8ffe9dfada : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1b619d0922314483ab7d93f4e6179534 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac7f4ee2ec6f45c88d64020e48f2e787 : SUCCESS in 23m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35c638c6c7d54aaba87c0f358eb36ecd : SUCCESS in 8m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/621c101fa52642d5a069f258f8aed756 : FAILURE in 18m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/edafa5da4ed049fd914073d8f659df4b : SUCCESS in 9m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f06a8674643348f3863feeac41afc12b : SUCCESS in 35m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92e23a2b385f413080fd6163994a81ab : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/213f00c677664a939b88a798a3b364f7 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb45622eb6d044bcbe1b8d01cb57ea42 : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":7},{"id":"1197fcd8fee706478ebc22927ca8d3ec2d564c64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 11:36:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/884591d1d44f4128875b52eebf197409\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46d2ca1ba7444b57b6fcbcf41b23fe20 : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":7},{"id":"a7924a0d89d64da6aa7cca6fc7d4b1515c825001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c9026a9e66ba525ca289232e91ea8d2f50ffd1e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:16:34.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9187cfdbd82e45e48fd5b52f319586d7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bebca842d1e245a0947f299c03e5801b : FAILURE in 1h 00m 44s","accounts_in_message":[],"_revision_number":8},{"id":"041eb3713e8eb76ee10a9f6ae57b82c080f76d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"8f6a770dd3b970a29bf901456e45e2b2f8a3b62e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:20:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5997d85692834cb9bb3e2c81b72086ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bbce9c5ecc2471d905b6d394ded8c6c : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee9b5d6f0d49420db4113c31100b1138 : SUCCESS in 9m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65a8a9aba6b34499a4c43a2c44d73a5e : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e12dac0be0ac41a99df66420516f5286 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/052b492ef7684c7c8539a733be52eca0 : SUCCESS in 11m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67c400b483f045bda449e3d8de119dfa : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/487ba012f9ab4b8797f47ffaf8810516 : SUCCESS in 26m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1f03b3702338441687983a2e98d27ca0 : SUCCESS in 18m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cce4662b9b1043929b729d2784e53125 : SUCCESS in 23m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/340d2547fc34485ba77400a5f65402f4 : SUCCESS in 19m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0301a965ad3b4e52ad99ae4f44bc851b : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/563ea4a0197e4cdfbd6f4e5f440b380f : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/674dfd8a567b4cf09eb629743a04e28c : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d7d840f7644441e9127aca57d1dea49 : SUCCESS in 8m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e19a0c1698d44253b86fe958f7a0b3d3 : FAILURE in 17m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a85929351f5f4ac29c299f6d65dccf2e : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d6cd8a50d9a4236a66bd860547d04e6 : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d11c37116b244dab9e7b45f84f582d54 : SUCCESS in 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e89bb178d3a42fda97a3a81d627dd8d : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a12329d523f4f81a7504de60ae7c68f : SUCCESS in 21m 48s","accounts_in_message":[],"_revision_number":9},{"id":"e2cf5fa61498cc355b9404527c2c54f0b1d04a87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:48:52.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8568392717d341ed972882ab65efd1d3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1190c977d87e4ea3a92c34306be8e68a : FAILURE in 1h 09m 29s","accounts_in_message":[],"_revision_number":9},{"id":"eae8e523c847c02b3443e610ebcb90e9110630e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 13:27:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"1d3d2d7d525550c8862c9b65e87fe05b657fc517","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:32:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a51431c15684a1c8b16b77981a1d203\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a6f9de305f44eb0b2b6b257e0003e38 : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":10},{"id":"68853a592c0185925f4f6f5a25acd28bedab3356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:42:52.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11c263d94aa44459862b488e2dc16a30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff64023d5d0343e8a8d79d12e10b7e89 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c969f62f665422f9f29439687905ce3 : SUCCESS in 10m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc3464c3829f427280cb058a02d645fe : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc4d6aeed0a1429585d46864c7c4b7b3 : SUCCESS in 22m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/917d2f42bc9a427f84c136c14e8988da : SUCCESS in 11m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/350db86fbbf4452183a5ca1522f4162b : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3685780cd5f049688a5bfcca355c7e91 : SUCCESS in 26m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/25d870f7745448c892a724a51214705a : SUCCESS in 21m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78c0e0367d44497e869c808bc9bcc185 : SUCCESS in 22m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3b5a91ddb6d146da8dee3013c39cdd8b : SUCCESS in 22m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bb7ece2bb00f4ec8b4c3f62b7afdc27e : SUCCESS in 9m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d7fc84d94b544408651ea87f888631c : SUCCESS in 9m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5e46270cbdec4ded870d966533491217 : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80c0732f0bc74213b382658b0775eeb3 : SUCCESS in 15m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6c608ba81864dcdb82e5d5544aee06c : FAILURE in 12m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3aaf3b2318d8444292cd464c675c2197 : SUCCESS in 10m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/427f60e8b1754e609e760e7f8c3279fe : FAILURE in 1h 11m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5edb75ff4646d5b674759cd9b1233a : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba61622c87384bd198bfe3925d303b5a : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc63bb5e1b694bdc88ed95a1d08eb81f : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":10},{"id":"55dcdcc5e2db74e497799783648a1c66c84c2b13","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 15:16:50.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cba7f5694b818c43849e120065873b3aac1bbba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:28:56.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86f68409f2cd4341a800c05340f17ef7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa14947f648e4eb494950f4ed271c63b : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/869a566631ba42f7815a05943bf3aef6 : SUCCESS in 8m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4513593c1db34eecb574c03f2ab807e6 : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80ec55d3da404a03ace81432de178cfc : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6adeacf809d14978820354c01baa86f5 : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3387e332e69a4237bf55da7c1e73d2fa : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40f470c6b3954649b8027b934d208b85 : SUCCESS in 30m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0f114841623f4fa590bb618a7f86ab48 : SUCCESS in 23m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a939f0c5d2194dfbbfbba3f338e1b38f : FAILURE in 39m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0af4cf59cb42412cb437e17845aa95fc : SUCCESS in 23m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/26e1e9d85efb45a2ac14e3729c4608e1 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/29102e5ea58d4465af182d2cadffc0e9 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6cca87d3c674bbd85e9c804550cb473 : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/706ecd964a814d569271d8e2db6f453d : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/866c8eb9374341948086b001a8c019cc : FAILURE in 22m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1487f57a99e443f1ad7438ec9e356210 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f53c479da81b4a848ae60d24047c760d : SUCCESS in 1h 01m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c820fe48ffb43fdac27944ec06952dd : SUCCESS in 32m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df17d4c6769c4253a45b4fdb8ef0eadd : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/610df39a2541429b95c7696d1d23fa95 : SUCCESS in 24m 10s","accounts_in_message":[],"_revision_number":10},{"id":"289bab62de73d785044000eb03e9d8f2ee96b838","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:53:06.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0204108507e4484588754070cf64d303\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c658430ef5f24dfbbd3b292d0fb5769a : FAILURE in 31m 36s","accounts_in_message":[],"_revision_number":10},{"id":"3e6005ddcf152726bd2ac9a5a384654093314fef","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-03 06:14:48.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6ab6d6a53fd914c19960a4c05739076a35e34b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 07:19:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea22cb1c0b2493fa78f1aa822787748\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae722a8c57ff40bebd6e9191300ebad8 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c959408542594bf689879b3017b9e540 : SUCCESS in 8m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b46eceef0754b229e80b85de62d1080 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb5435039b4d498daa71b8114ea09264 : SUCCESS in 22m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63597ab6ed2c46cf8dae787694b57d91 : SUCCESS in 25m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d273ea29996c4bd98302ebb7e116687d : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e1648ebebf64d29921bad6680eb5324 : SUCCESS in 10m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bdc32a778d8445a48afc232b992c1de8 : SUCCESS in 22m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/29966dc1d408459880c8a71ed4affd4c : SUCCESS in 21m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b229dac70fdf4cf29baeb4f96e638650 : SUCCESS in 20m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a2198d91b5f9436293f2d9a4579093da : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/579858052c31446eb2319d98d0db007f : SUCCESS in 16m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/489dc9176a8144aab57cc0a022db0fcd : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9fe6f747d1f248feae7b1208ca4a5f5c : SUCCESS in 7m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3f552f253d0a41579f111313ae420b7f : FAILURE in 9m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ffa30c0e4004efa8ea081bfb23e1004 : SUCCESS in 5m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8f74a58ffd64145922ee4ff86752b89 : SUCCESS in 55m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ca59f98540643eb911aca414d67c755 : SUCCESS in 31m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9aefb0f5b344c1b11c010f20a96865 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aae7929764bc4db598c503a909003a7e : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":10},{"id":"da565f012816a4d3727c89e8abc2d66c012d7107","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 11:09:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df6a778763b4f4e9732c867dbca770c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd355f35a44a48689738aa96e9a45c9d : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":10},{"id":"07cb81e8038d6f7a52f2eb395b19a825ad3b4946","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-11 12:27:31.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a5c11560e8b29bc2663fa3159574a77a4ffeebf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"623f4c7d94363ef924a1bc46b2e5e58bb193b8d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:30:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6b233ec14fd4876bd216714890f816e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22cafa18f27b4601a6b71201c7467a38 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b9fab97a0444003abcc5e37c9ee9663 : SUCCESS in 4m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a51eeafe0bf84f1482565a4c418c8f93 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db439bcf185b44faad048db1de0391bc : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddea2a1aa63b4e91b1aa0b28bf4c5fa2 : SUCCESS in 22m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/623dcca9bf7d4574b9aac0e04d7f006f : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fd966451b4b843bf99e8deb2d8fa8fd9 : FAILURE in 24m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d4d5417eca4642d19d27a372b42fd28c : FAILURE in 18m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e26578d55e04f2a8f2c807ad7c1391e : FAILURE in 32m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/035cc0e70684461184308300f4cf03f8 : FAILURE in 12m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d0aa229a362f4f3eb15251b4e55117d5 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/79cb5488f4c64a9b923f0ed10d9a076a : SUCCESS in 9m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6668f72bae9349cbb015a4bfc16f1d2f : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95ca9f8ddb3d432c9fc4b36a09b328ab : SUCCESS in 6m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b30008635a1c432091998989da671ee5 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b8e490c6b5744bae84fdbc10d3ea5614 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4131b45b6078462097c9dde71c6f842e : SUCCESS in 1h 00m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1346cd153afa458989b20f21e5b211cc : FAILURE in 30m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae038d949779427fa87ff7a274fb7e5d : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a028db71bc0b413fb8068d58bec2dcc7 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":11},{"id":"769a718d6b68f4e7cec926b99ff4ca4c305034fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"607323d4944ac50a48265b24b6fc6da8659eefcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:34:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/292d96e8468e41c3ae6a62de092834a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6242a047d266499e8606e4cfe52c0501 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be51e9090ce445a9856e85f9baec4bbe : SUCCESS in 3m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2caae82d34ab4df9bcdafdf1b6527f00 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21ccb0444bc74c00b379e3d2f16ded62 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6675030ce0b6423f8eb4d55ca8cc77b7 : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c52ff77d669f497ab97a42c3ae98d25b : SUCCESS in 7m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89d51b255bb44e919a8037181df0267d : SUCCESS in 13m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cf0d7a7bae03476695a2a68a02f861d1 : SUCCESS in 22m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b3b6ef356ead460685200f0d3086cf0b : SUCCESS in 11m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/923759d640d949e98d2b85ef9267c139 : SUCCESS in 20m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/147211650c524de18c63a9417346013a : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/785e8588a0544dc8a826328159a5d7a6 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/276eb068ee1440eb9dffdf02e73daebc : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39e28071b51340f3badb6cda4d6bf5f8 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c60f3f36d93a49f5bccd63d4bb31f55d : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff3f7325e27e4d1093cadb7ba507a269 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec59ea7bb3cc49999510d6b4852a4ed6 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abe5fb3d82ae496cb8f0e59d6e53d82c : SUCCESS in 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f2b7d7360147e49e16f9f1018478ad : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7535b3bb45634f9c96f5129db237eb90 : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":12},{"id":"ba527879e44c79c5734e03fab352257ec6c3ed1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:57:06.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c3787e95db44279b0f33e565735504\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63b4ba59359c48e3be6d07d5d5332e24 : POST_FAILURE in 56m 56s","accounts_in_message":[],"_revision_number":12},{"id":"6f9360b99187fe44092ffb5015a992ae517ba41a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"bfc6df265fba226cade6d778abf00339c87e61d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 09:14:53.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d4c3393c4946738ca5ac976f980178\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6e2f6f76b984871b0280a586c4f03a3 : FAILURE in 58m 24s","accounts_in_message":[],"_revision_number":13},{"id":"5aaad8f4160d155517fa1896a059141715873aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:41:26.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b091b45c78bf44f4a212301f2e0fa370\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007231035eaf4c4b8a43bf9457d64645 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c2dbc8a21864659877bcdb9f34f982a : SUCCESS in 9m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71255eb6d6b446ae9f559ddc93525a47 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a0f50647ac242f3a59d5f856c6affb6 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d44859b670264c22aa49314cb0b6cde5 : SUCCESS in 21m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62d54d961b4b4d459614ef684d7368e6 : SUCCESS in 11m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/27bf90f7269c4d5cbd1ce560a8d9f7ba : SUCCESS in 23m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81f5169595204de49bc34a172d1a48ea : SUCCESS in 10m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/da3fc28db6dc417a8545752f1bb5f49f : SUCCESS in 13m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3fecd0fa78ec46509c3620b57a826952 : SUCCESS in 20m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/56260a882ff6463f8a3c490f71319567 : SUCCESS in 9m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f1361bde31f34f839e95b17b333ebd33 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9667d93512d7425192266bffd30ec65e : SUCCESS in 12m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/502f31206e6e4e638ba2933f40ebc791 : SUCCESS in 9m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a3ad68625bc441e9dc9b5bd77ea1221 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9a1271c80bb45a5a2e88a0ea0034d28 : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a3a6bbe5dba4cfe820482f757ce3bc9 : SUCCESS in 1h 03m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80d4e001e2904438aff5f67e126ba637 : SUCCESS in 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f07596764e41b29823f9a8007e1421 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9124e48215564ee188ffc590fe6dc6fd : SUCCESS in 20m 26s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"9a5cb39ce8fe36011da7c64c7dbce645de2aab35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 17:22:32.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"1f3f4fe5fcbbf988ad00e61b388642366fd275ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 18:32:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56944929f87045bb9c2892cda6b51ca2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3573e1c76fd140b6b55464d2a8a4887e : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28871f6d3b944f7fb070efc49816cd56 : SUCCESS in 10m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1575ceaa3bcc48c0be6a32cca4daaf83 : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a01af4a32b73498394ec3b42c7516f8b : SUCCESS in 12m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96ffdfa7c5dc4a69b9c91ad9715c2631 : SUCCESS in 22m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a76163fd976f4ddb9385b19c5885fc03 : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/520846c4ca1f4c61acf9204e7173039c : SUCCESS in 24m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1a1818135554462d8ff01a0d678ac547 : SUCCESS in 23m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e04fab9d4804e48832a72b17742e100 : SUCCESS in 30m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0c9cad6b4b4247c188284f32024a321a : SUCCESS in 23m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/044dd49fef3249c08b546ecfa06f5294 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/774baa6b8d4d4aec8e98e25020d6f011 : SUCCESS in 10m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4a33c00b5c874f838f9fe8a6fb2d2e9c : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f284aaed72254f2fa566fa9f1daae642 : RETRY_LIMIT in 6m 26s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12e9ac11d412478d833f90c1ac01f169 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/93dfe656e8184975b9ea1e9ab517757c : SUCCESS in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/619c9653d5354690abcf7c12ba1651fa : SUCCESS in 54m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcdf0d345dc446d7960bff2ca1de5ad1 : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de36c510e20743339ef254b63a91ec6a : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97d0de57b4ac44c29e9f4b58563a1f8e : SUCCESS in 20m 33s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"b3f9b7f5abd49a655c7d07590f3de896e4382e04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 01:11:30.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07a02823f3542f0a194d677928dfe5d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/441b31fb3ab0420dae10de4560585da2 : FAILURE in 41m 54s","accounts_in_message":[],"_revision_number":14},{"id":"3fc55acf1e409cb4eac89d4eae4eca00c127f1b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"79448d87a919a39ca735a14fa2720f6c5dec9b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:33:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b72ced82054942b063882c96ddfe13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef1fee5714b403fac0a8f843f2c13f9 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34f996b448f349c9ace51bfff9e02e20 : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8dd0964a71e142f395d96aae5de1722c : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0160a2fdb2c648b7820046f85f2c6978 : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fa7f4b84d2b48fbbd500dfb1b0e915b : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f5def20f5d0047f4aa39127ccf090116 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d7707a82007c4e129292962dad2b3aec : SUCCESS in 11m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/02a3c3cb703d4cc38c9761496adce058 : SUCCESS in 22m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0635a7c90b0041b0a4fac71f57c6e219 : SUCCESS in 12m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9db608ae9b66439794964931ef66519d : SUCCESS in 16m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bc769b83e9774b8fbff74196239e26bc : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ad70a6916e5f441aa3e09fe637d15fde : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6f5712c6efdb4637905b0b1a9d9d4703 : SUCCESS in 8m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9bd4c247487e459da84a8f011827e913 : RETRY_LIMIT in 4m 57s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db47ca4d48e84d8e8a2dd2eda1f30ee5 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/35c79a16c31c4548a0ea41d380ea1090 : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdd0a9e8a9d3462e98f6561db7df9814 : SUCCESS in 56m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a8362717b043fe8db15c1e4353c975 : SUCCESS in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa918d58dc9c4107a5c375f1aeb6ed88 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d467976e94224190b2bb1816dcd27b9e : SUCCESS in 18m 48s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"c9ce885568ee538f309d4d0ade09726e84d423bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:43:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f097f4546aa42598e9d09f9b7677650\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b9bc4d56bba47df92a5f584c7b83dc4 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":15},{"id":"bb2f8621c3773ec92d7b24f2b2e2107f43b60e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"a980b7c0399c48e97f0a01c906e3c62ee65e2a9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 17: New patch set was added with same tree, parent tree, and commit message as Patch Set 16.","accounts_in_message":[],"_revision_number":17},{"id":"19a8b18b4949a73b0958ae61a1202fab1bb4d307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:28:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ff79ed9f5c049abbf3598740c6391b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adccca229c9b4ad3937df258887114ca : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d56611958d04ac7b61e7c779efd1e9d : SUCCESS in 4m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc34725c8bcf44a1b6d5c9265aa23967 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc9c8015c47479582d5b2bae7511886 : SUCCESS in 11m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/435e839d0a9442f183533f1223003d25 : SUCCESS in 13m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3c70b1ba08d4d06b7e75a0182882117 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/87f4f932c510400baa5a8ec77ac373a6 : SUCCESS in 31m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bd7a82ed0abf4baabb72f1c9e4e76d81 : SUCCESS in 9m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e775a3b6bbf64f00b154eb625204fcc8 : SUCCESS in 12m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d7df9b35099f4413b998715209169f8e : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e8cc975891384156a79fea9c72990f00 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cdba3e61bbae4d9dbbb99cfa9f979815 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0ba2f65608ee49a1883a8afc5786f6b2 : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a2edcb5f54b450ea3a59d11288de25e : RETRY_LIMIT in 4m 16s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/266eabd87e694e958da08547e1f4076e : FAILURE in 9m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e4c4d83718144d7a5fd98c94e8ba494 : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7c5271712cf4a08863c8094eab5143e : SUCCESS in 39m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a78cfda481434564aed6511f687f1e97 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea06027c1bb4bbcb965e6f2f9816fdd : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb3b36951d7a40e19b930ee5ad8500d0 : SUCCESS in 24m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"23be4ef6e64be2fc1a81700a1d4c33591ebaa1a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:44:04.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e3a4ef37a245d4bb3c31e5d0b4e6ca\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/395640f7af84496185a1749b39933c97 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":17},{"id":"85dae9b781d92acf63d8a36c26422f37a6e27436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 06:02:18.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"8aca47d91d89c1e7ab6fc18887b2743ce279b776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 06:46:53.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ed5597a64545ae9d3f15a3008191c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350c2c75c16648979c98414c2be5846e : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76e8ee9b6ecc407b9a0dec6336379e2c : SUCCESS in 7m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1800f565aea642a39f217af3f93cc01e : SUCCESS in 10m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/becd713f72da4042b52401a45ab20a32 : SUCCESS in 11m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e7455cd520a45e4a63c2af818860284 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/910c85675a144aebb1dc2ea8cd2c35b7 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d5bcd7cb9b5b4edc9569c1e9f8b66177 : SUCCESS in 19m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/badbddcaded64e9ab4c8359b1d80c7a0 : SUCCESS in 21m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/585a539fd41e4da486db1d82b7c68633 : SUCCESS in 19m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1e2cd925dea34c2cbe558f3f0bbbdd4e : SUCCESS in 15m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8f5b1f7734b4c99914c338e4d965075 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e37cf66cfcce4420b6effa30e0a64177 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/395c40188dc04ca7b8fd48a74e75d212 : SUCCESS in 18m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fc56bbea3b241a494e8776ef08943b7 : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e43c7d85a57c4bfbbd71387112a3e434 : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1613da5acfed40c8aec3c7037a40c3ac : FAILURE in 8m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a903c3565c049cdb1b057393990d7fb : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/143e00a47c5a4edcbf536c4b19e5f483 : SUCCESS in 35m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04eabc1dd6954e8aa4270223052b9243 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d990986349b44a76bdabc44ea97a28f5 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18c92efae8944e9d82eda8d271f08d4d : SUCCESS in 20m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"104c1cef63039eba8f313ba8d58546397a531492","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:52:51.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b0ba732afa4a8184be932deca3280b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57ad912fc28a428cb2e7531f419613ae : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":18},{"id":"de73f45f64c283982ab46691163b7019dcb49c01","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 08:48:36.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ed521112bf826df7c6529de37ce959b1bacdca08","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 14:36:52.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"fbe478cbcbef0cddf18a6ccb13bf791d2e0f47a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:37:12.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"0b031a6776d0e112a34322f162c717f8c57ecf3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:45.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4614e58e7b144acd88ca94c1c1aba831\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c3b77a72244825a43eed4cfabe93ef : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8074ce48a7364007983287580071292c : SUCCESS in 10m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/36fe16e375ab43eea6e41f99e36a1da9 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb34b6459dcb4dfba8a491f38d579478 : SUCCESS in 26m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ee6c50f6b1564da2b26f6d51693536fb : SUCCESS in 9m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78c921755e59433abee381fb665450f6 : SUCCESS in 13m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/252ed081124f40ea988f22112fc9b55f : SUCCESS in 9m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0540eaa09b3644f9873120ce52ea1498 : SUCCESS in 9m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c315467c11574547a456ff993fed6f1e : SUCCESS in 17m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f377b038fd9a4397bc410947bdda89b4 : SUCCESS in 8m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/703918552f124e0288cf0c196eb4269e : SUCCESS in 14m 59s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a1ca46b50d140e5adb70c6fa9e86e09 : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0986c07283540029403f530caf18ea5 : SUCCESS in 58m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6724184603c0482eb1cdfa16e927f2a7 : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/410bb6b2f9ef49d0aaf144bc11fac51e : SUCCESS in 3m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d62b8bda12984d38a18faa1b196b09b6 : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":18},{"id":"87ed326bf5a04393692b76391ce049cefadce570","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"64cb0597022b9c52c9218c08ae09c18ba530336d","revisions":{"b52a9aa5c9a14b877f2f0250774e017210e5ccc2":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:26:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/1"}}},"commit":{"parents":[{"commit":"054c188bae4290f3c20b5af82e86ef395d4a5f20","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054c188bae4290f3c20b5af82e86ef395d4a5f20"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 12:22:36.000000000","tz":60},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b52a9aa5c9a14b877f2f0250774e017210e5ccc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b52a9aa5c9a14b877f2f0250774e017210e5ccc2"}]},"branch":"refs/heads/feature/threaded"},"9c34e137f9cb6a1193575670c18eb898f2cfb9ae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-09 18:52:50.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/2"}}},"commit":{"parents":[{"commit":"0acc733f9ab70db5bca87e37923d05a9cba62541","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0acc733f9ab70db5bca87e37923d05a9cba62541"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-09 17:19:05.000000000","tz":60},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c34e137f9cb6a1193575670c18eb898f2cfb9ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c34e137f9cb6a1193575670c18eb898f2cfb9ae"}]},"branch":"refs/heads/feature/threaded"},"50e59e9211a1129c0621c84bb04d1bbec0db6eae":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-14 10:22:00.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/3"}}},"commit":{"parents":[{"commit":"32be266249603caf9b000a90d95fb8e2706ae84a","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32be266249603caf9b000a90d95fb8e2706ae84a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-14 08:44:43.000000000","tz":60},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50e59e9211a1129c0621c84bb04d1bbec0db6eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50e59e9211a1129c0621c84bb04d1bbec0db6eae"}]},"branch":"refs/heads/feature/threaded"},"6fd1d69e7513a253c0c324f44bf8b24a834af61d":{"kind":"REWORK","_number":4,"created":"2026-03-18 16:58:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/4"}}},"commit":{"parents":[{"commit":"32be266249603caf9b000a90d95fb8e2706ae84a","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32be266249603caf9b000a90d95fb8e2706ae84a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-18 15:32:17.000000000","tz":60},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fd1d69e7513a253c0c324f44bf8b24a834af61d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fd1d69e7513a253c0c324f44bf8b24a834af61d"}]},"branch":"refs/heads/feature/threaded"},"8291c154600a1e9fcc87f5ee420647d09da5c45c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-23 10:33:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/5"}}},"commit":{"parents":[{"commit":"5df0dce88aebdaf12a18dc761f182421de317964","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5df0dce88aebdaf12a18dc761f182421de317964"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 08:40:22.000000000","tz":60},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8291c154600a1e9fcc87f5ee420647d09da5c45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8291c154600a1e9fcc87f5ee420647d09da5c45c"}]},"branch":"refs/heads/feature/threaded"},"cc59137675c7795b5449b408b2e53bba932443f4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-23 10:55:39.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/6"}}},"commit":{"parents":[{"commit":"a9229de9a654522cca84d256c34236072dd67228","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9229de9a654522cca84d256c34236072dd67228"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:55.000000000","tz":60},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc59137675c7795b5449b408b2e53bba932443f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc59137675c7795b5449b408b2e53bba932443f4"}]},"branch":"refs/heads/feature/threaded"},"46bf8493e2575d95882e6dc4e8dea6579fe37ef3":{"kind":"REWORK","_number":7,"created":"2026-05-13 09:16:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/7"}}},"commit":{"parents":[{"commit":"c81dcb722a729b6725ddfcfd6649292f8f040c77","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c81dcb722a729b6725ddfcfd6649292f8f040c77"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-13 07:02:43.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46bf8493e2575d95882e6dc4e8dea6579fe37ef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46bf8493e2575d95882e6dc4e8dea6579fe37ef3"}]},"branch":"refs/heads/feature/threaded"},"bcd084d24cda871dc492dc9a038d669decacdacd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/8"}}},"commit":{"parents":[{"commit":"4f2e1814d9cbe83c4c29183e025f6f90f48f77da","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2e1814d9cbe83c4c29183e025f6f90f48f77da"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcd084d24cda871dc492dc9a038d669decacdacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcd084d24cda871dc492dc9a038d669decacdacd"}]},"branch":"refs/heads/feature/threaded"},"64e692f13931657f4b3479cdb3cfc0d5d56fe709":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/9"}}},"commit":{"parents":[{"commit":"e816f9971b5f513e95e5fa1263d9329b88d30e24","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e816f9971b5f513e95e5fa1263d9329b88d30e24"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e692f13931657f4b3479cdb3cfc0d5d56fe709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e692f13931657f4b3479cdb3cfc0d5d56fe709"}]},"branch":"refs/heads/feature/threaded"},"9ba8a36e705568f2b940871bc3c56c88e37f1d66":{"kind":"REWORK","_number":10,"created":"2026-06-02 13:27:47.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/10"}}},"commit":{"parents":[{"commit":"e816f9971b5f513e95e5fa1263d9329b88d30e24","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e816f9971b5f513e95e5fa1263d9329b88d30e24"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-02 13:06:29.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ba8a36e705568f2b940871bc3c56c88e37f1d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ba8a36e705568f2b940871bc3c56c88e37f1d66"}]},"branch":"refs/heads/feature/threaded"},"15164448224d5bdc9148f55c69433a07f75e5c63":{"kind":"REWORK","_number":11,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/11"}}},"commit":{"parents":[{"commit":"a6682f3e86c655ce985c00467bcdf53fc146276f","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6682f3e86c655ce985c00467bcdf53fc146276f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:17:19.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15164448224d5bdc9148f55c69433a07f75e5c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15164448224d5bdc9148f55c69433a07f75e5c63"}]},"branch":"refs/heads/feature/threaded"},"79710fed8bf3bcc710a9b9be70830b07e519dd84":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/12"}}},"commit":{"parents":[{"commit":"0d1b63517d2b67ad8925a302e0e9f1e3058c9474","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d1b63517d2b67ad8925a302e0e9f1e3058c9474"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 15:27:26.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79710fed8bf3bcc710a9b9be70830b07e519dd84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79710fed8bf3bcc710a9b9be70830b07e519dd84"}]},"branch":"refs/heads/feature/threaded"},"3eee8eca38e611f23de62b2e51a06ab01d557b5a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/13"}}},"commit":{"parents":[{"commit":"a2fb65494aa62246a30e23b106124bff21a2762c","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2fb65494aa62246a30e23b106124bff21a2762c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes eventlets imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3eee8eca38e611f23de62b2e51a06ab01d557b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3eee8eca38e611f23de62b2e51a06ab01d557b5a"}]},"branch":"refs/heads/feature/threaded"},"66d9e83682870a0455dbdf6ee8545c295c3e2dbe":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2026-07-07 17:22:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/14"}}},"commit":{"parents":[{"commit":"042c0b9331ef7516207cd52002676d72d842672c","subject":"Replace eventlet imports with standard modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/042c0b9331ef7516207cd52002676d72d842672c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes Eventlet imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol.SwiftHttpProtocol/SwiftHttpProxiedProtocol\nare not replaced because these are only used if eventlet is used.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66d9e83682870a0455dbdf6ee8545c295c3e2dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66d9e83682870a0455dbdf6ee8545c295c3e2dbe"}]},"branch":"refs/heads/feature/threaded"},"d39a71c524fd74f46912da929d1f318d1c1e1126":{"kind":"REWORK","_number":15,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/15"}}},"commit":{"parents":[{"commit":"5aa8334a7bc00894504a135964fab77411cc9634","subject":"Finish standard-module shim call-site conversions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aa8334a7bc00894504a135964fab77411cc9634"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes Eventlet imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol now subclasses the HttpProtocol /\nHttpProtocolMessageClass stand-ins from the concurrency shim and exposes\nits classes via eventlet_only(), so it imports without eventlet (the\nclasses resolve to None in threading mode). daemon.run_daemon() and\nutils drop their direct eventlet hub/monkey-patch use in favour of the\nshim\u0027s install_hub()/hub_exceptions()/eventlet_monkey_patch() no-op\npath, and utils.get_hub() probes the un-monkeypatched select module via\noriginal().\n\nBufferedHTTPResponse also takes the plain-socket path here, before\nGunicorn: real_socket()/close_real_socket() return and close the\nunderlying socket.socket instead of the eventlet-only sock.fd.\n\nbufferedhttp also disables email._policybase.validate_header_name\nprocess-wide: Python 3.14 made non-ASCII header parsing stricter (see\ncpython commit c432d014), and without eventlet\u0027s request parser the\nstdlib validator would reject backend headers that were accepted before.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d39a71c524fd74f46912da929d1f318d1c1e1126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d39a71c524fd74f46912da929d1f318d1c1e1126"}]},"branch":"refs/heads/feature/threaded"},"6c5728a51957961180c8c661ecb828209cf04ec3":{"kind":"REWORK","_number":16,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/16"}}},"commit":{"parents":[{"commit":"9f133fa677699b9b2e1e50c1193889583c96c3fc","subject":"Finish standard-module shim call-site conversions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f133fa677699b9b2e1e50c1193889583c96c3fc"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:18.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes Eventlet imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol now subclasses the HttpProtocol /\nHttpProtocolMessageClass stand-ins from the concurrency shim and exposes\nits classes via eventlet_only(), so it imports without eventlet (the\nclasses resolve to None in threading mode). daemon.run_daemon() and\nutils drop their direct eventlet hub/monkey-patch use in favour of the\nshim\u0027s install_hub()/hub_exceptions()/eventlet_monkey_patch() no-op\npath, and utils.get_hub() probes the un-monkeypatched select module via\noriginal().\n\nBufferedHTTPResponse also takes the plain-socket path here, before\nGunicorn: real_socket()/close_real_socket() return and close the\nunderlying socket.socket instead of the eventlet-only sock.fd.\n\nbufferedhttp also disables email._policybase.validate_header_name\nprocess-wide: Python 3.14 made non-ASCII header parsing stricter (see\ncpython commit c432d014), and without eventlet\u0027s request parser the\nstdlib validator would reject backend headers that were accepted before.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c5728a51957961180c8c661ecb828209cf04ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c5728a51957961180c8c661ecb828209cf04ec3"}]},"branch":"refs/heads/feature/threaded"},"3c33599a45eef1cbdf52ec74d00540b8528f7cb1":{"kind":"NO_CHANGE","_number":17,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/17"}}},"commit":{"parents":[{"commit":"cd248c42af3278db01ecd9e731a9519bbbed379b","subject":"Finish standard-module shim call-site conversions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd248c42af3278db01ecd9e731a9519bbbed379b"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes Eventlet imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol now subclasses the HttpProtocol /\nHttpProtocolMessageClass stand-ins from the concurrency shim and exposes\nits classes via eventlet_only(), so it imports without eventlet (the\nclasses resolve to None in threading mode). daemon.run_daemon() and\nutils drop their direct eventlet hub/monkey-patch use in favour of the\nshim\u0027s install_hub()/hub_exceptions()/eventlet_monkey_patch() no-op\npath, and utils.get_hub() probes the un-monkeypatched select module via\noriginal().\n\nBufferedHTTPResponse also takes the plain-socket path here, before\nGunicorn: real_socket()/close_real_socket() return and close the\nunderlying socket.socket instead of the eventlet-only sock.fd.\n\nbufferedhttp also disables email._policybase.validate_header_name\nprocess-wide: Python 3.14 made non-ASCII header parsing stricter (see\ncpython commit c432d014), and without eventlet\u0027s request parser the\nstdlib validator would reject backend headers that were accepted before.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c33599a45eef1cbdf52ec74d00540b8528f7cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c33599a45eef1cbdf52ec74d00540b8528f7cb1"}]},"branch":"refs/heads/feature/threaded"},"64cb0597022b9c52c9218c08ae09c18ba530336d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-08-21 06:02:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/85/979585/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/979585/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/979585/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/979585/18"}}},"commit":{"parents":[{"commit":"77f55de88c5e2cd930c12c4a0d0400a64536889c","subject":"Finish standard-module shim call-site conversions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77f55de88c5e2cd930c12c4a0d0400a64536889c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-10-28 13:25:08.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Skip importing eventlet in threading mode","message":"Skip importing eventlet in threading mode\n\nThis makes Eventlet imports optional if running in threading mode. Not\nall imports have been made optional, this patch starts with the minimum\nset of files to make it possible to run Swift API services using\ngunicorn without importing eventlet.\n\nswift.common.http_protocol now subclasses the HttpProtocol /\nHttpProtocolMessageClass stand-ins from the concurrency shim and exposes\nits classes via eventlet_only(), so it imports without eventlet (the\nclasses resolve to None in threading mode). daemon.run_daemon() and\nutils drop their direct eventlet hub/monkey-patch use in favour of the\nshim\u0027s install_hub()/hub_exceptions()/eventlet_monkey_patch() no-op\npath, and utils.get_hub() probes the un-monkeypatched select module via\noriginal().\n\nBufferedHTTPResponse also takes the plain-socket path here, before\nGunicorn: real_socket()/close_real_socket() return and close the\nunderlying socket.socket instead of the eventlet-only sock.fd.\n\nbufferedhttp also disables email._policybase.validate_header_name\nprocess-wide: Python 3.14 made non-ASCII header parsing stricter (see\ncpython commit c432d014), and without eventlet\u0027s request parser the\nstdlib validator would reject backend headers that were accepted before.\n\nThis disables zero-copy in threading mode and needs a follow up to\nimplement this properly when using gunicorn.\n\nChange-Id: I650d08d638479c9795e55bde7f8c58fd47af152c\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64cb0597022b9c52c9218c08ae09c18ba530336d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64cb0597022b9c52c9218c08ae09c18ba530336d"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
