)]}'
{"id":"openstack%2Fswift~979790","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I9a90376a67d084e943446ba1c82c3be197187155","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-28 15:44:49.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-28 15:44:49.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-21 08:53:10.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-28 15:44:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a90376a67d084e943446ba1c82c3be197187155","subject":"Add Gunicorn using gthreads as wsgi server","status":"MERGED","created":"2026-03-10 07:49:33.000000000","updated":"2026-08-28 15:44:49.000000000","submitted":"2026-08-28 15:44:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979790-eventlet-removal","meta_rev_id":"e47ffcd6635fd971ad49a8f9e2b0f42b9ad2ab84","_number":979790,"virtual_id_number":979790,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 15:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-08-21 06:02:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-08-28 14:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-28 14:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-10 08:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-12 18:54:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-04-23 06:30:23.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-04-23 13:20:28.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"CC"},{"updated":"2026-07-21 10:43:43.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"f05382aaa419dce16a597f5f118520d7d9ef9f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-10 07:49:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c901f27898cdeb6349b281517af4f3a5e65943e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 08:25:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b075406a2d14f53b34f6b379306eadb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f293fb5f18d4df9a369de14a15ed3b2 : SUCCESS in 5m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53210e3e479849a2bdba46146373cf63 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/504c9924b8154946af4bb2240c510ce9 : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5a653c5eebd43c5b5d06403ccc99a84 : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c72e812844194e5ab7d400035f5c9315 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7033a68311f34fd798cd5be6ed3d466f : SUCCESS in 24m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57997ccc15a2415ba4ebeae0ba4619a4 : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb7a713d2308423caae1884f23e83044 : SUCCESS in 20m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4563bb2263a54f28b33702d640f7a3aa : SUCCESS in 20m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3dbcf0e3034841b4af5b70663e429dfa : SUCCESS in 12m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1ce1ba3b1fc648b9bfe5bf6b991b30d7 : SUCCESS in 8m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c7c2eaa68364db0829e48e28c37003b : SUCCESS in 8m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5494b89d91414fc28d1f3b95b43f032a : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04df3e6d0d6644788197b16ebd4db22b : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c35eaf593214d8fbd450bcbda83ad5a : SUCCESS in 5m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf7d12fdaec64de9b7fbae8503387d00 : SUCCESS in 34m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53538e3e22634fdba20595c9467fafb1 : SUCCESS in 30m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e33113d0015242fcb8ee77e96a65009b : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d125541c85d4d1a9127e8bcfd5598f6 : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e0a172641b0a712e9cafb804c538a2133b90ae80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 08:40:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53917b863d554aa68af374932e686024\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f5c9b8d49e940f6abe996187f7b30b1 : FAILURE in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"830644bef2efafff2eabca8293c50ab7fcf201fc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-12 18:54:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2442d8ef24544a26b29895354b9036da487db54b","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-13 11:45:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"334da6a3fd3a577ef428557f00ed8432e2a2006e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-14 10:22:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ff47dc62e37181c200a4cb9cc1e7bcea9db1e22c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 10:56:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac220250020f475189b40fcfea28117c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6eeee70e361d447293c70504f2fe251e : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b1eecea05a7d22c60b5b917f85f723ddf01ca65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 11:29:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1caad671224f4dc9a38349128ce52da6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f6e48dc6ce43a9bfc9c666bc7e0e8e : SUCCESS in 9m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cb0fc3088054654a5073cbe0014afe7 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c1fb2d4da214403be7a4e286918249f : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96ecf9321bad4e769b1ec73746113c16 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f631cdd7bbd942ad83b166db6cd449b1 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/52fbfba80027455096dfab044cbd993e : SUCCESS in 29m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6cf43c0d3584e2c8b2c58b33f5cfb88 : SUCCESS in 17m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c562cbacd84d4a8db36f4a770abb22ca : SUCCESS in 29m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ef0bf652b12f4cdd9c39506de9f92005 : SUCCESS in 29m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3db62b8968334edca56bdce6a2f5ecfc : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/176c1defdf30422ba029ec1776f5cdcb : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14af6957f92d42f29982ce9d8fac3ec7 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1b7011a626e46dc9eacebb1e9e17d8e : SUCCESS in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95f4a62e8c3d46fcbf687c9d26d15799 : FAILURE in 15m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2e5b091ad1dc4c58a989e8ddaac700f3 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e4e4ac6685447e48e4980e49266015f : SUCCESS in 58m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf4f69a9fc33483e8f26355ebbcc2ecc : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe6e2e76fdc48a29c458720298538c5 : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed75a9f91e0e49d292f3c6292e2a8283 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e9e23ac4dce9f8d6306962acfb19d351ebc01230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-18 16:58:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"352258d7f77b4c42905624aeb8ba1620a04cd8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:02:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4424de6099a9471ca978bffac06a3188\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6311e140e70241108b1c6efaf8155a92 : SUCCESS in 10m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c50a8b2b17594d0db71e97c369e731f8 : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d93e6f1c7304479b33af34d91171e5e : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1f53e5ea68b4b41a26d86a774966d17 : SUCCESS in 24m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2ae59bf54f643078c0f74e52549d217 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f76b9aeae264f989d2c9ffe6730d61a : SUCCESS in 26m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9210e147c2284cf6b07e3c463dc7d2d3 : SUCCESS in 22m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/19ccc3223db442a6820afb1032de55a7 : SUCCESS in 31m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9bc80aa77c1c490789c329e1d5ddde22 : SUCCESS in 19m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b019649ec44c4f0c82ae23b0ecb4638e : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/33831add17244847ac7291caab39a6bc : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fcddd600dec249c88de173b27c8a6ffa : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ddc91e207ea7413fa2ef4f209ddaf988 : SUCCESS in 4m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78e979722b464e99814e17e0bf65ef67 : FAILURE in 15m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/191ebf03a0be4d9883d97166dd2252c6 : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f84be3645f64b50b9017a03c9e206a7 : SUCCESS in 54m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/326db9ee1ef14b458cec3437ff586647 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994a405e5de747ba99160e236f1c5289 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90fc5684002048e286aeea7b1a72798e : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":3},{"id":"bcdcd186b08209a511fcca2236d43b4b1b5461f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 22:02:36.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4be9f2ae754827b1cbde16ba33dbf2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d2c5cc2aad94ed79075eb4ec1d10c0c : FAILURE in 1h 12m 53s","accounts_in_message":[],"_revision_number":3},{"id":"d41b1e8bff8e58769322820eaaf35c3253714d03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 10:33:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1813a098fa7bf9205547a0aa377231ff31c00427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 10:55:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"dd41a843bdaf72ee5db22c05555ed00b1732d4e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:07:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0649c9260b394889855ebee4e4fa7551\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b856d01a4764ac59aaffd2fa229c8dd : SUCCESS in 10m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cec4422594e04e0a9f73631132b0f3e3 : SUCCESS in 12m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b76e41cac3b49d4979258e63a4878b9 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76cc8e0bbead4ab2b5350d90f1b0c67c : SUCCESS in 22m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19ad81baf3a74b0eb5d6d9a6ae1eb5f7 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f6686f281bc41b2bd991470ec9a499d : SUCCESS in 24m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/49795904f3b848f8a9b26e62baf2723e : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6b3b195bc3be43958e91f29071b6e703 : SUCCESS in 27m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/612fe9545aaf4646b884b89f518b43f0 : SUCCESS in 28m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/05162ca9e8424462a9dbaaeb0bd07729 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/929a1513be074c88b3aafb46c4697c5e : SUCCESS in 16m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/591397cbecbe47f681b87d29d1235ec5 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d911fd560b37445998a2de89481c5b32 : SUCCESS in 7m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c3b38b1ca4b423ab25b227ca8a3ed0b : FAILURE in 15m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7217a1144ec4834a646c5a52d9a54ea : SUCCESS in 5m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/771a6bf88e3c45d48eca67aeb63cc0c8 : SUCCESS in 56m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b79023c1fb8244b4aa8a17150e1fc474 : SUCCESS in 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59bb54792047487ab26c6c046c108bf4 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/685d6a92d5494006b4f6935aaf1bf7fe : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":5},{"id":"f690918780e8bf3e6a562d76e69520251da6ddc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:10:15.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ab46f148b841d48cdf0750495490e4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04cafe189cec4a66a24c1b17e685bef0 : FAILURE in 37m 57s","accounts_in_message":[],"_revision_number":5},{"id":"6bf73f707496642b15b9b1ddd26ac4aa1ff69780","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-04-23 06:30:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"efaf1ca9f62d02a3fb4283862d4d982df0386370","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-04-23 13:20:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3bb8f0179db1470ceeed3bdfac7f5277824dc5ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-13 09:16:18.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"29fffb39ea32f36c9faf7fb494e84e5dd5dd52b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:03:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11011cda85044270a330538cf9daafad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eada9e1275141df8b4d3d382cc0d729 : FAILURE in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f21975648d704cfaabb28d2bb0d04668 : SUCCESS in 7m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7df5183804ea4e61a874e69bec99a050 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9a8ee23deb3409796f1622532e1ba15 : SUCCESS in 25m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c85334548c74aaa9c4198a53797dfeb : SUCCESS in 23m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcf9bc566e364896a1a0b7e9b13d632a : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b097f05d92343f49ce0dc35b372efca : SUCCESS in 26m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9b88c2a529854957ba5c076b39598d66 : SUCCESS in 16m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9c394a5fd52b40eb946dbd48fc88e456 : SUCCESS in 36m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a517098eb20543609997a6b60f024287 : SUCCESS in 22m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f6eec05e22e24dc68ca2ef48b77a4e75 : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/38e1e2aae9e4454593794aa96b21ecd0 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/82de05ed3e4b4f4898eef6590e63447e : SUCCESS in 20m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ce7cc48abdb40b5b01e923516bfbed1 : RETRY_LIMIT in 5m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ca551800a0e34defadb84691372a3baf : RETRY_LIMIT in 5m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e4e388050764bf0ba634ee95d563a55 : RETRY_LIMIT in 4m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c26d92de381f47528bafd4ea7fa043d6 : RETRY_LIMIT in 5m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beac45ae226f4fee8bac5bf8d8e0f35d : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a2552b0fec457aad4fbda98dc625bf : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be20ebc3d51e411aaa1c314344bc2b9f : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6c3a1f8cd324502a7e50bc870aa45fe7322e7c05","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-13 11:24:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"39e977353880d5092e1d2e93b6326555653819c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 12:05:35.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8a49b18f2ae4ccb9b3ea622ab36f478\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da4e6167ec84ede89567881f99ed502 : FAILURE in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db57cb469bdd491c9fe0652cee2a07d4 : SUCCESS in 9m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93c65e4bb52f48879c12f9df40ebe11e : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6af7a4107f142c0be3e28b69a5db586 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1813f79911245f9ac9e442587660cf5 : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4979ced2782c491ca99a0237618e66af : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ac5a1ac0a104f98be8198298221e071 : SUCCESS in 29m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e9768c96ea5e43f096608b298d035e0f : SUCCESS in 20m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1a0009555f7b4b6387ae815f3244ba18 : SUCCESS in 11m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/40c6ebe8b33c4da289744363e3ff45df : SUCCESS in 22m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f3e03fe5076d44e99c5eebe8b2e786ed : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d1e671903c74434099d63a7bb6b700ca : SUCCESS in 11m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6626768e11cb4280938378f172330e2f : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f3b1abfe49a44d3a6755a67660b4da0 : RETRY_LIMIT in 4m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a2c71ce982943a7a1cc192b75714ff8 : RETRY_LIMIT in 6m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c8b315592c448fb957e035bb941fe53 : RETRY_LIMIT in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d67d19bd705b4c168d312ea72c1f4e21 : RETRY_LIMIT in 5m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a372dbcfa64947bd88e5014c64790b79 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1862976d5c7749e5a0fba5baf08b6791 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15200b92534d40168a51ad0139a81b82 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":6},{"id":"7b4f232ca66543bb4302f3984ae8c1130b27b8f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 12:21:56.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08447b9245964b82b54786367b33ba16\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c39c073789234d2aa59ee03c6d7cfcc4 : FAILURE in 43m 16s","accounts_in_message":[],"_revision_number":6},{"id":"e89199d5a127ff9265738e70f169c3d5efd2df10","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-21 06:01:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"85dc86abb9f401b9c16330908eca2228052a9e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 06:41:06.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/482cf6542be84c3d9ef1da9e4b85b88f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c92ffa9ae8468782a0104657d49473 : FAILURE in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66e3179fb2f54938999d60087b38ad2b : SUCCESS in 3m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c082d8c49d7a4f27a9fff09cba5940fd : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81b2c090b2944b36a28ea98e6629c546 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/562f2e62865643e4b70b1860e7ac8962 : SUCCESS in 10m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50ea111fef44413cb43254a2347aa211 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfa4a67bb7e64a038a771605e5e01f9a : SUCCESS in 22m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aff53b52ae29442189801a175756ea4b : FAILURE in 10m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8d0356d192f843a2992d0dfd19fd2100 : FAILURE in 11m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9eaf69608ad24af195d3e14c8265578f : FAILURE in 19m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/290ad1802c2a4573bc7a8f2c05614a15 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a6175e0f4ec24b8db65da888ef2edadc : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad8ad30d87a8471dbb46928700c93962 : SUCCESS in 8m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1860c125c3bb40a1aa1e014b168af429 : RETRY_LIMIT in 4m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6be5e376d45344a1a613c0cb3d9b018d : RETRY_LIMIT in 4m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/05d11a71727d473d98ab63fe285aa473 : RETRY_LIMIT in 4m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd41db1f0bdf458fbb9a7f1e5fb60591 : RETRY_LIMIT in 2m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5a10677161c4a0e8dbb207f338822e8 : SUCCESS in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2495b79aa0774633bb5710acefbe679b : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39b80829546449d594474756d4591d73 : SUCCESS in 20m 32s","accounts_in_message":[],"_revision_number":6},{"id":"7b101bd73541131c76249444dceaaf8cb89753c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 07:17:54.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a905d39934145cd88246707d84f6f15\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7350ee5d85fc4ff8a6281618d5cbe443 : FAILURE in 54m 08s","accounts_in_message":[],"_revision_number":6},{"id":"05aed9c7a44053a6b033cdc511ff2e562c9d1440","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-21 09:01:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7a87ebf42f93519c28e86596e25f591671e8f216","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-21 16:32:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45a38ed02522181d524ae871c931e7af0ea0b25a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 16:41:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"7d55672e730d64a4db93b279f2046da2b97c150b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"3ad260e69b959b99fb371a0bed39a468df8c17df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:15:22.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d020744e85d4a3e87d17eceef962247\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1febc49241dd46908d5df004731c8e7b : FAILURE in 59m 32s","accounts_in_message":[],"_revision_number":8},{"id":"f00031a9563658d14affbe2e171f82bc0c5f4a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"4d75b7f09428aa928fc98727f06ae4a7fa16f325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:33:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06bc5081f1a74d39b542337c3a9293a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0caa5b6e6404449a918a8eff2ebf836f : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3709537fc0941cf901853b6d95b9c97 : SUCCESS in 8m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56d29851503e426599f472752316a1c3 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f70a34c1e3a74d9bbeda5752843808bd : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/190e97ab9980475cb8c2dbff3bf9c736 : SUCCESS in 12m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7459eda85934866bb0497052d468ccc : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f708f52eb1f347ba8cb1fa0139e2edea : SUCCESS in 20m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/255cf78dffcf45dd8839649dcd2c2504 : SUCCESS in 9m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1f47c84318a547afbcdf0a6ab4561c1a : SUCCESS in 31m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e4276b652f004dfebff47db8244f2fed : SUCCESS in 21m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e81f05741a444ba8bd69e7c086376f43 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/11798c36d96b4b2cb13c866ca73bad93 : SUCCESS in 13m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9e5e4abbe7fb45f1a3ef6ead5f0d4bb6 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/25ee377318124ff6bd6510bcff8bec1c : SUCCESS in 3m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fcf4a7e2341e486c9d9d5ad20144eaeb : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c25cfb02a1f4eabbf6f3ccc3b174086 : SUCCESS in 6m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/db7f4808623349c18939537f65c88893 : SUCCESS in 40m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df29069415414f51bc13adbb95b6d99e : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744ad762d9fa45b18ba292e0c2e4ce04 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/739dcaac71df44ebbdb1ed7d30a04a37 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":9},{"id":"188e648468ca99a261b662b8518d5797c40ac205","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:42:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd36097536946f1812aef357e174739\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ecf0ea2062234562b2c9a953c3784aa3 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":9},{"id":"d029287b8487ca34d45de3c102b292090e8ff888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 13:27:47.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"cf91e910845d2924351176c2be4f63bd383ef7bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:49:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b40effe4259d494498c91bca7b69800a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77b1eb7a698457caf52a02fc14e836e : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbd4495f81ea4e99a65108405275a2cb : SUCCESS in 8m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f780fa498c22404c8e2604409d105481 : SUCCESS in 24m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3fc819df96c43a48aa72a9660ac74ee : SUCCESS in 11m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6416f9018f0646bca7d1e81309fdcee0 : SUCCESS in 24m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0f0428f2ea241649c0b2f2049dfaa84 : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4b20b9589424b51aa8b0d06a1a5016d : SUCCESS in 28m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5a2ab3e6310c496393ce7d3b8d426369 : SUCCESS in 16m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5ee2f13bddd447be978f4956436ddf10 : SUCCESS in 33m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/19d320c5052b4f14ab5d933cca3ea43d : SUCCESS in 18m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/75bffe3d1cb54fcf9ed5970f902fdc79 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/055df4b952b5458f91c1c08ff5736578 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0b03243a833e495eaf532a8b333fb1a6 : SUCCESS in 9m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3381de6eb314dc790ce768f0b86e726 : SUCCESS in 5m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fcc5c6b4e860498f9bce6bc87068a5b8 : FAILURE in 17m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41bed7b2a927431da4daffbcc96b7028 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/484b623aeece44c7ba461cfa3410fcfe : SUCCESS in 1h 12m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18ae7ae381364faeae6889e1f5bffa68 : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a449facd37034584938b61d640af0dc6 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab7ee260d78e42a18be9c56e955a17bc : SUCCESS in 21m 40s","accounts_in_message":[],"_revision_number":10},{"id":"c10b75921d8e1fa544728e69f4351852d53a628a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:30:55.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eda56497fd2e401aa110c7a0edaed653\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/abb4123dc97546e38fb2e46a4453c2bd : FAILURE in 55m 06s","accounts_in_message":[],"_revision_number":10},{"id":"60e7a92fc08e65df416f4bcc0e90311bdaddc4a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"02736806773302093e25637ddbdbec804a12dcd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:39:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/458095ab28aa4337aa1a5cef2bd28008\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e60c30ee8840c295fa3165a46ebae3 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acbb37c427d8450c85374b9d06d8e34e : SUCCESS in 4m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6fbd13337ff4389a4ec1e43532b44a3 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d144069c0aae4be08a23898e5c5eba5f : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a4df0f1c01b42a7a9005daf4a7ce61a : SUCCESS in 17m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b32423dcd354a179a45f0d514b60444 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d248da81c5f451ebc579a9db70ab26f : FAILURE in 6m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a7bb95a99ffe4b1794a9d8eb5e7b15b0 : FAILURE in 23m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2af2b486019f4e6cb7fb5f6f7ca71da5 : FAILURE in 31m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9799a26269bd489cb4cd318fb4bce58a : FAILURE in 21m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/55b973363d37448a900039c956dbf168 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bc78ea3714f440c5b2333a918f0a2651 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cb105b0227a04d6f8e4f71da3a6011df : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/805cd10c4c8d4767a3e598a20467ba79 : SUCCESS in 8m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f3efe6dfc6e41028af6823a48558e2a : FAILURE in 17m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/182253192b5c441b8648daeb8d17c6a7 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a1a96b3b290473a917a257d17ccd099 : FAILURE in 1h 07m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3902398fa8b2407b97f7e7aed53430d2 : FAILURE in 24m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48dc5820de564bd9bbc2ed1ade8a9105 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c1595cc5e094d959c1dc772cf3b0a92 : SUCCESS in 20m 52s","accounts_in_message":[],"_revision_number":11},{"id":"f0842ec5820664ba5fd62c259e1491901101cce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"48a785f39957f999d548aa7ca6c8651f725a516a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:59:27.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fa40566beb54a6c8df44247b944c2b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5198c2ba72c14df8b179f7611526a2da : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e399193af0e646ea82b92919d590403c : SUCCESS in 7m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/757604ed534049a7afcd94367f1e49ee : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f199d0cf49cb4dd29c7b03e7b16cc88d : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32a2e309b949498080fafe83f3901733 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50c874dfdcb84180816cbbd0577d099e : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/558f077d10a546e8a528bebb1c55973b : FAILURE in 4m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4b3bbbf6185b4528ab6b22c46cec9661 : SUCCESS in 21m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/943c925c61e74603914292490e785288 : SUCCESS in 30m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fb7bd39722c245ff95275e9834871e18 : SUCCESS in 20m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7cec3505e3b14d15ab30205c637e9b88 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a8ac9f044324a3194e071f31fcdae02 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2bc1332859a240339bf6e74c343514fd : SUCCESS in 10m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd9727874c58490e87c2512736384ffa : SUCCESS in 3m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/430c9dfa99bc45d79e86d3a0a086cca7 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f4b7ce1f1a5478084e1b52bc45adf81 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/766275a3230a42d19353e5a88f50e847 : SUCCESS in 1h 00m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2c844c3eaf14305af65f5b0d1e46427 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b43cc8f5944577b75895f46b08b1a2 : SUCCESS in 2m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f769df5225d48a0ae8b531582990e70 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":12},{"id":"c607b95d7474bce405aba9f5e9cc3ce615c1c488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 07:10:12.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"dadac4417fa50120767b3601b53e6626abfb8ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:22:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8167259d0560423db94f15f50a9195c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b446abed56df428794779f1c0d5df70e : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b699ba5d02b34427b85094628c61aba3 : SUCCESS in 8m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfa999891c0343f38afffeb7d264e95d : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b05024aae78a48758e18b179db309db8 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d38be135175e4d64907f66e8b38c0916 : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/697328ec0f2e478fac55b65c2f6bc81a : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/258c2091bc9b4c4eb16f0f3b57a984e7 : FAILURE in 6m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d82dbefdaa954efa8c2ede72c48971ab : SUCCESS in 15m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab9e7e211d31405db7d67abca07639cc : SUCCESS in 30m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4e49bf14b776483681e97b3012942902 : SUCCESS in 16m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34d5d54dcee849279c66ea53245d40a1 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8dbda97975e45d98c03ae68b0a88bd1 : SUCCESS in 13m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3375cf1f25ea4b37936e605cf8724b1f : SUCCESS in 13m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19e2d7e2d7864d05b7b658e45ae03b88 : SUCCESS in 5m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/990bd7319d50483da9511a913939eee3 : FAILURE in 14m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d984f1869f44eed8f0fb1b598ba2414 : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/42feb86f7d584cd9acbc2b0c7a967b64 : SUCCESS in 1h 05m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5058385f037c438f98429d5e2968298d : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036c0b60ff35482dba647cdc4a78a4c4 : SUCCESS in 2m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a4d1a1c84a742c2807e793a568d87ea : SUCCESS in 22m 13s","accounts_in_message":[],"_revision_number":13},{"id":"2b9a6eda9302f0fe979899caa5b2927b50259f14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:21:43.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b5f92eb55c4c65a59578da3f3cca6d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/face32dd4be741d0a9937b2d2fa01a68 : POST_FAILURE in 1h 02m 20s","accounts_in_message":[],"_revision_number":13},{"id":"aef50f4771efe23e8723af3a3a95761321a831e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-27 07:34:40.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"755f273883c84b10ba5efc51366534181cb02037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-27 08:37:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"dacf13d6378879f97f681ae6ba633c6f5c5e3554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 09:13:53.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60c185a88a2e4697aa90be68ea79410b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edeea52d4ac3485bb680f9acc656db5d : FAILURE in 33m 12s","accounts_in_message":[],"_revision_number":15},{"id":"01836a21254c525321cf696cbd78a5cd3bb7abdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 09:53:28.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e6bf56c44dd4ecea24a7e4878a76dd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186d1592398f4b55af079d92f36bd9d5 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c65d722cb5e54d869162f69b3d4d31a8 : SUCCESS in 9m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6abc1e75cc614926807096f2b254ce78 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce219456c3634aa6986543088e719930 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1f4146f31384486a8120df3a1f32301 : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8cfe751bc6a44782981227f76af6f926 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad0c8bc01bb64302876bf57e04be5643 : SUCCESS in 22m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6b7f3bbb95c14ddb85f9c5e34ea8834a : SUCCESS in 19m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4cb5b76a1467414483a8d26ef9c580f1 : SUCCESS in 23m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6d9c13bd9e6f4a0dace373c6f9c66f7d : SUCCESS in 19m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31b0ca6894084e46a3419e316d6510be : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/00ba46dc2abb4c72b05423588100af1e : SUCCESS in 16m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b41cb7135f4f4fc0adc94b510a328101 : SUCCESS in 17m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1e723dd814b4e4f901f8cd02cebe9b2 : SUCCESS in 8m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/217249e9327046fdb6e96d2b2a048558 : SUCCESS in 22m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9288cf1888d64062a37a64c4fc467dbc : SUCCESS in 11m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07793cb28374425383ddd7ba7b8ece0f : SUCCESS in 1h 05m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe69289c3434a1cb9cf8754f379b1e2 : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7701a6b49a654b77aa609c30446f5c54 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a81c8c8b40504dc4ae4416901eb800ac : SUCCESS in 8m 52s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/docker-image/run.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/buildset-registry/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .local/share/containers/cache/blob-info-cache-v1.sqlite found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"b0a53c556cadd2f3572f16c73e14ba07b638c77a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"4afe5960f26dea8e1274ac99062a0a4460c453ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 09:14:20.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38d973ab4d01437db19e65c006b5337a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7585f0a9e11419591c7a4a2a16b09a1 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":16},{"id":"c5edfa1ff1eb4c1508c98b001b543647f9e4a1d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:35:56.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ba7b58365d4c00b2818a3d99e4a279\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff47995d3da74ebf8357a5e11c3611ea : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eb6c0c8f61144729682df89e7fa1abf : SUCCESS in 10m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/338282f10f0a4f1e8096de2eefaf902b : SUCCESS in 11m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8390bd6180a74ed4bc3caea48cfdc352 : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c22e627339514f58adcc2d4ff8c4f38a : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34b45d69aa6d4589ae44b86a22555c31 : SUCCESS in 11m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b55a63e8176f413d95ea9f6d94e9371b : SUCCESS in 26m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/43420a5a2dad4b38abff66d44a115583 : SUCCESS in 21m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3e424a8b4dc14f489b2a4193af370cff : SUCCESS in 35m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ad100fc26e6e484eac204fa2002a1001 : SUCCESS in 19m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9bec1911d72c4f30b7c339042246de5f : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ef4a5e59272402797a738e0a81856de : SUCCESS in 16m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7809c0b4e19445d7b3da34e927b7ce81 : SUCCESS in 8m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b12104ac4dd43389f64b03f6ed0f2d3 : SUCCESS in 7m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31a5a81947834d7e860f1e9286ab9cee : FAILURE in 18m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bdf973dcf874bdb8742e2a6c6821158 : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cc8d7cf8c1b4e15a0d80be55d7296c1 : SUCCESS in 57m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c02c63c765c6437c9aa0da7c88a5acce : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4317b91b819f43f5b5ebd4db6c9237a6 : SUCCESS in 2m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f6c425dcd3347d28d8c567b34dcf5b5 : SUCCESS in 19m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"234db37f37971932f9352bbcc1d59a6fc6604e2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 17:22:32.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"1b55522f4130102c231821c6063fb95e6bcb644a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 18:45:15.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad536b362e44d37a0fb9a27cef692a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac08065649f4037aa77720f07825e25 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19737b2553ef40ed9bb12ccf1c5b5282 : SUCCESS in 8m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f905b70198b4c4997e06606159d8a30 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed1c4618be974f3e883a4719740b272a : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/818a32a8ba5e499f9e60e5c773ef47fd : SUCCESS in 15m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93f5cdf2c4294ebe92dbcf73a9730f47 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e450e9dfd17b458e80c86dab2bbb4950 : SUCCESS in 24m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/52da01f4c6af4c4e92b351f3b7cdc59e : SUCCESS in 17m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/47458d9c2028468c8d7c3f63b782445f : SUCCESS in 29m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/45677aded83144029e84894d54007a99 : SUCCESS in 26m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/92df54fff55447cda1c9b76fa55f65ec : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0506eea15d804785b0d8ba48104a6c7d : SUCCESS in 17m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/361b658fee4d490f86d84577062e0bf7 : SUCCESS in 11m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e908946ba94141f9a8ef661250cdc252 : RETRY_LIMIT in 8m 42s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6bc9b31d3dcd474e85e65c575cd811f3 : FAILURE in 11m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d67585b3bbf14a7d9b3352f150d32848 : SUCCESS in 9m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d443cbd0bb1a420993a761b2dba94e2c : SUCCESS in 1h 04m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/206aeb6fe4a04e3388b371397bc32dad : SUCCESS in 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e4597b8e4c94f8385605fc8b77015c0 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e694db70cfbc46bd84bf1f56a396f5af : SUCCESS in 22m 14s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"15b048f61d20002bc3743e96ba450af6dd23fad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 02:00:33.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f943a5fb874b67a0cdde4bde869b6c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cdfdb5eefe92472b892a1e4403e40f23 : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":17},{"id":"1ed210f598fcce28b1454ad649e6e7abd9ac239b","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-09 10:33:03.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1c88bdcecf4bbb43f208b8359d038890f8abe938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"72fbe48632e58c92b0a8c2fa2a6f626c2af85977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:26:54.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70e02c37a15a42bda765f64011999644\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d64903401d004e26be0ebec1ac36b17a : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a6ed16557d34d55a9ac13ea7b82a3aa : SUCCESS in 9m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7953970c400442d8b60707f1da4a6d50 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec87f9a260ef42f7a9585174a5edb019 : SUCCESS in 22m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2830fef2204d46fc94a0cb4b838a3744 : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2ab74b60f3e4094985f37c49674f275 : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d33400983def4eb9949705156f88514d : SUCCESS in 26m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/30fb61ed1dba4a21a450f9f0a803e96c : SUCCESS in 21m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3235a41a06404b2186931ebc22f782c8 : SUCCESS in 30m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ab0ab442b37248e2bb0011fcf603e4e0 : SUCCESS in 11m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/be2893a5d8814d49a729ef52363cbc67 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/45dcd376210949cd8bd7ec064b7037dc : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/16ac80ce9c0942958240566e565976b8 : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd87a079d5d54040b14509d40034416a : RETRY_LIMIT in 4m 04s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8decab52569e4dc985e203820e1150ec : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f443f95fc9c6447793f1a2a047c2e92c : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d3845d1bbe345538ac636c382ea6a6d : SUCCESS in 58m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc839aed58084d919a65e02c95384314 : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9453720affb4520b877b62255876fab : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86f0b2bc57594e8eb26d07c31e8de432 : SUCCESS in 17m 53s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"69ab29178966a3db260613fa229af788bd85ffbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:32:04.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bee4c8a2df644fbbd9bf2c047f7f26f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c19743f1c5846bda4ebe427e6142d32 : FAILURE in 1h 07m 55s","accounts_in_message":[],"_revision_number":18},{"id":"49d25243fc91c2702bb08c4631b695cc388a4464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"975aa5e1d9d97ba18942e0b1a2594aa73b06d49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 20: New patch set was added with same tree, parent tree, and commit message as Patch Set 19.","accounts_in_message":[],"_revision_number":20},{"id":"5c4ea0e08f5a00a579dc258913c63586bfadfc5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:47:24.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e9117e124cb4b01a2fe1b84b7b519c5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b1232e36f844656ad7a626336d1d80d : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":20},{"id":"5aec08dd738bd4bb235abc881422709e5d23b44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:51:13.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf0208d3536347c9b64d9fbf9edf56df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360dcab8ce8443b29dab5099e270e9e7 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a99ba11f11b345a79bf61f000cdf550f : SUCCESS in 10m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5bf99e2ee1f447da9e5cfd65d19fc660 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da4ab375ba3f47a2a9f133d21ff6449a : SUCCESS in 12m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10d16deab5b94062a2516455e3243778 : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b80e9abca4564cd6b231fec79937c50a : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50922df9b8f3488cb7214a2803dc6610 : SUCCESS in 26m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e59d63821a324a439e530072a1180663 : SUCCESS in 12m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0d2d9a23460047ce82b6501de700e241 : SUCCESS in 31m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6589174eaa4240389485146e898fcb43 : SUCCESS in 16m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e86676543fd48ddb69666eef200c6cb : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e18c0325005643de9bcfbc30deb47b9d : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5e6156250384a8aa5be61b01f500796 : SUCCESS in 9m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e10e69aedc9a47abaf89591ec5512c39 : RETRY_LIMIT in 4m 06s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63f219708b8c4cbbb89b13ad13eac31c : SUCCESS in 11m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c4d6ecd66fa4a9e81f84a7a8469ca88 : SUCCESS in 12m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24f0cd08cd774b13ac40034cdada5aaf : SUCCESS in 1h 02m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e738bf6f8ac84aa480e75ea31a0a06f1 : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6bae88a1d649238f7d6b278056675e : SUCCESS in 3m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88fcdaf21d574bf2a633b0b9c6d4f591 : SUCCESS in 14m 56s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"4346945b53942d7141ff263c360ebcae60c9be46","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-21 10:43:43.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"34ba94838fbcb1abcba43d603325991caaaf5c4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 06:02:18.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"7fd036ff32adb873635de6a03197ff5ac903baac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:28:58.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d97928e561a147f0aa5a85cc1df26b65\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937fef044503494c88a77ad034a34eb4 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d142760b4d3d447c858d238a1d713ec5 : SUCCESS in 9m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3508b4a56694b3981dd4037b629b0aa : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35c4590ceb6d4829b4d8e2267b39e920 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2ae8ccae4db4ba2a5ea6fe5ea39b1ee : SUCCESS in 22m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0d54b15e24946028eb4d26a2f09998e : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/520fe1f4814b483989001b90476d8da4 : SUCCESS in 25m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c213a461e29e4034a7597d4482437bc0 : SUCCESS in 12m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5e1e84ef6f174293bf1493ce217af02a : SUCCESS in 34m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/733ea3470185487da1e0324de4ebe91e : SUCCESS in 19m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e80a2dcd8ed413b8c6e67eeba46db89 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/52af451a96544a5180f7a5c92b49992f : SUCCESS in 8m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ea90217e0174481ca0c803af6296e2a4 : SUCCESS in 11m 22s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cef433a554941e7b21f34c42cbc08ed : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c922c10129bc4349b538d5994b08ab2c : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82dc0cbc7feb4ff6b85f9a1606bd148a : FAILURE in 5m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7cdd62fa92474cc8b82a9c2760b304f8 : SUCCESS in 6m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3071065965c41c68ad7e086b17e3639 : SUCCESS in 1h 02m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ad551040e054796ad170fcc59ede82a : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99386f6ea1154e8b81bdacfd7038e67e : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1396abe4e02541c88e7962af1cbca1b5 : SUCCESS in 16m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"ed3205b5d778788e49d54d388b6a9b6d603de902","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:52:52.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6da469c09cc34e37a4f634b7f8f9dd34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69f3405763744268ae75e5a172b9b8e0 : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":21},{"id":"5408be9f253200e34f0783dd53f4c7d4d013cb17","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 08:53:10.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"97e461567b4fa4c06db8cba53f30050b8572dac3","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 08:53:57.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"5825d80928a90a30af2a4b4249cfb0bc961b2bbe","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 14:35:49.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"90525a56a7e2dd9f8fdecd22b251a96ab6eb5a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:37:25.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"4f4b5c302e56c0f59bee7996772e32bac059866a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:48.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5db971f846b4c68b5f0bd9e076e8549\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0e5346174c4f56bafffd297ee36a85 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d057d884d4b47a4931904c968cfba3e : SUCCESS in 7m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6c891cef3d60492e8794ccfbbacf87b0 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fd630d37109a49518b704cb3fc7fe11b : SUCCESS in 26m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2f5b4e3d3648487981e481c632a6dbef : SUCCESS in 20m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f1555531a9964e6586b8fd0c070dbd6f : SUCCESS in 32m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/17f1d3d9a73f4d4e9f97785af1932abf : SUCCESS in 13m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dda74ec83b8540f5935f9d43a9d50f6a : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3c718bdb59f4f869c514543c06ae22b : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/77aa6591e74942399b3df1f034f436d9 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd9e15c159ce4d32adda31b0dad030c0 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0fbdcf6c41144f4892ae63d339195468 : SUCCESS in 9m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aadc6ec7b7d1439d879bec69f7894310 : SUCCESS in 56m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2a574a923074d7292baad67d027c2ea : SUCCESS in 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4837221f595747779d99b7d34c86f02f : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ede30d95c5684668978c45e66da57511 : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":21},{"id":"e47ffcd6635fd971ad49a8f9e2b0f42b9ad2ab84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"ba2dbede0ba313979d7802b8bda1adfd9499b876","revisions":{"da3b9505ffb2202b5c7b56d7f0ef27c866fefc43":{"kind":"REWORK","_number":1,"created":"2026-03-10 07:49:33.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/1"}}},"commit":{"parents":[{"commit":"8ce04f5f4ef2118593bdc5d61fc425f451eb3480","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ce04f5f4ef2118593bdc5d61fc425f451eb3480"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-10 07:41:31.000000000","tz":60},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da3b9505ffb2202b5c7b56d7f0ef27c866fefc43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da3b9505ffb2202b5c7b56d7f0ef27c866fefc43"}]},"branch":"refs/heads/feature/threaded"},"fd2f46e14c5ac000fa805e04ae7882a8058978e1":{"kind":"REWORK","_number":2,"created":"2026-03-14 10:22:00.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/2"}}},"commit":{"parents":[{"commit":"7ff5e599f177850a399ac6670f0bd345090f1062","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ff5e599f177850a399ac6670f0bd345090f1062"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-14 08:44:43.000000000","tz":60},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd2f46e14c5ac000fa805e04ae7882a8058978e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd2f46e14c5ac000fa805e04ae7882a8058978e1"}]},"branch":"refs/heads/feature/threaded"},"0c43af2cb5b7e69bcef82daed4c063a1e8d5bdc6":{"kind":"REWORK","_number":3,"created":"2026-03-18 16:58:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/3"}}},"commit":{"parents":[{"commit":"c3bac034bb5210a15c00cf36dc780f70007988de","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3bac034bb5210a15c00cf36dc780f70007988de"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-18 16:24:14.000000000","tz":60},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c43af2cb5b7e69bcef82daed4c063a1e8d5bdc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c43af2cb5b7e69bcef82daed4c063a1e8d5bdc6"}]},"branch":"refs/heads/feature/threaded"},"a630548727fd1c190739585a03905f19e495629a":{"kind":"REWORK","_number":4,"created":"2026-03-23 10:33:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/4"}}},"commit":{"parents":[{"commit":"47fd236072fc8841bb25d1875f6bd26eca00ef89","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47fd236072fc8841bb25d1875f6bd26eca00ef89"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 08:40:22.000000000","tz":60},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a630548727fd1c190739585a03905f19e495629a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a630548727fd1c190739585a03905f19e495629a"}]},"branch":"refs/heads/feature/threaded"},"21c1778f3d855dc42f863c6d842e8478cde525b7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-23 10:55:39.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/5"}}},"commit":{"parents":[{"commit":"cf8831e9f79508fdec20011adac2d9f92b188e4e","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf8831e9f79508fdec20011adac2d9f92b188e4e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:55.000000000","tz":60},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21c1778f3d855dc42f863c6d842e8478cde525b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21c1778f3d855dc42f863c6d842e8478cde525b7"}]},"branch":"refs/heads/feature/threaded"},"198b495d3eb647eea841c3d9a04a815628fa1440":{"kind":"REWORK","_number":6,"created":"2026-05-13 09:16:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/6"}}},"commit":{"parents":[{"commit":"89ee5721d111bf7391cf2638553ba603a541af74","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89ee5721d111bf7391cf2638553ba603a541af74"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-13 07:02:43.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/198b495d3eb647eea841c3d9a04a815628fa1440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/198b495d3eb647eea841c3d9a04a815628fa1440"}]},"branch":"refs/heads/feature/threaded"},"d72a801c3f3cf3a459e367fea717f6b9a4aa154f":{"kind":"REWORK","_number":7,"created":"2026-05-30 16:41:57.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/7"}}},"commit":{"parents":[{"commit":"89ee5721d111bf7391cf2638553ba603a541af74","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89ee5721d111bf7391cf2638553ba603a541af74"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@nvidia.com","date":"2026-05-30 13:47:04.000000000","tz":0},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d72a801c3f3cf3a459e367fea717f6b9a4aa154f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d72a801c3f3cf3a459e367fea717f6b9a4aa154f"}]},"branch":"refs/heads/feature/threaded"},"fc8bce3b6ce8fdb4ce3d80aa278c4ca0d60bdec7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/8"}}},"commit":{"parents":[{"commit":"be813aba5bd1a660dcdef721928752e54c8cf5a8","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be813aba5bd1a660dcdef721928752e54c8cf5a8"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc8bce3b6ce8fdb4ce3d80aa278c4ca0d60bdec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc8bce3b6ce8fdb4ce3d80aa278c4ca0d60bdec7"}]},"branch":"refs/heads/feature/threaded"},"52366a3b2a93a7bfc2492bd5c55b30fe37982446":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/9"}}},"commit":{"parents":[{"commit":"deaa5ce46a12301d0d0a9379b9e78ce46ae1e41b","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/deaa5ce46a12301d0d0a9379b9e78ce46ae1e41b"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52366a3b2a93a7bfc2492bd5c55b30fe37982446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52366a3b2a93a7bfc2492bd5c55b30fe37982446"}]},"branch":"refs/heads/feature/threaded"},"863058cbca62d4a585052b47e207a1a613f34518":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-02 13:27:47.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/10"}}},"commit":{"parents":[{"commit":"36435ac2912fd48af113fe49dd35f1d2cd7548ce","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36435ac2912fd48af113fe49dd35f1d2cd7548ce"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-02 13:06:36.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/863058cbca62d4a585052b47e207a1a613f34518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/863058cbca62d4a585052b47e207a1a613f34518"}]},"branch":"refs/heads/feature/threaded"},"1a35004f1a4e5945a62ed8590f2bc3c4f0a24c3d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/11"}}},"commit":{"parents":[{"commit":"11c65e0f75d1a427465e34108edec5b116ffc16f","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11c65e0f75d1a427465e34108edec5b116ffc16f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:17:51.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a35004f1a4e5945a62ed8590f2bc3c4f0a24c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a35004f1a4e5945a62ed8590f2bc3c4f0a24c3d"}]},"branch":"refs/heads/feature/threaded"},"0c2b5a053d09ab5295c8b08b5c3e5f9117f37a84":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/12"}}},"commit":{"parents":[{"commit":"27a768660a418392f71bf67d41bcd19d8ee9d974","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a768660a418392f71bf67d41bcd19d8ee9d974"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 16:12:39.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c2b5a053d09ab5295c8b08b5c3e5f9117f37a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c2b5a053d09ab5295c8b08b5c3e5f9117f37a84"}]},"branch":"refs/heads/feature/threaded"},"b5501dde71cbb508df6acd2b3f541c94bcd3bcbc":{"kind":"REWORK","_number":13,"created":"2026-06-12 07:10:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/13"}}},"commit":{"parents":[{"commit":"27a768660a418392f71bf67d41bcd19d8ee9d974","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a768660a418392f71bf67d41bcd19d8ee9d974"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-12 07:07:48.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5501dde71cbb508df6acd2b3f541c94bcd3bcbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5501dde71cbb508df6acd2b3f541c94bcd3bcbc"}]},"branch":"refs/heads/feature/threaded"},"e71f4f065037a5f9f554ec1a29288b943429246a":{"kind":"REWORK","_number":14,"created":"2026-06-27 07:34:40.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/14"}}},"commit":{"parents":[{"commit":"27a768660a418392f71bf67d41bcd19d8ee9d974","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a768660a418392f71bf67d41bcd19d8ee9d974"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-27 07:34:02.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71f4f065037a5f9f554ec1a29288b943429246a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e71f4f065037a5f9f554ec1a29288b943429246a"}]},"branch":"refs/heads/feature/threaded"},"2debb1fd451800bb50eb068f8f3aa304ac40b698":{"kind":"REWORK","_number":15,"created":"2026-06-27 08:37:45.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/15"}}},"commit":{"parents":[{"commit":"27a768660a418392f71bf67d41bcd19d8ee9d974","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a768660a418392f71bf67d41bcd19d8ee9d974"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-27 08:37:07.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2debb1fd451800bb50eb068f8f3aa304ac40b698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2debb1fd451800bb50eb068f8f3aa304ac40b698"}]},"branch":"refs/heads/feature/threaded"},"950b7e33ea29d6b74d73db9256f5d7e9e21cf867":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/16"}}},"commit":{"parents":[{"commit":"360c69385749aa2af89743734a90dca6fadf0f1c","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/360c69385749aa2af89743734a90dca6fadf0f1c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune gunicorns a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/950b7e33ea29d6b74d73db9256f5d7e9e21cf867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/950b7e33ea29d6b74d73db9256f5d7e9e21cf867"}]},"branch":"refs/heads/feature/threaded"},"830fd65e885a66608e83653368265daccf08ab4e":{"kind":"REWORK","_number":17,"created":"2026-07-07 17:22:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/17"}}},"commit":{"parents":[{"commit":"d9f8f5578c2c65d43ec32ac0169c1a97094b62d1","subject":"Close sockets properly in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9f8f5578c2c65d43ec32ac0169c1a97094b62d1"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nTo tune Gunicorn, a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/830fd65e885a66608e83653368265daccf08ab4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/830fd65e885a66608e83653368265daccf08ab4e"}]},"branch":"refs/heads/feature/threaded"},"596e7db80dab96fd1ce1c0caa0e2011033cf1717":{"kind":"REWORK","_number":18,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/18"}}},"commit":{"parents":[{"commit":"d39a71c524fd74f46912da929d1f318d1c1e1126","subject":"Skip importing eventlet in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d39a71c524fd74f46912da929d1f318d1c1e1126"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nBecause gthread workers now run concurrent replicated GETs on real\nthreads, guard the GetOrHeadHandler status/body/source bookkeeping with a\nper-handler lock (a no-op under eventlet) so it cannot race; the EC-only\nlocking arrives with the later EC work.\n\nTo tune Gunicorn, a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nReliable 408 responses for a client that stalls mid-body (read-only\nsocket shutdown, unread-body detection, forced close) are a documented\nfollow-up: they depend on the patch_gunicorn machinery completed in the\nwsgi-completion commit.\n\npatch_gunicorn() is idempotent: repeated calls (e.g. on config reload)\nmust not stack the wrapping patches.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/596e7db80dab96fd1ce1c0caa0e2011033cf1717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/596e7db80dab96fd1ce1c0caa0e2011033cf1717"}]},"branch":"refs/heads/feature/threaded"},"c7c6320e815a28e6ec33faf032ed8de340ad1bc2":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/19"}}},"commit":{"parents":[{"commit":"6c5728a51957961180c8c661ecb828209cf04ec3","subject":"Skip importing eventlet in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c5728a51957961180c8c661ecb828209cf04ec3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:18.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nBecause gthread workers now run concurrent replicated GETs on real\nthreads, guard the GetOrHeadHandler status/body/source bookkeeping with a\nper-handler lock (a no-op under eventlet) so it cannot race; the EC-only\nlocking arrives with the later EC work.\n\nTo tune Gunicorn, a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nReliable 408 responses for a client that stalls mid-body (read-only\nsocket shutdown, unread-body detection, forced close) are a documented\nfollow-up: they depend on the patch_gunicorn machinery completed in the\nwsgi-completion commit.\n\npatch_gunicorn() is idempotent: repeated calls (e.g. on config reload)\nmust not stack the wrapping patches.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7c6320e815a28e6ec33faf032ed8de340ad1bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7c6320e815a28e6ec33faf032ed8de340ad1bc2"}]},"branch":"refs/heads/feature/threaded"},"fcd99f7ce0be57304559724e62a466c72df9a341":{"kind":"NO_CHANGE","_number":20,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/20"}}},"commit":{"parents":[{"commit":"3c33599a45eef1cbdf52ec74d00540b8528f7cb1","subject":"Skip importing eventlet in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c33599a45eef1cbdf52ec74d00540b8528f7cb1"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nBecause gthread workers now run concurrent replicated GETs on real\nthreads, guard the GetOrHeadHandler status/body/source bookkeeping with a\nper-handler lock (a no-op under eventlet) so it cannot race; the EC-only\nlocking arrives with the later EC work.\n\nTo tune Gunicorn, a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nReliable 408 responses for a client that stalls mid-body (read-only\nsocket shutdown, unread-body detection, forced close) are a documented\nfollow-up: they depend on the patch_gunicorn machinery completed in the\nwsgi-completion commit.\n\npatch_gunicorn() is idempotent: repeated calls (e.g. on config reload)\nmust not stack the wrapping patches.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcd99f7ce0be57304559724e62a466c72df9a341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcd99f7ce0be57304559724e62a466c72df9a341"}]},"branch":"refs/heads/feature/threaded"},"ba2dbede0ba313979d7802b8bda1adfd9499b876":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-08-21 06:02:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/90/979790/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/979790/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/979790/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/979790/21"}}},"commit":{"parents":[{"commit":"64cb0597022b9c52c9218c08ae09c18ba530336d","subject":"Skip importing eventlet in threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64cb0597022b9c52c9218c08ae09c18ba530336d"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2025-09-02 14:24:38.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Add Gunicorn using gthreads as wsgi server","message":"Add Gunicorn using gthreads as wsgi server\n\nIf you want to run Swift without eventlet you need to set the\nenvironment variable \"USE_EVENTLET\u003d0\". For example, using swift-init:\n\n        USE_EVENTLET\u003d0 swift-init restart main\n\nYou can also uninstall eventlet before (re-)starting the Swift services\nto ensure it is working without it.\n\nDue to a more strict validation of requests, headers and different\nencoding a bit of monkey patching is required to match the current\nbehaviour running Swift with eventlet.\n\nBecause gthread workers now run concurrent replicated GETs on real\nthreads, guard the GetOrHeadHandler status/body/source bookkeeping with a\nper-handler lock (a no-op under eventlet) so it cannot race; the EC-only\nlocking arrives with the later EC work.\n\nTo tune Gunicorn, a new setting \"threads\" has been added to the\n*-server.conf files. It defaults to 4, but especially the proxy-server\nand object-server might be more performant using a higher setting. Note\nthat this is the number of threads per worker; the number of workers is\nby default set to the number of CPU-cores.\n\nFunctional tests (still requiring eventlet at this point) should pass\nwith this implementation, except for the xprofile middleware which\ndepends on eventlet.\n\nThis implementation does not yet work with EC; this needs additional\nchanges to support 100-Continue requests and will be added in a\nfollowup.\n\nReliable 408 responses for a client that stalls mid-body (read-only\nsocket shutdown, unread-body detection, forced close) are a documented\nfollow-up: they depend on the patch_gunicorn machinery completed in the\nwsgi-completion commit.\n\npatch_gunicorn() is idempotent: repeated calls (e.g. on config reload)\nmust not stack the wrapping patches.\n\nChange-Id: I9a90376a67d084e943446ba1c82c3be197187155\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba2dbede0ba313979d7802b8bda1adfd9499b876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba2dbede0ba313979d7802b8bda1adfd9499b876"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
