)]}'
{"id":"openstack%2Fswift~980178","triplet_id":"openstack%2Fswift~master~If3cefc156e705ed6a0877f04e0499e0dc8e5f299","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-14 19:21:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If3cefc156e705ed6a0877f04e0499e0dc8e5f299","subject":"obj: Use conn_timeout more","status":"MERGED","created":"2026-03-12 00:07:59.000000000","updated":"2026-07-14 21:32:11.000000000","submitted":"2026-07-14 21:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980178","meta_rev_id":"563ea5ec5ccb7869e3a69c92c9e9217e550be4c5","_number":980178,"virtual_id_number":980178,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-14 21:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-03-12 03:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-07-14 06:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 00:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-12 01:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-12 03:51:16.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"db7160ba692e1f24f790cfa0e34a7508abca72d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-12 00:07:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6826f69947e945a1fd81d2228a8ffe827348751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 00:38:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f2abf4beec847478bbfc443da6c224a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba8e85d63e3d433dbc7d2b5aee571122 : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d0dd774e5116c351789b5c5b3537069223e77fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 01:17:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f086b39db2a490ab9cb6268d883943e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab2584def924fedb5ba778110d587ed : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/72c0cf52ec524c82aa25f0af9b2e81f0 : SUCCESS in 1h 00m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a0f75c58a384354933eb7745779149a : SUCCESS in 1h 00m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c635517d80247e194d47e5b084388ec : SUCCESS in 57m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb6dce41606f4eda898e23b4b2eb4034 : SUCCESS in 33m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b817254246c24fe8abfc32b5acd5612a : SUCCESS in 11m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5fc5f4e098e47f3896009670e4e7828 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d31e0eebf9884185a3ebf889b1f119da : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1edfdba93b0643e6bbab2c49348b5228 : SUCCESS in 8m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f166ff0ef5ad4bc2bd1e66b79582d107 : SUCCESS in 9m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/209e450912bb4f029aac16aa5d1c0357 : SUCCESS in 22m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2f1898a344f54ddfa0a41abc6cff82c2 : SUCCESS in 10m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/deda1985407445c9ab55466c0f5f4e7a : SUCCESS in 20m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cea40811136448cd9bcfb44839cdb0ac : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/91ca972e7e444da696cfa94a053379b2 : SUCCESS in 9m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4a4e365f9383403ea61147d57e043cec : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6df1be0d6f714ad8858195140fbe6f17 : SUCCESS in 5m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/761fb8b8c4794c748ecdbc01bdad656b : SUCCESS in 7m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f20c54232f64411b7c42a43fc5cd5df : SUCCESS in 10m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f86b0a53a274943a279692cfc82e9f0 : SUCCESS in 59m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8dd29307392416bbff4d9af1a9c8927 : SUCCESS in 32m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82f7757bcde64d76b0c1847584ece3e0 : SUCCESS in 52m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/381fa06a638041cc8920d6dd15252d94 : SUCCESS in 27m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b9c080942734a0d8474ef5e2d66c65c : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/605d1037ee844dbfa29f24a6f1d9bdd2 : SUCCESS in 13m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50cb2372d1774136837c359758d13b09 : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6f452ab976f2c2a4f35692c9dc8966ca70e86677","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-03-12 03:51:16.000000000","message":"Patch Set 1: Code-Review+2\n\nI really like this. Breaking the timeout down into conn timeout while were actually trying to connect and then timeout to get all the data, feels more fine grained. And better we can just increase the conn timeout when we need to give more connection time.\n\nOnly downside we now give them a little more time to get us data, but the fact that we can move on to the next node when we can\u0027t talk to one in a few seconds rather then a minute is a huge win IMHO.\n\nNot sure how this can really be tested","accounts_in_message":[],"_revision_number":1},{"id":"1c9c6fe7eceae3e9d111d30ab075e55372ed5d90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-28 17:27:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c6469ae5787eb4e220714832b96aa76dbaa0a49","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-07-14 06:00:46.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daf9c167ff52c3baf01b9ccd28457eec6fac5a7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 06:01:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df30b51d98a494d03104491d72542fb28608560d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 07:16:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10d903e8b0a44f37a6609797d8545759\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47e00b1c8ea4f7eb8d328eb4b96e41b : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f7f84a7d12e4d5c8e0789ec39768147 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/22eab905ff3c42ddabe698b97e015958 : SUCCESS in 1h 10m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d734af06cf9405d9c2aa5701d266c08 : SUCCESS in 30m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc195d3cfa23428b980c590984339959 : SUCCESS in 50m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddfaeccca68c40f88e2ae6cf12cb58a8 : SUCCESS in 55m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/38ae08fb3f0f46b1ac66155999b73a4e : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3980ad7b12db4dcfa5902dd37dbc0dd3 : SUCCESS in 26m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a40377d0c7434fe6b244e709517663ea : SUCCESS in 9m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/124a3d6795824f9ea40192d9f24f4987 : SUCCESS in 11m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/97d13fb8aeba4320bf60f5dc0e5d331f : SUCCESS in 21m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4d06be504a3044b78ce0f9375094a016 : SUCCESS in 8m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5353c0e8b427450e9bbe8941d4690e16 : SUCCESS in 12m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6024b17fa9b54e14b15d6aa0291a62e0 : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6800acdb882d4bb3a9c5d5162c0be0d1 : RETRY_LIMIT in 3m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a43684088a45408d831aea25c1e050e0 : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/91e089ac591c4223874ae645fa87d56d : SUCCESS in 57m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0a79ae5e40c45738b664d6e46449f54 : SUCCESS in 53m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/86a6eeae75914e05bf40f9d55abdfa69 : SUCCESS in 37m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/331189a0f3d74d8289004f15c5aebe81 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a043c50961c45b4b7cd874546472f53 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/438d925a4f8042998626b9d887ff18c8 : SUCCESS in 25m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e239f083ad4bc391be75bdce3a722c : FAILURE in 1h 01m 11s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"19d5a934e2352ceb7262fcb13b7d65c730dcea78","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-14 19:21:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6eda9549e21cde9e1cd9db67659b7b192a732e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 20:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c71cf64e0414239b3d6d60db9df87d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb5b903e3f54aa8bcb294ec6ff013aa : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60cdd2781d1542778bc3edde5e693d15 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/25a8bee499984f9bb31c316fe86f4a00 : SUCCESS in 38m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/55ef1c8bc96f444f854eb18b2446fd4d : SUCCESS in 30m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee8dfa198d0c4c74a113592fe4290d17 : SUCCESS in 40m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7746240480d045e9987daf065f97ed23 : SUCCESS in 46m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8dc4e3b61854dabba6640b268f2e8e8 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44c1cc108ca742c3848403c5fe679fac : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8101e78afe2a493f96f5454e501cf5ce : SUCCESS in 10m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e91c5ca4eb546a2bcaa0ed37b0014f1 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/297e7d57eccd4130852e1f01bcaba651 : SUCCESS in 24m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/37603da0aacc49148611afbe53b45e8b : SUCCESS in 20m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/607c93bc9739465896318fef766d081d : SUCCESS in 13m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/24d6c40944ef4bb8ac6659f34e19dfa6 : SUCCESS in 9m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f9d6675b3a03492d9a31d359e3ff8253 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30432faf2b224301b4f99f344a6df411 : SUCCESS in 9m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/96b747c1274f4c7ab484db890af6b3a7 : SUCCESS in 8m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ffc2cc1968145878bfcf095cb7e64a9 : RETRY_LIMIT in 3m 53s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1b86921a5d62438a82612666bd4e1b5b : FAILURE in 4m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/36486bc76fc848968d04c03ce30e9be3 : SUCCESS in 11m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aecdbc203253432496c1695bb0f9259a : SUCCESS in 44m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/933dc5e5b4aa486bb1b210ff9eb7b6b4 : SUCCESS in 34m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e98f2c74d2740678536cdc90520c06f : SUCCESS in 50m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b03ab5dc7f4248629843c585ac090509 : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22f6e29e82a46c0885a38a436e8a232 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6bfd13037654148a286eb6709cd6384 : SUCCESS in 20m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5801f62ee96a4e06a429703050f66e6d : SUCCESS in 54m 11s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"6fd51bc9e017148d7af8acb8c44e00fc9e48c822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 20:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"830ee796b6cf3734f0a3017fc242e0f27e81e3d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 20:22:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0f7ed0d31d4ad6ab28b6e73a23dc1d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/437065251e584808b1f9c4594bcb9bd9 : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":1},{"id":"52e010f5dcefd12a10c7a278b5c470149c940da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 21:31:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f976fae02ec4b06bdf69ba37f34c5c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b27800766c341b5925c5af57bc18603 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28dfa01478f445e0944ca3c04d0e96de : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/735bb85bf9a74714b4ddf4dd5cc4f8c0 : SUCCESS in 35m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df1fef58194a4dbf9d2ae24e9540c403 : SUCCESS in 1h 07m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f2ee950b2a4491c82cd8966a1db1c30 : SUCCESS in 36m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/283708f1a4f6430d8bffa27f20fd237c : SUCCESS in 32m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/96b7527037024c3cb4de59dfd0d6530d : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5da0e29335954914be722efb920af08a : SUCCESS in 25m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/402e6cb709504eb59c5fdc9a5a92155f : SUCCESS in 22m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ef1c735b50cd4aa09d3e28b21eae384a : SUCCESS in 33m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b57f600a43d5400c9f178c2b92736048 : SUCCESS in 23m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3fb30b38b44d4195a1cd74d028ecd55a : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1deb830d553a4144a382757c1b3e4946 : SUCCESS in 9m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/075ad1abce31430eb9b682996ab40fc8 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fb49e09ef07498e81085cf8f0fe52d3 : RETRY_LIMIT in 7m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/97e7d7578d9f4120af649b572e2c810e : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0f8fd4381924e62a73896e2e92e7d4d : SUCCESS in 1h 02m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ee80a2b66a1486ea8dc974db01b8366 : SUCCESS in 22m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96845605f9bd45b894e2b21b8aac199c : SUCCESS in 22m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1c0b938da20422e88887f0a18ad2bff : SUCCESS in 24m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1066f8f3305348e9b663966ec3e7460b : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/530211c72e6443668459408e5f4097a5 : SUCCESS in 10m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae089444dbe4410097f3ba339cf0ba91 : SUCCESS in 59m 37s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b4a676ffcaa56431ff642fdf25e09f6f8f806fc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 21:31:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"563ea5ec5ccb7869e3a69c92c9e9217e550be4c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 21:32:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4268ba7b4c754fb8b71e3523661e5fbc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f1b78614274fcc984748efc16216c2 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/762b7e027cac40f99e6ae1bfd10a7b68 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53b8e6e19b4523c25044e8d5bb347f0a52d4f4d3","revisions":{"53b8e6e19b4523c25044e8d5bb347f0a52d4f4d3":{"kind":"REWORK","_number":1,"created":"2026-03-12 00:07:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/980178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/980178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/980178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/980178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/980178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/980178/1"}}},"commit":{"parents":[{"commit":"bc60bdff0a024c046597dceb90c68c7110d6ce5d","subject":"CI: Bring gate in line with Gazpacho testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc60bdff0a024c046597dceb90c68c7110d6ce5d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-12 00:03:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-12 00:04:39.000000000","tz":-420},"subject":"obj: Use conn_timeout more","message":"obj: Use conn_timeout more\n\nPreviously, we were willing to wait the full http_timeout (default 60s)\nto connect to another node. It was a combined timeout for connecting\n*and* waiting for a response, but if we can\u0027t connect quickly, we should\njust bail.\n\nChange-Id: If3cefc156e705ed6a0877f04e0499e0dc8e5f299\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53b8e6e19b4523c25044e8d5bb347f0a52d4f4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53b8e6e19b4523c25044e8d5bb347f0a52d4f4d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
