)]}'
{"id":"openstack%2Fswift~980387","triplet_id":"openstack%2Fswift~master~I5e30015fc1beca4a0372b0988d4c3942bb7af2fa","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-13 02:39:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-03-13 02:39:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-03-17 23:20:42.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"hashtags":[],"change_id":"I5e30015fc1beca4a0372b0988d4c3942bb7af2fa","subject":"A probe test to verify jitter timestamp works to prevent collisions.","status":"NEW","created":"2026-03-13 02:38:44.000000000","updated":"2026-03-18 00:50:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9b9d450e17cedd5b0195f0f18412d833934decec","_number":980387,"virtual_id_number":980387,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-18 00:50:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 02:39:10.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-03-13 02:39:10.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-03-13 04:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f480da2fb7555a25cfeccdbe9abded25f7b2522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-13 02:38:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d96ce74737d7dd7301f4495da26ccdb315a04fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 04:16:20.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/caddb3f0ecb8490c973ebe8f46118c97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f8348954cd432ca7a80c6e1b891c0e : SUCCESS in 10m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1077cd6544c5417985808658563c4bc6 : SUCCESS in 1h 06m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4266db5317a44f838760eb5cbfdeb083 : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2b687776edf137d459672251dc96f0d91403a991","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 04:24:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12e9f72bc6748e2aa5f5a8dc7a07df0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e7d0a7378524428a1df5c9cf4784ab7 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1f04dca62fc7240241c71121bf969efd50168b75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-14 00:19:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8c2299df96c76b7d5cc1aad33775573078a9e057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 00:19:35.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 968740,41","accounts_in_message":[],"_revision_number":2},{"id":"c1a986cf0706dbaa88d67dc785bdbe78604449fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 00:19:35.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 968740,41","accounts_in_message":[],"_revision_number":2},{"id":"4b45b913b811f94cf3831649ea86e1bd3cf2426c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-17 23:17:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c27f00b2b58a2a2106955d35daa73395e5d38ea","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-17 23:20:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22bc78f9a79332de5714ebf42918a1406a87ac49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 00:35:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/076c86e172014596a835882ef9a60148\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3662548eb47243d8abff257124a9bf79 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9b9d450e17cedd5b0195f0f18412d833934decec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 00:50:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53c19323d2df44438f2868f190c3ee9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01c17054b8c4386b8089c7b18bff537 : SUCCESS in 9m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aedf572e7c14d6d943aced870531d70 : SUCCESS in 1h 24m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d07616fb5114d01b0af543c8baae2ce : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bc5c725aaa7f7626f97026b385211f76e95e24de","revisions":{"765bf5e8bc117f36851c71a2f3a5c8321b873f6b":{"kind":"REWORK","_number":1,"created":"2026-03-13 02:38:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/87/980387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/980387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/980387/1"}}},"commit":{"parents":[{"commit":"57c4bf8d7ad2d61e5d4ae92556cd88532e1e63f9","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57c4bf8d7ad2d61e5d4ae92556cd88532e1e63f9"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-13 02:37:59.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-13 02:37:59.000000000","tz":-420},"subject":"A probe test to verify jitter timestamp works to prevent collisions.","message":"A probe test to verify jitter timestamp works to prevent collisions.\n\nChange-Id: I5e30015fc1beca4a0372b0988d4c3942bb7af2fa\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/765bf5e8bc117f36851c71a2f3a5c8321b873f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/765bf5e8bc117f36851c71a2f3a5c8321b873f6b"}]},"branch":"refs/heads/master"},"9f34b428f0be17bb85eadba1c2f462ada25ae33f":{"kind":"REWORK","_number":2,"created":"2026-03-14 00:19:08.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/87/980387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/980387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/980387/2"}}},"commit":{"parents":[{"commit":"57c4bf8d7ad2d61e5d4ae92556cd88532e1e63f9","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57c4bf8d7ad2d61e5d4ae92556cd88532e1e63f9"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-13 02:37:59.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-14 00:17:26.000000000","tz":-420},"subject":"A probe test to verify jitter timestamp works to prevent collisions.","message":"A probe test to verify jitter timestamp works to prevent collisions.\n\nChange-Id: I5e30015fc1beca4a0372b0988d4c3942bb7af2fa\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f34b428f0be17bb85eadba1c2f462ada25ae33f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f34b428f0be17bb85eadba1c2f462ada25ae33f"}]},"branch":"refs/heads/master"},"bc5c725aaa7f7626f97026b385211f76e95e24de":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-17 23:17:24.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/87/980387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/980387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/980387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/980387/3"}}},"commit":{"parents":[{"commit":"d09e4b5579a3db80f8bfee8486b9b9d7b2b00da7","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d09e4b5579a3db80f8bfee8486b9b9d7b2b00da7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-13 02:37:59.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-17 23:17:05.000000000","tz":-420},"subject":"A probe test to verify jitter timestamp works to prevent collisions.","message":"A probe test to verify jitter timestamp works to prevent collisions.\n\nChange-Id: I5e30015fc1beca4a0372b0988d4c3942bb7af2fa\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc5c725aaa7f7626f97026b385211f76e95e24de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc5c725aaa7f7626f97026b385211f76e95e24de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
