)]}'
{"id":"openstack%2Fswift~980785","triplet_id":"openstack%2Fswift~master~I1e0052717d9b25f97e3cf58508d005f43d306267","project":"openstack/swift","branch":"master","topic":"check-db-conns-harder","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-06-26 17:33:26.000000000","reason":"Change was submitted"},"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-06-26 17:33:26.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-06-26 17:33:26.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-06-26 12:21:35.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-06-26 17:33:26.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-06-26 17:33:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e0052717d9b25f97e3cf58508d005f43d306267","subject":"Close DB connections in metadata servers","status":"MERGED","created":"2026-03-16 16:37:58.000000000","updated":"2026-06-26 17:35:09.000000000","submitted":"2026-06-26 17:33:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":74,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"980785-check-db-conns-harder","meta_rev_id":"c6d20929422e8462eca3aac0dc1c1d2b0764d50d","_number":980785,"virtual_id_number":980785,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 17:33:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2026-06-26 16:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2026-06-26 16:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-16 16:38:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-03-16 17:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 16:45:34.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-04-13 03:26:21.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2026-04-14 07:25:30.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-04-23 04:28:37.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-05-07 13:06:36.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"CC"}],"messages":[{"id":"45a3820235651b8a6b143a5ba284694702083de9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-16 16:37:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"676909fd3a0ea6e21479949f4b6e2c264412d2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 17:52:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac99d93ddb8414390c442f4a4629b5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6024b936d74473fbd4d7724ae8a15f6 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/50c536f1aa34456b8e548f53daeddb99 : SUCCESS in 1h 02m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b16e1f895d814f02a2b21c6ce1feb175 : SUCCESS in 1h 05m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/88b360a57d084d84a5d2873cb01fe891 : SUCCESS in 42m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/174c5d8209b840e8b7d4888ab7024eaa : SUCCESS in 41m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/358559bd587542a19bfa702f492d40d9 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b571882f79a47e3b7415e00b3039856 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5271f06ecf0140abaabe5c7474beca1c : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/387e7c63040947d98ab787b2368563ad : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab258f304eb54b538320735757762dfb : SUCCESS in 30m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5c194fed5cc94330bce423b77dcbcd5f : SUCCESS in 21m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ca7976e6fdf248cab2a0d641c000c1d6 : SUCCESS in 29m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cca45a7be8a343cc907766b6ef161dec : SUCCESS in 18m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4f257480c65c4b13b122fb5562cba3b0 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4893ca6b770f4ede9b7ae1b2668240c4 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e8c9d176e21a43debc5c96ef003035cb : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9a782735c944c8681cb394d704b6aa3 : SUCCESS in 7m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ac273c3152847f9a24e480f9a7d3c79 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8722180df613429bb556ec4bcf228416 : SUCCESS in 5m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4bbc4a2f09e46ddbf8fb9b6dea0d92e : SUCCESS in 57m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ed9724090e847febbc482d00bf35158 : SUCCESS in 25m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4d1e0213c85474cbc70662783785e10 : SUCCESS in 33m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c22fd00b180049309d21a76de61a267e : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51867a95d4e4da9995dc8a18b12dcae : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17ae4c93253a493fa8cccf73bad9b48a : SUCCESS in 26m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfbf53bcbe124acd81fa76aaf03e627c : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1d93df83e9cdffa597252ec9776bc423ebef61b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 20:57:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3113f9f69e0f4cde81571f0e0953bada\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6b9ff5c1af34091b1dd279ade9c26ba : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bb18ea80fbb0b0599941a7d14b59452e318d123e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 21:57:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9e0544baaed999104a3558d828c321cef4b2eb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 22:19:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e1dffb7fd7a24d660e72319df5b980033e5dfd93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:50:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f76154f900144000b81c386250470a21\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65ec5ba49c7443c186e1ac369c223c6a : SUCCESS in 29m 52s","accounts_in_message":[],"_revision_number":3},{"id":"8d22ca7521135887aef26a5abb27a35c9b94166b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:29:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6880126cbbc94516836ba3cda7a2dd28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b0b657da754a13ad4e9771fb277b8f : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/eaded4bc207f4f64949f1370e310bbb3 : SUCCESS in 1h 02m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa7968daf3ef48dfa08e0e3c24dbd053 : SUCCESS in 54m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75ced0fc5d6649ad8d230cc927608c62 : SUCCESS in 33m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b13b80d5c1148b2a344df9254b573e3 : SUCCESS in 1h 05m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3b2682506c04dff9366815cd5e42c01 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3aa264e8ceb445c78351b7ee4afca585 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b8b10321d3045ec9f7ff2df63570ba2 : SUCCESS in 16m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ddca546c3a747a5863a50492cc7f621 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e0dc6170c8ab43cf8895a73bbeb008c3 : SUCCESS in 10m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ca316ecb36804a049a14ac7de6c28ce7 : SUCCESS in 19m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c81c741fc1064dcfb970991173c90bed : SUCCESS in 29m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fb8c6e650f8f41bb924f96f7705aab2c : SUCCESS in 16m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c55956c64f59413c8851f25f6313d899 : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3cc257bd9bbe4f81be049c8b245fc13d : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7dfbfc299c004740bd1fd735b8b3ece4 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7de066f4f8d47d6adbc7ca6bae954b8 : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4087888c288c4af99fc63063c9db7676 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/966e0b85e9b04947ab25ecf6c7fa532d : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d44a572943144180bd7924c9c4c93298 : SUCCESS in 48m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9e16554c3f84f83a691cf80b813266c : SUCCESS in 59m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6127a098ba15491fb653e05f2b27ae62 : SUCCESS in 51m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab7a58eaf5d64670869878cbb26e192c : SUCCESS in 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a51893bb8d4a4e894620b8cbaada02 : SUCCESS in 2m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4837a768032143668f579b691e4dbc7d : SUCCESS in 20m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f9e3885b87d4ea3a0e0a6b6466b6c7f : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":3},{"id":"13a66701e90ccca22d34b818ff9cc39faf247ce9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-25 18:33:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"160008002b4b8819508a34662f70551f35bcfed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-25 20:59:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"030c9244787950890190bd6ed2749eac4b2df212","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-25 21:01:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7f5aa79365bcfbd9809eb179e400f66fe590ac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-25 21:15:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"51a14501bda38c623bbecee8ee08d722a46e360a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 22:27:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ec81a6f9ee4efcb0f1f8683a4e52da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ee9b7fdcd64fe39d509112c2b6bd6d : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/dab435fd2aa54971ac480e74c56ab016 : SUCCESS in 51m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0db756476aa45b2ac2a820fda5d0abf : SUCCESS in 1h 05m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd38380000cc45298fc27182589078cc : SUCCESS in 51m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1384bd0d9c343239902167163a0cd93 : SUCCESS in 1h 06m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e03b23e39da491ab0a1a610985f1bf1 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69d68667a3ba46e487fec89b57937046 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73ca41892ffb4fefb1472c771c1c0b1d : SUCCESS in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e6d9719653146a99695fa2954f1bb3e : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08405b5a551b40d682aaef521d45d715 : SUCCESS in 28m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8bbe6446873b40a09b870cf3cad359e0 : SUCCESS in 22m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c18a1041ac0247fbab557d980866c995 : SUCCESS in 26m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/06e2179f138d4e9e8b7f26625c66cacf : SUCCESS in 21m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28b3042996324d4fba53ec9ecf642b37 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1dcd32c7adb944bf99c7da3fc8284af3 : SUCCESS in 16m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cb93764165664823809c80b7e8a7d81e : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45bfa1f4cfba48f781a4933e41fb3231 : SUCCESS in 3m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9274164241c14d5b8b169ceb26048e5a : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8df61734777d428f97c11e4e7e5405a1 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a10b8e56218e463c97932f6518899945 : SUCCESS in 34m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ffaea0824d45426d9bafe8bf653ddbcd : SUCCESS in 37m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d9514a422864b779e17aefa704ae848 : SUCCESS in 52m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccc562cf0869419b851ea96edd66324b : SUCCESS in 22m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca34504336c64f30b917ed589b254d09 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1dc71aaa838d494c9566a61b4a5f7769 : SUCCESS in 16m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc85dad53dd34857bf3f67482d3bc2e7 : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":5},{"id":"878a7f6a72a913abd6fc20b7beda20014e56a8a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 05:37:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dae5da1066d24415ab6cd55c0aa20113\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8208dc720f664e39b4e8e10cad6a0548 : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":5},{"id":"c4dc9499e7300936fc43478dc249de2e7a099971","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-26 16:45:34.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0e0e2b1ecf095d8414739518bf561be0ece16d95","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-30 23:35:11.000000000","message":"Patch Set 5: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":5},{"id":"117b05eae98b5bd885d4c327f931662fce2c1b4e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-09 18:12:54.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"55ae8e26db203b79a5ae64f9e76bb6e9baca55ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-09 18:27:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c93b446eac1aba61bce2dcbd3cb36397a9c804bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-09 19:01:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9be8e0b1eb4c7e5f566a309e14653af3eb2dc908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-09 19:06:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"63eca94b55bca5aafaa9e3a9a20164dd5cdcd0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:46:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3177956060e478393e775d9c4ec454e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33384592dc8d452ba157bc8a0d7174de : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":8},{"id":"f17f61c8e199cd91fdb51731e95fe3e3bf6763a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 20:17:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb6200c3cb242bab6051828f660444d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1161e3077c9b4114a9f1cc6a04679101 : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a3aad69da08544a090035643701e19f1 : SUCCESS in 1h 09m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f10ffb7a413949dab8b21361f6c8234b : SUCCESS in 37m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b7b96b7dd6f48a69fa8c2dd12b6ee4c : SUCCESS in 57m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a75a9c3ac6c6458280b7b4da7e939e23 : SUCCESS in 1h 02m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99ebe1265e9249a18a5402c850f51281 : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e27f8661915c43e7a68dcf6ba7b421dd : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df81677f781c48dba58f2a7adcf2f39c : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/185f345318424f5bb0c8d622ddf7b4be : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88337cc214b24dbdbb2095b6d0df2bb7 : SUCCESS in 12m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6ceaac7dbe794a568f0a102ab31c29e6 : SUCCESS in 10m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fbdd5b15aa2744cfa51bc5544b440a17 : SUCCESS in 32m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d61593cadbc14e338799c4f2e73ec38f : SUCCESS in 21m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e79bcac1e6f40cca0c82baec8b0657f : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/29c3c9a7494d40b9b57468cda1d7cd3c : SUCCESS in 9m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/95402895dc964cd09b36decea5d5921a : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b4cb516660f432f999bab612037d672 : SUCCESS in 7m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11ccfcd96ea04f5285299cc0e030cdb7 : SUCCESS in 24m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/03646d487fc643a1a684937c2c552a7b : SUCCESS in 9m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1edd42ca9b5e44c692bb300f0aa26a64 : SUCCESS in 39m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4246269b0b340e392f2fc9a1aeb56a9 : SUCCESS in 28m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a79759c3cceb44928ec00fecd608d4a0 : SUCCESS in 28m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ee6a671b6ab45d8946e241a75116da9 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6e2cdeef0e4a43bef4770fd66b29cc : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebf48d71afc54446b657f79b75f549e7 : SUCCESS in 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a0d30f6e4184248919bc9aa470891cf : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":8},{"id":"fbfdfef93a4c7ca362c03688073ff43f453a8c74","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-09 22:58:08.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5ad4c0daa3e4902c924f81a1f395a21640e425d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-10 00:03:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"d6cfe853960fd7ecb6255406a93b1355892428ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-10 00:28:06.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4384c0c6ab2e83a644112f29ea978eba13e572af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 00:44:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfcef183ec37479185e016f3ab7e7f93\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8361f57f6bbb4091bc791aa5fea87f35 : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":9},{"id":"b4986fbd7e413ae6ed85d5d0dc1be95a014b1166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 01:14:00.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78886751ca2e43e5b0c4051f060a6856\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ddec20f6f4c40aea1fcfd7217f8566d : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/606355296015469bacf44494a46a840f : SUCCESS in 51m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/edd7c8eddd514c7f836055e3230f4b5f : SUCCESS in 51m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/deffad6b11e04a788e2adb585342bc1e : SUCCESS in 37m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/056f9e4e9400418784bb5933313bfc03 : SUCCESS in 1h 04m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2a4af959f824c8a97aea612ee826c92 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bf25b96c02c4582bfaeb360fbae1b04 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6b4298abd2749fbbf112a1fe8171e14 : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/406fca44d90643bb986452fadf237446 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e1d2f07049a94f22ad45139d8bde2086 : SUCCESS in 24m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/40a1d59450b1460e88a0bf26b745abb9 : SUCCESS in 21m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/994bb7e039aa471e806816de505aa1f1 : SUCCESS in 28m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/233308e12bcb4abea506e994c7c16865 : SUCCESS in 21m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8516bb80ecd4b7c988883d15c014823 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f8d57bed671640c882e7e7dee5fd457b : SUCCESS in 11m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c8bb24fd722247b496877747dc4369ff : SUCCESS in 13m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/271bc9e4c3834211b6fcb4a1471304f0 : SUCCESS in 8m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3dfb4543e5d74bf9b22eb952a0772662 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fdeca6779a04f3e929b6b55483d841d : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ec64b6499f0460ea3e434ae10a72c3e : SUCCESS in 58m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92a1835df47f41818e58f13325f5c47e : SUCCESS in 53m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3217aa6af4044d79b0edd96612a04997 : SUCCESS in 57m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35c334d2979f4e0c8a3c3030ff2c3f8f : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1862ed0587e4f0688acebdb1bc18852 : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71fd583cf80748bda128b7f01fc31cfb : SUCCESS in 18m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3ca32ed251e4beaa92a388f1f5b83aa : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":9},{"id":"51d1e247bc40bdf5c6822176a8555e92cf154f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-10 01:23:20.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6fcbb39fb6d7ccf1b4097ca9db82098dabc85526","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 01:54:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c9e611d1d74ce4b706b1c6dfd5a7cc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc3f54e1f8554c67a702ee6a7f6d8133 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":10},{"id":"3e15e0a5a5dccdfaa84374e0159d7a30c9591f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 02:34:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d5f6073e62b4cbfa594d7033e006d24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ffc01f5cd94aee92d1a5b7fe85783d : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/6da887b1a2a74ca1a443e0fbbaf825e3 : SUCCESS in 44m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f424cc3aaa484304aa1ebb416ae5fe53 : SUCCESS in 46m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92fdc759dea84b15b64d5a74d28f8cdd : SUCCESS in 53m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2609d34ef124ae190e99e9a8ced1ecb : SUCCESS in 1h 02m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ef0007a24144b5488c4b606a93a12a8 : SUCCESS in 10m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ad1aa3351454caba19dd8765f1b22b9 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59954475ec1142cdbf04b8f049421544 : SUCCESS in 11m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3452640141342d291ed115a563ebfea : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29d3b85f62134695b4eb36a64331d8ca : SUCCESS in 30m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0cd6486c82ff45c094dd69fd7544f214 : SUCCESS in 17m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bce44b87354641ce8fea1bcfe2389733 : SUCCESS in 23m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/02b6da25055144c48528c486d943b942 : SUCCESS in 19m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c994cdca676e46988cc7eb7e34947be8 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d8093b4fb6464caab8a442e8ecc08f99 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c8a3c12eca44e84b98573286c3741a7 : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ad6d42d08f043c3ac4254ee77e6d70e : SUCCESS in 7m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e4eb60dc79841d7aded54a17cb7e254 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/46e224964e754468b0bc1a4953a06dd1 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d76ac0d65a14738be8db9b005cae1c0 : FAILURE in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f469d9db95d149e88d8ade4611d234a8 : SUCCESS in 53m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f3ca96fe3bc4796a472acb13f6e8d0b : SUCCESS in 24m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6e3dd2348ab41a2969359217bce9c91 : SUCCESS in 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3a05d43c284d539f0d524dac2f5391 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e6a78ee872740dbb4c7aad44a34e996 : FAILURE in 41m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5560548229a745738ea2a3c25e7c622a : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":10},{"id":"004a52933d9efa9ec616ec8fec13eac498a730d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-10 16:20:16.000000000","message":"Patch Set 10: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d489d2691e4b622ad52de272ec8c57946b778c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-10 22:01:30.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9c9a293d69b39a04e5686789a6e7a8dd3c428187","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-10 22:07:02.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ed3c4d096c07d92668e89a890299e0c20defdafc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:56:42.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f81045f13f84f87913ffef07b8b735b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8791e0b21ff345ae8350a2274c1afe03 : FAILURE in 28m 22s","accounts_in_message":[],"_revision_number":11},{"id":"eb1aa33e9e38f1c93d2375a16adc3a5655e11f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 23:18:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f645f452e3347e59c43370e8fef5270\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d408c98f9e246d29e14501cadf6f0be : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/78f9f6ba10c043c0a37dad10a239fdcc : SUCCESS in 53m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99c0010e04cf40c0834b3ab8307b4a99 : SUCCESS in 1h 03m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/10d7190628b343c2a50791d9fd4a93ec : SUCCESS in 1h 00m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/378376203f27414997a0e75bc3dcb7fd : SUCCESS in 1h 09m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c560c302ecc431c93556d7f1ae39848 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6222b4c96e546f3b3c9a59fbd882f3c : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b40c935d5cac427287cf6af43b4d0c90 : SUCCESS in 18m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5995cd4b670347139d97a3a9c6f3583d : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b041386296341b1849bde547adc6c55 : FAILURE in 25m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b0bbd9fb289b4185b4519522eb0f8ea1 : FAILURE in 11m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b78139dbf79b41c8bf1467f661b1d873 : FAILURE in 12m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d5d78596811a4343a737dcc3fb3e14b7 : FAILURE in 10m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/18205992d594419d948b4be9994088a0 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f3eb2f44558642109005caffc5b6fd44 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/19e1ab7687274c099f7ec9cf064ab080 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a953585a1084b8aa2b8d2b5ca5c0d7a : SUCCESS in 3m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49853d603b17471a900ec1cdd9cbb6f6 : SUCCESS in 25m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ce64cb45aab41a9a6c44b52c0c1e587 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/969ac84bf99d4dadafcbfa27886b9487 : SUCCESS in 35m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd62c6c5e01745f081620b9c7d360fae : SUCCESS in 25m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6ea34d4e8654b1a889351347438ca6c : SUCCESS in 52m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbcfc12088b443ec80cf0a01cd2997d7 : FAILURE in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c64980e1d44a8ba8edfe65d7ee2a2b : SUCCESS in 2m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39c41dea43184bba884ecec15c7d8dc3 : SUCCESS in 9m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0337577e72441d499a596c46c1c39e0 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":11},{"id":"1f9bbc0fb60ed78a4e54e4d29c42bbf8f6689216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-12 15:46:09.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"69c2f8c666b5f587ec76aed05b2a9d2b13109480","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 16:21:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162f7a0df349469ea7c233ae00edad31\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80a962700975427cb22fc1518e69ff03 : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":12},{"id":"e66a70a2dacea03af14ce7409f30d97dab303863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 16:59:04.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/318b8cd70c9546909893c4c489511f51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2112167efb304fbca2f5a86cfbc0cc93 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/8be45b46ae8446b585e230ba35b3bd65 : SUCCESS in 55m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f2746245a2245d8900be1233ee4ab4c : FAILURE in 52m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/512b2f6b81294f36a948d8710064a74b : SUCCESS in 1h 05m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b31d9811101941b694204f10ce80b260 : SUCCESS in 50m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b06d04f55bed40608d5add224b63f12d : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f83d4bdb495d4f17b5fc1045450c9834 : SUCCESS in 22m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2bfade2436d4d02b8914ed44a33725a : SUCCESS in 11m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f63e8489232140a1b873748ebd12991d : SUCCESS in 4m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/970ce094af8a4fbfbd746383fc271408 : SUCCESS in 11m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/db0f2df30ae24e208b6e14bd32333bcd : SUCCESS in 22m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1015f958179e4436a529b43b201bf42f : SUCCESS in 11m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f44c341ea04e4099933e0e65fc4ae77e : SUCCESS in 9m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bc53919aba044bcc86a9fa07e704df5a : SUCCESS in 8m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b2e3753fff034165b646960cd118420f : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/830cf795cb984a23954c60df72167a65 : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/109d15150f5741dcb1c1f80e08ef1c13 : SUCCESS in 6m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70c6539252954bc98a85b106de4e8219 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/42e3528946d84ea4aa943da9c1486ebb : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0231f0ae7c848faa7c7abe1c6d35a91 : SUCCESS in 1h 06m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23818906754347ceaebfa653c383b6df : SUCCESS in 56m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2f7bc75a3ec48778b31b1ec27438824 : SUCCESS in 54m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8d3311b0cd343319c6846970843c711 : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99d3637ba194896b555ad98d0fe9e38 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53c48d605a3446679af237407458176b : SUCCESS in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b261efb8e2d849d88518a449ac50ae36 : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":12},{"id":"b8925a0faf590ca2b6c4c275ee30f932b6e5c07a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-13 03:26:21.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b58c0a9ff06139f8ee202225de2fd655b4dc3464","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-13 03:28:01.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4766d4382e12ea1b54fe9f90c44aca83a245ed75","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-13 15:28:07.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"70cea91d2df5453a810826f76227af8af7ea9a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 16:47:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d644f7cf12534ccc98fe286c50456f2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6da0d58e7745aa9d4784af93af3d1d : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/d0b060cafefe40949837dec0e6fbe91b : SUCCESS in 1h 02m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2fc46ea608946b3aaa7ae62d4a4def5 : SUCCESS in 1h 08m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5cabf9983330410cb1934dd3ac4fbaf5 : SUCCESS in 59m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b821e33eb2644bb39ee8aa880229d366 : SUCCESS in 1h 07m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/860b3e5df4aa40e39b09e3e75e285af5 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa22db929bf44c1fb2210e2e803e5b21 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6389fe5bd474f1384670c42da2ffa7b : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f00e24d1eda44b56bdb802af631d4b88 : SUCCESS in 8m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/447f8735c52c490f8416d08dba92ec33 : SUCCESS in 25m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/03e50fb47fbb4dbc8b1391721aa6ef75 : SUCCESS in 22m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/db78cc529e224850b3a6aad1b797053a : SUCCESS in 28m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0166a793538143bf9792852171fbac07 : SUCCESS in 22m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/851b0a738f7240c4af44e28540422f5e : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1a2e4ee957a640c3aee546f1ff42e28e : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/352b4b1acdfd46a0a3101add351af918 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/190eedd5496c41328d8b32dbea11f592 : SUCCESS in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1cf732ddcc6b48e6b6a75ac60135a9f6 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d13e24f03b76449c9508283845fe551b : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6aeceb946d74efd83ec241ca3bffe29 : SUCCESS in 1h 00m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a0aa6e406994ddb892e1b27bd23f5e8 : SUCCESS in 57m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2dcac61383c4ef7a735f09de967fdd5 : SUCCESS in 52m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93cbaa33822b4812a0404e3a0c818b65 : SUCCESS in 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e60a46b181473b8953e95336fee1f5 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/475b145278814e1a96fb21f119868445 : SUCCESS in 11m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1086b7ad794d4a3b8a292cfe3d687d7d : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":12},{"id":"f5698025d98be968bc51782cab0e023f963380f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 17:28:23.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f23b85ed1343288d0102402705bafc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7287d3aece94b0abea9af008725e661 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":12},{"id":"5fcadaf98e120baa4f22f7811aede418718972e0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-13 19:40:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d142b8231d0ed32a006aeb5ca0e64dcce7315335","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-13 19:44:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5ddd5d5828e8638f0b96b599c8ed4799a6d90285","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-14 07:25:30.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)\n\nNice improvement and I like the db close catcher.\n\nPLUS this unblock modern python so win win to me!","accounts_in_message":[],"_revision_number":12},{"id":"86571bbc2de5e9b7852d70a13ee6800afe363c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-21 15:52:02.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"0c3208158cb120742385c1dc70a774005687a237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 17:03:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25fa9a3e14764c52933095d7a7da53f4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36a829f357614086ab701c34ef67c4cc : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":13},{"id":"96b84e3b0cf820dc8bbb6fd8de112fcbe8a0cb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 17:06:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2187b904cffb4948a4fe40970b0b3b01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ffebb7445ba4cbe85f5d2e6687ec8c5 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/00935b042cb74d479654a4f2b2bf8bf3 : SUCCESS in 1h 01m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9cffff27da7346969d89f4861cb9f869 : SUCCESS in 1h 07m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/368a719d1f164743b414876bfcd0a2db : SUCCESS in 1h 01m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38f004ecb5ac46eb8eb29693033d34b0 : SUCCESS in 1h 04m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f8672dc725a46c39f936475357eb8fb : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb6feada9bda4ae6b82a59f1c5b7c11a : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/894840fef4e2453996fa11e48ff7a70d : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56179a88f9044d2f8c9067e264acfe8d : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a63c13765a634983a85dcb62c2582db6 : SUCCESS in 28m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/10ccf30fbcd345eaac4cbe81b026efe5 : SUCCESS in 23m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/44f8dff9757244c68b51f3626385167e : SUCCESS in 20m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/47ed5245002b41bfa401dd8116c90b12 : SUCCESS in 16m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d35ea4b93a8c47a8a3f385e9a8c6e081 : SUCCESS in 13m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f666f5a6ec2f477d9d33cc9dcf417f31 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a9cfab646cad463c9a0d8070181930ec : SUCCESS in 10m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/171e657d048144d9ab787ca173ff1064 : SUCCESS in 6m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93de8fd9a2ce4020a3c7aab3bb2b2e8e : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/63768c55c1ca435fba963b2ee2170c6a : SUCCESS in 11m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab85370a55384ec98cb8175b86e61fd5 : SUCCESS in 58m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9cf3123545b46fdb735ba3abbf588e0 : SUCCESS in 56m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08b929cb6ec941308b555f3ec6e22d5a : SUCCESS in 42m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17c7fb6051344404af62f91f9f0c1f57 : SUCCESS in 29m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/787c838194d543cfb76fdd9876d57cbf : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa646ba6e4dd44308eebbded816b4a14 : SUCCESS in 21m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc03125c0b1f42fb975e7ebbe9a80471 : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":13},{"id":"cb3319e1815bf14704eb0b91421712bc678be5a2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-04-23 04:28:37.000000000","message":"Patch Set 13: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"37bb98ea31db308792c297320dfe03a3cd2938d9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 15:13:11.000000000","message":"Patch Set 13:\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"66e4be236b80177d17c779242523496189ecb696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 15:38:01.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"07d69bce74dc6f359af0bfcd0d7937cd36a1648a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 15:38:16.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"0ede7de21bb8e39e7b8a4e0b685c8ec0b299b1bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 16:08:54.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"9716169b3eb83572d7c6828438ded6108b707d0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:43:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae337e78ca3b448f8dcc6f31dcb1de69\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79f39646a7d244e8b31f9d6a724b2933 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":15},{"id":"d63b1621735412a473a559c48d7cc47f71f06771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 17:20:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573987ed351e4f29a4d56334420a6c3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16314f81445b49b2baf36fa8c2b829e3 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/325cbd9ec03948feb2ce97e0eab08957 : SUCCESS in 47m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8c2b0b7c6a54fd3b700a12df7e75c0d : SUCCESS in 1h 03m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7549683e7e7e4f26bd2ba2be6c39bf1a : SUCCESS in 47m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10d6f2983889497bad0cbfc694ba700b : SUCCESS in 34m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50b203e85a0a414c8bff7e36ef54dc7b : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8408db83931c42b89d1e391ca821736f : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2f84a205f474dfa8655b06062748396 : SUCCESS in 22m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b32c3b05acf43f886bd54cf0f75991d : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/25e6d16691ee41e0aea938262b21431a : SUCCESS in 27m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e9cd53d3f23c4d2fa80bbcce21c0c9f1 : SUCCESS in 22m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e1851c6ce4fa4f0cbf25841cfa4f7fb4 : SUCCESS in 27m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1309aaab29e14a80acaffdc6d5fbcf18 : SUCCESS in 19m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a6f102ac7174dadb57084586b99c75f : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bce2d4deb00746a082d4b8a3d5a23ee4 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/42ea17cd808c4b18a64cf3874aea5ab4 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4701c14e1625499a81e8a95e2374fc5c : SUCCESS in 6m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c0169a380bb4db9bb78108647040698 : FAILURE in 6m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a0adf0b324b945a498786ca9ba017f87 : SUCCESS in 10m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae585f6ebbc840e4bdcc6d5645825f26 : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e985811a1b748ff83bafccaea09eeda : SUCCESS in 23m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/67f19239246049eb8033dd225358b138 : SUCCESS in 57m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10d024004cc5423bb7d7eedd5a1332fb : SUCCESS in 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4ec5d8dc89413396e5eff5a68a0b74 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59b10f255ba74952bfe0c5c876d5771f : SUCCESS in 16m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6a6b6d971c5463d82ed78600a63966e : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":15},{"id":"991bef9dd922013dde4191a5391ab390fc0e4150","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-07 07:13:52.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6b9bc9b4a4562d3f5e147472ba91161da930e7bd","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-05-07 13:06:36.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"28d87b03d4326db3fe6745648d49ac67af8af6f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-07 20:58:25.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"e35ff8d1ef686dc66cbcff55f1e92b0ed7a4ff87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 21:44:49.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3622658689649d986b90404c7034c1f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1cc4c09163414f7dbdb7356dcc51de2d : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":16},{"id":"14f57866b1030cdecab0121bf3518372b1ad1f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 22:12:14.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f20616323fc4d58b4e943fd962b4481\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebf3d545eb54f818fc2f6c38a28ae5f : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/d2969256d1244319a166976ca8979b9d : SUCCESS in 1h 06m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a79b6f4c7dbf41e7a4ae6ff9353610e2 : SUCCESS in 1h 03m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9e652e0ea6d40cbbdd756c97e682654 : SUCCESS in 57m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07612c489ee544f3b4fe35a49d5e5ad1 : SUCCESS in 1h 08m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdfff1512480406d97ed3c2040320c13 : SUCCESS in 12m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4928f12ce3f74be7934c18ec9b74f4d9 : SUCCESS in 22m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58029ddda810430cbddd7d3b029e0771 : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96b6cf23d36044299fe13ac82f36e9ad : SUCCESS in 7m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4d4594940d82485c9fcd28b733cd8dd1 : SUCCESS in 13m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d71d5774cc81419db8134f4368c9337b : SUCCESS in 22m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f5ba86ed48f40109f14518cd532c9dc : SUCCESS in 32m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bb7a1f06cb2b49e1b8e0b1cdd6e81b0a : SUCCESS in 18m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/98958b1b71cc4364976045c47faaccb3 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/99f76adb034a43cca0e2e197313e3b87 : SUCCESS in 8m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30c83aa508ee481eb18652ed551abbeb : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4495620062544d1e9d7ca1b61edd510c : SUCCESS in 4m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ccc773b15084f278915ea5d7f6fcd1f : SUCCESS in 12m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4ca249da40d94b5b823641cd9a30220a : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/27ce75fdbdaa43aeaf54f363b3bc2d21 : SUCCESS in 1h 06m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/526aaa91beed4ac1b6b30dfdccc3dcdb : SUCCESS in 30m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5ed5626c39449c28345d2786be9f64d : SUCCESS in 26m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ed55f9f2d44494a0b189286e4f6d08 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ea95daac514ab8b37956888d9a3071 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2eea4a4776e421496e3e24e5c8f6eb0 : SUCCESS in 19m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad86443e0f194de2a72e3fb3cd2125f6 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":16},{"id":"ee8102cf7b5f5449e0faf965043efbda62a384b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-27 17:57:32.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"ae1c0aa0d64e7744946f8d03dc45520f9cce5e3f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-05-29 04:55:44.000000000","message":"Patch Set 16: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":16},{"id":"9e0bfacc8baef833dcc76640d4e9ab4617c8c3fe","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-29 12:21:23.000000000","message":"Patch Set 16: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1e41e18ba5e81c5dd5d2975b639a6553095f4d85","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-29 23:25:53.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"624c7a05f92e59da0997e4cf59c02a0ecf25d235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-05-29 23:26:34.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"28e6223e6997f6ae483479bef84c3055e9d6a7e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 00:39:03.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad39b895c3c4f2db01329b57dc62259\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdab7f2cf9fb435e9b2225d3866e58d8 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5504747ff8fb4ccb88059aacb2ed837c : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/7dcd1bfe66b94ff380e9fd745fc26945 : SUCCESS in 1h 02m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d1f67961312548358c9dd6f999175be6 : SUCCESS in 1h 02m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2bf819958dc24a51b0ccaf59f8bfd21e : SUCCESS in 58m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63a66edb49ff4a1b96d8c9204c30b4fd : SUCCESS in 1h 10m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/269d33b7f38045f2b5dd6c2583cc92e5 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b040b64eef04847bd04590678e451da : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2733938dfbf40cc967019df1c2caae9 : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/339f327db69d46ef958663f844daa61e : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a73d899869b444cd8474ac20ae2ea985 : SUCCESS in 24m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e1d394c6cab4dbe9b8e7b7c3f954a55 : SUCCESS in 22m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b95a3969c5be447e94018cb69823cec6 : SUCCESS in 31m 18s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8a6bc3006f2043d3bf1995fa06221bd1 : SUCCESS in 19m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/371a2dcce35b4022bacb58e36fd26de9 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f21b31f09c81465e932e4e1af1aaa913 : SUCCESS in 15m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3f9867eec7a4e2eb82192c080b7881e : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d26f2d459054c509a94e0ba2aadfd38 : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea7fd5f89ef041dc8d172fec31d7ef50 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d399582dd3714c5aaebfc63e834de937 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17d87ff2e25244f6b58f5d0c01865e86 : SUCCESS in 58m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c54d7f0ede54b27b4c8665017c7dabd : SUCCESS in 58m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3191da9a43e54842ae34b71086e3dea4 : SUCCESS in 54m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83cb333bbfd545da9cbddbfbf5edf038 : SUCCESS in 30m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1242eff65ed249bc89d2e417bd2810ed : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4b219f984664ff383afd87096c9306b : SUCCESS in 17m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cf39d9c13d24d068e73e599c8d872ee : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":17},{"id":"545be5c767c79882a4e6e0d9c21200f94db3de16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 00:48:34.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3cd35113396446b911289402e36b542\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22b957d1a9474b13a6e8b071c91cf0ea : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":17},{"id":"61effcb7ccf066ae90da55cc7f31a848fc713129","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-01 15:12:34.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"0ce18ee39a12d1cb7d047607270ff15c7da42a12","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-01 20:07:13.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b6b3e8097e746693b416ac77447204f52fa68c23","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 10:38:02.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"4512dd1247b686d7ab159962e5edf2e2390cd5cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 10:38:47.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"bf7f8e4b5ab66adf122ce8c07ae70a2cef296ad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 12:22:58.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3328411db10e4866bf4b1ca67470d172\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4385818e5cde49c2a6c18c94d113b1dd : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3a594616c354abb806fffb7d06b0c59 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/fca7922ab83342cb98a5dbf94111c8ba : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2346787479c4b168b05fa2babff8b35 : SUCCESS in 1h 05m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6edada3c6cdb4411a6c2d255ea164177 : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa61e207f82046dc8b6f1ee38c19073b : SUCCESS in 1h 15m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5aa33f04e7164800adac770a1cfbfa5c : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29d23176fcef4b3aaf0aa79ef0d14018 : SUCCESS in 15m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0834603f299f41539bc3b3602793c2ef : SUCCESS in 22m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/129406030ab94e01af694530fcf7f1c5 : SUCCESS in 32m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e11f5676f86941b0943fb2a574b69fcc : SUCCESS in 21m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/84be133d1961448684002c5922944bf2 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5026242992fd4608a42c9e530049d4cf : SUCCESS in 16m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1f730b2adfa64bfc850b94907570c176 : SUCCESS in 19m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ab369c57cfc49899afe4c707683933a : SUCCESS in 5m 07s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5a422593eef04a0a967d26fc0e0d381c : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/457343d6464b4361b8508e805cc37204 : FAILURE in 1h 09m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fdd39e8373b54e54a92db8de3bfb56e9 : SUCCESS in 57m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f48e36244d9045a58c69a17cc2598072 : SUCCESS in 51m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/475e70cf1c79477aa81158323492bd14 : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e2dd1b0dd44e60b48a5dc76104b3a1 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57651a010ca246bdbf42c4abdf1783eb : SUCCESS in 22m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fefb3f843724ce995d93d194baca542 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":17},{"id":"2dfb255d4e5697799c5d18c37ca72a1096061598","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-02 22:46:12.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a1cde2b7efc2b694d0e12db1455418e39d259dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-19 21:02:46.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"42df82ce01ababa2194e09cf2fa396e8f9dbf4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 22:12:24.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/266e174807c742ecb417b6be0f5752f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c898379185b469691dc3630a951ec65 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31267bede52741139854a1653f2831c2 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4e348eb0d6204ac58673e30590b2d05c : SUCCESS in 36m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e12ff2980b984889a82964dbb2fe8acf : SUCCESS in 1h 02m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/247dceebce4b461db569dd0ef5ed7e22 : SUCCESS in 58m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84f67058348f43ef8d11fe5675eef8dd : SUCCESS in 1h 04m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33d9d51f310e4168aad986faa88963c8 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c7f904ee76a48b0bb2b0a3fc9b1bfd0 : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45277e56c9bf4de6a84ab4613e1e4402 : SUCCESS in 11m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/363c34f06b424a2db129521120fef37e : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ec7072e18604955862e288e90cbf37e : SUCCESS in 23m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/853831dad6b64854b505a8154eeb87a6 : SUCCESS in 26m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dd6c87404a5645f3b3639735934a2659 : SUCCESS in 30m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a976192785cc4c01960b942546b9dfe5 : SUCCESS in 20m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c60bc537d7034d999c5bae19419fa0d4 : SUCCESS in 12m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/625d4ebf833a4bd294dbccebf596635d : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/af5346dcbbe642058d919b824e224062 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef006a2a8b254a9487aa5f751b75a8ee : SUCCESS in 5m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ddaa8d91ae674a67bc72120d923d25de : SUCCESS in 9m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f35bc2e015be4e8dac9bd7cba1dd490d : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2902b80b85c49308497ff61b610b705 : SUCCESS in 49m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7b9c156f2364f1d9dee680afb8c77dc : SUCCESS in 33m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98894440d3ef44a9a6cc7fa69f3c6acd : SUCCESS in 25m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/159517287aac4c0f9497809ee1ed3929 : SUCCESS in 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6ea47fcccc41b085b9763876ab51a3 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f560ae890a7b4fbba8267ca88270d6d5 : SUCCESS in 25m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9be9dddef8774ddca9443c1457bc2d77 : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":18},{"id":"2cd0739fa6cc66abbff9f6aeee976d7f2c89ff50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 22:29:48.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394943ec02384193b694eebe36a0c4a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a07de64e676648ac97eae0540d98ab2a : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":18},{"id":"303d96364f13a6aebb73d4476b2f53c728db760f","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-26 12:21:35.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"06132e031648482cbf270869543a8c3f2bdd58b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 12:57:51.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd3fd4d29e44665bb0a7e60809c35f0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a85818692474caf90efc2febff146e0 : SUCCESS in 34m 21s","accounts_in_message":[],"_revision_number":18},{"id":"610c8a427d5676219141bd13720d84237628e264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 13:33:52.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c62ff089221490c9e40d4fb90f67866\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922ae7d29df24183b9c4076daba789a5 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c673a102712d4c75a16e57ef39a0d548 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/9f3c35d143c04feba08cfcd88584d1ea : SUCCESS in 1h 03m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d9d98c96b3a455c915da21c966d6fc5 : SUCCESS in 1h 07m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96901374b0eb48fdb556705b7b13403b : SUCCESS in 54m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7b2bde3fcb44eef992e5d3c2549f279 : SUCCESS in 1h 06m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7efaed32001940e8af69a8d25c393af7 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cd951fc851c4c21b09f05af52a5f6c1 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2071e01a88445478b47bfe0fc0f724f : SUCCESS in 24m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4eafc084ff6443a38952d74360e57d92 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3dd95f2df664433abb18c9ee94f6c88 : SUCCESS in 27m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3d6079dd55b7447eb3ecfc35b381502c : SUCCESS in 17m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/398d54ba63214baca45ff908b15ea249 : SUCCESS in 24m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4a9fc1968f674ab7a4a4829d9a3c1555 : SUCCESS in 19m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e67d21b4fcde4bdb9398e72cacbe7f61 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd6d51d453f043f08bdae0001e294fb5 : SUCCESS in 9m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/872a73a681fa44a7a1f1815056ef39d4 : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d3005685b634d7dbb43582a51c5a124 : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1cdcb5a6c6c2446393f7b5be5a857334 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7b2274ab8983414db293c09c9a4b0ac0 : SUCCESS in 9m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0db8c7084d3414691c1920791631fef : SUCCESS in 35m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4c3c7c520bb4ddd83a60521045596c5 : SUCCESS in 37m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d27b1b93933c4854a8e6aa90ea78e1cb : SUCCESS in 48m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73899911e4234bf2bdb02787bf2fa0bd : SUCCESS in 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ff9e0de7354f3c8692350997eace26 : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa17bae43375432599aaacc63519a6b3 : SUCCESS in 15m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8a37eb1c96f44ec9a6be4d58360f695 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":18},{"id":"9a1e71164b0fccd2925d19779d772f5395cdb787","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-26 16:23:56.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"0e04878b2071bcab14db585a62991608590ce152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:24:28.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"c414ad3054d01e878b2bfa2d40b3a6cf1a7ad3db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 17:33:25.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4415ba929bc24e2db4b1a95dddd3e62a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819403590b764c898b71a75734a49bb3 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0af90059c002480582ecc1ae69a59d34 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/836f50d29d9e456e9dec6f9b2353a301 : SUCCESS in 50m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3db21b11c2c84b15b4765669f2156caf : SUCCESS in 1h 02m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17fe7ddcc0614aae8f80ad4bb70d11f9 : SUCCESS in 37m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c91d9be6db94f3cae5a0547a1057eac : SUCCESS in 1h 04m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ebc3ad9818f149188d64f1c6b1243ae6 : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/131d5eece298404eb1ce5bf5395603ff : SUCCESS in 13m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/427b08f61c7845f6af07ee43d956c9ba : SUCCESS in 18m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d400dde0fbd84dbfa105cf5b10b13479 : SUCCESS in 31m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/966bfb248e6b4735899c58582f580b0c : SUCCESS in 16m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/325e96bd04ce44e681dc5288ced0259b : SUCCESS in 9m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b6905f64245d4757b77deb0a0886b059 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/433ac4cd1c214670a9cf10d30f2921c3 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86cf22d827c74fd5a2710e0834c5c6ec : SUCCESS in 5m 00s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2553154dfe4b47b2872fa885b4b865a3 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eae087aa020d42fb9fa1dba9e52beeb3 : SUCCESS in 55m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d565eb428024691880f24898815657c : SUCCESS in 50m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6f4e337bad34a87a27bf67e6bcac706 : SUCCESS in 37m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc1bff29d3a742f0b8fe7f2dd99e1b0a : SUCCESS in 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e9beb81fdd4dcfbd3005ed791b2aa4 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1b8c13c478a4c21a6f837accfbd4f77 : SUCCESS in 19m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/886999ab7eb740b8a55b3ea57bcc6e1f : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":18},{"id":"8763a595933ee5a0b8561edc1d1fae396cf3917f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 17:33:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"c6d20929422e8462eca3aac0dc1c1d2b0764d50d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 17:35:09.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038163996c50496e9b486b683312fd91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d7d63060ce4952a656b6398db58bc6 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c0dc677fce4247cd92de4b503b5265e9 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"245ee33880477e61c5efc267adb17f44c11d153b","revisions":{"315ce337c29c539fc80045f6b82944bd0cce0567":{"kind":"REWORK","_number":1,"created":"2026-03-16 16:37:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/1"}}},"commit":{"parents":[{"commit":"178e8e2e70f8153ac27e343caec2e9391ca369a9","subject":"Merge \"tests: Fail loudly when kill_drive/revive_drive fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/178e8e2e70f8153ac27e343caec2e9391ca369a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/315ce337c29c539fc80045f6b82944bd0cce0567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/315ce337c29c539fc80045f6b82944bd0cce0567"}]},"branch":"refs/heads/master"},"84d29362bc67a237e0ae222491d0b71a8fc82b1e":{"kind":"REWORK","_number":2,"created":"2026-03-20 21:57:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/2"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:57:20.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84d29362bc67a237e0ae222491d0b71a8fc82b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84d29362bc67a237e0ae222491d0b71a8fc82b1e"}]},"branch":"refs/heads/master"},"fcc316a95c46412860b9d577657f2b5f1065f594":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-20 22:19:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/3"}}},"commit":{"parents":[{"commit":"eb1c6890cb263d7754d2223fb6f509a317a07303","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb1c6890cb263d7754d2223fb6f509a317a07303"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:16:49.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcc316a95c46412860b9d577657f2b5f1065f594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcc316a95c46412860b9d577657f2b5f1065f594"}]},"branch":"refs/heads/master"},"be9c050e48feefc3cf21744297dd32abcf3411c3":{"kind":"REWORK","_number":4,"created":"2026-03-25 20:59:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/4"}}},"commit":{"parents":[{"commit":"eb1c6890cb263d7754d2223fb6f509a317a07303","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb1c6890cb263d7754d2223fb6f509a317a07303"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-25 20:58:59.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be9c050e48feefc3cf21744297dd32abcf3411c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be9c050e48feefc3cf21744297dd32abcf3411c3"}]},"branch":"refs/heads/master"},"6871e28ed6807838605f26a21583ba0e0466dd9d":{"kind":"REWORK","_number":5,"created":"2026-03-25 21:15:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/5"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-25 21:14:26.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6871e28ed6807838605f26a21583ba0e0466dd9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6871e28ed6807838605f26a21583ba0e0466dd9d"}]},"branch":"refs/heads/master"},"ae71f81a9103bb59640b5a9056405372b100b8b0":{"kind":"REWORK","_number":6,"created":"2026-04-09 18:27:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/6"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-09 18:26:49.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae71f81a9103bb59640b5a9056405372b100b8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae71f81a9103bb59640b5a9056405372b100b8b0"}]},"branch":"refs/heads/master"},"4bf2c97c88bbf79c873912e33fcf313bb704c811":{"kind":"REWORK","_number":7,"created":"2026-04-09 19:01:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/7"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-09 19:00:57.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bf2c97c88bbf79c873912e33fcf313bb704c811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bf2c97c88bbf79c873912e33fcf313bb704c811"}]},"branch":"refs/heads/master"},"3baac2111c7c7a51b6019320ea18d0e236173df1":{"kind":"REWORK","_number":8,"created":"2026-04-09 19:06:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/8"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-09 19:06:40.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3baac2111c7c7a51b6019320ea18d0e236173df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3baac2111c7c7a51b6019320ea18d0e236173df1"}]},"branch":"refs/heads/master"},"81e34ec638e4fb3dcbf61c79e63d5b3c94ab6e07":{"kind":"REWORK","_number":9,"created":"2026-04-10 00:03:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/9"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-10 00:03:44.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81e34ec638e4fb3dcbf61c79e63d5b3c94ab6e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81e34ec638e4fb3dcbf61c79e63d5b3c94ab6e07"}]},"branch":"refs/heads/master"},"85cbe43bf6d900c4a8fa2b49ea7495d6847b1888":{"kind":"REWORK","_number":10,"created":"2026-04-10 01:23:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/10"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-10 01:21:20.000000000","tz":0},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85cbe43bf6d900c4a8fa2b49ea7495d6847b1888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85cbe43bf6d900c4a8fa2b49ea7495d6847b1888"}]},"branch":"refs/heads/master"},"024f114d977d836999cdf6f4e50aff9ab426e4e0":{"kind":"REWORK","_number":11,"created":"2026-04-10 22:01:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/11"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-10 22:00:38.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/024f114d977d836999cdf6f4e50aff9ab426e4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/024f114d977d836999cdf6f4e50aff9ab426e4e0"}]},"branch":"refs/heads/master"},"4d7825132682549e479d80cf7d1896ba46a72a09":{"kind":"REWORK","_number":12,"created":"2026-04-12 15:46:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/12"}}},"commit":{"parents":[{"commit":"020312b8ad30bcb7838198db72ce6166f9833724","subject":"Merge \"Fix rolling upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020312b8ad30bcb7838198db72ce6166f9833724"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-12 15:45:54.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d7825132682549e479d80cf7d1896ba46a72a09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d7825132682549e479d80cf7d1896ba46a72a09"}]},"branch":"refs/heads/master"},"292f09cbb84ac1361fe1d56286daa7cab04a8de0":{"kind":"REWORK","_number":13,"created":"2026-04-21 15:52:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/13"}}},"commit":{"parents":[{"commit":"a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1","subject":"Merge \"timestamps: add NormalTimestamp for ShardRanges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cfcd869642fbc7e6fda68bffabfeb5cd1098b1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-21 15:51:34.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292f09cbb84ac1361fe1d56286daa7cab04a8de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292f09cbb84ac1361fe1d56286daa7cab04a8de0"}]},"branch":"refs/heads/master"},"9b19594055260f9b8caf0672210ba8beb4aa326b":{"kind":"REWORK","_number":14,"created":"2026-04-23 15:38:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/14"}}},"commit":{"parents":[{"commit":"65e47b755b3a2b7399997133ed208fad97cefdca","subject":"container: Split DELETE code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e47b755b3a2b7399997133ed208fad97cefdca"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 15:36:29.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b19594055260f9b8caf0672210ba8beb4aa326b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b19594055260f9b8caf0672210ba8beb4aa326b"}]},"branch":"refs/heads/master"},"ab106e5f65561748c8bd8c0e4a1a3771f5f80aa2":{"kind":"REWORK","_number":15,"created":"2026-04-23 16:08:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/15"}}},"commit":{"parents":[{"commit":"65e47b755b3a2b7399997133ed208fad97cefdca","subject":"container: Split DELETE code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e47b755b3a2b7399997133ed208fad97cefdca"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 16:08:32.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab106e5f65561748c8bd8c0e4a1a3771f5f80aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab106e5f65561748c8bd8c0e4a1a3771f5f80aa2"}]},"branch":"refs/heads/master"},"05c11f44beef85a357ed5a2d98097da79a59dcdf":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-05-07 20:58:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/16"}}},"commit":{"parents":[{"commit":"1afe49b6dff786ad79cc9d8b16e584bddbcee812","subject":"Merge \"container-server: add test for invalid s-p-i header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1afe49b6dff786ad79cc9d8b16e584bddbcee812"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-07 20:58:17.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05c11f44beef85a357ed5a2d98097da79a59dcdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05c11f44beef85a357ed5a2d98097da79a59dcdf"}]},"branch":"refs/heads/master"},"5e283c47182ae365676d204569d72d1eb7f42a8d":{"kind":"REWORK","_number":17,"created":"2026-05-29 23:26:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/17"}}},"commit":{"parents":[{"commit":"18506712ca4cf0596398a8347a30f6bed08eaf7f","subject":"Merge \"Harden pickle unmarshalling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18506712ca4cf0596398a8347a30f6bed08eaf7f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-05-29 23:26:26.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e283c47182ae365676d204569d72d1eb7f42a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e283c47182ae365676d204569d72d1eb7f42a8d"}]},"branch":"refs/heads/master"},"245ee33880477e61c5efc267adb17f44c11d153b":{"kind":"REWORK","_number":18,"created":"2026-06-19 21:02:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/980785/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/980785/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/980785/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/980785/18"}}},"commit":{"parents":[{"commit":"bab1688e2c91cb9866c714ee77bb8c45f6885e7b","subject":"obj/diskfile: self-document _drop_cache offset for ranged GET reads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bab1688e2c91cb9866c714ee77bb8c45f6885e7b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-16 16:35:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-06-19 21:02:39.000000000","tz":-420},"subject":"Close DB connections in metadata servers","message":"Close DB connections in metadata servers\n\nWe\u0027ve seen issues with newer versions of Python keeping sqlite file\nhandles open longer than expected, leading to issues unmounting drives.\nClose connections more promptly rather than waiting on GC.\n\nSee also: https://github.com/python/cpython/issues/135117\nand https://github.com/python/cpython/issues/144345\n\nChange-Id: I1e0052717d9b25f97e3cf58508d005f43d306267\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/245ee33880477e61c5efc267adb17f44c11d153b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/245ee33880477e61c5efc267adb17f44c11d153b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
