)]}'
{"id":"openstack%2Fswift~980810","triplet_id":"openstack%2Fswift~master~Ibcc76c59494497648518ca51c5dc6b93adbacb66","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","hashtags":[],"change_id":"Ibcc76c59494497648518ca51c5dc6b93adbacb66","subject":"WIP make Timestamp jitter non-default","status":"ABANDONED","created":"2026-03-16 18:48:01.000000000","updated":"2026-03-17 17:36:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d79ee571a9ffb65df75998bb4aca652b8d9e2b4a","_number":980810,"virtual_id_number":980810,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-17 16:48:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-16 19:32:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-16 19:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1af2783d54f4967a7fac001ff27b8bb5f8a723a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-16 18:48:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c680179b5bad054513985bf1e0e187dc18e71524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 19:32:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc1c962a5987418db2635caf5ee1a51c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0502f055112a48959ed2e7163f6d566e : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3d1d363e395f82cb2423ebb54dcdb1616225aa74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 19:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd631843760e45a290357df88bfa1474\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075a13f300394727a93af8a048483e79 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/fb406379f16f455891f4db78307f3111 : SUCCESS in 48m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9036bf5137d948bbad3cce2cea68b27d : SUCCESS in 31m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/446ffc097034457fab823d550d71a89e : SUCCESS in 36m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b681fb4c7494dcf94ccff6eb9454e10 : SUCCESS in 1h 04m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8083af9197db4c01a95f69352651f951 : SUCCESS in 11m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/911e696bedfc448292a3b68c5934b80e : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/244b02eca02140cfa08aa99a0c1278b7 : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/036df2fb530c4e2f80b6548d19b90525 : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e5e1a6be27b4f52919c1347db3f740a : SUCCESS in 18m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ed0ebf3bd7d4442e8de4c09d82591ece : SUCCESS in 23m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/acdf9b3f93ce4b278dbc7419c416a2f7 : SUCCESS in 32m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/72c0148c0c7a492b819e3b0f07b8cb40 : SUCCESS in 13m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a47f291b40294d35a9bff26129c01278 : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40d132650fe74870b701e40a1f15c2a6 : SUCCESS in 9m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e515e07e8014c0e8d61bd58ff6bf8e7 : SUCCESS in 10m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a539de63ec2b4eca8d11e87a0c9e1b72 : SUCCESS in 8m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c67bd6e43d14f4081545516da633560 : SUCCESS in 10m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1a66e6a749e4b8e8dd322d9da548a67 : SUCCESS in 5m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50ffa83890cd469995537dc941f9ab39 : SUCCESS in 58m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/41007fce5539438c86ae422ddd05debc : SUCCESS in 51m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/705c897cd684430fab91d0c6293d64f1 : SUCCESS in 50m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/519d9ea6ac8a477db4f5f861d6fe0268 : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05369a5e4bed42d09c3665f101fe645f : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a16816dc8da44ef97f99537f8426e7f : SUCCESS in 18m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9cf599393db4ff28b93570ade8ba2c3 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":1},{"id":"be494c95ca8c80a8bf96120a6899e792cb34eac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-17 15:34:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1623686d5c01847e0abf1c84aca12189fdd8f2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:48:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd3609d3e0b4e28836a707610d576a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8efc35481f44b99af7cc6bb5ede4b7 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/d422dbc1dd8d437b9b7c4745b5e4e787 : SUCCESS in 1h 05m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26251199144a4360a292eb4261c44a55 : SUCCESS in 1h 00m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63ef1767e3ff4a2cb092b460d692ff18 : SUCCESS in 1h 04m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4722c445f2cb4a3d9cc23eaa2da56971 : SUCCESS in 1h 04m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8901b105bc324d4f92b53c49b55f6ec3 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c651601302347ffaf454bcb2fb866a6 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fc1c325ac2f449bb3dfd4e1cba72a85 : SUCCESS in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ededb2fa25840438f367ec2296e250a : SUCCESS in 7m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/529b19712f304ca9a8324674eefbcead : SUCCESS in 15m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c019bf49a4f846b8a2d994addc42b9ee : SUCCESS in 21m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b7ee2ee3958d4541be9fafb231cb6d70 : SUCCESS in 37m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b7451b236f21469c92e92a45df324a59 : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7036cbc32dc04a769e75f34bbf866cf4 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6876f0160c924b0fa472d99ee84c03ae : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a59bcf6377ed4c96b9d327484fdd0556 : SUCCESS in 13m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c677300d49cb41e297304d75ab75f82c : SUCCESS in 5m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04fed4e18d90452bab5ba2aefcea3462 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a86ad17b282d493c92d59eacb086f4af : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/884ccf92ee93460380fa6f75d77ebb18 : SUCCESS in 1h 00m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4707e2cd83e849558662baa4e92d831c : SUCCESS in 1h 03m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75f67f90a9a84f6e8bd7b4985adabf18 : SUCCESS in 37m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e828992d8b848cebb6a9b2500a89628 : SUCCESS in 35m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e523f3e3ac14202a3445d2c1093c9e5 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3dd71a1b52940eb9d2d5c2239bab06c : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60aedc34438b4934be5862efc855c480 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d79ee571a9ffb65df75998bb4aca652b8d9e2b4a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-17 17:36:05.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/967738/76","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d30c4e77838c4f6b584df60df6b488c0d97cf623","revisions":{"e900d4654ec21a5c00ac191df59b302d06284cb7":{"kind":"REWORK","_number":1,"created":"2026-03-16 18:48:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/980810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/980810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/980810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/980810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/980810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/980810/1"}}},"commit":{"parents":[{"commit":"3860917a3f03d7664a26683b0f5a9575a0b9de04","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3860917a3f03d7664a26683b0f5a9575a0b9de04"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-16 18:46:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-16 18:46:07.000000000","tz":0},"subject":"WIP make Timestamp jitter non-default","message":"WIP make Timestamp jitter non-default\n\nChange-Id: Ibcc76c59494497648518ca51c5dc6b93adbacb66\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e900d4654ec21a5c00ac191df59b302d06284cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e900d4654ec21a5c00ac191df59b302d06284cb7"}]},"branch":"refs/heads/master"},"d30c4e77838c4f6b584df60df6b488c0d97cf623":{"kind":"REWORK","_number":2,"created":"2026-03-17 15:34:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/980810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/980810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/980810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/980810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/980810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/980810/2"}}},"commit":{"parents":[{"commit":"29cc5848ca993ea5d3d11eb2019b5dc56dac281d","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29cc5848ca993ea5d3d11eb2019b5dc56dac281d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-16 18:46:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-17 15:34:18.000000000","tz":0},"subject":"WIP make Timestamp jitter non-default","message":"WIP make Timestamp jitter non-default\n\nChange-Id: Ibcc76c59494497648518ca51c5dc6b93adbacb66\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d30c4e77838c4f6b584df60df6b488c0d97cf623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d30c4e77838c4f6b584df60df6b488c0d97cf623"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
