)]}'
{"id":"openstack%2Fswift~981158","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I0430793ba29d1e051d9cffb5c47eced1ae6ecc79","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-07-13 06:04:46.000000000","reason":"Change was abandoned"},"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-07-08 15:39:48.000000000","reason":"\u003cGERRIT_ACCOUNT_39146\u003e replied on the change","reason_account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}}},"hashtags":[],"change_id":"I0430793ba29d1e051d9cffb5c47eced1ae6ecc79","subject":"Add helper to raise socket.timeout in unit tests","status":"ABANDONED","created":"2026-03-18 16:58:54.000000000","updated":"2026-07-13 06:04:46.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b83528d8e5eaead3d8c05fa13f58e08eabf5d404","_number":981158,"virtual_id_number":981158,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-07 18:16:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"all":[{"value":1,"date":"2026-07-08 15:39:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 17:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 22:55:52.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"}],"messages":[{"id":"bdeadee8cdfc34c4818e0568115fbf9ad0cd714b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-18 16:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c1ea0b40b072dcdb371bdfb9aeab807e440c34b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:41:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0927854ebe054a7fa60d865f36155a98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47f1d085adb4e7e81ea4591dfcd5b20 : SUCCESS in 12m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/387db77fffd043d5ac1f56ba3563abc3 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adf4858d5833460e9ec5ea0a2ce28fbe : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5f4acfa6d5745a9b034384041262643 : SUCCESS in 26m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42c8066ef4c244b88b346d930c7bcbe0 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c0472cc302c4490b25521efe4865be0 : SUCCESS in 11m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37013f86285f4fa28bf81cb8831de7f2 : SUCCESS in 22m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f75e5b866a8f4badae485c5bd212af5e : SUCCESS in 35m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9cba916c088443b081afc2e2e3704d50 : SUCCESS in 23m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5cccdef2ad9346f7a072fd0bd96cf4de : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b9af9734b3084703be180a472af95476 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d31b17ce52c48289bba605de9a534d7 : SUCCESS in 8m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d8cddf67ceb450fbb0fe287e8382991 : SUCCESS in 25m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875e59294cad4c7a8e5daf932d5d8499 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60698041c0a3477586cb23553b971859 : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"abb11322ad3e08f82d6203fb93b6c7ee727e3f5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 21:59:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16f0e155f924430688d7d39744c7cd0d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1328a780b84543a9907e596bfdafd1a9 : FAILURE in 36m 17s","accounts_in_message":[],"_revision_number":1},{"id":"00629bde40cd81310a4c6689588755639ad6db99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:12:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b017aef6f567f9af96abbf6d83a49926a3cd853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:20:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b367681c33fc6678d00b1a2bb75eb73922ad5aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:52:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7ad93efaa0945deb2810d297e4c9066\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e4f9da9eec648bc9429c0f2de26eb08 : SUCCESS in 5m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29ca1a57d02547a6b60206c824cb9cdf : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6c5da82ae5541eda1711d9869088ac5 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb2efa84023b465aa0ef1e80497f6da2 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c19fd8d211254239bb8c3836724835ce : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5838a94f6d3d406590558139a6cf80bc : SUCCESS in 10m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3364dd78ba5d4affbb9461cf526f955d : SUCCESS in 16m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ea9eec59101b42a5a1668fc06bbb234d : SUCCESS in 11m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1d9d73bb1e564dc4a39899a68f32fd26 : SUCCESS in 11m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da44a940a7394890a6f32633db0bfc94 : SUCCESS in 8m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9070a3eec49547f3ab0994217b27671c : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2a47f1b9b49f44b8b125a8b50adcbb78 : SUCCESS in 8m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aecca40b71394cd0889237b6e5c5c411 : SUCCESS in 25m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3be95d50b754b02beb7e05ced3ca8f0 : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9df6cc2f95e142ce900a5329d4e92080 : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":3},{"id":"596f056ff8d1da5ef88ec8dbf309f2ac5024a312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 14:10:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a66028a74f445c2b8852d96d7985c12\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd5270248ba54f58840de252487c2eed : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":3},{"id":"498ec54e452f90066d1f06b6addc8e2e608f27e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 16:41:57.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"89bab57ba2f6612d834fda9b23bfde997b648062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"81f33b65c4c6260d05829be3b5ec21cab0f6a759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:02:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4155f52774544fbc86f7f308ac52ad09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f59460a275d4723ac50274eb89c7f4a : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e227ba1598c4acfae7ce7a03bde42af : SUCCESS in 8m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d6c56a0513c46658a177b260443f1cf : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e70779cfb3834b5883cbc011aefc1153 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a64dc19d39f243a2a37334903318cd11 : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f40c8d3e60a54a1ca22c7c747f8503db : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dba530c94ff54be1905be4cd31b28bf7 : FAILURE in 26m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/963a2882b0984b93a013d6a5535b9e67 : FAILURE in 17m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9b96f1aebc5b415b9746c5f7be3c38d9 : FAILURE in 21m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/27820a0258104bbcb4d85d2fad33df08 : FAILURE in 16m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d3a80496df4e4a85890f70071281c340 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d0ebf19525254b4bb494b6d786b5e3ce : SUCCESS in 15m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fdfdae0df3b44085bf71f10396a548a0 : SUCCESS in 13m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b81d72dafac44a4b96b8219ba03f040 : FAILURE in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93a6cb08ed94decb70117a3630216aa : FAILURE in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03218c82be7440a7bc6a0f873085bd13 : SUCCESS in 25m 06s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_updater.py","accounts_in_message":[],"_revision_number":5},{"id":"198a1c14ec516fdd46f7ffbbc2bd9b356fe481f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"fe1dd6da2baf9886694d8fa0c356baf0c9a71dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:25:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e02377fd3f734e0c98aeae5bd70b1c34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbaa23673dc4ababd391bcf69ed8dca : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17fc90eca0114945add01d55f8e55369 : SUCCESS in 9m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14687fcf3dd94f1ba6ee803b31799412 : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2b21e4bfa264d17b5c150d8d129dad4 : SUCCESS in 11m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7812277216f24a389ca738e10d98f1a7 : SUCCESS in 11m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f68e9a4a20634d17b3f935f9d678da25 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4d5f8333bf124e30a6ce5a473766b371 : SUCCESS in 25m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19aeebbd0bcd446db3dd1d820ee5ef16 : SUCCESS in 22m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2f9580c035e14a44b1258b837aff4e5b : SUCCESS in 32m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/498d8ff637894f559c5a0b5fdd1fde39 : SUCCESS in 18m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7077b7c6b6014f2c848ab9764a8d02ca : SUCCESS in 12m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/599587212db14757bb31b12966d4c48c : SUCCESS in 16m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/80239b2f21fc4092a58fd7fcab10a35c : SUCCESS in 16m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90028bb027c24b6db9b08b4cb165f0c6 : SUCCESS in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce872ea9cce497cb81fabf229b632a4 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6aa726d495f04f2ab1ebe47b2772287c : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":6},{"id":"d3128dc2cbd4a7831f4fdb9e1721cbb592c3aade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 20:00:08.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c733d20751ad43769346ab983bc60ab7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e9d8a2770714a028960c8583c393d0d : FAILURE in 43m 33s","accounts_in_message":[],"_revision_number":6},{"id":"2dffed942500f0a772f2a93b53261aafb79d0475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 13:27:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"17915d8db3963365eab7a014968c80172a192e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:12:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221351ea96bf4eafb042746098f871ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646729a579a44e829e58b313ad2c36c4 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9c5d20d9b9b437a886c283f04c0096f : SUCCESS in 10m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e56577bc3261409c9db1e191efb1d741 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff5b9d48eb6d49508abe19cf01f0a666 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7289ef8fa79c45528a8eaf51325925d8 : SUCCESS in 17m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/091d48028ac8434a8203cdadfc3339f8 : SUCCESS in 11m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/db1e441909c24a4ea1727652eb73e665 : SUCCESS in 26m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a1589cef16224868857a68043bdf1627 : SUCCESS in 23m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9d8be072b0bc4ec5ac56cbf96ee8077a : SUCCESS in 16m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cfea5f6738e7410fb96b03b6acfc56f2 : SUCCESS in 21m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7e72bd456fa4bad990cd8e8908b966b : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/275d51d8127e416a8d322495e5170e63 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d75cfb93a0464cb6804be43def2be915 : SUCCESS in 13m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dac20be44694f5a856ce28686999fa8 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18e206e1c954dec8196e41d303c46bc : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a22bfee2e3142d3978cc6b8fc7026a4 : SUCCESS in 16m 29s","accounts_in_message":[],"_revision_number":7},{"id":"ceebea9123876b6ea397eb44079c2fdf33448d8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:31:24.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9db9f7eb691488fbdbadbd3730bd2ea\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aee9b0e0b05e40fc8b468c6ee65209b7 : FAILURE in 1h 13m 17s","accounts_in_message":[],"_revision_number":7},{"id":"7b5d671bcf2fb60bf58467e5bb28eada1fe40ee0","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-06-02 22:55:52.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"44b9aa25b2acd7104d57f72000f24e7c4e0808b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a083f9097a0390460daa19616dc035f3e785dab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:04:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ebe440917f4d5e8fc9ed91a0199036\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127ba3d79e9741e59ff28f63a16bbd39 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d81ed900b624895a9abd3fbfe50ba89 : SUCCESS in 7m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/616f65d9c46a48f293bbc16a1f8ad27e : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf895d5736a744788f730569caa3c390 : SUCCESS in 13m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2eb593011f04948833cdb107b9b372c : SUCCESS in 25m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a0a4c6e6ec741f4a2c494d459d069a8 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b48b18ccda8c4d7d986a6c0d83bc1308 : FAILURE in 7m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0b8c8522c7ab40cb8fa5b12dbefc2a75 : FAILURE in 25m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fb6fdf757ac141d58b3ab17f3895a8b4 : FAILURE in 31m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5a68095460d347e3b0f0f02f8ee17d1c : FAILURE in 22m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/244fe00d679c439aae7d4e0c11b998a3 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eccf4001a24a4dd4ab2054c82dd06e32 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/964b22860ef54cdd856f8544cf835bda : SUCCESS in 9m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd7d0397faab4adf9374e7d1ac0ec1f5 : FAILURE in 30m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe7a9c52f7049d084c5543166a92343 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6cd7d0f3f71743b3ba43687b4cd4de5e : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":8},{"id":"dbe7c6d26eb34d4107d2761924816c38a75f10ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"47108be3fe694f96a80d89511579b57220048185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:28:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab54b198f2874d7fa6417b39de5587c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6287da5dae46e9a0eb42a16ec981de : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f34ae83963be46a5a94bdd7c8a2f9055 : SUCCESS in 9m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c63737de10443d391b8a5f9f9778e4d : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20b6275b2cbd4a4ba497e00c39ee1b03 : SUCCESS in 13m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1651720fb100421abf3ec32e0ce0e18d : SUCCESS in 17m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80aaccbb1efd429f9a8d5062ec3e557b : SUCCESS in 8m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1584ff9bee614cf69e7b02787ec59c09 : FAILURE in 5m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0ac06a512e214f85bccece718271a4a2 : SUCCESS in 12m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1c1226b7dfd04e978cac6122f03d7da4 : SUCCESS in 29m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7b86386e85e54105bee9456b9e5f2455 : SUCCESS in 11m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/268c4a38006847de8ab22ebbe822fb5e : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4ee9701ad5c547aa9e26edbb50d6843e : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dc706cc4863b429d8c2b0866f78f6d92 : SUCCESS in 10m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03fb877a01af4908898d219e3f5e1895 : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9490ab20374493b5bc4a7e8f2f1e8a : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/921b6d4569594180bf8acb734a831fc0 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":9},{"id":"3a44aa0c75b698606f126ae576de9fcfcab6c03d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:41:00.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa2527bf0c9945c697a0f64b8b119264\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8036d244a1594aaba3ce002fbfe7f237 : POST_FAILURE in 1h 15m 27s","accounts_in_message":[],"_revision_number":9},{"id":"fcc0f4157b60a47418cd597b8273a8c8649c9fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"5c9c33f0cbb0b6af4b9acf02e76c9b2b1614551b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 08:11:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e2b4a8bc054a5c849120c4ac78829f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af90dd94e50e46ba84cd50e628659a55 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":10},{"id":"395e53a02dc8cfed82bd198f15bd0f7b3c021227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:09:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f3a775f4464309853fd3f522823238\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/283e5b5986c94d308f70deae9c7e8393 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17f8b930e69e459d8a3035ebf4a94020 : SUCCESS in 9m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e41a730dfbc428596f08c19545bbb1f : SUCCESS in 12m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36eca645207a40e9bbc0ecee3beb3e23 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7c71a8beb6e4754a18375853bccc415 : SUCCESS in 12m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9448fe6fa4a49f2a3990b4be34f61bb : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46292efe9bed46fe862cfd66c429cdb4 : SUCCESS in 19m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3fd83d26bded439e8eddda5ea76f79ac : SUCCESS in 17m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/447182afa1044c04bf9426f0ef6be168 : SUCCESS in 31m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c40e4b23bb394933b36090362fc4c084 : SUCCESS in 21m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f1eb50f961ec43239faf0a7d6823a05a : SUCCESS in 10m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90a03f3ddee74fcd9ba00e81abe1acbc : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f0d5283e11224eef83d556b92f6bc82a : SUCCESS in 17m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9434e9ddc53b4a94970666a30f1410b5 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2781e0bf44984d4788e48ed6e88f1171 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2757de2db08a405fa56b0b15fe522019 : SUCCESS in 11m 11s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"e55210aa81ae5b8cc9f88f603fb2d8ab0f2cf5f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 17:22:32.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"6c8ec151cebe80c42a6e81d9af5689b271716e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 18:16:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02bfe51e0bdc4d5f86103a325b11787b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2ca9c48c7f46d49dfe9d1e656881bd : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9fc8bad20484079b8e0cdd9163a5bfe : SUCCESS in 9m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b94b353b938542e7b5e246259c983dba : SUCCESS in 12m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3065abc93634482a93fa7cec44e75366 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2b8d812aaf64ab789fc209e007cf7aa : SUCCESS in 26m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a1fae1728004ec78049982b9a647366 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab08cb0ebdca419ea1f4c032a358ede9 : SUCCESS in 24m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0fbbcdcc95004618bccc5b79d2527abc : SUCCESS in 22m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a71d6be2e39843f9ac1154950d457852 : SUCCESS in 13m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9f6f914a5e024e5895f9bb7041cca71a : SUCCESS in 22m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ffd6b5dc15334365a76759e653e977c3 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90d755f35f0e4f0c8eb8aacc14298e20 : SUCCESS in 10m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a198a73210df4a93b27b14de2b8325db : SUCCESS in 15m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a2af178ea85455480033f1ee9c3dccf : SUCCESS in 32m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432e232ac25f4e9190631b1613c13943 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/73cf1aa94f9049b483c290a95a6fbce0 : SUCCESS in 21m 39s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"11b0e2784f42ae7318ec041ee5b7c9bd57f81cf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 02:28:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8683010f9a1247f7b9e190f809144438\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/485c0fe09c6147eba6318d43996f99a1 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":11},{"id":"cdd921df7de92d5e9e33f10b23109c87084921c9","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-07-08 15:39:48.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b83528d8e5eaead3d8c05fa13f58e08eabf5d404","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-13 06:04:46.000000000","message":"Abandoned\n\nRefactored - content moved into other patches.","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d495a2b4d1689932da0e01018496d72e71fffa4b","revisions":{"80e13b9987d34b6ed8c41e03a21763930e904eda":{"kind":"REWORK","_number":1,"created":"2026-03-18 16:58:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/1"}}},"commit":{"parents":[{"commit":"7f32d39f25ffb0e4d63bf7f1ad5f2a0de937adcc","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f32d39f25ffb0e4d63bf7f1ad5f2a0de937adcc"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-18 16:51:22.000000000","tz":60},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80e13b9987d34b6ed8c41e03a21763930e904eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80e13b9987d34b6ed8c41e03a21763930e904eda"}]},"branch":"refs/heads/feature/threaded"},"3394a7932dc877eb41fdb37e2964b0b615075ad8":{"kind":"REWORK","_number":2,"created":"2026-03-23 12:12:35.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/2"}}},"commit":{"parents":[{"commit":"2f98c3b3bb7a8b1b5428c20de27a58ed0ccff76e","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f98c3b3bb7a8b1b5428c20de27a58ed0ccff76e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:55.000000000","tz":60},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3394a7932dc877eb41fdb37e2964b0b615075ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3394a7932dc877eb41fdb37e2964b0b615075ad8"}]},"branch":"refs/heads/feature/threaded"},"3e661bdcc331ba609fc5a0878796d036f5d3a1bf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-23 12:20:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/3"}}},"commit":{"parents":[{"commit":"1e46efbcd791d621562c7fd667914955aba6c5c6","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e46efbcd791d621562c7fd667914955aba6c5c6"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 12:19:45.000000000","tz":60},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e661bdcc331ba609fc5a0878796d036f5d3a1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e661bdcc331ba609fc5a0878796d036f5d3a1bf"}]},"branch":"refs/heads/feature/threaded"},"9fd4af7dc9dc2e6466de2798cb6fa535fb7a4c0f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-05-30 16:41:57.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/4"}}},"commit":{"parents":[{"commit":"f70f3e8177867c5ed266ac8e16f21324e4104246","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f70f3e8177867c5ed266ac8e16f21324e4104246"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@nvidia.com","date":"2026-05-30 13:47:19.000000000","tz":0},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fd4af7dc9dc2e6466de2798cb6fa535fb7a4c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fd4af7dc9dc2e6466de2798cb6fa535fb7a4c0f"}]},"branch":"refs/heads/feature/threaded"},"2ee27a10a9cd36b54c1422b1d5fe53af96d813c3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/5"}}},"commit":{"parents":[{"commit":"3c2658699dccbb5a42458b45a641508a024482de","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c2658699dccbb5a42458b45a641508a024482de"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:09.000000000","tz":120},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ee27a10a9cd36b54c1422b1d5fe53af96d813c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ee27a10a9cd36b54c1422b1d5fe53af96d813c3"}]},"branch":"refs/heads/feature/threaded"},"834cceb84c53c69a347d8cc58e462314c31a5f29":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/6"}}},"commit":{"parents":[{"commit":"5369ec254553be8f035a606548c11081d435fe71","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5369ec254553be8f035a606548c11081d435fe71"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/834cceb84c53c69a347d8cc58e462314c31a5f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/834cceb84c53c69a347d8cc58e462314c31a5f29"}]},"branch":"refs/heads/feature/threaded"},"3c79781bd8048fd20755f0ed3ee8ae7426ba15b6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-02 13:27:47.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/7"}}},"commit":{"parents":[{"commit":"7ceadb1686a2bf8bb45d9ca2eda1d97bab175084","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ceadb1686a2bf8bb45d9ca2eda1d97bab175084"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-02 13:06:36.000000000","tz":120},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c79781bd8048fd20755f0ed3ee8ae7426ba15b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c79781bd8048fd20755f0ed3ee8ae7426ba15b6"}]},"branch":"refs/heads/feature/threaded"},"14220b47d2b5c7d52f4f682e80bf7f082d6e3d35":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/8"}}},"commit":{"parents":[{"commit":"b35f8eb301eb19e3b77bfe8638efc403251c84ce","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b35f8eb301eb19e3b77bfe8638efc403251c84ce"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:19:00.000000000","tz":120},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14220b47d2b5c7d52f4f682e80bf7f082d6e3d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14220b47d2b5c7d52f4f682e80bf7f082d6e3d35"}]},"branch":"refs/heads/feature/threaded"},"e6a946f1a6562f9f5c1c3d5bff5cfa0432f50a83":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/9"}}},"commit":{"parents":[{"commit":"55706965c048f6be92fffedfefd2852f396997c3","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55706965c048f6be92fffedfefd2852f396997c3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 16:35:27.000000000","tz":120},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6a946f1a6562f9f5c1c3d5bff5cfa0432f50a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6a946f1a6562f9f5c1c3d5bff5cfa0432f50a83"}]},"branch":"refs/heads/feature/threaded"},"eee262451b72b33ea3a491951b70cc2aea5bebdf":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/10"}}},"commit":{"parents":[{"commit":"ed0b1e859568c8ab64287b6a4b09ea244d4a7a65","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed0b1e859568c8ab64287b6a4b09ea244d4a7a65"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:47.000000000","tz":0},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eee262451b72b33ea3a491951b70cc2aea5bebdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eee262451b72b33ea3a491951b70cc2aea5bebdf"}]},"branch":"refs/heads/feature/threaded"},"d495a2b4d1689932da0e01018496d72e71fffa4b":{"kind":"REWORK","_number":11,"created":"2026-07-07 17:22:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/58/981158/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/981158/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/981158/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/981158/11"}}},"commit":{"parents":[{"commit":"32370b7d685104cf48c3ea6822fb0500dbbe9775","subject":"Make eventlet imports conditional to run tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32370b7d685104cf48c3ea6822fb0500dbbe9775"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-13 08:25:38.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Add helper to raise socket.timeout in unit tests","message":"Add helper to raise socket.timeout in unit tests\n\nEventlet is able to cancel sleeping greenthreads, and some tests faking\nslow responses are using sleep() to test this.\n\nWithout eventlet timeout exceptions in tests are never raised because\nthey are raised using socket.settimeout(). This commit adds a\nsleep_or_timeout helper function to fake a socket level timeout.\n\nChange-Id: I0430793ba29d1e051d9cffb5c47eced1ae6ecc79\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d495a2b4d1689932da0e01018496d72e71fffa4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d495a2b4d1689932da0e01018496d72e71fffa4b"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
