)]}'
{"id":"openstack%2Fswift~981159","triplet_id":"openstack%2Fswift~feature%2Fthreaded~I7118c0e374486f33c399575d88e7e88573d1dfd3","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-28 14:36:09.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"I7118c0e374486f33c399575d88e7e88573d1dfd3","subject":"Add gettimeout/settimeout to mock memcache sockets","status":"MERGED","created":"2026-03-18 16:58:54.000000000","updated":"2026-08-28 15:46:51.000000000","submitted":"2026-08-28 15:44:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981159-eventlet-removal","meta_rev_id":"519d7aebace85c95161cf1357d6662d1737490e6","_number":981159,"virtual_id_number":981159,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 15:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-08-21 07:26:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-08-28 14:36:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-28 14:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 17:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 15:19:41.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"da81e8ade4c2faf5a52040e6c38e12373fe9b3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-18 16:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ec69bad995ed58172985bb8022e0e2ddfd5763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:53:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aa8b4ada600410eafd97c00c24a364b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdf676cbf104167884c713b60677009 : SUCCESS in 6m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b70dd5e62a14faa827a8d51dc2d2f0a : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d52c5eff12d54e59b235397366b0fae3 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/877b9268f5194f9687ff788e57485904 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4d28202eb2b4d6e90bb2d708860647e : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1ffb721ed8f54dfdbb0eab3c984147e6 : SUCCESS in 27m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9a02eecb58ba4dcd9fcfe667a3151eac : SUCCESS in 9m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d2a13b1675240ac97b99981d0893cb3 : SUCCESS in 13m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5167a3192fc84e6aa2dbcdfe3def81cb : SUCCESS in 13m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/81a7017e51f74aa6b6352096a9444418 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ead6e9d82d064119a67e8b68aca2f512 : SUCCESS in 8m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6351e33ae5634555976b6645bb043891 : SUCCESS in 17m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfc47b5ce8954ffc8c203b6d6e3de646 : SUCCESS in 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af8005660b284654ad2a78a4b61fd5ea : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b09c590945234ff59b8f99bdd5345f43 : FAILURE in 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7a49839303361b811c92a826b8b02028b07980e6","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-18 18:00:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"682c09b45bf2ec86812559c6f10033dd0baeb1ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9e9075feab44a0aa4ee0fbe4d8d6fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdfdffc83d914ddcb68b6bc2e4fcc009 : SUCCESS in 10m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1aab95aaf29a43fc9dd807f1145e5eb5 : SUCCESS in 11m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1b25f411a7d41ef8a92ccf3fd60ad32 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e77dc6c9a7a042e48c0bae20a2976725 : SUCCESS in 28m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e3feace90414d83bcddca1887a9ef1c : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea26dea4f98a45458a7259a04cc33e57 : SUCCESS in 10m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62b56ff3b47646d78f99c0d223ab034e : SUCCESS in 19m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0ec71f155b7d4d53a27feccc5a70678a : SUCCESS in 12m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/da1af178c07349a085cbb2224f0ed18e : SUCCESS in 28m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7ea563f193d14c6ca965a30d2d4b3c43 : SUCCESS in 8m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e3a02341939743c9a71378493e7351b9 : SUCCESS in 9m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c07d282d2064506946fcf6f38e89961 : SUCCESS in 18m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be8a22e3537742d485f5dfe92b4beff3 : SUCCESS in 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/493c87411181472d8d20343b8efec761 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ff40a8c94964c28ab727b93dcaa02f5 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c3072e41563ac8d73d5f05d08d6ddcbf12bdd3e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 20:43:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/006b2f3f123a4969a1121e9e73304bbf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8051b3ef63904e2fb06c485d5a80ada6 : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3621d9a803069ec9d3d30145dc8aa6b1723e41e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:12:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9bf15c40f6d45a13cb1a63e6dbd55df058d7a113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:46:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/746695cf59fc4e07bec2a022c6c86038\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ae5c5ec54649609744dff3ed2b0dff : SUCCESS in 9m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7323a99647d14426b8b8d9dddabd4972 : SUCCESS in 10m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4156fd7bdf224d7d94671f4a1654176a : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4811b547c6d74b7293388e36fdc6fdc5 : SUCCESS in 10m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3973f4e915f446a4ac2b9e50a288d294 : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/23ba1e5837b44f8cba2b2b7f1144901d : SUCCESS in 10m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/27318c0b912645af9ebc3306c9e69284 : SUCCESS in 15m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40cecf68722048d0be3768493f98370b : SUCCESS in 11m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d452658b1efb405a92f8a9a2011b5f23 : SUCCESS in 11m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eac1b677f7814cea8bbfe7ee144c2201 : SUCCESS in 12m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ddda24bae8d14b6097f0440bb2a0c386 : FAILURE in 14m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5cfb962a332747fea91dcf1b585536e7 : SUCCESS in 15m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/353ba9d1fed64c7eb630ab02aba7daf2 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30afada02c01410a9b77910943455e7e : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a744edadb3fd48eeaf673461ca452377 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d65c69fedb94cfd944a7fd9a97145f1d52e1f932","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 13:22:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f31b71142fbcfdeb872422906b2cc93d337e1b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:24:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591513aba8c145d883ee93c40fc85fb5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6976c33850a24dcc9ed22f1520c37c4b : FAILURE in 34m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b5dda98b2d1c9f5bcf67136afa39f98c31e6b355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 14:00:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/171b2db10e7c4f499a83fbb395b8dec2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be583ded12ec4f9f9136c64d6edf3f47 : SUCCESS in 13m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8132ac74fe944eb8c3509a5045e2211 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dea7f941cbe040f1937ffc98f2f17b2a : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29a7e8fbaa6b4d24894b26e309437235 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b767a20569540a594d734c37f5058dd : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/066e348c08ed41cdaa970212980b9eda : SUCCESS in 21m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c590271c6abf42bcb884c4058110c3d8 : SUCCESS in 23m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3851ed00a9774f858ac2407e21c4c694 : SUCCESS in 31m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b71fcdf8dbd84859b457ad349608dba2 : SUCCESS in 25m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/054940091fc14ec3aaae537588b0176f : SUCCESS in 12m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/05fcf28ae08f4494ace31dd09ddbbfd2 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a3c6d77c6b924600bfb2cf327ce31b0c : SUCCESS in 16m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2cffe16f15d45338539d101390f1e95 : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1324f0785cf74154aa66c74f4f3fc490 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d6c342e3ed6496c96dfa67b3d480386 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a8970d8fc39bd24c2870d196472de3f2d36b1b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 16:41:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"41ebbd3c18d0809b558ae484ded5e0344d1556f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"7b8e1ff4f867410fdfb7e3cac8f123b6b4004a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 17:57:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e70657f574774052be3e99e84ef3feac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f187782b220a4a5482256bc05bc17039 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6b75b1ba334481493277fd8c3eae437 : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2d95bff8bbc49f8a549dad9e68086d9 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61d595d3d9924f3baa511266efd1a462 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7a373e6ab6842889c9f6e7b9149aea4 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64fb0013317644a7b80e28af49611a9b : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c8ab5a959d5429e8ec5ab3a37d5d251 : FAILURE in 26m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/08260b4b76634ee1a12f09cacaefb458 : FAILURE in 13m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1ae3e6d9b65f4ac69e158533b3a15805 : FAILURE in 31m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cf6c28c808a844268414027dbef6cc8d : FAILURE in 25m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6952039e82944849a729c3a2035dd4b1 : SUCCESS in 8m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3566d4e6fee84747b0444f0749af2c55 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cd433665e38d4a0cac7f0bd1f37107d4 : SUCCESS in 16m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b44a1c8d79c4c29a52b500646cc9c7f : FAILURE in 27m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c213fa6a63e9498996c393c1c387b7b5 : FAILURE in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e60b41a67db34172821059e3e6c4ebd9 : SUCCESS in 18m 37s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_updater.py","accounts_in_message":[],"_revision_number":4},{"id":"25d8189a089a7284534f9fbae5ce9e3a747d6cdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 18:22:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41662ab129c940918d739f1693e6910a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aed54759e8b94982a8dfb5c9174c8988 : FAILURE in 31m 47s","accounts_in_message":[],"_revision_number":4},{"id":"dd70c992f05023591947453958013b9659aaec21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"c3fa4017ff4f5f04f64e6b27ec1f4f5330b08c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:22:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff4a0568db44b14b8cb2150ea3e066f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cdfe9473ac04e8fae8d99030630b1f2 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75266985f2f74093860f62c94e629548 : SUCCESS in 8m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1da5836def77469fa3195306884eb28f : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ab9d693494d4515af8304908f756ed1 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/184b37728e5c4e1390f51ad6cfdbbce7 : SUCCESS in 21m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c1f576707c8448798f57a7af5bb0e81 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/94f42ffa29604c1185bb01adc3c5baf4 : SUCCESS in 25m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4ed42dba7e1b40998ca32a5e4b8a8944 : SUCCESS in 18m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/95005d63bb3842009870b4af12d3a587 : SUCCESS in 16m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f0a3425c98e647a08e20b15e6293251b : SUCCESS in 17m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4801fdcb821a4470900a8972cd505c22 : SUCCESS in 10m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/991db4f3e5914e93a620a8cf8c19f7a4 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e9cf481412a4a5abefbd2636bee31f1 : SUCCESS in 13m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59c3ee5419264b79a33187a9c96c7a8a : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee92bc4f2694c2cb073f7073b2c62c4 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92a7a6b3c5034976838977d1e19df7fc : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":5},{"id":"168ffa09b1a4eead9e1a67ec2ed93b31721b5f7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:24:08.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc22975ec184018bdd95de987a4f245\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c604a57164c4aaa87929b501684ee54 : FAILURE in 28m 23s","accounts_in_message":[],"_revision_number":5},{"id":"b602a4a1a4959228263b07b3584bee59b4af6e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 13:27:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"dbff165f7482c11244bce4aa0f47a8c291a2ffa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:12:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd11ec79eb8245f0a8b27677ad84559d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84c6733a9cb4340af03088488c9af64 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baba1064ea8e4636812d37edebd57d2f : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5b6d87edefc4c2abb583cf3c37df861 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ecbe4ede647499aabddf030e0f89902 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2921eade4015498a99ea14df3cec4863 : SUCCESS in 22m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34932fbd6a7149d9ab46daed6328b1b7 : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dbc8aab706684819af069a0775913459 : SUCCESS in 25m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5be36b66d38a4c05a8c9ecb26ec62e6c : SUCCESS in 10m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9b800e707df94d8fb7f4a219006b080c : SUCCESS in 21m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/67fd827dee14449aaeb7657bee9acb2f : SUCCESS in 22m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/82cb41eff8e04af9a2880c2020edfd38 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/97e834a94fa2405fa3f236abae5c3fb2 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8d2a9175159e46189b252e745837773d : SUCCESS in 16m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44989dfe92cd47e796b213a2b04d2a89 : SUCCESS in 30m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bf99018c9a47bfa4a9f9226c0d634f : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb3edffd62574697956f936d2d99c3d9 : SUCCESS in 23m 53s","accounts_in_message":[],"_revision_number":6},{"id":"9285f9920953021293cb3bf5e4b96875a195a5ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:19:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b1b5950bf1b4f4ca9453f9ef14ce771\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59999559b22845358714e80ffcfe3cfa : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":6},{"id":"5efba78ff329a8de37ec5866ea38693b9b17983e","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-03 15:19:41.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bc1b4667f1c9f96417d11ce9904e63d399fda205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"923f585cdba6b41519c766719f7df8651104e951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:27:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5184500351f5416880eddeafa4841906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8064f3ca5eb54dd1b59dc07d644a4ff6 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2be05083d6fe4160a4f0054a4ba0ac20 : SUCCESS in 7m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33d1173a1fcd4c92bc6b26d51b52854e : SUCCESS in 11m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ae288d18f2945fb9095e30a38710b8d : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25ea26f96ddc4d27a5d50a4f6e6c3426 : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bc4b313ff1d4dd19aacba421ec6398f : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/efc4ac8dfdd14d92b1f83a201077ecf4 : FAILURE in 6m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/369d232e2c9b433eb7dca71d5ef42e21 : FAILURE in 20m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0a8d47c8a03a4a8a9fe965b59fe4b3ba : FAILURE in 31m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b6bbd9b4d9a44aee83ea8530134731e4 : FAILURE in 10m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d2cb97671cc470f89faad39c15a75eb : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62319efae887416e8997f6ddfe0f8d33 : SUCCESS in 13m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7f62afd24d574d9fb3cfb0253059efd4 : SUCCESS in 8m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abef28f85d034c4da510024cb282558f : FAILURE in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d35c426069f4c15b5895b6a92a789d0 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92af96acdec84f67922f845c24435c0c : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":7},{"id":"2d257c4de2f7e06df4ab92fdc78a6eb4318097ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"7e666c1e11935071c9a2039cd8b71521bfa97b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 06:37:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e3e9ad77d7f489491003c753b250d61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/866950a10ce449d094e2bd9fd7c04dc1 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1d764034e2a428abfabfc9951046718 : SUCCESS in 10m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95fbd25df39146488948bf169a1f14e0 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/600ff80838164588a742861baa3ea9c0 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c53ca1b79d97458dac7a084d77254a13 : SUCCESS in 11m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9e7fdff2cc046778c43f3d6151f60a4 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fb506656ce81445c82d4d793865128c2 : FAILURE in 6m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0ff001d7d1c7410298a667bf492e8655 : SUCCESS in 25m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9ced757a902647ea9a83544bbcd6eda6 : SUCCESS in 32m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3e143fa2d3aa4f3aa479a64c4d5e7870 : SUCCESS in 22m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5471b6fe6a32469a888dafbb2a1f1bdd : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0e1d65b5df4347ff91b53e2e75c78571 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3fd0387682bb4f20b61ef51dfedbc615 : SUCCESS in 16m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d56e885780e4948b6c86eeb1cc83cb7 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53000d6a049f4eacae83e04e7acce922 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36bbb197f3de40238b0d5e18d22e952e : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":8},{"id":"ff82f1da7fc27a663fab504dd512e4cd4ca0e570","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:10:16.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d10fcb91f984122a1be3f680d722bf5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfea6a921d1f494b8a0db9cd8a70485c : POST_FAILURE in 1h 00m 18s","accounts_in_message":[],"_revision_number":8},{"id":"ecc71b5f934004b47034fa27d99e5fcd3cbf32d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"83a4543b1fbdbb5a619c396a1cce38521987805e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 09:53:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79dca46d42984717b37441cbb30cbbaa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/648a671dddaa4a838b5b22c8912fbad3 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":9},{"id":"2e7d0e584bb386c7ce1a842b43f5ec1c90490cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:04:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45572c8a1804f7f8978a04a4b54f157\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a42437110348b081c7e1bd95be2dd3 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e7f5e1015c486883b730a4012873a8 : SUCCESS in 9m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d67d38620ea4d1891145cb4eedb46f9 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efaea314a44743d88b8aa67738b9392b : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9bb945cd03c49c7b13901fb1d75e670 : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb1db1b5665b4f04ba6dc5932f352ab6 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1a4a14590db489cbba674f2889d0ea8 : SUCCESS in 20m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e9b61f24674e41379eec38c94fe5efbe : SUCCESS in 22m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/da7650ef174a4a10a5065abf44d4bf37 : SUCCESS in 15m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/12dc9ed90246474498fa8707f995aa98 : SUCCESS in 17m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7585fca0165840009a592fd636d65f67 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d9425e0c24345ba8d43ce3bbdde7841 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c254e6d969014257b61d384392fe3fa4 : SUCCESS in 10m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb638bf0aa84e328bb92c2fd048a612 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22164e07f93c444088e5457a6a41eac7 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7906061f38044b1d8ff99c1ee2546cc5 : SUCCESS in 20m 54s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"8383a7f6d90f33e1e4485a70947d0f3ffe15bbf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-07 17:22:32.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"3908dfefc2578921b880e9262c84d5355b57487d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 18:08:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee351634e6446d68d30633e518bb19f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cee226057a342b597f265137ada2652 : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9e5dd0006e944c5b9b0c6a620e0bc9d : SUCCESS in 6m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b770925285b4ec39373e60220778c2a : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9c2afe4573549349771a4d642d29f01 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dcaac88024e44c01834def939c33cac3 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4663ad5a021a4e76b96a3fc886b0c24c : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a3a8f110332430dbcf53184797cf909 : SUCCESS in 24m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8c4c763bc5ae4159bc97de8761145e1f : SUCCESS in 21m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8929b9996d114357827d5ac5fd23145d : SUCCESS in 29m 05s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/723c27198b6b489b9b5bffbe0cc18a58 : SUCCESS in 14m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/006449b7aaad4a229aba21ef8717cedf : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2cfc03f6ede040169c84f06ed68c45b5 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2a1f9d782c03417d9958d2421632bb5a : SUCCESS in 16m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48956ae4b50c48dfaded10a8fc8a7643 : SUCCESS in 29m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e26527573974ffa82598db41827d98c : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/987532da47b442d38671607516cdb694 : SUCCESS in 18m 38s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"92187442efa971761b4974b104ecfb73714c3c1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 01:57:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1453e82470a04d6fa6e545ce8ea7239f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e92cf270539426bb98cf318b45d04ed : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":10},{"id":"ee93a5c4260fe8240f04e37b86d2b80219cdeb41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:19:32.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"f426c6d8bcf5514be9e5d439fbaa0d43ea94fc41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:07:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f50a344921e4a6da14547974e1dec9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fe8bb77a7147bbb9d0a64e5b1f1b4f : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e692657a9b614689abe3739f576ba28b : SUCCESS in 10m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aed25fd394e446329527628bf8817b06 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d904c07ba3514f07b17a9b9b1b9aef37 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f2ee4ac85814f9587cec45dc2c22b94 : SUCCESS in 22m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40a363cc7da441318dcae60dcac455b5 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8a2c6720ebb541049ab5ddea107b750f : SUCCESS in 26m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c286665f47e444278d1d968f49628378 : SUCCESS in 21m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6100976628704fdc929d3c37538760d1 : SUCCESS in 31m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e2693b59529c4cb593bbd63cf2be68dc : SUCCESS in 26m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/314e8caae66f4ac59b1e44cbff38ebcf : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/601a4ee2d4314089915da105dac1fb18 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/321b102ba1aa4ae9a01c459163ee3cf7 : SUCCESS in 16m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/852c6da3395646e3886e16b786393d5b : SUCCESS in 28m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0aa78c40e64dc6b84798b9cf6ed686 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8303c4c5aa984fc5b9873af92ce848ef : SUCCESS in 21m 54s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"dd26e6eefce2e41b7cfc467dfd1cd26e725c1fa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 09:44:30.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1e01493d6a4140a6e6ba1b07c4488a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6eff213ff70a44a6929dd2f35403f9f8 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":11},{"id":"2131d9f07f017b20cc7f34be360dccde44bd3ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"af72ef33f983a8f302018706bfa55ac78174cde1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 11:29:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51994d9115a64f129cd995a913e01a40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3dc62ab906f47789bab8184fbbcacec : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fb0b756c67542a7b39fe579807f42d9 : SUCCESS in 9m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b61825a8deb41f79fecdfd2feac7164 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ca5c79215b6486a90750b0df7bf309a : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d5ad11716a640888e7eacfc192f17cf : SUCCESS in 23m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/523a84a7ca934e0d9fd329f9856e15b6 : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c1d08ac989549a8a444e67ad31e970e : SUCCESS in 13m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a82ef9a9af594100a12497f61a93e237 : SUCCESS in 11m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6dee517b81b84967a45fbeec3d90489d : SUCCESS in 11m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3b8725aac7d648afb19eaa49170a453f : SUCCESS in 9m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/66532b21629141a48804f56157eda612 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/23b1939166924b8db3a4196d7582398d : SUCCESS in 9m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ae0e6974fc6e4d628240b7a3cad1e3ea : SUCCESS in 16m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79a2270a0f05409aa6a842b0e28882a9 : SUCCESS in 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13214153e4c74c07bba1a26786fcc72e : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/919bef4be86d455c8a801d8427a28897 : SUCCESS in 21m 00s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"0daf4948b69dbfbfa05f121ccd15f21f2f987e0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"f22742246dd6ba5d3acd4ff2120891fc51d34ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:27:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34cb45d890674b92aea2c1e02d7db8da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c69f80ad345424291903dace6c0293c : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d15b18f5e0d4780a981a4876d514d0d : SUCCESS in 9m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66773e7aa6ea472ba90efeacb16680d4 : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0423e8084a6747a4a2150c4f75b698a9 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98143f372eb042028c37eab2fd4dadb2 : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7bde120d7545489cae267d2ac0877f87 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f2bf06264cd4fb690693902009a389a : SUCCESS in 11m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/58cd4e82afe64b0f8ab9d7e650714c2a : SUCCESS in 19m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/469c03008f6a4302b5b6721a6d4c4b03 : SUCCESS in 31m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/eecf086eceaf46ed973e1a447e1a4bb3 : SUCCESS in 20m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3ee1afc171bf4a4cb386bc2de5a6ca32 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/df2f5cf688374600b468144703b25b5c : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1558abd01db144c19555ecb724936951 : SUCCESS in 15m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c45dd65eb3dc47109d209483f1130010 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d06271b05554a77852bd91cfdb9dca9 : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7c04abe5ba34ed0a62da24fd9adc198 : SUCCESS in 11m 57s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"9b84e696e7a83795306c3d2861877c13df88f009","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:44:36.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a616e3d02a4aadb09467449cdd5b3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b61fa5d88b364595972fc67630526ade : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":13},{"id":"a6778e6d41e01986ffa76079ebc04d24c7fd9b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 06:02:18.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"40e9fe87e802cfceb1b2e03446e88a6155a8baad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 06:47:22.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/883fadcde09b4ad8a75ace342a5b6df5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75a254feb6249028f2fcb19332c5a0b : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cade6054bc52404c863e2e4cd040ccb9 : SUCCESS in 9m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c1763a8ed674e0b8c6e14f0de8e6995 : SUCCESS in 11m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4772b07c03e450cb99858a1351c4d35 : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4af271da61634d45b4709e5c028fca26 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/908bb9dfeec84c29be053855447d540f : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3514b5e941f34faa9cb9835f341ec8bc : SUCCESS in 12m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3326021c606e4c91bbca071369e506e4 : SUCCESS in 24m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/606ff6bd4e96450fab718643c3bd3105 : SUCCESS in 32m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/254685904c34430f92faaa5c6e91b042 : SUCCESS in 9m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d54594c2c5694c86acb3cfea31bf0b0b : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1dd1c439ba1a4e5f8b605db629c75ac8 : SUCCESS in 9m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4219734009ec4b55ac8c77dbb10bd08e : SUCCESS in 16m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/021bbd43cc27434496733308253a07cb : SUCCESS in 17m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84fbecf8765f4457af6282d2ecef525e : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8614b145e64913ab5b5a4e1d4b724e : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06506d9b4b614ff18e69b7ccfab9fe07 : SUCCESS in 27m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"136efd239333ab4120a7de5695f3399f708d52b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 07:26:39.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"6b66a59f228c27377dd19327f3b6366d084ebc16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 08:28:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93cbe3b80e9d4151917414279feb2861\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ecbb0ddf3845cf805abc9756c815d8 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58d9a573c73b47fab78446bc9616c460 : SUCCESS in 10m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e12364bb25884f5b8d6bb9f7e8db4572 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1649a31a0d042d3b0e62b142aee0af2 : SUCCESS in 11m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf0e699996b647398a47c6c9b54b6313 : SUCCESS in 21m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f64eeed49bec48598ffcf5d3e4763b62 : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce67e4ebfd3947a19de3129ad246f276 : SUCCESS in 22m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18bc61b7194b478f91b09eb48a7be1b2 : SUCCESS in 20m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7d577f4c107147fd96442332be7eab8b : SUCCESS in 21m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/36c02b5b8964432eaeaa3fa3de910fc7 : SUCCESS in 20m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e3e1159a64a44b8aa7c373b4097cec03 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f7fd827d439140778aa9cd9a0c3c40a0 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3b9a5d8ee4874ed0a33ace8b08633db3 : SUCCESS in 14m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cf065f4dd04472a9518e7ea001c8f19 : SUCCESS in 15m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2177a13607ec4c06be407d3b2142e8e5 : SUCCESS in 22m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986b4f50079f46cfa1fbb1a24ec880bc : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0284598b66ad46dd943d79e0761606da : SUCCESS in 19m 51s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"f3dcb75c909c982d85ce9abbc6edcc9f7a3bd46d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 09:11:57.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025e394db6f44f0299c8de12cc613109\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6746db9c1cbb4dd2ba0e0f09ea3ce138 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":15},{"id":"1f66c8a26a417038fc1046df8502ccfb1bdbd4ed","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 14:36:09.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"523b466c6eccda211d5bbffc20cf57bb22e3e0f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:37:33.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"70ef969c4b6a54178054ced5df238de8e0c7fcd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:55.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18de006ccccb4d92a24c059c0a88c4ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f14503e9eeb42aca5ca739c3d658583 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88ffba878a5c40388878dbcf4845c1b1 : SUCCESS in 9m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/35d4f5f1493b46c2b07efd29be1405d5 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b8cfcc8f1554b0c91d48971122cdc08 : SUCCESS in 26m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/558d9198b4394cbd9b498622f73ec958 : SUCCESS in 19m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3d7807f1751b4985a7aa6d585dde4946 : SUCCESS in 33m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5fee6b9dabd94cf8b7b53c071f7999be : SUCCESS in 10m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bfaabcfdb20048648b5d7b2bfb19d3b2 : SUCCESS in 9m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54a135269c2d4fea9b91de72ccb89d9f : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5587b9f03fa44614b8828639b63dec6f : SUCCESS in 9m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b7911a08b1048d8ba2e232839934b04 : SUCCESS in 14m 24s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd23e890fd7f48ea8c996907f6de85f1 : SUCCESS in 6m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0888ed53e5064b5c88e613e4c861f6c5 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab1303922f147b890a799d5ea50522e : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a36a5a1457d8425fab6ff1123562bd4b : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":15},{"id":"6664a8e3f3c18230fcfbfb2f483671067f15f7cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:44:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"519d7aebace85c95161cf1357d6662d1737490e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:46:51.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38fdcad772ac402c8921f732d6400160\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4064b39acb4442a5a3dabb7461009629 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7eb54697707042b8bb6354bb04655699 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"59c8fe3a5604f95286d055113a281055b33826b6","revisions":{"f9d4a894fbe118a7c01c102e7fa1775e2141d0b2":{"kind":"REWORK","_number":1,"created":"2026-03-18 16:58:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/1"}}},"commit":{"parents":[{"commit":"80e13b9987d34b6ed8c41e03a21763930e904eda","subject":"Add helper to raise socket.timeout in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80e13b9987d34b6ed8c41e03a21763930e904eda"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-18 16:51:27.000000000","tz":60},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9d4a894fbe118a7c01c102e7fa1775e2141d0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9d4a894fbe118a7c01c102e7fa1775e2141d0b2"}]},"branch":"refs/heads/feature/threaded"},"d52040da50768fd0bd1eb0f6022c817cf2f76af6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-23 12:12:35.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/2"}}},"commit":{"parents":[{"commit":"10c6d4a6bc6a6ad3da84238ed6e63091b78bf56d","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10c6d4a6bc6a6ad3da84238ed6e63091b78bf56d"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:55.000000000","tz":60},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d52040da50768fd0bd1eb0f6022c817cf2f76af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d52040da50768fd0bd1eb0f6022c817cf2f76af6"}]},"branch":"refs/heads/feature/threaded"},"1ae7c4cd48a5c51939acc68abc61e29f8acdea3a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2026-05-30 16:41:57.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/3"}}},"commit":{"parents":[{"commit":"36fe7dd95176838f7cc65acb9bed49f015684901","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36fe7dd95176838f7cc65acb9bed49f015684901"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@nvidia.com","date":"2026-05-30 13:47:19.000000000","tz":0},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ae7c4cd48a5c51939acc68abc61e29f8acdea3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ae7c4cd48a5c51939acc68abc61e29f8acdea3a"}]},"branch":"refs/heads/feature/threaded"},"dd9a24d7694590ca9a1101672782f76d633f909d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/4"}}},"commit":{"parents":[{"commit":"1c2ac2c74705f648616834b2e8fff884a0645ac4","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c2ac2c74705f648616834b2e8fff884a0645ac4"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:08.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9a24d7694590ca9a1101672782f76d633f909d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9a24d7694590ca9a1101672782f76d633f909d"}]},"branch":"refs/heads/feature/threaded"},"525f2c9dadfa29bd1ad439d0b544969c7b5eea8f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/5"}}},"commit":{"parents":[{"commit":"0c27d83db6afb55d87da1d7529ad0c737ed1a11c","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c27d83db6afb55d87da1d7529ad0c737ed1a11c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/525f2c9dadfa29bd1ad439d0b544969c7b5eea8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/525f2c9dadfa29bd1ad439d0b544969c7b5eea8f"}]},"branch":"refs/heads/feature/threaded"},"126c92b134e0c7e84b34182aa2c79e21ba1a9f24":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-02 13:27:47.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/6"}}},"commit":{"parents":[{"commit":"32aa3adb37065d5c66b7ccc810d15e558b974d45","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32aa3adb37065d5c66b7ccc810d15e558b974d45"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-02 13:06:36.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/126c92b134e0c7e84b34182aa2c79e21ba1a9f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/126c92b134e0c7e84b34182aa2c79e21ba1a9f24"}]},"branch":"refs/heads/feature/threaded"},"c39597feafdf430365eb9a89e3f0592534a50eb8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/7"}}},"commit":{"parents":[{"commit":"85f19120ca80b674ea2472acb8f077998df58895","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85f19120ca80b674ea2472acb8f077998df58895"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:18:25.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c39597feafdf430365eb9a89e3f0592534a50eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c39597feafdf430365eb9a89e3f0592534a50eb8"}]},"branch":"refs/heads/feature/threaded"},"62a260600d38212c7f0c01945466aa6b51f4f767":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/8"}}},"commit":{"parents":[{"commit":"5aba14d9bee5586e7f21ce83e7c69a596435c84c","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aba14d9bee5586e7f21ce83e7c69a596435c84c"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 16:24:20.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a260600d38212c7f0c01945466aa6b51f4f767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a260600d38212c7f0c01945466aa6b51f4f767"}]},"branch":"refs/heads/feature/threaded"},"cb25346bf7bbbb5b2f1cc069c811b990cdc516a2":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/9"}}},"commit":{"parents":[{"commit":"93bf383676d383eabebfbe9d48d9f4a49c8f43fa","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93bf383676d383eabebfbe9d48d9f4a49c8f43fa"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:46.000000000","tz":0},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb25346bf7bbbb5b2f1cc069c811b990cdc516a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb25346bf7bbbb5b2f1cc069c811b990cdc516a2"}]},"branch":"refs/heads/feature/threaded"},"e9f2107d8af6a8fb65895ad8198f186b48d3fe75":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-07-07 17:22:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/10"}}},"commit":{"parents":[{"commit":"bc608db50a9a90be5c215110f201ff11e0cd75cf","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc608db50a9a90be5c215110f201ff11e0cd75cf"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f2107d8af6a8fb65895ad8198f186b48d3fe75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f2107d8af6a8fb65895ad8198f186b48d3fe75"}]},"branch":"refs/heads/feature/threaded"},"9d22640976d9ac48ed699a7a84a8bffe79f36092":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-11 07:19:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/11"}}},"commit":{"parents":[{"commit":"8425211cc8e3dbb41da71103011fe8219ed0d72b","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8425211cc8e3dbb41da71103011fe8219ed0d72b"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d22640976d9ac48ed699a7a84a8bffe79f36092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d22640976d9ac48ed699a7a84a8bffe79f36092"}]},"branch":"refs/heads/feature/threaded"},"ef9b40ad56dc4754dac1f0b71198c96a853fae16":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/12"}}},"commit":{"parents":[{"commit":"de7a44c6faefd94fff6a2a0465c8fcb295d71e22","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de7a44c6faefd94fff6a2a0465c8fcb295d71e22"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:19.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef9b40ad56dc4754dac1f0b71198c96a853fae16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef9b40ad56dc4754dac1f0b71198c96a853fae16"}]},"branch":"refs/heads/feature/threaded"},"54fe934017650cb9dce47f2029aa89813b45050c":{"kind":"NO_CHANGE","_number":13,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/13"}}},"commit":{"parents":[{"commit":"58bad97a1b95fb51967ef6ad2b4378d837c53f09","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58bad97a1b95fb51967ef6ad2b4378d837c53f09"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54fe934017650cb9dce47f2029aa89813b45050c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54fe934017650cb9dce47f2029aa89813b45050c"}]},"branch":"refs/heads/feature/threaded"},"3515acac9e86ab1913eebb092c1215bcaf10c199":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-08-21 06:02:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/14"}}},"commit":{"parents":[{"commit":"d310f83fbe9c37fd3ca84592dc60e04324f94f79","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d310f83fbe9c37fd3ca84592dc60e04324f94f79"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:10.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3515acac9e86ab1913eebb092c1215bcaf10c199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3515acac9e86ab1913eebb092c1215bcaf10c199"}]},"branch":"refs/heads/feature/threaded"},"59c8fe3a5604f95286d055113a281055b33826b6":{"kind":"NO_CHANGE","_number":15,"created":"2026-08-21 07:26:39.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/59/981159/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/981159/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/981159/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/981159/15"}}},"commit":{"parents":[{"commit":"d310f83fbe9c37fd3ca84592dc60e04324f94f79","subject":"Make object replicator and reconstructor thread safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d310f83fbe9c37fd3ca84592dc60e04324f94f79"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-17 08:48:04.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-21 06:51:54.000000000","tz":120},"subject":"Add gettimeout/settimeout to mock memcache sockets","message":"Add gettimeout/settimeout to mock memcache sockets\n\nWithout eventlet, Timeout sets real socket timeouts via\nsocket.settimeout() instead of greenthread cancellation. MockMemcached\nand ExplodingMockMemcached are used as both file-like objects and\nsockets in tests, but lacked gettimeout/settimeout methods, causing\nAttributeError when Timeout calls gettimeout() on them.\n\nChange-Id: I7118c0e374486f33c399575d88e7e88573d1dfd3\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59c8fe3a5604f95286d055113a281055b33826b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59c8fe3a5604f95286d055113a281055b33826b6"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
