)]}'
{"id":"openstack%2Fswift~981400","triplet_id":"openstack%2Fswift~master~I18206720b1fcffe63e6c0e968ec7de5cfd7786ed","project":"openstack/swift","branch":"master","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-08-12 20:22:03.000000000","reason":"Change was marked ready for review"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-19 00:49:44.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-08-12 20:22:03.000000000","reason":"Change was marked ready for review"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-26 01:39:03.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38767":{"account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"last_update":"2026-08-12 20:22:03.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2026-07-13 13:34:09.000000000","reason":"Change was marked work in progress"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2026-08-25 20:25:32.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}}},"hashtags":[],"change_id":"I18206720b1fcffe63e6c0e968ec7de5cfd7786ed","subject":"add prom metrics to dispersion report","status":"NEW","created":"2026-03-19 19:17:24.000000000","updated":"2026-08-26 03:01:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":123,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"aa407ada250f4612b5200f0363c54afdbee7205c","_number":981400,"virtual_id_number":981400,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-26 03:01:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-19 20:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-19 20:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-06 21:29:15.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2026-05-18 19:09:31.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-05-18 22:01:50.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"},{"updated":"2026-05-21 02:05:10.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2026-05-21 02:05:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-05-27 06:36:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-08-17 09:56:56.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-08-20 16:36:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"24d18dd0bb1fe9da671777711e2fac417a74f768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-03-19 19:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98ff702c6b794ebfe453d4191e15e53047922031","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 20:20:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c385b2049a734451a46587a28c28fedf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bdbbcb9f17374817aeafb425ba0cf40c : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fc980a33f6320044e9fbb875ccc92a9f46d6c371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 20:42:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df7d75b8994e4cdba630e19ba9aab7c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a4ec18dcf3b4f80b68d0e121864f1fd : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4aabe4b5cbc249f8a7b9c51082743ffa : SUCCESS in 1h 04m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28a26ef13fe94e6e87886932ecdd0460 : SUCCESS in 43m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b26128ff82f7472f97c4c893946553d3 : SUCCESS in 1h 02m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd48bbbe776040bebfee12e927077091 : SUCCESS in 1h 05m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5490af524e124fd681f3fa6511a704ec : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6315d29db36f4d1dae05de8eb5444734 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7ad38bed55c47b2b946bbb6c5bdfbae : SUCCESS in 21m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/181c716b4333400992cb44bf92f972d1 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/02c9304c915849c9a7046287dff5ea10 : SUCCESS in 16m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2fbd75cecc50493b86c549a32904b9d1 : SUCCESS in 19m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dac567f383354bcd97f640336c813ae1 : SUCCESS in 21m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a03360c03b074d40906e74c0c71f6623 : SUCCESS in 9m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/afa9862370a54565b7ef35861db2570d : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d0cdeb18b6b416e8e0ce9be7f6d29e8 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c02d103c19b4e9bbfe7eae85f30c573 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2efd8a1d07114602aa03b602dceda2b8 : SUCCESS in 6m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51a79dc892fa48d1bffb80a83a1cb010 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/89dec02591ab4d7fa9b733a99fae251a : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a238a870e52f4ff4af3de94473c428b6 : SUCCESS in 56m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/173e0217c27e4227a2f721843a9b5fc3 : SUCCESS in 57m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/899434950b6d4fcd9e509d5e844f7c29 : SUCCESS in 38m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4a2e33adb6740f59022c881e734ce7e : SUCCESS in 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4646d561610b427795d2bdebb35653ef : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38576f33f2094cfdb1ff8c03215d48d2 : SUCCESS in 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67b65d9795a3419e9ebb11c0a7a0a79f : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bfd0682e13913d1063a10c6b146f3e4b8d1563cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-03-24 16:06:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"364505d0b86787db866bd833e1548b2d908790a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 17:23:51.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52792ff1964f409f8c7773d3e1446e7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05dd9315a254b618e203a08aa0ee479 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f26c5a9c645044b294357060a3eebb35 : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31c12da5d35d4bb1a7547c2e80756e84 : SUCCESS in 30m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3a61e1019c14bbdb8541b1315780212 : SUCCESS in 1h 07m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b4546dd01de47fcb7453014626e8df3 : SUCCESS in 1h 05m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72a7159e54b44291b3560b1121cc5f79 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d39b9cc2ef1f4aae9aac29af5fa05c9d : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a1cc58fdb10488badbf7f4ec382da76 : SUCCESS in 11m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25ca8c0e385b4cd59618a5831e1f246d : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a2372e8a698a41178108e187f539d511 : FAILURE in 22m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/652bc0bb39294b4ca5115b230c0af6d9 : FAILURE in 24m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/34716610e2e44455ab21104a709d7fa1 : FAILURE in 32m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ed53aae5313b44d996572915d069cdb5 : FAILURE in 11m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8ae9d4cd1d314f4cbb764c53b4155d6d : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e03154821ba4d64ac82b08fdbe36d62 : SUCCESS in 16m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6b88e396368d447e9d1bea8beb7675b9 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de8fa9551d9848d4a04ea781fcd1726e : SUCCESS in 7m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e9b1ca436b5940b6b3caef9a86cc09e6 : SUCCESS in 8m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/96491818d7c14e93ba63627018a22350 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b537c964a52040509605ce142e29add4 : SUCCESS in 59m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d813f837ac5047eea67bb3eb0b47b9d5 : SUCCESS in 23m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aeb8b1fca5d343a99187d4f1f6c30fdf : SUCCESS in 53m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e461a1ac7ece4b58810bf0bfd8054e45 : FAILURE in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afb673ff6fa4ca09a9dd098e4b9100f : FAILURE in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25d8d1e2c44646859a7cdc587f00d843 : SUCCESS in 9m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/186460970d7c41a3a46eb2e89ea58228 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3c0d8e1e8bd91a65c7b2ac0401f982fb2831b686","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 17:35:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd149acbce254cb281832f37171c78bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a4e581a9fe045359c6936b9f1664e68 : FAILURE in 30m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d5c5c8b5c213d5c61d91dd3bb5cadf4d5286b18b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-03-31 16:11:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/utils/prom_metrics.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"942ca7a40a70218cea6be8b6cb0faa2784178dec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:11:42.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56755f209b154e308f7453ab44918ff9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ffe50a5ed87e4cbd93ded3a33575822f : FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":3},{"id":"a438ad864995e4c4ebea82d1d85ba2a73f9748df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:27:46.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffc0846ee0554b8bb0a8422968371716\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781918b13910483a8ead5bbedea4ea72 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/fda938c0aab64f8d9311bdde256da6ae : SUCCESS in 59m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb1e26b88120401da02e5a1daaf7bcb2 : SUCCESS in 1h 07m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a13edb77fe24b679cc6ad0560a10fa1 : SUCCESS in 35m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da2695a031d94340abac2ac6e3b1308d : SUCCESS in 45m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/416785c94c1148a6b5503effaace38b6 : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1cce417b7b1421c96b4be95746b7176 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c5d32181d0f46a08113d4c086089f6e : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3adfb14758f34ac485532f55478d07b5 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bee298115a364e9f970f81e06b794945 : FAILURE in 8m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/956b799dda0348e6832af44d33dd46c5 : FAILURE in 4m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8ed244789ae0405c8e1895cdc3945661 : FAILURE in 7m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/77db99374cd3413ca66d2d4ff622a749 : FAILURE in 8m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/81ed5e5681ae48b2b06a8609d837619c : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e66eba172d5e4f2284097e652d949f4f : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/32ddc6d38ca34d6895f02a58f292ec09 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5f006df3a7d46df9dbf2c3444837ca0 : SUCCESS in 7m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e791e14c233b40f087d59d3f9901e269 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/378b7ade299249289bbedd115876fb47 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3632ca3fa75b46cfa900e1fabd9ce6e8 : SUCCESS in 58m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e34ab0380704422a334defc11f2e82d : SUCCESS in 48m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b45d2d3fd2b24922a4fe4aa967b3a721 : SUCCESS in 53m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b98674cd8c442c483d7823af9fd00f5 : FAILURE in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7927ba038e9a469aa443acaa10bf503e : FAILURE in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/526aa71c7c5a4bf1afa177cb3f22d388 : SUCCESS in 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e98616667b3741eea6d27a614912ca96 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":3},{"id":"276b94fb39c043c1eb53ffa8120ba4ad386ce66d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-06 19:20:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9ca4f24f3ccc6a2fa1deb24329ea7c95e24b103f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 19:30:02.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bedc659044f45b9b91f3ab93e65629e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17d8950bc8f24132ab1bb7a7dc5dbba3 : FAILURE in 8m 07s","accounts_in_message":[],"_revision_number":4},{"id":"3585a863e02c175be559ccb61354aa37724baeba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-06 19:42:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2854baee9ff7131ad8f2ad3be73deda140c7b6ba","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-06 19:46:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"d7c41630d6697e4539cbec1e56409a924fd5304d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 20:09:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b656457958ec48af8371d7f2087d5baf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d90604c2bcef440ab38725bd25810286 : SUCCESS in 24m 50s","accounts_in_message":[],"_revision_number":5},{"id":"dad5c916f7ec647283c81468a47941442ea8548d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 20:52:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ad6a60717e64327a3422c6eea20d207\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35cb40ab23cf4c289b7703b34e45bf21 : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/3af1901e20c84a16bc7d9b85acf81052 : SUCCESS in 46m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d8ad0e4fbb334b8cb57048c8c93219fa : SUCCESS in 1h 04m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c6f54f766f54318902feb3516dece18 : SUCCESS in 58m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4aa986876e3941ff994fc9c3a615ad3d : SUCCESS in 59m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd9953b71b6646caa506cefdc4f51191 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41b17d324de84d81add03eebf99e80ac : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4286a83bd81e4fcfb6f417117f2b3cec : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a48113646ff4620929cbef5313ed4c5 : FAILURE in 4m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea1881e0238f4290b14133aeb5385d1f : SUCCESS in 26m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/73c4ba3709fc4352ac475a1295e77b0c : SUCCESS in 21m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab204ce9378a4bf8a819684b1d9b3034 : SUCCESS in 20m 18s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fbe6c9a79ec44847b82a7447f1d2331c : SUCCESS in 15m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2c5d0aa55ca54d93bd92d865e4d8bc3c : SUCCESS in 8m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/01f4e2480646495680cd89bf0f9031ef : SUCCESS in 9m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/05f33bb6a9614cb2b0f99b7d2c573af6 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4bf679810cbc44a8a4a50b0a1ed92229 : SUCCESS in 8m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e60635846c04a519d8a1d48b13a3c17 : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/76f8b1a11b6b43ab961c87efc12befc8 : SUCCESS in 10m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/62c9dda25f9a4b76a47655029a581b38 : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/515fea9e3cf440bb976119d5e98b3c98 : SUCCESS in 59m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b555c010e6a54c5aa8831a8954a70910 : SUCCESS in 57m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f7be80fd8054f65a4fec90474406358 : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc02aa1a80ba47b79220cba130e2bc6b : FAILURE in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db49dcf3d1e24694a2445b08dd7f1f2a : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5447bda9be14a7285b4a6740b1b1960 : SUCCESS in 57m 35s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_prom_metrics.py","accounts_in_message":[],"_revision_number":5},{"id":"ce9eeddaa498d89c519713265979038442e6193e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-04-06 21:29:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_35790\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"_revision_number":6},{"id":"b8821ae4a380ecf769adadd49fde62400ecc0c57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 22:00:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1857a7803dbe48279ee4698dbfff9a0e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84174c7b4fe34755b1ad1f2149003529 : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":6},{"id":"9d59fba6e9d73be2d14dce0ff7b1a37d8f312c77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 22:44:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b143498e3b3540d3936df5658192cbe6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ed25b23dc24e59bf481e98ca008004 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/06edc5302fff43678c03db0b5185dcad : SUCCESS in 1h 03m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52ae3ffe676c4510ac1d7a7e5de33a2d : SUCCESS in 1h 08m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eec746132b364e5da581b4bfe4f7e794 : SUCCESS in 39m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42325ddcb6e24d0e89f3ac8e80585ae3 : SUCCESS in 1h 00m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2069099841d490ea220212b612d0484 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28deb2db6dbc4515b58411dbecb3a11f : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebe6212d55d74d729e316d35bb80434f : SUCCESS in 16m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2e2eee10e414c6b891599026d32f165 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7760c7bb1ff4ba3b9d77aeb761ecfc6 : SUCCESS in 24m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b7eeb62120594191968988bc531d9cac : SUCCESS in 18m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c4ab412f4b1c45188671a02357110eec : SUCCESS in 29m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/37256ca9f9984562b07b2c062375007e : SUCCESS in 20m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/41c65ac3c11a443692579571e25bdb66 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/492bcc596c91488091664793d0f08b92 : SUCCESS in 7m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e12fe3a2557645bd86744c097fdf5d0d : SUCCESS in 12m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6897a279a7af4b009469b82108100dc2 : SUCCESS in 5m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49795818c29749608b547c72d93de9e8 : SUCCESS in 12m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/17a38d56334d47999ac03c139ff45846 : SUCCESS in 11m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/13e0df9481c044a1a491ad19bc127155 : SUCCESS in 48m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc3444b3a61e45d788b270c206b838e3 : SUCCESS in 56m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/86903f0a5dcd40d2991762ddda556d3d : SUCCESS in 25m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/613b20da3c5e4744b2f156a629892bad : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f22b36cbf0d4ecd9a2f5706386e54e7 : FAILURE in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5ec73f242d54b44b4d35e372aac9cdb : SUCCESS in 15m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1e5f3ca675f446891aa911d2d9cd87a : SUCCESS in 1h 10m 06s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_prom_metrics.py","accounts_in_message":[],"_revision_number":6},{"id":"78d1209f83c6614c96405771af30e1f7613017de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-08 15:20:47.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f6cc4e392986d7a63d11e6dfaaeb8bf922707905","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:46:04.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e88904a5de49dcb5f97236af8390f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85cf80fdb9184ab791de9c3e29842c26 : FAILURE in 41m 42s","accounts_in_message":[],"_revision_number":7},{"id":"19e40db99f8cfd402cf85ddc81a068881a5c057b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:46:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/476a47c9312545f48373c9a88ccf6f85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ae97d261c442bd8b23085d6d5b533c : SUCCESS in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/d9da9b85225a4788893b36b8b3fc38e7 : SUCCESS in 38m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1056673c28de48a680228b15aa47cd4a : SUCCESS in 1h 17m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8abf4f7295ae439ba42a2ff03b2789ea : SUCCESS in 34m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45eabb19efc04971ab02cabe4866bfc5 : SUCCESS in 54m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03d0b876f1a54c72a1f3f23d7b2ca82f : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89c47d5928514a6a8f2b02c58cc0266d : SUCCESS in 12m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1154a4862714ff6b6635cedf83f2224 : SUCCESS in 11m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50c3fc592ecc41e182449f74c77eeda5 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2ffec2aa8a4486582c7430b9c5d1240 : FAILURE in 25m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dce40806ca004fa6b072ed717675bb7e : FAILURE in 16m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/17112987ffed46a1b2a1d7644d172330 : FAILURE in 32m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a8c3ea00c79446beb2be920dd83dc54e : FAILURE in 19m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e0f67e91bfe14e0d87779cac65529711 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/285bb89c2dfe4a2eb4f73aa391cde1b7 : SUCCESS in 8m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/291023a23f944e13ae5b4aa4c7ec9cf4 : SUCCESS in 21m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5dda105735cb48e58657325e5a2e53b8 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02b1dc7e1a854a41913ca0c85d49ff6c : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/499f4ab1c81c4f92b4862974cd0f2d1e : SUCCESS in 9m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/35e6f50a0c854d689b28f95cff907a71 : SUCCESS in 1h 04m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0331b49250b54b97a27e2abf1cad8c6a : SUCCESS in 24m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3f7fd4e5b204bed952e404621d413ea : SUCCESS in 53m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e27ffafd1d624b06ab7d294846a9e7dc : FAILURE in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef5b58f688b45d38e6adf0b955b051c : FAILURE in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51ae4ad2bfd04eac91aff20efde12189 : SUCCESS in 16m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24949a23b1854b3a9d774bcaf6f4fa66 : SUCCESS in 56m 10s\n\nWarning:\n  Comments left for invalid file swift/common/prom_metrics.py\n  Comments left for invalid file test/unit/common/test_prom_metrics.py","accounts_in_message":[],"_revision_number":7},{"id":"5c7e372fe71302cf92409c7f7c8c6effe6febe07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-09 14:21:08.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a7f578cb5ba3ac5c49434f241b3ec1c01dd23ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 15:37:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89b40c7c66e442cea42f8c180ddd0e0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7969f7a343a4460bc33c322e3fd7e06 : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/3e5a642933764f3a9c1a52f50752e56e : SUCCESS in 59m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/222f57dcafab42f7a419b14ad0355bc2 : SUCCESS in 32m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b34478d2472f4a6ab45ddea0def63dba : SUCCESS in 48m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adca56ef57254093a28960acb55f9620 : SUCCESS in 59m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a136813965af4a728ebb7e42c7609ca8 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c5df2fd83e641a8b388690fdd5ed39e : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6f6c61f4e7444b78a7f0b77bba13a46 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5361b9aa3ef148a2ba4f3b6d4ca8d197 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dff4293cd64c4688b228d13d37267dee : FAILURE in 17m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ec9466152ae2415c8d7b0b84c83f9aca : FAILURE in 21m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a1827f17caf944d2a63c6f7d59c72f1e : FAILURE in 19m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cc84867814094bb993a24c8889eb531c : FAILURE in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b2acce4dd9a433ab22d87bd8019b647 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1a84835b46a1417faed94caa884e72d1 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b91f1909893c4ee1a19fda45ed521fbc : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dfed94b685fd4f8e8e43d56947311253 : SUCCESS in 6m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7831407371664b74b392b17073edd56e : SUCCESS in 12m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/26d9f4c0c3a749bb9d233227c70c3396 : SUCCESS in 8m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e51e1ffed6224284af0008b30d13481d : SUCCESS in 59m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2f2067ec1de4922a0f528abf6e74ae5 : SUCCESS in 31m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97a0be0e108e4e9985fd2b7314f1f1e4 : SUCCESS in 47m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fec6149c93944a0bbff1fb6ed21c08c : FAILURE in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f68dca12ea4494b52e8e7a1ed5c55e : FAILURE in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/220ec900dfe0489da525dcdf97aef129 : SUCCESS in 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc466059b0e9464696cf7de660f97eee : SUCCESS in 1h 06m 23s\n\nWarning:\n  Comments left for invalid file swift/common/prom_metrics.py\n  Comments left for invalid file test/unit/common/test_prom_metrics.py","accounts_in_message":[],"_revision_number":8},{"id":"6fd12f9bc05947bdb1bb373487d079e1de6db0b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 16:49:55.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77337e1bd7ae4a2aaee8408cef5a9cba\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07550c7695fc455dadfec3c92cfac1c0 : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":8},{"id":"50ce50323cecd9d71e3c1a55d940094619a73b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-09 16:50:26.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"90e5cb9a62e6e0023a45d7271b4fbb375d980602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 18:08:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/027d1f207df946fbaae9040ea9447e23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b02cc69b254a3f822e7b27d78602bd : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/02429c110c764107a2b1d9bc730c3aaf : SUCCESS in 1h 10m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa87a792f3214536ba711bbe2cc06f6c : SUCCESS in 55m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3c407bc38ec478fb5f704094994e8cc : SUCCESS in 32m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f444ff4cbcb4e39ae04089ad9134fdb : SUCCESS in 50m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa56ce452ccf44b6bd52753e9805a5a9 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/add865959f2d49c4ae63500efc93b43f : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eaa542df636945eb908724981bb38ecf : SUCCESS in 22m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/459ea60aa6f140feb82b7893d3cba48b : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/460e2e7bdea7469d8f93ac03553b8536 : FAILURE in 10m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9c16953618ba4a42a9d33ece77b8f3c7 : FAILURE in 20m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6ee4ba9747e141cc8490bb7ec13ed405 : FAILURE in 11m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b7f0194722f84fa4858e3372e1d26d78 : FAILURE in 18m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb1d99177870482e9a16f72ab10c74e7 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9968cd6559b147acb16f2a249d4a9e82 : SUCCESS in 8m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0d0a3ae727a9404a8ea9b4861687e19b : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cdd4b77e3c6e42c2a287ae462d75a3d3 : SUCCESS in 6m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a85fa217e7434d558fa10d357e1f4da4 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3b926bf95e842a5a259c0b0380081dd : SUCCESS in 9m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/afe7ea23a68247c9a8cb690d06ba35a0 : SUCCESS in 1h 00m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02e7e621392e4ee4a3acf2bc6d8b593f : SUCCESS in 28m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ce289bb31ca45d1ba9a63742307b2b3 : SUCCESS in 53m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/248de9ea639d4a0bb4a5c7449ea9b9ce : FAILURE in 26m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c83ffa54f3416e9e31bd81a39e0d47 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e624d34aa74c4268956969c23269e7de : SUCCESS in 18m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a9afb3544e549ca8148083f58164f60 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":9},{"id":"75def122acb12de50f850f736bde7b5f333b2e75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:30:01.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bdbb2381f1042cc8f4be469caac8490\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8572144825c4656993a397037316326 : FAILURE in 30m 43s","accounts_in_message":[],"_revision_number":9},{"id":"d5bdf0d30caa088ec6292092e5c83680980ad59a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-04-14 17:41:23.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"b9f4bb404bab44225779efba9a4774ed9a7348da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 18:50:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30f8f07939e04cb283510f654c6c8328\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92ded78e5384553852d3bd57bbbf24a : SUCCESS in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/61cb18fb00bc4439b749ebb0fd6796f9 : SUCCESS in 51m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84fa9b6c522d4db1aee9991f2b430680 : SUCCESS in 1h 03m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db842b942d3545659c7adbbdf26d376f : SUCCESS in 36m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8889be322fc24fcdac3a7292c25e1db9 : SUCCESS in 53m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39c2c9ecfa3842f485bbe9dc8d8f7a3a : SUCCESS in 10m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c86ac52fe8494990a50c4fcbb4b9718e : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/989e25dee8ac4b92ac4b00f38140ab62 : SUCCESS in 10m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21b61564338d4d548ed159c5cf4f26eb : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/03feb2d9fe634bf982034c968cd13a6c : FAILURE in 20m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3761d2ddcbd34f5aabd5fbba6fc59f77 : FAILURE in 16m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/38cf5c20d87644a4aaa4d765f6bd746f : FAILURE in 30m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f8d1595608f34a79809cede03145daea : FAILURE in 21m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/142ef6ee7fbe44d79449912b40a1192b : SUCCESS in 8m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e5ed4848f534cbabb744f545c90edf2 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e1bb235e9dd8473dac7ea864539025cb : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20457b436005473aae379483ec1cb2df : SUCCESS in 7m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a75b218978e43b6916b14c0e4b4facb : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0506b6bd55da4eff897c06c84ae9567e : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e83dc961f67d480d9855a59226bd1c24 : SUCCESS in 1h 00m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0863efa4f6f40bfabeb9f3f76a07689 : SUCCESS in 55m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00eeae40058c40c98269643bb40e782e : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16915723a217490a8f5eb2ad2140e4ce : FAILURE in 29m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f451ba053d4fe5bf3dba51576c0be2 : FAILURE in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8043375913d54f1d8e0187759acd42c4 : FAILURE in 27m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62f3e3263fc343d19aa9e491b1e6ad68 : SUCCESS in 33m 33s\n\nWarning:\n  Comments left for invalid file swift/common/prom_metrics.py","accounts_in_message":[],"_revision_number":10},{"id":"0709203cd0123a5bea5ce5c9feee886d1e01fb7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 23:55:48.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008a2a2178424bd9ae88eed6b7fa6335\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8f23b31b2dd44f491b25e7b023eceec : FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":10},{"id":"088468e7f8c4cf4c362b1736279d73f06bac9439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-07 21:59:38.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"03f44c9cb12c068269a8c9cdd54a223c2539919c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 22:51:48.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cbbfdd29bed44e6b91030a727646b38\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8961d6d7e63042acab25efa5b274511a : FAILURE in 50m 35s","accounts_in_message":[],"_revision_number":11},{"id":"96f24577507b7be4cb0374fcdc13b58fdd6c80ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 23:12:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52d601f2e00448b49b8117f755fe48fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba02648a509f4d82a3c5edb4ae9c55f8 : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/c2d2114b962c4076a5ea1d9c3270308a : SUCCESS in 1h 00m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/744adc318eee4dbeabd8b1f8b75dbd04 : SUCCESS in 1h 05m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fedbae50c2364ca883ed421e7cf8d2da : SUCCESS in 35m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b77f61d689649e496cf112860282939 : SUCCESS in 1h 06m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfc309cb8a784455b43b0c77368ad141 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0b1f91290ad424598fa27e96e7c92e1 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ac68c283ff247eaa0bc1c58ac10c2e4 : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b25e3a54e95412b8312205b1f87a625 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/554995d6344e4ce292b684c36a0e67e6 : FAILURE in 25m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c2dedda753fd47b29aca61665ab465c0 : FAILURE in 10m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d2ca4d0f96704d0e83108e4296b492b9 : FAILURE in 31m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f79b46346fa74088b5b5aa15075b2eae : FAILURE in 23m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/66e5949410594d8eb51fa90d35f0966c : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d802ab77d7f4107bae6cfe70349bded : SUCCESS in 30m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/40b98978349a4d2aa3569023bd2c3485 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/935ff8c8298b4fd2a2c286cd26c57bcb : SUCCESS in 4m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51002f59e77246e39d6cffac93e0d6fe : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/34a9fd9b787d43b7b3f774d1bdd4d120 : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/773405b665224c6e80bcdd132b361cc4 : SUCCESS in 56m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a171c4481e81480fb254116a96cd3fcb : SUCCESS in 50m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa0cbdf8573e47929154676e8d38b4b9 : SUCCESS in 35m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89da0e9fcbec4abc9b3638c5f753d429 : FAILURE in 29m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9345bf97f7c048b494932a30d662eacd : SUCCESS in 2m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d445d927e5a4b93b275a2dbd5dff3b1 : SUCCESS in 19m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c02035a1d764fe088ce8333efc6e0d0 : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":11},{"id":"7348fbf14a58ddb0972e69cac7bebdcf34b493c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-15 12:19:27.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"2cb4b1ec10e930d1a0b668abeba386443adb9971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:43:51.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65208d07ac6c496e8860a9af445b0418\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5cd66d8fa0e4274ba10374e9ee43a6a : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ff4abbf6dc140deaa66effa04b0ec90 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/79b71696d50449c8b227b2caa77c74db : SUCCESS in 1h 11m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96cf5e4d2904485d9e88cc251c4e6221 : SUCCESS in 30m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f236977d7e740f28b6cb0c4fb3b1345 : SUCCESS in 43m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f2283d7e540436991355ef6e43fb82c : SUCCESS in 55m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0e142055c6b47beae823c99d22e349f : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8566eb5479fd4695a530f43f90c82271 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0000a899681449f8b362a9cfac2d21bb : SUCCESS in 14m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1281bcb7dea4c799addf916d967b4ae : SUCCESS in 4m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0673473a895c4c2ba2b0b03d5b8b6c85 : SUCCESS in 16m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/010688a051c3412381b7ec0406881ab5 : SUCCESS in 9m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/38a25e1d73cf40c09d74f16f09e0ab2d : SUCCESS in 31m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/07bd1e2437d1403b9e41a7ee084e70d1 : SUCCESS in 24m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/baafae45830043c8a0e4edf3f3fd7bc6 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff41ca28a8d34e6798ba7115d07cfc79 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9f92b666f7e84d22965502ee39cb74b7 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1bd045a23e524ad4acbd3e902d13681f : SUCCESS in 5m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b61c298e1b54c1a9e0e0256561f765a : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/545f157ae9b445df80cea9c011cbb759 : SUCCESS in 8m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fec1a2fd38524de8a9f4d012924cad82 : SUCCESS in 1h 07m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b73be27b04541d3bf3975d1b66d92f9 : SUCCESS in 54m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/857e855f26ab407f9da177a0020f0ebc : SUCCESS in 1h 03m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c20549e51eb64a4c920ab72e02536814 : SUCCESS in 28m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7320807483a94c42951bb162cd6e3f14 : FAILURE in 2m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4733e52d46342b091f39ff084479b52 : SUCCESS in 18m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06f5d3aaba154fa7947075256f9a1cb4 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":12},{"id":"d5831c6edebd6efb143e3d1a54488910e96b378e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 14:48:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45f987e629643beb63026b528bdb349\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c942d1f86bc4bb6876d202354540b5b : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":12},{"id":"a4eee725acab15cb9ea084a6e357330598770448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-15 18:29:52.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"1e80ff2ca28a195e0e84ddd11804c53175ee7718","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 19:47:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e78047d3b8446495b6ead37efd7e18\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2895561a21e942e7b6cdf52b5843cd4d : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":13},{"id":"66164fee3034f9eda22cecbb12bd4d13c59b0c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 19:50:55.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8997ab32cbec4bdb8f69b4b8fc629783\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06757462aca94cfdbfd150080cabe8d4 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed885cc076354f4784e5939061fa3e3b : SUCCESS in 5m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b8362b4de21a4e15b92808b127bf2a1e : SUCCESS in 1h 14m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3525028653342e396f1de60e115bc7f : SUCCESS in 1h 07m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d23fe2a89d045a8b21bdf8994b9189b : SUCCESS in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba579ec68889406798a57f84241f6236 : SUCCESS in 1h 00m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/503db93171de4acb9974601d2978c300 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbb0a7f9add2476a9487b5107c2920f3 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b54433885f1e4004882d0c69465d93e5 : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5bcf35c2e187404c8265713d43e105f0 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea400177ba1840c2a04418b5099f0cd2 : SUCCESS in 24m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fc6db3a5e74c4a4893acca2443021c7e : SUCCESS in 11m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/68214997d40542059b916c6c1fac6d58 : SUCCESS in 29m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/eec78108e0d94b488e4fac4683298091 : SUCCESS in 20m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2fd6bdca80514fb7bb4ebddbc9d87251 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f20eca77a88b4536b5e182f41635a85f : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/09958392543a44ea8306613a533b3524 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3327e70dc42e40de80bcf42334f2a317 : SUCCESS in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3616d2b5ef964646bea2cf7380654cec : SUCCESS in 23m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1fa64a5951d488f9a1295147512a99f : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1bdd6b3136c4d228cf751058bac4fb0 : SUCCESS in 1h 07m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67cafb89da9749e8990887eb9efb0954 : SUCCESS in 53m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd7ae0c69ba443569f7f19182f4694b4 : SUCCESS in 1h 00m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e8d040f575840f88a7a6cd14d4ba14e : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83af19ddd36e4cc7888b271b8d992185 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7e986b6df6445f89905246c542d026a : SUCCESS in 26m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77bb4df050cc4a68b389e236957e498e : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":13},{"id":"c36d34ba7ab384f3d592ac4775ecb5a642410a49","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-18 19:09:31.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ad3f89585e29271466d0cb47deeeb68215272dc2","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-18 22:01:50.000000000","message":"Patch Set 13: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6b88d06653b8e3d54930e7b9eb4bcf0bcf645627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-19 15:40:23.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"b6074b16a9c3d0c523e36b8fc4a2c984058c5ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 17:15:49.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/202f33970e3b4fb19c167d3b3e580619\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a02ea823bf74107afe935e2eb571c85 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9257d4c128384eccb288949a7f06d759 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/f4323e6944a94259a832b62d83113878 : SUCCESS in 1h 19m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ef9b07b9e164091ba3068936ddd976f : SUCCESS in 1h 08m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2f2b3b0a0424e96ba6fc40396b22bc1 : SUCCESS in 42m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e32e531d27514255b5f6b3f0f486531c : SUCCESS in 42m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/776f078edf2f48a0abb9b2edde383343 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e91a79a4faa44ca49ff733ead5126c38 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fca0b2c05b24f10b951dd12ffc40a5c : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c00abe4e2b745bca54cca5cde978fbf : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32dc7ea0de384c109785a5310804c77d : SUCCESS in 22m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3c5e28d3ac4347fb85d706d2b62c95d5 : SUCCESS in 23m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/06ac7e4d4a7c4c8db3ad1e8b56b38d08 : SUCCESS in 15m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/19dae072ee674688bb21c8d1c31029b2 : SUCCESS in 19m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f150f1faa2244fa287f32efcd0646b51 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22667e5616f642c89a27ff7509db8463 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a05385de98894a589a01c414a41a77f3 : SUCCESS in 10m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58a14e0f45d54364ad9fb0cfb505afbe : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3b48952d0e547188a8bf71a1fb46b34 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0407ebffdb9e4996b4eead0f402bf5e9 : SUCCESS in 11m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec3cb364907e448095c2ed9ab13c3456 : SUCCESS in 1h 04m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43313beedb134efeb18d46ae3907f968 : FAILURE in 30m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30f0139fa636490681225134da587e08 : FAILURE in 31m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de537bc8264845a3816467a47d2425bc : SUCCESS in 29m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e561daf05614346a0a2d6bfec87b5c7 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7989ab05643b4f2ebaa5aaa0263add69 : SUCCESS in 20m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7de0fa59ffab4f56aae18ba4ad500dab : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":14},{"id":"a62830ea2666c00c5bbbff13190503a90bcd74d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 22:09:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc49d1df93a54e51a981d82ad219a603\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce7df23455bb4d0791e692482ed5c1e0 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":14},{"id":"c6415408946bcaad0f1b663641a2b02b1a13313d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-05-20 22:49:16.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"732062e452d05f2f0f557c2c53dde4cc9e75ad23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-05-21 00:11:02.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"fb316d3f23a26cd8375a2251560188c5942a0ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 01:40:40.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae613707e1804d1ba74f5e05845a223e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589f9d7d2eea44bda40a249ba02cdeae : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed1d06a798d24bd69799927ca7a3b7a9 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c369a25577e44874b633f0e2408a38ba : SUCCESS in 1h 22m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02c15cd9e6c149bea86cdb179ab285be : SUCCESS in 1h 02m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1af080972af648c6aabdba898246e6ae : SUCCESS in 57m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e1ca416e7284a74b427a9c6a38549c1 : SUCCESS in 33m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66ca7e8468d14f1e910ef4e6266bdd84 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c8604c34d5541b5b7c7e09adca6bcf8 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7324be8b11d42a683df449f73f04c1b : SUCCESS in 22m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a236a64c4d20438886705a56eb190cc6 : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a4b88861a0445699530699e3db2baaa : SUCCESS in 10m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0fdefa88e76646f68f7937cdb6b348d5 : SUCCESS in 22m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/85e7bcac0bc54e5f8a28ba4e62e70b48 : SUCCESS in 11m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2cdb5227202a4bcfa78715d5b5bad5c6 : SUCCESS in 22m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6ad864d3caa5478988888e597751068a : SUCCESS in 8m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3d4c01a1231844d3bdbcb9fc9f9fc2e5 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d435ccbbef0143ff822e01e8809536f4 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff3e69ac0cc44806a035a0980f80befc : SUCCESS in 8m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1c0c3cca8bf484f9cb716d068a2f03d : SUCCESS in 11m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/98675d613fd14244973ba1db2affd37b : SUCCESS in 11m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67ad1c026fac41059220da859045022b : FAILURE in 1h 06m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29570539d88d44dda3eecab008eacde1 : SUCCESS in 1h 00m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1852040c0f8a4e07a771c5fb075e2aeb : SUCCESS in 34m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c457a5b6783d465ca2aba4696aa5c678 : SUCCESS in 29m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c3875c9086433ca8cbd5664f849b85 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8eb3b438f849404e8aeb7d7772c7feae : SUCCESS in 15m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b897ab1d6d1a495d9fce0e2936cc033f : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":16},{"id":"ec08556bc082dbf4c7572c5e0ff1d388a0f936c7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-21 02:05:10.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"eb69ab078ae93a96711c0bc733c8b49feacf6e74","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-21 02:05:49.000000000","message":"Patch Set 16: Code-Review-1","accounts_in_message":[],"_revision_number":16},{"id":"b8f159d07131ebd975c833ae3fe4669c1589524e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 02:05:58.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d107a097eec45af91179d383e7b7edc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29725bcd054a4eb3a01c00901de276ca : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":16},{"id":"c3e08cd5bf3119b0547abbde356f6fdedb898b44","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-21 21:49:26.000000000","message":"Patch Set 16:\n\n(13 comments)","accounts_in_message":[],"_revision_number":16},{"id":"bc704e33a02c17847d54522716345c73aad966f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-21 22:00:44.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"6bdd1c280d62e119fa3e07aa63221b324e00e992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-21 22:00:57.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"82cc3603e35bcff9f20529140bbcfdf29cc4c3ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 23:25:59.000000000","message":"Patch Set 18: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a198cd94e73475f93cd375224bac974\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614a66432f734e38af114e290e3bf5f2 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2f7cd5bd4644322b1a2433284ef8c63 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/1626ac61e4434875a1c8fc76b6ee9fc0 : SUCCESS in 1h 04m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9cd071ef0e41498c9979075fd5fd9089 : SUCCESS in 1h 06m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/998fd41eb767455a808e63aca10bb5e8 : SUCCESS in 35m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67db40a97c2249c99fa663ab43aa2bec : SUCCESS in 57m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32617c885137454a9c7bf39a1a62e64b : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95471933f804436f9f23b3d94a591196 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/975fd3b2f39e41499679474ce128d3e3 : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bfee15a744324de7abb7dc95da7fbebe : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c71340c20b24ac4805fa1ed4e7e3596 : SUCCESS in 12m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19863e84421440ddb77be1c3ed2814e1 : SUCCESS in 22m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9750d4535dce449eb09e5bed1ccb56fb : SUCCESS in 13m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/33ea5a1517784a928867620c1ee6f637 : SUCCESS in 10m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3f821a12e326408c86c03f26b38db98e : SUCCESS in 8m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/74d06bc512144e889d2c402b255d1312 : SUCCESS in 8m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c52862490b049d09fada54f067ef9be : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7849ebe9d1dc40eb87e9bceaacfaaf36 : SUCCESS in 4m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfce17447ab54febbb0dc00c4a9cac2b : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/19c0300fcd6c4fdbbdc3020a3883a5cc : SUCCESS in 5m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e42cb5377f9746559d5e202e44b2926d : SUCCESS in 1h 01m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d6cbf89ce944600a71eaecadb4b973c : SUCCESS in 25m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3040487f5c3248cf90919b0116375adb : SUCCESS in 25m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2c9cc6cf7f0490284aa2e9f82f222d8 : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f311807049b4f3e85060f152871d89d : FAILURE in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42b7767f388d4c2cb72d174e6465429d : SUCCESS in 20m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c0aeb1da2624294b374364d7a68f18d : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":18},{"id":"0263a291a48e8887e469e9591f462b99681bdc34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 00:49:45.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d07c639ae7e84b7f8a0576434099e3df\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a871159e7e048b2aa4888ad1f3681da : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":18},{"id":"62574ce3aed68fd3890911297881bd36da1f3212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-26 15:45:42.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"c09312ee99e0667263367b6824b5f60690e230d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-05-26 16:38:45.000000000","message":"Patch Set 20: Patch Set 19 was rebased on behalf of \u003cGERRIT_ACCOUNT_35790\u003e","accounts_in_message":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"_revision_number":20},{"id":"a2dd7b674536ca5c6036dbad85103326fba27ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:51:43.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cfadf73b77b4981ab8fec764079ffbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a04deb75bb4971abd49041cff05f49 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58d5aa783be419ab17e1afdb384c4d8 : SUCCESS in 3m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/14f169e5868549b188b3aa37d1c6b8c1 : SUCCESS in 36m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecd30476d88a4139bdff3064ce3f2ccd : SUCCESS in 53m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/624b4f03404f41408a6d5fce6fd8d189 : SUCCESS in 1h 00m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/353117eba76d44139d618eb549ae3586 : SUCCESS in 1h 05m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36c5551becfc446d96def106193623a3 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06a391f2e33c41eea9f731943a8a9153 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/063e602103dc4677bb3662b185e9905d : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db33fc33a2784347ae7855946f9f212b : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4d32dccb1141414196e91234f4d0b3a0 : SUCCESS in 27m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4d66de132660432c9b2e35f5f9119ec4 : FAILURE in 2m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f5c78ec296c4d60a6809c6bae2d9dac : SUCCESS in 24m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ecd759e2c34540b38dbee614c88ba466 : SUCCESS in 9m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/056c334a8762491783873bfd38e99285 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/064f6d58bd384eefa02d85d60c1cbc86 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6947293d0381459180ec7718c9af6fc9 : SUCCESS in 10m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/757c03b4b09e48d886783699ab6e1f82 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6694098a2184a469dd2d5ce38b5d4dd : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb4b84c1d6bd482d968f04f3c2e0c84e : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/654404e306ec4598b58ea9b804f7a732 : SUCCESS in 45m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6b9a67cdcc043539f18dfef9f6171a0 : SUCCESS in 22m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b82f31fc41844364a6e9f47d5df5baf5 : SUCCESS in 55m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb7a5ad4ac994be8905b1bf056bf80dc : SUCCESS in 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d5a7430b554985b3fc92d28d1a11fa : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb452bca408142f199290999df916ac4 : SUCCESS in 20m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b5e27c8b560447dacdfa40251d8c6f2 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":20},{"id":"e3850b5449abad5893b964d33ed0360fdc184551","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 23:18:27.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc3d0291a8c49148c8841f1e7c1afcd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19891948b9c6417abcacbe366069ed73 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":20},{"id":"d7553688a4dd66be2c1630cf986c5d31061ecb06","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-27 06:36:17.000000000","message":"Patch Set 20: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a6eae724401e3f84d7f9959df62dff974e78bbd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-05-28 15:13:01.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"86b98177ec3a974ebe08c8d06b2771ce69d44761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 16:28:57.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e6ed317b96840f0acd3c85a4f950157\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a787a903dd5f4156b90feba96c4d0d79 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df56360d722e4b1ea271e84c9633129a : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/71338e139293435fab83d755afd4a1bc : SUCCESS in 39m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c49873d3dc2147c48ea62c8c9b35f586 : SUCCESS in 59m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ba068094b2340ecbb988d0f0f55fa01 : SUCCESS in 57m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3105767948a4d84a19cb61944ef36f2 : SUCCESS in 59m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a067d8c87f6c48838f4e6c0a9b09965a : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33513c077e6141bd8f389f0e730fc5fc : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/598ef3f1cbe14aaf89ba4706ec64933d : SUCCESS in 15m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4242c89d671c4bf2bd5d92e48a8a8e75 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/66486104e3dd462eae0ac7ecfdaec04f : SUCCESS in 28m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7a2b1ba67a1647ebbb624010d5b890ed : SUCCESS in 22m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/91bb34025f40422d8f5d04bee490f9a9 : SUCCESS in 22m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ab00346cf8c042afbe03e34d01d0e87b : SUCCESS in 13m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff1f025f652747778c67b210e24160be : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/951ae586f2994876b3adedaf6e9565f2 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a3d2907c14754136baea41772bf55d1f : SUCCESS in 19m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6656083e93f447e28bbf11dac1f037ed : SUCCESS in 7m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e506aafa0db4ce694ec76d80221b89f : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/73dda66502654492bedcd34113692d2e : SUCCESS in 6m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/29db8c4bed754e42a4181abfdfd49037 : SUCCESS in 52m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/673a49ecc85c4d6cbdedc35ea4756c9c : SUCCESS in 51m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b9cb1d0edfe421980935ab6841807ca : SUCCESS in 52m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1125f1d1f25469baaf15d9017c4446b : SUCCESS in 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b7031b909b84173b216e1097ac5f26f : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4879bd7b84840bab6523aa9277e7792 : SUCCESS in 16m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2f7b73684db4c94b9baa1f443412f4f : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":21},{"id":"fb8a3dfdef7645fa996abe458b1b744ed9197a7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 17:16:41.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c7ce319a774f2fb3d71b341d53d8fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fa16d1bd25c494eaa30f4c2f8495c9d : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":21},{"id":"f9f47d5db97fa7e0f47b51de1371b28cf2e23a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-06-05 14:12:54.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"a19e649c40bd27ca675795b1fc53ddc0163074c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 15:41:00.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71c9d867ea2443228daa2f1644a59230\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60a03e9774774678985a9de09f5fca60 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":22},{"id":"4ab436f3f92ecf5f3d124807af3ef9a13c1bebd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 15:49:59.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab206adcf924c7594eab2e934a36997\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2f58fea0e347a0adb64113821e734d : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edfbd5512d1c48c9844f2fa9822c95ab : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/bc804a30c6984bc2a5ee3078ebb7674c : SUCCESS in 1h 01m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60f92177b6c749f49d599d7f2272719f : SUCCESS in 53m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/722a9186d6e64d6bb6d3b4cb6fc06fc8 : SUCCESS in 1h 24m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/904f658d76414d478bea675eacdc475d : SUCCESS in 53m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cdce273fafb4e69aa148f09df511da6 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69b6978aa27d4cc6be7415dba9d9946c : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7993e591920d4805ac8bed66ff8982f4 : SUCCESS in 22m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73fb48234702402aabd75cfd83b9a9dd : SUCCESS in 10m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4ff79010c6941f1ba5741f6598b5d70 : SUCCESS in 13m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4bba1bd1aa224893af9a13ca3ba38b52 : SUCCESS in 21m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c7ab5ef7111b4f54ad656b2752b0c54f : SUCCESS in 17m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/72ac05d83c374d2e85db7c4b176dc9ac : SUCCESS in 20m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/65ca57f2a55b438e8e791c6dbf24ea78 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/41483ad11de14dc5ba71111644e4bc72 : SUCCESS in 10m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/465d0c1d04154a7ab425dacae9456861 : SUCCESS in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2669f37ae175483e85138d135e11b37a : SUCCESS in 4m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0a1b9b1be644881ae702f1de0277eb8 : SUCCESS in 13m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f6823f4d9734d62aa363f7238048e20 : SUCCESS in 10m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a0989cc36c64fc5ad6212702a5132c8 : SUCCESS in 58m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a0a190796bb458189d84c4b7ddf8b43 : SUCCESS in 28m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fdbf40adc8134b879cf1b1c7849a7d45 : SUCCESS in 50m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de77283151494badb0cd59fbf33e8a3e : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ecc9fb05084c68a31c77f0994fd74c : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88532545043348f8aa413cae8d099a70 : SUCCESS in 21m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e435bb09a1be4cfa8b08de0dc57d96fe : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":22},{"id":"94564a0502b48e3ca71e81bbea393df5156850f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-06-22 14:06:48.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"ddb7043a350f80cca52ad868737a57f9b92eb9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 15:37:00.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab6575c983704f65a3c5d3007472eb4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce63608188a4d77b0a4ff09cd2a5707 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0ad5d4fd2b042938fd2f92bd2fd0660 : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5aadc96743904ff3b9d7dfb304a948d5 : SUCCESS in 1h 06m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25cafb664e3a429cb8a23edda993e9b5 : SUCCESS in 1h 20m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a4e7b60916da42ebb5ade32e42b0b8d1 : SUCCESS in 1h 02m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33e0bd7748014d7dbd33f0d382b1995b : SUCCESS in 1h 02m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7178ce3951644c94960d8971a5a89b78 : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8ddf9b602604ad682c77d5e317efbeb : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/720d1bd7507b444982c9161e181f1229 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d9b1a3792ef4dac91f622f1376c5a49 : SUCCESS in 7m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/964b36a96c45480e863d1630dbf1c9c5 : SUCCESS in 11m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b66c5a19072d466bb6a3edef6476cb24 : SUCCESS in 15m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a09f2a726c7142a5830d4b8322d3b789 : SUCCESS in 21m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e7c7cc264e964797bd6b23fb7586d418 : SUCCESS in 15m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c32d10eba1034b2a8720db0ad491ee94 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ec705cbd86643b89bc6769f7085f325 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df30309cd98e459b8448c7274e957beb : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/16b42cfc03624a7bb0c9b53663f19e91 : SUCCESS in 9m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82d578e2747542da85be42025e8dc051 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccfb379a35ea4473b658769644a66c6f : SUCCESS in 9m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/674da9b4a8f14e5bb997f790cf291301 : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bee6e4f3085e42cd918d27cb1fcc233f : SUCCESS in 59m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/820b8e9b43e949948ebd3f4fadc81e77 : SUCCESS in 57m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f826abdca15e49e9ad41c548262c8cb1 : SUCCESS in 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6001b91215943be9c746b21253e7a9a : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b106af83796f43eb827cd5defecb8efe : SUCCESS in 20m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e9d22d14e4240f3987da0c83fa259c3 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":23},{"id":"19a67ee2a6b8164d1235a0ea908147c77665d053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 16:10:06.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7363341cd15f485e840cf73bd9f6c425\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65b5ab713aa44138b8a7f5c0d0957162 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":23},{"id":"94a368618e515814a23b4300ab683ec9360516ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-07-13 13:34:09.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nThe following files contain Git conflicts:\n* test/unit/cli/test_dispersion_report.py\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"df536b10fe232e0d50e26d84e38c95707d214292","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-07-13 13:41:04.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":24},{"id":"ca8d1d9a43dec0c5df0c8303fa76c47b13e5554a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 13:48:27.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ab14d1a24c43449b9f317ca18767ef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3582c7a9450e4da0a19ff3139e87802c : FAILURE in 10m 07s","accounts_in_message":[],"_revision_number":24},{"id":"2abd6edb985b8318f8e2e9a085dd35c68ed68c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 14:45:05.000000000","message":"Patch Set 24: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc8d9d166864473eac5ee0613795db56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32d88c5a49e41418120f40bba91909f : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b727297229404df9844a9576b911ae12 : SUCCESS in 7m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/bc02ed1b67e84a67819df4a87914a307 : SUCCESS in 1h 03m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff85670a770048f2bdd22a49a8eb5c6f : SUCCESS in 1h 01m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85c65bbc38ac4c46aec52f88562f047e : SUCCESS in 45m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fe873e076e84a989619fcc7157bc42a : SUCCESS in 30m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed89a8cfc5ee4cbe880086e9c710b699 : SUCCESS in 13m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/909cf4209e2a4fd5a3803f9549c3c021 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db39f91d757f457493d7214270b4e8e4 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94f779a97d774f26ba66db14fee4cdfd : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3459c637a7d4d78a8f087fbe899850c : FAILURE in 8m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/efd66df2632247beb09177d35002b934 : FAILURE in 4m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04e4dfd5e10747509161951fd4d02b7e : FAILURE in 6m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/865badc952ba493292fdaba2f6a3b73d : FAILURE in 9m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3c93bebedb91427b94e7f15834a87a66 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7be45ba8332741989d8388a8a52243ba : SUCCESS in 13m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e4d8f95ab96445c9b9e822b3b6b04e0 : SUCCESS in 13m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eaf5bd29a90a4f6384f7ba3358346a38 : RETRY_LIMIT in 8m 22s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d99b1096c994d9eab1eadfda098b01e : SUCCESS in 9m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/afa8da3eff2a496cbff32a1d823c5a4c : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ab447eee54e4acc8cb0211b68e05842 : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5d4796ac47740119d05f573683f0b00 : SUCCESS in 48m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c115d7ef9f174018baa2dcf228d049f8 : SUCCESS in 49m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ef060c3dc2a4bc3a619abb68da01b81 : FAILURE in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1ce2aace474d4cb95bd507e9458e87 : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21ab7205384849e78c1fd5ebe4d855f8 : SUCCESS in 17m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/276817c225b645fda36b50c49488f494 : SUCCESS in 31m 40s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":24},{"id":"e0c9ccc08b7cff8a009b7c50de9a0e1d1b73f639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-07-14 19:53:12.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"035d9cae08c92464308c0030db9f6b6f74e67e59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 20:56:12.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5191167b4c54c0091d43a479ca74dc9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56f6721e20de4610a7b59c4a7620727b : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":25},{"id":"ed4e8103a55e3ab6e3cb254edb678dfb1958cf34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 21:06:48.000000000","message":"Patch Set 25: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffdd6cac53314537bc190824edd9d4b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a4b26e51544fbdab3393e554d511c9 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14eb67fe2a2c4652ac6e26aeb050ff3a : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/44e7affafc4b44dda26a4b23d4ec26a6 : SUCCESS in 1h 07m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62769b59b1224864982459df1841e0cf : SUCCESS in 1h 02m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/481e80ca74874b53801d2bc964046729 : SUCCESS in 28m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9aaf0037f3c40ada13fefdfd427e12a : SUCCESS in 38m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02020facdbfc4cd6961d85d2de0ceba1 : SUCCESS in 10m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4dc5ddeeac14bcfa49fe56e60068acf : SUCCESS in 10m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea63a0c9f6774576997e2c8285a411b1 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a55f2f73eff344a9bccd085de79a5f83 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6246c8edf8884d47a783d9ac7b793668 : SUCCESS in 26m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf4381430fb479f8262e2503824c1c3 : SUCCESS in 16m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0cc60e6faa2144d8bbe3aebf37f785be : SUCCESS in 30m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8793628be82548a3bbe3a6b1536d3339 : SUCCESS in 19m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6792c5cc075645fd96fee73ad1ee79c8 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f5748b537e3e480fac076779a4068971 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/adc3de2f0b98480c9c6c5ef3e73dd28f : SUCCESS in 11m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c23c0033c934c0b8ceadfa7fd85b774 : RETRY_LIMIT in 5m 29s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c80112ade89c46caa61de982e034f160 : FAILURE in 7m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ad0ab2aeed04af18f945c4d3be7249c : SUCCESS in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da9972b6416948e8a09b11cf72ff386b : SUCCESS in 1h 00m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4690f545306943e5babe08ce39353517 : SUCCESS in 51m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b09f7190ed1a4938bdff826cc3df7f7e : SUCCESS in 1h 02m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bdd2342570a4d0d9fbc9783b93bcf6f : SUCCESS in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5b7cddf62d49578cbf1a0619e37610 : FAILURE in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ec3f2d77d664de99f0b81973146db9a : SUCCESS in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb59d493fca545a38fcc8a7952055b1e : SUCCESS in 32m 26s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":25},{"id":"1423f06064b21424504a40a5d725f69e824378fd","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-07-15 02:05:55.000000000","message":"Patch Set 25:\n\n(9 comments)","accounts_in_message":[],"_revision_number":25},{"id":"0d4433547f5c0ee2a0a365799dca901644e93658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-07-15 15:01:02.000000000","message":"Patch Set 26: Patch Set 25 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"3f356b444847a4aee691879f10c79da5ead9e7ac","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-07-15 16:30:51.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"8352f51693c9f6ddd56502ad9d6e8af35f253835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 16:49:34.000000000","message":"Patch Set 26: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dafff5acdfaa402aad031c0563973401\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3549b3784b074a4a8c2b4e0a84b9738d : SUCCESS in 13m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45cf80d3c2b2404cb5504a7b6aa6a98e : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/904b1b55c48b4694b8808f9c8782d962 : SUCCESS in 1h 13m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/959811fde51342e6b842d4b788b9a16f : SUCCESS in 1h 19m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eff9bd0abe964bba883a6e5be72bedde : SUCCESS in 47m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfd1bc0269b747238743892e83d2d6b7 : SUCCESS in 56m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b1547e150a840bbad90403d7c302160 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/653baf19bfa84fb9aee8c02d54df3139 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe606c55fa6f452f8a07c242e119b202 : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3c81bddb11742d6a019400a12d41906 : SUCCESS in 8m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/059dd77ae1c1496383808e5fb6b471b0 : SUCCESS in 24m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/34e40bf620094d9fbfe6051232345961 : SUCCESS in 22m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/62d3c87eef174349bb4e7ab4f8351ca4 : SUCCESS in 33m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3da2578aa8eb4c2381772f286d80f0c9 : SUCCESS in 22m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fd69e67119664d71859284a97b7c778f : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a197e8cc15ba4e40968ca8b7ea050330 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3af151d73454c73aab8fd510ffb6c50 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/52b5e4e425a64a61b643e1c9e1c36e16 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58606c26279f479b877ae5e4dc705e87 : FAILURE in 11m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/59fe4f96f75941799823efb324eb93da : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/648d3a31998f4b1188f0b90714218065 : SUCCESS in 57m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8be9f69f5eb47479ffd409e80f5cb4d : SUCCESS in 56m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/754c1e4a39f14cc1bc6ffde3219c918d : SUCCESS in 54m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46a2cf8cfe5f4d30866b052f7882adb5 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5d2a87302547dc84346a46359f19d6 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c078ad33c67a4bce8a5f6ce130b9b038 : SUCCESS in 22m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97f71b5e794f471c89f149a49a142571 : SUCCESS in 55m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":26},{"id":"da0c5858f531f51ba284c04d8afd51e48d3a3027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 17:36:31.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ca213b706d4f4685e93804efe4ddb5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9078e0cf6ab45ba93f5ff4527d2d56d : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":26},{"id":"ef07956ebcc1c579f5d1b8bc160c57f12f909837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-03 16:14:50.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"e76b2db07289b51e6dec387888652025e0bb3438","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 17:19:10.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59cadc148734366be4b60e36f567e49\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3c94054a53f417d8141563958f97cb9 : FAILURE in 1h 02m 43s","accounts_in_message":[],"_revision_number":27},{"id":"194bba9ded73e7a780f93edd56692ad764f9cd8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 17:23:58.000000000","message":"Patch Set 27: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55c733be0e67477ab8804194eff4f98c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4edc51ab409547519f7754fcf8709c21 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be4bba1e1677439e9236e93662551c2f : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/139fd4164bd34b09a11158cedcadcb74 : SUCCESS in 49m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e24fa940ede431dac20eb54a6aca67f : SUCCESS in 41m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce6eeae33a6347bc8bbefa78fb06b4c6 : SUCCESS in 58m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dda0fbb1d29144ec80279a8995fe7056 : SUCCESS in 26m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/942842ae0d1344b58d31492882747e91 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73cee18ec37a45cbacc59e06db1e06bf : SUCCESS in 24m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fee125a0dbe34066a576c50bda6daeb4 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ad4ca0424ca4e3cb5e64dd1ae739223 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8b30e7ee093b4fceb4d5f4fffc5e6046 : FAILURE in 23m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2227b9aa5dd546c1a13ba75d3938ac50 : FAILURE in 22m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b1020baa1af74aa59a46a9207f9b9262 : FAILURE in 30m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c6550f635184432ea09600c98cbaba7f : FAILURE in 13m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/958c3da343244865a6e660eba3bbffa8 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6738cb9a462f48e2af459677f796b6c1 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cce5f775b45546e88330d7a3796bdf81 : SUCCESS in 9m 54s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7efb9f104b1e4e988fc31a28e30faa9e : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10e93e633a054392b9a2d3b5de1308f4 : SUCCESS in 5m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c28e3dbdfc754d338f8209ccb3643a16 : FAILURE in 7m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5fe3c26d2c5f412aabac08428b890e69 : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/227744f8d7444ad4b560f006843f9944 : SUCCESS in 1h 01m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0957d250c7246378ebdcac44190976e : SUCCESS in 52m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/610cfe4bb501416bb1ddeebb88d7ade6 : SUCCESS in 33m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e121bd33de3486e9f2048092acbdd39 : FAILURE in 24m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa4d3c7f357401b8307d510674d0fd9 : FAILURE in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d041e2a73ce4a869d4ef5816009fc08 : SUCCESS in 26m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f0a36b04650401eae1a4f57ae997085 : SUCCESS in 1h 03m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file test/unit/common/test_prom_metrics.py","accounts_in_message":[],"_revision_number":27},{"id":"8ca9b1f70af6abd2c863b449326657074f262767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-07 14:42:29.000000000","message":"Patch Set 28: Patch Set 27 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"c1209d470b0e1e467638c236f05ea31dddd93e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 15:46:01.000000000","message":"Patch Set 28: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/615da6142795442ab945220724311ba7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e0031f8dee41bea74b465d2604eab1 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2244f69d3f834e30a3708a8d4bb0cc29 : SUCCESS in 6m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f168986263994adeb4b5c6d98c1a058a : SUCCESS in 31m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32f1c22a1bb44606a1704e8e6b506931 : SUCCESS in 40m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2082895c3db495a95c0efdfa219e3da : SUCCESS in 37m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/411209c8c110442894c3180d961a576f : SUCCESS in 49m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed5217b106244b759f76b21983de7b1b : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39d11d0b0a3b4aaf8796ffa4fa0da8cf : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1aa97818bcc44126bd662ef6b66c2797 : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2c53baf392142ec852e2287b501ad42 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff118c4d4dc44c7694e1c61123135bca : FAILURE in 11m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/77bb7aef061f4c5eb968a1f2059df535 : FAILURE in 19m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5f51322f02f8441fa4302d3a4d7fda02 : FAILURE in 25m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8017f2d5438842eebf8245904598a294 : FAILURE in 23m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a1feb397130844b4955cf3ce1d6fd9ed : SUCCESS in 10m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/04115a1072384fbdaa1d8abc7791d4ee : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/215cbe963b8a491a95fabbe511b7339b : SUCCESS in 17m 45s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3ac8df751024fe79a0d40d5cf3f6538 : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/88700e5ffc0545688c556a287d1c02c2 : SUCCESS in 8m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c834d601bcd64f6b93d7ebf1ace9a26a : FAILURE in 8m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/38b60693b57c4aa1aaab8e95eb92f9c5 : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/74be02ac4fe044518979548800f6d6b0 : SUCCESS in 58m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8970e3bdb25741fa88ab82a3b31ab600 : SUCCESS in 44m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26f88124963e4459b70b30e1daf0b546 : SUCCESS in 49m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94db1620cc0f429f8999d9f3f2756486 : FAILURE in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93f9d1e9d4c4b2e8abf542e3f0cf31c : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/340cd6cf730642c6bfde8d380600c29c : SUCCESS in 27m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc98b63929c54fbb8cbe6f49f352350d : SUCCESS in 34m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file test/unit/common/test_prom_metrics.py","accounts_in_message":[],"_revision_number":28},{"id":"dc2c97af86f9dca425dbd688c5a7af4bae4f1c2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 16:10:35.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4021bbc7f2124a409e40c3da90d20d08\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85b6862524b64162b22e754e1b65f395 : FAILURE in 1h 20m 33s","accounts_in_message":[],"_revision_number":28},{"id":"e1c5fa84cd1194e03cae487f6ec6a0bc77bfd5e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 15:02:56.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"09e9c6671675adbfe3ef0c2e886b84e139566152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 15:16:15.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"9d88a08e4dc610299c621fbf20a81c95f9dbdc9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 16:27:44.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1bf2b91166a4828a706048ec030316a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4761c49087aa48858c21ed425ade6621 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8147363c6a1046489314b6b2182911ca : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/9cc8778c8cf64818822f1fa0c4814b24 : SUCCESS in 45m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5fa9921ca0cd4f0a9bd5c76576bd0150 : SUCCESS in 38m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d2e4dda14c44ff88e6d0eab2cfbbf8a : SUCCESS in 1h 05m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa0d6aa73471466db932432ae97ac834 : SUCCESS in 38m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2793dbcc5cd7444b82205e81cef72532 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5945c95f868408d957bcff9186a6d80 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea9110a61c634483bab1bdec4472f536 : SUCCESS in 14m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97e4d409adf744b49866c7e4da089df6 : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef8be70469de4ed0a9def1b3e7eacc87 : SUCCESS in 13m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a60e3272bbe94f4e84bdc22f0ebed447 : SUCCESS in 22m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/630a6290eeaf4926bd54f92d965d4366 : SUCCESS in 29m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fcc2a7f303b947bfb636f17e969bb4d9 : SUCCESS in 17m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e1f6b619bba74e3e9a2b69051958c2a4 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c5929015a624c3c8802f51e57b38cea : SUCCESS in 10m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/474510d025e94daab5e241d55816ff5f : SUCCESS in 15m 34s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/010ebed57c4646d6a44a6d2bb1ba8deb : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a65be5f888b94681b9d690116859af09 : SUCCESS in 8m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa9bdbceeb04434fb9cdac7245fc4089 : FAILURE in 5m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9eae0e9f9f842f183e99f2978b3f21b : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/23947444e79447deb557bd303674b42c : SUCCESS in 41m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39ef282d2d26489aa1f776448bd7eefa : SUCCESS in 56m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9ba8268c78c4742918973d39c05ad5d : SUCCESS in 28m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd4d9fa2c0fe461f9f1a01627321b38c : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5196231dced54378b489a91fae3431aa : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ba782e23df5406b8946a67d752cb776 : SUCCESS in 20m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a58e4fb50bd1421cacd6980e8d4c71f3 : SUCCESS in 27m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":30},{"id":"d89d96d0066df444f635dcb4a1075f0ae3290fc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 18:10:52.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"5f326350eb890c9aff7ec18de5bb2f2ac87f5ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 19:25:49.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d864d558e4743ff9f6f720e827feb32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1d7f155afd4e75b12d7ba0321ef047 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a598135f3224e0c9b6416bee6ed472b : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ebf01b30cbe9440eb30b2c9be4567e65 : SUCCESS in 1h 09m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95b85c4684fb4dac929c1b7d6615be80 : SUCCESS in 1h 04m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0737dec3a47b44aa883982b06f68f8af : SUCCESS in 54m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6210d42eeac84517aac0ab61113c5763 : SUCCESS in 1h 02m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7fd541166724d159b1a641b0ba0b11e : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8550b4c8fb74e3ab110022103c433cc : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e3ce69a74884486ad97e532c72d9fcd : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1166061278b3422aa41b16f656bdbcd6 : SUCCESS in 9m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ed8962f3c2844fb88726a79f14ca215 : SUCCESS in 22m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/64d7f84dfdea404abaddde98fbfd2e57 : SUCCESS in 18m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/76992047c1814233ad86ac32a92d32b0 : SUCCESS in 31m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e6afc23485bd4cba974b2bd05279298c : SUCCESS in 22m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d17446afc37443958aad10ffaeea7795 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3449702179364503a7ac3be6d96bbb89 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6178097b3aad4893bb4db39842a04716 : SUCCESS in 14m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5306f573355e4468a4e1802b7d83d738 : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c85e0d07f0174f8daaae4c214e6fe81d : SUCCESS in 5m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6745cc9ae4ca43e7bdd475eebe1f04bf : FAILURE in 8m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/566b57fd613f4582ae3fe7d74da80d79 : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4b47052b9eb4817884d641d9f51fc40 : SUCCESS in 1h 04m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02c514f6122043b9806ab31f28242443 : SUCCESS in 50m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d47477a0e04d4dc9ad1098cafbc8b90d : SUCCESS in 53m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e049a8c166194fec817febfffd99367e : SUCCESS in 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a9c6f478994cb1b3ffcabf2b950bb6 : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0345adb1d7ff4b7daa17f7223535c8c0 : SUCCESS in 16m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b32709ac108341b58bb8e215d667a413 : SUCCESS in 52m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":31},{"id":"f4b675b73574e7fd3378c73a59286d2770e3aed0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-12 19:47:14.000000000","message":"Patch Set 31:\n\n(21 comments)","accounts_in_message":[],"_revision_number":31},{"id":"545559fe7f7e41d890fc4d0cea6e7b422d263aab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 20:14:53.000000000","message":"Patch Set 32: Patch Set 31 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/prom_metrics.py\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"fba03acdeec4e01c6594392df99090a2b1a5731d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 20:16:50.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"29ad066842feb2518d76a029867cb22df23ed1b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 20:20:56.000000000","message":"Patch Set 34: Patch Set 33 was rebased","accounts_in_message":[],"_revision_number":34},{"id":"cf186ba3f437da9c5576b46226d23226d7484bbf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-12 20:22:03.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":34},{"id":"8bd18166488502ea75411cbf7d14e8c6e4a1936c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 21:34:11.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2582ee7f2a664d879a2343b2b2ee155c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e8f51cb85049fdad57f4420e672cdc : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55bcbce8079541f28ac6ff43b443f61c : SUCCESS in 6m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/f664e665180a4e2db2e908cc78532a8e : SUCCESS in 1h 12m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efdd18b51d5b48558d3333b94319e845 : SUCCESS in 1h 09m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d8167c0b515405286161baa6453cc91 : SUCCESS in 54m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b18b98dec13e441b8fdc08dbb63aa65f : POST_FAILURE in 15m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2a770cb768b4250a21dd0dad0a14381 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06cff0139d4e4aa181cbed7f53956bff : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0df1bff673254741832be11817dd2444 : SUCCESS in 23m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e657d058dde4b39a14622e07734f256 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/74283ed76aab49a59f4ccdf5e758a5b4 : SUCCESS in 23m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2aafbab779164f2faf6e211ca31864a3 : SUCCESS in 9m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7a2c83e68aa94dcda1a7e39ab439907e : SUCCESS in 19m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3f005184ace44b0f84d069e83ead7f47 : SUCCESS in 23m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/33756f2ff23a474aa4a5da58e7bcb01f : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b4116a37dd0d44d5afc5fc2f344e6b7c : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/82992fd2354d4c04b25f2c92c13d61d7 : SUCCESS in 15m 02s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8626759a4a6a4deea4afa81434b1f6ef : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78f55a49e6704026945e12427c28c23f : SUCCESS in 6m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d063994cb5f4830a0a62f3b7985511a : FAILURE in 7m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/acc18342589f4ee3bace859138c5a513 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/01006752594f4488b18d90a38930ae78 : SUCCESS in 45m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1fddea51f38343e59dd1d74dc78c406b : SUCCESS in 51m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd58d274d22f409cb05c97df7bbb9716 : SUCCESS in 22m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68741fcf635b4112bf84601596b8d530 : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ddd696e5c4a41a0a97e18efd56230ea : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a15066c4026b449798f5f0fc98dbdd1b : SUCCESS in 23m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/801c80393493471abea881b2268b0ece : SUCCESS in 34m 56s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file swift/common/prom_metrics.py","accounts_in_message":[],"_revision_number":34},{"id":"0f0e6336e98ef8308cbae24d011450f6b7d74ee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 21:46:38.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc4aaaebef1148bf9974bc5f1fb32a81\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c66fc625d4944a449a55d65c28e0b275 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":34},{"id":"62da7b8eec4728c1e7233bc78503a870d5444259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-17 08:38:48.000000000","message":"Patch Set 35: Patch Set 34 was rebased on behalf of \u003cGERRIT_ACCOUNT_35790\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"_revision_number":35},{"id":"0b3de40882f02eb8fc72c1aa8e0985bd43da00dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:30:15.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6041b255a8074f79b670b9b640e17f3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c59d3ac6e3ed4f5ca7446642a6370d73 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":35},{"id":"dd352a0bc1207a9b63b5c4afd4f7617651885eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:37:29.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d938e7ec6d86490e9a8bce7abded8578\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba6c447c9ee4812afce1ba0941022d1 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce7ff7faab2b4ef8ac0719b6188a5ae8 : SUCCESS in 4m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/fd378c83056c499fb417191ac05f22fc : SUCCESS in 36m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d3221ca32b444ca9a5654ed5646c622 : SUCCESS in 34m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/701108565d5c499fbe8262bd30a1ae7a : SUCCESS in 27m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c45f275f0674dcd99290bd784f6a967 : SUCCESS in 28m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2ccb9ba97cf4b6a841fbb33bf84cd6e : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7bea3ac4d70422d9868f266a635dc61 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/231aa970b21d48018c4f23dd5f07b851 : SUCCESS in 12m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f5c82701eab4234b60776ed70d84d79 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e2b85cafa6e248c982ced3e464575252 : SUCCESS in 25m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0b0ea082a9ae47248096bd7c297eae6c : SUCCESS in 22m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d5d71ba666d24b14af562bf84bbe2f7d : SUCCESS in 31m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3555fe46cd934605b7d567838886b224 : SUCCESS in 11m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31f3dd0f98e14efc81e5010da1c2b698 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/68fb4ec2b6384879b6fee9fb824771e2 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3c09496867fe46fe8be23a5cad29bdcd : SUCCESS in 8m 38s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/62e43981c00b42f885910237aee32cdf : SUCCESS in 9m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/72c633e68dd64418b04a9ccb017a889a : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc1f86d4afa54362b7dfb8647c59ad22 : FAILURE in 7m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa651da12aba495c92c0ce0592bf5ab4 : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c8215d857ec482781453da5ccce025d : SUCCESS in 50m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/41a7e7f1efb248e2aa0df4472e54ef1d : SUCCESS in 56m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c16cca57c43433090bb16f487175516 : SUCCESS in 49m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73a39d6292b146ff9907b927c982f54d : SUCCESS in 28m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2e1729d20b45b6a571291511c27fb3 : FAILURE in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24f1a7c20fa841488bb609931b71e898 : SUCCESS in 26m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce05dee94968433d9d62f16a440762c4 : SUCCESS in 27m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file swift/common/prom_metrics.py","accounts_in_message":[],"_revision_number":35},{"id":"94e9721088820c5a45e8bebabe75c50196537916","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-17 09:56:56.000000000","message":"Patch Set 35: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"87fcda1404be12041baad4f1a0f5b45868d053eb","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-19 00:33:23.000000000","message":"Patch Set 35:\n\n(17 comments)","accounts_in_message":[],"_revision_number":35},{"id":"7d6904271e684a9986ef63a8872f0c0f8b0b8881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-19 00:49:44.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"790e7a1fbf57e68001d1fde9adcc38c112398f37","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-19 00:50:18.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"54085969430911c652c699830664409e70249d43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-19 00:50:30.000000000","message":"Patch Set 37: Patch Set 36 was rebased","accounts_in_message":[],"_revision_number":37},{"id":"06a7d0e902d86e0e4b2d5c481637650fcbad9f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-19 00:56:40.000000000","message":"Patch Set 38: Patch Set 37 was rebased","accounts_in_message":[],"_revision_number":38},{"id":"75bd9828cb1760dca835a84568e31a66d749b8c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 01:34:06.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5964df3b78fd4fbdbcfdba5a42c227c5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e121eb1ba5064303a7510062156c7356 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":38},{"id":"5fa2cfc627fe894354da8ddecc017029c58ed4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 02:08:12.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/538ec1b6a2484fa892b0a8f88d87286e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93049fde56f451db96cab98f13fbca7 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/652a19384d2b4bb9ba11e6831a4b49bd : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c1d0821a77b84a5782cefb363948e7ad : SUCCESS in 1h 02m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5249e1c8a48c488eb464b20a9420f020 : SUCCESS in 1h 00m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48550711ac7648619ca33050525ea5a2 : SUCCESS in 53m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb4fad8b1d44506823a7ab9f30fae3a : SUCCESS in 39m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f782fd5b587544fb92202d3ab1ac927d : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd1a37995101497d8c0bb56acb1338fa : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a8c5585da28431f9695a0beafc2fa5e : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04a4bf6e94d54aedaa15a79f6273e9a9 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/03ad0148bd4a44c8b065bb3ebfb73adc : SUCCESS in 26m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f5677fb3c8de48fe9ecc4e9966c22bf1 : SUCCESS in 22m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3696e38b6e884152b755c1e07df06230 : SUCCESS in 35m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9e1c882476454bc19d5493c7aae90911 : SUCCESS in 13m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f18fd2b1749d43f8bdea11aff297ead9 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b270eafee04d4e31a8b15b607ccf846f : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3d74ef555fbb4f4e804290ca73fc5ce9 : SUCCESS in 12m 53s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/faf61aac3de446978755c827074f8f0a : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5205f69f747c4e85be01265594d3d8af : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/011a71ee35af4fd2b95473e9671d13f3 : FAILURE in 8m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c72dd77f6585462bb510b20bedb57f49 : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc4805cb34a843548a50f83fdaf7d6d9 : SUCCESS in 51m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95bb34cfa2444d4395b7fc4ab3db15eb : SUCCESS in 56m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a42f11f8ab5944ce9d2981ad9802e9eb : SUCCESS in 26m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eabeaf05ad144ac95f62849da72c235 : SUCCESS in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0f2be35df8435a911f3ce1e9fd62bf : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc0df526df324ea3b5afe84660100f20 : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b1add694121409584315cf2c2650d7a : SUCCESS in 1h 04m 02s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":38},{"id":"9c6154e2387e517c97e955e4be18f467f1539068","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-20 16:36:42.000000000","message":"Patch Set 38: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":38},{"id":"f379adeedd1e2eb2b777e9a3ca2464122a30b066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-25 01:29:03.000000000","message":"Patch Set 39: Patch Set 38 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":39},{"id":"9c5c412367ea44bb1742eed6d5fb8585894db023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-25 01:57:00.000000000","message":"Patch Set 40: Patch Set 39 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":40},{"id":"3952df0a1e13b9f1950982f81d44ce610c454ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 03:05:00.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7011192ae61487c9b7aa8afe092639b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf520f268a244db08bf554b437fb8d23 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":40},{"id":"738720e5803c56fa1dbdd9e1cd7e9855a0dccdb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 03:08:51.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/708fccb8961a4ab186a7d9fd4caf2d39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd54d5ecc69646ce9001da376ad064d6 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d435f97d565e425e86f6f2e6de7d419d : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/1262cee467494f22b6c8d8ea8ef93abd : SUCCESS in 1h 02m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e86c09cf95b34b8fa1cc607feb19b729 : SUCCESS in 1h 05m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e3ee8eaa83c447189e5834f6ef4ef728 : SUCCESS in 51m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33921484f81744d59485e8c03547754c : SUCCESS in 28m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/560723e56ced4b16918dc666845070b1 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/803b0b1afe254d1c9c7c82262d00efae : SUCCESS in 12m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70287a3b269749e68e1eadccd928461b : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b20b0afc610740dbb0759323a421f40c : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a61f621b733473082e979e9b57ab18f : SUCCESS in 23m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6721f555d7384bbcb0770ed5f0a601d4 : SUCCESS in 9m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ff98106fb0304a6997f89a5d95c9c0ed : SUCCESS in 32m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/399ba5b703834227aeb7ae00da63fcd7 : SUCCESS in 15m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1704bd42ce8b4bbc9a4a8487988a9f63 : SUCCESS in 12m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a2d8e47404d8499cb64d9a371ba5ef1c : SUCCESS in 10m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/25077f67467a4000b24c325b66e8759e : SUCCESS in 16m 55s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc8a1b43962044509c592de9f248f14f : SUCCESS in 10m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a680761c2ebe4c889b5677bb11663276 : SUCCESS in 5m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f704c05d2e3457e846ec8982dbd904a : FAILURE in 8m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f85517a23cbc420ebeecf284d70086ba : SUCCESS in 5m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba22f3c9e8a34280861063ba29f95191 : SUCCESS in 43m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/793f57d985464b328c72084bc8283eb2 : SUCCESS in 23m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c112388c9f64861a4b5d1a2c57b6de9 : SUCCESS in 27m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bb11be96f59442084127fb369b766c4 : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d720ff837434b869f07c85c325ef409 : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79835035e45d41808ce13f750ae17de4 : SUCCESS in 8m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a169c4419b44b6e95761f113e519e1d : SUCCESS in 39m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":40},{"id":"d9d6f299480079f24880cdc8eea4e25a2b370bdb","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-25 20:25:32.000000000","message":"Patch Set 40:\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"129881b35f2c305a07931638c432944aa96f15e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-26 01:39:03.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"03de672602903decf099f7cd65a7ba23c1b27c43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:53:59.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae1c2f5751a4f6aa86f7407f51d23c0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c99937ec1b5d481e8ae945a4e32ce64a : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":41},{"id":"aa407ada250f4612b5200f0363c54afdbee7205c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 03:01:11.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea0ca68af0349a1b658b39cf215fc5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877ddb00dad94432ab4074deaed877ed : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97b2e29a37444b6fa96de211d8daaa3c : SUCCESS in 3m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/0c385b46a8914ed38a8bc9315516fc64 : SUCCESS in 1h 13m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61fae044272e481c90a783d3fbe9b159 : SUCCESS in 1h 13m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4903916d0a814936992bfa8ca6e9a6d6 : SUCCESS in 28m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/833c17d00eae42f7a2d3c9ff8143e2c9 : SUCCESS in 1h 03m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8f210bcd8ff447683b8004e7a0904fe : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d31f9a0e4aa4dafba8782331f89f2b3 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e95417d41aff4e6baff2622cd06c7923 : SUCCESS in 12m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/893c939ee8dd4efa9d7247f5c504f2ed : SUCCESS in 8m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b61603a7655c4c96b8a2965fc6d3d6f5 : SUCCESS in 20m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/73b6afdf94804cf8bd3bb2276e41ab5f : SUCCESS in 9m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5a8411bdb70140628c27ba01a1b8359a : SUCCESS in 34m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0c312661893e4317a0e9f3c851096968 : SUCCESS in 18m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/234a8ce8bd684c9083374ed6edda054d : SUCCESS in 13m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a42235f7006946dca01db84bdc055861 : SUCCESS in 9m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6bcec76fbde4e6b8f5187bf617c2707 : SUCCESS in 9m 47s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b92a2350353c4649afd52398a44fefd9 : SUCCESS in 12m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/526eb4263828438e9999c271101547d4 : SUCCESS in 5m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf39dbd622514ecebe9a594ba1c476e5 : FAILURE in 7m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6c4f5b785af44c48927b85925a80258 : SUCCESS in 11m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d0419d65ebc422493f1e5265ee07805 : SUCCESS in 41m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee8c04255b664994be9879c9a41f4086 : SUCCESS in 23m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c68b353adc554598b6ab3fd61660fc13 : SUCCESS in 23m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77027dceb10a4202833de5b325eb5162 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029fa85a36e34ec5a89de8293bfcbae3 : SUCCESS in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/715a8df37bf54bd6ab191b93c7eb22ec : SUCCESS in 18m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea1a2736704e4b678af2632d8e6c027a : SUCCESS in 27m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":41}],"current_revision_number":41,"current_revision":"9a0b942af5882aa0fdd0adc2377510f135b04356","revisions":{"d27c62a7cb3a0eeeb62c7bbc64813766bbcc7e72":{"kind":"REWORK","_number":1,"created":"2026-03-19 19:17:24.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/1"}}},"commit":{"parents":[{"commit":"1999c38dab00890ce9f2226c398d0ceb5b0774ec","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1999c38dab00890ce9f2226c398d0ceb5b0774ec"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:16:22.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d27c62a7cb3a0eeeb62c7bbc64813766bbcc7e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d27c62a7cb3a0eeeb62c7bbc64813766bbcc7e72"}]},"branch":"refs/heads/master"},"05e80cc9a07ce9aa248d5e25c9b0ad3c2c15c1ae":{"kind":"REWORK","_number":2,"created":"2026-03-24 16:06:58.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/2"}}},"commit":{"parents":[{"commit":"1999c38dab00890ce9f2226c398d0ceb5b0774ec","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1999c38dab00890ce9f2226c398d0ceb5b0774ec"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-24 16:05:48.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05e80cc9a07ce9aa248d5e25c9b0ad3c2c15c1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05e80cc9a07ce9aa248d5e25c9b0ad3c2c15c1ae"}]},"branch":"refs/heads/master"},"368c97b14d5b6f4a11f0cdf7a69917298092c845":{"kind":"REWORK","_number":3,"created":"2026-03-31 16:11:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/3"}}},"commit":{"parents":[{"commit":"933a397980cc020f6445d7536a53eec3f7f6cc27","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/933a397980cc020f6445d7536a53eec3f7f6cc27"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-03-31 16:11:47.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/368c97b14d5b6f4a11f0cdf7a69917298092c845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/368c97b14d5b6f4a11f0cdf7a69917298092c845"}]},"branch":"refs/heads/master","description":"Rebase"},"14589ea8bff478d7d0cd88c2cce6a804bde0c2d6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-06 19:20:21.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/4"}}},"commit":{"parents":[{"commit":"b188e6f5b36efcb029683afd2d135c119751d912","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b188e6f5b36efcb029683afd2d135c119751d912"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-04-06 19:20:21.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14589ea8bff478d7d0cd88c2cce6a804bde0c2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14589ea8bff478d7d0cd88c2cce6a804bde0c2d6"}]},"branch":"refs/heads/master","description":"Rebase"},"de73359a1ee87e6506bdbbba97fe0f106cc9e74a":{"kind":"REWORK","_number":5,"created":"2026-04-06 19:42:46.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/5"}}},"commit":{"parents":[{"commit":"b188e6f5b36efcb029683afd2d135c119751d912","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b188e6f5b36efcb029683afd2d135c119751d912"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-04-06 19:42:04.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de73359a1ee87e6506bdbbba97fe0f106cc9e74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de73359a1ee87e6506bdbbba97fe0f106cc9e74a"}]},"branch":"refs/heads/master"},"236867e729ea5e296962c5e21ab3eb3e22288362":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-06 21:29:05.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/00/981400/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/6"}}},"commit":{"parents":[{"commit":"7d89e84e0cea0add748ca9937cd32326d33c0960","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d89e84e0cea0add748ca9937cd32326d33c0960"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-04-06 21:29:05.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/236867e729ea5e296962c5e21ab3eb3e22288362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/236867e729ea5e296962c5e21ab3eb3e22288362"}]},"branch":"refs/heads/master","description":"Rebase"},"e88daae5e3241519b6765519e70811508ab58828":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-08 15:20:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/7"}}},"commit":{"parents":[{"commit":"226b453a1d4cc68436e02641ca37ff564f6383eb","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/226b453a1d4cc68436e02641ca37ff564f6383eb"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-04-08 15:20:47.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e88daae5e3241519b6765519e70811508ab58828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e88daae5e3241519b6765519e70811508ab58828"}]},"branch":"refs/heads/master","description":"Rebase"},"da00e0a0b931a681ec3631570a827e087616d98b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-09 14:21:08.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/8"}}},"commit":{"parents":[{"commit":"9f0cf144c46d39e024e2584a124f5eaad18e966f","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f0cf144c46d39e024e2584a124f5eaad18e966f"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-04-09 14:21:08.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da00e0a0b931a681ec3631570a827e087616d98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da00e0a0b931a681ec3631570a827e087616d98b"}]},"branch":"refs/heads/master","description":"Rebase"},"b2820f6edcf2adb6ee651c1af21486b8ebef70eb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-04-09 16:50:26.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/9"}}},"commit":{"parents":[{"commit":"16cb7e13309e42b272d67b3ac2bc8b6c762e3497","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16cb7e13309e42b272d67b3ac2bc8b6c762e3497"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-04-09 16:50:26.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2820f6edcf2adb6ee651c1af21486b8ebef70eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2820f6edcf2adb6ee651c1af21486b8ebef70eb"}]},"branch":"refs/heads/master","description":"Rebase"},"4d8e9a6d3d20e7768e862849fb5912dc4bc6b071":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-04-14 17:41:23.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/10"}}},"commit":{"parents":[{"commit":"81b11074c834ebbc1fe521d6f15ee70c5f468fb3","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81b11074c834ebbc1fe521d6f15ee70c5f468fb3"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-04-14 17:41:23.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d8e9a6d3d20e7768e862849fb5912dc4bc6b071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d8e9a6d3d20e7768e862849fb5912dc4bc6b071"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"b2820f6edcf2adb6ee651c1af21486b8ebef70eb","theirs":"81b11074c834ebbc1fe521d6f15ee70c5f468fb3","contains_conflicts":false}},"13eaa219aa7acc9cd648766d428fe92872098f89":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-05-07 21:59:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/11"}}},"commit":{"parents":[{"commit":"5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-05-07 21:59:38.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13eaa219aa7acc9cd648766d428fe92872098f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13eaa219aa7acc9cd648766d428fe92872098f89"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"4d8e9a6d3d20e7768e862849fb5912dc4bc6b071","theirs":"5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07","contains_conflicts":false}},"bec52fc51f496db895a34c70279d80540e103abd":{"kind":"REWORK","_number":12,"created":"2026-05-15 12:19:27.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/12"}}},"commit":{"parents":[{"commit":"5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-05-15 12:18:33.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bec52fc51f496db895a34c70279d80540e103abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bec52fc51f496db895a34c70279d80540e103abd"}]},"branch":"refs/heads/master"},"d43b4af67d87eb1f98707fed6b3ff433f00a6d5f":{"kind":"REWORK","_number":13,"created":"2026-05-15 18:29:52.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/13"}}},"commit":{"parents":[{"commit":"5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-05-15 18:29:10.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d43b4af67d87eb1f98707fed6b3ff433f00a6d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d43b4af67d87eb1f98707fed6b3ff433f00a6d5f"}]},"branch":"refs/heads/master"},"1c74f689b6fe68153dd5e81c6ed2dd5365dc364b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-05-19 15:40:23.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/14"}}},"commit":{"parents":[{"commit":"397db0737c2ea3d58cb4da7c3d9e0f18193b3524","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/397db0737c2ea3d58cb4da7c3d9e0f18193b3524"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-05-19 15:40:23.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c74f689b6fe68153dd5e81c6ed2dd5365dc364b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c74f689b6fe68153dd5e81c6ed2dd5365dc364b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d43b4af67d87eb1f98707fed6b3ff433f00a6d5f","theirs":"397db0737c2ea3d58cb4da7c3d9e0f18193b3524","contains_conflicts":false}},"c528d172bb32864f7c3abde3f228d64086898573":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-05-20 22:49:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/00/981400/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/15"}}},"commit":{"parents":[{"commit":"397db0737c2ea3d58cb4da7c3d9e0f18193b3524","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/397db0737c2ea3d58cb4da7c3d9e0f18193b3524"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2026-05-20 22:49:16.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c528d172bb32864f7c3abde3f228d64086898573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c528d172bb32864f7c3abde3f228d64086898573"}]},"branch":"refs/heads/master","description":"Edit commit message"},"04622259f9d7182666e8f0c6b675cda3ffcf36f9":{"kind":"NO_CHANGE","_number":16,"created":"2026-05-21 00:11:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/00/981400/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/16"}}},"commit":{"parents":[{"commit":"f1e49db86dfba5732e65403c0b82a0b42e5a214a","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1e49db86dfba5732e65403c0b82a0b42e5a214a"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2026-05-21 00:11:02.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04622259f9d7182666e8f0c6b675cda3ffcf36f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04622259f9d7182666e8f0c6b675cda3ffcf36f9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"c528d172bb32864f7c3abde3f228d64086898573","theirs":"f1e49db86dfba5732e65403c0b82a0b42e5a214a","contains_conflicts":false}},"aa0bf8da76a3d880cb6132a188d0e38883530f2c":{"kind":"REWORK","_number":17,"created":"2026-05-21 22:00:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/17"}}},"commit":{"parents":[{"commit":"5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-05-21 21:59:39.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa0bf8da76a3d880cb6132a188d0e38883530f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa0bf8da76a3d880cb6132a188d0e38883530f2c"}]},"branch":"refs/heads/master"},"28e32e69d72fe9f886965e9ddc65d9c3d028bfb7":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-05-21 22:00:57.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/18"}}},"commit":{"parents":[{"commit":"f1e49db86dfba5732e65403c0b82a0b42e5a214a","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1e49db86dfba5732e65403c0b82a0b42e5a214a"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-05-21 22:00:57.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28e32e69d72fe9f886965e9ddc65d9c3d028bfb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28e32e69d72fe9f886965e9ddc65d9c3d028bfb7"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"aa0bf8da76a3d880cb6132a188d0e38883530f2c","theirs":"f1e49db86dfba5732e65403c0b82a0b42e5a214a","contains_conflicts":false}},"b4ab0180450a75386b6158eeb0528bfb4263499d":{"kind":"REWORK","_number":19,"created":"2026-05-26 15:45:42.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/19"}}},"commit":{"parents":[{"commit":"5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07","subject":"PrometheusClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0d0d3c8322ec5a4e8d2fdb9c9c5be3ab246a07"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-05-26 15:44:52.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4ab0180450a75386b6158eeb0528bfb4263499d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4ab0180450a75386b6158eeb0528bfb4263499d"}]},"branch":"refs/heads/master"},"a92db4071a3dcf4e4e7da3e0d0dc8be5b0c1e360":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-05-26 16:38:45.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/00/981400/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/20"}}},"commit":{"parents":[{"commit":"f1e49db86dfba5732e65403c0b82a0b42e5a214a","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1e49db86dfba5732e65403c0b82a0b42e5a214a"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-05-26 16:38:45.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92db4071a3dcf4e4e7da3e0d0dc8be5b0c1e360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92db4071a3dcf4e4e7da3e0d0dc8be5b0c1e360"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"b4ab0180450a75386b6158eeb0528bfb4263499d","theirs":"f1e49db86dfba5732e65403c0b82a0b42e5a214a","contains_conflicts":false}},"4892dc1247cfb9f212e1926968216810fe3fc658":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-05-28 15:13:01.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/21"}}},"commit":{"parents":[{"commit":"3a19d2fc785f99c1ff4b04f41053386182804561","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a19d2fc785f99c1ff4b04f41053386182804561"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-05-28 15:13:01.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4892dc1247cfb9f212e1926968216810fe3fc658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4892dc1247cfb9f212e1926968216810fe3fc658"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"a92db4071a3dcf4e4e7da3e0d0dc8be5b0c1e360","theirs":"3a19d2fc785f99c1ff4b04f41053386182804561","contains_conflicts":false}},"c39f37ecf2a7c277844183b1bf579083edfbc4cb":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-06-05 14:12:54.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/22"}}},"commit":{"parents":[{"commit":"31a1f30322a2748ca38cc9639bbbfde6298fd897","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31a1f30322a2748ca38cc9639bbbfde6298fd897"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-06-05 14:12:54.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c39f37ecf2a7c277844183b1bf579083edfbc4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c39f37ecf2a7c277844183b1bf579083edfbc4cb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"4892dc1247cfb9f212e1926968216810fe3fc658","theirs":"31a1f30322a2748ca38cc9639bbbfde6298fd897","contains_conflicts":false}},"03c57a11098cde4945060f74ba582791c28c3a03":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-06-22 14:06:48.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/23"}}},"commit":{"parents":[{"commit":"da2e0b87ad5c01fac66398a3765bf183b3e9b787","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da2e0b87ad5c01fac66398a3765bf183b3e9b787"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-06-22 14:06:48.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03c57a11098cde4945060f74ba582791c28c3a03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03c57a11098cde4945060f74ba582791c28c3a03"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"31a1f30322a2748ca38cc9639bbbfde6298fd897","ours":"c39f37ecf2a7c277844183b1bf579083edfbc4cb","theirs":"da2e0b87ad5c01fac66398a3765bf183b3e9b787","merge_strategy":"recursive","contains_conflicts":false}},"4df6e0d883fc0ef371be86f20a0ce00b0c86f3dd":{"kind":"REWORK","_number":24,"created":"2026-07-13 13:34:09.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/24"}}},"commit":{"parents":[{"commit":"b7e62281fb7e807745056fb6f58f589d19327a13","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7e62281fb7e807745056fb6f58f589d19327a13"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-07-13 13:34:09.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4df6e0d883fc0ef371be86f20a0ce00b0c86f3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4df6e0d883fc0ef371be86f20a0ce00b0c86f3dd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"da2e0b87ad5c01fac66398a3765bf183b3e9b787","ours":"03c57a11098cde4945060f74ba582791c28c3a03","theirs":"b7e62281fb7e807745056fb6f58f589d19327a13","merge_strategy":"recursive","contains_conflicts":true}},"5baedea203ef2c7548f71c0efe8eca4272d332bf":{"kind":"REWORK","_number":25,"created":"2026-07-14 19:53:12.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/25"}}},"commit":{"parents":[{"commit":"b7e62281fb7e807745056fb6f58f589d19327a13","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7e62281fb7e807745056fb6f58f589d19327a13"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-07-14 19:52:36.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5baedea203ef2c7548f71c0efe8eca4272d332bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5baedea203ef2c7548f71c0efe8eca4272d332bf"}]},"branch":"refs/heads/master"},"30c1b36c9958838490851bacdfac4eda68514755":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-07-15 15:01:02.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/26"}}},"commit":{"parents":[{"commit":"1cc6bd70b44f8332cb1565aeb5812bc2ba24de1f","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cc6bd70b44f8332cb1565aeb5812bc2ba24de1f"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-07-15 15:01:02.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30c1b36c9958838490851bacdfac4eda68514755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30c1b36c9958838490851bacdfac4eda68514755"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b7e62281fb7e807745056fb6f58f589d19327a13","ours":"5baedea203ef2c7548f71c0efe8eca4272d332bf","theirs":"1cc6bd70b44f8332cb1565aeb5812bc2ba24de1f","merge_strategy":"recursive","contains_conflicts":false}},"15f5958dc16a611fd3f233d3aa537c5efc746612":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2026-08-03 16:14:50.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/27"}}},"commit":{"parents":[{"commit":"e9b03d535fc7f406438b6e92157f9d323a8c6190","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b03d535fc7f406438b6e92157f9d323a8c6190"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-03 16:14:50.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15f5958dc16a611fd3f233d3aa537c5efc746612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15f5958dc16a611fd3f233d3aa537c5efc746612"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1cc6bd70b44f8332cb1565aeb5812bc2ba24de1f","ours":"30c1b36c9958838490851bacdfac4eda68514755","theirs":"e9b03d535fc7f406438b6e92157f9d323a8c6190","merge_strategy":"recursive","contains_conflicts":false}},"5b40b39d09698d7b0002608f5b321cf342e8be5f":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2026-08-07 14:42:29.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/28"}}},"commit":{"parents":[{"commit":"2848dfb97cd0da34ff67ec272192e921d4c124de","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2848dfb97cd0da34ff67ec272192e921d4c124de"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-07 14:42:29.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b40b39d09698d7b0002608f5b321cf342e8be5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b40b39d09698d7b0002608f5b321cf342e8be5f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e9b03d535fc7f406438b6e92157f9d323a8c6190","ours":"15f5958dc16a611fd3f233d3aa537c5efc746612","theirs":"2848dfb97cd0da34ff67ec272192e921d4c124de","merge_strategy":"recursive","contains_conflicts":false}},"cfa4ed3e346f72d66346c7e0c41d7a36ded048cc":{"kind":"REWORK","_number":29,"created":"2026-08-12 15:02:56.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/29"}}},"commit":{"parents":[{"commit":"2848dfb97cd0da34ff67ec272192e921d4c124de","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2848dfb97cd0da34ff67ec272192e921d4c124de"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-12 14:59:33.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfa4ed3e346f72d66346c7e0c41d7a36ded048cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfa4ed3e346f72d66346c7e0c41d7a36ded048cc"}]},"branch":"refs/heads/master"},"ac9f62e0175b48d17776e2d9f042614a8c0d43f3":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2026-08-12 15:16:15.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/30"}}},"commit":{"parents":[{"commit":"8eec55fe81540fcdca5c906eef739b5d43214b7f","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eec55fe81540fcdca5c906eef739b5d43214b7f"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-12 15:16:15.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac9f62e0175b48d17776e2d9f042614a8c0d43f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac9f62e0175b48d17776e2d9f042614a8c0d43f3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2848dfb97cd0da34ff67ec272192e921d4c124de","ours":"cfa4ed3e346f72d66346c7e0c41d7a36ded048cc","theirs":"8eec55fe81540fcdca5c906eef739b5d43214b7f","merge_strategy":"recursive","contains_conflicts":false}},"18867fed23f40bf2117321bc895bf8b0a31abd5f":{"kind":"REWORK","_number":31,"created":"2026-08-12 18:10:52.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/31"}}},"commit":{"parents":[{"commit":"8eec55fe81540fcdca5c906eef739b5d43214b7f","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eec55fe81540fcdca5c906eef739b5d43214b7f"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-12 18:10:13.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18867fed23f40bf2117321bc895bf8b0a31abd5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18867fed23f40bf2117321bc895bf8b0a31abd5f"}]},"branch":"refs/heads/master"},"efda189ebe8219d0f999f8d69e34931a8b43e836":{"kind":"REWORK","_number":32,"created":"2026-08-12 20:14:53.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/32"}}},"commit":{"parents":[{"commit":"fdfa779400e8051a6855fe9562efbe4122bb2a1e","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdfa779400e8051a6855fe9562efbe4122bb2a1e"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-12 20:14:53.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efda189ebe8219d0f999f8d69e34931a8b43e836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efda189ebe8219d0f999f8d69e34931a8b43e836"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8eec55fe81540fcdca5c906eef739b5d43214b7f","ours":"18867fed23f40bf2117321bc895bf8b0a31abd5f","theirs":"fdfa779400e8051a6855fe9562efbe4122bb2a1e","merge_strategy":"recursive","contains_conflicts":true}},"671325d8b97ce22efe51e93206e064d4b06a42bd":{"kind":"REWORK","_number":33,"created":"2026-08-12 20:16:50.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/33"}}},"commit":{"parents":[{"commit":"fdfa779400e8051a6855fe9562efbe4122bb2a1e","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdfa779400e8051a6855fe9562efbe4122bb2a1e"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-12 20:16:50.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/671325d8b97ce22efe51e93206e064d4b06a42bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/671325d8b97ce22efe51e93206e064d4b06a42bd"}]},"branch":"refs/heads/master"},"082902d49a5a445d7b05b88b70469270299de0a5":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2026-08-12 20:20:56.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/34"}}},"commit":{"parents":[{"commit":"d34ce7a3c3271062fd5a378ec8b6304035b3fa42","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d34ce7a3c3271062fd5a378ec8b6304035b3fa42"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-12 20:20:56.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/082902d49a5a445d7b05b88b70469270299de0a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/082902d49a5a445d7b05b88b70469270299de0a5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fdfa779400e8051a6855fe9562efbe4122bb2a1e","ours":"671325d8b97ce22efe51e93206e064d4b06a42bd","theirs":"d34ce7a3c3271062fd5a378ec8b6304035b3fa42","merge_strategy":"recursive","contains_conflicts":false}},"666504708c2d178654aa955830f581a8ec374452":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2026-08-17 08:38:48.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/00/981400/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/35"}}},"commit":{"parents":[{"commit":"ed5b8ff1596f1afd1bf2bcc650cb80c1f2b1871e","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed5b8ff1596f1afd1bf2bcc650cb80c1f2b1871e"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-17 08:38:48.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/666504708c2d178654aa955830f581a8ec374452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/666504708c2d178654aa955830f581a8ec374452"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d34ce7a3c3271062fd5a378ec8b6304035b3fa42","ours":"082902d49a5a445d7b05b88b70469270299de0a5","theirs":"ed5b8ff1596f1afd1bf2bcc650cb80c1f2b1871e","merge_strategy":"recursive","contains_conflicts":false}},"a86102b7e7914ede9014f1c30b87f663ea33654f":{"kind":"REWORK","_number":36,"created":"2026-08-19 00:49:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/36"}}},"commit":{"parents":[{"commit":"d34ce7a3c3271062fd5a378ec8b6304035b3fa42","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d34ce7a3c3271062fd5a378ec8b6304035b3fa42"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-19 00:47:22.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-authored-by: Copilot 223556219+Copilot@users.noreply.github.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86102b7e7914ede9014f1c30b87f663ea33654f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86102b7e7914ede9014f1c30b87f663ea33654f"}]},"branch":"refs/heads/master"},"24c51b08ea1667a8d11136a57f68f77f2039c8bd":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2026-08-19 00:50:30.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/37"}}},"commit":{"parents":[{"commit":"adb92c20f15507021e04a29367847cc1d7b9c48b","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adb92c20f15507021e04a29367847cc1d7b9c48b"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-19 00:50:30.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-authored-by: Copilot 223556219+Copilot@users.noreply.github.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24c51b08ea1667a8d11136a57f68f77f2039c8bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24c51b08ea1667a8d11136a57f68f77f2039c8bd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d34ce7a3c3271062fd5a378ec8b6304035b3fa42","ours":"a86102b7e7914ede9014f1c30b87f663ea33654f","theirs":"adb92c20f15507021e04a29367847cc1d7b9c48b","merge_strategy":"recursive","contains_conflicts":false}},"26f57d0f85308e1e6ef67aeaab127d713180929c":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2026-08-19 00:56:40.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/38"}}},"commit":{"parents":[{"commit":"181c1c05f48785e946122695c4e41cd9fd89d289","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/181c1c05f48785e946122695c4e41cd9fd89d289"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-19 00:56:40.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-authored-by: Copilot 223556219+Copilot@users.noreply.github.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26f57d0f85308e1e6ef67aeaab127d713180929c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26f57d0f85308e1e6ef67aeaab127d713180929c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"adb92c20f15507021e04a29367847cc1d7b9c48b","ours":"24c51b08ea1667a8d11136a57f68f77f2039c8bd","theirs":"181c1c05f48785e946122695c4e41cd9fd89d289","merge_strategy":"recursive","contains_conflicts":false}},"347b31a010c433abea612785358a5cfa083ac0e1":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2026-08-25 01:29:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/39"}}},"commit":{"parents":[{"commit":"01f1ad5d3bb4d03365b207afdae013a75ceca181","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01f1ad5d3bb4d03365b207afdae013a75ceca181"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-25 01:29:03.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-authored-by: Copilot 223556219+Copilot@users.noreply.github.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/347b31a010c433abea612785358a5cfa083ac0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/347b31a010c433abea612785358a5cfa083ac0e1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"181c1c05f48785e946122695c4e41cd9fd89d289","ours":"26f57d0f85308e1e6ef67aeaab127d713180929c","theirs":"01f1ad5d3bb4d03365b207afdae013a75ceca181","merge_strategy":"recursive","contains_conflicts":false}},"65e345df20983394029b5deaf602b92353dc58f6":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2026-08-25 01:57:00.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/40"}}},"commit":{"parents":[{"commit":"1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-08-25 01:57:00.000000000","tz":0},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-authored-by: Copilot 223556219+Copilot@users.noreply.github.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e345df20983394029b5deaf602b92353dc58f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e345df20983394029b5deaf602b92353dc58f6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"01f1ad5d3bb4d03365b207afdae013a75ceca181","ours":"347b31a010c433abea612785358a5cfa083ac0e1","theirs":"1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0","merge_strategy":"recursive","contains_conflicts":false}},"9a0b942af5882aa0fdd0adc2377510f135b04356":{"kind":"REWORK","_number":41,"created":"2026-08-26 01:39:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/00/981400/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/981400/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/981400/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/981400/41"}}},"commit":{"parents":[{"commit":"1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-03-19 19:05:20.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-26 01:38:03.000000000","tz":-300},"subject":"add prom metrics to dispersion report","message":"add prom metrics to dispersion report\n\nalso add missing test coverage for test_dispersion_report.py\n\nChange-Id: I18206720b1fcffe63e6c0e968ec7de5cfd7786ed\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-authored-by: Copilot 223556219+Copilot@users.noreply.github.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a0b942af5882aa0fdd0adc2377510f135b04356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a0b942af5882aa0fdd0adc2377510f135b04356"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
