)]}'
{"id":"openstack%2Fswift~981408","triplet_id":"openstack%2Fswift~master~Ic1b05e4d513ff3671718b1c3510f3ed7371d6a25","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic1b05e4d513ff3671718b1c3510f3ed7371d6a25","subject":"AUTHORS/CHANGELOG for 2.37.1","status":"MERGED","created":"2026-03-19 21:22:44.000000000","updated":"2026-03-20 00:20:49.000000000","submitted":"2026-03-19 23:54:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981408","meta_rev_id":"fc5c52e25450453feb841d3353a3708bf3cf5abc","_number":981408,"virtual_id_number":981408,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 23:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-19 22:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-19 22:28:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-19 21:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3b0c6a65273f9d2d0169f995ae16cbb349f2a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-19 21:22:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd33fef8fd77a055ca69c1ab10a9e415a6e03037","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-19 21:23:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f94885bdfd9a2d556f3a1d76a7fa5fa0ae05b274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 21:58:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a17a0c0c4b0473f966b39951aa78fae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fe23434be74361b8eecdbe8f4efd88 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d3b2d2dae974939a706d12dd71bc943 : SUCCESS in 9m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3366802752845feb164486abbe29baf : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f45f84ebe87e4dbab5597117ce9e2921 : SUCCESS in 9m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/64bb0d58d1944faeb122fd3ab57975f7 : SUCCESS in 23m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e943d7a13a47cbb6b9972ec96eda19 : SUCCESS in 29m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3a96c1553e874584b98b6a3326f1da8c : SUCCESS in 20m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84956db2fafd4eccaeb67ad82001a392 : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcaa44967a944048f90d787897a8cde : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c7620e53a96d5f03f1c7e17e56f874cd2f2e8d8f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 22:27:10.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/398e8d59dc7a45e293525679047e8f6c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e386f4e9a1647798c15059cfaac5606 : SUCCESS in 33m 19s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0c6f35f9bcb4631aef30e49477fbbdc : SUCCESS in 24m 19s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/719c486cbda14b968add1e5faca2349a : SUCCESS in 31m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91cd489bd1e44a67808266e468e73071 : SUCCESS in 7m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfdabe540ff3474280970073632c2da1 : SUCCESS in 19m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f04c9b514ab44341a9936541e982d18e : SUCCESS in 20m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a0de7df9aef4a1eb30fe785223138be : SUCCESS in 23m 59s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a52826e20f8d45328875054c84650223 : SUCCESS in 8m 43s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e65316845ec417d8da7b5010d952af9 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ef948c53ed74c46a5a91c39c6d5af0f : SUCCESS in 25m 58s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/c39229bba0184ad6afac138ecbdb2f05 : SUCCESS in 12m 09s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/efdcf8fea91349e7a2fb85751068079b : FAILURE in 4m 17s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/1a365d2ca639468e80682f33a9e704e3 : FAILURE in 9m 50s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/ea4eb2125ee949419379ddd8c7fcfdfa : FAILURE in 7m 39s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/939dddc4d92b4caf86e31de9bf051178 : FAILURE in 8m 03s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/dd474995fe2746918828daebb98ca833 : FAILURE in 8m 46s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/ceacb100cca544278239162cd9e4da1b : SUCCESS in 17m 57s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/2b39010d0b3845ffb6422cf4e8f7396d : SUCCESS in 17m 53s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/88aa045eed15408f8e8bd8958675d9fc : SUCCESS in 18m 44s\n- swift-multinode-rolling-upgrade-flamingo https://zuul.opendev.org/t/openstack/build/96fd29d39288430bb95e8ffad00212b9 : SUCCESS in 16m 36s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/49b54498387441b6a89b76cded5ca940 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b22dea32efdaab0a38120626aa586a4225999f87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-19 22:28:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc0a74811ac6464b559cf9f824979bcc0a794752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 22:28:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a702c81ecda66d75a04d35a22375d8f9e96f281f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:54:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b00328ee124641f9872ad4a35d9cd24b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831032795d0d4d8793e3d439ca6d1840 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d6d1fa7cdce44d1bc1de69acc94ff7b : SUCCESS in 5m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/215123d8660241d19cdbc16f6007f57a : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf2be2f36b7643909e0f499b4cf8b8a6 : SUCCESS in 20m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3fecbc9040d649a9b558bf048190ebc8 : SUCCESS in 8m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bc0e6bc238d847fbbcc03c6d72a19379 : SUCCESS in 17m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2e189cecaa7643a4913ba102f8672e0c : SUCCESS in 9m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd1583269b94417198a6d4c4d3924a0d : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6ef5f95ece2b405d9cbf8c0d72091958 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d291380c72634083b252c2f0b2686d33 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/246cbbce166c4e6a955c8989ded57258 : SUCCESS in 9m 13s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/58bbdb9ff2144118912a8da2fb31784e : SUCCESS in 7m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d13eb79fa66343c693b156d53f94642d : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748a3b0bdfe0425dbb63ba151c8c1e5d : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3a05f90da579c42862841705f786dce18cdb36ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:54:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eba8d08bb59776cdcb1beb8ab48673efb89e5775","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:56:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/131f107c2c534659ad15afc865ccccea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583ca906c82f4d7abf2cc223bf7374e8 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9eea0144fc347c7b703572567550727 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ee78117021a142e1ac1fb8e7e76bac30 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc5c52e25450453feb841d3353a3708bf3cf5abc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 00:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd055bdcfa5b4a58ba4d3d5fac46769a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/249b60da62fd41fcaa35ca7c55da16f7 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ceefd82d5b110a6e7d6c26f29b0c130065860df3","revisions":{"ceefd82d5b110a6e7d6c26f29b0c130065860df3":{"kind":"REWORK","_number":1,"created":"2026-03-19 21:22:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/08/981408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/981408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/981408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/981408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/981408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/981408/1"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-19 21:14:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-19 21:22:28.000000000","tz":-420},"subject":"AUTHORS/CHANGELOG for 2.37.1","message":"AUTHORS/CHANGELOG for 2.37.1\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ic1b05e4d513ff3671718b1c3510f3ed7371d6a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceefd82d5b110a6e7d6c26f29b0c130065860df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceefd82d5b110a6e7d6c26f29b0c130065860df3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
