)]}'
{"id":"openstack%2Fswift~981462","triplet_id":"openstack%2Fswift~stable%2F2026.1~Iffbd37f1295445760c107fc2bda7e3070278a4dd","project":"openstack/swift","branch":"stable/2026.1","topic":"create-2026.1","hashtags":[],"change_id":"Iffbd37f1295445760c107fc2bda7e3070278a4dd","subject":"Update .gitreview for stable/2026.1","status":"MERGED","created":"2026-03-20 08:59:39.000000000","updated":"2026-03-20 21:01:30.000000000","submitted":"2026-03-20 20:57:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981462-create-2026.1","meta_rev_id":"77162bae36cdd8ceae8f0f5a25fe144dd4b8f6d5","_number":981462,"virtual_id_number":981462,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-20 20:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-20 16:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-20 16:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 09:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 16:59:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"789bdb7547b4a3e6e32c8828029c6127b03ba599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-03-20 08:59:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"231bd591ea92e5840ce4dbece9df7c0cbabfa106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 09:39:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36e5c088f67445f885054b995aeb7559\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358daef8d7154e4895fbf42726d6fc94 : SUCCESS in 4m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/adcacb987bae42dca503c5d3fbec0079 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/048267083d89477eacee63ec79791170 : SUCCESS in 22m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7e6d12b9e13149f5b66e867d920f8e28 : SUCCESS in 18m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/82422e6cc10d47ea8eb82bd520a3d8f3 : SUCCESS in 33m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b086d44c937a47e984d7a609ca2c6e2f : SUCCESS in 19m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/187da6af3cbf44b6a09e03d0a51da3a9 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01b5ab0d23246bf80052b727b054530 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4f7ca4cf7551015fd26389b9874883cd86afbfbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 10:56:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7766e21caeb044129304db80ebfef7d2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/adbc635cc447455db6a343ffed420d05 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":1},{"id":"087ac28bddbbb849b7d01e2cd5e2d9e6df7283cf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 16:59:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d3c0adf181a88ff60da38b8031bd95e142d368d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:00:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca3b361fef906d445b0b0f4e81738e6ecfc1bfc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:57:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce3b84d41e24d90833fd8ebc494e7df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a02026b5798412e81e7c6c3a3d7870a : SUCCESS in 9m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cfef75d3d73d480782f3a132c5165126 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef0c6dfb692f47b9b4e59b7a986be0af : SUCCESS in 25m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8937b19c109c4404bf5214b4286378bf : SUCCESS in 10m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/814a9b7d35954127b679eb320d688297 : SUCCESS in 30m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fcd66db8005e42baa902682001ac83f4 : SUCCESS in 10m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a0c3e838cc5c4d23aaa19e75753ff650 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d02e07b2f26407bbccc3429141c543b : SUCCESS in 16m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fdb9d3f90f9d4a81918fa265d2479ea4 : SUCCESS in 16m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab500b4170e34c688dfe90df6d63b8c5 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cec62bc8127f44129d0906a14862d1bc : SUCCESS in 9m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3c9d16efb5c4e9babc9e4b2fab2ec6a : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a9c26d37410409587c223b24a642ea4 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0d365c37daf47e27e339867b8b29d9089da6d3fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:57:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"77162bae36cdd8ceae8f0f5a25fe144dd4b8f6d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 21:01:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/732bfd6ea37b46c48295dd3e60623398\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543f68f2cd09434da343cca5fbb6996c : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/32ffcfcca4bf4e10862fa7aaad977c64 : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8b94bbc913c20f867f5b0a880e5c349b430b52d","revisions":{"a8b94bbc913c20f867f5b0a880e5c349b430b52d":{"kind":"REWORK","_number":1,"created":"2026-03-20 08:59:39.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/62/981462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/981462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/981462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/981462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/981462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/981462/1"}}},"commit":{"parents":[{"commit":"ceefd82d5b110a6e7d6c26f29b0c130065860df3","subject":"AUTHORS/CHANGELOG for 2.37.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceefd82d5b110a6e7d6c26f29b0c130065860df3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-20 08:59:39.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-20 08:59:39.000000000","tz":0},"subject":"Update .gitreview for stable/2026.1","message":"Update .gitreview for stable/2026.1\n\nChange-Id: Iffbd37f1295445760c107fc2bda7e3070278a4dd\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8b94bbc913c20f867f5b0a880e5c349b430b52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8b94bbc913c20f867f5b0a880e5c349b430b52d"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
