)]}'
{"id":"openstack%2Fswift~981463","triplet_id":"openstack%2Fswift~stable%2F2026.1~Ia0d65818cdb443de8a3678b68a829635987994c6","project":"openstack/swift","branch":"stable/2026.1","topic":"create-2026.1","hashtags":[],"change_id":"Ia0d65818cdb443de8a3678b68a829635987994c6","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","status":"MERGED","created":"2026-03-20 08:59:42.000000000","updated":"2026-03-20 21:02:45.000000000","submitted":"2026-03-20 20:58:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981463-create-2026.1","meta_rev_id":"2b42f95fd412075c82be809a8656a03cd7541a44","_number":981463,"virtual_id_number":981463,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-20 20:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-20 17:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-20 17:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 10:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 17:00:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5461801713612ad6f554864db68e75dc73dd47f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-03-20 08:59:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"040d128f7186d8255a544099907d407859b0d110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 10:17:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12e57ba07c19422dbeb9b351f7f87acd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e452222f376445c3b5d633c9fded294d : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/712bb04e71cb4344b071458b64e2bfec : SUCCESS in 8m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06671f46fd6b4c6884d1e680c64b84b0 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b16a1e1cfbe54126a7342f2d780a9f82 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bf74d8fadef4d9495d5ec74b29204f9 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/506db30a5fa14fad8282e28ae1668329 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/319d1c3e3ea14b21ab3444edf9c48937 : SUCCESS in 25m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8ef388646ac74fbe935e2f60b40cd093 : SUCCESS in 8m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d60424217c5d40318aa87f9cf1f54007 : SUCCESS in 10m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9d13b0fa438844a8b6e9d1e71ceca77a : SUCCESS in 21m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9ccc7b667ee2445b82d3e478c4d917ca : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d521da4a4a9049208254c5ab960b3c9b : SUCCESS in 19m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/986d541ab32045c4bdeab4ef59fa90d7 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9e1341a50e7454e8b2027a371e10f17 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47a281d3171c4ed5bd05797313ea3730 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/841a1ec7dbe943b794f2da738fdc425f : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f1c86dc0fc348a4bc4f632b37d479e0 : SUCCESS in 1h 07m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1227a65c5a6f456c82948947e4061882 : SUCCESS in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57bd857fb7934b4c8f75bbe88bd8cfcc : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a6a250f1d1414109b23db95594588115 : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9fc247f254602125c64022703cb3141ee824f8ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 11:27:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31cefde29f3240ce827b3f7ce495628d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afb29312cc4c4138b57b2029d2fe86a3 : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"834a4e6b0d6ec0a95401a9131df73a4b6fedb28f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 17:00:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abd49695ac5f5d229b4c82cb6ede63f9204092cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:01:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01c6d4457e6da4dc85392787a01e32f661ac4dca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:58:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776299d54960451894758a3a14b257c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87de332fe8fd4731b4280c712072c9af : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89f8d6f6888046f2bb3c2b2fd521f9a9 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/71c22101f2cb47c9949c3b8cd962ceaa : SUCCESS in 1h 06m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1529f60975474d9eac83eba990a89cff : SUCCESS in 1h 06m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89eb7d9b3bc34763af9f452df594e4f0 : SUCCESS in 54m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/021146173c5349a08fa666ca9e4b45f5 : SUCCESS in 1h 01m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3a04ff367ea0441a88b481ffa3d4c4d7 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6abd12b9a9574312bdd5da9e596fb383 : SUCCESS in 12m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aa0ab8bd31d840659751b38f84eb7fb1 : SUCCESS in 22m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4ae1bf007daf4931a38df6713ba63129 : SUCCESS in 13m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/171fae6dc06d41478f98005f37a4ef68 : SUCCESS in 18m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d311b0b3139b48d1b1c07c5518ba6d2e : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/db546c57127c425d932e3028e29da141 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7c15a12a4f2f485683a1d55f92a653de : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60715713ae5e4c1e804ba46158630107 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5f97fff0799f483d88d9bc4763d8bd1d : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b529b3c797f64b308712eac8416a1364 : SUCCESS in 1h 05m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80ec83e96a6d4165911a58c6bdba355a : SUCCESS in 24m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae49c3dcefe94d5182de09757f68043d : SUCCESS in 24m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/829b58318e324a0095f0d6801e4cb139 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eff1903b646a40d690dfb73185ca1ff6 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/346c6eced1e348a4b193e323f6b24558 : SUCCESS in 19m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1cacbd91df34ae9b5542ba2ff8dfbb9 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ad6820bbff8610d6aa897eb9986f87095d817249","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:58:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b42f95fd412075c82be809a8656a03cd7541a44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 21:02:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a17d5a180b34caca4d5d43b8b46f554\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30074683b174a74bd22a8263a46d45a : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aba49887b9d84238b585c7c9202615f3 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2c401c38314a47ac8c45d8703d2dda2b : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78b20c6f29802a35cd559d8c63ec3ab48b75bf81","revisions":{"78b20c6f29802a35cd559d8c63ec3ab48b75bf81":{"kind":"REWORK","_number":1,"created":"2026-03-20 08:59:42.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/63/981463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/981463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/981463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/981463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/981463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/981463/1"}}},"commit":{"parents":[{"commit":"a8b94bbc913c20f867f5b0a880e5c349b430b52d","subject":"Update .gitreview for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8b94bbc913c20f867f5b0a880e5c349b430b52d"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-20 08:59:41.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-20 08:59:41.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ia0d65818cdb443de8a3678b68a829635987994c6\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b20c6f29802a35cd559d8c63ec3ab48b75bf81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b20c6f29802a35cd559d8c63ec3ab48b75bf81"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
