)]}'
{"id":"openstack%2Fswift~981531","triplet_id":"openstack%2Fswift~master~I4b619bc20316eed61ca257a151fcca4f18c9af76","project":"openstack/swift","branch":"master","topic":"p-versions-listing-marker-fix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-30 19:43:24.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-03-30 18:25:57.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I4b619bc20316eed61ca257a151fcca4f18c9af76","subject":"versioning: fix version listing marker name","status":"MERGED","created":"2026-03-20 15:01:16.000000000","updated":"2026-03-31 10:34:18.000000000","submitted":"2026-03-30 19:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"981531-p-versions-listing-marker-fix","meta_rev_id":"2ac36a80178cd7d6fb0852d9869f4dce9b53f1e9","_number":981531,"virtual_id_number":981531,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 19:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2026-03-26 02:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2026-03-30 18:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 16:02:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-03-20 16:02:56.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-03-20 16:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-20 17:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"779352919def2aef85d86916ebd600e2fd1c0e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-20 15:01:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95f5ad87ef1cbf42dff6bb2f8e5fd3aec7ecb959","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-20 16:02:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e87955cf312c1ded93629ab4c786c4b5eff36d71","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-20 16:02:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c3acf98d55d11896b00ee50da1b10735ff231375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 16:07:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1a4fea7fe24759923a9ff09da4a3e3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50aed836fc76492fb2747eb4cf00a611 : FAILURE in 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"50065e50353a6ad642dfaa8c63e8964cf240b575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:36:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1975c73dbcaa4a63bc305ff84ed9b76f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a6c40dd53f41828066aa89ce603a6f : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/def184328d0b4d90895c6401c3a23a5f : SUCCESS in 1h 05m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db7f54706cf04dd4a5934a469df27e3b : SUCCESS in 1h 10m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d45578fdaba8446282644d721865aa3d : SUCCESS in 55m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fc90d42873f4c8f87e5bca7b09d6bf7 : SUCCESS in 1h 07m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9abd686efd6546689d3519d32bf9b242 : FAILURE in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99c00afcffef4388a50314944ac0229b : FAILURE in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65199073be404ac9add2e2c648816056 : FAILURE in 21m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2327cb5281eb44d8befa569c69686357 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0e77986168ef43a79ecf30976b7c327c : FAILURE in 11m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a6259dd44b2740a291969fc6ff332fd1 : FAILURE in 22m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7db244bbdd684a1ba238cd7c3e45e886 : FAILURE in 28m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d815880d80c54ced975976c0d058371b : FAILURE in 10m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ad582eaf8fc647568131fbc7a9e5bd22 : POST_FAILURE in 18m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2274af09b6ef4a1c8d3c489ecd406523 : FAILURE in 15m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1a6587e9af341899294bfdf5fe81778 : FAILURE in 21m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07215aeba2004048a9ea7be46795f19e : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c873320eeab4471b29d464753703099 : FAILURE in 6m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/200ce6bad5ec43d6b678f7e6eb110450 : FAILURE in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a910d489bbd9438695742f7be81063f9 : FAILURE in 49m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b69cefe6a2204fd9bf5da47bc2eb6a18 : FAILURE in 13m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a5da7eecddb4021b61fb6f2d2a88057 : FAILURE in 28m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6906a914901d4d19b7857062237af5a3 : FAILURE in 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9c6c041a35466488785cc6b3174e72 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b50f4d5172ff486b8d276d153af68002 : SUCCESS in 18m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99d4a3dbccb34fb8b30fc04c07b14004 : FAILURE in 51m 08s","accounts_in_message":[],"_revision_number":1},{"id":"aab7b891b275cb25b7827f5cf47ce7986ac421a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 15:50:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4e74ab90487925f674279381873da9d7479db0ea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 15:56:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8661d1338d69101cb61c179759f3758f132bea3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 16:58:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92519135c372436281f891a218ed23fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff6d5f86a5334e2c918c24444eecfedb : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/be503a89cdd1499d9ec1ad530e368158 : SUCCESS in 58m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed7fcb26e5db45d6af4fd597f2375bcf : SUCCESS in 1h 00m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/552c9ce3588c4873bd97221a5aafc742 : SUCCESS in 51m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a9cb1cc6f8444e6a1d3787b94f0ce68 : SUCCESS in 1h 02m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/672cb7272d554e7caa5891da7b19c3e7 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58adc3661cb048e696c469f0c35da569 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e665b96cc1624b5e80ad7e21314d63c5 : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c896b5805364291bb23122c1122d9f6 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7576db60e450486997a75d8cfc15f969 : SUCCESS in 25m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e15a6f5f4c93453c884b659087023ba5 : SUCCESS in 21m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/936ca5344be8478c946512c85b90806a : SUCCESS in 29m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/37d7a2a48bce42ceb3b762d1673cc4d9 : SUCCESS in 20m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/398f766e9f0340af86e36f925d24513c : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e3cc7e1003e46c7a2a7040f2163f8ce : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cbca9c9786dc404caf647fe78a6e0d9e : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdb80cfe6cb6489a9331b60374dd9053 : SUCCESS in 7m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c40db4615654043a3308632ac657a70 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa66083d7da347bdb462b230f5f60aa7 : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bacdae08b96492eb7f49624ed05fdc7 : SUCCESS in 54m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6c947f520cb404fb51dde34c839dd19 : SUCCESS in 51m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/108250d6c50a43288083b171437c1ca6 : SUCCESS in 51m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62cfecb793eb4404a9b600fb3a586996 : SUCCESS in 27m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12456c403068469aba3a1737ce12234b : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca6b797ed77f49bfac36df6107d648a1 : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee8e9fc55bda451eb38e8e3cffa2a598 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d7f635fff83975ae84c4382f0469a77ecd1d3d76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 18:32:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3042dd6cab6048f99c385f618c46b0a7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab231d79ad1a4b01a5abd6ec778e3e09 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":2},{"id":"814e98ce87f835c5cd858e80653b21a4771f7d39","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 18:39:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"74973b9909e1e94b3abbb7df891026fbe2407b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 18:39:26.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cedc16e761f773391ba2286493e8623d92a09614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 19:51:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2ef1c414ed47be8c7496026566a556\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28fd6d9086f468c900c831f09fa4b3d : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/44ae345f8ed04de88388f4c9da5f3ad6 : SUCCESS in 36m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5112d6a8e5b407fac9d754f185d5d4a : SUCCESS in 1h 04m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3cdbaebd77142fb803e4281eec3d80d : SUCCESS in 40m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3852b1525ea49c8b29f9de9301ac760 : SUCCESS in 50m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3ec736551514a769535dfa7fb1f8e5a : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26efb1445ec54e5f8c259ba02db2b7e9 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8dd794054a164d5d986aaa6ab61e7672 : SUCCESS in 16m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/991e4d2325594ac3a1301c14b4f741ad : FAILURE in 5m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9be27cb7498848f3958f074dcb665330 : SUCCESS in 24m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8794385bb3c944b6bf55928bcd743ee8 : SUCCESS in 21m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c328d2b4ec1b43c18d5c51b3749e1813 : SUCCESS in 12m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b59f7334b56f4941a935565a3e0e8918 : SUCCESS in 20m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3de0e882231f49fc9f5ee6237de538c7 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ee97086f4c94f78b3e560ea87c6ed18 : SUCCESS in 11m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1b19e82ec0e7474c800d16806db23fe2 : SUCCESS in 20m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57861f1a5e0c4cdbbfcc851579b1a16b : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a7a901b6fb1479689f04bd7bfcdb74e : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4249842e2324742bcab2fd59371e1fa : SUCCESS in 9m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dda7e5c77264ce38edff82b6b6df748 : SUCCESS in 57m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49e4e55c90f3433383647e7608cb7862 : SUCCESS in 57m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da219115e6864a77aaeffc43ecaf6060 : SUCCESS in 52m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a603525c3f54ceca7a37b144701630e : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8105519ead954275bb43c4f839d0f58e : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a35986e970d2455b87780c2a6f12e7be : SUCCESS in 10m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d10fea6d032c4147a696b51fd29166e1 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":3},{"id":"f5c24cff01b94e0c33c9a090de146e13480eb778","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 20:32:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1013faa841c84fbf8a3fa009d70a3785\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae26545bdc1f44fa898d69b42db6c561 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":3},{"id":"513e215bc8ef5f99d5d31031a74847a855228932","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-26 02:50:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(18 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9f738f7d7e1a2a7935efecef539543ac2445f87c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-30 12:26:44.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8864523008cb73349d3ae0cd1d8a729b765f73f4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-30 18:25:57.000000000","message":"Patch Set 3: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bdbce78a3e17453b0804e24c2e73dafcc2bcdbe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:26:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"538a5c9694c3bb6b35572f6c46fb8371d41aecaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:43:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2640dd66d746728ee228b9aa7f7bfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b18005c24b4cdf929dae5c278bb4bf : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/344a44309ce5456898f5fc9c4f359547 : SUCCESS in 46m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a7fecdeff844fe0ac37488dabf79bc0 : SUCCESS in 43m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/714d858f87d941e3acd4caa66e3ec322 : SUCCESS in 37m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c151316e607940c1a16327b272c96fad : SUCCESS in 1h 11m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b8b9a45547b44a30acab8bd1265f8c84 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa836aba8f824e67ba7f4c9a3933bfff : SUCCESS in 21m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c7bc6edde00a405daba1f115ceea5609 : SUCCESS in 17m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cbd1e91de1da4cbda16527d6ed99b073 : SUCCESS in 20m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7d7e43778aaf4b7b9bb8ab21e4b5f585 : SUCCESS in 18m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/98a7b6e8136446318da9ff9ec08d69d5 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/264ac4c9924a428cb2b162320fcd83a0 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab66310fdb214a84934a13d7684821e3 : SUCCESS in 8m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed3a5842895c49109d7f502129917926 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e91c3662b9dd4242b17614a8b3514bfc : SUCCESS in 5m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd65888d8e9849758165ef29609a961d : SUCCESS in 41m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9078f9bd68104b0d930e908a446702c6 : SUCCESS in 56m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c11edd622604ce6be26331dc5878efc : SUCCESS in 36m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbfde79b46664deebca6d4fdf018666d : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f75a8b1369423293c4bc7ecdc40b16 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5fe30a287309418ea41856ee1b53b4fd : SUCCESS in 16m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41aac90633e341e4ba79ebe06cd4c175 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":3},{"id":"0059663c20dd74d19e374cb9d38c7bea99b1777f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6e2706c9299bef85f098f7143c826756c020f5c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:44:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba65cfc9a7b746dabbb97669cd0b7212\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500c528aaf5743d780e64b22528dafd9 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/61047760b0404605b91e8015149e8a55 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2ac36a80178cd7d6fb0852d9869f4dce9b53f1e9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 10:34:18.000000000","message":"Patch Set 3:\n\n(11 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52abfba659d1b08c65f7a6dc4d8267c02d0b450b","revisions":{"d06cf74a36ea23e3645d5c3a51f9dc0caceebf50":{"kind":"REWORK","_number":1,"created":"2026-03-20 15:01:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/981531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/981531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/981531/1"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 14:58:32.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 14:58:32.000000000","tz":0},"subject":"WIP: versioning: fix version listing marker name","message":"WIP: versioning: fix version listing marker name\n\nTODO:\nlaunchpad bug\ntests\n\nThe marker should *always* be translated to the equivalent versions\nobject name.\n\nPreviously version-id-marker\u003dnull would cause the version marker to be\nthe original object marker name which will sort higher than any object\nion the versions container, since the latter are prefixed with\n\\x00. As a result, no versions would be listed.\n\nChange-Id: I4b619bc20316eed61ca257a151fcca4f18c9af76\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06cf74a36ea23e3645d5c3a51f9dc0caceebf50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06cf74a36ea23e3645d5c3a51f9dc0caceebf50"}]},"branch":"refs/heads/master"},"724697dbc4499191645274ef3f362005b4da8b1e":{"kind":"REWORK","_number":2,"created":"2026-03-23 15:50:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/981531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/981531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/981531/2"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 14:58:32.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 15:49:59.000000000","tz":0},"subject":"WIP: versioning: fix version listing marker name","message":"WIP: versioning: fix version listing marker name\n\nThe marker should *always* be translated to the equivalent versions\nobject name.\n\nPreviously version-id-marker\u003dnull would cause the version marker to be\nthe original object marker name which will sort higher than any object\nion the versions container, since the latter are prefixed with\n\\x00. As a result, no versions would be listed.\n\nCloses-Bug: #2145423\nChange-Id: I4b619bc20316eed61ca257a151fcca4f18c9af76\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/724697dbc4499191645274ef3f362005b4da8b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/724697dbc4499191645274ef3f362005b4da8b1e"}]},"branch":"refs/heads/master"},"52abfba659d1b08c65f7a6dc4d8267c02d0b450b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-23 18:39:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/981531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/981531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/981531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/981531/3"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 14:58:32.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 18:39:08.000000000","tz":0},"subject":"versioning: fix version listing marker name","message":"versioning: fix version listing marker name\n\nThe marker should *always* be translated to the equivalent versions\nobject name.\n\nPreviously version-id-marker\u003dnull would cause the version marker to be\nthe original object marker name which will sort higher than any object\nion the versions container, since the latter are prefixed with\n\\x00. As a result, no versions would be listed.\n\nCloses-Bug: #2145423\nChange-Id: I4b619bc20316eed61ca257a151fcca4f18c9af76\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52abfba659d1b08c65f7a6dc4d8267c02d0b450b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52abfba659d1b08c65f7a6dc4d8267c02d0b450b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
