)]}'
{"id":"openstack%2Fswift~981532","triplet_id":"openstack%2Fswift~master~I2e01264b176ff05131b4dcbbda32aa8a3901634e","project":"openstack/swift","branch":"master","topic":"p-s3api-fix-next-version-id-marker","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-25 17:45:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e01264b176ff05131b4dcbbda32aa8a3901634e","subject":"s3api: fix the version listing NextVersionIdMarker","status":"MERGED","created":"2026-03-20 15:04:46.000000000","updated":"2026-03-25 17:46:48.000000000","submitted":"2026-03-25 17:45:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"981532-p-s3api-fix-next-version-id-marker","meta_rev_id":"73a780a236d1f3ae48d5a5bd0076e63ffe34e98c","_number":981532,"virtual_id_number":981532,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-25 17:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-25 16:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-25 16:42:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 17:08:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-20 18:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-25 16:42:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"e9bbd5296eab8e4be45eb9d303e22564e23c752c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-20 15:04:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"444cc61d0f9d9b7b67b917b2d91085195362c214","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 17:08:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ddbd78de8447d78d769b95ec25ce06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a341e72a21a40cebe2bec30b85b75f8 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"abb858cff440b252a69a3ebff66e5d3874f240ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-20 17:13:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dab3437ca52945fce57c295970fd652cda1d6cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:18:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/340ce1208b8a4243b06a7e32564e9170\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f3242cc4bdb4d04a7744f6211ba4649 : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":2},{"id":"702f5d67e9249e442498b82d3c77b354689027c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:25:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0327c242c6c4329bbdd11f4f1eda7f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219b28e78c044db38987d0878006b8c7 : SUCCESS in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9b1265dbb7ee47ce9aced25bca8be81e : SUCCESS in 47m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a8f9ed68fb5940a5b9b7e544615d24e9 : SUCCESS in 57m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79d64908c48f4d7f826ca84e159ca68a : SUCCESS in 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cc76f96659b4ea38ba3cdcf672c2eac : SUCCESS in 33m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ee766d5026b428ba9bd1d4dc808de6d : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/538ed765e97b4784bbc71ac5d4851c08 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/baee7e41465a4426ba162b31fb0e5c6b : SUCCESS in 12m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a19fbce93bde438ba18b136d2e9a4b42 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d081e4879b544f928e6a42d9831f0350 : SUCCESS in 18m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f7cdc71d50384ed3a13842b81624692f : SUCCESS in 21m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/42e167ebb3774e4facc65d872e670428 : SUCCESS in 12m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9282ae1ec8014ed3bac96070fd7d1dc7 : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d91917981a8a4c8981a0b5f6715c4cb0 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/02b59fbb8fc64d01858478c64cbc5695 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b3847f70954246ea9a8e91ee70d024f1 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f6c497e9cad4458a0c880de341b4368 : SUCCESS in 3m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6882d90357b448a484e5e06c5ee726c5 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f20c2836b2cb4713951dda9dbdab2ec5 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed1e07a72fa84f47b184dde86547363c : SUCCESS in 1h 01m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76adfdc91bc547dfba15516a71fb9ebf : SUCCESS in 59m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37ad6328c6de493f8d21f35fb6fde91e : SUCCESS in 41m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8f4e23226fc40b6acafad36f7cb028b : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c19a70b4d44400d8f03fe5f55b3f7ae : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10f385f93ea444e5af682f55894b7c74 : SUCCESS in 18m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18cb827ed7264decae39b3aaf75d2ff8 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":2},{"id":"860457e134c877d799ad63fbb97340bdef12e7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 15:40:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"de742705dc0a141ce462dec1012a5ed2ad014bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 15:44:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c93a595408f9c9a7a61ffed70999dc3b955309fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 17:17:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4c7f2ed5c843ee9908567da95c7998\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de3018a8efc4ab984eee5c885bca171 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/4d298a7cfee849cdbd67588bea9f2b51 : SUCCESS in 1h 06m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf09543cfe0c4986ba031fdb5ae003bf : SUCCESS in 54m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/408fa5c0e442441898fce2d375f92338 : SUCCESS in 33m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ded8a5d0245f42fdb5addd53d31151d9 : SUCCESS in 1h 04m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/180c064ecd154e3b9cb40b2cfb4c526a : SUCCESS in 23m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8db038c023a4552b5087aa13a92c9fb : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93f4418c0fc049aaae22269646327e29 : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6bcfb606d484c4f93506deee2fa4cfe : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0448fe71b7a84c2288b63d6242a76f62 : SUCCESS in 26m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0fe32f8fa16041cfabdf3eb55e524d5a : SUCCESS in 11m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/280eda90cb9b4a8d9caa2790c482943a : SUCCESS in 27m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/317e19c7e1d543e1862ae8c788d04c50 : SUCCESS in 21m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fa50eb3085804435ab3ec64365f9d268 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cc3ae72cb0644b42a204ac04ee49f574 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c2f05ebcb674ec0a6acb5be354c2939 : SUCCESS in 13m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c941e5b7de843a7833218865bda7584 : SUCCESS in 7m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9eaaf00cc3234cfaab19166ffd08aaf1 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a544b22347514112926b74147a3c1c7a : SUCCESS in 12m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/876f94e91fae4e9185c47d00bc71baa7 : SUCCESS in 1h 00m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8975e6cc8204b69b8b094c7476e6655 : SUCCESS in 52m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5567c4f02494d8485fce88412c0a7df : SUCCESS in 49m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/557770155e214fcfbe9cf7b0325dad7c : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f80665c7b747fda28e4b208c1a9da4 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ce211847f16415ab59e408310b1b6aa : SUCCESS in 19m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24fb48bab02e491b95b2109bdf61ab1a : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":4},{"id":"b43abb514a465fe1acdb75c5835bfad9de2e3890","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 17:36:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd29b4e2cd744d184cd32a66d2a7665\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/416b54e7aaaa4e7aaa860258aea58332 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":4},{"id":"15d428a88da80db4c81a6fa43f9b204ecd787c07","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-25 16:42:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e7b0606d32d1b60fb48c6f0d350d374c405df581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 16:43:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"edbcabf783f938b9aa95fc9f99e9e3752c77495b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:45:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece8b14b079c46dbb8fc2ea7b5146fe6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecfdfa21ad249e2817c25233e95514a : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/bab16655fb6445d49911efb06fa8ab18 : SUCCESS in 56m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c054ed592e884f29bfc9eb0afe1b6231 : SUCCESS in 57m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d366d2455b94938891e5ca06f773789 : SUCCESS in 33m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/925a521d7c1b42cf941f834a5cf48b48 : SUCCESS in 50m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/12a439197fd4460da4402312a68057bc : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0eb5ace4dff2449089ba50ac70bd61ed : SUCCESS in 22m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/53fb65ed73b14ca1bb25c92468498ce1 : SUCCESS in 19m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ac401ac41617456a8df0b9239fbb18ae : SUCCESS in 27m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/41fefbf24ce9412d81de968ec70db76f : SUCCESS in 15m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/267945e965d0479fa234c5d365532777 : SUCCESS in 8m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/205ee7251e7c4ec2b7927a8ba33b90e7 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d04d8658017344078037a059ecd0fb72 : SUCCESS in 10m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e20241067f544cd6a30f329549657e47 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b4eff508781f4e7bbbb457eee2e58cff : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/47ea5cb4053c4dfabcbf65608e84ff03 : SUCCESS in 36m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67295f2cfef948d5890f90ba2b09db77 : SUCCESS in 24m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ab4ce12fb82486592a646d53c7fda4e : SUCCESS in 25m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b91e9e281cc9454aab900481fc9ebc9a : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b639ddee90f74c4a9c201a6ad1d1ab2d : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00e765b5c059492283011cf9f1958272 : SUCCESS in 17m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c22080be758b4dfab7ea5e18558d6fd9 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":4},{"id":"caf780bf9d4e176fad170992b0e572fbe056b7cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:45:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"73a780a236d1f3ae48d5a5bd0076e63ffe34e98c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:46:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad0f9f364f2e4baeaf6cb18c7457f832\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b369548af742fda2d622e46178de1f : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c6b2c4dc880c4610b73ecc7c80b1b654 : SUCCESS in 1m 07s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"95ca0af6cf5dbfbbb54bd390f9c86e2f0b188a9a","revisions":{"543f9f0743532bd8b2433a0972959a8087b77914":{"kind":"REWORK","_number":1,"created":"2026-03-20 15:04:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/981532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/981532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/981532/1"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 15:03:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 15:03:20.000000000","tz":0},"subject":"WIP s3api: fix the version listing NextVersionIdMarker","message":"WIP s3api: fix the version listing NextVersionIdMarker\n\nTODO\nlaunchpad bug\ntests\n\nThe NextVersionIdMarker should use the Swift version listing item\u0027s\n\u0027version_id\u0027 key, not \u0027version\u0027.\n\nChange-Id: I2e01264b176ff05131b4dcbbda32aa8a3901634e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543f9f0743532bd8b2433a0972959a8087b77914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/543f9f0743532bd8b2433a0972959a8087b77914"}]},"branch":"refs/heads/master"},"f2ac580b3af1c92ebee75dd87bfd36e007b6f55c":{"kind":"REWORK","_number":2,"created":"2026-03-20 17:13:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/981532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/981532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/981532/2"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 15:03:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 17:13:12.000000000","tz":0},"subject":"WIP s3api: fix the version listing NextVersionIdMarker","message":"WIP s3api: fix the version listing NextVersionIdMarker\n\nTODO\nlaunchpad bug\ntests\n\nThe NextVersionIdMarker should use the Swift version listing item\u0027s\n\u0027version_id\u0027 key, not \u0027version\u0027.\n\nChange-Id: I2e01264b176ff05131b4dcbbda32aa8a3901634e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2ac580b3af1c92ebee75dd87bfd36e007b6f55c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2ac580b3af1c92ebee75dd87bfd36e007b6f55c"}]},"branch":"refs/heads/master"},"7064913073a11c72531751e9903b711f833291bb":{"kind":"REWORK","_number":3,"created":"2026-03-24 15:40:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/981532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/981532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/981532/3"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 15:03:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 15:39:54.000000000","tz":0},"subject":"s3api: fix the version listing NextVersionIdMarker","message":"s3api: fix the version listing NextVersionIdMarker\n\nThe NextVersionIdMarker was erroneously translated from the Swift\nversion listing items \u0027version\u0027 key, which does not exist. As a\nresult the NextVersionIdMarker was always \u0027null\u0027. The correct key to\nuse in Swift version listing items is \u0027version_id\u0027.\n\nCloses-Bug: #2145449\nChange-Id: I2e01264b176ff05131b4dcbbda32aa8a3901634e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7064913073a11c72531751e9903b711f833291bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7064913073a11c72531751e9903b711f833291bb"}]},"branch":"refs/heads/master"},"95ca0af6cf5dbfbbb54bd390f9c86e2f0b188a9a":{"kind":"REWORK","_number":4,"created":"2026-03-24 15:44:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/981532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/981532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/981532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/981532/4"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-20 15:03:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 15:44:27.000000000","tz":0},"subject":"s3api: fix the version listing NextVersionIdMarker","message":"s3api: fix the version listing NextVersionIdMarker\n\nThe NextVersionIdMarker was erroneously translated from the Swift\nversion listing items \u0027version\u0027 key, which does not exist. As a\nresult the NextVersionIdMarker was always \u0027null\u0027. The correct key to\nuse in Swift version listing items is \u0027version_id\u0027.\n\nCloses-Bug: #2145449\nChange-Id: I2e01264b176ff05131b4dcbbda32aa8a3901634e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ca0af6cf5dbfbbb54bd390f9c86e2f0b188a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ca0af6cf5dbfbbb54bd390f9c86e2f0b188a9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
