)]}'
{"id":"openstack%2Fswift~981611","triplet_id":"openstack%2Fswift~master~Ie284481139401acdc520db4c68fad3171ae8fb50","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-05-05 07:35:01.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-05 07:35:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie284481139401acdc520db4c68fad3171ae8fb50","subject":"tests: Allow probe tests to pass with encryption enabled","status":"MERGED","created":"2026-03-20 21:57:35.000000000","updated":"2026-05-05 07:36:03.000000000","submitted":"2026-05-05 07:35:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"981611","meta_rev_id":"2b18139187e29a7c4f5234edd5238f0a6b626bf3","_number":981611,"virtual_id_number":981611,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 07:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-05-05 05:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-05-05 05:11:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 22:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-20 23:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-24 15:56:05.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-05-05 05:11:13.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"5dde98d5b6557e073e85990e5841ec5222056611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 21:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"925d24caf63130948b9584b0c403d3159daa56b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 22:19:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2108a232deada1fd73cd2218fbe4e512687366d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:49:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3a2786e54f47bb9f01ceee4a9865ab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8669e6722b574e6ab76432f8b4c04668 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4aad4741e3a41b025ab8c63528c06735749cfc0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:11:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee48a0e8baa64f63a4e44015f6a85756\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f25e3e2a473449ea5f92605da8458a9 : SUCCESS in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/0d1cf7047dac40cdae20fd45f3b7c368 : SUCCESS in 29m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e11dd1f28a95406d84584f800b01019f : SUCCESS in 30m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d5ee4fbded0492b820ebdbfb810e2e5 : SUCCESS in 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/533f739452a44e6ebe12d642b7fa492b : SUCCESS in 50m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/281e7e54fca24e83abb23a78cb1c61bc : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fddb40ef32f34ec681363c45a44f60ff : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/611e3db1ec9e494681ecbd689420dfb2 : SUCCESS in 17m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f0206bc7094489a831fcd31852106af : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d0a2b586d6594df197bfa0444249b13c : SUCCESS in 11m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/703ead59a33e4f18a6c89f8f52bbc058 : SUCCESS in 8m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b6af1d5fd7ae4398a2def2ec0d0f20da : SUCCESS in 23m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/18369a9cd30445eca8aacb967e0be1c4 : SUCCESS in 15m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/11f4be5151be4190a366e01753bdc169 : SUCCESS in 8m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/468300d1354a4237b7b6ba156de3b872 : SUCCESS in 8m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/81cf21cf3ec046b59edbbf1a128e19a5 : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/caf997158b9940579fdf189cc6930a6d : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9109a17d2e84c8d99246548937bb20f : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/db2e05980a56459eb399b05e1eeef6a9 : SUCCESS in 5m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fab711df37cb47e1bb6d5dc5fa307d95 : SUCCESS in 35m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2e49d8f1ae84755941465d163206ebe : SUCCESS in 25m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2b1d59e554c427d9f08d5f23363bf26 : SUCCESS in 25m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5658643d4424beb8270c921b22ce9a7 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a224bf67c64cce85dad293a3c64472 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12287fd65acb4bff88b71d516c9ab693 : SUCCESS in 8m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af1fb64826c74e749f6bbf1435c348ae : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9c04e56ea3804e07bbee198ff2a4f75270be39e2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-24 15:56:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5c59e5cdc1c647579cdf2cdae07d3b3e68b37f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-25 21:15:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7bef975a47515633516a3e61f92d2210aade33a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 22:27:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f6b50aeb5e444a5bdbb2300f59a245d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5022f045f64e1aa756a5b25c9d3b81 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/074fc9eab00b4ee281d3adbf4fa52963 : SUCCESS in 1h 04m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4a8a602b33842cfbcbc3a75824e3e83 : SUCCESS in 1h 03m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ade5b63a77454977a45264ef4a428d00 : SUCCESS in 54m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e918e56e74e04faeb7b1f26b9a422dc9 : SUCCESS in 1h 02m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f066f79949e4bcb9e44e95932d3f41f : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a58fb35229ea41dda52935a3a5a37af2 : SUCCESS in 26m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03257d6e2b25411ca425f633ad6c7c84 : SUCCESS in 23m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b93680ca1fb48f7bdc17cf7b2c85134 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/42644f7c75614f3e9b5482a2befc8c48 : SUCCESS in 26m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e481ee76d29e4116a38bd8e0c4d5d3d6 : SUCCESS in 21m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5c9cb8667f214ffaba1c371bc77f0504 : SUCCESS in 32m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1dd790157d6641a68041b0016490f1cf : SUCCESS in 21m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d4311eea1e3b4e18b1178a7adab3eca5 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f991976a2ce347fa8f1abd0c51b78963 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0da4d322d7b544d4aa9eafc03d3df2d2 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3cebc27e1c8416cbc85f5c6ae174bf5 : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/718df683b8134efda27f72a6151d1a91 : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/22356f2070574c7891c689a66ab61ea9 : SUCCESS in 7m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b727d0041054456be91bb07b894b809 : SUCCESS in 1h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/baa945ee15ec4ac2b1aafe14c9c54672 : SUCCESS in 48m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/122a1c3798d949059e55abedc167ef47 : SUCCESS in 53m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18d685f0cdd044dd9658252528598349 : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a79a76715bf4d74a5a061efef0bbee8 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dbed61e715b542b5aba113af8bc880c0 : SUCCESS in 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d34861ea28d47baa53ae3df940518d4 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":3},{"id":"59cd9180da165386c9f81bc3aa7b96f6263a3462","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 07:29:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e754ef212b8c44cea892ca151bd4628b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5df08616aa2245df851cc5e0e72719ba : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":3},{"id":"432a16984e6cc0644ec224f0c1999baa9b42ebfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-20 22:17:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ee23d1bd174bbe88038a6edf6177eb4729df5660","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:15:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f49a44ebbb545d0a942e266a91e7b34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc74b07e21f84aef814950381ef12036 : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":4},{"id":"b0855b469f2485c25bde77d3358334e72f59c611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:31:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110c58f218bc43339ea96d17054ed876\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9cadcde9f648b59d57e12aa37ddc12 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/59dee6d26d824ccaa48f6328ad7a373c : SUCCESS in 1h 08m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53ca5d6a0d1840198205ec10bbcb3b11 : SUCCESS in 32m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7981c3d252e641dd8430df6fefdc84b3 : SUCCESS in 32m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be57e1bf744b48ba80e1af0a6ea9008b : SUCCESS in 39m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f05c3125f730491e8e0f8d6f174afc4c : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7692dafe5d841d08497929ea0360dbe : SUCCESS in 23m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8381a4788c0d407eab187ec6e42593d6 : SUCCESS in 12m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea8e05e8219949e382daf431dfc4917e : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/176decbefc4b4399a7d095c1343fb093 : SUCCESS in 26m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19a1a55bfd834b4aa8b2d9a02e97a6e4 : SUCCESS in 21m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2d856b8b18f84818b84961698c32ec61 : SUCCESS in 29m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c8142e34b25f476bb54ec94cb673799a : SUCCESS in 15m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/caaa160f7a2b4121927bb7987e0d12bd : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/697faad921dc4cac96bcd3b4b53d5ba8 : SUCCESS in 9m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/185e369595894933a1039e409d3bc5f9 : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53de5665d7fd4d43a43850961654ba12 : SUCCESS in 6m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a147dfd552c744bfa06ceadcdb8b5bc2 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/89bc6510e63f40d3b446e09af90a64dd : SUCCESS in 9m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e9c0c1a73a64baaaa3fca715a4d234d : SUCCESS in 39m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48a20842b2014186a72892f4c5785047 : SUCCESS in 54m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78ce81adcecd4752b3f365dc7aed3cd0 : SUCCESS in 49m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b301f26d2b45b98b9eff10df26f62d : SUCCESS in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39aa46c23344293837160068e9f2b19 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0f32efaa6644febb15792f40300c3ff : SUCCESS in 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09af91fb3e1749ff9a5a9cf6a9ec5dc2 : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":4},{"id":"64aaabd130dccbf6704eb13c934199f35a73fddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 15:38:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"43462dd606ac2517922a84fd08b2d2d5f980dc55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 16:08:54.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c8c0a9a31f490fb1d1055c0fac4f477fb3654ab3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 17:21:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240a6f47d9494497bd3f2f2b9dcfb227\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d97866e6a3ad4bd4b03c3d3430c54d55 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":6},{"id":"097efeca4e8b8db258a7d3790f740ce28bc3c289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 17:24:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45a41d047f09428ea8e33e6020bea684\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c4c7a06b1a4c0e9bc0893416a9e488 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c0e6cfe81a474aac8ffccf0517e273cd : SUCCESS in 1h 08m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b132f623ff054e808b3fab3b5118cd20 : SUCCESS in 1h 06m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aed71044149c40e3b8e4c5173f1c66bd : SUCCESS in 48m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d0f36d55863495696037025bf0338b9 : SUCCESS in 1h 03m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b56d576f827434b804c73b3e4a46f89 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac94eff2abd249259d41430fcd438036 : SUCCESS in 13m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/766e11da0bcd4050bf59ab3cde518e59 : SUCCESS in 24m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9170e0125dc842228b9bbb8da75dd17c : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf19230f73ac47ae826fcf377d1a4205 : SUCCESS in 10m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2cee14d0d52e4dd0b37b70a19a172160 : SUCCESS in 22m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/60fe560eec454cff8c22f1488219cfd7 : SUCCESS in 11m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d5f104cf2fae416ab9a6fb0486e781b0 : SUCCESS in 21m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e8d79fa1822f4d3d8c5b93f8f565a40d : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/70a13d9c97ae4e4c8e55fc4d6585f933 : SUCCESS in 12m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0c8cc82913e3471cb90cdf09b1a817fd : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec7f22a220634363871b702c584b07e6 : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a74ca0998a54653934260b1d0b69ded : FAILURE in 3m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/71d17ff2a1274cdbb33268a9ef2fb06b : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/18a05ebf55514db1a0795e10d84430cb : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d6f4f7e746e40cab3689420c197ff46 : SUCCESS in 37m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75e978b936864364b5bc13659c5048f2 : SUCCESS in 56m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50aabea0aeb345e59ef4809aee408994 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed77e51e1b564e83a99f5692a0827891 : SUCCESS in 2m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/daa0323ef88146a49676b66a28037a40 : SUCCESS in 19m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fc2e365f1984b1dbd377109bfa94ca5 : SUCCESS in 49m 07s","accounts_in_message":[],"_revision_number":6},{"id":"dca1681fa2966b8dd04dfd3e73dc327bf58827e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-24 17:46:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"01775821049426ee0574c99544ed285bfe0cb215","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-24 18:53:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"79c3829d505559dcb5961bde42852237b5aa0ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 18:56:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1cdd71396e640eca92f58d6a9f3dd79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c343bfc7aa234a4faf8e7ae6e2682a44 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/314c027c9c22481b9debe54b6ce78aab : SUCCESS in 1h 06m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1243dc98f6f74f34a340b6329bf5fd39 : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/582128dc69384996906a20acbf321d87 : SUCCESS in 56m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1b2ccfbd0d1462fa090739ed587cd25 : SUCCESS in 1h 07m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7fb6d313cbf41278be4f800df8d179e : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0273c569c544ea4a7302f5d3c6395ee : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10b243a8f39a4690b60e4eeea8ee7ad3 : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c15a6938d574e46a077c0a58f386b1c : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc00656de90a43ab86e3669bda17d119 : SUCCESS in 29m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/47b9e6ae23f74365acc0a26e4b7817b8 : SUCCESS in 22m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eef433d67fda40c098ed9834d91de84f : SUCCESS in 26m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/55e7e7e2aadb43729d4eac46ac7209a0 : SUCCESS in 21m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3fefa6d18ab04fdabdb95d24b18fd337 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2ba20d877908452c9daec36522ac9dba : SUCCESS in 16m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/535e098620e44b709aa32e2e054424e5 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/15347ffc8b004c17a62a11859a51cd8b : SUCCESS in 6m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81650fe299ce43598b80e7c918e3e8af : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cdd54067a1e24deea230c67f510f0f12 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ed90a8adafe48718b0b94e4428e7065 : SUCCESS in 35m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c32d9f9e55e43c0b266b6722eed8183 : SUCCESS in 59m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95221448dbf04a2e9750f032f94f988e : SUCCESS in 55m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3076696d150b4dce83e9a274fedcab16 : SUCCESS in 33m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c6acc890c44881a1521beb54f04461 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d1ccf2f8cb5547c8853035bbe4f8d547 : SUCCESS in 27m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2a7846c79494002a1895a03df7daf23 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":7},{"id":"33641e2f9ccdd97c081fc9066c138da81aaa45af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 21:48:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2276b79dacfd46c49f2941b84a60706d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4220f04c2755460684d61f51c2fbdd5f : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":7},{"id":"ff514d95a76e3feb7e7a00190775d7739272ae20","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-05 05:11:13.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"dd12ec5a3822f1d225c77caba076ab8f416d7f2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 05:12:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0991e127ce3b7eebe948308a309b6bc8d3f8bc6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:35:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8e8f86b5548471fabcea83e1135b809\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a199f8bff814425aa8d46c66b60ef200 : SUCCESS in 5m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/e37b77cf86bd418e930adb5eec8bc923 : SUCCESS in 1h 14m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/516f1c8b04324d119c3293f8548aee2e : SUCCESS in 1h 04m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dfbeca3036c045aebda545bf66e4ea22 : SUCCESS in 33m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46940f76f58c46d3ab4cc3607aee4ac3 : SUCCESS in 1h 03m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f00a9d2c93cd430b8edc0a88dd0b8121 : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4116f2a044f045aab47859b70117455a : SUCCESS in 25m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a539d0c13c224d6198eb6708f9f8a839 : SUCCESS in 24m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d2f680160a1a474c97e4eab1960e00e2 : SUCCESS in 30m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5e798fcada0a43c797db8b70852a4e2d : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/401812fb1a4043e59aa04280e0269c5d : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/166e4b200e5b4b4281fa2d863c0847f1 : SUCCESS in 17m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/662bf62a0ee14c7dae37ae979f0a7ffa : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c0fa757ef7344b78b223358be023a1a : SUCCESS in 5m 13s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ec3e4d62bd843f8a1f25a0e3b4ce738 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a675af8b8db44b39a25f97140f374bf6 : SUCCESS in 1h 02m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c12b52df20a242abb279571aac386d62 : SUCCESS in 22m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2ccfd55ded674f5d9a8b073ec71fcee6 : SUCCESS in 34m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b49d53c4d5a45c588c122db31e7dc48 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829b776c7ca14cc8a7dfb0d971444c31 : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d14918c05d8a4be0b90dd291f624f0a5 : SUCCESS in 18m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c541b11a021a42efb1302563ac9542d5 : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":7},{"id":"121b04b3d902ef4f2f773f63de3215f1b8218429","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:35:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"2b18139187e29a7c4f5234edd5238f0a6b626bf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:36:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f163fd3882ba4a15849ab2b7843cba77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f66d2e14be4444f802eb6f3d2260122 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ae17443ce6b24ec38f8de7d85ec67246 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4f37e5ea777363b54fa473623dc1cc9f25a76b38","revisions":{"5127a4ad2312a7b736651db8b8dba6cfdfd5f377":{"kind":"REWORK","_number":1,"created":"2026-03-20 21:57:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/1"}}},"commit":{"parents":[{"commit":"84d29362bc67a237e0ae222491d0b71a8fc82b1e","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84d29362bc67a237e0ae222491d0b71a8fc82b1e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:57:20.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5127a4ad2312a7b736651db8b8dba6cfdfd5f377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5127a4ad2312a7b736651db8b8dba6cfdfd5f377"}]},"branch":"refs/heads/master"},"2102492e29b028be18fcf017dcad20224bb47534":{"kind":"REWORK","_number":2,"created":"2026-03-20 22:19:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/2"}}},"commit":{"parents":[{"commit":"fcc316a95c46412860b9d577657f2b5f1065f594","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcc316a95c46412860b9d577657f2b5f1065f594"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:18:54.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2102492e29b028be18fcf017dcad20224bb47534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2102492e29b028be18fcf017dcad20224bb47534"}]},"branch":"refs/heads/master"},"5ac30a408d2429bda50d7604ac3ad68d2f9f737c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-25 21:15:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/3"}}},"commit":{"parents":[{"commit":"6871e28ed6807838605f26a21583ba0e0466dd9d","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6871e28ed6807838605f26a21583ba0e0466dd9d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-25 21:14:31.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ac30a408d2429bda50d7604ac3ad68d2f9f737c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ac30a408d2429bda50d7604ac3ad68d2f9f737c"}]},"branch":"refs/heads/master"},"2332a7696804b0a6154293390a88441f68ee27fc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-20 22:17:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/4"}}},"commit":{"parents":[{"commit":"4d7825132682549e479d80cf7d1896ba46a72a09","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d7825132682549e479d80cf7d1896ba46a72a09"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-20 22:16:47.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2332a7696804b0a6154293390a88441f68ee27fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2332a7696804b0a6154293390a88441f68ee27fc"}]},"branch":"refs/heads/master"},"176e2d0e4f4d0c344f52614fd4f390378bc729f4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2026-04-23 15:38:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/5"}}},"commit":{"parents":[{"commit":"9b19594055260f9b8caf0672210ba8beb4aa326b","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b19594055260f9b8caf0672210ba8beb4aa326b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 15:37:20.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nDepends-On: https://review.opendev.org/c/openstack/swift/+/985896\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/176e2d0e4f4d0c344f52614fd4f390378bc729f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/176e2d0e4f4d0c344f52614fd4f390378bc729f4"}]},"branch":"refs/heads/master"},"9883159cbfff35a3b68d31da4220b22404a87237":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-23 16:08:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/6"}}},"commit":{"parents":[{"commit":"ab106e5f65561748c8bd8c0e4a1a3771f5f80aa2","subject":"Close DB connections in metadata servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab106e5f65561748c8bd8c0e4a1a3771f5f80aa2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 16:08:46.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nDepends-On: https://review.opendev.org/c/openstack/swift/+/985896\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9883159cbfff35a3b68d31da4220b22404a87237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9883159cbfff35a3b68d31da4220b22404a87237"}]},"branch":"refs/heads/master"},"4f37e5ea777363b54fa473623dc1cc9f25a76b38":{"kind":"REWORK","_number":7,"created":"2026-04-24 17:46:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/981611/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/981611/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/981611/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/981611/7"}}},"commit":{"parents":[{"commit":"3948d71adc6a25d4671d1a81eb2b25ef127e9bf4","subject":"probe tests: Pass when there are multiple EC policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3948d71adc6a25d4671d1a81eb2b25ef127e9bf4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 21:30:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-24 17:46:00.000000000","tz":-420},"subject":"tests: Allow probe tests to pass with encryption enabled","message":"tests: Allow probe tests to pass with encryption enabled\n\nGet rid of the hard-coded internal-client config in probe tests; use\n/etc/swift/internal-client.conf instead. However, probe tests require\nallow_account_management be enabled, so update the sample configs to\nreflect that.\n\nChange-Id: Ie284481139401acdc520db4c68fad3171ae8fb50\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f37e5ea777363b54fa473623dc1cc9f25a76b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f37e5ea777363b54fa473623dc1cc9f25a76b38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
