)]}'
{"id":"openstack%2Fswift~981612","triplet_id":"openstack%2Fswift~master~I03afdbcf78e65d00c8bea988a19a93ce7b746baf","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-05-05 07:35:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03afdbcf78e65d00c8bea988a19a93ce7b746baf","subject":"tests: Enable encryption for probe tests in the gate","status":"MERGED","created":"2026-03-20 22:10:22.000000000","updated":"2026-05-05 07:36:56.000000000","submitted":"2026-05-05 07:35:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"981612","meta_rev_id":"32fa8984eb65661a8251314bc545eb6c332cb659","_number":981612,"virtual_id_number":981612,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 07:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-05-05 05:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-05-05 05:13:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 23:11:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 05:13:24.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"8f05604177f1de12e10948423569ff2623a8448e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 22:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4fac3f2037079f669156fcf3e6dd4fa718be663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 22:19:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"54616982b2f6dd0f87d831c19c15d3db4523e097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:11:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67551aed9ec34981880a86db5057db8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fa95aadbe1414bab0122a3f2f4ae69 : SUCCESS in 8m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67f0777c7ec84136b2e9146b1f29ddf3 : SUCCESS in 7m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ea33533869b4ae391f8844191d36e25 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6a0fb8c8bda4abc8c53d1faa2a42b35 : SUCCESS in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/783a465798dd4c019439902a7e3f7f41 : FAILURE in 51m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a7da2d1c9afb5c46b8cb17a1109bb1f673d8a7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-20 23:57:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"445bf3ab2e801f8b95a8fd75ea8c13d386536c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 00:34:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc8d82dc4d8d40f4b073859bc9116247\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85436d6e2e304ba6a20c399b4d84cfb5 : SUCCESS in 9m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9f587e8ba954578b7e18b41a3f2fcd1 : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f9e88775c3d428a947dd63b5792375f : SUCCESS in 13m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4a189229f5c64cb08905a8fc7c94a306 : SUCCESS in 5m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4aaa1107a204ed6bd73a2686fb85172 : FAILURE in 35m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4ed45628448e746a72fad66af38622b7718d6d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-22 22:07:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"027c6789b04dc9a56b16b77ce1ac91d8492fef55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-22 23:00:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c01a74ec8dd40f5aa7223e0da8899e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558a5c7a7ed4419eada5600a59bcca5f : SUCCESS in 9m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f890dcc2b04640c8910cf993a6997f69 : SUCCESS in 3m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d75b82a996c41e4bd40616f36b391fc : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/109225a6336a4552931b5b074c3bf669 : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b84db75fb1846bfafbeb25a740dff2e : FAILURE in 44m 54s","accounts_in_message":[],"_revision_number":4},{"id":"e7e49ee29d68735bef2921002eb506ac2ae1fb74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-23 17:02:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b9797c81b43c3207c5bc3cd01f115b6ce5ea7c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 18:07:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2f4c4e2028498da70c48cdd0b0e70c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222049f02a2b4604b037caddcad993a6 : SUCCESS in 9m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/437ab78244544e4a85cb0ec968f1ddfd : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/906caa45378246d591c8b80270f6bc04 : SUCCESS in 24m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e3682ac36e14c58990047b7db128d8f : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39feafdd0c9546e0a9238c1de783caaf : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":5},{"id":"cfe5286d747f44174cbd4aa16176c2b6b43c1317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-25 21:15:03.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8d11dcab4e7e820ca9255117d46cac81e3cac227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 21:52:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4043d47202d44cfbb4cfa6988f49b60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e937ad5967407cb0fdcde74f62493d : SUCCESS in 8m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d6ddb103ec243f18fefd9e7113f7f40 : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0a42da8b70644d5abf56448a86d409a : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f282c68b2c324401af780eaf92485d68 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c5feb29b03b42dcaa7fb525bd1af25c : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":6},{"id":"0a511af37200f364d658fcc4a6329ca4386dfb98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-20 22:17:12.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"32ee28378b77d28c9e9c7d836e1f4f91ac6c38dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 22:53:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8919bbe4370a4f30ad2601f16da0d0db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef4c7e412a444508695b8106171a1d3 : SUCCESS in 9m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ccd464d124f4c3d8d853b32a966ce95 : SUCCESS in 4m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ed6f251d9ce42e5a110701fc0aefb7c : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/20d1326f3fe84be78cd122dbd9834033 : SUCCESS in 5m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fb1624e93cd46c7ae8fb428868be575 : FAILURE in 35m 45s","accounts_in_message":[],"_revision_number":7},{"id":"6094c6de3e951aae2d373ad2a78b92bf6b2be167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 15:38:01.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d5134c145212f1e2a441b3d08cc88519116092ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-23 16:08:54.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"146a6add6d19e5e45b99559e52f0c94cceaf7205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 17:22:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f616264fb9a1439d93319633a587c9e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ca80a56c364f738f9a56354f30905a : SUCCESS in 10m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1190721b71c4aa6a72af471a0f98659 : SUCCESS in 6m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/34ff2c94c13c469282301144d3a4f33e : SUCCESS in 9m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/018286aa3f0a4c6aaa480405fbb443ee : SUCCESS in 9m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/575ed28c9fb44df397a2514823aac3c8 : FAILURE in 1h 06m 42s","accounts_in_message":[],"_revision_number":9},{"id":"adb096c95dd33cf37a9636753a047aaf5bff2730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-24 17:46:08.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"ba5994d7ea0e4c794b2d7c3f9b2bd7b91e2d15e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 18:23:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877058b1d02a40d9a8f8ecef80604b51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6770b275af74311b80c09621f91a562 : SUCCESS in 8m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c012ef27d0a04dcaada10d880bf6e567 : SUCCESS in 7m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f731efe1fee1491aaecb194983d49cd3 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/52947178040547c385df1f63dcd07588 : SUCCESS in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f1a738666f346ef9de1c4d9ef433478 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":10},{"id":"e8d33556ba3ba96a6b35433b915fdff0f803e19c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-05 05:13:24.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"745876a2944ec8bb06045b366b3d718f9c518e57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 05:13:43.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e32886959e1a5ffc2b460ab69e7c6a8daea7ee2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:35:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fb1513eb9e4aa2b44566042d05d73c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf846a5eecd4ab4bcb3462cc094564f : SUCCESS in 7m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a6727080b32410b93e82e4a6ab0bbf2 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/50e994023f904c4f8bbc64bd556f302b : SUCCESS in 5m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff869e1be6be4686bae4a2e9ab79bd49 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":10},{"id":"b6bdecff6657e2b1c08d2fb72f3dd39a8b8fa307","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:35:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"32fa8984eb65661a8251314bc545eb6c332cb659","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:36:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/289fd6bb177e42f585061335bf9803d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c52c78c4e14e279412207e2a5ff573 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b30efda2dc274fb891df572f141211c2 : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8747d5f11f122f2b49688c2c31b6353c06a83964","revisions":{"5a302c9a10b299fa9b3e0cf8954d2ee8f97aec8e":{"kind":"REWORK","_number":1,"created":"2026-03-20 22:10:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/1"}}},"commit":{"parents":[{"commit":"5127a4ad2312a7b736651db8b8dba6cfdfd5f377","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5127a4ad2312a7b736651db8b8dba6cfdfd5f377"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a302c9a10b299fa9b3e0cf8954d2ee8f97aec8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a302c9a10b299fa9b3e0cf8954d2ee8f97aec8e"}]},"branch":"refs/heads/master"},"a73891bc6ec16e17101bed79b1c795083c4b0145":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-20 22:19:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/2"}}},"commit":{"parents":[{"commit":"2102492e29b028be18fcf017dcad20224bb47534","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2102492e29b028be18fcf017dcad20224bb47534"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:18:54.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a73891bc6ec16e17101bed79b1c795083c4b0145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a73891bc6ec16e17101bed79b1c795083c4b0145"}]},"branch":"refs/heads/master"},"be6440b85baa557b840fbff24e6ab4b9a86e235a":{"kind":"REWORK","_number":3,"created":"2026-03-20 23:57:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/3"}}},"commit":{"parents":[{"commit":"2102492e29b028be18fcf017dcad20224bb47534","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2102492e29b028be18fcf017dcad20224bb47534"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 23:57:42.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be6440b85baa557b840fbff24e6ab4b9a86e235a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be6440b85baa557b840fbff24e6ab4b9a86e235a"}]},"branch":"refs/heads/master"},"00713e319cff30470f1ba0567d4631cbcb7360e9":{"kind":"REWORK","_number":4,"created":"2026-03-22 22:07:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/4"}}},"commit":{"parents":[{"commit":"2102492e29b028be18fcf017dcad20224bb47534","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2102492e29b028be18fcf017dcad20224bb47534"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-22 22:07:42.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00713e319cff30470f1ba0567d4631cbcb7360e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00713e319cff30470f1ba0567d4631cbcb7360e9"}]},"branch":"refs/heads/master"},"8b722df51af79dd5f3be578c86382eacc357efbc":{"kind":"REWORK","_number":5,"created":"2026-03-23 17:02:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/5"}}},"commit":{"parents":[{"commit":"2102492e29b028be18fcf017dcad20224bb47534","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2102492e29b028be18fcf017dcad20224bb47534"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-23 17:02:20.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b722df51af79dd5f3be578c86382eacc357efbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b722df51af79dd5f3be578c86382eacc357efbc"}]},"branch":"refs/heads/master"},"59a86a4d021924d84047e262c80890b8355fa408":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-25 21:15:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/6"}}},"commit":{"parents":[{"commit":"5ac30a408d2429bda50d7604ac3ad68d2f9f737c","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ac30a408d2429bda50d7604ac3ad68d2f9f737c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-25 21:14:31.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59a86a4d021924d84047e262c80890b8355fa408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59a86a4d021924d84047e262c80890b8355fa408"}]},"branch":"refs/heads/master"},"12f069cdc562ddab6da66290d50406c4b30fd89c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-20 22:17:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/7"}}},"commit":{"parents":[{"commit":"2332a7696804b0a6154293390a88441f68ee27fc","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2332a7696804b0a6154293390a88441f68ee27fc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-20 22:17:05.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12f069cdc562ddab6da66290d50406c4b30fd89c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12f069cdc562ddab6da66290d50406c4b30fd89c"}]},"branch":"refs/heads/master"},"fb003c6fe7b90e4408f82279f6b9ff045c96f863":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-23 15:38:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/8"}}},"commit":{"parents":[{"commit":"176e2d0e4f4d0c344f52614fd4f390378bc729f4","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/176e2d0e4f4d0c344f52614fd4f390378bc729f4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 15:37:51.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb003c6fe7b90e4408f82279f6b9ff045c96f863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb003c6fe7b90e4408f82279f6b9ff045c96f863"}]},"branch":"refs/heads/master"},"40957121fba54b9a7dce4735ed105844bccab643":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-04-23 16:08:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/9"}}},"commit":{"parents":[{"commit":"9883159cbfff35a3b68d31da4220b22404a87237","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9883159cbfff35a3b68d31da4220b22404a87237"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-23 16:08:46.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40957121fba54b9a7dce4735ed105844bccab643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40957121fba54b9a7dce4735ed105844bccab643"}]},"branch":"refs/heads/master"},"8747d5f11f122f2b49688c2c31b6353c06a83964":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-04-24 17:46:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/981612/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/981612/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/981612/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/981612/10"}}},"commit":{"parents":[{"commit":"4f37e5ea777363b54fa473623dc1cc9f25a76b38","subject":"tests: Allow probe tests to pass with encryption enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f37e5ea777363b54fa473623dc1cc9f25a76b38"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-20 22:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-04-24 17:46:00.000000000","tz":-420},"subject":"tests: Enable encryption for probe tests in the gate","message":"tests: Enable encryption for probe tests in the gate\n\nChange-Id: I03afdbcf78e65d00c8bea988a19a93ce7b746baf\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8747d5f11f122f2b49688c2c31b6353c06a83964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8747d5f11f122f2b49688c2c31b6353c06a83964"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
