)]}'
{"id":"openstack%2Fswift~981750","triplet_id":"openstack%2Fswift~feature%2Fthreaded~Idb6eae3f000aa90c7fa39688cbfbc54c409163f4","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-08-28 18:11:07.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-21 07:54:26.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"Idb6eae3f000aa90c7fa39688cbfbc54c409163f4","subject":"Patch gunicorn to unblock chunked ssync reads","status":"MERGED","created":"2026-03-23 12:12:35.000000000","updated":"2026-08-28 18:11:07.000000000","submitted":"2026-08-28 18:11:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981750-eventlet-removal","meta_rev_id":"932c6dc9b33285ae95ba4a7e6a4b4c2ba1d21698","_number":981750,"virtual_id_number":981750,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 18:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-08-21 07:26:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":2,"date":"2026-08-28 16:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":1,"date":"2026-08-28 16:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 13:25:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-06 13:55:18.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"9b304fe6733b1732734c32568f84e45a1f3e3749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:12:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7da021f01eca8fe9d57b61086ab0f96ea91d7cb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:20:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e37a1b2d557a9266d9ebbd965b464106940323f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:25:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9395db2d862c4f9a9b8dda521c86237f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231804b42b374e4dad2e529d363e229e : SUCCESS in 10m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a65d447b43d442e8bd288a677000f5f6 : SUCCESS in 11m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01cdea17f916406981c103f31b2ac5df : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b975087ce5d4f05add35611da14d94e : SUCCESS in 12m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eaa0567c551c42e691de4ce1978167ef : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a031e7f1c1d4483ea92f99608eac160c : SUCCESS in 25m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d34457687f643e1a15a1b289f2c1190 : SUCCESS in 22m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b4b175866d74586b3e295dcd38a3376 : SUCCESS in 13m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/17716322a7dc4dc294c07107a418230e : SUCCESS in 12m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4a38720cbc584a79874ac48179a1acb1 : SUCCESS in 8m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dd0460e4e2764915bc9ca95f38e65a1c : SUCCESS in 16m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d64a44185a464dff90b84e2d5f1f018c : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/faee3df35e7a4f4c9dfade6a5703c1c9 : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8079d31e858a4ec99df8b723b530e5e0 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a55fb13d94ea43fdbfb24a3be08d42b1 : FAILURE in 12m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b2881a4c9644778b5720e394d2dc9c8 : FAILURE in 56m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dda6dad303854c5081e1825d25e5df0b : SUCCESS in 29m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874fe7453628408581db8f3bf20ffc34 : SUCCESS in 2m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e3c426966fa4d12b3eb9411cbc9090a : SUCCESS in 22m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4778e11a2f39b2e892564955df3e58bdc0ce0c89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:56:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff445d7dd824fa99544a58a47ef2546\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f55cd4d901a94ee299834adf36a48378 : FAILURE in 41m 44s","accounts_in_message":[],"_revision_number":2},{"id":"86c9de176204392a72a7756f25dbc6b74abe5fc2","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 14:01:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72191324f1528856bad68ec1d14dd4d763784203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 15:07:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf86c00286444ac998cbafa1cbbd5bdf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ac809ae07f44238d1dd59b659bd650 : SUCCESS in 6m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f73078d6b4c84dbaba64ec58c2516c74 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c59c530961bf43e6a7d2caeb6dcaa20c : SUCCESS in 23m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4063c723509a4e4aa4b500d265f9dadb : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0db4f88656114331ad8fb763a6b2784d : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eec3f28cdfdd4c1eb255c2344438012b : SUCCESS in 27m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e76911f7ab642888c23803fb2d9f341 : SUCCESS in 22m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ecc998daf9af407b85ae64d9ab98157b : SUCCESS in 49m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/428d24583b9749bd8f20c0040e7e6ed2 : SUCCESS in 31m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1771afd5025c4395921f5c8e480927c1 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3925c0f4b03b4036974071c109f6a1bf : SUCCESS in 10m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/98ccf1f1ca854257b9e738e82063def4 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e76d579a71584a1d95613bccca6dd2fb : SUCCESS in 7m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/abd20020acf7400f937e348a5c3df36e : SUCCESS in 11m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0545f80f55848109da660397da7aefd : SUCCESS in 9m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d0f26958b3d4484aed81fbaf8409196 : SUCCESS in 59m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52137424f4e046fdaacb69a13b164f59 : SUCCESS in 29m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9595db0e9d994949854baef6802161d7 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1da26db2e682470886f0e7a0c4ed991e : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":2},{"id":"16258ae6a9ba77282e377e2466367707acaa42a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 17:50:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635b65f6013742b1a4d5fbb391783415\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f026d2b81d46475dac8d7dabe84f95d8 : FAILURE in 36m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0f1d479e4d91bc71952dd52100a16080869f6a49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 16:41:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"70bbfb9f2a0192521f19206b34d45d00fd16616d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 17:02:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e68dcce7b0916a3ec5748143e13a89bc1399dce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-30 18:24:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"79d1a316987b659ca90ec63b0b1251f5a35aed8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 19:21:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/951b48aac9604d9599757f7766aded0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3cbe1397c3443ea9fc49c76c6404ef : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b64dd0c49414f9bb345ed87e7f744c0 : SUCCESS in 7m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ce6989d05054224bbd1c3ff0e595ecb : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/723fa2b8818d441aab97742b6782e3da : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb0cfba4b86f4941a6ec834611866fa0 : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9173ef89ca204b18acd3a0ed8745cb45 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/78435c68933949c8b9f919a58df85988 : SUCCESS in 26m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9eec5f87f50a46b6907f213fe4983003 : SUCCESS in 18m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/13088ec5a5ab465ca3dbd412c2c65bd1 : SUCCESS in 24m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2513eb1fa30f4b8fa6a9f98e368c32bc : SUCCESS in 14m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/db6a9e9614bb480a9f0370777db6f111 : SUCCESS in 9m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4c47516f0aa42c484316e67ea4ba242 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8db7b07cac5c4fab8728324db0b32c12 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad3d97cd08264850adb77470b542e948 : SUCCESS in 6m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a53a7ac35a444469b6aa1040dc6db50 : FAILURE in 17m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/33dc279916044d9b9324cc1cd220012a : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5706793d21c54d0aa709643a3b992a36 : SUCCESS in 41m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4d9d7d14fec49cf80bb01e4708196b8 : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2df10d1e7a440eea274260d5c1d4248 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6a97208e92843fcb76737cf09b2a6d3 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":5},{"id":"cdcc57d47d481d76487f489ed838f856fa6f7aaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 20:36:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21f0ddcd5e6a40ad88680642cad1bf07\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0fee49e073ae4047a850c11d81a97b49 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":5},{"id":"ea41c25e17b52830cc9c383e8ffb814af24c3ce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-02 13:27:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"7e6bc0f07f58b0ed993b75da3c4c249b2916443b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:01:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1e277ec6a24f468f9039dd8f123f40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d806193f1b49238075f2256625fdab : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40a30da159ea43dd84cdb97543bacfb4 : SUCCESS in 4m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05088b1909cb4102b06b9f349d67bb58 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e21040302dc14693960a4a3a1e8f0c37 : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e29c3507a054850a33affebbb934ac7 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4746321e1944664a187ff41a1d71f24 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08f36bcfa3244da99bbbd52b0de691bd : SUCCESS in 29m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5ff44321b8ee47a09154650ef6421646 : SUCCESS in 26m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e97f19d7ceab49c7b41f6291558fa240 : SUCCESS in 13m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ac74eb40e66f4de79e0d24b9b79d00ca : SUCCESS in 22m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b84ee537fb648f8ba60bdfa32f99675 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/80797995684a45638842b5d1003b2a31 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1780861911664102b4c21466247625f2 : SUCCESS in 18m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fe24ad3585945e086275cb58f271ffa : SUCCESS in 9m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc8c9bbb9b0a4820aa42d1bd85bfddbb : FAILURE in 11m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/02f404d61bb841338984a7a065b7642d : SUCCESS in 12m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a6db486f5fa47c79b097a150b71818c : SUCCESS in 1h 05m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2381d686b67c4b5896ac5d044d100051 : SUCCESS in 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1c1ad1e7c041f7bfb631132248c4a9 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78424e4644994d5e81ccedb8ac159697 : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":6},{"id":"25ef0c26560ca6a3434fbf904e0dd4064139ee7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:57:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eec8632caff468dadaee711508e2b62\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42f108442399406ab11b819bbd26a767 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6d2752dfd2d6bb6ab4a81853a15461964b9be463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:21:52.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"69d553458b40e55d474d87191bde6ffc888f3cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:47:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b44c3da4e8f441b82e142c1be509b87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885da61f96d7415fa8b90174cc2852f0 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/298ecc61d4bb41bd97f7158e4c2e89a1 : SUCCESS in 10m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91e0361733cc4fe3b92f02d5673fc0bf : SUCCESS in 13m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db5e2474a7b34975aa123fff6b6502f9 : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae6255c0172d4ec38a85cc6ce6fe0d48 : SUCCESS in 16m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b06395c2c7e64445a34c6ee9885ea652 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2afcf66627f64565b86e4bc2479e0823 : FAILURE in 6m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/536591cc0e1745be8a64d29abdca9c90 : FAILURE in 22m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/19ed195102944d46b1090067928376f2 : FAILURE in 31m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e5ab8c9ae7cd45d5b2e709354912672f : FAILURE in 21m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c3e837e44922481e8a8358334fb32e00 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/868d667c84e14fa5b9ed8803dd556cb1 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cf6b41153c6f4745a1dff9c5d622b835 : SUCCESS in 11m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ccac6cbbd5b04bf1a32f5a6dd91c6628 : SUCCESS in 7m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6d88a6e216f4f999d51b26fe0541621 : FAILURE in 23m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b8dfff01451a497e98dc881653f595f6 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb1b8c139ded4453aeeeed0b4f22e535 : FAILURE in 1h 04m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/668f4c47a03c4446b02d3b0b738e437d : FAILURE in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69728083cf254f618c089e8199e27207 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b92ececbecf2473e981ad3a95d0c4dde : SUCCESS in 23m 15s","accounts_in_message":[],"_revision_number":7},{"id":"603336a06334a3f1130b28fc54c8ea861f1f5480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-12 05:48:55.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"8f2dc30d1323baa912f45b2ee092a356b9bc183d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 07:02:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a590aa187df64edda9b7a86a65d5ad2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052bb4ecc750478db45ed03d5d72a409 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eccf340cfdaa44d5aa0d976a4bb0da01 : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5e09639fa8342ab9b86f71bab8d81ee : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75bd6605e0194969966fb18bbb3fd71d : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3df65fe8785940568e905041241aa306 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa6068a937524f05b52625dab2bc8134 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5a2a6a68d1f74353aecb2fda7123e238 : FAILURE in 5m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/95dc0b4e8a3a460fa09a2aadc4f33971 : SUCCESS in 19m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8277d68b8f984fb98dc17acafea8e1a4 : SUCCESS in 34m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fc2c80b48b914a09a35d7fb672909963 : SUCCESS in 21m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/485f9b1d2bc8459bad462353e5b3fe4e : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/551d6ee814dd484599a0b6aa9d95a87a : SUCCESS in 15m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b579680ce6a8469eacb98712abc31ede : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/711f81ad5e3d4fa19d12450faa07d5c6 : SUCCESS in 8m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2804d279dae4e02810153e4a7296771 : FAILURE in 19m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e8891773292b4f6c9d85f477753d6647 : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/927e41d70dcc4a1daaaacd9919124738 : SUCCESS in 58m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa244980dc4443ebcc06ae55bf7fb5a : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55b4a822e4ad4218b7f0fa263230f81f : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/504cc0bfe1c743c7805aa590c5c683d2 : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":8},{"id":"3c35e1ed02badaecc42406d22d0b22bcf830255c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:05:49.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb6866f36d524333847071e5a4373d58\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbdc567501e446e0a9081a383c929a17 : POST_FAILURE in 1h 03m 44s","accounts_in_message":[],"_revision_number":8},{"id":"0feda05f7e3a9f1c287839e0abe0e67428d68a5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:22:52.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"43c3c39afcd7a4cd5140167acb42df665400a5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:10:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5bed0d7b9d44266ad0a3395859b1696\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45c84b2f50840d380dc7be9f244f381 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658cb52ac59946a9b8eca8cfa043a25e : SUCCESS in 4m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d43a5258a0e4da093589c1388a5bc4f : SUCCESS in 12m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cb2ff5a0d95446eba39dff51ddc961e : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6ba5bc532c54e1fbf9010d7a05d8521 : SUCCESS in 22m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea9aaff6dd7341a881c5aa827f5062e1 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3bca6fa56f1d43cfbc8c952382934fba : FAILURE in 8m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/de1148de11e64de8b32c423e2dbd7539 : SUCCESS in 10m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1d6a597219dd455493ce006f5a4cc508 : SUCCESS in 21m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/70242a48c18c43658264201111bad9bb : SUCCESS in 21m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/181219330c5843f695c40efa62e1f804 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fcfe93205d41474ca389c57007f4a076 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe8325381dba414bb922d06a706f6177 : SUCCESS in 9m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e977d904f1b2442091c63701bf461e49 : SUCCESS in 14m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3c72aaa80294770b0bf8e4c393b8c95 : FAILURE in 21m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/809d551a272447b3a6f68d0e959f7669 : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e564906cebd146c896a3e7ef9563b55a : SUCCESS in 58m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd7934e1f8f14b20a84021c7e88d9e80 : FAILURE in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ca485d4cf84f988d7170d35f78fdd4 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9015f80dd6b145b0b0e8f0626bee6a31 : SUCCESS in 10m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"08fae326503f4b03d1b17a1dcb3ca9e5dc270f21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 10:16:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7ae4838a1fc47cda4c565e5a1320ddc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21354a67ccaf44408296293a6a018a6d : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":9},{"id":"0fd629a866c8190d70af4c70e9f38fe61a385ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 12:01:36.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"330e86baa0e1f4fc85d570d6f981b347ffd5860e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 12:51:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba93b95dcc04614877aa96dfd04ec14\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e520af3f185a491f90b0bbf6d0f237e0 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":10},{"id":"eeb1e8245a657de5454b1e40a97d2247fa110da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 13:17:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e5c65f213a43bf9139f2bab0d3aef4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/671ccd3da1c3433599dd2848cf18361e : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/477c1415fb8a4c719e3466de6d0b926b : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2983152bf15d481da495ebb13f7107cf : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc7f6ced0a54488b58a3ba7b708b18b : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4726a2e1a1543d59111a21fb6ae2d4f : SUCCESS in 11m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51c50ca627264770885ba37629cd1708 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e8f5f761f92941edb192f20d8d4e7c56 : SUCCESS in 26m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/47e650f76b4847228cfd853004ac2235 : SUCCESS in 22m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cae45a3465f841f1affb3acc208608f2 : SUCCESS in 30m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4b44ff25f97641f7bdc387800eae696c : SUCCESS in 19m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c7ab60161a124a5eb6ecc6a1e96020b0 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b0244426743442e3beef26acbf550d22 : SUCCESS in 8m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/55f3e3b2cbe4402586547b5602768876 : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fca9ee7fe11744c0b7557276e07dc7f3 : SUCCESS in 7m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d306e21b8f38469fbbe39971d5a1c98e : FAILURE in 21m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec975ad6d55d4010b82958d8d08ead54 : SUCCESS in 8m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f3801d586d44149a76f7238d240f778 : SUCCESS in 59m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/793e240f24cc4b7699d55ea871ca800a : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5001cb7cc90491eba2f7e0857a1d983 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/801dc6c433da4564ab1db6d69149ae31 : SUCCESS in 19m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"29608746846622edd54550a61858c8c14db9c795","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-06 13:55:18.000000000","message":"Patch Set 9: Code-Review+1\n\nCopied votes on follow-up patch sets have been updated:\n* Code-Review+1 has been copied to patch set 10 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"69730a0fb1d6ba0dc783c9b29d5f322cfdbed713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-08 05:01:57.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"f7f41545a4d6fabb8e90edac45c61d52ff681b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 06:13:49.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d2cbd0dc2fd47afa21fcfb2fdef05ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b1d5ca493d4e4c9227f623c1052fea : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/578a8ca536e841a1a3efa03e8afe67de : SUCCESS in 4m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/285bba9bec384a01872bd9495300721e : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf5c4747aa4c435099dd67c0511cd179 : SUCCESS in 12m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05530896eba544e4a1c6cdccf03c09a3 : SUCCESS in 12m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c577b5cba9d547fd9beee3e247165098 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8b1c2ff862cd47788571f6b456a28659 : SUCCESS in 30m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/caceab24f95b4084bd8b74201d5d1a52 : SUCCESS in 23m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d7bc6edbb5404350a7053ddcf00c4016 : FAILURE in 15m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/59c316deba53433a860f1d344a6cb53b : FAILURE in 16m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b0bf044f04f54d53a2c91b053dfd6cea : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a22178b83c374160ac5c327f37b08b2d : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/45354f0bfbbc4910b468f031c14aecb3 : SUCCESS in 13m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c0f58ea6aff94392ae6017d28bba2c1b : RETRY_LIMIT in 4m 42s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0baa678534274f14ae6d9457ad9af28a : FAILURE in 10m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a74113a24304623885030335619e29a : SUCCESS in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d56f3b5ee3914014be1a08a43ed4feb1 : SUCCESS in 49m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edccaa14e28b49faa92e908c684ddc71 : SUCCESS in 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23fa10a24f6d49878857688ea975bda0 : SUCCESS in 2m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26482d43688f41439d6a3e28ee7c3459 : SUCCESS in 10m 35s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"9d5a0ce5e65e85b044ef8562840453ac8dd24dcf","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-08 06:39:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8b73e32b39021012d42ac3d819247394989bba37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 07:56:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58de7d91ab2745a3a394da77c5c5d09a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487fa07756c449a2970bed1bcdbb11a1 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8693597b02d48c1be30aecff772f63d : SUCCESS in 10m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01a8106a0751477c99bc0318e1b665c2 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06fd3638ee3642cc9629065e7af40da7 : SUCCESS in 11m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db811e0d95bf42368e1608de32e58e53 : SUCCESS in 23m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a5b0a3f9b8b4593aadd83dcc592974c : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/691b5b3936ee407dbe38a5e3673c4d13 : SUCCESS in 11m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/680a05cf71b3479ead2e8da4f6f6340c : SUCCESS in 13m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/17549fbfac714e74ad107c3a62fb829e : SUCCESS in 12m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6104915d59e44289a3b7ea91f1762e3e : SUCCESS in 22m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/636fe2a5fc614d7f871340ce0c9587f5 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/462820f2c0b749ebb6774b1f550d4921 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/31542f6f99a64d7aab2704e0a542ddd3 : SUCCESS in 14m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f51a7e9e12f349dc8ec3c5966540ff80 : RETRY_LIMIT in 6m 38s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0719a8dd6a19426a8ba3a9c843e94d38 : FAILURE in 22m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef3be282265d4c128a8799fdfc18eba1 : SUCCESS in 9m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3277a48e9a36429a97e4d93776a92139 : SUCCESS in 1h 04m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a153af215cf4df4add999309118e83f : SUCCESS in 30m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2724aab3e94379aee5d69a2d5e38ea : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e651a815a28a47528414f2331fba0dc4 : SUCCESS in 17m 16s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"e9acc4b215abba29c30f76670600b9100826de3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 08:00:59.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87806819879e4e4da0cce787514021c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1592fb043095400e8b9cb97dd9e90d62 : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":11},{"id":"48859bd8bdd317d1d4818c0fc7cf757eaaddd312","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-09 16:00:04.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"fe91030b2ed6f78a9e8968a24b78ade287fd3147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 07:29:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"bd1c0dbbc5c227ae6a3ff95cbb81b73eec762e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:48:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e89ad6ea4b16467594edd08d14ed594e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e87e61af80145ca82878bc9a54a51d3 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5962dcfc92db45aebba86c962fffaa1d : SUCCESS in 4m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3cd39f9a6e24fb29a40634be9701686 : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68631b1ed51b4031a21520cfafb97232 : SUCCESS in 22m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1327b9a1dc5d46879a34959006ed15dc : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d84f83a7fbf34dd9a8a46c5e2d26951f : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b61eadf4232a4f5e9f1cf5c9292ddec1 : SUCCESS in 26m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/300c9ddb28ac420ab5974b70f898d9bf : SUCCESS in 10m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/25d91ed1c5d8471fa21864d380bf7030 : SUCCESS in 32m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/62441234a26347c38adfa95239a6f8fa : SUCCESS in 11m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/350cd157cbc14a16819c1b45cd4d142d : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/827313016dfe49158212e98a705aa16f : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6a4bebe4c1f94d52888e5b85725ea7a1 : SUCCESS in 9m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/72a6091587cf4d42b83d537b2c1cb3fb : RETRY_LIMIT in 5m 57s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd3157c4a6034ec68d21d7a5e0cb9d96 : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbf03673baf94770ba07657b3fa60ec8 : SUCCESS in 11m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3c35166b8be40ef9e4951142527fd2c : SUCCESS in 1h 05m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/815c1862a8f14477b899884d6fb7f3ce : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce9ecb7123443b587bdb8394079f101 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61dd298dd119469b81991140692f5f82 : SUCCESS in 18m 50s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"055fca67230cd5ba2bb055e13e21dad5b957a8a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 09:46:08.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36dc0fc731734ef49eef7c049934f07c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32956273245144d0aaf0c62631ba455e : FAILURE in 1h 15m 17s","accounts_in_message":[],"_revision_number":12},{"id":"0930e4c91a4e6b5a1cb0b12e70415a5fcf20eba2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 10:44:32.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"b4075ecd362ffb13b7c39767ba2ad321d8e235bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-11 11:34:41.000000000","message":"Uploaded patch set 14: New patch set was added with same tree, parent tree, and commit message as Patch Set 13.","accounts_in_message":[],"_revision_number":14},{"id":"e1cbaab6e3e6dee37398288f3e6a93cb2b8c47d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:46:53.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c004dfb1fd6480d98fa00ded483a4f6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2e813791f6d4667a240978a05addbcb : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":14},{"id":"7a4b29c253b559bf1e3988b5cca52c1544f21106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 13:02:13.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e945c9570ea54858b7b313f0ac51ec2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e303422a810342df84c117219ce18a0c : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e35345cdec8464788e316dea45d8eb1 : SUCCESS in 8m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/153b7a45fc2a46b897c8b5000409b687 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c49456383aa456ea59fb5ce7c25b675 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9643a86ab324a2793242d743fdf74ba : SUCCESS in 18m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ad11d7d73d94ad5b774bf5cf9fbdc91 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0fea3e2914044e06ad2ffe5cf0c08ea8 : SUCCESS in 11m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4be0bdaa29804814b290b4026f6d5fa5 : SUCCESS in 15m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9ea5660858374a4bb27cfd96619f8266 : SUCCESS in 28m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dc9f261ee10f4b69af78d628210b60bc : SUCCESS in 16m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c075982a478948fb8a6ef2c094fc32a9 : SUCCESS in 10m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3319a9ff2ab546439f4e3e588ffedbea : SUCCESS in 15m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/75a28a8090c04358aed474cf4e5e8d41 : SUCCESS in 13m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3b1767d159c4800871a790753503447 : RETRY_LIMIT in 3m 40s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10e67f5943d34466a0c81a17df8eacc7 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a35d687b2d9c4ec1823f195a5b7f8da7 : SUCCESS in 9m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bff98b33bff64c4aa8a4a393855c61aa : SUCCESS in 1h 01m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1ec22ec6a2400ca626d04bdc90ff06 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f4e547317545f8b88ba1013708cbbe : SUCCESS in 3m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fbf0cbe145d49139c57acc6ddb26290 : SUCCESS in 10m 03s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"65af980c80559ea1c1e82fbcdb0717cf6ffc8010","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-07-21 10:56:22.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"3e64fdc8f048f675564c58d915151c59f45b5094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 06:02:18.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"ddb42a404fe5a03fc50d31134b0f32f710169cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:15:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360f637620594c12a2c279afd8ee8cb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab9f89407f141118efaad30dc31c2d7 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82aca30d9f0446469382678914b9099e : SUCCESS in 11m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3dbe802b5bb74ea19a228b5944c711ea : SUCCESS in 11m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c89a29bf0c54d49bb65373e6cf6efd3 : SUCCESS in 12m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bd53bfaf5974928a8e4d720b2ddd8ba : SUCCESS in 13m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e5f0a0b52834cb2836c181e3f148144 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4fb12aa6d3d43a48a9c84a75d33a8a0 : SUCCESS in 12m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fc2ca891255d4698ae90a1e9492613e4 : SUCCESS in 22m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ae77445897b349bf91150b94db03fc67 : SUCCESS in 37m 35s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/30c8df91d21a4b4987dd83d9011f1036 : SUCCESS in 22m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/889a40a95e7047fd90102806b7ccd96f : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/09a6ebc0fac74be88ad18ae27fb5843f : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/73258e22965d4eb1adf60bb7cbff4c1c : SUCCESS in 12m 54s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ac5dcabeaae45639064399538c3eaf8 : SUCCESS in 21m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d36f13646cd14e2b9bd3cc6293d5226e : SUCCESS in 4m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8f394aa8e714616b35d230d432aabd6 : FAILURE in 7m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/682474aa246f486cbc75ff4750d5cc06 : SUCCESS in 5m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a07e42076b7d408193cec68c6830aaf9 : SUCCESS in 1h 04m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c7b9c3d16c9487d88f6dd89f5a24ba4 : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5426b1331be4876a0e2b365b691fdb4 : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c97ea39d2b14976a34b151a05345ede : SUCCESS in 20m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"81bb360d01dc64b781489b89e3176eb701747f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 07:26:39.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent tree, and commit message as Patch Set 15.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"3b2b45ec10b80e2502c66ac0334f31238862cff4","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-21 07:54:26.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"04a21dc7e97728c7a900a92b99709beed6e441a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 09:06:34.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdf5edb1ddc046a3a5d7f813ddba0466\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/178213e50cba4a0198ec61c47c6502e7 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":16},{"id":"15c7cfe3d37f7e7fb25d884e069d0950ae5ff1d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 09:13:09.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78058d5fbe8f482888f66f40bab54139\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3190989ced47839f21ef877ba5e541 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd40e43ed64b4057916b71201e07e470 : SUCCESS in 5m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a4bafb2d3714d80b1af5b1ca4d662a6 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b19b1f6c3efe41b3b1a035d56a6a505d : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe25434af8c74ae29b5562d70370f5bf : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca1b2234325d4e9aad770f8273ad5a3a : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f53f95853e4c46cabec7c3edc8bf87f4 : SUCCESS in 11m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6e280e073b214bd7b82d50c9cc53cc36 : SUCCESS in 11m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/065513f22cc44647a53ed5441f51e90b : SUCCESS in 35m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a3564fea4540407890edf5eeea96e025 : SUCCESS in 18m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/945ef0cf0f5a4827b445789fd317e211 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa160149eb1e44dfb9ad8d6f40f16e47 : SUCCESS in 13m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dcd16e0094554fb98fb09ee65c35f620 : SUCCESS in 15m 54s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bcad59820564c2bb3bce6c5e53d090f : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e83ce2ab276244ffb1f025b110258a16 : SUCCESS in 9m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a82ec9feb1943e38baa55aa6f2cdf01 : FAILURE in 4m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8cc79b34b47a4d3a9a1ad2b8f367d6bd : SUCCESS in 9m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9bdf472182b4469a48a6af83116c0bd : SUCCESS in 1h 05m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/150ed18197834be2ae5077a2d8a1e37b : SUCCESS in 30m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0acd264ab6c246df953e6c703572da73 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13a45b5759e84dee88a63c720babdf80 : SUCCESS in 27m 04s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"63f3ffb161e812674354392f70b6736fc74ca9fc","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-28 16:31:43.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"fdb6ed15c42fa1460284027cf4535b507c68a6b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 16:32:05.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"4984d4b90e6829261fe80eaf79e849911f4fbb16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:11:07.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6af4a0bb334341b4d1eecde9b25ca4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41615669afb480f862c465622eedba6 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74cad9653dca43d49557681c9790849d : SUCCESS in 9m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6e3b6f2386da46608e2093c629cfd305 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b9c9d0945b741428932e5998d20beed : SUCCESS in 25m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/96223e72d2f44fb2a4dee4fa4fdb4b83 : SUCCESS in 25m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d4404265e5764f24bd1dbf363c5b0684 : SUCCESS in 11m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bd938068340544999370218809d4e5c8 : SUCCESS in 21m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d981d037847c4f238333f69275dd9999 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35d3ec2d1bc447f191292dd58e1c6da7 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/09ddeedbe49747d0bd2d9cc453464dc2 : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53e7017ed1f34715bcdf8cd44dafaece : SUCCESS in 3m 52s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a2b999342aeb433ab8323d609915897b : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/46e8a30c0e50433f85a4b38eb376e450 : SUCCESS in 57m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd128c865e7a4c80b4bf331e0f37aa20 : SUCCESS in 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf690c4e2fa24ac99e99d1b79b1dcf3a : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c92da4f9ca0e4730a7f2a68a62ceb89b : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":16},{"id":"932c6dc9b33285ae95ba4a7e6a4b4c2ba1d21698","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:11:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"04c1cb172dde13e6c992639eac8f126bebd850e1","revisions":{"bd7e036389a63818eefd7c877906ea8967979991":{"kind":"REWORK","_number":1,"created":"2026-03-23 12:12:35.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/1"}}},"commit":{"parents":[{"commit":"6b02352e7db0ec8fbdfba95a3e39e639b37e74a9","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b02352e7db0ec8fbdfba95a3e39e639b37e74a9"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:56.000000000","tz":60},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7e036389a63818eefd7c877906ea8967979991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7e036389a63818eefd7c877906ea8967979991"}]},"branch":"refs/heads/feature/threaded"},"95ce4658d77a6c57ce71311894cb60cb03527b03":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-23 12:20:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/2"}}},"commit":{"parents":[{"commit":"4714a3057765977a72901bae09b29ac8de6c984f","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4714a3057765977a72901bae09b29ac8de6c984f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 12:19:45.000000000","tz":60},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ce4658d77a6c57ce71311894cb60cb03527b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ce4658d77a6c57ce71311894cb60cb03527b03"}]},"branch":"refs/heads/feature/threaded"},"99eb317481edf2db925ba89d52fd15c20b638c99":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2026-05-30 16:41:57.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/3"}}},"commit":{"parents":[{"commit":"6e7a2267535fde9f4fd466c60b7f6ca8f93b9abe","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e7a2267535fde9f4fd466c60b7f6ca8f93b9abe"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@nvidia.com","date":"2026-05-30 13:47:19.000000000","tz":0},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99eb317481edf2db925ba89d52fd15c20b638c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99eb317481edf2db925ba89d52fd15c20b638c99"}]},"branch":"refs/heads/feature/threaded"},"0905f29f97c5908b52c853939295e77c79c3c43a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-30 17:02:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/4"}}},"commit":{"parents":[{"commit":"2a5be5e390c53798c85aa7e9dd6245427420788a","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a5be5e390c53798c85aa7e9dd6245427420788a"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 17:01:09.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0905f29f97c5908b52c853939295e77c79c3c43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0905f29f97c5908b52c853939295e77c79c3c43a"}]},"branch":"refs/heads/feature/threaded"},"09f141dea4b8c5a4e501e54d06f8aa9b39769ea5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-30 18:24:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/5"}}},"commit":{"parents":[{"commit":"bcfb8e0752f393d7b0180923dd7418fbdd73ac36","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcfb8e0752f393d7b0180923dd7418fbdd73ac36"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-05-30 18:18:21.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09f141dea4b8c5a4e501e54d06f8aa9b39769ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09f141dea4b8c5a4e501e54d06f8aa9b39769ea5"}]},"branch":"refs/heads/feature/threaded"},"e1cf99e500a4de778dc9bb3342e193618c379d89":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-02 13:27:47.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/6"}}},"commit":{"parents":[{"commit":"db1331733e21a199c09b870c9363c1217befed81","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db1331733e21a199c09b870c9363c1217befed81"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-02 13:06:36.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1cf99e500a4de778dc9bb3342e193618c379d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1cf99e500a4de778dc9bb3342e193618c379d89"}]},"branch":"refs/heads/feature/threaded"},"d9064961a3e4cbff5fff0d6e54dcef3a82e08a3e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-11 13:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/7"}}},"commit":{"parents":[{"commit":"fb2d1fbc18173f56caa368560175b75b111bb894","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb2d1fbc18173f56caa368560175b75b111bb894"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 13:19:48.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9064961a3e4cbff5fff0d6e54dcef3a82e08a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9064961a3e4cbff5fff0d6e54dcef3a82e08a3e"}]},"branch":"refs/heads/feature/threaded"},"3de169e0fb429b20f342b78c24ce190852fe96a9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-12 05:48:55.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/8"}}},"commit":{"parents":[{"commit":"418635184cac029588fd3ed595bbd26c75fe4e0e","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/418635184cac029588fd3ed595bbd26c75fe4e0e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-11 17:37:53.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3de169e0fb429b20f342b78c24ce190852fe96a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3de169e0fb429b20f342b78c24ce190852fe96a9"}]},"branch":"refs/heads/feature/threaded"},"b4838596262a6193cb295b10b1e2c341e813281e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-07-06 07:22:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/9"}}},"commit":{"parents":[{"commit":"1bacaf817954b4ceb9b4349b3a9fb4e3b3f41ca3","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bacaf817954b4ceb9b4349b3a9fb4e3b3f41ca3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-03 13:01:47.000000000","tz":0},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4838596262a6193cb295b10b1e2c341e813281e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4838596262a6193cb295b10b1e2c341e813281e"}]},"branch":"refs/heads/feature/threaded"},"feb5bfc79f791ed2096e50fa16d030b3a2b8c653":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-07-06 12:01:36.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/10"}}},"commit":{"parents":[{"commit":"ad7607b043e8de14a468d95f864f9da1a5d14b99","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7607b043e8de14a468d95f864f9da1a5d14b99"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-06 11:57:16.000000000","tz":0},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit useds the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receivers while-loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/feb5bfc79f791ed2096e50fa16d030b3a2b8c653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/feb5bfc79f791ed2096e50fa16d030b3a2b8c653"}]},"branch":"refs/heads/feature/threaded"},"c48b2d503de7c06aff1b4cad957a51458263d9e2":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2026-07-08 05:01:57.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/11"}}},"commit":{"parents":[{"commit":"7c79eefdeeb15eab9f65994cace593876337c23e","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c79eefdeeb15eab9f65994cace593876337c23e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-07 13:26:01.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit uses the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receiver\u0027\u0027\u0027s while loop after the first successful read to\nallow sending a response afterwards.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c48b2d503de7c06aff1b4cad957a51458263d9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c48b2d503de7c06aff1b4cad957a51458263d9e2"}]},"branch":"refs/heads/feature/threaded"},"cf667f305f5ae29b14f84c13e2426f50d6c40f27":{"kind":"REWORK","_number":12,"created":"2026-07-11 07:29:34.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/12"}}},"commit":{"parents":[{"commit":"94bdab89d444552402eb586e3af1a5820367077f","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94bdab89d444552402eb586e3af1a5820367077f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 07:16:27.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit uses the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receiver\u0027s while loop after the first successful read to\nallow sending a response afterwards. Empty parser yields (input pausing\nexactly on a chunk-size boundary) are skipped rather than returned, as\nreturning them would be a false EOF before the chunk data arrives.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf667f305f5ae29b14f84c13e2426f50d6c40f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf667f305f5ae29b14f84c13e2426f50d6c40f27"}]},"branch":"refs/heads/feature/threaded"},"7a913f2412c86197b2da7515dcdd453126b528d1":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-07-11 10:44:32.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/13"}}},"commit":{"parents":[{"commit":"ef687f8c98ed58cdb4dc0ed41210165b456e98f1","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef687f8c98ed58cdb4dc0ed41210165b456e98f1"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 10:44:19.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit uses the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receiver\u0027s while loop after the first successful read to\nallow sending a response afterwards. Empty parser yields (input pausing\nexactly on a chunk-size boundary) are skipped rather than returned, as\nreturning them would be a false EOF before the chunk data arrives.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a913f2412c86197b2da7515dcdd453126b528d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a913f2412c86197b2da7515dcdd453126b528d1"}]},"branch":"refs/heads/feature/threaded"},"f3bc71f9b33219780b0b9d477972c0975a4f09b3":{"kind":"NO_CHANGE","_number":14,"created":"2026-07-11 11:34:41.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/14"}}},"commit":{"parents":[{"commit":"062428bdcec85b5acce460e83bc9d4d8e0475a24","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/062428bdcec85b5acce460e83bc9d4d8e0475a24"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-07-11 11:34:32.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit uses the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receiver\u0027s while loop after the first successful read to\nallow sending a response afterwards. Empty parser yields (input pausing\nexactly on a chunk-size boundary) are skipped rather than returned, as\nreturning them would be a false EOF before the chunk data arrives.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3bc71f9b33219780b0b9d477972c0975a4f09b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3bc71f9b33219780b0b9d477972c0975a4f09b3"}]},"branch":"refs/heads/feature/threaded"},"09313716ecdeada110944e23101026c5b05ec6a0":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-08-21 06:02:18.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/15"}}},"commit":{"parents":[{"commit":"25de1858f292d643d8dd6214dda42886bed06049","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25de1858f292d643d8dd6214dda42886bed06049"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-20 14:23:11.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit uses the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receiver\u0027s while loop after the first successful read to\nallow sending a response afterwards. Empty parser yields (input pausing\nexactly on a chunk-size boundary) are skipped rather than returned, as\nreturning them would be a false EOF before the chunk data arrives.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09313716ecdeada110944e23101026c5b05ec6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09313716ecdeada110944e23101026c5b05ec6a0"}]},"branch":"refs/heads/feature/threaded"},"04c1cb172dde13e6c992639eac8f126bebd850e1":{"kind":"NO_CHANGE","_number":16,"created":"2026-08-21 07:26:39.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/50/981750/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/981750/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/981750/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/981750/16"}}},"commit":{"parents":[{"commit":"0e53bdf5d5b44e839867904ff35eb83ef898f07f","subject":"Patch gunicorn to speed up chunked requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e53bdf5d5b44e839867904ff35eb83ef898f07f"}]}],"author":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-03-16 15:48:58.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-08-21 06:51:55.000000000","tz":120},"subject":"Patch gunicorn to unblock chunked ssync reads","message":"Patch gunicorn to unblock chunked ssync reads\n\nGunicorn reads chunks until the requested buffer is filled. When using\nssync the sender waits for a response before continuing, while the\nreader tries to get more data before responding. Eventually they end up\nin a dead-lock situation, waiting for each other.\n\nThis commit uses the code from gunicorn.http.body.ChunkedReader.read,\nbut breaks the receiver\u0027s while loop after the first successful read to\nallow sending a response afterwards. Empty parser yields (input pausing\nexactly on a chunk-size boundary) are skipped rather than returned, as\nreturning them would be a false EOF before the chunk data arrives.\n\nChange-Id: Idb6eae3f000aa90c7fa39688cbfbc54c409163f4\nSigned-off-by: Christian Schwede \u003ccschwede@mailbox.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04c1cb172dde13e6c992639eac8f126bebd850e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04c1cb172dde13e6c992639eac8f126bebd850e1"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
