)]}'
{"id":"openstack%2Fswift~981751","triplet_id":"openstack%2Fswift~feature%2Fthreaded~Ia279a22726afe59ba46e97a4818b5177a2f41553","project":"openstack/swift","branch":"feature/threaded","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-06-11 13:41:38.000000000","reason":"\u003cGERRIT_ACCOUNT_38496\u003e replied on the change","reason_account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-07-06 07:59:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia279a22726afe59ba46e97a4818b5177a2f41553","subject":"Add a wrapper to mimic wsgi.Input for ssync receiver unit tests","status":"ABANDONED","created":"2026-03-23 12:12:35.000000000","updated":"2026-07-06 07:59:04.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f230032e12c95358c8d19ccfa48b01c7bc743856","_number":981751,"virtual_id_number":981751,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-23 14:01:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"all":[{"value":1,"date":"2026-06-03 15:56:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 13:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 15:56:01.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"}],"messages":[{"id":"59206f9201e32f9d55cba20ba1ed614dd48fbec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:12:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dc7a38c6cce81ccc5eca487558380725d704159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 12:20:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"bd7b0180210ffe7b915680dac91f716ee419af40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:10:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bce8074093fd440489db6f3c0090b1ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3c9149bf064bc38773b40e187cc7cd : SUCCESS in 10m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f11ca4207db54878b784ac34a212ff74 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cc1e19d8ec94c9b919c564dd01ee9da : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7890b48be8d143ad87acb0db1b071899 : SUCCESS in 10m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/674f1541ab17417eb1606c96aa7b649f : FAILURE in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d209d78f9a674db1a177fd9652977fc6 : SUCCESS in 26m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b0957562830d45fbb5380f2584a2eda2 : SUCCESS in 21m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d4fe5e9b684c48248ba0e971ea1cbf42 : SUCCESS in 21m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/51468042d4a240dda387aefca3edfb3a : SUCCESS in 32m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b8368a339164eef88e132a6eb663ad2 : FAILURE in 16m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b23970896de430c9d660ee15cfa7150 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/35f3f0eae13f4fb3ac9cf376a9245bed : SUCCESS in 19m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2fa9d7fe06d4d50ac40dee4f559990a : SUCCESS in 24m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3009db61e7044ce94bedd10551204fd : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca6b19a301b041ff812185d14e8816c6 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":2},{"id":"36e715dd035e36adf1d6a93f68d532afb77e5eae","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-03-23 13:24:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92da458186db2b2e389208e2f5915ffcf9f68564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 14:01:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03fdd0588c364dda90fe63b6a06d224e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71645791701d46fca7304e0f1424ebea : SUCCESS in 10m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c85e5a2745cc4f93aecfe57dedbb3764 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1c2d6b0d147483291bacaf808df47bd : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddbf48d3b32f44d49b638a389c7483d6 : SUCCESS in 12m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78e63fe4fa184673891a3abae24b5b07 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/90ada2c62f3d4780bab5460e49fcc7d0 : SUCCESS in 27m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f752eba2db0b4a5991807c76d8db2649 : SUCCESS in 8m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed95c282898c4f2584a96811a2181456 : SUCCESS in 13m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/83cf2c9dffee4b99ad3fbdba303fb4be : SUCCESS in 30m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/889af9fbde734d95967adb975a7bd92c : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/868a97a0e97d450886258160e2cd953c : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/405f6db01dee4524baf1ac49127fcd06 : SUCCESS in 15m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3638446ef2c47cc9b2c86417a53b822 : SUCCESS in 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed9841036464ed0acd3b2928bbccfa6 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f7e3c73b0fd449e9651a35d4c21c5fd : SUCCESS in 20m 20s","accounts_in_message":[],"_revision_number":2},{"id":"16b1531f89022f13fa9b3c8a44a1e0b82874d1af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 14:53:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eb52d5ce36e4084b6eee383dd0f7220\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/433ce800b05a480696da22928c8a6dc4 : FAILURE in 54m 46s","accounts_in_message":[],"_revision_number":2},{"id":"431ff7fc40bc92796bdeb9a776b406d571d1b918","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-03 15:56:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aa9064e16b580442e189e14789372ebd0b3421d8","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-11 13:39:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"36c686856c3ba103683669a468f764bc236bba3d","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-06-11 13:41:38.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f230032e12c95358c8d19ccfa48b01c7bc743856","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-06 07:59:04.000000000","message":"Abandoned\n\nReplaced by recent refactoring","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e3a08d026a5315f757635aa68f62e99b7b5400f","revisions":{"54779efbf0172c1b64a90a8892c6d308a26841a0":{"kind":"REWORK","_number":1,"created":"2026-03-23 12:12:35.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/51/981751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/981751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/981751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/981751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/981751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/981751/1"}}},"commit":{"parents":[{"commit":"bd7e036389a63818eefd7c877906ea8967979991","subject":"Patch gunicorn to unblock chunked ssync reads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7e036389a63818eefd7c877906ea8967979991"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-20 10:26:52.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 10:54:56.000000000","tz":60},"subject":"Add a wrapper to mimic wsgi.Input for ssync receiver unit tests","message":"Add a wrapper to mimic wsgi.Input for ssync receiver unit tests\n\nssync unittests need a wsgi.Input wrapper to use gunicorns\nChunkedReader/Body pipeline when not using evenlet.\n\nChange-Id: Ia279a22726afe59ba46e97a4818b5177a2f41553\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54779efbf0172c1b64a90a8892c6d308a26841a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54779efbf0172c1b64a90a8892c6d308a26841a0"}]},"branch":"refs/heads/feature/threaded"},"1e3a08d026a5315f757635aa68f62e99b7b5400f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-23 12:20:54.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/51/981751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/981751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/981751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/981751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/981751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/981751/2"}}},"commit":{"parents":[{"commit":"95ce4658d77a6c57ce71311894cb60cb03527b03","subject":"Patch gunicorn to unblock chunked ssync reads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ce4658d77a6c57ce71311894cb60cb03527b03"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-20 10:26:52.000000000","tz":60},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2026-03-23 12:19:45.000000000","tz":60},"subject":"Add a wrapper to mimic wsgi.Input for ssync receiver unit tests","message":"Add a wrapper to mimic wsgi.Input for ssync receiver unit tests\n\nssync unittests need a wsgi.Input wrapper to use gunicorns\nChunkedReader/Body pipeline when not using evenlet.\n\nChange-Id: Ia279a22726afe59ba46e97a4818b5177a2f41553\nSigned-off-by: Christian Schwede \u003ccschwede@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e3a08d026a5315f757635aa68f62e99b7b5400f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e3a08d026a5315f757635aa68f62e99b7b5400f"}]},"branch":"refs/heads/feature/threaded"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
