)]}'
{"id":"openstack%2Fswift~981825","triplet_id":"openstack%2Fswift~master~Ib776ef1f00ac6bc9b409027abe1943239a9a66e2","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-23 21:39:40.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib776ef1f00ac6bc9b409027abe1943239a9a66e2","subject":"WIP: timestamp/constraints wrangling","status":"NEW","created":"2026-03-23 19:19:54.000000000","updated":"2026-03-24 13:52:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b78701172b21799cb389dc35f2e5fca16314088","_number":981825,"virtual_id_number":981825,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-24 13:52:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 21:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f7c165693e59d38ef33c32c5f3720525fdfcad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 19:19:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12dad66252656679dace52aba9c00ef0c4d8f810","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 19:21:06.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"23b958afbb362bae7f6961275b4b6077385c7462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-23 19:31:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77e211429176826dbb07bcd613e66728a0e25ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 21:39:40.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3258fca2b0be432ab24d97b49d21b007\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8264923800445789b051e31f4d25b2c : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/104582fe21b4446a9a900dafcc8c934d : FAILURE in 33m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33d05ff8a55f42868388f364ca05bbf8 : FAILURE in 27m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69949bcfac4e4985a622dde624eb488e : FAILURE in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9c55c7a954b45ce89601e5fe1ab9c73 : POST_FAILURE in 15m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/096c4f798e414a529157cc85c88d305d : FAILURE in 4m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63caa1cfea904feebd4f891c77e0a265 : FAILURE in 5m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffffae0f8f56432887a2124803cb77b0 : FAILURE in 7m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25d35a4d9e324ea89f700f0e0e17f327 : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79bc9e18f81747789cff606353d15ac6 : FAILURE in 41m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a36cd26c84bd43ed920e2d7d76ff6bc8 : FAILURE in 43m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7e557a052a504eb8a0dbb176c6c74568 : FAILURE in 30m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/aa5e80a361cc4da9818dc45f130cd02e : FAILURE in 54m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/748851ba65e548d1a9949219f400e52b : FAILURE in 3m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/17b922075a7246738a277a7b813c71e9 : FAILURE in 3m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bcac6e15c2214e8e9975a61babfe6a3f : FAILURE in 2m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c78228bad32c4a2fadab40166f9587bb : FAILURE in 5m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bf15eed7d7e4c9f85de748251f104cc : RETRY_LIMIT in 3m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c3dc1a0cde140429b8464f27b98549e : RETRY_LIMIT in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/582bc24034d6489286fd0cb06a79a461 : TIMED_OUT in 2h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5e368397dc34ceb9bfe8e89aa6a3bfe : RETRY_LIMIT in 8m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/06a19fc2d46b4024b499f3b305bbf989 : RETRY_LIMIT in 8m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e17d57ddc5249d0ad99a4c8b3351daf : FAILURE in 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5464944455bd41e09b5f43f971cc0247 : FAILURE in 2m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4ed9a8e3d2f41ee9050b5f209527a36 : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/606ad3af68af4b24a52743aeacd72863 : FAILURE in 24m 35s","accounts_in_message":[],"_revision_number":2},{"id":"caf0c55afc189874492a3f314b3b7c3491067ee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 21:40:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eedaf67ced2c4bedb4b93b8e1f82a6fc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41cfbfe3bccd43c99b2dc1a188b8b61a : FAILURE in 1h 32m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0ca968ad35b2369f7729b435209a20046febf8b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 12:36:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"44fa3cfa08434299c9f4169ad1e39508fe7c0406","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 13:22:49.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0777f20b9234ba6bd3ee90967851f10\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13701cd0f9d74d29af93afb0e673904b : FAILURE in 44m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9b78701172b21799cb389dc35f2e5fca16314088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 13:52:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b02a726c56634736ad43a3de9a578074\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45e3088616b47c194b6b023f53c5d9a : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7ea79e9c07664d208cc36ab2ff15ad76 : SUCCESS in 31m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a36cad4ed6e402e9db4c968a18fe7c1 : POST_FAILURE in 1h 05m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d85488abb40f4959ae318da336776390 : POST_FAILURE in 55m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2901aa234c69482ab7f52a11f4a3897a : POST_FAILURE in 59m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f4ae57fb9584fd1a8920069279ae81b : FAILURE in 4m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60bd55701ca14b0a9f1d09968bf2a69f : POST_FAILURE in 3m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41b42d7c8408474d99af2da24894ac11 : POST_FAILURE in 4m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36b428ec348a4e02aa7bb0ce37307dda : POST_FAILURE in 7m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ced85e9dba33418a9877a32659d51fd5 : FAILURE in 59m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/765d99f1db4548a8a68fc170cbf72aa6 : FAILURE in 39m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6374f8db476443328461eed2901f07a8 : FAILURE in 52m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6cbb77e8c92c456aa0fba337d32a45fb : FAILURE in 35m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ced36563586a4d04ad87cba39e4ceed8 : FAILURE in 3m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30f65914c16e4466abb509dfa5ee8dfa : FAILURE in 4m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e4420277f2245669d6aa8141575968d : POST_FAILURE in 3m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aabad47f35124d2d8992c24bcc186250 : SUCCESS in 6m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/560fe66422ca4477ba8305d30ebaa4ca : POST_FAILURE in 15m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/48723198b5b64679b6aa729a61b9d714 : FAILURE in 11m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/58b02b073cb9411cba60e5a763a9b39c : SUCCESS in 50m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc286066ea404e3b8528ca3cca315b27 : FAILURE in 32m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/799edecc1e5a41138beeef871711da72 : POST_FAILURE in 19m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/029af64c4c784d018156c3423b79115c : TIMED_OUT in 41m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/311be26683a54c369b1544161f1adf2d : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8164cb3b28814cd9a1952de9c06e3a6a : POST_FAILURE in 10m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8c186e3f6f04a87aa9ba4b03101f7f4 : POST_FAILURE in 1h 08m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"15168b188d685b0bb00e684257f9c5c0b3b5a48e","revisions":{"b783394863257e7a5a867fc9218da96625e7be16":{"kind":"REWORK","_number":1,"created":"2026-03-23 19:19:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/981825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/981825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/981825/1"}}},"commit":{"parents":[{"commit":"1d6813c656835bc81acba9504c428c5c41e988c8","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d6813c656835bc81acba9504c428c5c41e988c8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 19:17:38.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 19:17:38.000000000","tz":0},"subject":"WIP: timestamp/constraints wrangling","message":"WIP: timestamp/constraints wrangling\n\nI haven\u0027t figured out how break the circular import yet. It\u0027s hard to\nimport other swift stuff into swift.common.utils.timestamp!!\n\nChange-Id: Ib776ef1f00ac6bc9b409027abe1943239a9a66e2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b783394863257e7a5a867fc9218da96625e7be16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b783394863257e7a5a867fc9218da96625e7be16"}]},"branch":"refs/heads/master"},"88675146871debd5af64f926603a3545b5ef3ba9":{"kind":"REWORK","_number":2,"created":"2026-03-23 19:31:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/981825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/981825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/981825/2"}}},"commit":{"parents":[{"commit":"1d6813c656835bc81acba9504c428c5c41e988c8","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d6813c656835bc81acba9504c428c5c41e988c8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 19:17:38.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 19:29:17.000000000","tz":0},"subject":"WIP: timestamp/constraints wrangling","message":"WIP: timestamp/constraints wrangling\n\nTODO: make me better\n\nIt\u0027s hard to import other swift stuff into\nswift.common.utils.timestamp!!\n\nChange-Id: Ib776ef1f00ac6bc9b409027abe1943239a9a66e2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88675146871debd5af64f926603a3545b5ef3ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88675146871debd5af64f926603a3545b5ef3ba9"}]},"branch":"refs/heads/master"},"15168b188d685b0bb00e684257f9c5c0b3b5a48e":{"kind":"REWORK","_number":3,"created":"2026-03-24 12:36:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/981825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/981825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/981825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/981825/3"}}},"commit":{"parents":[{"commit":"1d6813c656835bc81acba9504c428c5c41e988c8","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d6813c656835bc81acba9504c428c5c41e988c8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-23 19:17:38.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 12:33:11.000000000","tz":0},"subject":"WIP: timestamp/constraints wrangling","message":"WIP: timestamp/constraints wrangling\n\nTODO: make me better\n\nTODO: with USE_V2_TIMESTAMPS\u003dFalse a *lot* of tests fail because they\nare mocking the wrong timestamp class.\n\nIt\u0027s hard to import other swift stuff into\nswift.common.utils.timestamp!!\n\nChange-Id: Ib776ef1f00ac6bc9b409027abe1943239a9a66e2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15168b188d685b0bb00e684257f9c5c0b3b5a48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15168b188d685b0bb00e684257f9c5c0b3b5a48e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
