)]}'
{"id":"openstack%2Fswift~982735","triplet_id":"openstack%2Fswift~master~Ifdf6b7cb527cb934eb77ddd666d89cc472f0a1b1","project":"openstack/swift","branch":"master","topic":"just-raise-diskfileerror","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-03-31 16:46:49.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-03-31 06:05:22.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ifdf6b7cb527cb934eb77ddd666d89cc472f0a1b1","subject":"raise DiskfileError on \"bad\" useful_bucket","status":"NEW","created":"2026-03-31 00:04:59.000000000","updated":"2026-03-31 16:46:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0ccf126d1f5f76474a02f4160013cf94c1d31b2b","_number":982735,"virtual_id_number":982735,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-31 01:13:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2026-03-31 16:46:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 00:04:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2026-03-31 00:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-31 01:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f085b46da0b0ddc5b67eb9404eca59e3e7b3e948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-31 00:04:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47fa12606c7ce602f7ae23e440728cb2fce696bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 00:35:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604818f50d9a495c88ba2378331ba675\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3b9cf13a37f749df8c38e873e74cb698 : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ad9fbdda6a2a6671b3ff6d7301957272f9a43ac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 01:13:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ee7717b7c7433e9006c6a82d231d39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899e45ff3d254fcda03d90470fd2a3df : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/004b71a1ac9d43bc80ff4c49d262c77c : SUCCESS in 41m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79ea5f5a64da4fdca3d4eb09bbb91757 : SUCCESS in 41m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ec39570ed40427db1ee61f649951865 : SUCCESS in 54m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ced0ccd42a94044ad2ba9e21954d9ad : SUCCESS in 36m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce8e2cd6d0d546faa337e29f1ed817c3 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c14a7c06b95a4e6facaeab890368cc87 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2882b20ebd724a75a95a4283eeb7d157 : SUCCESS in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53cda16ada80442593ddd34be2eca6ba : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6bb2318b59204782836536da560f41d6 : SUCCESS in 12m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d301f7300fea48b8a169bc88ebeee9b7 : SUCCESS in 8m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/240f3dcf968f46e487faf8c19e7f59bb : SUCCESS in 28m 18s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/efb925eb1b5e4f7ea0fa8b75a9593510 : SUCCESS in 13m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ad045b77e5bf4098b3ae2f7cd6e9c728 : SUCCESS in 9m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44cc9e3b7e3140e9aa97e93c13b1d7f0 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7496dffda53b4416a1196f934a464da7 : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2758ca1685f4b84bb184c2073dca748 : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/19a420235baf4bb99f141e1f1e8af451 : SUCCESS in 13m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/74ac36cc33884659a812dd4899733a1b : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8f9be44e0984319b6ce3818188338dc : SUCCESS in 48m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3d00ca45ff14a3b97a50aca8165c2f7 : SUCCESS in 54m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/586d412e05a34cc98345c600ef8b8d57 : SUCCESS in 56m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f710c1aa6064cefbe35cba2f9920c88 : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d1308ff6174fb9a13defb2cbfda068 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7286a48231144da09a93d08351b9111e : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b33b2c5f0e884b93ad55467818b10848 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e25b47bf8b7fc0affc9f2093e76fc301ea363149","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-31 06:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ccf126d1f5f76474a02f4160013cf94c1d31b2b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-31 16:46:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e2f36fff3d1fe8c19e730e8ae58617e820b1bab","revisions":{"4e2f36fff3d1fe8c19e730e8ae58617e820b1bab":{"kind":"REWORK","_number":1,"created":"2026-03-31 00:04:59.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/35/982735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/982735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/982735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/982735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/982735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/982735/1"}}},"commit":{"parents":[{"commit":"d0b321a8e6a197531abca9f442fd9c5538f10d62","subject":"reconstructor: log warning on timestamp/etag mismatch during rebuild","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0b321a8e6a197531abca9f442fd9c5538f10d62"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-15 21:37:04.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-03-30 23:37:10.000000000","tz":-300},"subject":"raise DiskfileError on \"bad\" useful_bucket","message":"raise DiskfileError on \"bad\" useful_bucket\n\nCloses-Bug: #2143202\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ifdf6b7cb527cb934eb77ddd666d89cc472f0a1b1\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e2f36fff3d1fe8c19e730e8ae58617e820b1bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e2f36fff3d1fe8c19e730e8ae58617e820b1bab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
